)]}'
{"id":"openstack%2Fnova~829386","triplet_id":"openstack%2Fnova~master~I914ee01071b2b3f5c0c11de2171a7405c199ddb8","project":"openstack/nova","branch":"master","topic":"bug/1942329","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-04-25 13:03:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I914ee01071b2b3f5c0c11de2171a7405c199ddb8","subject":"DNM: trigger nova tests with the neutron fix included","status":"ABANDONED","created":"2022-02-15 17:28:30.000000000","updated":"2022-04-25 13:03:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5bb199db897211f3358f5e88a94fa08eced76221","_number":829386,"virtual_id_number":829386,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-15 19:03:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 19:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 19:34:35.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-15 19:57:23.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"}],"messages":[{"id":"54f1e1f616167ada2a86321d7f1e6c8dac85fcc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-15 17:28:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d4dc5645ee05ba20310c9c10c39f1a66cdec6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:03:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0d9497e0c7a14b7dbcc6e18e7d5f97cd : SUCCESS in 1h 23m 55s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cf016051e7f42b49ee830a418c8a730 : FAILURE in 1h 31m 57s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36e2b8e25c8a4ade8891f1dee54c80ce : SUCCESS in 48m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43856ca66488434ba9566acd76667997 : FAILURE in 19m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd9e2aa1429741baae2bc2a53b531be4 : FAILURE in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302eb0ca9d4c47dbbf4ead2f309ee423 : FAILURE in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30097340f63c420c9faac3a5ae4d8e6c : FAILURE in 16m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9e2eed8bec0426faa70af978ccf9792 : FAILURE in 18m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c7876ed07c4507b665a1f7d922829d : SUCCESS in 12m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/962c9f4249e64e08b117ae708594e144 : SUCCESS in 1h 12m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21b4865305e7484ca126b869690a90e7 : SUCCESS in 1h 19m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d9e933583c87472e82f9032aed5e6b41 : SUCCESS in 50m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d172942d7b13473d97fb09aa6b963a11 : SUCCESS in 53m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c55c3efb897b44f3b38b2b39eec17d4c : SUCCESS in 1h 27m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/1566bd29e55e4fb88eb8fb7d39273687 : SUCCESS in 1h 11m 33s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/06048a52f5344bc4ac35a699906d252e : SUCCESS in 39m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0772eb9a985c4060b2afd99726d9beac : SUCCESS in 6m 03s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/cf5bc5a79f37415f866b6f4406f8b2e7 : SUCCESS in 23m 46s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4f875766de241ad8d78806592c9249c : SUCCESS in 20m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d315031d592243c1b5cdd7840311ff3f : SUCCESS in 22m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/68b7c5d7c0234399a2403748d939d05b : SUCCESS in 56m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7d3767b55814c2c86e50c6a6e56ad9f : SUCCESS in 41m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/10fe1335c8c34674aabc15b384ad3681 : SUCCESS in 27m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9372f78b8d1241a081b5ba3d92e43650 : SUCCESS in 32m 36s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/c165ccb82589490180bf4a8b8b7ab7ca : FAILURE in 1h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7caa889c6997c79e7e6b302caeda99fc8e9aa4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14a2c2d6568844cdb77d694d8ca3255f : FAILURE in 39m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4544a4b5aed543b0a31f2124dc37ed43 : FAILURE in 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fda888285aca7b1bca5020c9853870f9754ed236","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-15 19:34:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/86/829386/1/check/tempest-dsvm-full-focal-py3/1a61e92/ : SUCCESS in 1h 59m 27s","accounts_in_message":[],"_revision_number":1},{"id":"542ddfed882d7f5de3da0b460e360167e46903f3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-02-15 19:57:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/829386/1/iscsi : FAILURE in 2h 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5bb199db897211f3358f5e88a94fa08eced76221","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-25 13:03:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ba10f355e8b524a21de6508d033d260c729d18c","revisions":{"2ba10f355e8b524a21de6508d033d260c729d18c":{"kind":"REWORK","_number":1,"created":"2022-02-15 17:28:30.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/86/829386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/86/829386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/829386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/829386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/829386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/86/829386/1"}}},"commit":{"parents":[{"commit":"9c4d0c9b171f07cd70c1660b63198d258135890e","subject":"Record SRIOV PF MAC in the binding profile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9c4d0c9b171f07cd70c1660b63198d258135890e"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-15 17:26:31.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2022-02-15 17:26:31.000000000","tz":60},"subject":"DNM: trigger nova tests with the neutron fix included","message":"DNM: trigger nova tests with the neutron fix included\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/829247\nChange-Id: I914ee01071b2b3f5c0c11de2171a7405c199ddb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ba10f355e8b524a21de6508d033d260c729d18c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ba10f355e8b524a21de6508d033d260c729d18c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
