)]}'
{"id":"openstack%2Fnova~829860","triplet_id":"openstack%2Fnova~stable%2Frocky~I17f4d7d2cb129c4ec1479cc4e5d723da75d3a527","project":"openstack/nova","branch":"stable/rocky","attention_set":{},"removed_from_attention_set":{"29074":{"account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"last_update":"2022-02-22 13:55:57.000000000","reason":"Felix Huettner replied on the change"}},"hashtags":[],"change_id":"I17f4d7d2cb129c4ec1479cc4e5d723da75d3a527","subject":"Gracefull recovery when attaching volume fails","status":"ABANDONED","created":"2022-02-18 08:51:39.000000000","updated":"2022-11-11 18:31:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"92f4045c290ff10210f1ea4c6ebc9b84be32d589","_number":829860,"virtual_id_number":829860,"owner":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-22 15:49:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-18 09:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-18 10:54:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"64ec89e824c372a7a2d09d090095ca2b62e26355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-02-18 08:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94faca33ba8bd6e9c699ac20cb8cc4b35d22d5f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 09:55:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe70365c809483f8138f3db8c0119a6 : FAILURE in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47e655697260471f923b2c3e85ad5fa8 : FAILURE in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e50858baf3df4cba94698be456ad6572 : FAILURE in 5m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8960c02cf2994df8ba44b3b4b1e4d459 : SUCCESS in 12m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f4be72f25d974868ad5a97e18b2c7689 : FAILURE in 28m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/caaaee2a9d5a47259e9c42352bc9abdd : FAILURE in 50m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfd907dcaddd4719bcb4acc6dba773db : FAILURE in 29m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/956e146c45ee420e9118197554eb0774 : SUCCESS in 15m 15s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/46cb8951bc064a0bad9df60ce3ee9255 : FAILURE in 25m 37s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/b172b87036a141b8b5edcf5948785684 : FAILURE in 23m 53s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/3adfa634c63344318c5919a8c38ced71 : FAILURE in 41m 39s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fb297f554a90471bae7653e75417843a : FAILURE in 29m 00s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/ff68a3360c1048d8938b5293d34afab0 : FAILURE in 22m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/b05b614576724ce1bef61bb62cc1afc3 : FAILURE in 21m 14s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/02b6850e3d8c47819ffb72a172a71e15 : FAILURE in 28m 54s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/7cad317d10134ef581fd396322ce1ca4 : FAILURE in 5m 46s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5dc95670b671402087f15e68f49b7380 : FAILURE in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66c29343dcf04ba7a539a949d9042b5d : FAILURE in 4m 05s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0dee962fe95d483cac57ae65c6ab3b4f : FAILURE in 45m 21s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f2138defcb3142ecb424914273e072ef : FAILURE in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4dc502b815dff2e594f5db10ae8c74478e79560","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-18 10:54:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/60/829860/1/check/tempest-dsvm-full-bionic-py3/788ef36/ : SUCCESS in 1h 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cece2cf6e3facd4d057bc77b97a00da485bf06bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-02-22 08:07:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a57c38e7644abaa90f7ae447f0a0df2135c73199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 09:58:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8286a109dd634e119ae6554c67051c4b : SUCCESS in 9m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/251ae68e255f469496ed55b1c3c2e6f2 : SUCCESS in 13m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a55b223c87224de38e50dd553994a94d : SUCCESS in 15m 08s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/09c44c3936ec4bf4840e71aff65817f6 : SUCCESS in 10m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bec8db4a57124beb87dd4af9e1241de5 : SUCCESS in 1h 14m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b90474c89a0246ba99ecc724a6017c63 : SUCCESS in 42m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6d97d13fe87470dac1ffa952964a68c : SUCCESS in 1h 13m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f96c2a90ced4eeab7039b9ca7afd7c4 : SUCCESS in 22m 58s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a24b3d1c66444605a49dcea285d69c1e : SUCCESS in 41m 44s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/655bd33d0ec7448389a6409231a40da9 : SUCCESS in 56m 58s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/92533ab9d9a14be983a4f398341b65e7 : SUCCESS in 53m 34s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/83d0e4789f9240e0a02a5aae92d522e1 : SUCCESS in 37m 11s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/92f49af40386444492d26e6d8d905131 : SUCCESS in 54m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/8d51ff72f65d41f48c1c1621a628689c : SUCCESS in 1h 07m 39s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/d1447bb941254abe9a0c164d2178f416 : SUCCESS in 1h 40m 50s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/85ff29472ce142668913ab2851899970 : FAILURE in 16m 19s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3d6292013d6a428a8f859e206d2a4893 : SUCCESS in 18m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0860c9edb99a4d08aa33ed438db1704c : FAILURE in 3m 44s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e04754c04cce4869a31d820017d99849 : SUCCESS in 54m 16s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c7ccd7df86434bbbb0fed93f5ebf41b7 : SUCCESS in 1h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4183412e9d3d472bba05f63d6977667c292d139b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-22 10:59:54.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/60/829860/2/check/tempest-dsvm-full-bionic-py3/43bd099/ : SUCCESS in 2h 14m 58s","accounts_in_message":[],"_revision_number":2},{"id":"fb52fc68a4a16cdffc22b9b0e45623e605c45d2b","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2022-02-22 13:55:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"849b85b9679ac07a187c77617063e12dda59963e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:49:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe0689c232b4428a260149543feb1db : SUCCESS in 10m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b83e55071c149f9b713aa3dac65d7fe : SUCCESS in 17m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/efb6ee1bd5284c9d9051539a4284e360 : SUCCESS in 14m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8383e44a744d4361abf1bd15cdebc413 : SUCCESS in 10m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2b612e1651584118aaeb7b8ae4256931 : SUCCESS in 1h 16m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/20d9149b4e8b42f2bed84547ea98116a : SUCCESS in 52m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6db04782ea444de6a784ddc15f2b88e2 : SUCCESS in 1h 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfa8027ab85540ebaeed7ef75241c95d : SUCCESS in 18m 54s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/49e559c100f5406e806bffa72c668ecf : SUCCESS in 39m 49s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/f0e2ac643f184598bb0b41f01f1096b1 : SUCCESS in 56m 04s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/8b82ea0905564147960ecf27db131daf : SUCCESS in 1h 17m 06s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/085ee4f4f5c84dce86382319c70cb24c : SUCCESS in 45m 22s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/7cea2584ec1740dc87744d64d575b26f : SUCCESS in 1h 00m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/82f6218762aa4d158ce40ae11707eb80 : SUCCESS in 1h 30m 47s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/961d074623794190a7ccb3a0e6c77a6c : SUCCESS in 1h 42m 29s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/2362aa51731b4b199cc9e1b8dff37f66 : SUCCESS in 18m 52s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0764ce6b63704e438feb7f87f2716de8 : SUCCESS in 16m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dc183498eb04d10ba4d09750ee96d10 : FAILURE in 6m 03s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/087f2e11f17d4b01986ca938941e8681 : SUCCESS in 1h 09m 37s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b413d5b4b89549049047bc82718beb22 : SUCCESS in 1h 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d77504ba1e20e0aa7059eb0c476762e081ebddb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-22 15:53:42.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/60/829860/2/check/tempest-dsvm-full-bionic-py3/07c96a0/ : FAILURE in 1h 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"92f4045c290ff10210f1ea4c6ebc9b84be32d589","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-11 18:31:06.000000000","message":"Abandoned\n\nThis branch transitioned to End of Life for this project, open patches needs to be closed to be able to delete the branch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b6933e54394cb47476f3c40227ba9367116cb374","revisions":{"00f38c8d2c789e388562bc1bd66c023a216dd525":{"kind":"REWORK","_number":1,"created":"2022-02-18 08:51:39.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/60/829860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/60/829860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/829860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/829860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/829860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/60/829860/1"}}},"commit":{"parents":[{"commit":"32200feb78e3a4a8c5d205756007f3b6a3c12981","subject":"Merge \"[neutron] Get only ID and name of the SGs from Neutron\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32200feb78e3a4a8c5d205756007f3b6a3c12981"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-02-09 11:03:15.000000000","tz":60},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-02-18 08:51:31.000000000","tz":60},"subject":"Gracefull recovery when attaching volume fails","message":"Gracefull recovery when attaching volume fails\n\nWhen trying to attach a volume to an already running instance the nova-api\nrequests the nova-compute service to create a BlockDeviceMapping. If the\nnova-api does not receive a response within `rpc_response_timeout` it will\ntreat the request as failed and raise an exception.\n\nThere are multiple cases where nova-compute actually already processed the\nrequest and just the reply did not reach the nova-api in time (see bug report).\nAfter the failed request the database will contain a BlockDeviceMapping entry\nfor the volume + instance combination that will never be cleaned up again.\nThis entry also causes the nova-api to reject all future attachments of this\nvolume to this instance (as it assumes it is already attached).\n\nTo work around this we check if a BlockDeviceMapping has already been created\nwhen we see a messaging timeout. If this is the case we can safely delete it\nas the compute node has already finished processing and we will no longer pick\nit up.\nThis allows users to try the request again.\n\nA previous fix was abandoned but without a clear reason ([1]).\n\n[1]: https://review.opendev.org/c/openstack/nova/+/731804\n\nCloses-Bug: 1960401\nChange-Id: I17f4d7d2cb129c4ec1479cc4e5d723da75d3a527\n(cherry picked from commit 9eb116b99ce32bc69c4abf8ec3b0179ef89a8860)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00f38c8d2c789e388562bc1bd66c023a216dd525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00f38c8d2c789e388562bc1bd66c023a216dd525"}]},"branch":"refs/heads/stable/rocky"},"b6933e54394cb47476f3c40227ba9367116cb374":{"kind":"REWORK","_number":2,"created":"2022-02-22 08:07:06.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/60/829860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/60/829860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/829860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/829860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/829860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/60/829860/2"}}},"commit":{"parents":[{"commit":"32200feb78e3a4a8c5d205756007f3b6a3c12981","subject":"Merge \"[neutron] Get only ID and name of the SGs from Neutron\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32200feb78e3a4a8c5d205756007f3b6a3c12981"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-02-09 11:03:15.000000000","tz":60},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2022-02-22 08:06:59.000000000","tz":60},"subject":"Gracefull recovery when attaching volume fails","message":"Gracefull recovery when attaching volume fails\n\nWhen trying to attach a volume to an already running instance the nova-api\nrequests the nova-compute service to create a BlockDeviceMapping. If the\nnova-api does not receive a response within `rpc_response_timeout` it will\ntreat the request as failed and raise an exception.\n\nThere are multiple cases where nova-compute actually already processed the\nrequest and just the reply did not reach the nova-api in time (see bug report).\nAfter the failed request the database will contain a BlockDeviceMapping entry\nfor the volume + instance combination that will never be cleaned up again.\nThis entry also causes the nova-api to reject all future attachments of this\nvolume to this instance (as it assumes it is already attached).\n\nTo work around this we check if a BlockDeviceMapping has already been created\nwhen we see a messaging timeout. If this is the case we can safely delete it\nas the compute node has already finished processing and we will no longer pick\nit up.\nThis allows users to try the request again.\n\nA previous fix was abandoned but without a clear reason ([1]).\n\n[1]: https://review.opendev.org/c/openstack/nova/+/731804\n\nCloses-Bug: 1960401\nChange-Id: I17f4d7d2cb129c4ec1479cc4e5d723da75d3a527\n(cherry picked from commit 9eb116b99ce32bc69c4abf8ec3b0179ef89a8860)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6933e54394cb47476f3c40227ba9367116cb374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6933e54394cb47476f3c40227ba9367116cb374"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
