)]}'
{"id":"openstack%2Fnova~839275","triplet_id":"openstack%2Fnova~master~I516b9d732ccea6e12904a1612530bce273d06587","project":"openstack/nova","branch":"master","topic":"drop-c8s-testing","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-04-26 20:36:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I516b9d732ccea6e12904a1612530bce273d06587","subject":"Move centos stream testing to centos-9-stream","status":"MERGED","created":"2022-04-25 23:15:06.000000000","updated":"2022-04-27 07:34:18.000000000","submitted":"2022-04-26 20:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839275-drop-c8s-testing","meta_rev_id":"92047f9e4f77706ddaec91707e5c91f8a8f1f8e2","_number":839275,"virtual_id_number":839275,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 20:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2022-04-26 16:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2022-04-26 07:17:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2022-04-26 16:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-26 13:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":2,"date":"2022-04-27 07:34:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2022-04-26 16:17:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-25 23:15:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-25 23:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-26 00:11:12.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2022-04-26 00:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 00:56:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-26 01:06:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-26 01:26:48.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-04-26 07:17:51.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-04-26 13:24:54.000000000","updated_by":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"reviewer":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"state":"REVIEWER"},{"updated":"2022-04-26 16:16:34.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-04-26 16:17:06.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2022-04-27 07:34:18.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"471ec3a0990f3c89a3fe8a51cd079d57dce53105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-25 23:15:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4625f1d0a7dad0ebe3f26d484979f348cfc9033a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-25 23:15:18.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42042/","accounts_in_message":[],"_revision_number":1},{"id":"3a31d94fb9bff1adb825b424e5f26ea311d63bcc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-04-25 23:17:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42042/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"66aaa68f1f99d61f273e79c83e5dea128d54c385","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 23:49:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39859f8eb598421c8c67b04df664a99f : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6962fa6299174da2983807509f67b668 : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39adc98486be0db2379e3cb125873d732b9a4781","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-04-26 00:11:12.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/839275/1/iscsi : FAILURE in 54m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b8ec195c266c4cdc8a702d174fcca6e3bf279671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 00:56:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/87bb31e8ebd8404aad0759580e60fa94 : SUCCESS in 1h 25m 02s\n- tempest-integrated-compute-centos-9-stream https://zuul.opendev.org/t/openstack/build/c65c6ce621804317bcdcd921c52a3aca : SUCCESS in 1h 20m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4a25893cd4248ceaea2a88c474ad2de : SUCCESS in 59m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09251b4b36c04c07877feb076cef1e3f : SUCCESS in 17m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/320ca942c49849a9909fe83ab3169bfc : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3257bf9879de46bd8e8b88971f01a274 : SUCCESS in 13m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/337a71e0b29c4aee8aca88a096056133 : SUCCESS in 14m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd55f6a904b94073afb0fdf0526342c8 : SUCCESS in 15m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2688156a90c84c62b9c323e0dd7602fc : RETRY_LIMIT in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeb2eaec932c4aa3b5bf49121ff53072 : SUCCESS in 11m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c9fa2073b4fa4a32b90cec05ba7271ca : SUCCESS in 1h 06m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d739e561ec5747b795fcbc9d50eaaefb : SUCCESS in 1h 12m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aa7f35c5dab9462685cc4937cd97839f : SUCCESS in 45m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/abc63bbb882b4078b8b1721ae3a29b01 : SUCCESS in 46m 16s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c3405b727d7046afb6d0ccbf22a06f8c : SUCCESS in 48m 00s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/393a70e68222418595dea899d050973e : SUCCESS in 1h 32m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/71211d74d21b4e1199f0701cc179fd71 : SUCCESS in 1h 38m 41s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6159850cbc404235a4c426de5714d837 : SUCCESS in 49m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/00ebb8b57aa04f6fabfba89a3c07d5e2 : SUCCESS in 7m 07s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89aadd0733b34c8795af49c4cd7a9c52 : SUCCESS in 19m 21s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e990da3191e400e81d1eafb5f7559b4 : SUCCESS in 22m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c68db0593a55417ab0514a1bf28c8ec0 : SUCCESS in 1h 04m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62536c1e5b7844e194c93dd2a0c44651 : SUCCESS in 56m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c8b0177d630d4fe8912f80a54afb02de : SUCCESS in 26m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0cd69493c2a14f6fa1ab6e9c7bd69bc3 : SUCCESS in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"100465e211cf8308de5c8e12e89984d1084a2275","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-04-26 00:56:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/75/839275/1/check/EMC_PowerFlex_NOVA/6821a5e/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"f34408c0becff4f7b53a0cdf0ba467e2a2d7109e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-04-26 01:06:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/839275/1/check/tempest-dsvm-full-focal-py3/5cf84e0/ : SUCCESS in 1h 45m 58s","accounts_in_message":[],"_revision_number":1},{"id":"34c66918b30196bb5b40e18b1952a654f2417ab9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-04-26 01:26:48.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: https://github.com/LightBitsLabs/logs/tree/main/26_04_2022_00_56_23\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"2f9b7d9979c6bdf2899b8ee16f64c211b93f735f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-26 07:17:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"93af23d83c991bb19ca71c5cb958efcc47063e7e","author":{"_account_id":6062,"name":"jichenjc","email":"jichenjc@cn.ibm.com","username":"jichenjc"},"date":"2022-04-26 13:24:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6364876f7aef84a551ea218b8fd749d7fb76d29","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-04-26 16:16:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2c22cdd66642adc594bf1e353ac39a3707c37ca","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-04-26 16:17:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05be95fd328246bba9353855ffa5e049ac867a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 16:21:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc8e831693818a0142c3daddbe7292e7c8cbb1fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 20:36:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/92f1e7ee98f14a2a8e49db691da27af0 : SUCCESS in 1h 17m 27s\n- tempest-integrated-compute-centos-9-stream https://zuul.opendev.org/t/openstack/build/937ed9c857a4433dafa977c4d6d4a7f4 : SUCCESS in 1h 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/331a62d0a62547169963261aadf99e7b : SUCCESS in 58m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e8d61bca68426bbe5aadbc0450a28b : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa735880d31f4d9d9ebf9633dc7e6278 : SUCCESS in 14m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b2b4f7085db46958f03f5cc0dcbb976 : SUCCESS in 13m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72ba2f2157fd48a184c93a781908e435 : SUCCESS in 15m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816c3c5cd5914d01b0e5d46ac2bcedfc : SUCCESS in 11m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8f69464068d041e3a7af5b7e7b80b0bb : SUCCESS in 56m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7c6c9a12a506439dba1936f72ca6c33b : SUCCESS in 53m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd5f319b73ee4a519dba9bebae7af9ff : SUCCESS in 20m 45s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ba21e48518004ac29fbe0ac507edd73a : SUCCESS in 20m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f7cddf119c174f8391e23faa93cfd236 : SUCCESS in 1h 22m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/d2eb7f990362473f86c061ccbdadf213 : SUCCESS in 1h 31m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fe9e93b77aa445c8aabc8f7f6aa0ce17 : SUCCESS in 7m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c01a6b2d20948e99205aa4096bac120 : SUCCESS in 1h 08m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae26b4654b454a889ff32e89914ffccb : SUCCESS in 1h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edbd6340d79a431eb2a402b1b7797f6f : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b7bb2cc3ea553870e755db8dd89ba1d37197a4af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 20:36:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4249782ea15313e69f3e923b7b8f8f3935ff4c90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 20:40:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff7d856ca6048169b43f34b081591a4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1},{"id":"92047f9e4f77706ddaec91707e5c91f8a8f1f8e2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2022-04-27 07:34:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f5551448dcfcde26095963e223f973b90e6f637","revisions":{"5f5551448dcfcde26095963e223f973b90e6f637":{"kind":"REWORK","_number":1,"created":"2022-04-25 23:15:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/75/839275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/839275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/839275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/839275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/839275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/839275/1"}}},"commit":{"parents":[{"commit":"56b5aed08c6a3ed81b78dc216f0165ebfe3c3350","subject":"Merge \"Deprecate [api] use_forwarded_for\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-25 23:11:12.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-25 23:11:12.000000000","tz":-300},"subject":"Move centos stream testing to centos-9-stream","message":"Move centos stream testing to centos-9-stream\n\nIn Zed cycle testing runtime, we are targetting the centos 9 stream\n- https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nWith dropping the python 3.6 support, project started adding python 3.8\nas minimum, example nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nwith that, centos 8 stream job is failing 100%\n- https://zuul.openstack.org/build/970d029dc96742c3aa0f6932a35e97cf\n- https://zuul.openstack.org/builds?job_name\u003dtempest-integrated-compute-centos-8-stream\u0026skip\u003d0\n\nQA is dropping the centos-8-stream testing and support\n- https://review.opendev.org/q/topic:drop-c8s-testing\n\nThis commit replace tempest-integrated-compute-centos-8-stream job to\ntempest-integrated-compute-centos-9-stream.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/839274\nChange-Id: I516b9d732ccea6e12904a1612530bce273d06587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f5551448dcfcde26095963e223f973b90e6f637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f5551448dcfcde26095963e223f973b90e6f637"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
