)]}'
{"id":"openstack%2Fnova~844750","triplet_id":"openstack%2Fnova~stable%2Ftrain~Iab455e22aa075bb1b3fb9b7d1870b1eb7314e320","project":"openstack/nova","branch":"stable/train","topic":"bug/1939545","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2023-09-01 16:55:15.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-01 16:55:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iab455e22aa075bb1b3fb9b7d1870b1eb7314e320","subject":"functional: Use tempdir for CONF.instances_path","status":"ABANDONED","created":"2022-06-05 19:58:10.000000000","updated":"2023-09-01 16:55:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e7275344faad1a4e5d1298a4a56a26430305198","_number":844750,"virtual_id_number":844750,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-17 18:16:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-05 19:58:10.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-06-05 20:42:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-05 20:42:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-06-14 01:33:22.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2022-06-15 16:32:49.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2022-06-17 17:18:07.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"75d378e4fcebd0b824cb30b3f03513881eb9edec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-06-05 19:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9882317873684ab70cf940c0fff6c9caf094cbff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-05 20:42:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6da0c38214a4cec8ba6877411bb8ccd : SUCCESS in 36m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f3709959486452996bcdc7145e3f8b0 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/474a31932181409bbdea2db0d0a7a351 : SUCCESS in 12m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b76a546fc6b4af08d46d22b6f502705 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f1d1889c0f4c23a2f15d0c6c8e5cca : SUCCESS in 12m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30a3f957ecb74167992b207787b371ab : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852e579c8fda49acb5115a5e3f6543ef : SUCCESS in 10m 13s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/14af49d82f5d4d3d877580a809665051 : SUCCESS in 16m 36s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e7b730d1fc743c997c62d495398cee8 : SUCCESS in 14m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/54c9a8faf370483f8b91eeec9030d9b4 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b7afdb09ab9af829c277d449e8cce3bac22bbf0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-05 20:42:17.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"69b53391b44df8bb0d7f7a224a534aca02b83e51","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2022-06-14 01:33:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a308e009e6276b18a8251db616b2200e62732213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-06-15 16:32:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"47ba2e46ce9ef389716ac68743e8c846fa430f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 17:38:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1848f4388f664b06b510f8621cce4669 : SUCCESS in 59m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cc65785037546738538818aa10dd86e : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91fb8a97a81b4979a8e93af7c55f7c6b : SUCCESS in 11m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f2dcffa251a493094cfcf3630d3fd60 : SUCCESS in 13m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62843d9b3b3b494189f799b721b4d9cb : SUCCESS in 13m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3a0464b54994df381d6d7d02b01a713 : SUCCESS in 16m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ccba54cfc44dd2903644947a71341d : SUCCESS in 11m 15s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/a724ce64e227465ab1cfe74a1abe5c4b : SUCCESS in 19m 34s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e480adffb24f4b84b3d9be442bf000ca : SUCCESS in 20m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/393a44fbe4ee444a8fcc29b78b969edb : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa853c2ac72bb60b6ff5345419c3ca136b7ff3aa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-15 17:38:19.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"f6ff974ba6348ed266b193385842f00a93cc0716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2022-06-17 17:18:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"58ca84fb18a5a9aaedf5febc745b9811e8d12d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 18:16:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b872691c4d740329079c646cc10b9af : SUCCESS in 54m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b76f0408b7c4324bed5db8cd22dda71 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d814932f8fe542dda5c84c4d641ff785 : SUCCESS in 10m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8c84ca329e442ae937330c6b3290fe7 : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b245c4728dc40e49175356004416c9a : SUCCESS in 13m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc28525ca8794e41a3fe909a15addf55 : SUCCESS in 14m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820f86665032497087393b62c9165816 : SUCCESS in 10m 01s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/c8a92051df604e198ea386eba07711d8 : SUCCESS in 22m 39s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5210fbfcda5c41a4b8b2e01bd9e8cbef : SUCCESS in 22m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f522b1922e9e4949b69dc83ffc163a36 : FAILURE in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cc2c7d1aa10c22fe6e69d2a7c8907ee88db0e412","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-06-17 18:16:55.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"0e7275344faad1a4e5d1298a4a56a26430305198","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-01 16:55:15.000000000","message":"Abandoned\n\nstable/train branch of nova projects\u0027 have been tagged as End of Life. All open patches have to be abandoned in order to be able to delete the branch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef66e1a44ac4544e37a7cdced86444d6e2ddc9a8","revisions":{"ed83edec0c3ebd7c7d38db6e72e959fe372f7a3b":{"kind":"REWORK","_number":1,"created":"2022-06-05 19:58:10.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/50/844750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/50/844750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/50/844750/1"}}},"commit":{"parents":[{"commit":"e55fa2597b4fa49cc8a8b85157e1e1d571e7ba28","subject":"tests: work around malformed serial XML","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e55fa2597b4fa49cc8a8b85157e1e1d571e7ba28"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-19 12:09:20.000000000","tz":60},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-06-05 19:54:10.000000000","tz":0},"subject":"functional: Use tempdir for CONF.instances_path","message":"functional: Use tempdir for CONF.instances_path\n\nThis was being set in an ad-hoc manner throughout the code, but it\u0027s\nnecessary for all kinds of live migration and evacuation operations.\nJust centralize the configuration.\n\nConflicts:\n* nova/tests/functional/libvirt/base.py due to 6c4d1a6aabb changing\n the context around the line this cherry-pick adds.\n* nova/tests/functional/libvirt/test_numa_live_migration.py didn\u0027t\n  exist in stable/train.\n\nNOTE(artom) This is needed to prevent the live migration test added\nin the subsequent patch from failing in the gate (and only in the\ngate, the failure could not be reproduced locally) with `No such file\nor directory:\n\u0027/home/zuul/src/opendev.org/openstack/nova/instances/tmpif84g_jd\u0027`.\n\nChange-Id: Iab455e22aa075bb1b3fb9b7d1870b1eb7314e320\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 980711f3d3f2db9dfd792a8c6fd15cb45a85f615)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed83edec0c3ebd7c7d38db6e72e959fe372f7a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed83edec0c3ebd7c7d38db6e72e959fe372f7a3b"}]},"branch":"refs/heads/stable/train"},"3aacc07d734398b324f69e5982ec689b1571f064":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-15 16:32:49.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/50/844750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/50/844750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/50/844750/2"}}},"commit":{"parents":[{"commit":"08334dae2959b664b5fd395728ed299543bca4da","subject":"tests: work around malformed serial XML","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08334dae2959b664b5fd395728ed299543bca4da"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-19 12:09:20.000000000","tz":60},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-06-15 14:41:39.000000000","tz":0},"subject":"functional: Use tempdir for CONF.instances_path","message":"functional: Use tempdir for CONF.instances_path\n\nThis was being set in an ad-hoc manner throughout the code, but it\u0027s\nnecessary for all kinds of live migration and evacuation operations.\nJust centralize the configuration.\n\nConflicts:\n* nova/tests/functional/libvirt/base.py due to 6c4d1a6aabb changing\n the context around the line this cherry-pick adds.\n* nova/tests/functional/libvirt/test_numa_live_migration.py didn\u0027t\n  exist in stable/train.\n\nNOTE(artom) This is needed to prevent the live migration test added\nin the subsequent patch from failing in the gate (and only in the\ngate, the failure could not be reproduced locally) with `No such file\nor directory:\n\u0027/home/zuul/src/opendev.org/openstack/nova/instances/tmpif84g_jd\u0027`.\n\nChange-Id: Iab455e22aa075bb1b3fb9b7d1870b1eb7314e320\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 980711f3d3f2db9dfd792a8c6fd15cb45a85f615)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3aacc07d734398b324f69e5982ec689b1571f064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3aacc07d734398b324f69e5982ec689b1571f064"}]},"branch":"refs/heads/stable/train"},"ef66e1a44ac4544e37a7cdced86444d6e2ddc9a8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-17 17:18:07.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/50/844750/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/50/844750/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/50/844750/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/50/844750/3"}}},"commit":{"parents":[{"commit":"a863d808260547e62c8eb09452963b8fe07af713","subject":"tests: work around malformed serial XML","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a863d808260547e62c8eb09452963b8fe07af713"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-19 12:09:20.000000000","tz":60},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2022-06-17 17:13:21.000000000","tz":-240},"subject":"functional: Use tempdir for CONF.instances_path","message":"functional: Use tempdir for CONF.instances_path\n\nThis was being set in an ad-hoc manner throughout the code, but it\u0027s\nnecessary for all kinds of live migration and evacuation operations.\nJust centralize the configuration.\n\nNOTE(artom) This is being backported because a functional test failure\nwas observed when running functional tests that are added by\nsubsequent patches. The specific error was:\n\nNo such file or directory:\n\u0027/home/zuul/src/opendev.org/openstack/nova/instances/tmpif84g_jd\n\nThe failure is not always reproducible, but never happened again with\nthis patch included in the series.\n\nChange-Id: Iab455e22aa075bb1b3fb9b7d1870b1eb7314e320\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 980711f3d3f2db9dfd792a8c6fd15cb45a85f615)\n(cherry picked from commit 05c88a4e21b523ad30ab133093b627a0fae78adf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef66e1a44ac4544e37a7cdced86444d6e2ddc9a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef66e1a44ac4544e37a7cdced86444d6e2ddc9a8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
