)]}'
{"id":"openstack%2Fnova~858383","triplet_id":"openstack%2Fnova~master~I9660d42937ad62d647afc6be965f166cc5631392","project":"openstack/nova","branch":"master","topic":"bp/allowing-target-state-for-evacuate","attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-02-06 21:28:13.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-01-31 13:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-02-01 11:49:38.000000000","reason":"Change was submitted"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2023-02-01 11:49:38.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-01-26 16:56:34.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"hashtags":[],"change_id":"I9660d42937ad62d647afc6be965f166cc5631392","subject":"compute: enhance compute evacuate instance to support target state","status":"MERGED","created":"2022-09-19 13:48:24.000000000","updated":"2023-02-06 21:28:13.000000000","submitted":"2023-02-01 11:49:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":50,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"858383-bp/allowing-target-state-for-evacuate","meta_rev_id":"5a110063c3f78428cf9bc8b82f9b5ab2b611a990","_number":858383,"virtual_id_number":858383,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-01 11:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-01-31 13:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-01 09:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-01 09:54:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-02-01 08:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 14:18:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-19 15:40:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-19 16:14:55.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2022-09-21 16:04:35.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"CC"},{"updated":"2022-10-04 11:46:06.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-10-04 12:28:20.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-07 08:16:45.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-18 11:47:32.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-10-19 11:48:51.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-01-25 16:59:43.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-02-06 21:28:13.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"}],"messages":[{"id":"d50e8eccfa19671a49954e964eb03ce9dfb85774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-19 13:48:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"835caffa69745e056290804f90c515a27e7f03f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 14:18:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b336a4cb0bf4838b44630b8468e7a35\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e465588716524dfeae6bcd7be921d41c : SUCCESS in 27m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a57dbb369ef48c1a700e1869416b1f8 : SUCCESS in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57ce2620850ef70073285c7e50bdcc9d3391f521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 15:40:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8d1b580edf84da39828c4b8955a9ac4\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5b9705bc4b20417987c3620786670f9c : SUCCESS in 1h 23m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2222c2efd0ac44ebb2d86fee5d6870f0 : SUCCESS in 1h 21m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e8392413184e8bb190be6050579eb4 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eac9cdb88c450f92f2ff2d26ee8e66 : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aef2cb824c5c483695650c9a16f82206 : SUCCESS in 11m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/094567abca7f44908882d44b9f89b6f4 : SUCCESS in 13m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a86321b5bab41dfbd47beeb009b632c : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb581f00c64e4cf181f610d4c907d078 : SUCCESS in 11m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4e1dfee527db42efa9b3e2b601a6c3ff : SUCCESS in 1h 07m 57s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b492ef3152f434e8393f8e6d603b588 : SUCCESS in 1h 14m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d876f5ca68f74818ab8caa10c6a489e2 : SUCCESS in 57m 49s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b30ec763bed0473e92d686861b5626ff : SUCCESS in 54m 58s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/642a2348949248a6be259661f5f42777 : SUCCESS in 1h 33m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/369eae3c45ed44bd91dac48f67e2a866 : SUCCESS in 1h 48m 31s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b91ac4d10d6f422885ab0426582a56e5 : SUCCESS in 51m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fd56a802ab744ae082b0826477c05a2f : SUCCESS in 5m 34s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f6d4b4be148f466c9c0882de4aed5a0d : SUCCESS in 21m 16s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b17783cdcde44bca801bd179e2e44906 : SUCCESS in 22m 40s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e67002346a045018ceb1a9553bdcce4 : FAILURE in 23m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3670d530d50b44be8de7f74da2ba9ae0 : SUCCESS in 1h 17m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a237300265d241c0b157e8afb9a5f9e3 : SUCCESS in 1h 01m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/22f83541a0c54e09839ef3aebf2aaff5 : SUCCESS in 37m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/234d74092ad8499189a1febbfee089b4 : SUCCESS in 46m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92505b803064b1c90d760505401660b069807334","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-09-19 16:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/1/iscsi : SUCCESS in 2h 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d60024cd2a640d281782b3f73eac2f933789f1ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-20 09:58:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1052e48b1291cd7b1b83432610783dfaf49d79f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 10:32:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc892bc9857345dc9c93962d1b9c441a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15971e08e1ff42fcaa073edf63adcf5c : SUCCESS in 30m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f50dadac73d4e538cec71cc27be3ce6 : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37abd116e8aa6ffa3696df945c928ca3c0c1f0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 12:06:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b5a97fbc99440478191eee4003e029d\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/93c8fff57b8449269ef727a92c5f6e72 : POST_FAILURE in 2h 01m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7b079e788af471b99267e0cdbd98773 : SUCCESS in 1h 22m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79475864440a433f812cf88c64dba70a : SUCCESS in 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cbeae1e9a594aec901c05840a4b33f0 : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f13b91b3f284db4aa1e63035f6b7afe : SUCCESS in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa6e4cefea204219830261160e52fe05 : SUCCESS in 15m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63541606f38f45c5800045ca9c8a6163 : SUCCESS in 13m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f3b7909ee840729f88b834c1a19ee5 : SUCCESS in 12m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6028623df29a46269b0ab072b46fc4e7 : SUCCESS in 1h 12m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/201254d0e28b4c78a86fb6dfb1807930 : SUCCESS in 1h 15m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/85267e40e57e4c1b9695c81a7e764a9c : SUCCESS in 59m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/40dd532286ed4e68841e5eb474e5a1cb : SUCCESS in 54m 40s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1d103cc59bb24efead00587e9d736530 : SUCCESS in 1h 27m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/ed3878f8712e4c75882de9043d511f0d : FAILURE in 1h 49m 31s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8eb78c4471b447b19f6d3a9a3801c2fc : SUCCESS in 1h 04m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fd53199b1d5848d0b9c24cbae5005c86 : SUCCESS in 5m 04s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cf5b03dd0284ec88753c251c25243b0 : SUCCESS in 20m 46s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a1f48035df8b46029356e16ad94a326d : SUCCESS in 21m 34s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15df2eff7d0545e68a1fa808b2c50fa3 : SUCCESS in 23m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/05e3c14e505744bbbe65fcdda0d8eb06 : SUCCESS in 1h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f39c2d4d7d33497fbc9ba3283224be47 : SUCCESS in 53m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8e9f12452dd9498cbe5ddd88516db15b : SUCCESS in 39m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/31b6e15894ae48079e71c2885ba508ae : SUCCESS in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76f0fb8d019930c8c23d1268625183d46049ad0b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-09-20 12:27:16.000000000","message":"Patch Set 2:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/2/iscsi : SUCCESS in 2h 27m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a65c979a07fba9433c9e9249f2e52de59e5c69f0","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-09-21 16:04:35.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9fa055b76c02167a443475fd67b3a95692b0099d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-03 09:32:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fa302b9c2242dca99d51dee21fa923ad80f72389","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 10:13:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f1987bd46d4b3fb035b3582a1d8c0a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ef43be5a606447eb8c4da7cde2a1a1d : SUCCESS in 38m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59542f749c5640508ec2e8849fe2d2f7 : SUCCESS in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"83e7ef2a7d59e555902a925c0ea27cb0aebd778c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:47:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d69d902f6dc4f039e575eb9b42f106f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/42fa39e7dd954391be85be81ecb375ca : SUCCESS in 55m 12s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/faf13d9ea0c74b2cb02f1c800e2235fe : SUCCESS in 1h 29m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c33c8dc80754b488f73bfa4332b7135 : SUCCESS in 46m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96007bc728f14f6698b5ece7734e3516 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/397d4c7a2bbf475baac63e8c6471a12e : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e64c3a281a1497dbc89534e496b0411 : SUCCESS in 12m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b6cff82a863400ea02cc802093f3f20 : SUCCESS in 14m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234cf06aa787428199bede2e0332b51a : SUCCESS in 10m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2f02511ae3ce4775a5d963fbf31bb07f : SUCCESS in 1h 10m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83675d2b914a49ff905f1518c88d81e6 : SUCCESS in 1h 19m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/85209074c52b4274a237ad17d4249622 : SUCCESS in 58m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4308ffd00038427fa6967d21f330009d : SUCCESS in 57m 49s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/92ecb48752464ba280c709f934d6a3a7 : SUCCESS in 1h 24m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/9a87056e41d34eb8bf32691d00e8b5c7 : SUCCESS in 1h 57m 47s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/48a8d227c69d4b368f80999dadf68240 : SUCCESS in 1h 03m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/04dba04f86b341cea504dea8afeffe3f : SUCCESS in 7m 07s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af697ec482cc475b93d2781ff0b65e03 : SUCCESS in 19m 57s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51e3028b8fbc4b96bfbac5280e6d57e1 : SUCCESS in 25m 36s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/886251d6a0de4a188a3e02314aa0a5a8 : SUCCESS in 21m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e36d188175843658299ed0a37e99f61 : FAILURE in 1h 11m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38b856e53e9f4fc6b40c1ca96b47f75b : SUCCESS in 57m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/65e0a1adc20144e29d7892ecaa1ee81f : SUCCESS in 29m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a9e73e6d3d014c12a3bdbeb4bac1dd53 : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d888a7d807a0777e7bd9bfe4976ed7d1eeecc8c7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-03 12:21:31.000000000","message":"Patch Set 3:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/3/iscsi : SUCCESS in 2h 44m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e0eaf4125cb6269167a955820b1c9aef1f32a65f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-04 10:51:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95bcd1998570ed97c4efc2349d29da86d0bba559","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-04 11:46:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c514a1771859795c0b50fbe98d27fe06c6141ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 11:59:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a012769f014212b2350239a350ff19\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07ec6d4344a840398eb74d9a2bfebf23 : SUCCESS in 30m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ae688bd6bd84903b0d4cfd7de3aab7f : FAILURE in 54m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7b608d17c7d25c0e79ceb534d13881a725c644cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 12:27:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93ac644121b4fefb382dc5dde769ce1\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/857f7d7b375f40aeb2a4f93f2627db3f : SUCCESS in 1h 16m 05s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/17a8280deb834a04a887f8def844caa6 : SUCCESS in 1h 28m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd47424052f6483ca13d5fff6fe16b83 : SUCCESS in 58m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01f06ed73d1646b9ac83aa6db63af425 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c410469e258046ab9aec94589c212fe8 : SUCCESS in 12m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8ccbc00a8784d0ea1124d7d0d9a44fd : SUCCESS in 14m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb21a3c5afce47c0a739509c1b0fe906 : SUCCESS in 15m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50bd1c89bb564bd2860a381591514f2f : SUCCESS in 11m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b5f199580a954dd99d5d1b6eafd8ba7b : SUCCESS in 1h 16m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c2d26e753fc847949c9be577ad96ada2 : SUCCESS in 1h 01m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5e267a17b56a4e5e9ce653ab36c691e4 : SUCCESS in 1h 02m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d2ebed1fa5a446dbaff5a53e7781b00c : SUCCESS in 58m 30s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d54dc80d506143a7b795b7996d71a670 : SUCCESS in 1h 28m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/d1d6f105583646259dfc0e7080faaa22 : SUCCESS in 1h 34m 32s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/bba2b4fd8d79417a8b223bacba0a2c3c : SUCCESS in 51m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2daae3a322824f7e8c7472cda49d9294 : SUCCESS in 6m 35s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e87badf7125490784dc3682c735f755 : SUCCESS in 21m 22s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cda0f04846446658d0d8a7695c04bb8 : SUCCESS in 23m 14s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9af5c773a9c04b7fa73fe39c70d5e577 : SUCCESS in 22m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/115ec53834a443fe989d65519a7fe65f : SUCCESS in 1h 11m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a859534ba7a242f4948af978dca35724 : SUCCESS in 55m 07s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/77407e307e0f4a64aa46b2e00a16a9a6 : SUCCESS in 36m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4f6026497bdf45968078622a65b35ffd : SUCCESS in 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d4326a535144cf612077f5ded74f728a7a9bb253","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-04 12:28:20.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/4/check/EMC_PowerFlex_NOVA/87fc499/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"ba1272db416c1535deb94a20d2437fe1100e8701","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-04 13:13:13.000000000","message":"Patch Set 4:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/4/iscsi : SUCCESS in 2h 20m 54s","accounts_in_message":[],"_revision_number":4},{"id":"3f149ee23ce08c5ff1b36a3e9c753514ac2b5c28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-04 14:36:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"27ed5e11eafbbff3d0740028eccb4502422164bf","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-04 14:37:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ef9eebb275c9fc00a737c7981efbc04d6ce2c652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 14:54:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b859e6ce6af44b6802f91cbbadb8e2e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e80864b2f0af4713bfc4195ff4feef90 : FAILURE in 9m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bf36ac22a66a42769013899a3edfd747 : FAILURE in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4805005b0b0c6a25aacd07fee473e2e6c4e11f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 16:37:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6062098d4cd41e08bda06774bb4957b\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/745d2f4c3a7c4b1484c1f1121cc7e9d4 : SUCCESS in 1h 00m 39s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8b1088c2bc83421884ee26afcae72285 : FAILURE in 1h 50m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40741f892b634c6797e0e8cb28b16a8f : SUCCESS in 59m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e4b5afcd6684a4e9a889e4b10ff4c6a : FAILURE in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9e3b2e49774f99bd1059f705721c92 : FAILURE in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0af69da9185f4ce88187333484a6ce3c : FAILURE in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f30ba2d97cc9422ca6182de50657e7ac : FAILURE in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2d45952702047beb1005a48026cff8d : SUCCESS in 10m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f005bc80c56940c8946f3df8c8b7726e : SUCCESS in 1h 01m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd95b42acf074786876207b4d09d7dc7 : FAILURE in 1h 18m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e48f66aaa3e44f1a2e8febdd794edfc : POST_FAILURE in 1h 01m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/44e6e600ccc140ec861ded47ea34f51d : POST_FAILURE in 1h 00m 56s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f15fdbfdbc1246a7b9588f78d565f274 : FAILURE in 1h 58m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/4a8620e87a884b93b2af01b6ef56cc79 : FAILURE in 1h 51m 32s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/06a35b0c8e784e92a236604dc2031f0f : POST_FAILURE in 1h 02m 45s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8d46975986e14996a1a2e20ce10088b0 : SUCCESS in 5m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e1555fd63414e96bb3bcec780b8152b : FAILURE in 22m 53s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67157bfbb778475091b8e4afef293f53 : FAILURE in 21m 21s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/064c15b174304604a8194624971ad42f : FAILURE in 26m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b1e98168c284f6990548558227a9fca : SUCCESS in 1h 10m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d32f431e8fe3461d9f324b65ec75e092 : SUCCESS in 48m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b3a75a66adfc4bc7b5589d09c08b3971 : SUCCESS in 30m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3d5bd8a64edc42f1a99813db1ea46f67 : FAILURE in 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70a663d777a4160573eec316a7021b7b2bee76da","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-04 16:55:46.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/5/iscsi : FAILURE in 2h 18m 18s","accounts_in_message":[],"_revision_number":5},{"id":"8fc092ba0b586ccf9a21de8e3e4af6e2f1bb559f","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-05 06:44:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9de1eaaad0c8dbe03a869fb1620918ed0e06e16c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-05 08:58:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"31395bd702124e3c1facb91ae2ba69d0d35919a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 09:28:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdca53cad8ab4a20961bfa19d90a84bc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82696805c173456aaa6ec755003f655d : FAILURE in 16m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4eac826ee8ab4e06b2e3967414bce4a1 : FAILURE in 28m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"76cd174900fbc3cc4b80e9b148596a95eb93cf12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 11:12:36.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9892da1b7808475cb8d757ecb8e6676f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/30bb98e6942440d285928ba34600a376 : SUCCESS in 1h 30m 39s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5181d429c3674c0790ee6c2253721f35 : FAILURE in 1h 36m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aba06c45a4dc4b0e8d1cf2690f06bdcd : SUCCESS in 1h 04m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed4389fe174647efaa9071148ec0a42d : FAILURE in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df049b48c167487598744af2c4c791c3 : FAILURE in 13m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd7f9bdf16444a2abc5c548efe22649c : FAILURE in 13m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/090b97cff6864d8bb8e3473d76594574 : FAILURE in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a42335f1404b5da0d01566f26d459d : SUCCESS in 10m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4d4e58bcf2dc4af48c232821c1b3043b : SUCCESS in 1h 06m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/41ec627a1ec74c2388681a609ccdf093 : FAILURE in 1h 36m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e8ed09dbf19438cbe7057efa88d9399 : POST_FAILURE in 1h 06m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/59c771cad9354c10915fc25823e38173 : POST_FAILURE in 1h 28m 41s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8496b2966b5b490082a7cf0ebf83d388 : FAILURE in 2h 07m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/3f3e38398fc64fa0b45ec4a1c0e93d06 : FAILURE in 2h 02m 19s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5d2e15f6a88d426680216edd9a145c15 : POST_FAILURE in 1h 08m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/56a819bed3464301885d8a19bd8e838e : SUCCESS in 6m 15s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/791d71342cd44d839e638adddc4273a5 : FAILURE in 22m 11s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31a0cc22ae634999bda149c91d1446a3 : FAILURE in 26m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2f178cc63b24f0a8679f7fdf1c174f5 : FAILURE in 27m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/463a61dce229495da3cd24b0123de9ac : SUCCESS in 1h 14m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db892fdeeed844c4becc7607ae603f21 : SUCCESS in 56m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6f1faea676fd4a6d809d09b3981d478b : SUCCESS in 40m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c3e7b798124d4a5e8ca7141572cc5899 : FAILURE in 51m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5f0ad2e5ce64c2278f2a778940bc5e3e005c227e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-05 11:20:29.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/6/iscsi : FAILURE in 2h 21m 14s","accounts_in_message":[],"_revision_number":6},{"id":"0b242c0cec1040c688b262b630261b944199f212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-07 07:53:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bc6150a6259b0081408a6239c8ec04932df2e78a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-07 08:16:45.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/40e4791cad05496b85d42586be28e485 : RETRY_LIMIT in 28s","accounts_in_message":[],"_revision_number":7},{"id":"3947cfd205debb321d25433d7a57f61118584ec2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 08:32:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c3648d4a134e00b275739a1aae40a6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/565e7db0a3064556811794daab5ca3f5 : SUCCESS in 20m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e0f09cbc63d469aa16863fa0f1aadda : SUCCESS in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"29110fc0f88fa452778fa42d9e2706b83a2032f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 09:40:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bef9594cdddf49158c81a68e3db925e7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bbf08da0fe8f4ccbafd1585c9442a64e : SUCCESS in 53m 02s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/52ce776b510d457a8e0610bdab7f53cd : SUCCESS in 1h 19m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb0a7e3c590f4f06bc8d3d94e9d0886c : SUCCESS in 48m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99721ad9ffb64254a6cda948bf1c4abe : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd294eb9d644fe2b53ed10a20e9ed26 : SUCCESS in 11m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/668d3987da394773a4bbefa2948d10f3 : SUCCESS in 12m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec049282b7bf43e2bcd1466b0a1c6dc6 : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1885c0ddd24175a4885e51c4b46987 : SUCCESS in 9m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/241b3d6defcf4a838bb29470bf0bdcf6 : SUCCESS in 53m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56c03ada24914d11bc47c5e582175666 : SUCCESS in 1h 06m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/61ab4480c56a41b0926843d6798013da : SUCCESS in 41m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e9ea01f9bf704edea48ee759c01c6eb8 : SUCCESS in 45m 23s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/88391c48ef2644bc936c9df40989531f : SUCCESS in 1h 39m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/d96a904fdb714582b0d69b8f7e17d9a2 : SUCCESS in 1h 35m 26s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9f947c041a62401ab00415cd43bff06d : SUCCESS in 51m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fa6ec4251e014b60955f64c496f9efc5 : SUCCESS in 6m 08s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d223e4833134d2a8a60720051621eb0 : FAILURE in 20m 37s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/107887b5ef6b4a2187a3ecd16544223a : FAILURE in 20m 00s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ca35e0735b74d959a2da3e9eee02fd2 : FAILURE in 19m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd890775689847c4a5b7fd49b241a363 : SUCCESS in 1h 03m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bdf09c3f92d4ebcbc8f98cb69417d36 : SUCCESS in 43m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c845d4f44d2d4515b03102bdc5d88184 : SUCCESS in 24m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/62fd617cacdd401f85dbf8e21155a267 : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"974901b850c855dc157a74cea8fb0a0005d15566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-10 07:22:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a017272538cf02a0360616541e9299c8afe457cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 08:20:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9159a94a969a4a8790369e2839518b47\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11451420ca854958bc95c1451d09eda0 : SUCCESS in 39m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a4eb682fcf8d45d6983c54ef2120b02a : SUCCESS in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d99bb45f617b34b216086525df72034e1175d6e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 09:16:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9aee584e8c14fa08be0e1afd9a0b95e\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/97d3d865ff7748ed9df4fc10243de703 : SUCCESS in 40m 28s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dd8ee8f2d88a42faabf12baa0b8fe6d4 : SUCCESS in 1h 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8438369a47724713b0a12055cd7c61ba : SUCCESS in 52m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd1b58493264468d966bd2f9df45b8a9 : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa04706c56f4e2c9f5786666e160be5 : SUCCESS in 12m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37840827ef364d7baa513109003d4044 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/478764121820455d8228cc9ffe863c7a : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047f7ba3f9c84102acf421964a8cc9cf : SUCCESS in 10m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cc2070e68b4b4077acc53cbb010996e4 : SUCCESS in 1h 07m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b7e017ea8a1f424b894a8aaad1ebae39 : SUCCESS in 1h 06m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/26502dce1d3648029c7f0cfe73a73110 : SUCCESS in 51m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9a8636db65344ac4a859115d6f25b4f5 : SUCCESS in 49m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3daa17ed9503424c8350181e9e82316e : SUCCESS in 1h 44m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/76f82e0ac8114f2aa757e77d48feaca1 : SUCCESS in 1h 32m 55s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/af69b70be1ad466c9a332cd86b0f4c48 : SUCCESS in 55m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/16e3931e56f34528b40b62eb1d1ff05a : SUCCESS in 5m 56s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68555b71a76147c0ad58dee6f6ab07df : FAILURE in 20m 19s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f7841e8df80457caee0a3a70ac9c4b2 : FAILURE in 19m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88e2934ef3e245dd8616bd2d31f1bbba : FAILURE in 22m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/98a87e35736d4952aed41d173e87c8c3 : SUCCESS in 1h 01m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff2abbf67c364c24a8b1b82b4a1055a8 : SUCCESS in 39m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fd9bdcee3f1e4ba6a1013521ed4f204e : SUCCESS in 25m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/772532ee56e84a3d9a5984b989736677 : SUCCESS in 39m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cb6ce2eabdc16a1e9809683b0d7cf0d32aef97b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-11 12:58:15.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"519e3bb3cab57bb27534531843c8e2eb72d35d1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 13:23:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5459af871dc1493c8cc4bfca663f7621\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb2500e120a143c89e8644f0c31accb2 : FAILURE in 18m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df571a298b8942f7ab2e2dd759901ab2 : FAILURE in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a7e9010706003f531ff11d55d1c9eb3626709eb5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-11 14:39:35.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/83/858383/9/check/tempest-dsvm-full-focal-py3/f9cbb66/ : SUCCESS in 1h 36m 01s","accounts_in_message":[],"_revision_number":9},{"id":"add0935b5da6b33434ab17746ab6e38f3768f00c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 14:54:00.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bbbba5e3cfa4b69aac9a44ba81bec69\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fe2c105397c94dd6bb0b5ece42d4d461 : SUCCESS in 52m 53s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bdb6db4a91484f6f854b4c1af4ffe33e : SUCCESS in 1h 18m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbf81d060b294353b41cdeaf1de01af6 : SUCCESS in 45m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6f059e3a2d5432ba690b40e04ce345d : FAILURE in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bfd595ec5dd45c8ab3d6d9a5e8f367b : FAILURE in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46b35ebe93784b5fb7dba581d7a9414d : FAILURE in 12m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/976dc923fe424a42ae9cd9ef4dc16760 : FAILURE in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076131e23df44f1cbf9b309df3fed614 : SUCCESS in 9m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cda32bc34e144b218ba52c84f97d6695 : SUCCESS in 53m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36a2d0392db143ce80026e108a941430 : SUCCESS in 51m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9375aad8681246158ca357213b5230fc : SUCCESS in 41m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a017592b235240f9b3c7ed1f30c6a5e0 : SUCCESS in 45m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d21f36a2c65849779bf8988ea486ef23 : SUCCESS in 1h 30m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/741ada2ab7f1415285d8eab4c1582ee7 : SUCCESS in 1h 30m 07s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c057c2cbde6e4cd9a751066372b1fb49 : SUCCESS in 47m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f71281979ce747e7a7719cd59e92fc9c : SUCCESS in 4m 51s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1640734aa46e4658b7ca123e76f03a46 : SUCCESS in 24m 54s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28d36b4f99724509a9115ed6678510ed : SUCCESS in 20m 19s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b43071c95ac347dcaa8dec33146dffd9 : SUCCESS in 20m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ca39bf9769b4b78875a5417d4a2bd3d : SUCCESS in 49m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80b9d48795244a31839f526ddbc45cb8 : SUCCESS in 49m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6c8fa05f6f3d4534beb6f808466af907 : SUCCESS in 21m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9af0eb43e2da43d195438da62bda0a85 : SUCCESS in 30m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"42e604d6c73209fef2c2402ea488a80adec2d040","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-11 15:10:47.000000000","message":"Patch Set 9:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/9/iscsi : SUCCESS in 2h 11m 16s","accounts_in_message":[],"_revision_number":9},{"id":"7b1fd6dd83dfb7e5bf08c2b94cd77789a5df0bd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-12 07:57:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3c16bbcf86f0c0b4c92e3d89104c17b3d99cbfa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:23:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7803d4aa6a11487aa6e2257e94600795\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5be8564f037a4054b379202213b8cabf : FAILURE in 20m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/22862c95d63f427f91671b638009aa50 : FAILURE in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ce30f817e0476716415126e08885bf610430baae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 09:24:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9ca64694d55446daafee3b965b197a3\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d9a42f6f55394d0599522cd52fa4933d : SUCCESS in 43m 53s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/472b6d4e8b1d46cd8eac0fde07c33eba : SUCCESS in 1h 15m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2bc8eb1a71945369669902bc0f01c2f : SUCCESS in 38m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89b33c7486b845458c05f2400b55c5fb : FAILURE in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1215f0a11244da86c7effb32ece8d4 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bac7d8f81b74e1493caf0a97b067198 : FAILURE in 13m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5c751cfdd714023ab6394a8c4a89b7f : FAILURE in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a8d9c0dbb54ef5b9edbc4c9a666ed3 : SUCCESS in 10m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fdfc658e78de4e3cbdd865bcb5fc4ef8 : SUCCESS in 49m 51s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b4dbaa01bc0f42e8bd10548427ae29b3 : SUCCESS in 1h 01m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7b4d27dec8274fe2a35f7b2e456aef27 : SUCCESS in 38m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f4331b07080b4beb99dd24516fdebe62 : SUCCESS in 43m 48s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/692ed52657b449628c5f17575c80e76c : SUCCESS in 1h 08m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/c54dfb2a1aac4431a9e5233017e9fc96 : SUCCESS in 1h 18m 56s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c3c7503268b243f4af9d14306fc0dfb0 : SUCCESS in 59m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/92b7a3708cee4f44af4cd45b3a4f6314 : SUCCESS in 4m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54563d662ced46e2922d9b2265a4e389 : SUCCESS in 20m 44s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82d3ac28a950434c97eee8495144811f : SUCCESS in 20m 42s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c594219165d14e76881af4f7188006ed : SUCCESS in 23m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4be0c750a3a4af4b813cbbe8b0ef1f5 : SUCCESS in 54m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/289c0336c3894b069b7dbe0dd77b667f : SUCCESS in 53m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fb7b82d9cbf14992987a5ab2f67fc608 : SUCCESS in 24m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2b0273a4f39c45fbac59ea5ad25c88e1 : SUCCESS in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"691d145471f997da022c805b1fc7139f6839acd9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-12 09:38:11.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/83/858383/10/check/tempest-dsvm-full-focal-py3/22388ff/ : SUCCESS in 1h 37m 11s","accounts_in_message":[],"_revision_number":10},{"id":"3f7eeec46fa22bc94622b653a53412b5e202cfd8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-12 10:22:52.000000000","message":"Patch Set 10:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/10/iscsi : SUCCESS in 2h 24m 26s","accounts_in_message":[],"_revision_number":10},{"id":"0888814330bdd3180f6c4b9fc3ef50143ff2d770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-12 17:55:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fbb658e1a8541e7f5790f2df895c20936145646f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 18:22:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f7e5973027f414b95d3b47b8a47af1c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d4d6b80836740069eec374955bb8be5 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f989be3d98148d6b2a7b32194b63ca6 : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5f07d6ca3d4aca5b128de0f619bea90d2ed56c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 19:29:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c87a954bf6814071a3f2464b336360dd\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9c3cc777de14e41859e6252b8d29d24 : SUCCESS in 52m 50s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/009fd345aa71464bb9798777eb662802 : SUCCESS in 1h 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b6139bc3d99476193eb72f0d4fbc30a : SUCCESS in 46m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adcc4476c98041b1bcccf9475c99ea34 : SUCCESS in 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58fc88baf0854e0a9b85491592b6811e : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/765ead3041aa4d4e990d1c0199cca2f7 : SUCCESS in 12m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01e9d1c830ce4c03835031e2036c7318 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcea82175d0e4dd1a849b5ea190c248b : SUCCESS in 10m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/48f8ae2bfed3473a8c450ceb07e428cd : SUCCESS in 53m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b190c1e02f542e39273880f501b1c80 : SUCCESS in 1h 08m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b6af3fe6221c4ab79c2727962ca2c8b8 : SUCCESS in 47m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9a5b35483afd46d293bb78f37231049d : SUCCESS in 43m 15s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5cd6f9e76bd34d9184a990b9b2f49e8a : SUCCESS in 1h 26m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/ed8966945734432e81b2ce837628f71b : SUCCESS in 1h 31m 47s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/864c9053dcf14e1b9caa1f70fd2cd070 : SUCCESS in 58m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/efbfae7a33594083b7d66be36acf6fa0 : SUCCESS in 5m 24s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a631e8db89d48b0bf225996e70ce5cb : SUCCESS in 22m 52s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7222e51168374e50ad7e3d837a3e81e8 : SUCCESS in 20m 55s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f47fee2e7644740972930ff34dfc48e : SUCCESS in 22m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5180af0c95b0414894cee8b2288a010e : SUCCESS in 1h 04m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4790d494eb20470c88f78b7549fc053b : SUCCESS in 47m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/621ee3bcf6544834aa815679b363a2a7 : SUCCESS in 30m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ab7d37a4130244b985a77a9c76683456 : SUCCESS in 32m 42s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"065e7338395caecabde2cf1b56dbf93a144053ef","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-12 19:30:07.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/11/check/EMC_PowerFlex_NOVA/bfd521f/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"2b349880ed5fb550979430cabfe9d4da8e48bd3e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-12 19:42:16.000000000","message":"Patch Set 11:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/83/858383/11/check/tempest-dsvm-full-focal-py3/87139a1/ : SUCCESS in 1h 42m 57s","accounts_in_message":[],"_revision_number":11},{"id":"056beb88d3fb467c1f1ef97e168c006af888f822","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-12 20:19:01.000000000","message":"Patch Set 11:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/11/iscsi : SUCCESS in 2h 22m 17s","accounts_in_message":[],"_revision_number":11},{"id":"ffe920446f16ad8428baf14a31d36fbd1e18d583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-18 09:44:32.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7765044ea53b9166b8e2ae198714e7b7f2966c0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:23:38.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd6cf2223384e35aaa911e0d4159615\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f417c3d48b64284ada84da64a4d6741 : SUCCESS in 30m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0efcbcba96254005a40b7183f272fe34 : SUCCESS in 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e5c98a302eb167b6dfa30bcca26aa384e8e0c9e0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-18 11:34:00.000000000","message":"Patch Set 12:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9ba138489b0a439cb3d1c8d0207f7dc2 : POST_FAILURE in 1h 44m 27s","accounts_in_message":[],"_revision_number":12},{"id":"3b4bdd4e58403fd3395577072ce5f0085084a387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:38:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9136ca91881e4a6eac02ece778b9dba2\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a884f33a36704d18b575938193d3aa83 : SUCCESS in 44m 22s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5e4d4a694901430d84ef7dc6a368d9ed : SUCCESS in 1h 09m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4bbf499d67a40efb40a0724b52aaa17 : SUCCESS in 48m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864e987d16d5485cae3af5439e4ba93a : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b32b68fdb94ff48bc9b61488d6ba98 : SUCCESS in 11m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab41938f8bca4014a3dd4f4e28fad060 : SUCCESS in 12m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9aa35ab00bc84a658e5d214e0245486a : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd01fc57318a41f4a75b4e9dc142ce3d : SUCCESS in 9m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8155c044017f4fd08e023b953ff25287 : SUCCESS in 47m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77d9f92dc21c470483b3723467d5c310 : SUCCESS in 1h 09m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/471ea46d9449488ebfa7897a0f6429b3 : SUCCESS in 43m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/67dce61f61b34e3b8047f63d7407d914 : SUCCESS in 35m 36s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c6dac65b16b6468587be30eb1518d518 : SUCCESS in 1h 21m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/b0f4eed922cd4a29b2c3d239328bd305 : SUCCESS in 1h 46m 20s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/21f5fef63289482b987be3773e5fddf2 : SUCCESS in 40m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b8577d4f5554436e9ae5c9318c8ef97e : SUCCESS in 6m 01s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/628f251932fa479faa5f463e6daf609e : SUCCESS in 20m 26s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a272d4ddc03f4e2088d84344fc05e7c0 : SUCCESS in 21m 23s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95f5c5d88f414841b4f26b2ff89600aa : SUCCESS in 19m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d0fa903158944c598f38061541eb28b : SUCCESS in 1h 01m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e8cf6ef082244989b74fdcb3ad2b1fd : SUCCESS in 46m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b4d035adc26d491c8377c48b9e27ac07 : SUCCESS in 21m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/74d34b42e7bb440496cb4f441e3dfdb7 : SUCCESS in 38m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0e6fe2b8e50f45a9e9e4cca64d875ec8448ed111","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-18 11:39:20.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/12/check/EMC_PowerFlex_NOVA/6bb7236/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"0caa6958ee0955db9bd6bf8b1172470b26299929","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-18 11:47:32.000000000","message":"Patch Set 12:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"0e665fd4d476cb3533128db193ddc55dc7ce5c5a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-18 12:07:22.000000000","message":"Patch Set 12:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/12/iscsi : SUCCESS in 2h 21m 43s","accounts_in_message":[],"_revision_number":12},{"id":"4d7724fa0d336bde5d42b36f9c37b964ad75ab83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-18 12:27:54.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"50246232cbbcf030a155a8bda923c8f72ed6fc51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 13:00:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40f03f0361d643408f34de0fe5e9571a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70c983dcf4094e0ca3b942396e0bb145 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3701afecd56a4e069744f2ce9d771188 : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"874237ca652e764e4afef839770a2cff1f217f85","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-18 14:21:19.000000000","message":"Patch Set 13:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/164aaa383bc94c34a01c05175ba39f13 : POST_FAILURE in 1h 49m 13s","accounts_in_message":[],"_revision_number":13},{"id":"75d81a216b08c7ee494f225ac2f57bef735551f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 14:26:20.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be5d891f08dc404188417460268454da\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/777045cfd55647b5841cb030a7b7ba7d : SUCCESS in 56m 21s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c229d097e6774235843c5c926302be54 : SUCCESS in 1h 29m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a602cd4ba3874743981a69f82b1ea0de : SUCCESS in 47m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aa7d260c0b241b59126b7daebab3b15 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdde5769ab344a0799b04bb8f7c43f27 : FAILURE in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c51a8a748a2e4b5f8ca61af5e9cac4e3 : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ffd8eed565a4d84aa7a4368d8b268ed : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc1deb339bc463d9aa5970487cd5df1 : SUCCESS in 8m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/28fecbbefeb84b7ca4f6c14859ecf8e1 : SUCCESS in 53m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0a6360688f443ac9bbd8bdc80d84fcc : SUCCESS in 47m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1d20b4158bee433e8870b21160221e16 : SUCCESS in 43m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e4107e76674a481b925dbf67de14c557 : SUCCESS in 40m 11s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/00d34b496d1c49d4b4e282b73f04d6f7 : SUCCESS in 1h 22m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/ab12048b47f742ee8b1441d3f2fb40fa : SUCCESS in 1h 50m 56s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/420bedb054da4be1a6b4226dba3480ce : SUCCESS in 41m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d2512e4584b64aaf8cd372801cea7028 : SUCCESS in 4m 26s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/436f30051f7b46ebbcb80b10c700a0c9 : SUCCESS in 21m 10s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4511d6d2fb3d4d9099629598d9cc7e01 : SUCCESS in 20m 35s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4fea259b130b42a7b5af506de5690616 : SUCCESS in 19m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9050e3c528e7429e9fc0782772d2e4f5 : SUCCESS in 1h 06m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06ea26f1a0314433b04e35fb58a1ce10 : SUCCESS in 43m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2d8644a755044a91ab744a5780a6f16f : SUCCESS in 20m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/50089237e0c7442984c4519750ee2c25 : SUCCESS in 29m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"555914a403beb74bb940c4f8d1db5ad95223e696","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-18 15:08:43.000000000","message":"Patch Set 13:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/13/iscsi : SUCCESS in 2h 39m 10s","accounts_in_message":[],"_revision_number":13},{"id":"8cd5d8626939451f36e5c15bcf4f336d391dcdf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-18 15:12:19.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"79304fda734927bb1c98579998b44d583eb64339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:57:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a89d7bbb71c4553915fe8bd75e3fc74\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f139bd8ab94436abc58031e2e20aad9 : SUCCESS in 37m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/db9d66a9942f47c0b9721909b152f0b4 : SUCCESS in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"88ef88fda256464496b6dd9960153b16447bf5db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-18 16:58:27.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b01039c48783433282923629aba66d0d : POST_FAILURE in 1h 40m 38s","accounts_in_message":[],"_revision_number":14},{"id":"b1db989d9d384db3a471248a8b72a2390604d4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 17:02:54.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb1e923d3a04aa2a3c4940a5ae7730a\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4fa65063f6ad4ff6914e6581f619586f : SUCCESS in 37m 59s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d67fe181a2b54a0c8888c84150fbd9bc : SUCCESS in 1h 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f87c43dfb954a6e818be59b3ec009b1 : SUCCESS in 57m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65e9b62add6d45e582233c9cbceea73d : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5107df8547b4181b21edd4875481f5c : SUCCESS in 12m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95639da899764d8ca1313904c519e0ee : SUCCESS in 12m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83e9efe06887400fb64aa6f8149e6b93 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7710c7ec15e4d9998e65ae7ffe264a5 : SUCCESS in 10m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66aab3ab1ae14de187c4290eb3c9c5d5 : SUCCESS in 1h 00m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b3ea0ec357ff4a1db15874a2247da477 : SUCCESS in 1h 12m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/feaebe28326b40318cc72fe2fe15dedd : SUCCESS in 48m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2f5cf12082bf42c4ae959308b8351c0d : SUCCESS in 50m 51s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/232127ca24194cf3a56593311fdf570a : SUCCESS in 1h 38m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/980e35f9f22646409021c09453758d22 : SUCCESS in 1h 43m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c12953ba6ccd412da54ae5ab312e15e6 : SUCCESS in 55m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d9a47a8566f9460c807968c44f3a1f24 : SUCCESS in 4m 41s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/469139500fef4b91a1769e3134009bbe : SUCCESS in 20m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9dd0457c6b364c08bee5b50ae8ea3cfa : SUCCESS in 24m 26s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d0f72fdc72024ba5a956a73f69503ddf : SUCCESS in 22m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f177b8337cc1414ba338caea205e490a : SUCCESS in 1h 15m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3d2fa7ee16947499119f57a8736ad88 : SUCCESS in 52m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ee47142d5e7544d5b088ab2d818aed76 : SUCCESS in 28m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cb71522c833346138364b5b6ac44cc0c : SUCCESS in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"8900ad97bae6a5652e2d1322db0830496610a357","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-18 17:03:15.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/14/check/EMC_PowerFlex_NOVA/9b4820e/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"3663f1cf06c22308a3b433bcda9825bc384daebb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-18 17:10:51.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":14},{"id":"20eeb9c3b770331476769e57ffe293363a5f90e6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-18 17:54:52.000000000","message":"Patch Set 14:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/14/iscsi : SUCCESS in 2h 40m 13s","accounts_in_message":[],"_revision_number":14},{"id":"69401d5ed4b65110d59ac8599eac210d63cae30f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-19 11:48:51.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"8ccc7a5a2c14800c30683d584002c83e5ad46059","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-19 15:33:58.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"037df8766c8f71a8136a4e946547806944d4e510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-24 09:34:56.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"79a52bdccfd58879bdb4e7e0735ff6ec80512e65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 09:57:37.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffa6d3de0734bb8a5338ed0dbfd50ff\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecd6551e73a348babd3e2dfec18b48c6 : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ca8b1f25e54432b872206e1a255a6fa : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a0343f3fd1be28e076d17811e553b290922cd3f5","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-24 10:07:51.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3f1231c12b5fe61d6513142d29cb12c6ac9e4565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 11:26:48.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d1db4925e614b42a96ede1c61d82391\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a58dc04da49d4310b8d7c9956abdcefd : SUCCESS in 1h 03m 19s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bcb98c454672446ca1a0f39045ad5f55 : SUCCESS in 1h 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae1b3146a90b491983e41e1199146997 : SUCCESS in 56m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee101f22915d4ad39d0995c90e25dbcd : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b29df53422340e482e82e616a818472 : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f95480d60044454d824c3dcac86a96b6 : SUCCESS in 13m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ead5f6ea98046dbb3e58ce62f985d6d : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c047306e0684f42874fef8a648b0b7f : SUCCESS in 9m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/30a466494f464292ac70bbcc51e3d5a8 : SUCCESS in 1h 03m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b26dcc55b0394c998420713c4d1f0851 : SUCCESS in 1h 10m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8b9cec4ce41c4ddd8fed777e844a7c69 : SUCCESS in 53m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/75b042b5ed5348d8a6817fadf1cbe5f0 : SUCCESS in 56m 56s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/298bbdbe23ad44559c4e956fb03bb258 : SUCCESS in 1h 44m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/fdaf37266007406fa6da0eb034ca6a32 : FAILURE in 1h 43m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/685f93f024c14cf8a108d905afbe1f97 : SUCCESS in 58m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5f52ef4f598c4dcfa489d9ff51f11e7f : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1cd08435854943a2a8133a1bce660b44 : SUCCESS in 22m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/888fc70f96054bfbb2bdb1fdb810f2a6 : SUCCESS in 22m 37s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/224fd651153342e69e52deeb2a9b1095 : SUCCESS in 23m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1995d50d8b954c00be1daf17ed0ecfe0 : SUCCESS in 1h 12m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e17b26a354148a39102d0bde56062fb : SUCCESS in 52m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bfb7cabed55f40858bf9d2bfc6c4ba4e : SUCCESS in 33m 29s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/399150f736804383b087c3c02ca66be8 : SUCCESS in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"01522af26befd4e444e92b95da336ae514e1236a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-24 12:01:43.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5fb6e84070df441dad2fdcca35c88f82 : POST_FAILURE in 2h 21m 16s","accounts_in_message":[],"_revision_number":15},{"id":"674e93b8e29466cf869cd86da35fe6ffd8b0fe32","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-24 12:04:47.000000000","message":"Patch Set 15:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/15/iscsi : SUCCESS in 2h 28m 35s","accounts_in_message":[],"_revision_number":15},{"id":"6fae48622a731a2433c4e33555f77ffdc780b6d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-27 08:12:47.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"973c848dd5ee84bf54623262da25be08aa90a448","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 08:43:59.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/087656b586c341d1a7766af06338b28f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4a66b91c79f4eb28670b78935d6f09d : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bcf9bc9eda2c44f9b81ae08b218903d8 : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"38db91c081cc94aff2d8dce7629571a86afa6caf","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-27 10:01:48.000000000","message":"Patch Set 16:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fa1914e98443453ba5e8275783ba1a6d : POST_FAILURE in 1h 43m 34s","accounts_in_message":[],"_revision_number":16},{"id":"bb41b5203a0fcb0abd43ec1eadae19f828349adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:01:49.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86d16acbb9c649449bc64cd4c3a90406\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5294b31a249b4d2bb8b38081e4f75a1e : SUCCESS in 40m 59s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/005e1934b4b14737a5506b1692674abe : SUCCESS in 1h 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cfd0493618d4348a2e6832d17f32985 : SUCCESS in 1h 04m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9320ccb7a2a3404ca637fbccd60d74f0 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec06b2eee0ba4160b5bbbc2f4911dc99 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeb921abc3c84a548953608ed1d8edac : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d05c7cca56ab4168a2429439582db0f5 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bba84f237144780939fdd15b77a83d4 : SUCCESS in 8m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f5bb4d0a92f9425b8c5b9f59b6131c81 : SUCCESS in 45m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8aeef7658b5c47299ae68d81697b37f0 : SUCCESS in 53m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f77b3fbb1d134604a4ef08bc3280ab55 : SUCCESS in 50m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b6d25a33afd14cd783f8c3405322a56d : SUCCESS in 48m 43s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/67372ea732f8482790d15333ff732470 : SUCCESS in 1h 42m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/b0cfbfddfbcc4f7080d4344066af8bce : SUCCESS in 1h 36m 44s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/fd8d1b4a045644fda3ba317abf47bf44 : SUCCESS in 58m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6faae0baa8f14984aaf333543b8b2cc7 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b22acd522a54bfcb958bd3bbc91c4f4 : SUCCESS in 18m 32s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2540744e14d41119d74c5118c0a0142 : SUCCESS in 19m 54s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0867d3febf2145009445d9e3925177a4 : SUCCESS in 18m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa5e438d61dc44a48726d7ca163f8c25 : SUCCESS in 59m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42f8253d8a2347498f4f27a6ccfb5f3f : SUCCESS in 45m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/783b2bc3ab4945c09df45785f63134ab : SUCCESS in 25m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f17764e3f97147d9b56e6c800441687c : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"30723567db841adc422e6710424605aded38febb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-27 10:02:12.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/16/check/EMC_PowerFlex_NOVA/9e3b6b1/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"27b569939c4988130a6722ac96464103507218c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-27 10:10:34.000000000","message":"Patch Set 16:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":16},{"id":"99d8a3df7572c43502a19a299515d11f8a5a3235","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-27 10:56:28.000000000","message":"Patch Set 16:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/16/iscsi : SUCCESS in 2h 41m 16s","accounts_in_message":[],"_revision_number":16},{"id":"6d81c462bef87d2047dca5b5918b64ab517a3033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-31 14:27:44.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"344e7e1c8d5dd162f7d06fa8e5a3f013b8849fd1","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-10-31 14:29:10.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"d444c2d1ad99cbece1376fa156a277277055a58a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 14:54:06.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a80b9da92364d9db869f3c3a84e84d9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8fa7a08ceef4534a2b7a33a089d889f : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d921715624a453bbbba5dd9305687f4 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ee032178bae3efbc7c382df38a502e7353537cf1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-31 16:12:21.000000000","message":"Patch Set 17:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/0968c8a08f444cb1a8a3d66f2432f983 : POST_FAILURE in 1h 39m 33s","accounts_in_message":[],"_revision_number":17},{"id":"284d1a9e487fcb08c2a9e2064890143e41418c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 16:27:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77e8133322c45e88c9e10b448f4aa9b\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cabd0918cd0047e49e6c585d73ba8452 : SUCCESS in 1h 02m 23s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f1a6a75edb064e68b60d11f364003b94 : SUCCESS in 1h 26m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d334f0eb9a144e48706219863f59119 : SUCCESS in 53m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a52d3084f5404e5c9f7f208bf3d54196 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de020843f2594a5ba3e998e387edb4a1 : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee92955403b64c4bad0d60496d97dae5 : SUCCESS in 12m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eecd346748743b8813fb7076e3fa147 : SUCCESS in 13m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9efa77a6ec4cb0aca2596d643e98e8 : SUCCESS in 8m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/763cadca34024392bd07183ff4e9cda4 : SUCCESS in 1h 11m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b9e9a4cdc1f4d0c8dc9334f0bd71f78 : SUCCESS in 1h 21m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/55be0db36c7e43279737e820793178d4 : SUCCESS in 53m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b4bea148eed54602b3939c322c26bca2 : SUCCESS in 59m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0f6b6daec76d4c8d830e6762e5663206 : SUCCESS in 1h 34m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/18bcdea199f8417590bea44b29fffb59 : SUCCESS in 1h 49m 06s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3dc5478130cc47918804390e7bde0107 : SUCCESS in 1h 02m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4e48fd1d7f4b4ba8975b4350d7bc75d1 : SUCCESS in 5m 00s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a31cd5b8ca1483c877147285856dab0 : SUCCESS in 24m 03s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cee63b91d7b544cb84a281688c585d90 : SUCCESS in 23m 18s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fd7843de02334347b0e42a3c70a16490 : SUCCESS in 24m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4842901d6ec4acc938c027a6dfca354 : SUCCESS in 1h 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb620001dfe4952a211d114a6babf6e : SUCCESS in 55m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3f7cb9491ddc4641b71823bcf78f511b : SUCCESS in 33m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7b15cb185dcb4c35bbea584766f685d7 : SUCCESS in 41m 08s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"56950d87f8192f31c348b5cbe9c7f6497c9da3a8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-31 16:28:11.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/17/check/EMC_PowerFlex_NOVA/935598e/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"c16c37b566cf11115938f4fe059fac438ccde9c7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-31 16:35:55.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":17},{"id":"31d345757ae2d0f75995629a4d6fbc0a7dd587ce","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-31 17:01:37.000000000","message":"Patch Set 17:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/858383/17/iscsi : SUCCESS in 2h 32m 42s","accounts_in_message":[],"_revision_number":17},{"id":"6a1b03864fd2570b77fca3286cf2c4097d1c1998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-21 13:19:41.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"a3d14e954e074a1de60786cfc67a32cac31cabfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 13:57:07.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13cbfcfd1205454f80c6e92e1f94a601\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a295a465f77741c4b602236168d4bb54 : SUCCESS in 29m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1a734549c8414df9bcd491e50c85d54f : SUCCESS in 30m 11s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"eaa24a0a8d89b11d61448c0acf850a6c92e23950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 14:53:04.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9baff77681174c0392d00b9c7b5138ef\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4a4bfb2a12cc452c81f06081bcebfc07 : SUCCESS in 45m 59s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d3f6f616c911416caca2f7d341d551f6 : SUCCESS in 1h 25m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b78913ed73434a7492b9c7d3385856fb : SUCCESS in 49m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40172643055c475584353f17da40d000 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc6437842254674b5e6f5976afad414 : SUCCESS in 11m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8648169847c3483fb79e1724dc9267b3 : SUCCESS in 12m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60186a4f8ea54e9ba4261c16d48cd662 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e62a9f19456443db275146476a2db00 : SUCCESS in 9m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee70199b9c4744cab16321385a95242a : SUCCESS in 52m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ea8448856dc4c8cbbdd629cc7d61b7c : SUCCESS in 1h 01m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/15c4391a89274745895c9d139731a8de : SUCCESS in 44m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d36319e15e894ba7a68e7ab37b412c24 : SUCCESS in 55m 04s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/871b47a90d2d44559685d8b265452d95 : SUCCESS in 1h 23m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/6b8040049d6d4ec888f60b387629e48b : FAILURE in 1h 30m 27s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a8fd09542c9d4582ad829ced759093f2 : FAILURE in 46m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8656618c204f46a3bcd9e6cb17864caa : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b711858e86c34ce693db0328668d16b6 : SUCCESS in 19m 19s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7907e6367d294962a87467a259dda538 : SUCCESS in 26m 00s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/062d3dd31e5844b884bf48e6b92b623e : SUCCESS in 19m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9c21c61d8a740438f70ff814f68a2ff : SUCCESS in 1h 02m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84140661e1bd41cfa9bade5529d1dd4c : SUCCESS in 40m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e0b64ebaa54d4eb0aba26b397f7d75bc : SUCCESS in 45m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0da243e66a0d4465bc181ede07c0f5f5 : FAILURE in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"f094371acaac821decb7d004da18a32db969c74f","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-22 07:08:53.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ca0eeda1ff4902037ed0b319c62ec12241e6aa2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 07:38:36.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ae073f9555480387ced12152cb5af1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95fe9bc1ab9e41cb8a572739353da979 : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3a9e6deeaa9454aa1f75aac67cdc18b : SUCCESS in 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"8cdf59c28b3ee4acbd63f6ba37f5399ec4780312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 08:49:49.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2d67d2532764d958a91855785fa8f39\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4b41574d1ed74c46ac9b867e6f84cf5c : SUCCESS in 42m 50s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/914c286262874ed5b856360c5fb8e651 : SUCCESS in 1h 25m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e906ef937e134b23accc349fefd06894 : SUCCESS in 50m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff3b5e98594046ee8aa6875e4805747d : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a155993df940aeb74afb8b40942faa : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da5e506124444e9bb7a73b0e03eb58b3 : SUCCESS in 12m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d493e536ddd642ecb90849e30f1fea08 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/116cfb0cbdc24ea9a927fc5b7070d4f9 : SUCCESS in 8m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a0e95a24d770453987eb54435fa8c830 : SUCCESS in 48m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/643df849e361480c8519c7a01206a25f : SUCCESS in 56m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c38c47bcb934be9bdc731f4486a8592 : SUCCESS in 35m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/084eb79828aa492582191a9a85e1cc99 : SUCCESS in 37m 02s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9dad9c7eefbe461cb960811915e56308 : SUCCESS in 1h 27m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/e6992b77922546dbbdfe1ffe699a3e87 : SUCCESS in 1h 38m 13s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6cd913b7c1284c75b79fd0269d40f2f2 : FAILURE in 45m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4c151f53f06442ad845485a5b2aab666 : SUCCESS in 6m 17s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2357efee7f4a4ac68fb43b63850fcf6a : SUCCESS in 19m 48s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a347812d1b2f472dbbd7772489afebe0 : SUCCESS in 20m 12s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d7817d77dc7441db9c867df125ac6cc1 : SUCCESS in 20m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/56383d9add7c410fa591bb36947344a7 : SUCCESS in 1h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/303eb3c76244420b8cc0d6877de0ad18 : SUCCESS in 40m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9d35139d80134486bbf65e6b698deb56 : SUCCESS in 24m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6c01edfe858a415bb3597858eb674449 : FAILURE in 32m 42s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"659b2dfdb3b61e2ad04372d5cd12ca042415741c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-24 07:02:51.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"ce090cf6f7e00e9bbf399768e5190a6c8915caf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 07:53:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1aebea8fa4643d7ad611c3c36ebc397\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/579d0219fcab4b528a21daa6a03e2e2e : SUCCESS in 32m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ebe3d033db44430d9633e1691a53d86c : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cabf11a30187b4bc261cf8cab231a25208464e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 09:03:58.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f375cf10ca3d42ad82b380422f7b9bf4\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9d654e4d67224242b017f3c444a37984 : SUCCESS in 1h 07m 06s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0c0bee1430d3481486bf17e89fd80f8e : SUCCESS in 1h 25m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf5fb0f139594fd0ad64fd84a79dd17a : SUCCESS in 1h 01m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8023f006cc0416787fa00e2210d07de : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c3847937d74c0da2257f4b8a51f17f : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59690997f81a45638f4d5622c6471ee6 : SUCCESS in 13m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af391f053e44b658bf71fb469aed856 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1446bf7081de49cea6a83cdbe429fa3e : SUCCESS in 9m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ae0f3115af984c62829de76b964dcae3 : SUCCESS in 1h 13m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/adcbd26a72bf48168d4da65404497066 : SUCCESS in 1h 03m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/74b50d5231bd4ffa89d91c931b0cdd0c : SUCCESS in 54m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/16c2ceb9d41241b488b55eac26a52662 : SUCCESS in 42m 40s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d1c1f08b84a74e7592a4ac29b70b4eb0 : SUCCESS in 1h 22m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/d61de317a66b47719b7132059ea0b78e : FAILURE in 1h 52m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3f15ef4513104c1c95c08864212788af : FAILURE in 46m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/074d7f0a0bb24b87a9e7eb69e0a57f83 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7de151a3d5bf4fb69dfc582e0f09c785 : SUCCESS in 20m 29s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6084e4cd0a6748149f20717302e7bdd5 : SUCCESS in 25m 53s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66d511c7c134414bb79dd60952c481b5 : SUCCESS in 20m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/da5ec79655194e0ba6d9d735803005fc : SUCCESS in 1h 05m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e513a895b92040aaa6b0f57d9fa0b353 : SUCCESS in 55m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/07b10a74190d456081e42330d47256dd : SUCCESS in 30m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/363e6216ad62412ca348b4fa4ef4d794 : SUCCESS in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"9cb8de0e28fa57e370794fa704dea8c120c45107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-24 19:20:36.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"ee264feacca8b5a66e5f14a23c2aed425d5d8049","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 19:49:14.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1f5c0fd6624e76b9a560fe7ae654b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa48ca7e2dec4294be86446c692e4586 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45599f53b91049b1ba8a8f9481ee7418 : SUCCESS in 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"116a75a5e3e9e6890374876ab222a5491b933064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 21:42:55.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7599e9000620471d8b408eb0bc9fe2ff\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d604af927c824cc899f3c9771a841f8a : SUCCESS in 52m 58s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/33fc16db523145aba1abe689c568b45f : SUCCESS in 1h 09m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2271b06eca9841acb59beb75c530b31f : SUCCESS in 40m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e3e556b008f4a1d92b347dbb55ca2c4 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e68fab01d01495c9265228e883c50ed : SUCCESS in 12m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff2518851a674a53b000bea4c30ab08d : SUCCESS in 14m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc971a4b0ee140efaf7f1fd85f676451 : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6dc417ad2cc49929f3b0f57e979c9ef : SUCCESS in 9m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/563f6b5526bd484e9bfdca211d50539e : SUCCESS in 47m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/69fc96e2341245928acc7a34e16bbb96 : SUCCESS in 1h 05m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c2fe06e72e8a487987bbc705642d7f5f : SUCCESS in 43m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/72ff66a162c3407d8c25251c90e2b628 : SUCCESS in 53m 09s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/81815626b32746dba9f935a4544f3c3f : SUCCESS in 1h 17m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/1a30c1788e5247cc96f5feaa4a51e7c4 : FAILURE in 2h 14m 52s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/7d911078f6a44f6cb8cb881bb6f2d487 : FAILURE in 44m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3576aeca739f433e86eac686e3de9705 : SUCCESS in 3m 59s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fcb9343cd8114ad5b72a0e9319280575 : SUCCESS in 19m 16s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/145df87cae5445a994b5a19526eb3712 : SUCCESS in 22m 37s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/839398a0c2f045ee985ac880bd9122a0 : SUCCESS in 20m 43s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f10178f1003d4a3eacba166555606d15 : SUCCESS in 49m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0682d0326ba44f8e97b9d1ccaf73d020 : SUCCESS in 40m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/aa83cfe75b9e40b1a4d108344fa4c7a2 : SUCCESS in 23m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2a3b448993c440cb952b98a301a7152e : SUCCESS in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f3a4760fa472ce17e6ec88089507ca720c81b08e","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-25 09:35:46.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b4bbac7113496ba7cf0ffc49952ff4efa0c60b6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 10:09:46.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b09f5f78da14d6f999f8c1ff6d81537\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/445679e824794a95a9ee6ddace24bd7b : SUCCESS in 28m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/521606d9fd564139b7b32c2de6c492cd : SUCCESS in 32m 09s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"b62af04a1fce147c2164eea85b415b6e6243fb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 11:17:20.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67225714e5304b4d8e35b1172703f63f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/462e5d5560574e57b01c5cd8a48643ad : SUCCESS in 1h 04m 39s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d0f01594e1094cdba7f27f39b28ed204 : SUCCESS in 1h 27m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a79cf9b6c0ff46d7982dd1b06821552b : SUCCESS in 57m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c65c511ae374b0785452cdad6c2082c : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b698397633d3455895f6c10e19d094d3 : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d695b6f1deca4e1393eb115e8a331c74 : SUCCESS in 11m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4df6cda164024ba9956b9cac4cfc8fc4 : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4d0b5376704ee18c3a058a7f3576be : SUCCESS in 9m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5931cf30c83e4f58a5a429c272c87684 : SUCCESS in 1h 01m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c6d852a0da274dc6afb9642e9267a1fe : SUCCESS in 1h 11m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/37efcd3645f54871a5a388708651c515 : SUCCESS in 47m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/33eeeb97fb4c4e95954ce1420aa68cd8 : SUCCESS in 48m 40s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b0bef3938aa643d0bf28d824baf59b67 : SUCCESS in 1h 33m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/22b2c1fe6c854ddf98184a866989bf64 : SUCCESS in 1h 36m 19s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1886e4420e6c4f6e91de2c5ad9b122ae : FAILURE in 1h 08m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f5efe55230994b238a75deb4e2b5b600 : SUCCESS in 4m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aaa1c9b2f61d42ec96c35cd18438bf67 : SUCCESS in 26m 03s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e4ba8102bd84e28b72a83a72ea5870c : SUCCESS in 19m 44s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/02093ddcd1934e59a8de0485857cb052 : SUCCESS in 24m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/195ac50fb8e240dea74f8de7dfbbaeb1 : SUCCESS in 1h 03m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dce287a5269045d2b7545c41138839c5 : SUCCESS in 50m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7aeaa0b27f6440b9b39d3928643b6a8c : SUCCESS in 23m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/38b6baee07244ebd9b74413abfc10cae : SUCCESS in 44m 13s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f8e0600e3c35a81d2d10a2b9e0b0421239a41b65","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-28 11:20:43.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"152f58f9f211cae8e942db0b1050f12d159a50ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:12:27.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb75b013a777496b8039a3002d4078d2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fe7d922aa7d451fb45b9842c1b866f1 : FAILURE in 43m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bf4204b01614f56ae57001e214a87c2 : SUCCESS in 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"c3169b555b44334f2440a24fafaad9790b6c1469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 13:17:37.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f64e81400c7d4ca188bab32565e03b06\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/593898b30fe5443d9fe8c827f0485712 : SUCCESS in 52m 15s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/35adc41972b14093971251a1bb03d648 : SUCCESS in 1h 23m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fd8245958354d709535c2f9da72f3f4 : SUCCESS in 48m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e7c80a470c34ac89a515cc5c9f71c93 : SUCCESS in 15m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fe627c9db5410189064bd9f956bf29 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/624da0a331144e0180bc052b0ddb426b : SUCCESS in 12m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82eaeada583f42e3a5d2a54569ef649a : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e231dd3f18401ea20e66bb2c7d13e5 : SUCCESS in 10m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8abd9e5213dc4c0a8f25dbedbd5612d3 : SUCCESS in 56m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/213e712bf8f4413797270666a4c7ff1e : SUCCESS in 1h 04m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/40858c2f6677418baeebd56651c42f68 : SUCCESS in 43m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2856890f617d4eceb6a5f2066b54dff5 : SUCCESS in 55m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/29f8cf86de394296a71160e2c5aa509b : SUCCESS in 1h 36m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/78094e45b3d04285b49e5ad56a6864a3 : FAILURE in 1h 49m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c87e2cff29b4481a851ce6e3b9de3e56 : FAILURE in 1h 01m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a948e0eb41654c6eaed36838bc26e228 : SUCCESS in 6m 00s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ee6e4347d094693aaf88d6d38a03623 : SUCCESS in 23m 34s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f26dd54cdf6b4177815a701b54b90a9e : SUCCESS in 27m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3eb02c576af14e0d80210d9d05cf8740 : SUCCESS in 24m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/966c3bca23f4447f8f690f9e2ecc8415 : FAILURE in 57m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c6f60e27a6b4a94b3037239673a5918 : SUCCESS in 50m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6bd7e9f77eb64ab2bb91de9fd1f1b64c : SUCCESS in 21m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ff0b1f0846e945faa73c1e51c36b3f42 : SUCCESS in 45m 49s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"97a60e1fc5a66e26a2ba0dc7f6cbdd4212110aae","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-29 07:44:48.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"444d19ed65bd6f31b32da2ec9bfb1bf1c9b68392","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 09:08:06.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d42bbfe082b74cefb368c7ef59134817\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd510f0a56f54d229d0884d658a8e649 : SUCCESS in 36m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98c67bcb545840e997a0cb2d62fdd4c6 : SUCCESS in 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"fd33052389b9a5166e6c1fb4dadd061d49bb0f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 09:45:37.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa810239b44e4dab824cf7f49ec7b0c7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0d701c081f4a4ccaa16d839bd47e1a19 : FAILURE in 49m 25s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8ff743be52914b50b2933a9ace5a34c8 : SUCCESS in 1h 38m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bf9d271c8f24470a28bcb2b2e29a955 : FAILURE in 54m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6324d9a6f7fb432bb3fa600a01f34412 : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e3a6dddf1eb4009b6cd246ac20400c6 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a87be75358aa45dbbabaa31f424526e6 : SUCCESS in 15m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21a46f794d08446b8f68fef40f29ae6e : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774cff3937b34141b6181f97e1e46279 : SUCCESS in 12m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/065be9c031774a319a11fbc3388a7efd : SUCCESS in 54m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/539218da5ff04ea78e22be24e0a13f43 : SUCCESS in 1h 12m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2ae3f3592d3b420294e3c394b6e714b4 : SUCCESS in 53m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4c9254085729475b8d8282351138b858 : SUCCESS in 43m 00s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6a86426aa56d4b1f84575eb4ce8b6e14 : SUCCESS in 1h 53m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/2514ade662474e4ab734267e780b6b61 : SUCCESS in 1h 36m 44s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3571a0eb82b5471e954ba10759eb795a : SUCCESS in 1h 04m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e1b9d8d7c22e4a32ba97c77e45531b42 : SUCCESS in 5m 26s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cdb442df8cc9416d8d89fdb0d2c128fe : SUCCESS in 21m 01s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7f509cce643428b953dcc7a75163ee1 : SUCCESS in 21m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2eaeb27a34949a88c6e5146e41acd6b : SUCCESS in 25m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/563f35954cc44e2e99f307f372ac023e : SUCCESS in 1h 02m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d18ced964b0943f59da5321bdd76c5e7 : SUCCESS in 54m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c5fc5f6b3b9142459d63343a3c25a3ab : SUCCESS in 24m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b8ce02bb0ede42c89018da73b89f229c : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"a73a9b93f7694fccdd2fb57fd1c5286bda4ec4db","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-11-29 11:04:30.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"091290edab8045fa512c6596d58ed94d2d1340a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 11:35:03.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa6ee9f07e30476dae26551e4f094943\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df95d2b72c614a6a91469e9aa40fc759 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f979ec119994de7b9de6aeec1f321af : SUCCESS in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"eb2cf3d9289047fbf61b296498c83bf7ac563bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:37:52.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e5edd33f8e44118dc6f93dfbbcbfd8\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/61634b731ec84083a32caa0dabf29927 : SUCCESS in 45m 15s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/288365564e7b4b8aa029b9035fd69c64 : SUCCESS in 1h 22m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34e0e71ace41437da6b45558327023d3 : SUCCESS in 44m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85714eb1686c44fdb99c072a96ab148b : SUCCESS in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5ae99c5d8647268314c2925de55bd7 : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8160281e19c4afcbed0dcb6dca467a8 : SUCCESS in 11m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fa0c3ff9fdd4b6495a39fb627ac537a : SUCCESS in 14m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15e5706746c4408aab62ea07ce633a1 : SUCCESS in 10m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9868551df4be4be4aeb88f9e5d274ba7 : SUCCESS in 48m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a90ab9ba73f4aad9bc54821870564b4 : SUCCESS in 1h 04m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6bf417febd2e40328e837694f33bbe38 : SUCCESS in 44m 12s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/55082064857b41418e6e2cacf9973795 : SUCCESS in 50m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/823aa21337264827989cef735a3ef166 : SUCCESS in 1h 30m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/fb19fa63096849d088fe0d6ba0185e90 : SUCCESS in 1h 30m 33s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2caa265bdd3c435481b4c9761b778be3 : SUCCESS in 57m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/eb2b7c84db2c44f5a71169a6486b93b5 : SUCCESS in 5m 44s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/941394caef774c8d90bec2c0dec4eb50 : SUCCESS in 22m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/323d05fa3965495989196489792c163a : SUCCESS in 20m 21s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52ef144f87f341b28ce154e72f27fb4c : SUCCESS in 21m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d62e4d264f234e3082092256a65c240c : SUCCESS in 1h 01m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/765278774a5c46058b91241ad56c6707 : SUCCESS in 46m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/136d6de6b6b8433b8c034afc1f0a088f : SUCCESS in 35m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/296f6b1a68ba49a1b24c7b176826c99e : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"ed9e88e95130be7397542d5529569ad431fb555e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-11-29 12:38:22.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/20/check/EMC_PowerFlex_NOVA/2299932/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"a6de0f3909bbe22f63e4a285545eb01930b637b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-12-14 08:35:40.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"3ac89b4dbdaadecae550e78b6e80a90e28f0659f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 09:14:02.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f568570533ab4adcbfa28e7e797353b6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72e1e2bcb9184b3580e4d81fedadb4f9 : SUCCESS in 36m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/601829107e954517bb1b4997aa6471cf : SUCCESS in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"669bd2b26eca49ab0f471498da620d46163316e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 10:49:13.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea9192497b1448b78da023e0fc207192\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4e97a2bfbbbb43c7a87ed5f146a648b3 : SUCCESS in 1h 03m 11s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/456b7b51177841de9412c7fb0d09e3d1 : TIMED_OUT in 2h 04m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9065439a4bd4d52b48ae053dc8d7f3d : SUCCESS in 59m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1188224af58542769414253864ee4b0b : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b032587507e4c2b948c1443ef98f0f4 : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06abfc4cac244b619d4ed7bb99092646 : SUCCESS in 14m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7af65ff1c8b343348c6c035d40649220 : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93892e4247734c1bb4ca71d58f1f30e0 : SUCCESS in 11m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/adad56737184480da5ca490f7769f89a : SUCCESS in 1h 06m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5d93a8679f394c2298bcc09a301eaef3 : SUCCESS in 1h 11m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e0828323183e40f59f5690d3b55bfc1f : SUCCESS in 55m 49s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b76cee49e2c24bde99ea7f3426f24587 : SUCCESS in 55m 57s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/200dcf637e964ff592c00ca984525a80 : SUCCESS in 1h 34m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/3f53db2dbe354adc8e346ed445f2ca3d : SUCCESS in 1h 48m 52s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/471df64a6b5a460ab181cdf310f165c9 : SUCCESS in 56m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/70e69a8517554175b71ec3c67c9d0657 : SUCCESS in 5m 12s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/659afc5aef0247238da00b1ac144bff7 : SUCCESS in 24m 55s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b53cd5b1ed2344aea5932a3445ee8ccf : SUCCESS in 23m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d299d70cf019445aaf64e4701865fdac : SUCCESS in 23m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f34e182d8cea434b99fdf7583568ff14 : SUCCESS in 1h 23m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f47b9b518a2e4d338e4aada6fd764d8a : SUCCESS in 56m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/588faf8955cc4714ad2534c06c365dc0 : SUCCESS in 35m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7645148763d44e19ac985df38603b0a8 : SUCCESS in 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"6089ccc91513bb30432978720f358cae876c36ec","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-12-14 11:15:57.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2d5d39985622f7aa19bf709fa352e0fea042c02e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 11:46:13.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9603b1e84cde415a868ee6d6e39e5ec7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acd6efa33b0e468b8f4182f49e6d6c15 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4702beeacb54b0b88e504e1ccb64602 : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"cc7586b63dfe745d7bd1425e8d877ac1d4021c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 13:05:28.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa3618055aa45c395da7efb50df84c3\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/094068437cbe4bce96d9ec460e2c558f : SUCCESS in 46m 09s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2563ee92cc4e4e4faabd68ec8e50a1c2 : SUCCESS in 1h 17m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eb6b116f5524cb6a8c50625b16b9863 : SUCCESS in 47m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ccc9f49e0684f75a9337d46ecc7dc05 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29e52a7df269443aa4dab2f20e4dcccb : SUCCESS in 12m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bdbc78f7c6f4d2c93320a2c721b1118 : SUCCESS in 12m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40e1a7c04d80487c8f45dca83d29be99 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9aa388aa23543fd808bcaa0c30e2cb4 : SUCCESS in 9m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e2a985b9bc204727a70d0e74a085d4ee : SUCCESS in 1h 04m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70d361a074fc4b53bc6c096a40c9d3aa : SUCCESS in 1h 04m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/26257a936cdd4039aa7aac3bca51b4a0 : SUCCESS in 37m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/600fa894c91e4e9cb41df5252f45c987 : SUCCESS in 48m 00s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0434af97aae944cf91db3affcbd64364 : SUCCESS in 1h 12m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/1837e009241b484e80226cf805307d6b : SUCCESS in 1h 48m 30s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a927e5170ccd4e548059e4ca01632739 : SUCCESS in 49m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6e36b147b9aa45228ef291aff3baae0f : SUCCESS in 4m 56s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1d8ad4f45ec40119ff68d0970f3f662 : SUCCESS in 22m 49s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/393cd4905a4f46d49a0c656dcc2fc92f : SUCCESS in 25m 19s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b534801e1f714e08ab3c011e3f30648d : SUCCESS in 22m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/356df824caa845588ae71f4036ec5a15 : SUCCESS in 1h 16m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6925a442e16940d2a83adfc3ae38f965 : SUCCESS in 45m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d8f9ebcea0fa464f91dfbc234738b4b0 : SUCCESS in 19m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2817c85f86134377a0c8b01f279983e5 : SUCCESS in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b7ac255a7953ea83bd598c4ac513e5d009766213","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-12-14 13:05:55.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/21/check/EMC_PowerFlex_NOVA/3ae4726/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"a65cd56c59556915f301974dec7aabf7fa5b2bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-02 10:29:12.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"4483550b99735b5a00de0862083aa4b93666621c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 11:23:04.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8be002370655410ba22ad3e6fb268b0e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb79878c4c5f457eab2d2d9dcc7ed660 : FAILURE in 36m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b7ad758e1864dcf9233c42fca4ee04b : SUCCESS in 51m 24s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"a70ab5f6775dc121f9fa0f1bdf2573c5e67be859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 12:41:04.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdc120f1dbb943998070a178c60a25c6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0bd94ea006b646b7ad44c062f2aa7189 : SUCCESS in 1h 15m 45s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e335364e80b947c99cfc814ea600b353 : TIMED_OUT in 2h 04m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c58da88716de4a0ba8c13c1330292f4a : SUCCESS in 1h 07m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c24ca0f2c24d42831988a0968fb9ae : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674f2f5dc7a94f8db83506ac5b475ef5 : SUCCESS in 12m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a11726b4ef54aab819372a638b2c815 : SUCCESS in 16m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64c98bc0f9a24ba89a5a10b52061eef1 : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962005bc92bf4cb4b797fb6f3899e7dd : SUCCESS in 13m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ab61dc1799544a3f8acd663905010ed0 : SUCCESS in 1h 13m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8c1b3a841c845668ebef0a224ac5f6c : SUCCESS in 1h 04m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cdac8272883c4ffebaea24f7a5c0e24a : SUCCESS in 59m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a009f5a071ce438a9637706b78f83d06 : SUCCESS in 47m 55s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6d26ace284904007b338eb52f777d626 : FAILURE in 1h 53m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/732ed6ff6c454344a79b23fed1221daf : SUCCESS in 1h 44m 51s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1d01a210e97f43048a553a9dda9fdfa4 : SUCCESS in 38m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a8d9d4a1dc1f4801a4b315bb390214c5 : SUCCESS in 4m 13s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4319dbcd8be74326953ee1ef82a242cb : SUCCESS in 20m 44s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/976cafbd50b24a3e97246326638494e0 : SUCCESS in 21m 27s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/04aaf870952042f097405aa60ae3d7d3 : SUCCESS in 23m 49s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a9772457be141db91d9264c02f79dd9 : SUCCESS in 1h 24m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4d5601312a54a9e98d3252f146b1ecd : SUCCESS in 1h 14m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c44151e0bc1548e3a6cebfe9e854b6d6 : SUCCESS in 1h 05m 07s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7ec1f721c2614d949a3529754b0d4461 : SUCCESS in 18m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/925a8907c50b4c018ec2d01b14bf20e1 : SUCCESS in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"182567f3e68c4efa3dca2a9ef13b6cdcda12e186","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-02 14:01:23.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"185dd46d7317ab6244bc852f2ef47d1346e67ac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 14:51:22.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c5cb8037c3142549b2518d6565eb322\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b4554ede3604ef2b630aff725e8c955 : FAILURE in 36m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84adc3aed0a6462e951d22f85204a326 : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d6015e7ec55a9f93c95356e5fba6d3ec9a96eb56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 15:54:52.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff2d9fc7730e4b4c9a40f4be47adbf49\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/417d89e4a0ec4e849a02cff4cc25189c : SUCCESS in 50m 47s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/74ef9738d832429f9d778b19bb27f3e9 : SUCCESS in 1h 31m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fab47b8b6474dfb852b55992704534b : SUCCESS in 57m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30cb309887cf45d3815a21f92ff9276f : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63cde23a85343009b05bfd8dc9c957f : SUCCESS in 11m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9860c2f9813c4b0cbf146c9675ab5af7 : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7adb120ede0d4a02b344fd9683f26b90 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffeb40dd75747729feaa426b5d95357 : SUCCESS in 9m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7e84759a7a6e4a2a9653e459874684b1 : SUCCESS in 59m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eecd29f2d9284a728720d935b95dafb1 : SUCCESS in 1h 11m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d439e84bbd044f00880273d51adfb083 : SUCCESS in 50m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/db1da1546a9d4383a4e874a59d31c811 : SUCCESS in 50m 08s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2464905bf37b4246b029c8e40453edfe : SUCCESS in 1h 39m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/b4b771de46f54a81b74f9d21dfecd31f : SUCCESS in 1h 47m 59s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6b1c5009d5de48578435c9aa310824f1 : SUCCESS in 48m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/daa25204b14945d1a2b796973bac0a95 : SUCCESS in 4m 40s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f5f369b42f34af9bb1b8dcc4eb21c2e : SUCCESS in 21m 17s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2ac0451cbe94caea593b4b26ccdd719 : SUCCESS in 20m 54s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ecbabbdb4294039959d402bb317dbd7 : SUCCESS in 19m 17s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d1eb77c817143e0ae55692de059b242 : SUCCESS in 1h 25m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2791fa381ed4cfba1f8ff5cdff43d15 : SUCCESS in 1h 04m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6373fc16a35546d2a1f7b7ae03671c1a : SUCCESS in 38m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/75bbeda3613b4a9d814200d77bae2cfb : SUCCESS in 29m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4a5e26ae1e1c4e7682f77daf0204d128 : SUCCESS in 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"91f59a229521ab8507060be80d78c94ba01b5759","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-13 10:29:02.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"30a1eea8a16748da6e52f23e759b9f717974d5e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 11:21:11.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69d28d15ca514b32b21aa9ee01c54b2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4295381b1d3c47fc95d17c5b394053bf : SUCCESS in 32m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5226b0f1f634ce4b3a854f606681109 : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"186b68fde625be987472d5d126aa11db3f722560","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-13 11:34:25.000000000","message":"Patch Set 23:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/8cb8f3c8c6144f0c8174174591f90c9d : POST_FAILURE in 57m 42s","accounts_in_message":[],"_revision_number":23},{"id":"b3dca25f53ba98f7c098ba1b6debc7d90438fd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 12:48:35.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f817e9cac9c46819b4ccdb5cac1903f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9ad096bee1a540419aa3cb56c9795e03 : SUCCESS in 58m 27s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1888a6328e1545cdbf9a49c2b0164535 : SUCCESS in 1h 20m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45474d6e305f488e9521e373279b90ef : SUCCESS in 1h 03m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9fe34dffc4441b89ccde396062d71ac : SUCCESS in 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0149f8a538f44aabd93474e8a358532 : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf801e8a5c9c4fc7a001ffc6a0f6b15f : SUCCESS in 15m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d14ec4197f814a06b4e5a287e35a9db7 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df80cf19985d49acb39b71fdaa7910ff : SUCCESS in 9m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7dca57da38ab49feb31430c9b8d261bc : SUCCESS in 1h 03m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e9734a598fc841e99f90ceaf295046fe : SUCCESS in 1h 35m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/32930cbd87f84750bccdcb7b5b3470a9 : SUCCESS in 54m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1eb2b104f40e4b60af1b7d69bc2bf7c0 : SUCCESS in 57m 45s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ecb31ac259d64d66b87828453444d66c : SUCCESS in 1h 40m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/5b4882b2ec2b42438bcc2682888d2dee : SUCCESS in 2h 01m 15s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/7e53db4f671949a09a7a8e0fdfb2dd55 : SUCCESS in 54m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3962d61b4a57419db3e826ddfe6cd801 : SUCCESS in 5m 14s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b235d0a8b4714322a2f90f59cd5cf731 : SUCCESS in 26m 59s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e75bfa94ac164ef0960306290c3b22f5 : FAILURE in 28m 33s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5e9cec8404d416c93ce33abc4398443 : SUCCESS in 24m 20s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/776ff3aceafb4abdbe01a51ddb83e7b5 : SUCCESS in 1h 17m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c04bb2bf141643c1a573411319329828 : SUCCESS in 1h 10m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0925f7f02a864b80a55f2753aa0cad7b : SUCCESS in 36m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3ea3096c1b5e436f85f3f5af670d51e8 : SUCCESS in 35m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1b297e13bf8841ee8d14345ef857ef9a : SUCCESS in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"5044aa29b86235d2d9660d463fb08e75ae1cde17","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-13 17:00:43.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"03edd4b83848d93f2abffdbe2af6c4612800f9e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 18:34:31.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627fce559c2a4bbdb75c7f65b4e1913d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2d55abfe5e04cada4899451256a268b : SUCCESS in 43m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bf05f7d8a454b3caf0e040061d87c32 : FAILURE in 43m 36s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"44baaabec133b6c8a92bec1a27f5dadb17ebcf36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 19:13:51.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08288af8482044d3b5fa4853f9520b51\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ab3910dac9b248e8a3c54e7f2f390220 : SUCCESS in 1h 06m 30s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f7b2bd63801345e3913f5907512e561a : SUCCESS in 1h 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0536c34a41746dcb4a06a68c2ca79b9 : SUCCESS in 54m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afd6d39945094c569dc0129d3b2879e9 : SUCCESS in 17m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d018b240b94dc3a2e11040f8524f74 : SUCCESS in 12m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63fac3ab560249948fbf113a2b4031e1 : SUCCESS in 12m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42da718f5f7f4b3e9d92f24b8b00214e : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cef77bed08c40b1b72b64dcc5aca1b2 : SUCCESS in 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/88f5803ae5024bf58f8286f45339b972 : SUCCESS in 1h 06m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed48354bdbce476aaef641d7bac339fc : SUCCESS in 1h 22m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f912264aeb404e3d846cb53f1103740c : SUCCESS in 51m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0eaf2cb1e848434d856c646b07522543 : SUCCESS in 53m 07s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d8486dab6dcc4847bea802f82e1c8599 : SUCCESS in 1h 49m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/161cf9b3cf8542eeb075eb0d8715f060 : SUCCESS in 2h 05m 21s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/927b4e2155974925ad7bb8acff365e52 : SUCCESS in 48m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6e681100ff1c46618e7eeca0e5d1011a : SUCCESS in 5m 24s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51c873abb21a4d0db11904a8f5ab5035 : SUCCESS in 21m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b86a667a9084ff8ad0dd0feb913dc2f : SUCCESS in 23m 46s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c91af7c518a47b9a10a919629e06e7a : SUCCESS in 22m 20s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6893813b3364475dabf764049c0f7d28 : SUCCESS in 1h 29m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9e581d94a5a4f73be469da248806f45 : FAILURE in 41m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63925de8704441da82e4521aa0d338ff : SUCCESS in 49m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6188c736efeb47a59c080ca32d9e8563 : SUCCESS in 33m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/79fb77c9c24a4d3fadde439b4299d37c : SUCCESS in 41m 25s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"bc668e78f6d9ef394dc3068d926e598ead43e75e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-13 22:36:10.000000000","message":"Patch Set 23:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/9f7101d8c68b42c1b46377f72f477bd3 : POST_FAILURE in 2h 13m 18s","accounts_in_message":[],"_revision_number":23},{"id":"c5e1a3562914e92751f079b77dbaad58e570941f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-16 10:14:43.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"b8e58270110bb748bbc16f8386faf13ceeeca69b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:10:10.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66d5b20be63445b495bd7acae3f9a0a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdab044aeabb4af8a04401c105fedcd1 : SUCCESS in 41m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4d72f81343349c8ba4c529f85b3408e : FAILURE in 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"8699363d35f4d3cf9da6ffa14d4eccb84b839c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 13:26:57.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a34771f17ef43368e1f6824df75adb5\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/630e2d22d63340fd8c8888051b6bec38 : SUCCESS in 55m 35s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0be375dbc022421f8bfc436bb1a287be : SUCCESS in 1h 24m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e896b2675aa48868e85c9ab46d96770 : SUCCESS in 47m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87850120f3594a2eb0e207d40f605250 : SUCCESS in 15m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e65a8eaa23e4f1583859ac11fcfd7a1 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/579d8cf060374cd0908a0e74ea2463f9 : SUCCESS in 17m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7e61e43a65e4bd2b8b701519526428f : SUCCESS in 15m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb3eb77f814455385c992bd8778ea8f : SUCCESS in 10m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6458447df76a456e97943bed0b0d5bc8 : SUCCESS in 54m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8587e6a4b6de4edb93de5c064928788e : FAILURE in 57m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4103bef4966a41639f46f90758c93e30 : SUCCESS in 48m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/69936c5ab54b441ca5594b2bed45f34c : SUCCESS in 48m 03s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/accd2fae742d4ac8814617df4d8040a7 : SUCCESS in 1h 46m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/d9b69d5e5f3c4f5293d41f27849a7b17 : SUCCESS in 2h 14m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c4c4399155904afeba72ec302bb4c05b : SUCCESS in 37m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/158511c6255a4271a5400c08b7611797 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0d40109bf304d9194ca28d8cd05af84 : SUCCESS in 22m 44s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05db71373ad140cdb5330a0a01a03cad : SUCCESS in 24m 07s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/022a3537eb714d35b776421dd0a1c022 : SUCCESS in 22m 48s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72ecd1d601ec4514902f42fef0b119ab : SUCCESS in 1h 28m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff0fd3a11aa34d4f9386fb5f801b4e9c : SUCCESS in 1h 16m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd573bc97dbd4b59a468819d4f97f215 : SUCCESS in 45m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cf7b51ec6e7b45468cd3fffb634163c2 : SUCCESS in 26m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6fe805c8d8a541a6a00fb32bed2348bb : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"1b238926059b2cb584b4c70a7a64be4dfa3f0b4a","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-16 13:51:23.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"43b3471d990e6deffb575e2ede74cab76f1f7379","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-16 15:45:35.000000000","message":"Patch Set 24:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/eb22d13058b0462187a080118db8e7ae : POST_FAILURE in 1h 46m 03s","accounts_in_message":[],"_revision_number":24},{"id":"1e44e6d5bcf8399fad8a77c56610a48cac603d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:00:03.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/831c117587604d5ea2fcb507357b282b\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/98f255b44a79482795b70132d035d5dd : SUCCESS in 58m 54s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3c44d562b55b487a934eac4b3978d0f9 : SUCCESS in 1h 37m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af981ea1b2204c9d8dd77d4f3d6577d6 : SUCCESS in 54m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ae36ca49523414faaadddc8be81699f : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd90bd9b1494a2483746dec0d7c63bb : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03fd900f66144deeae81dce9cfbe683f : SUCCESS in 10m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/800f64eb9ee14b4bab047f191b8ff461 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c63f019d2f48e0b641093475cb0af5 : SUCCESS in 8m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3ae625b035574131aa4890a04f4698b6 : SUCCESS in 54m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8e28b968104e454eb7ecd2344ac6b36c : SUCCESS in 1h 03m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8f082d6111264879860247f13a6c596c : SUCCESS in 43m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cbb9127222134be7b80ac90bcfa359a5 : SUCCESS in 36m 54s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ee1c58b5f793404fb6b93ece6c8cdfe3 : SUCCESS in 1h 39m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/c58a339e860745439422182bad7b4cfb : SUCCESS in 1h 41m 29s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d13397ace6f4447b984cae575ebe67ed : SUCCESS in 52m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d80436b783554038a34db15c6c89c632 : SUCCESS in 4m 49s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3e86cf331b24be9a0d35a4025d89d30 : SUCCESS in 21m 40s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7918a065658f4d5e8972a314320fd894 : SUCCESS in 22m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ea22392176c4ab3a2c5f9549258a76d : SUCCESS in 21m 50s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db22ebcfea3f4947b56527fa959b085e : SUCCESS in 1h 20m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0bcc3c10a2045649e167b551d02b8a3 : SUCCESS in 1h 10m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cebb1a012064c879b673b40ee2895d8 : SUCCESS in 47m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1c8c4d362afc4cb4a92e25118574cc83 : SUCCESS in 29m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5ccf2412500646318f348ce9876e8184 : SUCCESS in 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"82a64c09ef3c8f6270f9d025e9aef8a6936a6058","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-16 16:00:27.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/24/check/EMC_PowerFlex_NOVA/046a661/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"702c720dd4266cb2a6bfb8a55a8b6a55fd8d0f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 16:03:11.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be06484604c647ebad2be2f0d502f3f2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a30b9628589b455a9d6ba55511aa7a88 : FAILURE in 39m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5cccfcc442254a669d97236faca24a2b : FAILURE in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"f022827151d90a25ae61cb3474f774a441e2df79","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-16 16:03:35.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/24/check/EMC_PowerFlex_NOVA/f1c15ce/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"44975025b3945829a0dd359dcfd65df1fbc91129","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-18 05:58:22.000000000","message":"Patch Set 24: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"03814ea8162eb94935993638305ad029dde3768b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-23 10:00:26.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"51f1c28c1510e6b9b73e4cd9cd608141c3d56add","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 10:40:04.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3923a9c5da384d9c8d9132f7603838e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cfcda9786fe4d04be31ae8e2b4d29bd : SUCCESS in 36m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c88d79a356b48b69ea2f79161ce20f8 : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"c7c28bbf315e2352b6cbc8315f271ece6fb7f5cb","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-23 11:17:39.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"9711d6cf9e2bdd124c973eb2d9fc6c99ef66f34a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-23 11:31:51.000000000","message":"Patch Set 25:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e22df5f1affb4b7abf8ab7048b3a9982 : RETRY_LIMIT in 1m 02s","accounts_in_message":[],"_revision_number":25},{"id":"7cc9d955c08cecc230ab7a21d08a475d041f2f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 12:22:10.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a38982325ad94b67b57af5b6d3eb11ef\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4d89304a5f52406984b331fb16e4740d : SUCCESS in 52m 14s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/25f0e67328124f95a200b4775bdc7d78 : SUCCESS in 1h 37m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a24dfa125ac74ecdbfeb10ca6b211430 : SUCCESS in 50m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a51c46dcbd5a45dfb21b31b83188695e : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f44ac19c9246779d47adaa4e63eccd : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa32ece82404e2d9a1367d270c27f2e : SUCCESS in 14m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1d44cc18bdd497ca679f2f3ca45579c : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab28b53290e44a349e258c8024063f2a : SUCCESS in 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/82bf47ed87e84227982be734acea74d2 : SUCCESS in 50m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45dad3ecb83f4269ba8228ccda0e2c12 : FAILURE in 1h 04m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/230469b563d84e94a79d6ed7ddda7c4f : SUCCESS in 35m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d9646ffc15af443b9cb159c2465fe989 : SUCCESS in 45m 48s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8892af5ff1e74e469e6656086b32ec38 : FAILURE in 1h 14m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/2dfd092b4c1b4adcad033651d29d2d2a : SUCCESS in 2h 05m 38s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/19d008c9c8a84395980692bf65783841 : SUCCESS in 35m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/71d5b11912ce4ee7b6c5c43a590a81f4 : SUCCESS in 6m 03s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a4bea48177c4004a40ae68c840a844c : SUCCESS in 18m 38s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c135a35396984e1080f8831b475468be : SUCCESS in 20m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f33557fa5c7645208c9e9dd4157fd350 : SUCCESS in 22m 47s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f936be151f4bc7aa322d21481899f5 : SUCCESS in 1h 16m 04s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e6356c9d404e43e5a980167bffb27220 : SUCCESS in 1h 36m 49s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/36c1a0f1a59e4a21a57af9f92a202546 : SUCCESS in 1h 12m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d8c0e4eeb04651a9efefda81981d34 : SUCCESS in 48m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cc8db871eedb47849089a8e8c395d3ab : SUCCESS in 24m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/13eb89eb6da34bf0ae5ee4c475f8eae7 : SUCCESS in 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"bb48b4372c92fffa0ce2a8d199762b8629e2660a","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-23 14:09:14.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"93d66a0af3f7709fe1cba9a17be251e043570879","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 15:18:30.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3790f44c95408ea560bffa18926850\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d65b2d3ea7d429bbf639d7a4417aca5 : SUCCESS in 32m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b903e857fa04aa29e5c5c0918b3babe : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9573a47c4db440f5fa76e652cf417147954589a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:04:31.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87a8d0e8579f4b569b5c100b258b45f2\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f370716fcf804d61b4897125185d854b : SUCCESS in 1h 03m 00s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7aeba655de8f4647ac3ece0002364b0c : SUCCESS in 1h 31m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7dd0814597849eba7f9a3ddb6513f94 : SUCCESS in 55m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd7d9e51e3bd4e108794a0731e1e16f5 : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/702c1823ee2644678933816f70329e07 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9aa8cf41b8454ea89af1b5e70e04d547 : SUCCESS in 14m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3962b84009940d1a4a4a7c48d910987 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/113bd332a2954b158b3bfbabc763d019 : SUCCESS in 11m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/27fdf09a61b1401b82814d8c5acc6490 : SUCCESS in 59m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd52033824e94726ad6264f2b1a31c49 : SUCCESS in 1h 12m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/782249e24f2b4d66966bcf70a98c2503 : SUCCESS in 43m 32s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c8c6366b5e394eb4b827ae2cc17afa49 : SUCCESS in 49m 40s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ca26a6cce2664056b5019d8a741367d1 : SUCCESS in 1h 29m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/17ab8d131b2941e6a39a8a4ef37bb0f0 : SUCCESS in 1h 37m 23s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/eff2c3de17d84e3b92b75706e0be7757 : SUCCESS in 51m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fe334c1f493a4b8a94bcbccc8284a6af : SUCCESS in 4m 45s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7e56e2e70da54de6959c7d049150468d : SUCCESS in 21m 39s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c587d7edc194df88dc81b7230122b91 : SUCCESS in 22m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c23bff55f7e4c2f9affe65bd6718ecd : SUCCESS in 21m 28s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9efdf74c26740d5ab6dfd39cb07e899 : SUCCESS in 1h 23m 30s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/49861984ddf24313bd618e9c69fb3831 : SUCCESS in 1h 32m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0eb76238387432cbb0d0d71ec12f844 : FAILURE in 1h 09m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35f5137783334f34872de30c97b0ea1c : SUCCESS in 54m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/45f3b40578ca4277a38bc3749b79276e : SUCCESS in 29m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/db6da582fa6d40ed940ad1f31ccc4508 : SUCCESS in 43m 34s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"b40e0f76775f260959cdad4c2481b6c8aa5ab6ad","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-23 16:18:16.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"1fa41799b25ae21ecaf43b7120c859bd50b479e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 17:02:00.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98fe1b4988214aaeaaadcea46126c304\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b324ffcd77054f0692377ebe7e8df2cd : SUCCESS in 40m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08dea120ef9c4511b2ad58ecba2906f5 : SUCCESS in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ea855c025119e938f6e7a3630354d7b902efb8d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 18:08:39.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/376bb460eea645bc9ec9f0efd1a2bdf9\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/205d89b75ee945a29b5666fc94ab16fd : SUCCESS in 1h 04m 38s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4db1e51d48e2405584f8274a42646881 : SUCCESS in 1h 26m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44f5b8d5692044f290f89ed69d645e3c : SUCCESS in 57m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a12348c48f3e4025bd2a898b7aba0719 : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d44f97acecd4265abf9f1412d0de34f : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cabb3bff6cf941b790613c2a10eac4a6 : SUCCESS in 11m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46b826cc32a4482bbb56d530510f6f2f : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ab1e1f4be741499ed3d7a016822279 : SUCCESS in 10m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e6926408ff5d46dfaa81c1c35c84c950 : SUCCESS in 59m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aeb15bc92a774a919503a2b47a790e5d : SUCCESS in 1h 08m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c57438577df9441292695d1ea26613ff : SUCCESS in 47m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c069ed9865a84e7c8cc237c0d4eedc92 : SUCCESS in 49m 37s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ebdfc0b23a6b426090e9233f3e9062a5 : SUCCESS in 1h 29m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/5c7aff8154f74af8975892d98d5fe4e0 : SUCCESS in 1h 39m 01s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/54ab05f07680454c8cdb598c2e738792 : SUCCESS in 52m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/269f48f8b0594e3d8c316c4bfe9f0502 : SUCCESS in 4m 33s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77006b1cec5848b4aa6fe83f149fa15d : SUCCESS in 22m 25s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb5f5db465ae436cb6bfeb81c2c92f9a : SUCCESS in 21m 44s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c4978fbaa3c491b9a8edae415cbdbec : SUCCESS in 22m 58s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4426d3186898435b9b71937ce736234f : SUCCESS in 1h 31m 53s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c4f564a67ae649928a7d22ddf9764b1b : SUCCESS in 1h 28m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6af960a1a3df4246b309c412da651137 : SUCCESS in 1h 11m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce1bf8a2d38345d49a5ae034ea691e64 : SUCCESS in 49m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a21260a843754cf5b173ed47de729b6b : SUCCESS in 32m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7900811bc4ea4cb3b7f77843aff90c3b : SUCCESS in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"577a0421e09b9e607a0b8cb83bbfb86234f5266b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-23 18:16:36.000000000","message":"Patch Set 25:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":25},{"id":"8627e7231407664fb82db7ec3af371269bf55cd6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-23 18:17:06.000000000","message":"Patch Set 25:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/cdfa70629ec6477299d58e599b4a752a : POST_FAILURE in 2h 12m 31s","accounts_in_message":[],"_revision_number":25},{"id":"a88ede4bd18f24f094c0b6fb53bad920e88d401f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-24 10:47:36.000000000","message":"Patch Set 25: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"234bce7819afd2bd9db58856967457a6a7c989ea","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-25 16:59:43.000000000","message":"Patch Set 25: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":25},{"id":"61379957b36048f499cbd38872af8621c6c487ad","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-25 17:06:47.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"88bc8a9f026f6c449543340af322f365f27181b2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-25 17:20:26.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"d508ef1470732a3db27a6d01b1eecdc3d61f7709","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-25 17:21:14.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"1922df4d2d0a143476d8e700f59b0fd5adb9db0c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-25 17:24:32.000000000","message":"Patch Set 25: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"55e3d7c9436a3e352f50d6bcdf852d0fa0653299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-26 14:26:12.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"f28a75246638cbe682cb33b87b4c4f7da2d66970","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-26 14:52:16.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"d3867a4f9e30a726c9502e9b763360f645233a59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:08:05.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4586b3a6680847a99edd3deb77dbf1cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3ecc1e257854047a3d7e216f503565c : FAILURE in 34m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8cfb5cb847004319aae1f3be91de6de4 : FAILURE in 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"d2e2782dcae0b6e708e24b2ff3a2462fea14d1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 16:19:39.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e4130ac8aa2432187b642baf658f381\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f7892bb44984ef2b30b60f486de814e : SUCCESS in 40m 21s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fe76d1de26d84735a68714f1dd34fd97 : SUCCESS in 1h 30m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a197350d05854286ba84ad1624968cac : SUCCESS in 41m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/307755aabf7146fb97079ac7fbd93ec2 : FAILURE in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d401b5c9fb4dd5b3e5d62b373d42e1 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d438d3649f3241539bb4c7c81195fac4 : FAILURE in 10m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da1de40c2f0a4f5cadb37b1c496c328d : FAILURE in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e19937630404d82838f2e0966b9f1e8 : SUCCESS in 9m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cf3ca7ee0d8742e498e7e3e05095805b : SUCCESS in 47m 35s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd1817ab76c74a63ab465e656ad6cf23 : SUCCESS in 1h 09m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/77bbbcff4ad2469da71b2ef1355be021 : SUCCESS in 38m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/96b53725d85b4b51bf30179dad93b093 : SUCCESS in 39m 40s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0fab237e7f2e4c19b7f042c984c02c5c : SUCCESS in 1h 26m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/a8d0082850a64795b1e7833492b8566c : SUCCESS in 1h 46m 54s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/54412fe8854b40f1bc25f629cac72b23 : SUCCESS in 50m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b6e6e7f30fad4dde8b32c0b4ffbd7691 : SUCCESS in 4m 42s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6c89de84d58e4d0ea90d921726a9dd61 : SUCCESS in 21m 45s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb84f1205a3346e8930db37a00884543 : SUCCESS in 22m 07s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b217306eeb6d40389b1d4215d46525d4 : SUCCESS in 21m 22s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f52b69b0307644a2b22b1e581f48d10f : SUCCESS in 1h 26m 16s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dd6dcbbbe7e241db93742db6168ccccd : SUCCESS in 1h 28m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d57e84a8e444741abff2a6c58c3d8ba : SUCCESS in 1h 02m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c47564e3099e45798603a572b4998d1b : SUCCESS in 49m 54s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8deaa25f3ecb41ee85ebb8e785b648fc : SUCCESS in 29m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/788ac22addc4480d8e3dd20790a727ea : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"3f2688f8bbc0eac90e7ead25c6e99addf72e00d6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-26 16:56:34.000000000","message":"Patch Set 26: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":26},{"id":"022ffd3def482a6e8fd94a2f64fc10ff16a721b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-26 17:11:43.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"67a607a523dcc89f35e7f30d33f2e7871a88dee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:48:05.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecb036aea031437fab095b00467d3501\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/497077dcc0b54f16acbb4c0ce5b999c2 : SUCCESS in 34m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91b5139a2e6e4446ad861b75b54b5276 : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"63f4b3e93a04814e6f095a0de1a2ac3e6611cabe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 18:44:11.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9bc0e8ac774e46bbeb2c6b593cfed6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4d4c192e9c3948a583321301bd51fb78 : SUCCESS in 1h 14m 03s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/919152016db94f009e74068bf73c47ea : SUCCESS in 1h 19m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c9ca95dbaac41b7ae896af3cd663ad9 : SUCCESS in 49m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97e5e7f490474c18aaab186eedfe27cd : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27257cecacea41c3b97cf2c29a0d37a4 : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef35beb656a14a4fb69b0b20a89ee0f2 : SUCCESS in 10m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efe6192fee934630b62e99572d5d2f69 : SUCCESS in 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18cb0704cc9541f8bbeb1055f48a3b65 : SUCCESS in 10m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/af78fde999e3476e812111c84b828d99 : SUCCESS in 45m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c1e15be50524b3a96d7f34337b809d8 : SUCCESS in 57m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/de99220dfebf4a5b841c32be42ed12bf : SUCCESS in 37m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/09644451940047119b104175997beeb9 : SUCCESS in 42m 10s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5057230f11bc49d49c22e659b9bd854b : SUCCESS in 1h 26m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/c7540f8c69fd4101bc7fe8b030c70146 : SUCCESS in 1h 27m 45s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/af8e8b38cd0245f1a85041ff3c567ec5 : SUCCESS in 42m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c94035de13434a1caccf640f18fef82a : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/abe1723426214be8aa5340969e3efece : SUCCESS in 19m 24s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7464782d12a047d8bb61d07aa36e4c43 : SUCCESS in 24m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/919dafa14b5041bda8154f989ae72520 : SUCCESS in 24m 05s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/028e552fcefe49bfa7c89596b0b488f2 : SUCCESS in 1h 29m 54s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f28ed7c7f564b73b48e9913e2390811 : SUCCESS in 1h 28m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d67976351d57461cbfd81c81c3bec66c : SUCCESS in 1h 06m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1cc8368c3cc444ab6ece355073b9f98 : SUCCESS in 48m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/989829ee675f40cc99732a7b8fe2545b : SUCCESS in 23m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/78f98e42fa1643aea813289af931288f : SUCCESS in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"fad18748fe87dde28318f72981db92f45681b22b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-26 18:56:26.000000000","message":"Patch Set 27:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":27},{"id":"81b01c693a592c7bf5e73afcc56d2741f53e19ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-27 07:10:54.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"270b611096bcff211df9ad19b1f50ccb5caace44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 07:26:50.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06a5a3c7490345908100d806b08ba3cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54b2b4381b084a46b5914ec1541a6ea0 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3bdb1443b3f642dab5a856bde3f68717 : SUCCESS in 14m 21s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"369796d9c1c35cd12defc22f24f343589edd64a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 08:54:16.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16bca901e5b4894bc25dc905429c1ea\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4bdd21b75bce4a5ab12db3b48b90f634 : SUCCESS in 58m 40s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/375eeb61fad34682a93b21e4ab3b736b : SUCCESS in 1h 18m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c30546d48db4d25b5c08971a14f2a16 : SUCCESS in 54m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4aefe174459419da8f9eb04aff16518 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79714ad4e2244f6488fe51723197f857 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d81d0d10dbfe4828a1e471fa1baaac52 : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1fd54b2218248fe8e9da90984c82720 : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d310113e8f44d50abca510ee298d97f : SUCCESS in 10m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a23919a03a5744a0803c9854ce3767fb : SUCCESS in 59m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c615779c1fb42079c3d0085c365f4af : SUCCESS in 1h 00m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8581c9f7068b43b8bfe4b83498e7a72a : SUCCESS in 48m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bc609dc3dc014cda8d13ef06defa5b78 : SUCCESS in 49m 04s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0d3b267659b747f2b799da2b251ea69c : SUCCESS in 1h 21m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/94cac03a5e0d4f9f9dec43e9df984f20 : SUCCESS in 1h 36m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8318a8a3deee4c8c98fcbd69b476f246 : SUCCESS in 38m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/320fa1d776294daea653b58792c4bc20 : SUCCESS in 4m 23s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c31ade2d49974e3b8ce9df00c0811e3f : SUCCESS in 21m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6713055534d64574a936df5b7475b183 : SUCCESS in 23m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c3b91e5bd1e348feaa3e78e67f341fb5 : SUCCESS in 21m 30s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0cf0fc6cf44dc198dce0f9e358a4cc : SUCCESS in 1h 04m 43s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/55fe602f758a461b9368c319d625f967 : SUCCESS in 1h 21m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/041f36a25ba14fe4bb4c1cace8172f96 : SUCCESS in 1h 09m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f65b43e829824b618e250d72c867a7b8 : SUCCESS in 41m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1d40d7947ca94060b0a3ccfc3f978f3e : SUCCESS in 30m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/28c86a383c964a09807d5c6a24db7337 : SUCCESS in 43m 19s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"22575a1ba32e283574735af7331da2165dd1007c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-27 09:02:18.000000000","message":"Patch Set 28:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":28},{"id":"cfef9449856e3795f9eccf6abf34566ce558fad8","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-30 15:21:37.000000000","message":"Patch Set 28: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"f8363c578805cf5c6649402914b8e71d34f6b5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-31 07:58:59.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"0917b34e04fb3d6fd08ba658f6b3eb6dc92f3ba2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 08:12:49.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8142c64fd124611b53be2f922392240\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b343843a9764c57b3c9691c560574d5 : FAILURE in 10m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5ad7feb115604f44a3845a1a5d8d1488 : FAILURE in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"6c4c65aac81f3d00151e6bd08f3b1f0e5ce4ec2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 10:35:53.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d5ba9967268437988b710d5cf7f3304\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f185396720440188366ddea3db64f33 : FAILURE in 19m 38s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b2df1aa4901f49daa73f1ddba2eb537f : FAILURE in 1h 25m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df0042d2999f4a449a18d3db255fbf6d : SUCCESS in 47m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c960ecb4d8874d5f91fddf681a447f3f : FAILURE in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f42725c84e14b3ea284ec0fdf65e665 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42cf3aa818174697a2e7fcebc3924dc8 : FAILURE in 12m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce8b1c08c1f146f3afc7fc2d0ba8b184 : FAILURE in 14m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047bd399f22c4d9fa541b948f481fb25 : SUCCESS in 8m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4d0cd7bd00c24a55925dd27de4c803da : SUCCESS in 54m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f46684e1eba94b389785a6fa38539fbc : FAILURE in 1h 45m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/619c6ab117744f42a5898ba7f49afd2f : POST_FAILURE in 55m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3a6704cecbc64421a0c87c5742537293 : POST_FAILURE in 1h 04m 50s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0d78344cb1534c79b14896695079a265 : FAILURE in 1h 52m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/7fe2c36a6b064885825d5a756c923726 : FAILURE in 2h 27m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9c8da48a7a6f4e938f4fbe44458d25b8 : POST_FAILURE in 44m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/df19bb58555d4faaa84465dd9a58bf9d : SUCCESS in 6m 06s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7e2768342c747c5977762b3b4fd57e0 : SUCCESS in 21m 47s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0b2b5c779bd64263a1522204b38a77a5 : SUCCESS in 22m 04s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c57432610ce456d8c9f6d05e6652525 : SUCCESS in 22m 46s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f785f17dadf3449faaa3cbb2729ec59b : FAILURE in 1h 14m 01s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6e59ecdf9dbf47e0ac80f6f1b0b4cf00 : FAILURE in 1h 28m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1756a7cc62e84367acf4892a1da7b6f6 : SUCCESS in 1h 08m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7147b2a1d8f94d17bd2a5a4724038917 : SUCCESS in 54m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bfbaf1f6462a4eaf817963fbd1fa01d6 : SUCCESS in 28m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7e943fcc425f4a979f741966cb70a010 : FAILURE in 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"8d857cde870b512c77040490cce3dba546522bd7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-31 11:06:46.000000000","message":"Patch Set 29:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/926f43b9dfba4162852ef5431e388ed5 : POST_FAILURE in 2h 28m 32s","accounts_in_message":[],"_revision_number":29},{"id":"a2bf0fd3d970080f0fbafc338617c59675a15b80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-31 11:51:00.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"9f8b0e9526f28611d5bb4bd81b350297b340c9e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:07:12.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd8815533a94eb9824b229b989fc4f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/567292d524f843bda78a11180d65361b : SUCCESS in 14m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c1b569c9a70c4a25a3462d809b61fed7 : SUCCESS in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"0239a01902a2b5d8a33f37476b741dc1d0dd16a9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-31 13:15:31.000000000","message":"Patch Set 30: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"4706fa617d586b935ba70745d4f776a4866ff04b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 13:56:39.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b38ae14b156449cb80157c804a583ce\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e9d55cf9bf3f42659466b0bda4f63301 : SUCCESS in 1h 05m 29s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a6e54d5b0c6f4fb68e86125e76ab90c5 : SUCCESS in 1h 23m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31071d5e0d2843b0b13429c66ca1a12e : SUCCESS in 1h 18m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b896d29f54246baa6eed5269e25bc8f : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490905a09d4748f598624d271864d9b0 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82da46f8cfab452ca09a6ae1848b946f : SUCCESS in 12m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8525f44e16fb4678b522ca7632d3b454 : SUCCESS in 14m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbefdf87b8f8442c848aa9a95de25b50 : SUCCESS in 11m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e36cb14e9e5b44f4965051f042bb563c : SUCCESS in 59m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e5c3e91bd9b844feb2ea92cf56bc176e : SUCCESS in 1h 14m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e5a27743e304da58934ccbda8b9285b : SUCCESS in 53m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2b45c72a7cc84e2ebcd93ca2e6e9439f : SUCCESS in 53m 11s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/af44c15163d544299610c28e596da7d9 : SUCCESS in 1h 28m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/14f6d066e8f3494798f5915c941131f0 : SUCCESS in 1h 59m 51s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d934abd240da4db2878bcbbaf8e8143f : SUCCESS in 1h 05m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/349f2761535a41fabcd7610b3523cca3 : SUCCESS in 4m 36s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7892d8f1fcd549b0bbcf0e57aa61c286 : SUCCESS in 22m 44s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6c65ddb39c74418fa2cbc6a79c019638 : SUCCESS in 24m 48s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c68bd26ee3b24ddf998a5565cec20433 : SUCCESS in 21m 43s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd9b84ff563741ed846dfa3e3e0c039d : SUCCESS in 1h 47m 39s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/598927972358426089adec711bcdc9b3 : SUCCESS in 1h 58m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf06f3ccceb74341acf3d8b07f1f754a : SUCCESS in 1h 27m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a30f5ce278b3459c932a5ef7fba0d9e0 : SUCCESS in 43m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/037bc1fbf72e4054b41c6ca52a0479fd : SUCCESS in 27m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/17222350b0af4164944610557149087a : SUCCESS in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"0bda5c2f40d62711b981fe98073fb58b3e373993","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-31 13:57:04.000000000","message":"Patch Set 30:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/83/858383/30/check/EMC_PowerFlex_NOVA/c751947/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":30},{"id":"94c06faf106206b2ad09c4093a6ee851a15ad99a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-31 14:04:50.000000000","message":"Patch Set 30:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":30},{"id":"28739975b433f79c595554524f40db6ade510bf2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-31 16:08:41.000000000","message":"Patch Set 30:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c40fa52f98e34b6fbf5f32ea5319e36d : POST_FAILURE in 1h 58m 38s","accounts_in_message":[],"_revision_number":30},{"id":"a86f7a4c75554176f8c31cca1dc1a9fc35a490fd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-02-01 08:32:36.000000000","message":"Patch Set 30: Review-Priority+2","accounts_in_message":[],"_revision_number":30},{"id":"48b0d74c38c0ce194d913bd7633988af66e7aa73","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-01 09:54:25.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"b216e61db949d53d2768b7deb0d2a2ca276adc0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 09:55:19.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"b01c5cb83b84fd6855372c311d3f1396b5dfeed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:49:37.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd640c729c0c46c38c580cfa7fc72b6c\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f73776e3f0994e76994dddf27fa82163 : SUCCESS in 1h 39m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca3ede72fd7841c9a62dc46c6b1048a5 : SUCCESS in 59m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87af7a1a479d4066a6e74659ae8f30f7 : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/532b40838e7b4f62939383721d2741b3 : SUCCESS in 17m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5193d937aa684045854bfd6b0a3cb02d : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f66b372aa4e34400bf022ba62a5676af : SUCCESS in 13m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6f93f637057844a89c8928c870e15ba1 : SUCCESS in 46m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/fb195fd6b3454750be9d22277a7b0c5c : SUCCESS in 44m 44s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f335f8d525a483481f0bdccc9df6231 : SUCCESS in 20m 51s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7bf915fd4114c25b337649adc4c5843 : SUCCESS in 22m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3850c7e351ff4cf7acd48094efd2c247 : SUCCESS in 1h 26m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/c56de9e60ae94a3bbccc7a6b7b8ee0da : SUCCESS in 1h 31m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bc373f8f53da4bfa974c18600e71229f : SUCCESS in 4m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/76fd463fea5c484ea6b3d50c5b84bde7 : SUCCESS in 1h 06m 34s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da5d59c5da8e413d8f191854de21f448 : SUCCESS in 1h 22m 55s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d555a37d099d4279b4e118e5d8bea09c : SUCCESS in 1h 27m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cfb47a5fc2347a1be0611170ad3bb35 : SUCCESS in 1h 11m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09135911ec140f4a43d51122f40c32c : SUCCESS in 49m 46s","accounts_in_message":[],"_revision_number":30},{"id":"5ba7a80661bf4fa6b22a233aa630170fb8c98b87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:49:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":30},{"id":"a74ee94a17d067dd685ab0662daac29a5720c0d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:51:04.000000000","message":"Patch Set 30:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8440fc2ec5f04a549473b2601f266c8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5c02386cec45dc845b5f056f403642 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":30},{"id":"5a110063c3f78428cf9bc8b82f9b5ab2b611a990","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-06 21:28:13.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"8c2e76598995f0d417653c60a63ea342baf4e880","revisions":{"82975610e16f1ff19dfa11904e5702bd9fc8fe7f":{"kind":"REWORK","_number":1,"created":"2022-09-19 13:48:24.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/1"}}},"commit":{"parents":[{"commit":"4fb4f6832c156907b786571f214984894703bf16","subject":"Remove mentions of removed scheduler filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4fb4f6832c156907b786571f214984894703bf16"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 13:48:07.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82975610e16f1ff19dfa11904e5702bd9fc8fe7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82975610e16f1ff19dfa11904e5702bd9fc8fe7f"}]},"branch":"refs/heads/master"},"ded0a7a107f80076077dc3182cc705705af1b8f1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-20 09:58:31.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/2"}}},"commit":{"parents":[{"commit":"9a0a9dd53f2c0aecf16e84a72ed00221d3d3a14c","subject":"Merge \"Move uwsgi-gmr reno to the proper place\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9a0a9dd53f2c0aecf16e84a72ed00221d3d3a14c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-20 09:58:20.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ded0a7a107f80076077dc3182cc705705af1b8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ded0a7a107f80076077dc3182cc705705af1b8f1"}]},"branch":"refs/heads/master"},"1fd21f053724b15c5aa2acb63a0bd66930c3f8f8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-10-03 09:32:11.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/3"}}},"commit":{"parents":[{"commit":"aad31e6ba489f720f5bdc765c132fd0f059a0329","subject":"Merge \"Update nova-manage doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aad31e6ba489f720f5bdc765c132fd0f059a0329"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-03 09:32:01.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fd21f053724b15c5aa2acb63a0bd66930c3f8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fd21f053724b15c5aa2acb63a0bd66930c3f8f8"}]},"branch":"refs/heads/master"},"a4cb0791dc36e9060e4212cd49051ea88aeff0de":{"kind":"REWORK","_number":4,"created":"2022-10-04 10:51:05.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/4"}}},"commit":{"parents":[{"commit":"aad31e6ba489f720f5bdc765c132fd0f059a0329","subject":"Merge \"Update nova-manage doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aad31e6ba489f720f5bdc765c132fd0f059a0329"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-04 10:48:50.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4cb0791dc36e9060e4212cd49051ea88aeff0de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4cb0791dc36e9060e4212cd49051ea88aeff0de"}]},"branch":"refs/heads/master"},"885715a09905b09be39224f1ab057a25508acba4":{"kind":"REWORK","_number":5,"created":"2022-10-04 14:36:16.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/5"}}},"commit":{"parents":[{"commit":"aad31e6ba489f720f5bdc765c132fd0f059a0329","subject":"Merge \"Update nova-manage doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aad31e6ba489f720f5bdc765c132fd0f059a0329"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-04 14:36:06.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/885715a09905b09be39224f1ab057a25508acba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/885715a09905b09be39224f1ab057a25508acba4"}]},"branch":"refs/heads/master"},"e3d6043e7a6c0e5240b3c0b3817b103fccc2af67":{"kind":"REWORK","_number":6,"created":"2022-10-05 08:58:06.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/6"}}},"commit":{"parents":[{"commit":"aad31e6ba489f720f5bdc765c132fd0f059a0329","subject":"Merge \"Update nova-manage doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aad31e6ba489f720f5bdc765c132fd0f059a0329"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-05 08:57:55.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3d6043e7a6c0e5240b3c0b3817b103fccc2af67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3d6043e7a6c0e5240b3c0b3817b103fccc2af67"}]},"branch":"refs/heads/master"},"4d8890c45e88cf00bb0da3f0e4a66665e1471b5a":{"kind":"REWORK","_number":7,"created":"2022-10-07 07:53:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/7"}}},"commit":{"parents":[{"commit":"03d2715ed492350fa11908aea0fdd0265993e284","subject":"Merge \"update default overcommit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03d2715ed492350fa11908aea0fdd0265993e284"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-07 07:53:32.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4d8890c45e88cf00bb0da3f0e4a66665e1471b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4d8890c45e88cf00bb0da3f0e4a66665e1471b5a"}]},"branch":"refs/heads/master"},"81a14cebcbd78270191a54e5d11a1f830d48dfce":{"kind":"REWORK","_number":8,"created":"2022-10-10 07:22:51.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/8"}}},"commit":{"parents":[{"commit":"03d2715ed492350fa11908aea0fdd0265993e284","subject":"Merge \"update default overcommit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03d2715ed492350fa11908aea0fdd0265993e284"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-10 07:22:36.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81a14cebcbd78270191a54e5d11a1f830d48dfce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81a14cebcbd78270191a54e5d11a1f830d48dfce"}]},"branch":"refs/heads/master"},"3f1c05568cce155c3bf19ae7fdd9be48c727a168":{"kind":"REWORK","_number":9,"created":"2022-10-11 12:58:15.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/9"}}},"commit":{"parents":[{"commit":"03d2715ed492350fa11908aea0fdd0265993e284","subject":"Merge \"update default overcommit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03d2715ed492350fa11908aea0fdd0265993e284"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-11 08:52:53.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3f1c05568cce155c3bf19ae7fdd9be48c727a168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3f1c05568cce155c3bf19ae7fdd9be48c727a168"}]},"branch":"refs/heads/master"},"b4b4b5ead57704fb46d5cdd578efe08f814c60ca":{"kind":"REWORK","_number":10,"created":"2022-10-12 07:57:21.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/10"}}},"commit":{"parents":[{"commit":"03d2715ed492350fa11908aea0fdd0265993e284","subject":"Merge \"update default overcommit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03d2715ed492350fa11908aea0fdd0265993e284"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-12 07:57:07.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4b4b5ead57704fb46d5cdd578efe08f814c60ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4b4b5ead57704fb46d5cdd578efe08f814c60ca"}]},"branch":"refs/heads/master"},"bacb4ae9df0775e032b65552221ad01a245d2692":{"kind":"REWORK","_number":11,"created":"2022-10-12 17:55:39.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/11"}}},"commit":{"parents":[{"commit":"03d2715ed492350fa11908aea0fdd0265993e284","subject":"Merge \"update default overcommit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03d2715ed492350fa11908aea0fdd0265993e284"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-12 17:55:20.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bacb4ae9df0775e032b65552221ad01a245d2692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bacb4ae9df0775e032b65552221ad01a245d2692"}]},"branch":"refs/heads/master"},"7827685c2a64ddf698d6f6f8ff74bfc0e6cac611":{"kind":"REWORK","_number":12,"created":"2022-10-18 09:44:32.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/12"}}},"commit":{"parents":[{"commit":"8aee01f1286d1f9ec5190328d24e9eb6b4edb626","subject":"Merge \"[compute] always set instance.host in post_livemigration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8aee01f1286d1f9ec5190328d24e9eb6b4edb626"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-18 09:44:23.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7827685c2a64ddf698d6f6f8ff74bfc0e6cac611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7827685c2a64ddf698d6f6f8ff74bfc0e6cac611"}]},"branch":"refs/heads/master"},"30aca326333e9e1d5ed3da06e8ae2ac1e618865b":{"kind":"REWORK","_number":13,"created":"2022-10-18 12:27:54.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/13"}}},"commit":{"parents":[{"commit":"8aee01f1286d1f9ec5190328d24e9eb6b4edb626","subject":"Merge \"[compute] always set instance.host in post_livemigration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8aee01f1286d1f9ec5190328d24e9eb6b4edb626"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-18 12:27:43.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30aca326333e9e1d5ed3da06e8ae2ac1e618865b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30aca326333e9e1d5ed3da06e8ae2ac1e618865b"}]},"branch":"refs/heads/master"},"18f004380fe230528b866675303e62a311326014":{"kind":"REWORK","_number":14,"created":"2022-10-18 15:12:19.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/14"}}},"commit":{"parents":[{"commit":"8aee01f1286d1f9ec5190328d24e9eb6b4edb626","subject":"Merge \"[compute] always set instance.host in post_livemigration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8aee01f1286d1f9ec5190328d24e9eb6b4edb626"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-18 15:07:30.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18f004380fe230528b866675303e62a311326014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18f004380fe230528b866675303e62a311326014"}]},"branch":"refs/heads/master"},"6c02f2eea1a97b4ae728d1a0cf4ffc8cf37013b6":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-10-24 09:34:56.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/15"}}},"commit":{"parents":[{"commit":"b1958b7cfa6b8aca5b76b3f133627bb733d29f00","subject":"Correct doc references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1958b7cfa6b8aca5b76b3f133627bb733d29f00"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-24 09:34:47.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c02f2eea1a97b4ae728d1a0cf4ffc8cf37013b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c02f2eea1a97b4ae728d1a0cf4ffc8cf37013b6"}]},"branch":"refs/heads/master"},"492b071781aebf2f5f56fd7c11708985fd524be6":{"kind":"REWORK","_number":16,"created":"2022-10-27 08:12:47.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/16"}}},"commit":{"parents":[{"commit":"b1958b7cfa6b8aca5b76b3f133627bb733d29f00","subject":"Correct doc references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1958b7cfa6b8aca5b76b3f133627bb733d29f00"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-27 07:37:38.000000000","tz":120},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/492b071781aebf2f5f56fd7c11708985fd524be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/492b071781aebf2f5f56fd7c11708985fd524be6"}]},"branch":"refs/heads/master"},"71f45794e30ca8255bd80d12441391ff20299e2f":{"kind":"REWORK","_number":17,"created":"2022-10-31 14:27:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/17"}}},"commit":{"parents":[{"commit":"b1958b7cfa6b8aca5b76b3f133627bb733d29f00","subject":"Correct doc references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1958b7cfa6b8aca5b76b3f133627bb733d29f00"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-10-31 13:30:40.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/71f45794e30ca8255bd80d12441391ff20299e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/71f45794e30ca8255bd80d12441391ff20299e2f"}]},"branch":"refs/heads/master"},"8b685a8efb80d77ddf25168909c49344975c8083":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-11-21 13:19:41.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/18"}}},"commit":{"parents":[{"commit":"596772d5224e8329f558435205f942c9c776f08b","subject":"Merge \"Replace \"db archive\" with \"db archive_deleted_raws\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/596772d5224e8329f558435205f942c9c776f08b"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-11-21 08:57:12.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b685a8efb80d77ddf25168909c49344975c8083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b685a8efb80d77ddf25168909c49344975c8083"}]},"branch":"refs/heads/master"},"8dc904b704443ee4ca4e4cc0299e22f70c213f7a":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-11-24 07:02:51.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/19"}}},"commit":{"parents":[{"commit":"17d8b92ed1db708eca2a2887eb19239735e4f6ee","subject":"Merge \"Reproducer for bug 1951656\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17d8b92ed1db708eca2a2887eb19239735e4f6ee"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-11-24 06:50:20.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dc904b704443ee4ca4e4cc0299e22f70c213f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dc904b704443ee4ca4e4cc0299e22f70c213f7a"}]},"branch":"refs/heads/master"},"e106d18ec719286a84439e03bd87ecf030510501":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-11-24 19:20:36.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/20"}}},"commit":{"parents":[{"commit":"2d774541c549bd43bff6cafba36d2351d4649a2f","subject":"Merge \"Bump minimum version of os-vif to 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d774541c549bd43bff6cafba36d2351d4649a2f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-11-24 19:20:23.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e106d18ec719286a84439e03bd87ecf030510501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e106d18ec719286a84439e03bd87ecf030510501"}]},"branch":"refs/heads/master"},"02c54d99e50a49b741f218a80d909531f23a6097":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-12-14 08:35:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/21"}}},"commit":{"parents":[{"commit":"8b4104f9f78d0615720c0ba1e3e8cfced42efcc5","subject":"Merge \"Split PCI pools per PF\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b4104f9f78d0615720c0ba1e3e8cfced42efcc5"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-12-14 08:35:32.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02c54d99e50a49b741f218a80d909531f23a6097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02c54d99e50a49b741f218a80d909531f23a6097"}]},"branch":"refs/heads/master"},"d4fe2fc3748a6395ec332f959047c80fdc82259f":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-01-02 10:29:12.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/22"}}},"commit":{"parents":[{"commit":"00a62d218d80df9c5edc7f26e025138cc9d710aa","subject":"Merge \"Func test for PCI in placement scheduling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00a62d218d80df9c5edc7f26e025138cc9d710aa"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-02 10:28:59.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4fe2fc3748a6395ec332f959047c80fdc82259f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4fe2fc3748a6395ec332f959047c80fdc82259f"}]},"branch":"refs/heads/master"},"3c3a8f84e21b61a1feec244e11c18b004096d922":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-01-13 10:29:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/23"}}},"commit":{"parents":[{"commit":"8e3ffb851b3faa8144894a3299f84acd8eed38dc","subject":"Merge \"Allow enabling PCI scheduling in Placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e3ffb851b3faa8144894a3299f84acd8eed38dc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-13 10:28:53.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c3a8f84e21b61a1feec244e11c18b004096d922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c3a8f84e21b61a1feec244e11c18b004096d922"}]},"branch":"refs/heads/master"},"ed88d1e3f56b811fcc627c8297f53e1abb524fbf":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-01-16 10:14:43.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/24"}}},"commit":{"parents":[{"commit":"79a513d4ea3f94159ce6803234ca5d33b974ab1f","subject":"Merge \"libvirt: Report ephemeral encryption traits based on imagebackend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/79a513d4ea3f94159ce6803234ca5d33b974ab1f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-16 10:14:31.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed88d1e3f56b811fcc627c8297f53e1abb524fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed88d1e3f56b811fcc627c8297f53e1abb524fbf"}]},"branch":"refs/heads/master"},"a9fa5068c43d23db5a1e44622dbdaca565dc8690":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-01-23 10:00:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/25"}}},"commit":{"parents":[{"commit":"d8b4b7bebdc0f55353cd99f372044b9e30315a6d","subject":"Merge \"Microversion 2.94: FQDN in hostname\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8b4b7bebdc0f55353cd99f372044b9e30315a6d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-23 09:52:23.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9fa5068c43d23db5a1e44622dbdaca565dc8690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9fa5068c43d23db5a1e44622dbdaca565dc8690"}]},"branch":"refs/heads/master"},"a2036aec4ce15a5fd82a016cdf7d68c37a598e99":{"kind":"REWORK","_number":26,"created":"2023-01-26 14:26:12.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/26"}}},"commit":{"parents":[{"commit":"88ab68c72df96ceac3f9bae180e915f054f1e6db","subject":"Merge \"Handle InstanceInvalidState exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88ab68c72df96ceac3f9bae180e915f054f1e6db"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-26 14:25:55.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2036aec4ce15a5fd82a016cdf7d68c37a598e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2036aec4ce15a5fd82a016cdf7d68c37a598e99"}]},"branch":"refs/heads/master"},"6741b648f251144b7c64d60415916f1576331df2":{"kind":"REWORK","_number":27,"created":"2023-01-26 17:11:43.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/27"}}},"commit":{"parents":[{"commit":"88ab68c72df96ceac3f9bae180e915f054f1e6db","subject":"Merge \"Handle InstanceInvalidState exception\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88ab68c72df96ceac3f9bae180e915f054f1e6db"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-26 17:11:31.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6741b648f251144b7c64d60415916f1576331df2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6741b648f251144b7c64d60415916f1576331df2"}]},"branch":"refs/heads/master"},"7ab1748cc1e3ae3b743e7f5fcbfcffc7d94b860f":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2023-01-27 07:10:54.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/28"}}},"commit":{"parents":[{"commit":"a2bbbf2139a2958973bd416608b07ff2b9a6d9c2","subject":"Merge \"Pass service ref to init_host(), if exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2bbbf2139a2958973bd416608b07ff2b9a6d9c2"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-27 07:10:02.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ab1748cc1e3ae3b743e7f5fcbfcffc7d94b860f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ab1748cc1e3ae3b743e7f5fcbfcffc7d94b860f"}]},"branch":"refs/heads/master"},"8a22bfe6a78ace3a7d77d24229c9c6db8e60c60d":{"kind":"REWORK","_number":29,"created":"2023-01-31 07:58:59.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/29"}}},"commit":{"parents":[{"commit":"23c5f3d585ff44b106a27fd54db5e012002611d3","subject":"Make resource tracker use UUIDs instead of names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23c5f3d585ff44b106a27fd54db5e012002611d3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-31 07:56:01.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a22bfe6a78ace3a7d77d24229c9c6db8e60c60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a22bfe6a78ace3a7d77d24229c9c6db8e60c60d"}]},"branch":"refs/heads/master"},"8c2e76598995f0d417653c60a63ea342baf4e880":{"kind":"REWORK","_number":30,"created":"2023-01-31 11:51:00.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/83/858383/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/83/858383/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/83/858383/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/83/858383/30"}}},"commit":{"parents":[{"commit":"23c5f3d585ff44b106a27fd54db5e012002611d3","subject":"Make resource tracker use UUIDs instead of names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23c5f3d585ff44b106a27fd54db5e012002611d3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-19 09:35:45.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-31 10:29:01.000000000","tz":60},"subject":"compute: enhance compute evacuate instance to support target state","message":"compute: enhance compute evacuate instance to support target state\n\nRelated to the bp/allowing-target-state-for-evacuate. This change\nis extending compute API to accept a new argument targetState.\n\nThe targetState argument when set will force state of an evacuated\ninstance to the destination host.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I9660d42937ad62d647afc6be965f166cc5631392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c2e76598995f0d417653c60a63ea342baf4e880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c2e76598995f0d417653c60a63ea342baf4e880"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[]}
