)]}'
{"id":"openstack%2Fnova~862404","triplet_id":"openstack%2Fnova~master~I9ce18cbba5083c55d15d9b7c2a89133d227754ea","project":"openstack/nova","branch":"master","topic":"862404","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2022-11-03 23:57:04.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-11-03 23:57:04.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-11-03 20:56:29.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"35421":{"account":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"last_update":"2022-11-03 23:57:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ce18cbba5083c55d15d9b7c2a89133d227754ea","subject":"record action log when deleting shelved instance","status":"MERGED","created":"2022-10-21 20:14:23.000000000","updated":"2022-11-03 23:58:15.000000000","submitted":"2022-11-03 23:57:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862404-862404","meta_rev_id":"0260e68aa6aa61ac7e4a19fa6b30ae235200e037","_number":862404,"virtual_id_number":862404,"owner":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-03 23:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-11-03 20:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2022-11-03 20:56:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-11-03 20:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2022-11-03 20:47:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-21 20:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-21 21:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-21 22:24:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-21 22:34:54.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2022-10-22 15:50:42.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"CC"},{"updated":"2022-10-22 17:22:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-22 17:30:31.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-10-22 17:35:04.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"REVIEWER"},{"updated":"2022-10-27 12:40:51.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2022-10-27 12:40:57.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-10-29 15:29:43.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2022-11-03 13:48:09.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2022-11-03 15:17:45.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"6bb2130a045d4c7eb01b0c38cc6ac5348b67cbcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-10-21 20:14:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba6c86fce3f3c5eb15e1d940ae8ba76ff7a1c2f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 20:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c5f6951ce4f46229032f02634809f77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bc8f2773c8c4f5691bb010077c0327f : SUCCESS in 23m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d24c5103619240338166062a32eae0c3 : SUCCESS in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7db34d80dd049c4b4c80354b110921ed98917281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 21:46:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1aa1f2a05d2945edb9c3751496fdb6b1\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/97bce665acae404fb22747cd7806fd20 : SUCCESS in 43m 03s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f3a6e3890acd4c88a8864bd3d0cbca4c : SUCCESS in 1h 20m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/014fd404ea3b4c48806635d2d45c2e53 : SUCCESS in 36m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b73f7c98a729490eadad7dc4d113c9f8 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2afbecd5c608410581469b8ee87ca974 : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/270fcf4b39044304ba4bad0279c70d64 : SUCCESS in 13m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adabc39e8b1c47debc9c82018d0587c1 : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6052e39247894b06b3de1a40e304278c : SUCCESS in 10m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5063693e0612499db7e20f7f460a6293 : SUCCESS in 43m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c1fb5bdbcb447738fcaa5d278f48d60 : SUCCESS in 48m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9606c42b775c467a9b89c04390ead33e : POST_FAILURE in 49m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/18f13d8b3a7045f1a2436a350026add8 : SUCCESS in 43m 51s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/98f29e53401c436d89e2fba8489894f9 : SUCCESS in 1h 09m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/9facb38aba69496fa9fd95ebabb534fd : SUCCESS in 1h 29m 47s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/18339e93e075462c99436ec7262055db : SUCCESS in 44m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/18e053978ead44469ba141904d2acd57 : SUCCESS in 5m 22s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53d33064400d405dacda340cb4b3bbaf : SUCCESS in 22m 33s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cdaef1532e36442195ec2c0b135d0879 : SUCCESS in 20m 37s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/64f36124f0224221b10958b152571dfd : SUCCESS in 19m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce7bdba9ffc14d69ae17c777e6045093 : SUCCESS in 50m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98074c458066458e9296d066d8d37553 : SUCCESS in 40m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0b5302e2fcec4a2884c2c859ba19d61a : SUCCESS in 20m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/746107363e504acfbe0fbf2c61cacd26 : SUCCESS in 29m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f81b18cfef0569f5ffd7ea5aea7dfcd3bce97f76","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-21 22:24:03.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/5853d2bb847b40b1a7d481045d9ad677 : POST_FAILURE in 2h 07m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a2203bd20ea46d38ef2d7fda85e2167236709c64","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-21 22:34:54.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/862404/1/iscsi : SUCCESS in 2h 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"74045c300c82b4eb5138a90e297b434d6e24b5a8","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-10-22 15:50:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c72659835b22e8aed039867191e16728da8807ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 16:22:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9162de2575a9464ebd0775ea1ff2abd5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/374aa3ccba824f229b2b48e672bb8d85 : SUCCESS in 16m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c075a1685fe4cb48e11f0e9d44b2473 : SUCCESS in 29m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f65b551a806c0077afe38caf45aacb910a983dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 17:22:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbda8478ce664514a81b5ff72efa1b34\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5015c7f8da7d43b69114c89278f52491 : SUCCESS in 38m 18s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6c247bb366f14aa19e44a02ca5f4d57d : SUCCESS in 1h 04m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7717abed2ff642f1b78bf5de27d7f3f6 : SUCCESS in 46m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b33c409d22b74e6ea469fa02d948ac9f : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b2c80f965641458dc9bb9350245b19 : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9a830a4a0e4003a50b04b6358428e4 : SUCCESS in 11m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a737e1a560524c47b2500777eb8677a3 : SUCCESS in 12m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3a2807fe4040789e49412e711508df : SUCCESS in 10m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/83feaee4cb00443da7cee64d9b1e9a22 : FAILURE in 1h 16m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3fe23ea2e9eb4184870f4bee748bcbb8 : SUCCESS in 58m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5e2e9243bd034fecb1057579a5bd494d : SUCCESS in 46m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/359da8447df04f2fbcae9c520dae0313 : SUCCESS in 45m 33s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7a3c32ec40cb47ed9d17d7f245c804ff : SUCCESS in 1h 14m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/4873c8d2ea694108b5a2067bbdcb9ddc : SUCCESS in 1h 28m 41s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/59e76f63e9ba4988afaceaa0b08639a4 : SUCCESS in 46m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d39beec66f404b3ab9b2fb42e0e4f7d9 : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b12d1da6d3214bdd9acaee1bab6dddb3 : SUCCESS in 18m 45s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fac1d985fd6648e2b1f30cacd4ab8e09 : SUCCESS in 19m 47s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a904bdbc60e406192fab51105fb0b14 : SUCCESS in 19m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e19d6edbdab469dae417ea672bd5066 : SUCCESS in 59m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59975ebc6f1f46a9a43a86551bdab61a : SUCCESS in 43m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc5b55c669f54df2a3a23b74a82ad68b : SUCCESS in 24m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a6798c6f64844cc880c9230eee0885b : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0180499db47078694cd7d6b8c4bf975a3ce9849","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-22 17:22:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/862404/1/check/EMC_PowerFlex_NOVA/bc02945/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9a4ec2bca8ce9fa193345471b9529d3ed13c9593","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-22 17:28:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2d3857fc976a4ae39248abd463da5106 : POST_FAILURE in 1h 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"73cf63f20d250f3af76f1541b5eb32ce6b5f41dd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-22 17:30:31.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"a8847ec8deec8a9a35fc12e8759c5ab23625580b","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-10-22 17:35:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f872d06a3625045e264a0c00c489f31674af631c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-22 18:20:10.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/862404/1/iscsi : SUCCESS in 2h 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ad6f6c5529964eebd93f2a485153d44415cb0f4a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-27 12:40:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42d1333cfecb623c626b9cde540414cb19d7f762","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-27 12:40:57.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9f1f85b81f2aed4b65b99725dd53d3cac51c92c5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-27 12:43:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae55b2f130b549c37282c949669308a824558ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-10-28 17:17:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8eedabca6c576a11fb74e1ce64b542bc137b9c97","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-28 17:29:57.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46b76835e30a2696f104ff2e7af24e988e6d4aab","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-10-28 17:32:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0643bdca7be839aa4c6b2919d72c827a56e76132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 17:40:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9007cd75d7544082a09f5d41244b0ce2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af6bb60b72cd4ce2bb273287e468f721 : SUCCESS in 17m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c56a370572d3426cb80f0c5723e80bbe : SUCCESS in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"543db97c41f2c7b14e1b97430eee7f48b3441584","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-10-28 17:53:39.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"922a2ab54452e250f813f772f850fa5215aa6eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 18:33:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe10014389b4481b73ec16758cac972\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3fd689810db44f09b1ff06cc2d87f1b5 : SUCCESS in 51m 31s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4fe42705d7da4822a1f30509746ca82c : SUCCESS in 1h 03m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e510883efef8460ba5f8c95077de2ba9 : SUCCESS in 42m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7842c7a5bd9046609dd03e9be60ed24f : SUCCESS in 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf98a1fd91cd48d0b1fe713db0eb9eef : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3123de999554d5381ff6c59f66c261e : SUCCESS in 12m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cfe9d3cb3a2499dba166dade683cf96 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6160ef5dc8b344efbe576e0ddff12524 : SUCCESS in 9m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2c0c0973ea0e47fdbed738155a208d75 : SUCCESS in 45m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b1e98a9b90b49a797a4da9556fc5bab : SUCCESS in 1h 04m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9d5329f04a974c649f76eeba550fad47 : SUCCESS in 39m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/31c5117913be4032a1e8a8073a2f671f : SUCCESS in 36m 49s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/770be2939959414cab5cf5c7d19a3f01 : SUCCESS in 1h 08m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/284fe1da6460443f8b0d491bd43683cc : SUCCESS in 1h 12m 36s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f6715d09a31142c3991c2d58e4c574db : SUCCESS in 47m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/65b038e4d21340cf9021c5bbd9ca5e08 : SUCCESS in 5m 12s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cf33479e73e454196fae2028169735f : SUCCESS in 21m 14s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fd93cc696e104cb2be945d07b37237fb : SUCCESS in 20m 33s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ee1af6c6ef74d57bd4501b35f4d3011 : SUCCESS in 21m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0e3dbbe57a44984ab222c1bde6f66d0 : SUCCESS in 1h 02m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de10c8821b8b45b6b0679c0fbec2a6e9 : SUCCESS in 45m 32s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6b6b7db986db44569ef7042134ca5659 : SUCCESS in 33m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/77c738a75cff40d087777c869e30179d : SUCCESS in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"daa00ace34c7a90c738fc04b5a2458897062bfe2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-10-28 18:33:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/862404/2/check/EMC_PowerFlex_NOVA/09a598a/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"16bb88d9fdd3d637a0893580d7ca936a8d8aafa3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-10-28 18:40:54.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"3b7fa648012987b8ef6d2ca28f1b7df4a3e95d15","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-10-28 18:45:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1e2b12362cbd69d7b6564c35eddc335d35ada5f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-10-28 19:02:10.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/e80b759c47444024bb4933d3a9c3ba5c : POST_FAILURE in 1h 40m 58s","accounts_in_message":[],"_revision_number":2},{"id":"fd16a58a8d35e352f0de41f53f0cc886bee49ba7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-10-28 19:44:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/862404/2/iscsi : FAILURE in 2h 24m 29s","accounts_in_message":[],"_revision_number":2},{"id":"00ef56210526a1f10fd429907a9b0a721ea32d04","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-10-29 12:34:04.000000000","message":"Patch Set 2: Code-Review+1 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"4d454a9e5f064896ddf1a120bfd91b195d039785","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-10-29 15:29:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b49febeba7e6f73f006ecf6cc2f78e50d5c9eb5","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-10-31 12:35:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"709dedb838ad76d280caa9a651941d8b46be99c7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-03 13:05:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74cf37a1ea13f27dfa89b92db3797f04fea9f72f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-03 13:48:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40a4d9e4bf25ba863b9f9f6daeeafc625583259f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-11-03 15:02:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d2f0de89aebc4482896ec7163c2a3fbab51eb718","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-03 15:17:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be8acc544a1aa209f91994342eff5c2d4da9481f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 15:42:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e184ec07e5a451da2bed60ee1643f5c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4138ac1e68484d13909f855263cd370f : FAILURE in 37m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8b85f5c059f4abc86e5a23a50940fa8 : FAILURE in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0edfd9e73fe77c50cf56a4b0bd709cb68c316d12","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-11-03 16:09:43.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d1693402b0c403c15f168d19ceda1a90fd86bd8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-03 16:42:33.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/0980cba64bc14aefb3b62a1154cb1271 : POST_FAILURE in 1h 36m 58s","accounts_in_message":[],"_revision_number":3},{"id":"e46d57a27df1162eeff2463b3a503f301ca1324f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 16:58:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9290816df49411482d34a9986d7df69\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4465cfc0245445b383e87815c076d26a : SUCCESS in 1h 06m 21s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f1b91d9a3f1f48f48f30b664d200c612 : SUCCESS in 1h 29m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dd77999594045ccaf834ba9cd376661 : SUCCESS in 58m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f800b3a46bd340c3bd4b5013ade177db : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/621ff9ade051430a9e9f552042127312 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6a1d2acc0d34b4c819298c212483d6b : SUCCESS in 12m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8be7d633dc3f466fa1de1e4905cfef0d : SUCCESS in 13m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ccdd0d27354a97a6f8a5b8cad0ef26 : SUCCESS in 10m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/756649a9269b4ccdb8b6b3fd4d392d5d : SUCCESS in 1h 09m 43s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9dbd9c24aa744ce4a31b44e043a79dd7 : SUCCESS in 1h 12m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/023064607e65459ea2eed0acf7c2bf66 : SUCCESS in 53m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/db8dc63c8b504415bfec3868117bdd88 : SUCCESS in 56m 28s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7a802b05659d42bb8e3b9b856c3c2059 : SUCCESS in 1h 33m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/09fe8be5a04b4f6f8a0be40d317a5493 : SUCCESS in 1h 48m 45s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3fd9c89b053d422e969a143ae0abcf78 : SUCCESS in 41m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a7016017bc2044098c262ea661fbd921 : SUCCESS in 5m 56s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53b2fbbe01784959a984654873022946 : SUCCESS in 22m 02s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce49bc3e77a04d2f86a4c7126ff59d3f : SUCCESS in 21m 25s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d3a2bde11b44d6aa0a221618aa60d81 : SUCCESS in 21m 54s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/34d24bbac7264033aedaeb8b3ac6a3f6 : SUCCESS in 54m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a9229baaf4a4b26915a30e550351b07 : SUCCESS in 48m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6f10bfdfc33340aca50a1df0deedd7e0 : SUCCESS in 23m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f6e422261119446c890f01d2b4a01bef : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"776df78452830026f425a48c71bc12ac8f83430d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-11-03 16:58:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/862404/3/check/EMC_PowerFlex_NOVA/c2f8610/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"5e6cde2744f53cb337e618d030557f66044b7772","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-03 17:06:38.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"679ee7670ebbcb7f712a99e9102f262ef6b6f586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-11-03 18:19:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f1777e105944835438f78d46f2ad69e9e5b0f3b7","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-11-03 18:20:27.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0c8acd53811bf4b2fd58e4eb2cb20f4e9c1baa16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 18:36:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad26e79b0f814a92ba66166d62e6fc3f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7948c6ecbed46c993f378d3a79ff21c : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/642787f1c7044c01b4059793905dc874 : SUCCESS in 17m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"577962d2f521e6ca707cde17fb643b9baec5629e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 20:01:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56a1da6fbc9e42e9a5dcf37208454f9d\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fa0a470c73fd4ca1b123e9c2aab23563 : SUCCESS in 1h 10m 17s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1afc89a3c5b24fbf9b6a68e1e1be9fca : SUCCESS in 1h 37m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf2f56fb027243b8b40d4a4404c0cf5e : SUCCESS in 58m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f74c10f1c5247ab89a7505a0581cefd : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ea8bcf48144957a7369b7a03ed9f6a : SUCCESS in 12m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc446b32f1334b0bb9785ed299f843d4 : SUCCESS in 12m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9606124f2a649b4be44679b9b34a7fe : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c0a5a32568403e93b5d61ecf618ee8 : SUCCESS in 9m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/94dbecd9f872458da71687efc1bb2542 : SUCCESS in 1h 09m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3f56c86d430a432fbcb70baa9b79949a : SUCCESS in 1h 17m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a5c91944d2de4633bf06ee1f954ba43e : SUCCESS in 53m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/64e763d6637b4c82bb095aaae04416ad : SUCCESS in 54m 32s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3e6b922f81da4cd4a53811e76a491854 : SUCCESS in 1h 06m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/5bc1c58f0dbe48c6b78b2de5d0710f40 : SUCCESS in 1h 33m 39s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b5d40216679e4e8ba63bacb8f8c7e75a : SUCCESS in 49m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bce235d0652441f9b05d9c0d513f54f3 : SUCCESS in 4m 43s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b17427d585f74156a95f44a387ac4c94 : SUCCESS in 17m 31s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37002d87522349a18be653f0b6c97bcf : SUCCESS in 18m 05s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a4be512489ad4944939b92d39f80599c : SUCCESS in 20m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc6d92b4e12f497fb3b990fdf0253c21 : SUCCESS in 1h 11m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd24f441ef244ce79547e9f188b84e07 : SUCCESS in 35m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/47f3688313924a3d8b4008cfd0de5962 : SUCCESS in 23m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/266c75a50af14d8da8adf9d534192f36 : SUCCESS in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"327e229df1797cd0b9fadfcb968dea69f8743955","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-11-03 20:02:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/862404/4/check/EMC_PowerFlex_NOVA/69873cb/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c5d706e846de4b0ecc77c41ab688ec4cdf0a1f58","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-03 20:07:46.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/72beb630d49d4ca1be6518fc843c9b4f : POST_FAILURE in 1h 45m 43s","accounts_in_message":[],"_revision_number":4},{"id":"5d3388774a10d0c7c1e9ee56eab6c38239a75d19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-03 20:09:45.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"22623efddb4a1cfeb3370a510322e46e59c0d1d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"date":"2022-11-03 20:47:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f27dfc643638015fb5a85ae9ec1b3c6fc5b5bdc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-11-03 20:56:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f1a8c05a0022dd35c3f8e9a3beee5207008d81f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 21:11:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f09d4868065543de8baf58fbecf12167\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bb4c71007ac47e785fd68b6fe032f6d : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c74a04c5bbf24003a4a770d5cd6f11ad : SUCCESS in 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2f2f209830c5059672ae817833f6ab9783db16a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 22:15:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6cf669dd7ac47cd88e5231764815520\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6d842fd332f542928eb150864bbae7a0 : SUCCESS in 54m 51s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b000b663b693480787712e867d11b234 : SUCCESS in 1h 21m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e696191de80844138a0ab162fa486db9 : SUCCESS in 46m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09abe3d0984a48a7bd8232044a6a08be : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a8c15a293842c9b33f7b932e919273 : SUCCESS in 10m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dceb3e09ca04d67ad63aa9f703b42a2 : SUCCESS in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3950af9ffb2648ef994a856509a7f788 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa15df6789bb42d799a5086829e5ff93 : SUCCESS in 9m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b7a9124656d24000b481913a0fd107fe : SUCCESS in 52m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/33d4fdfe72784a489bb9565b0f0f4df3 : SUCCESS in 1h 04m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/066dec24482d4c968653d9ea43dd61dc : SUCCESS in 43m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/da2d5502df4947e196104fc1f1cd97e5 : SUCCESS in 43m 26s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/735210da4f01454a8a720f5835d17252 : SUCCESS in 1h 26m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/3cf8faf63288407ab2ad5cf953526363 : SUCCESS in 1h 21m 58s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/43380642b4734d3fa1e2e03df7fd1011 : SUCCESS in 46m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/440eadacc6fa4c2199e7b82d251f6d06 : SUCCESS in 7m 04s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45d92c68ea7d428da3cfb1d66bb085cc : SUCCESS in 21m 46s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b0e264a8090417eaa851af2e52e1939 : SUCCESS in 22m 40s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c548c7c476444f92b18f7204005c658c : SUCCESS in 24m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/afc8f94cdd1f4aa89628a28ab0d49f48 : SUCCESS in 1h 01m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/defa00eb03d045649285fa37d60cfad6 : SUCCESS in 42m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/172ac4b6867c4c22b98ba49cf0377bc9 : SUCCESS in 23m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bf2e16e1b468428c9c67ee9aed1eaf44 : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9204684701b051565d3d2abe3d11fee7ec24099d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-11-03 22:15:49.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/862404/5/check/EMC_PowerFlex_NOVA/113a206/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"0eceecb58ef5f0c065ee6d97d02754e04dfff4a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 22:16:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b2440e3bc3b870005e69859f3a2f36a6fa2ee82c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-03 22:25:58.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"36c3b21f1e69a67c4360a7c7be4a30ba6ae40a2f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-11-03 22:33:49.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/904637f085b648a6987a6f39e96af56f : POST_FAILURE in 1h 42m 37s","accounts_in_message":[],"_revision_number":5},{"id":"7df524126e06be9799081d6a4f61e7b05ec82fc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 23:57:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3b3498ff93459186a106719fff7149\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5e0dfb4f52c0467cb95e4cb5cc1dfa79 : SUCCESS in 1h 09m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e22c6c434e44a4b80aa206a1b8af631 : SUCCESS in 37m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f884b292f7be44fc9a66b31701712f54 : SUCCESS in 8m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e48d43e750846d09cdc061b8c8bf5c1 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/033a8e496ee34f04be91f80e628c50ff : SUCCESS in 10m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da36243bb5bf437ca15691b2d9f19453 : SUCCESS in 9m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a9264d8766f84fd6a1411cfdba1bc019 : SUCCESS in 35m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a4923ed9b9c94436bff9670eff413fbe : SUCCESS in 46m 36s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89f26a8e25404c0793e1358da5de3685 : SUCCESS in 19m 25s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2d9d447c43b493982e1540dee0e3e5e : SUCCESS in 21m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3553210b3f094b8e91ad89123cc20669 : SUCCESS in 1h 37m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/678d11e9ca3d43aebcc0fcb718d8ae86 : SUCCESS in 1h 31m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/25edfd7e50c045faa1c16ec9b074bc50 : SUCCESS in 5m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b33c0bfa2237472891f27b6528a145ed : SUCCESS in 1h 02m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/23bf96d5c452427989ec7bab9ff4a52a : SUCCESS in 1h 07m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18f840506a274d5aa3f04c7afbc51bbe : SUCCESS in 43m 42s","accounts_in_message":[],"_revision_number":5},{"id":"7cbca370b1a58038e990d740a3e2e96e4a0f6acd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 23:57:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0260e68aa6aa61ac7e4a19fa6b30ae235200e037","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 23:58:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aae78b82ee144b61bd799b690db438c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37a00699d164aab94455dcd17ceb3fd : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c97507dfcd57cce9d76670d3b0d48538900c00e9","revisions":{"aaf82370a79dde4f0ca662f8a02987cff45eef42":{"kind":"REWORK","_number":1,"created":"2022-10-21 20:14:23.000000000","uploader":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"ref":"refs/changes/04/862404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/862404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/862404/1"}}},"commit":{"parents":[{"commit":"b1958b7cfa6b8aca5b76b3f133627bb733d29f00","subject":"Correct doc references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1958b7cfa6b8aca5b76b3f133627bb733d29f00"}]}],"author":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-21 20:09:39.000000000","tz":-180},"committer":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-21 20:14:17.000000000","tz":-180},"subject":"record action log when deleting shelved instance","message":"record action log when deleting shelved instance\n\nChange-Id: I9ce18cbba5083c55d15d9b7c2a89133d227754ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aaf82370a79dde4f0ca662f8a02987cff45eef42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aaf82370a79dde4f0ca662f8a02987cff45eef42"}]},"branch":"refs/heads/master"},"5ac4a90fed3a6e39dc67467e284d3c8f33ce4f6e":{"kind":"REWORK","_number":2,"created":"2022-10-28 17:17:39.000000000","uploader":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"ref":"refs/changes/04/862404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/862404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/862404/2"}}},"commit":{"parents":[{"commit":"b1958b7cfa6b8aca5b76b3f133627bb733d29f00","subject":"Correct doc references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1958b7cfa6b8aca5b76b3f133627bb733d29f00"}]}],"author":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-21 20:09:39.000000000","tz":-180},"committer":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-28 17:14:58.000000000","tz":-180},"subject":"record action log when deleting shelved instance","message":"record action log when deleting shelved instance\n\nChange-Id: I9ce18cbba5083c55d15d9b7c2a89133d227754ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ac4a90fed3a6e39dc67467e284d3c8f33ce4f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ac4a90fed3a6e39dc67467e284d3c8f33ce4f6e"}]},"branch":"refs/heads/master"},"6e0fb6194c6c4024c76656f09816402850b71548":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-11-03 15:02:05.000000000","uploader":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"ref":"refs/changes/04/862404/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/862404/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/862404/3"}}},"commit":{"parents":[{"commit":"b1958b7cfa6b8aca5b76b3f133627bb733d29f00","subject":"Correct doc references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1958b7cfa6b8aca5b76b3f133627bb733d29f00"}]}],"author":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-21 20:09:39.000000000","tz":-180},"committer":{"name":"Alex Chan","email":"alexchanwork20@gmail.com","date":"2022-11-03 15:02:05.000000000","tz":0},"subject":"record action log when deleting shelved instance","message":"record action log when deleting shelved instance\n\nBug link: https://bugs.launchpad.net/nova/+bug/1993736\n\nChange-Id: I9ce18cbba5083c55d15d9b7c2a89133d227754ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e0fb6194c6c4024c76656f09816402850b71548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e0fb6194c6c4024c76656f09816402850b71548"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ae4ba7907bd29f90de7b1d906911a5c92e1a7433":{"kind":"REWORK","_number":4,"created":"2022-11-03 18:19:14.000000000","uploader":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"ref":"refs/changes/04/862404/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/862404/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/862404/4"}}},"commit":{"parents":[{"commit":"5ad1555a8502ba32949f7f76363f5bffec24a7a5","subject":"Merge \"Enable glance location tests on ceph-multistore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ad1555a8502ba32949f7f76363f5bffec24a7a5"}]}],"author":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-21 20:09:39.000000000","tz":-180},"committer":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-11-03 18:18:53.000000000","tz":-180},"subject":"record action log when deleting shelved instance","message":"record action log when deleting shelved instance\n\nCloses-Bug: #1993736\n\nChange-Id: I9ce18cbba5083c55d15d9b7c2a89133d227754ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae4ba7907bd29f90de7b1d906911a5c92e1a7433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae4ba7907bd29f90de7b1d906911a5c92e1a7433"}]},"branch":"refs/heads/master"},"c97507dfcd57cce9d76670d3b0d48538900c00e9":{"kind":"REWORK","_number":5,"created":"2022-11-03 20:47:21.000000000","uploader":{"_account_id":35421,"name":"Alex Chan","email":"alexchanwork20@gmail.com","username":"alexc20"},"ref":"refs/changes/04/862404/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/862404/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/862404/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/862404/5"}}},"commit":{"parents":[{"commit":"5ad1555a8502ba32949f7f76363f5bffec24a7a5","subject":"Merge \"Enable glance location tests on ceph-multistore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ad1555a8502ba32949f7f76363f5bffec24a7a5"}]}],"author":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-10-21 20:09:39.000000000","tz":-180},"committer":{"name":"alexc20","email":"alexchanwork20@gmail.com","date":"2022-11-03 20:46:48.000000000","tz":-180},"subject":"record action log when deleting shelved instance","message":"record action log when deleting shelved instance\n\nCloses-Bug: #1993736\n\nChange-Id: I9ce18cbba5083c55d15d9b7c2a89133d227754ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c97507dfcd57cce9d76670d3b0d48538900c00e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c97507dfcd57cce9d76670d3b0d48538900c00e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}}]}],"submit_requirements":[]}
