)]}'
{"id":"openstack%2Fnova~864275","triplet_id":"openstack%2Fnova~master~I14c2684edf5e9a04c673d039d340afe23e448ff9","project":"openstack/nova","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-12-22 08:38:26.000000000","reason":"\u003cGERRIT_ACCOUNT_31827\u003e replied on the change","reason_account":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"}}},"removed_from_attention_set":{"31827":{"account":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"last_update":"2022-12-22 08:38:26.000000000","reason":"\u003cGERRIT_ACCOUNT_31827\u003e replied on the change","reason_account":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"}}},"hashtags":[],"change_id":"I14c2684edf5e9a04c673d039d340afe23e448ff9","subject":"Remove the redundance code in HostState.update","status":"NEW","created":"2022-11-14 03:56:51.000000000","updated":"2022-12-22 08:38:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"1af32c0672436258c3ba7ab372513421e56862d1","_number":864275,"virtual_id_number":864275,"owner":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-15 04:02:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-14 04:30:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-14 04:37:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-11-14 05:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 04:10:47.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-12-20 12:47:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"d025954aa03a2b35b4e61c4f53b861057c8f5fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2022-11-14 03:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9437692e5e2d0d623300d146e4bc641c2a36020","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 04:30:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787c07adaa4e41f3910ca666be53ea79\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23599940817f45489c55e9f6414c6eee : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a52499dfb0704ecfb25fd30365f839bd : SUCCESS in 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"372a419304be848ca1beea4b4b0157a6bf10c965","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-11-14 04:37:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/75/864275/1/openstack-sriov-nova-master-direct : SUCCESS\n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/75/864275/1/openstack-sriov-nova-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"6181fe7847d380e800c538c895ab3f0cc612dc42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 05:18:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3eee61d1a32245138b17f2e41a3697fc\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/096dd1167ba140d9a872e7a458a42fca : SUCCESS in 41m 54s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/70361da3d21b4689a8ba68c7083226c3 : SUCCESS in 1h 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb2b90f06cb14afb9491af6131741831 : SUCCESS in 40m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/286f391dec5746d4b7b2ed2d40a8b736 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b89f2d979941b8ac7db9da93cba332 : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78df234993d646e2abd59aa41419096b : SUCCESS in 12m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a43c329f9b54e308c158ff3678fa0ac : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ddb4799de049f19f9f63a0d83e22f9 : SUCCESS in 7m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3f5bb01976bc4302af055206a2a4b236 : SUCCESS in 42m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a81dd75460084653b3fe2f866b3d742d : SUCCESS in 57m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/933e12828be04decb33d364d3ff9dd54 : SUCCESS in 32m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b4ccaa749b9e4c3e843a50817549a783 : SUCCESS in 33m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/be09d4877b864ed08f3e69d87398aaa6 : SUCCESS in 1h 16m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/251d2b02fcfa4741b3be4211a4ecec5e : SUCCESS in 1h 08m 48s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/31a5cafe8eec409fb4384e89c58c745c : SUCCESS in 48m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7c4a35df21d342eba310b71fa9d88c7b : SUCCESS in 4m 13s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/57ffde55b67547a5a3a02e66d90abb7e : FAILURE in 21m 16s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/614e6452a5d74349b8194f7997e1b030 : SUCCESS in 20m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a688e8eea7ff4fc98e2b4d2073a892c0 : SUCCESS in 21m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/83a317c414f947498f4f51b462ca5a81 : FAILURE in 1h 05m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fadbd4ba0ac74ef382d28eae496e9687 : SUCCESS in 45m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/78cfbf8cb24242b384983f4623430323 : SUCCESS in 24m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3c89565f4b6b47deab858229dead8729 : FAILURE in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87284d7911142926d0b14c67bf11bc5c06889729","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2022-11-14 07:08:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc902aa74bdf00ad2c28187c0cf3ff812a322a32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 07:43:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0ead3413364730837b06bc55863bdf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d6d1d2cdf4d4c3ba3dadeba94e011ab : SUCCESS in 28m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41d75eb12ba7484791afb73501791a78 : SUCCESS in 29m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62480a4023fd2c1803f8403cce366bb3b67321e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-11-14 07:50:03.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/75/864275/1/openstack-sriov-nova-master-direct : SUCCESS\n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/75/864275/1/openstack-sriov-nova-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"558378aa32765c18000fdbac3a14a0422bc304f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 08:29:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d25d9a122424b3d8078e1aecff72877\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7ed6f3f7f57d48779c53c3415803a895 : SUCCESS in 52m 59s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8a3183e09ec54357a30666658c912e7c : SUCCESS in 1h 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c9388a01bb346509440fbe467a55b65 : SUCCESS in 47m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4018d8f8a5e40f4add07affbc20e356 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b4f0b7d11540c6ae6189f2e9a5e9c0 : SUCCESS in 10m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a51a9fa56cef477d879b1fc7c3f09b0c : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ff2d3947ad148bd8f9c1b5d7e6dc447 : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c0e9976c2e45908d9257283604b3a3 : SUCCESS in 8m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2eface3ed654444aa6e1ddc34b952bff : SUCCESS in 54m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/da49696ae5ed44d083376548d65f8ca1 : FAILURE in 53m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9a8d4515952845958a3c5d48df392089 : SUCCESS in 43m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/438123e033f9402a81931315cd722675 : SUCCESS in 37m 12s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a39e64c0ddbc425099be4bb5844fcb19 : SUCCESS in 1h 19m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/cdfcfb7a247747469ae1772a5e5b6300 : SUCCESS in 1h 19m 10s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5aa13941c654430a9be0499e89994db4 : SUCCESS in 48m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5f6aa6906cff4617a794abacd1654b90 : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b76ebd01856f491e9a17664a5e1d9121 : SUCCESS in 19m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0fc96968580846fb9f48c60166c7ec67 : SUCCESS in 19m 07s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae864928ebce4b4590ca404f3dd3a4a3 : SUCCESS in 19m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ba87ee5ff8f46189eb5f355d524ddfd : SUCCESS in 1h 02m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01036ce4115f4a368910ba646fa4d340 : SUCCESS in 46m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0d33326383384888ab00dbe353c91936 : SUCCESS in 24m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/659e9cb39b8c45e9a2493398f24efae7 : FAILURE in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"899a9da742472d8753c24f3132d664d80f61e612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2022-11-15 02:18:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"21f767c4abd2ce4ae493ec7b1a1c52d527fe5342","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2022-11-15 02:18:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e406733db194f8962063a5bb5609a8dffd35c12f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-11-15 02:19:01.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/75/864275/2/openstack-sriov-nova-master-direct : ABORTED\n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/75/864275/2/openstack-sriov-nova-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"0b0678c1dec21b31fb57bd58c0c5a752344131e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-11-15 02:59:37.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/75/864275/2/openstack-sriov-nova-master-direct : SUCCESS\n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/75/864275/2/openstack-sriov-nova-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"4ec81ea34218330ef08c23c1678a677e36619d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 03:13:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2b1499a9584b81a5e83c3dd34f0c37\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ba370367886446d84865fa4a73167ca : SUCCESS in 34m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/16c56d7ce6224339a3d59d74b9cece05 : SUCCESS in 39m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c72a6c59f0e0ff22e259e6bfdfc943da11674da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 04:02:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87ded8ff61b4455f820118aabcbd70d7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8c43ef3987454a8c9660471976f5b35b : FAILURE in 50m 26s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2b419fa5085140e69f9ea9f3ad5177fa : SUCCESS in 1h 17m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4030d026e6a240bd9d53512af171fb3d : SUCCESS in 49m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec344c3991d24277a41ea1761c588774 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f90051ee233457b872b6e0b6aafeed6 : SUCCESS in 12m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7804e3c5167b477ea1236267b04effc2 : SUCCESS in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b2a68e33664487d9db99f4891533e7b : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd66070c81df408aad0dd3d100e9cec9 : SUCCESS in 9m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/37ef7264ff2d48b882f427fdd2dce6f6 : SUCCESS in 1h 06m 48s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c90be4c0d2b47ec83afb41bde519043 : SUCCESS in 1h 22m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9fc72b82bc464778a825ff2d111c0a19 : SUCCESS in 53m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b4d57bf63a8549fba140af49c592fafa : SUCCESS in 1h 01m 30s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fe50afe336fc403686d17cd2b8e30be4 : SUCCESS in 1h 26m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/da4b3b8d154a4d68b147f8b4d6c89a61 : SUCCESS in 1h 38m 07s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b68c4055cad841b88b6ddc016b6e95cf : SUCCESS in 47m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a6fda9788e0f4869b4cfa01af33b790a : SUCCESS in 7m 05s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e458356655041018fcce803b2d4a749 : SUCCESS in 20m 36s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97c41e728a1f40cf88075797a3f77dc5 : SUCCESS in 22m 01s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c4b45fd0ddb4ee584535af7a4aa78d7 : SUCCESS in 20m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f033f46984274902802becb8bc69c630 : SUCCESS in 1h 02m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5597079028ce402faa16521d130cd123 : SUCCESS in 45m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8cdeaab408d24374ab1cfe26218256d1 : SUCCESS in 27m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7025bf125b714cb1868ab910e3257773 : FAILURE in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28ab7b5aea33fa4d5c7a40ef4feeda28b4a7c51e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-11-15 04:10:47.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"bb0be2af7a254f0b2b5c73d37b7710d357f0b825","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-11-15 04:53:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/75/864275/2/openstack-sriov-nova-master-direct : SUCCESS\n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/75/864275/2/openstack-sriov-nova-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"d1e74602fe5cb046ecc28317d05fa8a581e0c2f3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-20 12:47:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1af32c0672436258c3ba7ab372513421e56862d1","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2022-12-22 08:38:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12f04a3017aaa5d86729cf640f637f7b0b8d59b5","revisions":{"a061c0c63b4adbb6b9b9ca3156c5c2a49d54a1b2":{"kind":"REWORK","_number":1,"created":"2022-11-14 03:56:51.000000000","uploader":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"ref":"refs/changes/75/864275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/864275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/864275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/864275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/864275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/864275/1"}}},"commit":{"parents":[{"commit":"8aee01f1286d1f9ec5190328d24e9eb6b4edb626","subject":"Merge \"[compute] always set instance.host in post_livemigration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8aee01f1286d1f9ec5190328d24e9eb6b4edb626"}]}],"author":{"name":"dengzhaosen","email":"dengzs09@chinatelecom.cn","date":"2022-11-14 03:56:06.000000000","tz":480},"committer":{"name":"dengzhaosen","email":"dengzs09@chinatelecom.cn","date":"2022-11-14 03:56:11.000000000","tz":480},"subject":"Remove the redundance code in HostState.update","message":"Remove the redundance code in HostState.update\n\nThe HostState is local variable in HostManager, So the updated is\nNone for HostState.\n\nChange-Id: I14c2684edf5e9a04c673d039d340afe23e448ff9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a061c0c63b4adbb6b9b9ca3156c5c2a49d54a1b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a061c0c63b4adbb6b9b9ca3156c5c2a49d54a1b2"}]},"branch":"refs/heads/master"},"12f04a3017aaa5d86729cf640f637f7b0b8d59b5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-15 02:18:29.000000000","uploader":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"ref":"refs/changes/75/864275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/864275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/864275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/864275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/864275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/864275/2"}}},"commit":{"parents":[{"commit":"2eb358cdcec36fcfe5388ce6982d2961ca949d0a","subject":"Merge \"Test ceph-multistore with a real image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2eb358cdcec36fcfe5388ce6982d2961ca949d0a"}]}],"author":{"name":"dengzhaosen","email":"dengzs09@chinatelecom.cn","date":"2022-11-14 03:56:06.000000000","tz":480},"committer":{"name":"Jorhson Deng","email":"jorhson_deng@163.com","date":"2022-11-15 02:18:29.000000000","tz":0},"subject":"Remove the redundance code in HostState.update","message":"Remove the redundance code in HostState.update\n\nThe HostState is local variable in HostManager, So the updated is\nNone for HostState.\n\nChange-Id: I14c2684edf5e9a04c673d039d340afe23e448ff9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12f04a3017aaa5d86729cf640f637f7b0b8d59b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12f04a3017aaa5d86729cf640f637f7b0b8d59b5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
