)]}'
{"id":"openstack%2Fnova~869687","triplet_id":"openstack%2Fnova~master~I0f0520b0d01c85ff4a5f9fd7974262a8daf5e360","project":"openstack/nova","branch":"master","topic":"sync-all-bm-nodes-to-aggregates","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-09-29 10:41:10.000000000","reason":"\u003cGERRIT_ACCOUNT_14525\u003e replied on the change","reason_account":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2023-01-10 17:03:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-01-10 18:03:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0f0520b0d01c85ff4a5f9fd7974262a8daf5e360","subject":"Process all nodes when host aggregate changes","status":"NEW","created":"2023-01-10 11:49:52.000000000","updated":"2023-09-29 10:41:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"feba481ba6153f89f80687f73ae141e790687986","_number":869687,"virtual_id_number":869687,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-10 18:41:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2023-01-10 17:03:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-01-10 17:46:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 11:50:30.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-10 13:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 17:03:47.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-01-10 17:46:44.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-09-29 10:41:10.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"CC"}],"messages":[{"id":"9a5d23934762130f6a37d439a5cfb503c8f0c024","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-01-10 11:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e226b0fef4f740b8c4615a030b15aba4eab1f76e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-01-10 11:50:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/87/869687/1/openstack-sriov-nova-master-macvtap : FAILURE\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/87/869687/1/openstack-sriov-nova-master-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"2f7f0de074cefb435e21c5cdba02e8436612675e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 13:29:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0ae3968aa5148d9b7d461638b532b58\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/97cac76da28b48d89ed90f405e63d7ff : SUCCESS in 47m 42s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e05f8f88e47441faaa541538068d7e6d : SUCCESS in 1h 37m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47ae2c43194741ae8d5fb5cbd678052e : SUCCESS in 52m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/892312838c5b4711b0419c7cc3819db4 : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82bdc667de34e93ab3190a815c14fa8 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2ce5f4195904eabb38f4df8b903bf4a : SUCCESS in 12m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1163baa170d40a9b6f641bfc33eae65 : SUCCESS in 15m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0a26176f974a338deceab712ee4c33 : SUCCESS in 9m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fd2a4d65a3b743b5a061d4914150f719 : SUCCESS in 51m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aafc570957724832bd50cd61dbb61b8a : SUCCESS in 1h 15m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cf821dc0480e466cbb9264ee3de3cd52 : SUCCESS in 36m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8e5d3f004d594070882a0540ceba5799 : SUCCESS in 54m 20s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/319482e5277942c3b13b9f9eec1c069d : SUCCESS in 1h 29m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/14febfc16e244e7597dbdbd30f5a79d8 : SUCCESS in 1h 20m 29s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f8ff40c576b34c9d85727c9dd3f577b2 : SUCCESS in 59m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7eb663211cd8419b87035012ba3c1b8b : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c39846db244d4a17bf8a51bc8bd8cbfd : FAILURE in 3m 46s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aeaf7a68005240ba885f1d693d031066 : FAILURE in 4m 55s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c16b68d6cb0c437999ae49fb42159619 : FAILURE in 3m 45s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/660ae434c7e449e783d449ba81512ffb : SUCCESS in 1h 30m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/42fa906ede444e999aa46d05a4c60abe : SUCCESS in 1h 08m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c0c28f75ae346f4acfda057dcd115ec : SUCCESS in 40m 32s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ad31f8415dfb4dcaa6fbbed82312a466 : SUCCESS in 22m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1ebb2bd7aa504ccba1c6595b3d19a4bd : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f98ac7acf877c8f2b7b67e006c702f595880e7e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:05:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d11d94f54b1044f2a17fc11d88e9714d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3565012429e94fc9a1cf81d5a2867d06 : SUCCESS in 32m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b048b12a3974d9f97c2028a30d1df93 : SUCCESS in 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a23e2a47fff1e78fdbcc4c79d4dd313a80ca783c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-01-10 16:13:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10e61df646c3b19f2c434c5f02d4fbc50fd8077f","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-01-10 16:13:51.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f391ddc46213abf9930d0cf27098c5ca8156ed9","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-01-10 16:14:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d336aa6fb44b4c7643dc7050c033d3129abe300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-01-10 17:00:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"657fbe209e0011ef7fcd90358985b405713c5109","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-10 17:03:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0a2bb0fa032c0abaff3395d64367d8b73bf132a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-10 17:46:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e06eb39e32508f81454eff7df16bf7e6160d5144","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-10 18:03:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10a6b6cf3b9b31387372bc706920a7687cd2a08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 18:41:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/986f934348f443d5abb9b24c195f6426\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/41cc6b13f5ff4de8aa0eb5c157d29c16 : SUCCESS in 1h 07m 47s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/82e4cd75b0444e7285332299029df6c1 : SUCCESS in 1h 36m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddcf815f57e24aa2b948efdd53b3a650 : SUCCESS in 1h 02m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c96970da2e6435b84b5e8db28c13588 : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb91546d0b004886aa599e3c7ce70a3c : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fa22d5c993643668a6517d5811cbd17 : SUCCESS in 14m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3316383b685c4d2b8032cd800571f3c3 : SUCCESS in 15m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc3c5e1845e450eaac4489248754ff4 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bed4dc97546a4580ab868806c690c56c : FAILURE in 21m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/76d98e4e96bf49dd8539a22647b755a3 : SUCCESS in 1h 06m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0bfca593c5dd4e9e965b4cc57d54f51e : SUCCESS in 1h 21m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/608a268fcee64aeca38fb7fbe915fe78 : SUCCESS in 49m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6b49f631bd8442a6a186cc4fe06ae441 : SUCCESS in 51m 48s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3ae529d0d98546718f2005c0f6826908 : SUCCESS in 1h 19m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/4ae0566a2b054c998f323316eb0df03b : SUCCESS in 1h 28m 17s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1285f4a9100a4ba894d1be0530652b91 : SUCCESS in 43m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/79e28b7479ff4fabadac1b9686da7e38 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7d1326a19b744adbabc385c64f6a47f : SUCCESS in 23m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/191d161c1f454f6a8f41839d323e4d83 : SUCCESS in 23m 28s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3c3219c74da4fca93c66f07b80c76bf : SUCCESS in 22m 17s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89612ab0b93a4aceadafbe15790583ab : SUCCESS in 1h 18m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2032331f0119446394e3ee612d961e29 : SUCCESS in 1h 10m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e722d2634bb344f49a2e2d4fe745d1bd : SUCCESS in 42m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4b25ce6ca94b45088fc705db62c770df : SUCCESS in 36m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2171319317a64bf992c705aa2054095b : SUCCESS in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d81c5da652c465dc2a6b820120d2619a94abe688","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:27:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b32938abd70486e8b2ea7a27e662a33\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28a98f3d44494054b26dd39193b577d3 : SUCCESS in 37m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32894bccee654180adab296df91dad32 : FAILURE in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"feba481ba6153f89f80687f73ae141e790687986","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2023-09-29 10:41:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15dbef21563581372fcfb2de976fc7f307ff3967","revisions":{"046f366ad654b816521119e74f00d4ed0ffcc7a6":{"kind":"REWORK","_number":1,"created":"2023-01-10 11:49:52.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/87/869687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/869687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/869687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/869687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/869687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/869687/1"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-10 11:33:58.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-10 11:33:58.000000000","tz":120},"subject":"Process all nodes when host aggregate changes","message":"Process all nodes when host aggregate changes\n\ncurrently when adding or removing host to aggregate, only the first\nnode from this host is added/removed to corresponding placement aggregate.\nThis is confusing for the ironic case, since there multiple compute\nnodes are part of single compute host, and thus only some seemingly\nrandom baremetal nodes are added to placement aggregate (one per\nservice) while the rest are not.\n\nChange-Id: I0f0520b0d01c85ff4a5f9fd7974262a8daf5e360\nCloses-Bug: #2002400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/046f366ad654b816521119e74f00d4ed0ffcc7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/046f366ad654b816521119e74f00d4ed0ffcc7a6"}]},"branch":"refs/heads/master"},"15dbef21563581372fcfb2de976fc7f307ff3967":{"kind":"REWORK","_number":2,"created":"2023-01-10 17:00:02.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/87/869687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/869687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/869687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/869687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/869687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/869687/2"}}},"commit":{"parents":[{"commit":"5e5b6751d32aa50dc372c55b5c2044d0dade51f4","subject":"Remove basepython def from tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e5b6751d32aa50dc372c55b5c2044d0dade51f4"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-10 11:33:58.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-01-10 16:59:24.000000000","tz":120},"subject":"Process all nodes when host aggregate changes","message":"Process all nodes when host aggregate changes\n\ncurrently when adding or removing host to aggregate, only the first\nnode from this host is added/removed to corresponding placement aggregate.\nThis is confusing for the ironic case, since there multiple compute\nnodes are part of single compute host, and thus only some seemingly\nrandom baremetal nodes are added to placement aggregate (one per\nservice) while the rest are not.\n\nChange-Id: I0f0520b0d01c85ff4a5f9fd7974262a8daf5e360\nCloses-Bug: #2002400\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15dbef21563581372fcfb2de976fc7f307ff3967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15dbef21563581372fcfb2de976fc7f307ff3967"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
