)]}'
{"id":"openstack%2Fnova~870794","triplet_id":"openstack%2Fnova~master~I8ef9db851b37c5249d2efbe09a15a1ddbae8205d","project":"openstack/nova","branch":"master","topic":"fix_compareCPU_usage","attention_set":{},"removed_from_attention_set":{"6962":{"account":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"last_update":"2023-01-26 13:48:02.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-01-24 10:32:22.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-01-24 17:37:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-01-26 13:48:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ef9db851b37c5249d2efbe09a15a1ddbae8205d","subject":"libvirt: At start-up rework compareCPU() usage with a workaround","status":"MERGED","created":"2023-01-17 11:19:55.000000000","updated":"2023-01-26 17:01:24.000000000","submitted":"2023-01-26 13:48:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870794-fix_compareCPU_usage","meta_rev_id":"730028725c64ecc9507743ce180216443131523a","_number":870794,"virtual_id_number":870794,"owner":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-26 13:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-01-24 10:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-01-24 17:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2023-01-24 17:37:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 14:11:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 15:51:26.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-01-17 15:51:26.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-01-17 15:51:26.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-01-17 17:39:36.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-17 18:42:51.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 15:44:41.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"f30d8b366e2cd25cdd214e6cef9762c93bd623f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-17 11:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"634386061ec5cc72a2cb527e6443133e2c4557c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 14:11:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5df3c769a634508bc80bfafd439e12a\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/664cfd3cf01b4f63b669be7003a58151 : SUCCESS in 53m 50s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/920addb496a44ac4bd9b05f4b68e0510 : SUCCESS in 1h 22m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/111bc940498f49e0b1ca9c12e93fab42 : SUCCESS in 55m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c22354c552c54573a6b6b9aa7af8c9b3 : FAILURE in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b825f83faa40476eb35e5fdbe61fe1a8 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc6861a6421c432cb00dc7049ad23282 : FAILURE in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f78111501c1f439b81fb2dc438fd118c : FAILURE in 13m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d73605bda84ee094b5bb1efd727165 : SUCCESS in 10m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6afdbe85f9fd4b4abde5a6a0fc003543 : SUCCESS in 1h 04m 15s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/567d7893ad1c4eeeb6642995e873bae6 : SUCCESS in 59m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9c9f4d74ed1043e5bf69e10664fd490a : SUCCESS in 46m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/417380781b994925b644be575ad8f15f : SUCCESS in 52m 02s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/717ad47f8eec4a999e08af47d0d7ba3f : SUCCESS in 57m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9079c4d73a9c4aeaade7e02cce8483d5 : SUCCESS in 1h 51m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/43f4669aa3184377b138a3fa20c9134e : SUCCESS in 1h 35m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3be4ee77e90a4b1bb543679d6e3fde77 : SUCCESS in 55m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fc00ea7ed9304d5d8bf7f746cde9893b : SUCCESS in 4m 55s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b97f72335c5647a6b5cbf4a04dfbb0d0 : SUCCESS in 23m 14s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cef33ad4cd594cf187a831921014967b : SUCCESS in 23m 47s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7381d444e0b54a7ba2e8faae2b0a1128 : SUCCESS in 22m 21s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/635a84bda191434ca93b46b436ba6f0e : SUCCESS in 1h 09m 24s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c263be9f0f9541f9a82539bf4f4e6436 : SUCCESS in 1h 21m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/08d51166c17242b5886104546a698e64 : SUCCESS in 1h 08m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c71fac731bbb41d9a0b171aca5b5e6fc : SUCCESS in 50m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/84b95b19907d4c76889f706229720eeb : SUCCESS in 30m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/151d99b9b3974c4f8495077a6dc5e642 : SUCCESS in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"831b7e63c3db1d3dddb155053ef75031a31a73fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 14:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567144e7ee504f0db89999586ec3cdb8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69ccc0cfc0e8464f95acf03319dbf322 : FAILURE in 52m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea44a2ae068d48e48462ca9d5db1e2fc : FAILURE in 36m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9edde252a0317d427f8722ef7585f23c341ca35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-17 14:58:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77f14e2c82f31a06da8b130e3bb5271195586fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-17 15:37:59.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1a8e51ac43c207e73fb66a408b72a2b7a91734fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:39:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08f955e594964781b0062f77908ad0d6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c90682fcbf754cde8a83171d800db2e6 : SUCCESS in 51m 24s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/518f3d1e730840b2a045f9c59817916d : SUCCESS in 1h 23m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd06e007eca440abac7f4790d24ed807 : SUCCESS in 53m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2763d787fa74b3fa6af52d8e13d4f6f : SUCCESS in 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6b4536128b4ffab8937e76c636ce5b : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/380d0305ddcf4401bc063e67643b34c1 : SUCCESS in 13m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c53bfa90efdb41c1b7902245bde1aef9 : SUCCESS in 12m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a2ac17e7b54abcb01e7ff53d199159 : SUCCESS in 11m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6b73c46778cc406f8b28210edcde1c95 : SUCCESS in 1h 00m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/686a2eb3577648efa20ff7299db32a40 : SUCCESS in 1h 10m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e990a04821c2460dbe4aaf838c45ab3f : SUCCESS in 46m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bbdfb8d23301463e977ac632cee2f56f : SUCCESS in 43m 11s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/bd36b1d236fd425db577b487c90e7f7f : SUCCESS in 58m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/63ab3e56d1cc436187c1d24574fbed4e : SUCCESS in 1h 35m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/c025770ea3ea46fbb1f3f88acc35418d : SUCCESS in 1h 57m 38s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/bbcd3872e53948a2906c6fee274d81e5 : SUCCESS in 44m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/303cde0c7efb493385e55f1474aa9ca8 : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f9cb3c16c0f46babccc4957e23bfc26 : SUCCESS in 23m 09s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4b50e45802e45879b12910b4aa440ba : SUCCESS in 22m 34s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a80bfc79c09846cbb6269484205e355c : SUCCESS in 24m 57s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4039c0d315cc4ed99e1b3a54ed0d8f2f : SUCCESS in 1h 23m 58s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9b196a0c3f214f828a6f65b72dee9e3e : SUCCESS in 1h 37m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7f8761761f640169fbfc33c99b1ae18 : SUCCESS in 1h 08m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc9f460c6ca647ec987f1b430c3afca2 : SUCCESS in 59m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5c75ede9189d41d38e3ff314435e1f5d : SUCCESS in 20m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ad32751b89564360aa09cc8b183d34ef : SUCCESS in 38m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"805355ecc1bad7e3973f6da7cb0afcf13ed97c2d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-17 17:39:36.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/94/870794/3/check/EMC_PowerFlex_NOVA/a5bee7b/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"b0aa23d1b158238abe7b2f5221f8ee6f6882a3b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-17 18:42:51.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/fea67540472f42f686bee246a7128d85 : POST_FAILURE in 1h 58m 16s","accounts_in_message":[],"_revision_number":3},{"id":"2d534db8889a94ee4092f6bdee0846ad4b227abc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 20:42:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01fa2edacf464cf5b2e20ca4ea10ee86\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26615d0991f04192bb50273c612f8aec : FAILURE in 45m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e27f0b772e934701aa4c12a6f583d400 : FAILURE in 48m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30e15e6adb3b08bfa02145ce5940faef7612e4b7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-01-17 20:43:08.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/94/870794/3/check/EMC_PowerFlex_NOVA/9263bcf/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"68350f88ff252750e95875e436a267f58e51eb9d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-18 11:34:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ceea3fa49f5dcc100be1361bd9c44307112940ad","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-18 13:34:33.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a4c4988125deb0fedb835d8614e7e7e24d33331d","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-18 13:39:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9f281b30f23cf3a1ade14c4209725c5527dec587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-18 16:40:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ab5218c5f01478117eda0c2435b8cc6e8c2e9a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-18 16:41:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e1a94cac623f637600adadef6839a7422f3fec17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 18:56:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0187d7534ab42bbb5e857c4752599ed\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e03626079b8f4bce9617c5b402e45790 : SUCCESS in 1h 00m 14s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9912ef9d690a422598707a6f5c022da5 : SUCCESS in 1h 48m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55298f9001604b06a1da94b5d4a7fe74 : SUCCESS in 1h 05m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90292ff8b080490cba356d30efa2bba1 : FAILURE in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f174dcd6639480e867fb3739cf49493 : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f83b7fc3a0a0430c9ef2047fdef0cb5c : FAILURE in 12m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a90880abff274e379b8576610ed29238 : FAILURE in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19dd25ac078c45109764e79a5885a86f : SUCCESS in 11m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8c5895c8cdd0441a93cc16a6a6297598 : SUCCESS in 1h 00m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8bd3ab170904f898dc6492372063b33 : SUCCESS in 1h 14m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5c66fa077b614ee18d12e6435917b88b : SUCCESS in 56m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/215e68c9f2d04e9181f8413d5857df8c : SUCCESS in 44m 32s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/151c3b3a29354e4da918877e90a116bd : SUCCESS in 1h 13m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/08a5c97b2b4d4a53872850e440ff2d6f : SUCCESS in 1h 37m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/626ecd4905304ac2882f98d7789181fb : SUCCESS in 1h 39m 13s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/00f7b91886a842b0aaae0d5509cf46f0 : SUCCESS in 50m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2140025694604b7eadca74e0943739e9 : SUCCESS in 4m 44s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32dcd323ddeb44cc8b20b84f5c53b20a : SUCCESS in 27m 22s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2473410174d54cfdbbf73b2436f4a9b6 : SUCCESS in 24m 47s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/46995a36aec84ccda6bd89132caf994e : SUCCESS in 27m 46s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d77b819adfe04fe8a6cc248ec325a6ed : SUCCESS in 1h 29m 20s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/685cab09476f4415ae4d140bbeb531fd : SUCCESS in 1h 39m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/44ef1e58993243c09b3b8594f2bbed00 : SUCCESS in 1h 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e9adfb7dba242b2904d599939a15d2f : SUCCESS in 46m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e0b2aacbbe75479cbdbe030d7f8b0be1 : SUCCESS in 26m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5eb920959cdc488e84b1ffa60c44a4ea : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c1b4ca82d8d1508071e22ee6458cbeda23330cac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:41:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7667da119504bd5a7a473ba43b9d83e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa0e632ea6754d92bfb3ad9b8c8b239b : FAILURE in 43m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18140b67273840afb0a9c0a7bc90325c : FAILURE in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5c226b21f82589dd5b95d9aa08f9470f5d73010a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-19 02:20:47.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/1eef9f38c11f495c98631f1c7f86bbf7 : POST_FAILURE in 2h 18m 22s","accounts_in_message":[],"_revision_number":5},{"id":"2511b8b0960ffb0ce1c7f4af08c350fb9c354d88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-19 12:57:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8001aaaf8fd4be4ceda40c903a3519e4998baf60","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-19 13:01:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"64b99c3c6d1e13f986fa80a458f2a048335ea4c4","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-19 13:02:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8d754e4f4a7de308f2923de206424cff204f87b0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-19 13:51:13.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3be2162caae4414bc95d6b58ffab722c9a306e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:37:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3188c60801c47eca7c80bdcc712d306\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9df2aaeafe44f189ba3de8a04ded08d : SUCCESS in 55m 56s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0af8b8cda72a4d1d9faf2877fb74f451 : SUCCESS in 1h 27m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8861236a692646e38c18852a5d45b76d : SUCCESS in 56m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e3ba9c8b3c9407cb6e26d1714b17670 : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e076181bde2f455d9c4657e4cc6ce580 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dc37cfc78ff4d5b9b3bf9c9a135ed54 : SUCCESS in 11m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ff9a56cf56947098f37df9d435e02e8 : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc76f0e176bc4c44a170e03fc77b8bb5 : SUCCESS in 10m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/28e0c7fbf7be4e4583b761dcc21543c2 : SUCCESS in 59m 43s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfe4d98dab8d4810b477fc754ad29e26 : SUCCESS in 1h 07m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1f300f2472d94cc69f5fcb23b3b7a060 : SUCCESS in 49m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c7f53ad1d8fa4ca291dd367e080f8d0e : SUCCESS in 44m 38s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8c891379850d465e9ebbd4ca298a4273 : SUCCESS in 46m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7fcbf5b84b1043f89ca5198c33ac7b26 : SUCCESS in 1h 24m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/5b89a09505a244a28033e87d3f2fc659 : SUCCESS in 1h 49m 21s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9d96ad3205354dd5868a75323c3bd9ef : SUCCESS in 50m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/379209dcf8c5473bbab8ca663b9c708b : SUCCESS in 4m 42s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/86aa2fbfadec40699f068012576a2d11 : SUCCESS in 21m 24s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62832adda12547bc89ab77efcf90dfed : SUCCESS in 21m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/413aedc804124b35845189a27a50dd17 : SUCCESS in 22m 17s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76cfacd697724bd28e7fb4b6d295af11 : SUCCESS in 1h 45m 44s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7247783629d54849b630fd325125acbe : SUCCESS in 1h 24m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1ba741beae0445480f4bf1f0090397e : SUCCESS in 1h 10m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7866c2a191c9417a9ee238afffba0a14 : SUCCESS in 59m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4b8eb3e312274fdeacae579c8006f098 : SUCCESS in 22m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4c7a8b1a95c04fd2b80cf8b3d19a5b99 : SUCCESS in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"27f5158bc34bdc007ddb13ece01870e086ace9ba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-19 15:44:41.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"71c6cbc95bee262652dbba1a285f29aa7f328087","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 15:55:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69ecefb5781f4468a34c3dbbf0700d62\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a45b96eb99d4dec96d29ce3163b1265 : SUCCESS in 48m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab1b45ed4154494ba005b06c01778c17 : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"be5c5805a595468d27f7227b9e7eec51d4598a7d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-19 19:37:38.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/eb269de21d89446f9d6fe16788cee403 : POST_FAILURE in 2h 00m 40s","accounts_in_message":[],"_revision_number":6},{"id":"6457e1632b315affc60d6897d9a5a5d7e6a5d9df","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-20 10:49:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e378550aef7284cdb8ce74926b43da590ec719b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-20 11:49:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d1d95b9eee295c047f2cf0d3ee0ba844db119bd7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-20 12:36:14.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e651264357477c12843034144c64a14f2f957427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 12:38:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54a41a6eea2f4ed4bcdf5becba68a798\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f62c6586d1e34e608e824f5814948230 : FAILURE in 40m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c208c0d192464df2b570affc9c03aa7a : FAILURE in 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"90c5566a362e2c2ac741ee47da90ff1f201c83ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 13:49:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24c9fecccece4ea89a5bbe948486ec92\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f7e3fea3e33f440bb1da0b20908f6deb : SUCCESS in 47m 49s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5df2503a44204413bb80a4abdaad6de0 : SUCCESS in 1h 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67872a7d945c4c9a85c081cbc1fa8f74 : SUCCESS in 49m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0e150ad862143998856a0268c0b1970 : FAILURE in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4f403bd9734248801291a3c1002d58 : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58826b505cd64f12ab3cc934a72d650f : FAILURE in 11m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cce9314a8d504a61b36ce213ed292cee : FAILURE in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc7a6190b254426ab46eb30f645274a : SUCCESS in 8m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a92d8b856d814cf28cf9e22db4b0ced8 : SUCCESS in 51m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef2b930f0b594444adc24cd8c99baa32 : FAILURE in 55m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d6f4a1b1dcbb420f8a8c87c271e1121b : SUCCESS in 55m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1780dbc3b2ea4bb0a9ca50fff4ae2e60 : SUCCESS in 40m 19s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b93a6ed96a964320be1ab533bb7dcce2 : SUCCESS in 1h 01m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8d2926b65e624d5d96d34be19e79439c : FAILURE in 1h 35m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/d312c9b596954affafb65db4d05faee4 : SUCCESS in 1h 49m 31s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/15b317ea02cf4052adf74cc98d740aff : SUCCESS in 54m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/21520a532f784824beb7d4bed070262e : SUCCESS in 5m 14s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae564daefdf04e2a8f5bf8f967144618 : SUCCESS in 21m 11s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/598c3e1e10764c619c688776d98280ed : SUCCESS in 22m 09s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c88f22d28fdc49a592a279c232396e13 : SUCCESS in 21m 27s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/047951b071cf40d4ac021ab61c155edf : SUCCESS in 1h 31m 45s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/80301fbd1e2c460d85d30e4b3ef33ced : SUCCESS in 1h 27m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd6a60cd6cc14505a985ce5cccb31e0b : SUCCESS in 1h 21m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d315d568c9694858bd63e0d4387aeb50 : SUCCESS in 54m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b315a2abd547404da994a6154ec07f01 : SUCCESS in 30m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/12126db4609642f3b005ad8fdd769591 : SUCCESS in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"db1e2e33ebaaa524666b68709de9296da2940194","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-20 14:17:46.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a9e217530aaca46763329444a8776bc1eb1948d8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-20 15:27:54.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fde77f1d6fc927b0556502d74d5eef09d5c801b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-20 15:38:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b75c4c5216dd685ded312baa1e98ec3cf9c39b45","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-20 15:38:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"27a36f07f152ceb2757f8694e4e5ab07e0f2bfd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 17:10:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ec3058f27a74d8abcb4e807a1ce49be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb4cd6a7ddf446678bbe7e0727b974a8 : FAILURE in 47m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/00a8ea8eeb5f499da1f0b12a66a468c5 : FAILURE in 43m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d2abaf475a0f3b332ddd513119223cb9f0e57b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 17:48:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61865ae165594d57bac74fd6f92e4665\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aaf6cf0b6cf6490282542fd861fbae5b : SUCCESS in 1h 22m 51s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/31703a69ca04453e9322e74f897e0b1d : FAILURE in 2h 03m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/771b0965f24240179b9d2d6252b541c2 : SUCCESS in 1h 01m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22ed87ef25d44b8baea7fc8aca42b3d0 : FAILURE in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b56cb087fc346a7a7cfc6963a2fdac9 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42cccae5e7b64091a9a63ed0f808c980 : FAILURE in 12m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9b3110e684e4c56a2f0cdc8a0dedaca : FAILURE in 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8cfdb96ddb24cf78229abd962c48014 : SUCCESS in 10m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6c32af276bd043bfa07ec499e9b091aa : SUCCESS in 51m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cea46458924c4397b5592439a6e4ca16 : SUCCESS in 1h 06m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e7790f1df2894e45b0cbde74fcbf5a11 : SUCCESS in 45m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4c7f278f62364cc4a7942635d7b8cae7 : SUCCESS in 48m 04s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b5c132d04045424a8fcce9169b70ccec : SUCCESS in 51m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e8b767fc45814ecbb8f46d4450277b31 : SUCCESS in 1h 49m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/c8c46fd9f8c54e5a98eb743617a1197b : SUCCESS in 1h 40m 19s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d207688832e147aba61c0bbcc1568941 : SUCCESS in 48m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b1b19661b7d4428aad7c943722800a07 : SUCCESS in 4m 46s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aea6a13e881c423098089b5eae460dd8 : SUCCESS in 26m 23s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2dde609a8194494cae63517a711bcc56 : SUCCESS in 24m 04s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f62636b5cb6547a3a3c6e2d0592774f8 : SUCCESS in 22m 42s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/441cd96f7c4c4d04b001ab514b7ac71b : SUCCESS in 1h 38m 10s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cc3760b33093401896737f6c15cae4f2 : SUCCESS in 1h 23m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d86f9478beb24634a026593dff3b24b3 : SUCCESS in 1h 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a898aca4fa894d1d831e136d7b6943d4 : SUCCESS in 47m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f1ffbcc9f1584d72b067c31e978b30ba : SUCCESS in 23m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/45bff16e30ab472da332d6486b9e238c : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"84c24b2565b1b646ad1f1a3c088aa6af5d411ed3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-20 21:01:15.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/2be433237e9947abbb2fe7989b8bc837 : POST_FAILURE in 2h 07m 11s","accounts_in_message":[],"_revision_number":8},{"id":"a341f81a2c18076b007b1004359a42728012c08f","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-23 13:23:33.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"175779ee93eeb3d2b18437293268b9636b2bb26e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-23 13:37:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"68c25526c14ed9a3e90cf20cd71914b630097733","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-23 13:45:51.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bbe72d749edef00473752eb26390afb689974650","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-23 14:40:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dfe03dfee59a3204e1d0ac85e30d21e5c3e3b9e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-23 14:43:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ecaa33757c46c38bb105a04e14ea833c5bc2c3a5","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-23 14:44:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d6b9c321a577795cfec869465bbef6d766b02a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:09:21.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/282206b35c6a44e79b87b7228898dcf8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab2d899729ea47aabdac6ac87214e63c : FAILURE in 49m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05a17dfa8a0d486ebf6daa0d13433de7 : FAILURE in 50m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"59c203e7dd5319cf20fb9f9816a59e20304c2f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 16:45:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08cedf46d4ad4314ac08cd296d8017d2\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d9118385d3f243bd90e8159553c2964c : SUCCESS in 1h 19m 01s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ad5ec6224c6440949c5fd9d007508d95 : SUCCESS in 1h 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2f33835b3e34403bc386560125aecde : SUCCESS in 1h 13m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34a054aa8f3e4adeae8c1e2107bbae06 : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96164e92b3ee4fb98c958631a21b12f6 : SUCCESS in 13m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8884be63b884ce5b1b2341e45f91290 : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94f8aa731d9e4423bcc2e0997b42aefd : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520ed45d5d2e42dea46e4737176446c2 : SUCCESS in 13m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/345cd1eca0f74ea8bae6e66f5edda54f : SUCCESS in 1h 00m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44ab498795304d059d82d19086152853 : SUCCESS in 1h 17m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/83002cd06c964a66aab913bd27049da0 : FAILURE in 59m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cf7f31fe6a43444da0179840423f7364 : SUCCESS in 47m 40s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/747efb22e423491484792cf4aee62dab : SUCCESS in 58m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d4ea4bd5cead4eeaa59692d5b127cceb : FAILURE in 1h 56m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/921edba5e17c452bb6b0b7e455e560bd : SUCCESS in 1h 36m 36s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/732df594977c41579daf71c4e10c1088 : SUCCESS in 38m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1c94b7c40e4d4e6c92317c1b5fb6c743 : SUCCESS in 6m 01s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/268ef22166f645dfab893ebd7a7b8283 : SUCCESS in 47m 09s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/49dfc670f72141e8928e3ea42955e637 : SUCCESS in 21m 19s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/622f7ffe470445a79d6154f0c675eca7 : SUCCESS in 22m 20s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b49a0f91c2149f39f3e3271c612f952 : SUCCESS in 1h 15m 44s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/64fbac9e26944e67af7a7b8b11b500a7 : SUCCESS in 1h 28m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e6b2abb86cc47f6ae6f90d962d1afe9 : SUCCESS in 1h 10m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5ce7899eaba42db9f5dc3735f3716c3 : SUCCESS in 40m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e3e7bf231a974b3ba3f2c6ce44f8fb7b : SUCCESS in 23m 15s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f5c189aff3f14677933d16021e3df07a : SUCCESS in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"89fff76d586532bd395fe4f757bf31f8fec4ba05","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-23 19:23:14.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/c64d3b0142254ccf993883548b0199c1 : POST_FAILURE in 2h 30m 49s","accounts_in_message":[],"_revision_number":9},{"id":"16b26fb02319f3d5c39dad5ff49222c334363426","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-23 23:59:23.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a4d17934e8df0d7f05a16b054995fb03592b7914","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-24 10:28:30.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"05b9c4ae28edb24713ba6c5c881adfd644580e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-24 10:28:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9c11116196deeed037af8005baee1f60246e31d8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-24 10:32:22.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a5cfcc0604cab394e8d5187a66b6349d6f00560e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 11:17:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77df0e2a4418401a8c6a5f2b4a22a309\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bff345298f244098e8f1b5d0414a4ab : FAILURE in 45m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/583f9656ee164273ab4dfb57b8e5c26f : FAILURE in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cefa6e0a6da0ca72d0fe0fed81c6ae39a062e0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 12:29:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/900e5e940d4c4a84a9795d33ad4ce6b6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/27e599ea302340e1a98cb9ed76f47304 : SUCCESS in 59m 58s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6f4bb0aa0db5403699f336e48e0b3799 : SUCCESS in 1h 28m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d6e972cd77140adb91f75f2e99fd57d : SUCCESS in 54m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d109e91774c4d83ab93ac40df2e3c9d : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e6664b16c844da92baf7c85014b291 : SUCCESS in 10m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78c74846c6c9449082a29d128b19f127 : SUCCESS in 12m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec7c94831d0449db81fdd2bd5577c1d0 : SUCCESS in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270a178f5f8147169185054119d6605c : SUCCESS in 10m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/445a88436e6340b49f5abc4e99b469ea : SUCCESS in 59m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62a93a925b3244c8850fe87fa9ec2e80 : SUCCESS in 1h 03m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8950227e462d41ac868ef786908d34a3 : SUCCESS in 44m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/338aed407db74147a5a9bd2e0ff5f555 : SUCCESS in 40m 48s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/aa4babad96674347a3def2517656c075 : SUCCESS in 1h 13m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3709509505b94ddba53b76cf252873e4 : SUCCESS in 1h 30m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/870114cfc38447c7b14c072a32793c40 : SUCCESS in 1h 38m 56s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/dd509eef620542c8864780085b7e5eaf : SUCCESS in 41m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/98b64f4f4bba40e88e22b4945464b8bd : SUCCESS in 5m 23s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/246d5a77aa74411da1b6bc9915e91f88 : SUCCESS in 20m 33s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36714803fdd6400489ffba4a47525c0a : SUCCESS in 26m 48s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36a6c2273ab14eefab89ff248151c0ce : SUCCESS in 21m 25s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2477fb11e0f442ab792dd29e3c2ce86 : SUCCESS in 1h 25m 29s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d864872edff1491c9717dd4fe536f342 : SUCCESS in 1h 25m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e24aa5f47b4449e9a2e77dcf1b503269 : SUCCESS in 1h 09m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c2c5f188ad545ce8f06645842064ae0 : SUCCESS in 43m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/612fad91fc2f4f99b63ef98665c7c276 : SUCCESS in 18m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5cfdbe1c223b458b8f4034663e78425d : SUCCESS in 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c75f9cbf32f1d808aa19cc4a1ec0cf5f20715e8f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-24 12:38:16.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"2820ca5680dc054db634a45a15513461eff3c1a7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-24 15:48:13.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/32027445da884a6c94ce3c9bccf3db52 : POST_FAILURE in 2h 52m 53s","accounts_in_message":[],"_revision_number":10},{"id":"37bf79ddff505f764418ba6fea4c395c9ebbdfdd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-24 17:37:24.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"b8fed81bbba73dec5d06ea675908f635f686738f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 17:38:23.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"dde9bd1d2b03feda0063644a57080ac0bc5d0235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 19:50:11.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1efa2408bf7e4037b4b687580ff17b49\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/453f991eeeb34498b132eb84de3301db : TIMED_OUT in 2h 04m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdc590cbb0ed4a68945fdc37387c5b51 : SUCCESS in 51m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed78ba7f16442819f1dcbe39d102f3f : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f745ca3717e04e3392f6c1c518cc2496 : SUCCESS in 11m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e627c602197c46ca9dbf436efb29919e : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e8f4978d944c3faa0d707b92a6e592 : SUCCESS in 9m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a4471a8450cb4012bb6e7a699ef6de9b : SUCCESS in 38m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b5fd257e78944412a5bb63540c1f2dfb : SUCCESS in 49m 25s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/128bc1715a244e5cab2a9d40a0a14727 : SUCCESS in 22m 35s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c1754b986954dcf8c67475e58497ddf : SUCCESS in 22m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2485fd16de4147ae81a1fd5671f1c245 : SUCCESS in 1h 30m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/eda1b59d80d84c4583b804177a4b41b1 : SUCCESS in 2h 00m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0622973112af4bbfb933a8d8418294b1 : SUCCESS in 6m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5a5b8a35dc94796a672dd252b0c46b3 : SUCCESS in 1h 34m 19s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8332ccd950774998bcbb18685bba1b5f : SUCCESS in 1h 18m 52s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/316454feb62643bfb66991f2f7897fb2 : SUCCESS in 1h 28m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c21f641376df4d9b98ef9e207909a327 : SUCCESS in 2h 05m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/158151d8780d43f5960ca13e9c8a5b0f : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":10},{"id":"078c9f19a6512c0dcf1cf8f98711cfb7c8e56598","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-01-26 09:36:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0f1ae6980d1109738dbbd281fe855f6a72664e24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 09:53:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cfe569b94554101887d64269409e403\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/456f1d2db997444299db1301299cd965 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef564192d0664ef49d5ebee4f9a0b857 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5bac2c23f11a405316fc6a8978698ab3c34c5fe8","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2023-01-26 10:52:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0a8b67392a5057ee659fe1f4636356e27c198ce1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:06:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4712943b65461ea0fe836f4e255cbe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcbcfa0239474bb9af605be582a1503f : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be1a36421c8b4df5bdd950211742eb6f : SUCCESS in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2e873339b515fa767db98e4b98b3fcd34ad9a76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:30:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbc005cac14478eb334e8cd6c891ace\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/274d93c478d448a3b7cb77f75ff09b1e : SUCCESS in 1h 02m 54s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d74e7f9131e043129536264ebf097ebe : SUCCESS in 1h 47m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/884f2336c95c422c8866f23431e581dc : SUCCESS in 56m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a95ee6316ff44c6abde6d0488335892a : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7226971022074f8095d02222623b37df : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e559bf5733be4a1d96b5cea816ffaeb3 : SUCCESS in 12m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa099aed90824cbab030b9fadf40f92a : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4bcb41331541428d65561a95b8b546 : SUCCESS in 10m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c85ef118a20405c947cd704d8ae1042 : SUCCESS in 42m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2ef47930b66c41adba7d0b4d3b402ba5 : SUCCESS in 1h 03m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/94dab39a0b3146b3ba464ab02e821628 : SUCCESS in 50m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/98e78dcb11304414940a9bf3e8dfd795 : SUCCESS in 48m 23s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/81cca70f20644e54a12cebc0db513926 : SUCCESS in 52m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/91c8ef9471fd4ff79aed52856e518857 : SUCCESS in 1h 29m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/16ee776b0d6c49f29408a8b86e63bffc : SUCCESS in 1h 45m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8d792fb306244d8ebbaefdf85cb9cc29 : SUCCESS in 43m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/19b290ea541541069fa83db0edb0f772 : SUCCESS in 4m 50s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39c12e71076b4a87913b4efff9940eee : SUCCESS in 23m 08s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c2f1624af5a40ae99b574fb684adb67 : SUCCESS in 24m 10s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be98dd7c2d274282b8e68ba017ac8dd3 : SUCCESS in 23m 06s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7bbe402b37540e4a3017f8018a63ef8 : SUCCESS in 1h 24m 03s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f7416aa5e9248c5abfc23f9db9ef9e3 : SUCCESS in 1h 20m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f17ec41c952341cc89583fa032290b22 : SUCCESS in 1h 18m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60e46d68cd7f4f729b2016d5597b904a : SUCCESS in 46m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d6ebb03975d94d8297dce118c0636254 : SUCCESS in 18m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/17371e714cb140439ccff473534b1976 : SUCCESS in 32m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"031ce5bf3659d10214aa80659cfd42dcd1a6d710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:31:12.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"9a3493e1bb9aa821015d2882310628696c743adc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-26 11:37:46.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"f00196412c59af2e318c64826c2d60d1e7c425a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 13:48:01.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0504698e904b2eb53aec213c10fe88\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/780203d05e05492e8db90a7b9e73d632 : SUCCESS in 1h 25m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc80e245d0954327929349ec6ce0bd22 : SUCCESS in 1h 03m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c14f334b5d40e69f9d898b32d6a884 : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/903506f67e5f43b984dc60cdf80bc3d9 : SUCCESS in 12m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eecf65247d9c465b92e0ab2482c2f971 : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3fd34f99f740af827cc5c41ef442e8 : SUCCESS in 10m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/67638fdd2d5741129d3f49aef83407b9 : SUCCESS in 45m 32s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8d0ea4714fcd4511ba56f3bed95084b8 : SUCCESS in 37m 35s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f7e4ec9ec7ca4afeb458d91ade94d38b : SUCCESS in 25m 14s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b55753e365bf4b5c9686c40f60a958c1 : SUCCESS in 25m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2af662d587d94f309e7ac609f3c8fb95 : SUCCESS in 1h 16m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/e41575a34e6545e39483420081090d99 : SUCCESS in 1h 51m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c7612b89d34941b3bf95596d8d8779a9 : SUCCESS in 5m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d428019ff890483d807c57f5de1b7e6d : SUCCESS in 1h 12m 29s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b3b32ae71164b5cbc8c0c4822f8f462 : SUCCESS in 1h 34m 51s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b04d574878ee4336b28824813b5c6902 : SUCCESS in 1h 42m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae27c1f231c34887aa5fe38dc6cee641 : SUCCESS in 1h 20m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/effb27ce86da4ed7b0a613289006f523 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":10},{"id":"0d333c95b5c45c42dc9184fa91595166d0fc7e94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 13:48:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"6079da2cd84a67be8e0f3a43af78b2dbf45c6d73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 13:49:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb60ad267ec740879d7109840d92ca9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d9ff55623a425aa641e7b35aaa2d74 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":10},{"id":"730028725c64ecc9507743ce180216443131523a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-26 17:01:24.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/abcbf099f9ce4ea1ac8336acbde552ea : POST_FAILURE in 2h 22m 39s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"9caaaf1f221063a4329c72c8b67a6015648644a2","revisions":{"455186e679b456806b75ba7f74179b53c7a9c0b8":{"kind":"REWORK","_number":1,"created":"2023-01-17 11:19:55.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/1"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 11:19:09.000000000","tz":60},"subject":"libvirt: At start-up skip compareCPU() with a workaround","message":"libvirt: At start-up skip compareCPU() with a workaround\n\nThis will skip libvirt\u0027s compareCPU() call in\n_check_cpu_compatibility() and libvirt handle it more correctly.\n\n(QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the mininum required versions, and\nupstream Nova satisfies them by a good margin.)\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/455186e679b456806b75ba7f74179b53c7a9c0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/455186e679b456806b75ba7f74179b53c7a9c0b8"}]},"branch":"refs/heads/master"},"fa4104e56b7f0b723f7f9a9a86456a6bbdc56b85":{"kind":"REWORK","_number":2,"created":"2023-01-17 14:58:16.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/2"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 14:43:23.000000000","tz":60},"subject":"libvirt: At start-up skip compareCPU() with a workaround","message":"libvirt: At start-up skip compareCPU() with a workaround\n\nThis workaround will skip libvirt\u0027s compareCPU() call in\n_check_cpu_compatibility() and allow libvirt to handle it more\ncorrectly.  This will avoid certain classes of bugs.  (QEMU \u003e\u003d2.9 and\nlibvirt \u003e\u003d 4.4.0 are the mininum required versions, and upstream Nova\nsatisfies them by a good margin.)\n\nIn the \"near future\", we should remove this check altogether.  But\nbefore we bake that in, let\u0027s put this in a workaround to reduce\nunintended impact elsewhere.  Further we should also introduce the usage\nof replacing the older libvirt CPU API with the newer one[1].\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa4104e56b7f0b723f7f9a9a86456a6bbdc56b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa4104e56b7f0b723f7f9a9a86456a6bbdc56b85"}]},"branch":"refs/heads/master"},"6885b0ac7edef66b9dd35b6dec32971f94a7adaf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-17 15:37:59.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/3"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 15:37:43.000000000","tz":60},"subject":"libvirt: At start-up allow skiping compareCPU() with a workaround","message":"libvirt: At start-up allow skiping compareCPU() with a workaround\n\nThis workaround will skip libvirt\u0027s compareCPU() call in\n_check_cpu_compatibility() and allow libvirt to handle it more\ncorrectly.  This will avoid certain classes of bugs.  (QEMU \u003e\u003d2.9 and\nlibvirt \u003e\u003d 4.4.0 are the mininum required versions, and upstream Nova\nsatisfies them by a good margin.)\n\nIn the \"near future\", we should remove this check altogether.  But\nbefore we bake that in, let\u0027s put this in a workaround to reduce\nunintended impact elsewhere.  Further we should also introduce the usage\nof replacing the older libvirt CPU API with the newer one[1].\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6885b0ac7edef66b9dd35b6dec32971f94a7adaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6885b0ac7edef66b9dd35b6dec32971f94a7adaf"}]},"branch":"refs/heads/master"},"fc6a865dcbf2aa72060fc71c0caec28950a3cb88":{"kind":"REWORK","_number":4,"created":"2023-01-18 16:40:18.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/4"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-18 16:40:12.000000000","tz":60},"subject":"libvirt: At start-up allow skiping compareCPU() with a workaround","message":"libvirt: At start-up allow skiping compareCPU() with a workaround\n\nThis workaround allows to skip the *first* call to libvirt\u0027s\ncompareCPU() in _check_cpu_compatibility() and lets libvirt handle it\nmore correctly.  This is because the first call does not evaluate the\nextra CPU flags provided by the user into account.  This will avoid\ncertain classes of bugs.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the\nmininum required versions, and upstream Nova satisfies them by a good\nmargin.)\n\nIn the \"near future\", we should consider removing this compareCPU() call\n(and subsequent calls elsewhere) altogether.  This workaround is a small\ntesting ground before we bake that in.\n\nAs a follow up, we should proceed with the patch[1] that replaces the\nolder libvirt CPU API with the newer one[1].\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc6a865dcbf2aa72060fc71c0caec28950a3cb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc6a865dcbf2aa72060fc71c0caec28950a3cb88"}]},"branch":"refs/heads/master"},"59b9d151eb266f7f5eaaabee3282a8f33eeb0a59":{"kind":"REWORK","_number":5,"created":"2023-01-18 16:41:19.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/5"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-18 16:41:11.000000000","tz":60},"subject":"libvirt: At start-up allow skiping compareCPU() with a workaround","message":"libvirt: At start-up allow skiping compareCPU() with a workaround\n\nThis workaround allows to skip the *first* call to libvirt\u0027s\ncompareCPU() in _check_cpu_compatibility() and lets libvirt handle it\nmore correctly.  This is because the first call does not evaluate the\nextra CPU flags provided by the user into account.  This will avoid\ncertain classes of bugs.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the\nmininum required versions, and upstream Nova satisfies them by a good\nmargin.)\n\nIn the \"near future\", we should consider removing this compareCPU() call\n(and subsequent calls elsewhere) altogether.  This workaround is a small\ntesting ground before we bake that in.\n\nAs a follow up, we should proceed with the patch[1] that replaces the\nolder libvirt CPU API with the newer one[1].\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59b9d151eb266f7f5eaaabee3282a8f33eeb0a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59b9d151eb266f7f5eaaabee3282a8f33eeb0a59"}]},"branch":"refs/heads/master"},"6fe080c7fb00e3ebfb96f2e60bc3e11ecfea853a":{"kind":"REWORK","_number":6,"created":"2023-01-19 12:57:01.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/6"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-19 12:52:06.000000000","tz":60},"subject":"libvirt: At start-up allow skiping compareCPU() with a workaround","message":"libvirt: At start-up allow skiping compareCPU() with a workaround\n\nThis workaround allows to skip the *first* call to libvirt\u0027s\ncompareCPU() in _check_cpu_compatibility() and lets libvirt handle it\nmore correctly.  This is because the first call does not evaluate the\nextra CPU flags provided by the user into account.  This will avoid\ncertain classes of bugs.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the\nmininum required versions, and upstream Nova satisfies them by a good\nmargin.)\n\nIn the \"near future\", we should consider removing this compareCPU() call\n(and subsequent calls elsewhere) altogether.  This workaround is a small\ntesting ground before we bake that in.\n\nAs a follow up, we should proceed with the patch[1] that replaces the\nolder libvirt CPU API with the newer one[1].\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fe080c7fb00e3ebfb96f2e60bc3e11ecfea853a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6fe080c7fb00e3ebfb96f2e60bc3e11ecfea853a"}]},"branch":"refs/heads/master"},"cd0eccfff4f1ae30c67e8bbd06c798d152267c36":{"kind":"REWORK","_number":7,"created":"2023-01-20 11:49:20.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/7"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-20 11:48:44.000000000","tz":60},"subject":"libvirt: At start-up rework compareCPU() usage with a workaround","message":"libvirt: At start-up rework compareCPU() usage with a workaround\n\nIn this patch:\n\n  - Remove the first compareCPU() call (called via the wrapper\n    _compare_cpu()) in _check_cpu_compatibility(), and let libvirt\n    handle it.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the mininum\n    required versions, and upstream Nova satisfies them by a good\n    margin.)\n\n  - Add a workaround to allow skipping the remaining compareCPU() call\n    in _check_cpu_compatibility() as a potential future-proof (because\n    we cannot test all possible CPU models and hardware).  Unlike the\n    removed first call, this call takes into account the extra CPU flags\n    provided by the user into account when evaluating guest CPU model\n    compatibility.\n\nAs a follow up comes the patch[1] that replaces the older libvirt CPU\nAPI with the newer one.\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd0eccfff4f1ae30c67e8bbd06c798d152267c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd0eccfff4f1ae30c67e8bbd06c798d152267c36"}]},"branch":"refs/heads/master"},"01389198cdedf66dc2de0e30831c58f0d8098c7d":{"kind":"REWORK","_number":8,"created":"2023-01-20 15:38:08.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/8"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-20 15:32:30.000000000","tz":60},"subject":"libvirt: At start-up rework compareCPU() usage with a workaround","message":"libvirt: At start-up rework compareCPU() usage with a workaround\n\nIn this patch:\n\n  - Remove the first compareCPU() call (called via the wrapper\n    _compare_cpu()) in _check_cpu_compatibility(), and let libvirt\n    handle it.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the mininum\n    required versions, and upstream Nova satisfies them by a good\n    margin.)\n\n  - Add a workaround to allow skipping the remaining compareCPU() call\n    in _check_cpu_compatibility() as a potential future-proof (because\n    we cannot test all possible CPU models and hardware).  Unlike the\n    removed first call, this call takes into account the extra CPU flags\n    provided by the user into account when evaluating guest CPU model\n    compatibility.\n\nAs a follow up comes the patch[1] that replaces the older libvirt CPU\nAPI with the newer one.\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01389198cdedf66dc2de0e30831c58f0d8098c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01389198cdedf66dc2de0e30831c58f0d8098c7d"}]},"branch":"refs/heads/master"},"93847aaf3a5bb64bf3d15423432fd3d9b80e80c3":{"kind":"REWORK","_number":9,"created":"2023-01-23 14:43:07.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/9"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-23 14:41:37.000000000","tz":60},"subject":"libvirt: At start-up rework compareCPU() usage with a workaround","message":"libvirt: At start-up rework compareCPU() usage with a workaround\n\nIn this patch:\n\n  - Remove the first compareCPU() call (called via the wrapper\n    _compare_cpu()) in _check_cpu_compatibility(), and let libvirt\n    handle it.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the mininum\n    required versions, and upstream Nova satisfies them by a good\n    margin.)\n\n  - Validate the user-configured CPU models from\n    _get_cpu_model_mapping().  And take into account all the CPU flags\n    before calling _compare_cpu().\n\n    (Suggested-by: Sean Mooney -- thanks!)\n\n  - Add a workaround to allow skipping the remaining compareCPU() call\n    in _check_cpu_compatibility() as a potential future-proof (because\n    we cannot test all possible CPU models and hardware).  Unlike the\n    removed first call, this call takes into account the extra CPU flags\n    provided by the user into account when evaluating guest CPU model\n    compatibility.\n\nAs a follow up comes the patch[1] that replaces the older libvirt CPU\nAPI with the newer one.\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93847aaf3a5bb64bf3d15423432fd3d9b80e80c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93847aaf3a5bb64bf3d15423432fd3d9b80e80c3"}]},"branch":"refs/heads/master"},"9caaaf1f221063a4329c72c8b67a6015648644a2":{"kind":"REWORK","_number":10,"created":"2023-01-24 10:28:51.000000000","uploader":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"ref":"refs/changes/94/870794/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/94/870794/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/94/870794/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/94/870794/10"}}},"commit":{"parents":[{"commit":"ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6","subject":"Merge \"Support cold migrate and resize with PCI tracking in placement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ba9d4c909beff4e9ab86911a35dd5db8d8ce08d6"}]}],"author":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-17 10:15:37.000000000","tz":60},"committer":{"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","date":"2023-01-24 10:26:56.000000000","tz":60},"subject":"libvirt: At start-up rework compareCPU() usage with a workaround","message":"libvirt: At start-up rework compareCPU() usage with a workaround\n\nIn this patch:\n\n  - Remove the first compareCPU() call (called via the wrapper\n    _compare_cpu()) in _check_cpu_compatibility(), and let libvirt\n    handle it.  (QEMU \u003e\u003d2.9 and libvirt \u003e\u003d 4.4.0 are the mininum\n    required versions, and upstream Nova satisfies them by a good\n    margin.)\n\n  - Validate the user-configured CPU models from\n    _get_cpu_model_mapping().  And take into account all the CPU flags\n    before calling _compare_cpu().\n\n    (Suggested-by: Sean Mooney -- thanks!)\n\n  - Add a workaround to allow skipping the remaining compareCPU() call\n    in _check_cpu_compatibility() as a potential future-proof (because\n    we cannot test all possible CPU models and hardware).  Unlike the\n    removed first call, this call takes into account the extra CPU flags\n    provided by the user into account when evaluating guest CPU model\n    compatibility.\n\nAs a follow up comes the patch[1] that replaces the older libvirt CPU\nAPI with the newer one.\n\n[1] https://review.opendev.org/c/openstack/nova/+/869950 -- libvirt:\n    Replace usage of compareCPU() with compareHypervisorCPU()\n\nChange-Id: I8ef9db851b37c5249d2efbe09a15a1ddbae8205d\nSigned-off-by: Kashyap Chamarthy \u003ckchamart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9caaaf1f221063a4329c72c8b67a6015648644a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9caaaf1f221063a4329c72c8b67a6015648644a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
