)]}'
{"id":"openstack%2Fnova~871800","triplet_id":"openstack%2Fnova~stable%2Fxena~If01acbb8a110b8580ad0701ce92abb7491020382","project":"openstack/nova","branch":"stable/xena","topic":"wallaby-pin-tempest","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-01-29 02:57:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If01acbb8a110b8580ad0701ce92abb7491020382","subject":"DNM: testing tempest pin for stable/wallaby","status":"ABANDONED","created":"2023-01-25 22:00:00.000000000","updated":"2023-01-29 02:57:01.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8857530e7cb12559422c8267a7be8d0a706826d9","_number":871800,"virtual_id_number":871800,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-27 08:29:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-25 22:26:55.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 22:35:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-25 23:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd020e4aba3e87431188b039cbc4133bb6e905a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-25 22:00:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55f821e2909e22f96a5e5ef2559d9d7378879078","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-25 22:26:55.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"4f16b99d0381801bf82cab496b5f54367e33a81d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 22:35:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9acf7693b82044cdad8318b4d0bca179\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83600ba4be7646388dd8ddc50a53863b : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5f4709c73b147949d3771128269ac46 : SUCCESS in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93506ea6346379d9183884243e17d54c69f944af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 23:38:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42959a0cf3404153b411619e33663ab9\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/71512898fa154b178110d9a129ed1f89 : SUCCESS in 1h 04m 28s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4b975d07c6e4ac4beaf386a06a650d7 : FAILURE in 38m 52s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cac88d0620d94a1186673b0b2815ba7e : SUCCESS in 51m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be6d78b01f9645b6b4f46613164e22b8 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/befc762aec674890bc773ca623c9a830 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ae71525b734ad8906f3f60da6bbad1 : SUCCESS in 12m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb6ffbea97bf4643bf412e73312c7f79 : SUCCESS in 11m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e461a4285114d36a0098f24770fd5db : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00315c64dd6342978601eceb1d828366 : SUCCESS in 8m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/07b2c1fe80824ac8a056dc5e9205ca11 : SUCCESS in 1h 05m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c6d724ab94b44e9597f773a6f8b46026 : SUCCESS in 1h 31m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/253d471d6dc148139b89bfdae96a3de6 : SUCCESS in 50m 03s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ed155b3c15444d1fb2c2a21c1ecc5a81 : SUCCESS in 53m 34s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c4976e4860c54fe2890bb222bcb47efe : SUCCESS in 1h 04m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a42808aecef4477e92328f9c711551e9 : SUCCESS in 1h 29m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/c6539799864f495487654aa0dcec65a0 : SUCCESS in 1h 27m 51s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e87fd6a508a740df8990f785572fead0 : SUCCESS in 56m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/05685b48ded54f8eaef3385c184cb1a4 : FAILURE in 5m 49s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/076dc7632afe4bc7ad614b1458c23f9a : SUCCESS in 22m 10s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16452792238244cba34cce1d4f01f971 : SUCCESS in 21m 02s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/caa2da873d7d418c9cd7173fb75544df : SUCCESS in 21m 25s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b075ae3b67864ad6b6158a71df574d1d : FAILURE in 1h 14m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/356885223f7b4601bcd2c2f2f676863c : SUCCESS in 55m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a4bf6544dd034f5fbc09d5633cffcb7d : SUCCESS in 37m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cc410c3019fe4dadbd6e11782980bc03 : FAILURE in 40m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"152bf56f312765f4e4d38872b4177f344dc631af","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-25 23:50:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7943de3c6c071ca2af703528e2e9cfda56cc00bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 00:20:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34138208ac6a45fbb415b0d943a9d8a1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a186a9b71cce4567861a65609d088672 : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8510ebf6c530407a8f56e90aea152df9 : SUCCESS in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2f9c630f0914df7aaef99aaaf9c234a5984977a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 01:48:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b16a40fa04054bb3842ddf15762f917e\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/701e10112fa14c6dad8a194d8e76f364 : SUCCESS in 1h 36m 29s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef9054e7c3f64d1cba87701c9a49ab07 : FAILURE in 38m 20s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/363218160d594a01bfbd59a059a899cb : SUCCESS in 48m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80dd2fd096e04f65b6925bb4ff5b1eda : SUCCESS in 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d967e7fb39479cb4fb7e4c405f6442 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c0d3344e5c4676871cd9b0cc7ea7ca : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b14bdb3ee249ee968fb0be87d352d9 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19548eb924b44e29b4e9403de43cae82 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4993e96ee24eeaa9b6d1fd381a2819 : SUCCESS in 9m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d2b06726637495e81a07fa0884cc111 : SUCCESS in 1h 02m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70719f9e24ad4203a0be4ad953997f2a : SUCCESS in 1h 06m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/afd6a0eee01b4b6e882960ee32989c17 : SUCCESS in 56m 51s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/64c91bab646142859b72f2b68bd95203 : SUCCESS in 47m 15s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5f7a91f4d39f4050b3278e6037de2bb0 : SUCCESS in 1h 07m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e8fc418ca2ca45a6b4ac0d16bffa8f10 : SUCCESS in 1h 37m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/9519fcabd81f473ba6eb7ac28385d1fe : SUCCESS in 1h 26m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4b8415e2ef3c4ad789f7859de3928271 : SUCCESS in 56m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/016817c6981841ef820d8a8925bf74a4 : FAILURE in 5m 42s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/db236bb191834468a7adcda8e082b710 : SUCCESS in 22m 47s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/03207ada15664af4b8470ab023e9497d : SUCCESS in 19m 53s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4cef8adde17f4a03abbe49630d924b4d : SUCCESS in 22m 03s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/60090218f202418cb1b0873ebc9f6651 : FAILURE in 1h 09m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/974fa04cfb214921a8f71fce60d3671e : SUCCESS in 55m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dc386ba8ad08491c891086ccaf329ca0 : SUCCESS in 35m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ceda6e6b0c1f40a3af04f4e746c7be0a : FAILURE in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb6fd5fa46d329e0acea75ac807c8d83f042206c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-26 14:00:35.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/7ef3aa4362e347e4a91d096766a0a19d : POST_FAILURE in 1h 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d4778545d33f2c6c94492032239dddc04b8682ce","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 00:38:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff7b351ec262e863ed460c9fbe2ee38a849ffb38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 00:52:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db07cec4a0d4cdcbe42741ab42eb6ec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd4ecf4d76e047669a669015985cf410 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/286d00f608a64e75a9e26b5cdce36ff9 : SUCCESS in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0b51a12ecc4df2395cda640f60947ab85f9709c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 02:23:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57952770421b4317afa10b68baee61fc\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4c6a7f7617e848558ed5b84b4628488c : SUCCESS in 1h 19m 58s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/be2cfad2be20459e99cf41a74b862671 : FAILURE in 36m 21s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efc9c35f9f0444569f6baf9ecb55ffa3 : SUCCESS in 57m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/746bc8575f814ae19fe7db3e705f8eee : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c66c797e5714b6f9a99482465a3423a : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36495b8576c449988f45e08ea4c91665 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e337264878d5485d926abbba822b79b9 : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adbc497948e7492aaf34bd9f91cfa9ce : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c1860d864b4f2aaa61ec447dccd716 : SUCCESS in 7m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6cf6d2fef2054ae0a1657eb0264344ab : SUCCESS in 1h 00m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5aabe58d4bdd4085875d55d68de874c6 : SUCCESS in 1h 09m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/01d5a59cf4a3475e8cc11db6d634d394 : SUCCESS in 46m 31s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/65240d2cf5d94d45b9e1d11bb50e22fa : SUCCESS in 42m 40s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/29bf34930590454e8820f412bf01c23e : SUCCESS in 45m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/97eede30254c49488d9d4d96af2dd90c : SUCCESS in 1h 16m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/46b65feb866b4527920db8e096839dd1 : SUCCESS in 1h 28m 20s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8e04cbce5be347869c7aa11962d85bed : SUCCESS in 45m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7b6555671641487f9d316488e9cbf28d : FAILURE in 5m 38s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/bef52b8d2668493f8ac95f42b5b499fe : SUCCESS in 26m 16s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6bd469e18bd44f4b0a4ea0601c81adf : SUCCESS in 17m 32s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07a8d561e939484085ff1b16d8ff8a32 : SUCCESS in 21m 25s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb7b2a8b562c42bab4c741819f5e9732 : FAILURE in 54m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a36ae5cabaf430d90fb0631156754ca : SUCCESS in 32m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc01b6ee361f4b0592343800f5e1ca67 : SUCCESS in 34m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/df810a484ea342c1b25b59e634311574 : FAILURE in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f372afc50da94b2c748bdc35f2440ff281aa1c3e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 04:16:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2235d98d4a98b68e506268e5d069466d3618932e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 04:39:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f310069bdf8fa052eafca2845a1431e0d99a4400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 04:41:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e266499c2be04d667771213faa9b68ba83328158","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 05:36:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296bb1e22416446a93ce5861efb2bdc0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b27474489e254d528aeb4f72c57794d6 : SUCCESS in 52m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daf59b71d8594b91a2b55bbf2400c0d0 : SUCCESS in 45m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ee48211a9323c276a6382f971313021c2cb3e7f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 05:56:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b96b52cf8e39624a9073c283d602115e2b32441c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 05:58:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c0ce4a0549a2a82aff4493cd01090533ff946bf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:04:19.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871782 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"b07b79dd7d965e570be8e1a471183db7855de711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:04:29.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871782 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"414246d983e3e6b1c0b7f80f39c50455c4998e74","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 06:07:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"275526fca35e831ddbd80553a9c91ff94189c8ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:38:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0620721c4c6f43dc907f930db633b121\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc7fd808a6a949418fc761cdf7195479 : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc31480702274176a8fb168ebc7fb982 : SUCCESS in 13m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b479a12583e3436b2f280ee0c1f622ee190a4585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:54:05.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871949 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"578740c549ea7acfff0225aa45a6cc14204b2334","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-27 06:54:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"afa856a123749cf135d706cff60a00a270369305","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 07:07:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbdb0cd1869649f8813a9cd4c46f412f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3880bfdd5535473f89c629608ca823a1 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4a6c6590cfd4799bb9b10a3ae3e47a6 : SUCCESS in 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ccf7ecdec5be9b7392d18be921bd78b10120a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 08:29:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54cdd1f651314967bef0616e311f96a4\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c8f1832a1c6445ad954efd4ae08bbda8 : SUCCESS in 1h 22m 03s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/676e34c3d23e464d8d7fa62f6c298283 : FAILURE in 38m 42s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c49334a0b0b24d6193b37ff45b68f13e : SUCCESS in 1h 01m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b58d3a800c9d411abe7e576e0a117803 : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/423697cb5bde47d59d7be53a4dbd284a : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed55c16371d64834872293739c4d6516 : SUCCESS in 13m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da048d8a9b01436cb5a506bb25718959 : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2231059b5586464c96a102f433980ef8 : SUCCESS in 11m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23479127ce9b46a4b90017cd0cf7dcb5 : SUCCESS in 9m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dc62dc5b537b4973be7bdc6671ec3536 : SUCCESS in 1h 06m 29s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/916920abb6d44096b761f788bef59454 : SUCCESS in 55m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d454d87f9ba24d06b59593f643d4e76c : SUCCESS in 36m 50s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/70832ac5ecf54fa2aba24a5054572e0e : SUCCESS in 50m 57s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/305ad103d35d4cd7a00c5424ba261863 : SUCCESS in 1h 10m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c5cd90cdb9da463583f3e8dc87a663da : SUCCESS in 1h 22m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/53a696276a1c43cfb502767f229b80da : SUCCESS in 1h 29m 32s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4176a7c3248441ffb1138358d8ba5877 : SUCCESS in 57m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/71badb3e1f5d4d7ea2542aad9c75f6b7 : FAILURE in 6m 51s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/cd56bc06a51443e5913b807fc714558c : SUCCESS in 21m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2666a54a3cc148c4bce6ce8ffab392d7 : SUCCESS in 19m 48s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e0a94960f5f4707915fdf668ded0846 : SUCCESS in 22m 25s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f1a1416979844baaa8e9ff7da9fe3e2 : SUCCESS in 1h 17m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e3e89d456dd40dc8fb86a11b7bc2285 : SUCCESS in 1h 01m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b0854201bd5b4bdc9b27175b8a72479b : SUCCESS in 37m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a62d4afbe6a24d718c1d8519ad3449c8 : FAILURE in 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8857530e7cb12559422c8267a7be8d0a706826d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-29 02:57:01.000000000","message":"Abandoned\n\nall good here after depends-on merge","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef77d3f2476ef6106857d3624dfacf0bdf38a9ac","revisions":{"1fb17a6f49153a63f6ffe662a1cc8b40e81e7774":{"kind":"REWORK","_number":1,"created":"2023-01-25 22:00:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/871800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/871800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/871800/1"}}},"commit":{"parents":[{"commit":"ac3728e1239fe2aeffc2a06bbeab6a24eff28ae9","subject":"Merge \"Gracefully ERROR in _init_instance if vnic_type changed\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac3728e1239fe2aeffc2a06bbeab6a24eff28ae9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-25 21:59:29.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-25 21:59:29.000000000","tz":-360},"subject":"DNM: testing tempest pin for stable/wallaby","message":"DNM: testing tempest pin for stable/wallaby\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nChange-Id: If01acbb8a110b8580ad0701ce92abb7491020382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fb17a6f49153a63f6ffe662a1cc8b40e81e7774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fb17a6f49153a63f6ffe662a1cc8b40e81e7774"}]},"branch":"refs/heads/stable/xena"},"40b2fb066c512faff61827a2cc1edfdd67ac9fd1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-27 04:41:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/871800/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/871800/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/871800/2"}}},"commit":{"parents":[{"commit":"ac3728e1239fe2aeffc2a06bbeab6a24eff28ae9","subject":"Merge \"Gracefully ERROR in _init_instance if vnic_type changed\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac3728e1239fe2aeffc2a06bbeab6a24eff28ae9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-25 21:59:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-01-27 04:41:23.000000000","tz":0},"subject":"DNM: testing tempest pin for stable/wallaby","message":"DNM: testing tempest pin for stable/wallaby\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871945\n\nChange-Id: If01acbb8a110b8580ad0701ce92abb7491020382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40b2fb066c512faff61827a2cc1edfdd67ac9fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40b2fb066c512faff61827a2cc1edfdd67ac9fd1"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"ef77d3f2476ef6106857d3624dfacf0bdf38a9ac":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-27 05:58:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/00/871800/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/871800/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/871800/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/871800/3"}}},"commit":{"parents":[{"commit":"ac3728e1239fe2aeffc2a06bbeab6a24eff28ae9","subject":"Merge \"Gracefully ERROR in _init_instance if vnic_type changed\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac3728e1239fe2aeffc2a06bbeab6a24eff28ae9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-25 21:59:29.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-01-27 05:58:00.000000000","tz":0},"subject":"DNM: testing tempest pin for stable/wallaby","message":"DNM: testing tempest pin for stable/wallaby\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871782\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/871945\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/871949\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/871955\n\nChange-Id: If01acbb8a110b8580ad0701ce92abb7491020382\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef77d3f2476ef6106857d3624dfacf0bdf38a9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef77d3f2476ef6106857d3624dfacf0bdf38a9ac"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
