)]}'
{"id":"openstack%2Fnova~872116","triplet_id":"openstack%2Fnova~stable%2Fzed~Id4c8c5f3b32985ac7d3d7c833b82e0876f7367c1","project":"openstack/nova","branch":"stable/zed","topic":"lp_1978958","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-04-13 20:22:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-04-14 09:34:00.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-04-14 09:34:00.000000000","reason":"Change was submitted"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2023-04-14 09:34:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id4c8c5f3b32985ac7d3d7c833b82e0876f7367c1","subject":"Fix rescue volume-based instance","status":"MERGED","created":"2023-01-30 12:46:49.000000000","updated":"2023-04-14 09:35:15.000000000","submitted":"2023-04-14 09:34:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872116-lp_1978958","cherry_pick_of_change":852737,"cherry_pick_of_patch_set":10,"meta_rev_id":"a2c0abefcf6222f97d85de90101fbacc19d07541","_number":872116,"virtual_id_number":872116,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 09:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2023-04-13 20:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-04-13 14:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-22 11:09:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-04-13 20:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 13:26:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-30 14:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 14:53:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-01-31 12:05:21.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-01-31 12:05:21.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-02-20 16:21:04.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2023-02-27 12:23:53.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-03-22 11:09:18.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2023-04-13 14:49:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"2b894d54374eacadad0f25c7aad8c23c6f1ce4f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-30 12:46:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da299513ab2d73fb9ed54c792e64a9079c128c61","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-30 12:47:27.000000000","message":"Topic lp_1978958-stable/zed removed","accounts_in_message":[],"_revision_number":1},{"id":"76f20beb156d30871f12ed8a348333a68e05a50b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-30 12:47:33.000000000","message":"Topic set to lp_1978958","accounts_in_message":[],"_revision_number":1},{"id":"e04b252600be755518eb295282c35a38cadbaf51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 13:26:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e4359eddc84448b73ff4a7b3c0bfd3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75d3088c91a742b19752cf9fc60f1cb3 : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee43b69ace0a4eef82d874317a4644ff : SUCCESS in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"849a5ca719f923d1b525cd08afdac665a9fbdb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 14:46:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f056c2e797d443087957718d8859eaa\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/867f6046be5349f2bcdb8e2698e0f398 : SUCCESS in 1h 31m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06c5cc5779f44242898f5c73df3db520 : SUCCESS in 59m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13de51612947447aba711b9257addc7a : SUCCESS in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0fdeda058e24d908ed97adeb41cf859 : SUCCESS in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ca77f1923945849c353fe8a10f16ef : SUCCESS in 13m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4f2143bc5154465880fe55f44111458 : SUCCESS in 13m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/399c5718783d4f499f9b032dd127b650 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee785e2b4ff4b26945d62df59037be5 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e275773e265b4dfe9a31a6e625ef2bae : SUCCESS in 21m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/92ce45ac974b4a339d4801498b30dfd0 : SUCCESS in 1h 01m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/256f724b580c481cac9ddb9f30a78482 : SUCCESS in 1h 14m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5de74b5147d44345a19c3b8535c1e2e8 : SUCCESS in 47m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f0152fbed2514876922a35ec6cfbde35 : SUCCESS in 36m 42s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7f59d9d2730f4236befd1eb8c9a504ca : SUCCESS in 1h 41m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/a50f6cbb00f3400e9f6e5dc696161152 : SUCCESS in 1h 34m 39s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/110a3ffbdea54ceeaf3a362effe0ca05 : SUCCESS in 1h 03m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bf12702beeb346cebf52d3bd491efdb1 : SUCCESS in 6m 25s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e1d6571da99458ca6f985ddb935286b : SUCCESS in 21m 21s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11d6e5c8c6044bfb9aec3adb5343ef0c : SUCCESS in 22m 23s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f90fa020af2432585402c3dfb621016 : SUCCESS in 23m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/515b227c59fe4b03abbbfe17e1f369cb : SUCCESS in 59m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da1d9d4f8fc74f199253c170485d4a3b : SUCCESS in 58m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d2e0cd8ffe57470e8c132328c0ee655e : SUCCESS in 36m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6fcb04638f6745c8be3dc76103e63659 : SUCCESS in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57ac552d29546762e9ad9a31192e38775e538ec8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-01-30 14:53:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"92024b99133502289d4f72f12b092f6ef4c9d7cf","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-22 11:09:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"268539c19dd0d68a5d7d56fca67596284705d077","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-13 14:49:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2995038df89c7e9af3e0b3b94040775b158fde3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-04-13 20:22:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8cfb5888928fca604f546bbce4a05cf62d66886d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:23:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af4ccde21097067d9789cd2734bb243cd269dd99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:17:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af711644ac8a46938762c0e80fb87d50\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ca338ed548dd4a19a11aea1e3288aed1 : SUCCESS in 1h 29m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72aecb14f4a14e678c3c7eb93d3600bb : SUCCESS in 1h 01m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c910e9b0d7f4a2ebee84858efe4a392 : SUCCESS in 13m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed4f629ad5b64ddcba4cfb68d84b37a5 : SUCCESS in 13m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a4ad1dc9cc4602bd392deda2165715 : SUCCESS in 15m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc93b32df274491b311f85983b3d8eb : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a1598e6d01e4cbcaebd40089238bfb1 : SUCCESS in 24m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c46795dc60264af5a6ecd789cee3729b : SUCCESS in 46m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/148149f4b77a4334937a89c4b5072e28 : SUCCESS in 51m 54s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ed2e7ecd4744ddf8e8ec5d1bf7538f7 : SUCCESS in 22m 40s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25ba6a9c89e44e4eb06f6ee7eae074ec : SUCCESS in 19m 17s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2c8cf29bf2874e29ac5aca9c45d72b5b : SUCCESS in 1h 28m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/5d8fa19f59d3470ca6937d1012b89e53 : SUCCESS in 1h 46m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4cd6d86edc884e6eb3066b52753fef80 : SUCCESS in 5m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cec64a0c119d45169d664a37d836402a : SUCCESS in 55m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/37bd7e6ca7a640bcacc844b90c3886d3 : FAILURE in 1h 15m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d381dfa2558742a28b72692d62f33fa3 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"80478361090f4937f2e4f19255a2b42e0d3d4526","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-04-14 05:16:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8fe398e5bae6c38332b16fa4f5179afbfeff0ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 05:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25342171abf349b6879127a4bc1e988d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d8e3544b2824ec7a767b2fd542286a5 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d34f364dc23643cfa30b7207358530c8 : SUCCESS in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55d7d9f9ea5afac56ead8c072dd50ca7470b6278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:23:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de61e86931b0449bbee6a559864b3d0a\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9d6d4e0125cb4a509193e0e16f697751 : SUCCESS in 1h 19m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86c8b8fa1bf74e47a7ab137c449ebd4c : SUCCESS in 1h 00m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fc6f7ef78cd44018f64a9ac353a2a13 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40139617b4b940f0917a066e319559c8 : SUCCESS in 12m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9350d8f5d7fd4b86b2d07a7f4fa2b1fd : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9597670e585c4f36b69e2775b976b5e8 : SUCCESS in 10m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c881e4570aa04619acef494963727ba8 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac314bc2f2504d01a67f30c136bdbf46 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55288ad95f341b0a177b069329d644f : SUCCESS in 26m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/16f59335c71142d6b75643e6bbf7a4e8 : SUCCESS in 54m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3af5cb2174d4da18e11f0b518a4fe79 : SUCCESS in 1h 10m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f23608201b7c4075a19f3429e69c8f72 : SUCCESS in 54m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b61fb852ff234b59af00df50c82443d0 : SUCCESS in 38m 20s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f9de7ed99d92437f9915efb7d558e362 : SUCCESS in 1h 38m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/d6c8dbf05a664d28a40c65ef8cc7aa77 : SUCCESS in 2h 01m 05s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/fc687e839ad64b09a5daa7c6e4c717a0 : SUCCESS in 50m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3c0ca6d228cb4ad281353f4de67610a6 : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd0ffd0752094c1d9700098e6bacff45 : SUCCESS in 24m 23s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bd5eeb46a294552ad180086ec8d9a63 : SUCCESS in 24m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78b096b51e2d4f9ebe12fc407e5a069e : SUCCESS in 21m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b71358d1d994916b5562ec095bc6332 : SUCCESS in 1h 04m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bbc13ef300d479088897476f0b0f47a : SUCCESS in 1h 00m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ddc6446b2d064d4f9542035e53d33a10 : SUCCESS in 37m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/13c00cdedd0c400ea1657dee6871442f : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29cbec63e64d62736faca63174792441a66939d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:24:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06f6d07a1660c919372a1a7d01ba515230fae05c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-04-14 07:32:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"6941c19ca62829e50b958ef61d1420af532fef44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 09:33:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b382c488ef4e4e9abaca8bb941d94b4d\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8f7e70b1e5ab42f09163a5fd6fb4200e : SUCCESS in 1h 28m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/689403e3bea54a49aa29fa933aabb62a : SUCCESS in 1h 03m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98a365394bc408887f6ff1169d9d681 : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30c4235bb3bc49b38287c3a438f9002c : SUCCESS in 12m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc825fb8548f41fca1fb8d810f9b21ea : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265e6a7e6fd74ac09acd4625b919a71b : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e559279fc4194a93b0d97e09d4c7b6a6 : SUCCESS in 22m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8b70e70318d64d519e5ead7fe5e0a27e : SUCCESS in 59m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/30f4b665cf624941aae2d396f1be79bd : SUCCESS in 55m 28s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b24e3686ee046eabf1d07b1be28a683 : SUCCESS in 23m 03s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f27021a8b1594b38835c11eb47d23b3c : SUCCESS in 23m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dfc93fde48a64c8a980ab20fc95c70b2 : SUCCESS in 1h 38m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/eadc21f040b84bf49f30f95f4a6f51b5 : SUCCESS in 1h 52m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5a7a45d800f647e9ad68df6f4f2b9bec : SUCCESS in 5m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/05f4f3c860484717b8eb32432d4b694d : SUCCESS in 1h 10m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/25575688133c4495b921001685eb73cc : SUCCESS in 1h 11m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eabceaac8a1b42d0b946e1c277320f43 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3571eb8f85620a36f784c5227df53b819ce0873e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 09:34:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2c0abefcf6222f97d85de90101fbacc19d07541","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 09:35:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13603a87fdf14b378f8a78eff000e7d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938060a4ac37471db1156c32a9d86e02 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4178fbc992c476bb2c1b8bfec188182 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d00a848a735f98b028f5930798ee69ef205c8e2e","revisions":{"d00a848a735f98b028f5930798ee69ef205c8e2e":{"kind":"REWORK","_number":1,"created":"2023-01-30 12:46:49.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/16/872116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/872116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/872116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/872116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/872116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/872116/1"}}},"commit":{"parents":[{"commit":"9bca7f34a0d1e546bff8e226cbe45ccbc520c878","subject":"Merge \"[stable-only][cve] Check VMDK create-type against an allowed list\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9bca7f34a0d1e546bff8e226cbe45ccbc520c878"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-08-10 12:45:04.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2023-01-30 12:46:49.000000000","tz":0},"subject":"Fix rescue volume-based instance","message":"Fix rescue volume-based instance\n\nAs of now, when attempting to rescue a volume-based instance\nusing an image without the hw_rescue_device and/or hw_rescue_bus\nproperties set, the rescue api call fails (as non-stable rescue\nfor volume-based instances are not supported) leaving the instance\nin error state.\n\nThis change checks for hw_rescue_device/hw_rescue_bus image\nproperties before attempting to rescue and if the property\nis not set, then fail with proper error message, without changing\ninstance state.\n\nRelated-Bug: #1978958\nCloses-Bug: #1926601\nChange-Id: Id4c8c5f3b32985ac7d3d7c833b82e0876f7367c1\n(cherry picked from commit 6eed55bf55469f4ceaa7d4d4eb1be635e14bc73b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d00a848a735f98b028f5930798ee69ef205c8e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d00a848a735f98b028f5930798ee69ef205c8e2e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
