)]}'
{"id":"openstack%2Fnova~872118","triplet_id":"openstack%2Fnova~stable%2Fyoga~Id4c8c5f3b32985ac7d3d7c833b82e0876f7367c1","project":"openstack/nova","branch":"stable/yoga","topic":"lp_1978958","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"6962":{"account":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2023-04-20 11:08:21.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-04-20 09:01:05.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}}},"hashtags":[],"change_id":"Id4c8c5f3b32985ac7d3d7c833b82e0876f7367c1","subject":"Fix rescue volume-based instance","status":"MERGED","created":"2023-01-30 13:02:06.000000000","updated":"2023-04-20 11:10:32.000000000","submitted":"2023-04-20 11:08:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872118-lp_1978958","cherry_pick_of_change":872116,"cherry_pick_of_patch_set":1,"meta_rev_id":"fa1df90ae52bdeaf4a5d13327f41ceaaf6239b86","_number":872118,"virtual_id_number":872118,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 11:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-04-14 14:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-22 11:13:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-20 09:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 09:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 13:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-30 15:11:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 20:33:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-31 12:06:18.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2023-01-31 12:06:18.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-01-31 12:06:18.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-02-08 10:03:43.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2023-02-16 16:16:38.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2023-02-20 16:17:02.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-02-27 12:24:23.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-03-22 11:13:22.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2023-04-14 14:12:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"17de8b79959fbcd489fa6f948a8e40d1fe01257c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-30 13:02:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"c2ee0b72a9f36ca731232e1417cca7160b53abbd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-30 13:29:08.000000000","message":"Topic lp_1978958-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"65a8a887cd962981799b84363e3e021c0ed13d50","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-30 13:29:11.000000000","message":"Topic set to lp_1978958","accounts_in_message":[],"_revision_number":1},{"id":"d5f906a3c99ce14afaea7ecdb79e69128cdb3e9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 13:36:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d93cb45f5c946a386077fbe6263228e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea1bad4b2c294221ac1ce399160d3b86 : SUCCESS in 24m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ba798684aa648c880e7bd7343cfef7d : SUCCESS in 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67066d19878ef7964f86ce108e959344d8f49c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 15:11:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d990ccdc5854eca9b38c6b81a0b41a0\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bc98d7393fc9483d92aa35e69c229b2a : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8ce0d80548234553ba3e741a2652aaac : SUCCESS in 2h 01m 56s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac196c2aadd94e3da4d88c81a941aab5 : FAILURE in 44m 51s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a81f1a87b02b4e70a29869685c9b938d : SUCCESS in 54m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/404441d06d3645e68a9c763d552eb08b : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f36bec793142f08c0257acbd89a7cb : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3117d42ada304401b5e1aab75c17307d : SUCCESS in 10m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb185ab738b64aa88b9e4a3052430930 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52db56272c5d46f68336ce64f6db530e : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb89588706404d1c860180e90d036bc5 : SUCCESS in 21m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a3b8c4ed6a854f6b93fcbf0ae4705efa : SUCCESS in 52m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/422da5b78fe14c5b90548268cf467fbf : SUCCESS in 1h 10m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/581dfe095d2342a9936473a7dd68c140 : SUCCESS in 47m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d40f1581f164428396cfb8697091e31b : SUCCESS in 53m 10s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/23a7b001cca84cb79b5def960fef31d1 : SUCCESS in 1h 13m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/b8ebf8572d1d4d8fa4fc1cacb74ae8da : POST_FAILURE in 1h 34m 00s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ac05da98b5d4470db5d3d55711311847 : SUCCESS in 52m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4dfdcad67a944a7298697cd336127f66 : FAILURE in 5m 04s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/013f6f62947647f9b3111b996373fcf6 : SUCCESS in 21m 39s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55e1f245da4e4a8b8ede3ca07b57b09b : SUCCESS in 19m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96467901f5e449c79033817ee6f01879 : SUCCESS in 19m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/25fecf28a95140d4950e8506559c8418 : SUCCESS in 1h 02m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6733441b6bc94cd69525338799fcea12 : SUCCESS in 56m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4758ef5a6adc4b28b4055b98f301c68e : SUCCESS in 31m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a0e443e5c14142958260eff8e12d28c2 : FAILURE in 43m 00s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/e0404ebcf8b245baa3a6190fecdebb6f : FAILURE in 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab6481ad1d694a60ec883e601ff86f20743cc4f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-30 20:33:24.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/4e23fbcfa87f4c1d844a7f2ae3b5ea1d : RETRY_LIMIT in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d5e34bbf3c9a4092893e37454bcabb81b99b9b0d","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2023-01-31 12:04:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53297344e7d3a6527d0930349146c52ca5796172","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 12:14:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7fbb60773b9467abc6052019b899fe7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f05108318bb4a01b86dbda9d329ce08 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/749b666687c24daf82b6b98511438cfe : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93b9fc7867052e6de48827ee1a9f666654f27ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 14:10:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6604cf40a9d4912b232beebed7ce6b6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a0787650f3b240dc934bdcabd1f950b0 : SUCCESS in 1h 07m 39s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/97c4c853de4647a48bfc0901eeb896d3 : SUCCESS in 1h 42m 36s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/e623a0c15e074ade8757676b58f507d6 : FAILURE in 30m 59s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1da4d69b4bd643f1a9b8088f9d4b5417 : SUCCESS in 1h 03m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0de1835f2b4d6fb63b55618430f487 : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/707ba4d4a52040e8a31a9f2ceb3f2788 : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e724ae5c5f0479cadf16e4aa8177d52 : SUCCESS in 12m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f9822cfca4748b0a398ecbd4005cc39 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26e7b5f683e444882fd10a6f038dec1 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bf02b46186645e28d90d9a64230ed79 : SUCCESS in 23m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2fb5b4cc1e5840ff8461696eed0552fc : SUCCESS in 51m 13s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/937997d7a8884ae8824c6473d0527589 : SUCCESS in 1h 02m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/824f99f6dcb04dab9cd7f94d1485fd97 : SUCCESS in 44m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/90f4d1fa33e94de9b845f7191d52a089 : SUCCESS in 37m 28s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bbc4e136e40b4f5abe4bfefeaa48e6ea : SUCCESS in 1h 56m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/36b065b659024962aaeebcc9641abf26 : SUCCESS in 1h 47m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/7281b627072349d6a5fa7c40f2eb875b : SUCCESS in 45m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3445dd9716974ffe81b6128b8eea2b66 : FAILURE in 4m 51s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/6027a1e135e4405aaa8c88628aaa8be1 : SUCCESS in 21m 47s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89a44dee4d75442b889e37a4ab45fe59 : SUCCESS in 18m 49s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6323147e1c3246b6a2a5132066071943 : SUCCESS in 21m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/88353c5ce7f74b87bafd4e450c42bb69 : SUCCESS in 1h 36m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed7bfd5d178e4b24890ce21ae4519e87 : SUCCESS in 45m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d8be29740b114aba8e9526760bf66b05 : SUCCESS in 35m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/77dd32248f6a47adba384e5a827291ad : FAILURE in 32m 01s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/5c82da62dce246db9c42cb1a006dc34d : FAILURE in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3bedc0429ae67d303fbf98573b281e5919ffc2e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-31 14:13:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/815e0c88be134e5faea6cb50c7e0d7c5 : POST_FAILURE in 1h 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f2d8954dfefe28b22afaac485185d449e85c5602","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2023-03-22 11:13:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fbd6b91cb7d75af4f65590805386fa6913d13ec1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-04-14 14:12:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f519406cb4a8c23bd153183c54fb04ebcce50d1f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-04-20 09:01:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93fb71280e8c787c3daf840142d1d7ae2495994e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 09:01:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4810628730ae9822c36cecd230567e7361100b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:08:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6509e6a95cac4f5bbff1496b21387f15\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b24e831c3e6c456eaffc0b6e46e7cdda : SUCCESS in 1h 32m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3e42f5e51934765a109c222ce0458ce : SUCCESS in 45m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57da8016aceb4be3aee82e7edc00d544 : SUCCESS in 10m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/375c1d1f912b4e3dae69ac92a746dbeb : SUCCESS in 12m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdec9364449846419b1eb94010c6a849 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd6fbaa4c2a49d796884fb2cdf2428e : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f4c89975ae54590835b5bf99e5cbe94 : SUCCESS in 22m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/60d7706522bd4319a44d0e35ed456357 : SUCCESS in 38m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/53761aef56624c33ac8acea731fcf827 : POST_FAILURE in 24m 57s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/b95179301f6d4cfd9ba8770725fc5716 : SUCCESS in 21m 21s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a65530684b842b3abe4cd0c946b86ea : SUCCESS in 18m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8897ccb2c1c45faa066d0f551fd53b3 : SUCCESS in 19m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4773487a36dd40809eaa7112343a1f3f : SUCCESS in 1h 19m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/b01814ad5f00462d8130a363994fb717 : SUCCESS in 2h 00m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/80dac48cb4e34c7d8beb9cdcfd99cf0c : SUCCESS in 4m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0cd07e34fa2414aac04e86c001cb496 : SUCCESS in 1h 02m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f10653f73884942a6439394ac0c3c5e : SUCCESS in 59m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b361062d14234eb2a268760cc21a2627 : SUCCESS in 57m 23s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d01b8e58c3e4e8ca687edb329f78b54 : FAILURE in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7f956e95003e165c1795683bd021c653264f2e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:08:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa1df90ae52bdeaf4a5d13327f41ceaaf6239b86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7035c6825e104ba08c8a2a704e87e0d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49f0b6e57e74cc29b89be75ada8aeb4 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bcbb074e4e74cb69c49e2bea5c596a6 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4073aa51f79be54e2e6e8143666a7c1f9a00e03d","revisions":{"4073aa51f79be54e2e6e8143666a7c1f9a00e03d":{"kind":"REWORK","_number":1,"created":"2023-01-30 13:02:06.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/18/872118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/872118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/872118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/872118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/872118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/872118/1"}}},"commit":{"parents":[{"commit":"c07495d9d64dd0635d72fc7ff67d73a656a40d13","subject":"Add a workaround to skip hypervisor version check on LM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c07495d9d64dd0635d72fc7ff67d73a656a40d13"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-08-10 12:45:04.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2023-01-30 13:02:06.000000000","tz":0},"subject":"Fix rescue volume-based instance","message":"Fix rescue volume-based instance\n\nAs of now, when attempting to rescue a volume-based instance\nusing an image without the hw_rescue_device and/or hw_rescue_bus\nproperties set, the rescue api call fails (as non-stable rescue\nfor volume-based instances are not supported) leaving the instance\nin error state.\n\nThis change checks for hw_rescue_device/hw_rescue_bus image\nproperties before attempting to rescue and if the property\nis not set, then fail with proper error message, without changing\ninstance state.\n\nRelated-Bug: #1978958\nCloses-Bug: #1926601\nChange-Id: Id4c8c5f3b32985ac7d3d7c833b82e0876f7367c1\n(cherry picked from commit 6eed55bf55469f4ceaa7d4d4eb1be635e14bc73b)\n(cherry picked from commit d00a848a735f98b028f5930798ee69ef205c8e2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4073aa51f79be54e2e6e8143666a7c1f9a00e03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4073aa51f79be54e2e6e8143666a7c1f9a00e03d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
