)]}'
{"id":"openstack%2Fnova~872204","triplet_id":"openstack%2Fnova~master~If83ce14b96e7d84ae38eba9d798754557d5abdfd","project":"openstack/nova","branch":"master","topic":"bp/stable-compute-uuid","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-01-31 16:15:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-03-27 13:42:44.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2023-02-07 02:17:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If83ce14b96e7d84ae38eba9d798754557d5abdfd","subject":"Protect against a deleted node id file","status":"MERGED","created":"2023-01-30 18:53:52.000000000","updated":"2024-03-27 13:42:44.000000000","submitted":"2023-02-07 02:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":46,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"872204-bp/stable-compute-uuid","meta_rev_id":"5664d0907f96b37325af8c9a4d150bc071fd2055","_number":872204,"virtual_id_number":872204,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 02:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2023-02-03 22:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2023-02-05 22:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2023-02-02 14:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2024-03-27 13:42:44.000000000","post_submit":true,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2023-02-03 22:38:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2024-03-27 13:42:44.000000000","post_submit":true,"permitted_voting_range":{"min":0,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":36932,"name":"Xiao Feng Ren","display_name":"renxiaof","email":"renxiaof@cn.ibm.com","username":"renxiaof2000"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 19:07:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-30 19:39:52.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2023-01-30 21:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-30 23:09:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-31 15:24:09.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-31 15:24:52.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-01-31 15:50:08.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"CC"},{"updated":"2023-02-01 07:22:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-01 10:54:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-02-03 22:38:41.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2023-02-05 22:09:29.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2024-03-27 08:03:20.000000000","updated_by":{"_account_id":36932,"name":"Xiao Feng Ren","display_name":"renxiaof","email":"renxiaof@cn.ibm.com","username":"renxiaof2000"},"reviewer":{"_account_id":36932,"name":"Xiao Feng Ren","display_name":"renxiaof","email":"renxiaof@cn.ibm.com","username":"renxiaof2000"},"state":"CC"}],"messages":[{"id":"0afeb8e3567b782a585a5c2b4fa1b3e35bc15c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-30 18:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"203fdde270f377a553de89682e957ad05683ea16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 19:07:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6677242170fc453f82596871fd8040aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcb0c0072abe4b4cad3ed560d89fb7f8 : FAILURE in 11m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c73066f2cb1454c821160a6c689085f : FAILURE in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d35de7bf625d433e1222d2e6d09e084b73eff0b0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-30 19:39:52.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"31e3737588549300ec8e9341261abba0b257a78a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-30 19:47:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6eab64bee2b966e03e7bfba24243f26d1ca8e72","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-30 19:50:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecd0ec0363d21447c6b29c5fbf8b5e4f8cc0c5e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26c6b9f21db447ada5513a763954ce0c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3bf6ae5d7d5e457fa195dd8b64e9f37b : SUCCESS in 1h 08m 34s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/01d7bb162c1240b5ae1e1e2a8a8ccec2 : SUCCESS in 1h 28m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/500724e977a84cb792cec7cc7830c0f8 : SUCCESS in 1h 07m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/960f59592fdc4a7ca9cff48ccd9835b3 : FAILURE in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5810226b7fe94164b4eb1f30765b3dd1 : FAILURE in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6e594e16304459fadf995d76fae2bd8 : FAILURE in 11m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3ceca7922e340289cd91381c7210928 : FAILURE in 12m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4848b28cbdd443a390db97043d85c8f4 : SUCCESS in 9m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/beaff5219dd840f78d597345676ef41e : FAILURE in 55m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/157c628d92d04ca9acf7716e8708d5bc : SUCCESS in 1h 10m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/46f90ff26f9c49078e366265ec57dca4 : SUCCESS in 48m 32s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3368040564b6443e80c09b92951b05cf : SUCCESS in 47m 38s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d101e46beb334c9ea9c7234ee4135fb0 : SUCCESS in 1h 23m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/b0958b1cfac9470185d1efcecab5739c : SUCCESS in 1h 41m 45s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8fe300390ba8470385c845422b0f0adf : SUCCESS in 59m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bc707dbe09d34c99b7b84ad56307df20 : SUCCESS in 4m 40s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea0cc55b6b90467aae124182804cb2ce : FAILURE in 23m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9e93dd5bb0842a48c09d2c703a6518f : FAILURE in 21m 07s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5de4b77d27424e33a5f84daeb2627b55 : FAILURE in 21m 44s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bb42615c4954251977e04a73c49f526 : SUCCESS in 1h 10m 30s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0fa0d9b5a0ad4596b707e50c43d6ea0d : SUCCESS in 2h 02m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/abee6b6c2a2544ecb4c68cb104fadcc6 : SUCCESS in 1h 04m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75e5cc3e670c406082ffa2e0cf33b096 : SUCCESS in 49m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b17d62e536c24187835cf525bf475058 : SUCCESS in 30m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/526d4ea80d0b43efb103de9f74c535dc : SUCCESS in 52m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9df430b13d04973db76dbe493cf3b86e4c8f29a9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-01-30 23:09:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"96cb4371b0267be0d67b9c69e5bf62fa7712cf53","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-30 23:14:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"253c83e7c4f2ca2ade3cbe486e9679d802b5f715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 15:23:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df28cd68cfc3d4e256d8944229baf63209bcd616","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-01-31 15:24:09.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/04/872204/2/openstack-sriov-nova-master-macvtap : FAILURE\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/04/872204/2/openstack-sriov-nova-master-direct : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"b18d5b2e50b563cf2b2469f15ae6fabaf82a2cb6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 15:24:52.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab010e60318085d3c07ab71a5632858d957db92b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-31 15:50:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f28ef23c692ae8037ea45e7943b9463b3c78833e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 15:58:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"401219f59d45caa48b286573effe5bb62d8b862a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-01-31 16:00:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95bfab9cf6eed5ab054e4e44314969f04c0a320e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 16:15:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa6f158eed9308600fee656fabd30729c1961922","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-01-31 16:15:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c238265548795e72c786d1cf344aabd5e37ac352","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 16:28:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52d1f00a417041009ab0c7ec26240fba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/585cb163229b4fc48047272647ed2a2d : TIMED_OUT in 1h 01m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/926964e74586486cb5ea4c5b3ac3247c : FAILURE in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92cd696bade5a368f443dfed6a17f659611bb896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:12:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d472015251bb49ffac700a7d836f0dfd\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac9f4b3af3b04121a48d8860ed134830 : SUCCESS in 1h 03m 52s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5f2d6e181f134bddb7674a4f5c1b31f1 : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8e89dcfc21e4d4fba5b67098f2857a7 : SUCCESS in 1h 00m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6111b1a91f6d4d30a8a4f50a1fa5c86f : FAILURE in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ef90e0f92748869ab46b5a1826cc52 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7e6c0667584997a62cab909b6d5095 : FAILURE in 13m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14ec8c229c5045b8a3bc84239343537d : FAILURE in 15m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35a2abd6f124256b415311bb0eb0d41 : SUCCESS in 9m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3dd5781a74f8474892c5b7564f03c28f : FAILURE in 51m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a71ddd7e0ff4670b11d5a362987f956 : SUCCESS in 1h 03m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6403d1e70e1e44638efbaf9e821cc81c : SUCCESS in 47m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/896a37d41d3945fca3cf2e9d59ec9c56 : SUCCESS in 42m 05s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ea65455c719049e4b492f6fb0e132400 : SUCCESS in 1h 31m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/6f34c302fe004e068c7e51924d1b8d23 : SUCCESS in 1h 34m 39s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ddce7f34c2ad450e89bb175d4b9d15c5 : SUCCESS in 37m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1447916e057849f9a739d6fafc90b1d9 : SUCCESS in 4m 25s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2f98ec85c7c040cc96a4f7f22a79e2f7 : FAILURE in 25m 19s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0673fd0c1e714de8bd69ba9c3502801d : FAILURE in 24m 38s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18ca3883fbc94bea999141d44422fcef : FAILURE in 25m 17s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f422c220f444b6cb3c8c2c24bec7fa7 : SUCCESS in 1h 19m 36s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d69a3492bcc04334a8dc902b05450244 : SUCCESS in 1h 29m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c40156dec3342c899847c1117899019 : SUCCESS in 1h 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f124ad246b348df8ac05e49f1bf676f : SUCCESS in 57m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1c06d143048e4fa18203cfd20f69f621 : SUCCESS in 36m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e7914a1f04ff48fca90c8a69b979e33f : SUCCESS in 38m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"289a17e8182c6d6af6d52f8f8e78d3acb3b1ee83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 19:31:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8cae049d18e017775f1149dcedfa462c8361a7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 19:55:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b35cf44940097ccd5faaab44b1edd765e86629c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 20:11:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5d3c06535770f8b2ef6acac4e799a2e15556f445","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 20:22:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6674fb8e6a044989b2a76a06e0c4c019\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cf66552231042bca3267fd685c1e8c3 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0854777efb4b429dac0d4b39be3bf1cc : SUCCESS in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5205475c981c1e370800dec5901fc36156bc233c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 21:53:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf6bcb2f13034fbab5573c13ccf57fa2\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ececac4799174782bbea143bbddc02b6 : SUCCESS in 47m 13s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0487317c24d04f6fb2118e6a832bca51 : SUCCESS in 1h 08m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9a36ba948ab4ade88277145e432abbe : SUCCESS in 48m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d89387b7d5640a98b6976d8bb394eaa : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abea1ab8444648b280ef7512966d1ac4 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fdc7d599a8047658810b6bd766e9f9b : SUCCESS in 11m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee2a4514beca4e3a8fb6039c7e4fe025 : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5195905ebc7d4549937bcb4fca95f9df : SUCCESS in 8m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/509c18a2203c4166b6ee3ac46a399297 : FAILURE in 43m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fbb324c179274177aa3c424963e6b66d : SUCCESS in 57m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bdecef9b196d4f52b17262756b6c9b7f : SUCCESS in 48m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/879e6b771215414db34fdef8abde58c1 : SUCCESS in 39m 36s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dbe93c86af204112a37e0de497487704 : FAILURE in 1h 39m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/2abfaefbd45d45a69663994413f53b63 : SUCCESS in 1h 15m 04s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/cf9cfeb85271490a8dc018fdb87e0fde : SUCCESS in 50m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/36c755887bec4a089403e7855063ee95 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c961300f422346bfbfcc8cf32e873870 : SUCCESS in 17m 02s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c76354ff6a614da287f8ff492bc0d82a : SUCCESS in 24m 15s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aedd8493508c4432b9298d222c4847cd : SUCCESS in 21m 44s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fce0cb31117542788ff08b9b193862eb : SUCCESS in 1h 32m 22s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5cf98107712140dd9485612df1ae7bd1 : SUCCESS in 1h 35m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f7648dcc8d34ff98bd30b0217d68221 : SUCCESS in 1h 15m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3afe7127aab4654b3f7451f06cb579f : SUCCESS in 39m 54s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bc5927dbf5b94a5fbdbe3b98bea2f3c5 : SUCCESS in 24m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c375b77d25d242a4b5b521f9c41c9275 : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"55564bde3b7c071e1e9171611b2e907e3e046a65","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-01-31 21:58:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7ad6d7beaf9ff9892ff0120bd4cb512acc22df8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 22:15:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89cde26bf7e247528d5b122f9e72799b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/abf3cf10c9314a23bebea1e1457351be : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96f765209cc44b1d87216d0a72b9a547 : SUCCESS in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91953bae020d9fb1074627fd47a6d6efee934f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 00:37:14.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeb10739079d46a0b975316baa8ab586\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5c7fbcc5768349d2862a80f0b1dcc40e : SUCCESS in 1h 06m 19s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e23e2d0db2504c2a9c9fd450f5ffbd76 : TIMED_OUT in 2h 04m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05cafa2c523b4aadbb829125c4444848 : SUCCESS in 1h 19m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53f26a676df64c14b81d439f4801f049 : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369db3b7068f4527b07b4212a7caef38 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fdaa6e58a9f4a4d94253a975252c653 : SUCCESS in 13m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1031317c250e4efbb4f086f5655e3d75 : SUCCESS in 14m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4c20028a2b4fbbb78b2887b6e4455d : SUCCESS in 11m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3a613b1260fa4d0d86d39335bccd308d : FAILURE in 56m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f66f6e9438434b9d8595bfd56561f89b : SUCCESS in 1h 55m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/36237840ed3d4ac69ca36f32a3e1df3d : SUCCESS in 1h 10m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/da77bbfa3e8d4977b1aba351317acf83 : SUCCESS in 1h 10m 48s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9a17d4f7affe49eab97345c929c43869 : SUCCESS in 2h 07m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/3f3cf3bbdda54a6094f64710bd4edaf9 : SUCCESS in 2h 29m 17s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/142ba4abbb7f499c9f003cd552fed4c9 : SUCCESS in 1h 01m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2f583803b86c40e5a7e38244d0467026 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2631b6e72e2b429a9517a0232c1bd2d3 : SUCCESS in 25m 14s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47a0d1f60dcc4510b059d0cde4649294 : SUCCESS in 26m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/080a4f20ef654420afd1a05ded8016f1 : SUCCESS in 31m 16s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/289a39440500499bb8901c78236d342d : SUCCESS in 1h 45m 03s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/84983b3ab9fc430f8567197801729a19 : SUCCESS in 1h 56m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbb744023beb42e0990ccf04e48ac081 : SUCCESS in 1h 51m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce45c4d890dc464f8ec89fbdbf57c90f : SUCCESS in 1h 02m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/615d07a229d840e8b9d6c11db1b574bf : SUCCESS in 46m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/70f5ebd172d844538ebfd8eef89494de : SUCCESS in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"35e80260f0c54e94cc84e90ed6a394cb392323e8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 02:33:53.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/0a23868b325b42d6a3be123d3e2a2d61 : POST_FAILURE in 1h 52m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ec7bcad3259658e4605816f168c04e921986c545","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-02-01 05:23:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bc7ab219475083b3fde2362828ff154d064e0331","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 05:39:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2076091c38154de9a7d43bbca51a76d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/974d3176cb8c4780ae8bd8ea876d1d83 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce483f82770f41a2a0063b13ec6bbcbf : SUCCESS in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7c5309cdfc32c52965a3bc392d4ebedee34e19fb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-02-01 05:48:48.000000000","message":"Patch Set 5: Code-Review+2\n\n(10 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2bc37d008a450e1ec552d9a4e3de632f3951b7b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 05:50:18.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"22986f6e7b25e4d29b1d58bf8a63932298006bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 07:21:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5b6609d4544ba4a2cf15ba80941dbb\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e4a4a12e9b244bcf885f128a4ca5d0a3 : FAILURE in 57m 05s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/07e46c11d24f48e295df6b4d40c22d2b : SUCCESS in 1h 26m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6584fac3f90e4cddade5908e8c2c5e38 : SUCCESS in 53m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7770286e52c49c3b5a7552c07868ee0 : SUCCESS in 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/034673aab3704afc9eafe33d310ff5e3 : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fde44c7224ed4d67ba3982e762531b60 : SUCCESS in 16m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf727ba2dfd8429fa1e779909d3bb758 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41867360d2a49bea935214e2eb28521 : SUCCESS in 11m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ed91afe82b2d4edb9c20f2ae80be366f : FAILURE in 33m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/90268138ed844206bb2a973f8b1b9617 : SUCCESS in 1h 14m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c5c3b25f2aea407593419c1c8a6d5cf9 : SUCCESS in 54m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/008548c5b3264485bd7556fbd6ce3f5d : SUCCESS in 46m 58s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9babbd79fb854eeb807daf3d295ee82b : SUCCESS in 1h 40m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/48de74e8255d46ab9f8b3c267d15637a : SUCCESS in 1h 31m 10s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1529afe238f24f59836f8d0bf6eeb671 : SUCCESS in 58m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/92465c87a7804e79a29a6aa81e8340c9 : SUCCESS in 4m 18s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/520180871e7546158cbfa3ef606ecd74 : SUCCESS in 24m 23s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35ec5bd5ca3b43cda63ac51fa9795410 : SUCCESS in 23m 56s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d3d9d0b7c604574ab0f6ad6e756b02b : SUCCESS in 24m 49s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40cea47176354238b9ab9173f2a39938 : SUCCESS in 1h 28m 21s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8bd0f2828a4d494d84a7f33444516761 : SUCCESS in 1h 39m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e33c40a4b45a4166be55797665f30de1 : SUCCESS in 1h 07m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cfdefb515c742298cf4f1b51d7e9d4d : SUCCESS in 53m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/51c20df6585a4edf99dc9eab46d7d84b : SUCCESS in 32m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9521dc4a826d4225af0a755698507b6d : SUCCESS in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a2f51220c4231b65367ff0653337438c5d0c6a13","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-01 07:22:10.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/872204/5/check/EMC_PowerFlex_NOVA/229fbe5/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"1c8729d3330bddca01b7aac0dd682153a5befbbf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-02-01 10:54:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a5edba82b4fa75f7dbe902265b83bdca0ab9d725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 11:11:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5f6a3cb21e17289bb77f5b848e1cbcbc6adbdcab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 13:00:55.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/872220 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"e51b4d12cdcec79b5115f93f68752486648fe8d7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-01 14:36:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"213d1198d62a5e54b60c50831b6e15ee596a4b82","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 15:02:52.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"a65fd1cc8cbb754d9e51ea848b69d8d612553a9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 15:13:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a8f57b71e54058ba869e69d0a5b6fd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c251b3451b654faea9c90f33b9adc1cd : SUCCESS in 31m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b272305cb9d4195911c60658c1228e8 : SUCCESS in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2fde5b9c3e58a73643013a496407b0dd8ec12226","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-01 16:11:47.000000000","message":"Patch Set 5: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5202d8776d79d65ad5a640e6c05ad1a139634ce3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-01 16:26:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e014526a28f12850047658c8a27b5a71f2b459d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-01 16:47:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"01c79667f19849a098c9e9234df174edadbf910e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 17:13:12.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"6824152e587ad4456fb410fa2ca9d1145e13a0b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-01 17:23:42.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"bb5891f82e8fdd8395bd87c07201ee5161838392","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-01 17:50:53.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"0faf2debc175f13c717c160c4c5c31985d3a9e9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 18:06:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca49a6bdfffc4276aaf49a2842548312\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f9f1aa59976438296f500bc18fe7e8d : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/336387851aa3416aafccf3cd32014280 : SUCCESS in 40m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ccb9fce9de8b5a2afc7fc5ec94de499a192cdadd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-01 18:53:29.000000000","message":"Patch Set 7: -Code-Review Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"aabf135dc020033d5915585cc80e0531fe9f18ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 19:24:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb708e2d02cd43648598a975b2d7e40b\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/494f4dc848194317b36ca61d11bfe76a : SUCCESS in 1h 07m 25s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1217263f651d47f0b2e433028c040a23 : SUCCESS in 1h 23m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0233b622ff624ee88b6dbd32362c79d8 : SUCCESS in 1h 03m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7abad1bd811d415daf5144504370e954 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde079e21b7042738f929122b20990ce : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/760e45d9dc2147c1929fe232400dcedd : SUCCESS in 15m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28989e7e540d403896e51cb842947730 : SUCCESS in 23m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9b6144bab64d11b2dd5a4960820fac : SUCCESS in 11m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2073d226ef5e44e8956593e4c3990b9a : SUCCESS in 1h 19m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/068a1593def349a5b2a17205bb338308 : SUCCESS in 1h 23m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/91c5b8939cd44074a073097807f7440e : SUCCESS in 52m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e2682c7def1f414d834b919dd880bbbc : SUCCESS in 56m 53s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e9f525755e9f4f6f993781467643b6ab : SUCCESS in 1h 16m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/920cea16b5b248dbbcf845d01d6b833a : SUCCESS in 1h 37m 36s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/49576ecbb4e8423ca6b646837cc0c119 : SUCCESS in 42m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a913a3acf8d94c0bbffa631dcc547bd0 : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6f86459bfd9447aaa4b0e75277fbd4e : SUCCESS in 23m 04s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e05654722544fc4b0f3d20b20876dfb : SUCCESS in 25m 09s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03d77cf472c7405ba660d9f7514489fc : SUCCESS in 24m 00s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58c20fc34ea44fb6908b36b852309304 : SUCCESS in 1h 36m 38s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e5fb17bc47904fc6b1c6ef8375829e70 : SUCCESS in 1h 46m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/edfda349d9dc4523ae9ba7a96e203ab6 : SUCCESS in 1h 17m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/364d59f06de942fc95a8d92148c99033 : SUCCESS in 58m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c26931c2ce174abc8522d25212273cf8 : SUCCESS in 31m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c650942f71ed4d8a9e93fe9d8b97344c : SUCCESS in 39m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d3950fe00f683eea2032997aff51f245eccd2761","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-01 19:25:24.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/872204/7/check/EMC_PowerFlex_NOVA/47775e9/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"266fe4e41c71a994320e6b5b1ccfdb9728a1d310","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-02 14:24:50.000000000","message":"Patch Set 7: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"847d57347c61c1a74e89abf33a6e8964f1c4201c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-02-02 14:42:57.000000000","message":"Patch Set 7: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ad762f24cf317df4d323d0ac8c8ffd8bb4765c7c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-02-03 22:38:41.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"52e49d8d38eb1ed64350ce22a8671bd01628c494","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-03 22:39:40.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0c104c3af2caaa178f65b612263ebcb65817e0f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-04 00:42:34.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d27dd68586e04dbb8a3394ed6fa98b6f\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a2b3b3f2331941789a3f20543a327120 : SUCCESS in 1h 32m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d53eeb7f03d4b61b9395f94e0b9add7 : SUCCESS in 1h 02m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0837044ab774ec49f9d4feac0f6d27d : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30eadba3dc624ee5a7ac6b7f8c2c6bd0 : SUCCESS in 13m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21c6355a545243d1b74cb1866677677f : SUCCESS in 17m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057664c1c5c04aa5b684e50c15234dc0 : SUCCESS in 12m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/56791f93c99c4c12ae6dad72a1cdbe76 : SUCCESS in 1h 05m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/24419b4879f449268831603930e6184c : SUCCESS in 38m 35s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1cf038331b5f44f09aa63fd797e8bf61 : SUCCESS in 21m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d0194e48c194b439185fc4e908abb29 : SUCCESS in 44m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bcefdad368f44a2f8b58b6eed30c77ae : SUCCESS in 1h 18m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/125a581c0da64324964d13e4f5390d66 : SUCCESS in 1h 56m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/59661275644e4392b64ca06befdbea36 : SUCCESS in 5m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b7eb1e1966e4792b306980b5a21a402 : SUCCESS in 1h 08m 54s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96f7f763cd1e4e0caff0ae5c619677d5 : SUCCESS in 1h 41m 33s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6782de10d0fd453b9f58ce11e1345290 : SUCCESS in 1h 44m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d077a76bd9a4f5d982d893dedd4129f : FAILURE in 1h 41m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43612cfa0780427aa5e5f1876f7b66f3 : SUCCESS in 47m 14s","accounts_in_message":[],"_revision_number":7},{"id":"9ee3e41a783224ec803a59e4031abe99bae6853e","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2023-02-05 22:09:29.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"641c3cb6802fe783839440a37d6666088ee9ff14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-05 22:37:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6505f9c432a456390636bec0444b0da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd9301c8272d40df95696752a196011b : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4e34bd540c0406b8eddb6589fcf929c : SUCCESS in 18m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5c2560d01d42bcad5a40e2dbed3b4f5e21990d95","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-06 00:34:35.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/bce8f32a3e394137bcf866ab2abe36e4 : POST_FAILURE in 2h 03m 06s","accounts_in_message":[],"_revision_number":7},{"id":"242c038ae2880a8615aabd0e178e520063e928e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 01:28:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e84ce9b054d4ab4beba01ab2ef34bbf\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/af41f313e6ea4bbcbdc0f8acb5ad7ef5 : SUCCESS in 1h 07m 08s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/090e366cb0f54e10806ed22e539bab01 : TIMED_OUT in 2h 04m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/540e800feadf43cfba7ff45dad787a48 : SUCCESS in 1h 16m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/158a1281288b40e299cc99ff7e860c48 : SUCCESS in 35m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9cfbf858014af29666d402d2f390da : SUCCESS in 24m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f78bc3524fe94633b10605048a163cde : SUCCESS in 21m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3807b00dbe27435f95ccec4ff8f52e68 : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b028a0dd8a4ca9a85ea13ec68c9aa6 : SUCCESS in 10m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/52e99a64cb3841a0bae37767e247dc92 : SUCCESS in 1h 04m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a398ea35a6249b8bb86472a7d34e4dc : TIMED_OUT in 2h 04m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b2fcf5fa01a845eeb32d3601d904ddd5 : SUCCESS in 53m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e2bcaf64a2ec4ffda6f344628954746d : SUCCESS in 2h 05m 17s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f57f40931d6e45f788654e5d5ea33671 : SUCCESS in 1h 25m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/95d31ff55a49469a85b642a838416453 : TIMED_OUT in 3h 08m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d7f5268a48284798bf06a6a6ec3c6c2f : SUCCESS in 42m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/80b7380bcbce4ad3ac884cc97fd5d71e : SUCCESS in 8m 32s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/315e305edfa647a5a1e8930845104b97 : SUCCESS in 33m 37s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/969e22f8b0e743c9a48e26832b946195 : SUCCESS in 47m 10s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c61b715049bf441f9514956ccdcd13bd : SUCCESS in 28m 48s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64f5e3e7a3b140f4b6103f8a5eb027ab : SUCCESS in 1h 46m 16s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9fadda8d7d854c18a1e2c2893fb8112d : SUCCESS in 1h 47m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/03a974fe0f294500b0f980801999ba75 : SUCCESS in 1h 27m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cfd8d606bcc467d8f080d9ae9ae5ac2 : SUCCESS in 1h 04m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d674313e6d244ab89754ea51ae932932 : SUCCESS in 38m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5c187345c48e461a98ef639c4f564c6e : SUCCESS in 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7445d66e5cb7e5e037a27190ce0c16b674a42d49","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-06 14:34:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"47f1a5486c3b2c2bc7536003f7affe196c85438b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 15:05:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0522e210114ddfae741064b68a3021\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46d73d3a4dac48d6bf8bb0111979ab04 : SUCCESS in 29m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1352cff2e9964d38a83bc988422e9677 : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5b6f3fe897e54787ccce0e78f4d3a5a7205ca303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 16:33:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e9d8893f3145c88002b8f5ff9f9093\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/aa056d1eb0214b868505ab5e5688cc8f : SUCCESS in 58m 52s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7e3b4dfcf1a842cd85262a2a16cab587 : SUCCESS in 1h 52m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c7dbd09484c4d6e944233c281ccce44 : SUCCESS in 57m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b1f0e64a0264c8cab809083696671a6 : FAILURE in 13m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a005c9307e6649ea81814ece4b9b3fc7 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36acc7873e6b40498fd1ad3f3d16bd36 : SUCCESS in 13m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9861a6d8cf6d4bb184e166aced57a275 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf634f1a941347008d89bb7979b34715 : SUCCESS in 9m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/54772e3f77ff4fd09ac1439b6617d097 : SUCCESS in 1h 11m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac26f0c36d2845979afde684d948d53c : SUCCESS in 52m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6aa5e4e8e217426f8bcffc3db07a4814 : FAILURE in 1h 14m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ea43f021956143fa80f4336784d2d4cb : SUCCESS in 49m 34s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0c560042dd294cbda7467cbfdaaedfae : SUCCESS in 1h 21m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/8e4e12a643ed455fac685d9e1a0e0164 : SUCCESS in 1h 11m 45s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1527d558046243c58ad93e329a7b7ccf : SUCCESS in 36m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a1dc658a8c8f4187bfc207108a728025 : SUCCESS in 8m 20s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bc2049bffca04a669ac0d5b1aed5dcd8 : SUCCESS in 22m 18s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d31650dc95ff4580a318200649a0cbc4 : SUCCESS in 19m 52s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41446b17c56445189c0f6e88347c730e : SUCCESS in 23m 02s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2d7ec1414d444739d3109d6e524e9d2 : SUCCESS in 1h 23m 14s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/181815ef1c2c4b2290527c678dfde38c : SUCCESS in 1h 27m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4780e2566a0240a68a6bc7a62f9273ad : FAILURE in 50m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f0e1e95903744dbab7d613d2b3ff1f2 : SUCCESS in 48m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/04b72a5201c047c1b3c74800783485d9 : SUCCESS in 30m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d5271f3df96241aa9f55d1848b379739 : SUCCESS in 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d320aa4a8d529c0825c29d841fed21727846fc65","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-06 17:52:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a429f126d3fe2e8cda03d7526cd88c227ccae1be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 18:25:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a08df41f4e80488cbc31e3e3a4e5c14c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/650284872b274848912e94fcca90d4ba : SUCCESS in 31m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8312393816a4280a2a9fb6a8325b247 : SUCCESS in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4347936a97b391ba3ebeda08525a1b8d91218915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 19:44:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84db934db93544ac91e7e4b2bb1d2a5e\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fc0d76eb70c44a1e8961de55cb9ae274 : SUCCESS in 57m 05s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4a906556d0c8424f8e67a1f67446f69d : SUCCESS in 1h 51m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf23515688cd4d629e124b80ee7f2b56 : SUCCESS in 58m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6658f9708ed4b1b8ea0af142b5a5c25 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69928c388fc648e183136d311c677ed2 : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520a0c83f93640048908041d30166a94 : SUCCESS in 13m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/471e28cb66d84a3495a3e6ee22874b4f : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1673a08247924a0cafa65619daff5cf7 : SUCCESS in 14m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/60ec27c48197401fa29916740433fd01 : SUCCESS in 56m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e09c9147de294bf3bde97f6a5fd533ce : SUCCESS in 59m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a60659068d244d4aa6068c4d13e9d58e : SUCCESS in 52m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2df686cbc1434e67852bc85ad2010993 : SUCCESS in 48m 38s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b2ef9efe1fbd4c7ca35e67d018f1c7bd : SUCCESS in 1h 05m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/8c23a728721f424ea535ae9eb1061249 : SUCCESS in 1h 43m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f32a863eca5946c1bcd5d166d66c4642 : SUCCESS in 39m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c29eae49533d4da29db4b62a86a863dd : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a38d4344bbad414f87cb9d5680626c24 : SUCCESS in 23m 43s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f5ec63892620433aa84cfab5d8f8f4e2 : SUCCESS in 25m 56s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/739c5ceb3d524555b053610e285d9857 : SUCCESS in 22m 26s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a482cef53c814a2fa50725777e1a1398 : SUCCESS in 1h 37m 57s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cf4770f956b14261b48b7ffb917fefb6 : SUCCESS in 1h 41m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/97150ce2140445e79c539bd04ca077a3 : SUCCESS in 1h 12m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/256f0eefb9cd495e84151cf9cc00caef : SUCCESS in 54m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4e5fbc23f73c44d4a8332146bfc6fd86 : SUCCESS in 18m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0d68cf5b831f4d46bf4c563fddd79a24 : SUCCESS in 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0bc27b044f5315ad723446a5c4e7e06452ea7455","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-06 19:45:05.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/872204/7/check/EMC_PowerFlex_NOVA/0f2de29/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b1870f6b0a8092acd7f701de9ce8f86284e2604d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 19:45:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a2e1b29a5ac1aeb5dc6abc6657940ab2706f05e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-06 20:07:09.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/b56b311635664dbe9ee65b25101a2905 : SUCCESS in 2h 42m 15s","accounts_in_message":[],"_revision_number":7},{"id":"a2613e3b9f23551b63e3c03f23e144e578128a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 21:55:43.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e23eed65472e42dd846c46d1ab023416\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/64155ad984394524b47478abb4b6f756 : SUCCESS in 1h 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45f4ef13f75946908ac03923a8aba84f : SUCCESS in 50m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/335b3d2d9267478e9645485f7d146a6b : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7747ba65cead489fbe3e6e59e19552ff : SUCCESS in 10m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/481fecf4a8b44ed2a5ea2a96014a5cee : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71660e07d2344fd68b7dbee05ab9425a : SUCCESS in 9m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f5fca40ae2aa46bca86a3620b69f8b6e : SUCCESS in 37m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3a83c26b77be43f09291d81dbba324ff : SUCCESS in 1h 01m 30s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8856f6360c3441afa4f407d261885e48 : SUCCESS in 22m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b4002ce56bd43fb9821fffa98fb12fc : SUCCESS in 29m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6f76136de483464b867b4fcdf1680935 : SUCCESS in 1h 12m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/b6ebc0be87bd499c936ef8e8859965de : SUCCESS in 1h 32m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f544b81c894141e4abbd28aebd802da9 : SUCCESS in 6m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38a2a1eaa15740ba9e448a3e51866fbf : SUCCESS in 1h 22m 49s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b7c83822a4f40b09cc9ee7c0d7be95b : SUCCESS in 1h 28m 53s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/de302d41f65e40559beb7a75953379a2 : TIMED_OUT in 2h 04m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa32d27c35e74cec947ee31e1d8bf502 : SUCCESS in 1h 22m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d779953a0ced4d7e8276b43c9e36bdfa : SUCCESS in 43m 40s","accounts_in_message":[],"_revision_number":7},{"id":"0119f8f785baebd07deda6f2d747cb0cbe1dfe29","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-02-06 22:43:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3a3b0f81981f12b67a1b985b33dca8d4d67db5bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 22:55:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6b584185d24134b8c3017f5da3bd40\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd1e6324c89c45a5938a863820d93c8b : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10e7aaf2c8c44a81a2858bc8561f18a5 : SUCCESS in 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d4b78a6199218dfe1048734f3bf1685082c614d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 00:43:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95ab0a58789943ab8d6e8cb1076f6e9f\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/61bb81d9dd8e4262a048e67374e23513 : SUCCESS in 54m 36s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/150ec4a724374672aa101d4791a501cc : SUCCESS in 1h 24m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3364bdcb62147b28750576bee4ebe96 : SUCCESS in 52m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9441ef3c30b54deb8ccfb227942484f8 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ca29fd4494443cbae3219769b95c9e : SUCCESS in 12m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ecfd390bc5847ce96baa0ae749117d8 : SUCCESS in 11m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/695600440e0d4963b7d04eb42f9d7c79 : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77a12dae2f94f0abc0fa498ad78b517 : SUCCESS in 10m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/317634286c8f4e4381e57bbf77684cdb : SUCCESS in 45m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee2f36f77147480fa76b4d363ca2c2bd : SUCCESS in 1h 21m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/06ea1c6643b641b8969002eb21a215d3 : SUCCESS in 51m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a6c97d1db2d041adbabf80e4a21b5a35 : SUCCESS in 57m 27s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/440f21cc400844e586867b9cce7929c6 : SUCCESS in 1h 14m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/264eda006f6d4d5cb4f04228b8978804 : SUCCESS in 1h 30m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5f939d0b1bbc42449cc956565aaabf43 : SUCCESS in 43m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/27c02eb2a866416c9e595e16228f5b3e : SUCCESS in 11m 37s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59886f77c1ba4d569a3d3bf18a80766a : SUCCESS in 19m 45s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/437a9a0cfec344f5a9c14dc5177fa6d0 : SUCCESS in 22m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c6daaffe9d942b6b9258057b4e16342 : SUCCESS in 20m 15s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bbdd884776b4ec0a416b90ce2a98819 : SUCCESS in 1h 38m 57s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dd580af893254412bf4e3c11e7ade7cd : SUCCESS in 1h 15m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3d1c3538de44a08a45437fe1eda48b7 : SUCCESS in 1h 02m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4dc6f368594fe68c023388c8e42782 : SUCCESS in 39m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/654f1a0acbc340919bfa72c793d5d743 : SUCCESS in 23m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/08c4473a47374ae8a52df4bbbf179131 : SUCCESS in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0821fcc5a9cf3d6ef8f69e7672efe6cb42badd0d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-07 00:44:22.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/872204/7/check/EMC_PowerFlex_NOVA/81cb374/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"026a0c81d3b9d55963146ec28ff6953832a181b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 00:44:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"00ae1b60405392922c26e0bee2cd0b08ec43ed90","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-07 00:56:08.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/66f5b9f042124735912b3c23b9150cf7 : FAILURE in 1h 58m 05s","accounts_in_message":[],"_revision_number":7},{"id":"6c48597cce6476fc3a4f93fea67b065bbc9d9674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 02:17:41.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941b1b41622d4d2dafa2e85113778140\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d2d8f41b16e94bd6814564fe9b511425 : SUCCESS in 1h 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/495069c48dbd4e4b94376c10cc3a3c70 : SUCCESS in 42m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccdfecbb15e4100a409df492e8aa770 : SUCCESS in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/080eb333a281496eb8dac4c87fa83136 : SUCCESS in 10m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fd0568dcc8a498eba7819a1ab097315 : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/931ecb4953de48c0b38144abd0324367 : SUCCESS in 10m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/27dca8822ca6472b94adec09f5f06a64 : SUCCESS in 39m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ee1674f987f74cbc9fca24f174e652db : SUCCESS in 33m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1646e51cdf0e42c8b72f61739899fd40 : SUCCESS in 26m 26s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0a9fcb3d83284426b681fd3856f05f67 : SUCCESS in 26m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9c8d393104c04ecbb90204ab5879e2be : SUCCESS in 1h 25m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/1a66f49c420d41ba8adf0a3d43582cef : SUCCESS in 1h 30m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/006f54592fa24ade8983c258f2cfe5cd : SUCCESS in 5m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83fbc16ec7cb47188e155b9e1b7ccf4c : SUCCESS in 1h 06m 36s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1fc0123183346bb9413ccbd8965fc47 : SUCCESS in 1h 25m 42s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b72db1cddf64fe2a512531956bf543e : SUCCESS in 1h 22m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/171f91c7bc0d442a90fe75aac75656b3 : SUCCESS in 1h 09m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2671c078f9a34d0c85c4878df026735e : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":7},{"id":"ae3d3cb9316df7bdf7c350174f8a7e96b50a1632","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 02:17:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"e3ce5d4d1fcf27a7fdbd65e9c715295ded6ab0e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 02:19:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a62568cb7bb40e4b336b020400daf42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4cfdc81d2aa4b0080e6dc3b87b71f3b : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":7},{"id":"e0dd56be2d3ff2b18f91e8a2d0f4b04bf357ba90","author":{"_account_id":36932,"name":"Xiao Feng Ren","display_name":"renxiaof","email":"renxiaof@cn.ibm.com","username":"renxiaof2000"},"date":"2024-03-27 08:03:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5664d0907f96b37325af8c9a4d150bc071fd2055","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-03-27 13:42:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d892905904fd8ed7b5bb916ca4a119997ea0c9a7","revisions":{"4f5da7e9663f1141204864b8eb777b59b21e6770":{"kind":"REWORK","_number":1,"created":"2023-01-30 18:53:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/1"}}},"commit":{"parents":[{"commit":"b4105e47f16dcc9bdd8fe87e9653204b886742df","subject":"Detect host renames and abort startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4105e47f16dcc9bdd8fe87e9653204b886742df"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-30 18:53:12.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-30 18:53:44.000000000","tz":-480},"subject":"TURBOWIP: Add a host/node consistency check","message":"TURBOWIP: Add a host/node consistency check\n\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f5da7e9663f1141204864b8eb777b59b21e6770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f5da7e9663f1141204864b8eb777b59b21e6770"}]},"branch":"refs/heads/master"},"b70ab28597be38b1649ce120bf0d3e97987d35fe":{"kind":"REWORK","_number":2,"created":"2023-01-31 15:23:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/2"}}},"commit":{"parents":[{"commit":"195219a3990b76ac8aa57f0feed31bde3ed26f8a","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/195219a3990b76ac8aa57f0feed31bde3ed26f8a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:15:40.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:22:43.000000000","tz":-480},"subject":"Protect against a deleted node id file","message":"Protect against a deleted node id file\n\nIf we are starting up for the first time, we expect to generate and\nwrite a node_uuid file if one does not exist. If we are upgrading,\nwe expect to do the same. However, if we are starting up not after an\nupgrade and not for the first time, a missing compute_id file is an\nerror, and we should abort.\n\nRelated to blueprint stable-compute-uuid\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b70ab28597be38b1649ce120bf0d3e97987d35fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b70ab28597be38b1649ce120bf0d3e97987d35fe"}]},"branch":"refs/heads/master"},"cdaae36007d4d4a57c474093db3cf9672959531d":{"kind":"REWORK","_number":3,"created":"2023-01-31 19:31:50.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/3"}}},"commit":{"parents":[{"commit":"d226b6409a3ec8dc5de5fcd6e927890fed06cc3c","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d226b6409a3ec8dc5de5fcd6e927890fed06cc3c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:15:40.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 19:29:38.000000000","tz":-480},"subject":"Protect against a deleted node id file","message":"Protect against a deleted node id file\n\nIf we are starting up for the first time, we expect to generate and\nwrite a node_uuid file if one does not exist. If we are upgrading,\nwe expect to do the same. However, if we are starting up not after an\nupgrade and not for the first time, a missing compute_id file is an\nerror, and we should abort.\n\nBecause of the additional check that this adds, which is called from\na variety of places that don\u0027t have the stable compute node singleton\nstubbed to make it happy, this mocks the check for any test that does\nnot specifically aim to exercise it.\n\nRelated to blueprint stable-compute-uuid\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdaae36007d4d4a57c474093db3cf9672959531d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdaae36007d4d4a57c474093db3cf9672959531d"}]},"branch":"refs/heads/master"},"a87b98c1cca6936ea8c4cbe67a5d25761e55c86e":{"kind":"REWORK","_number":4,"created":"2023-01-31 19:55:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/4"}}},"commit":{"parents":[{"commit":"d226b6409a3ec8dc5de5fcd6e927890fed06cc3c","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d226b6409a3ec8dc5de5fcd6e927890fed06cc3c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:15:40.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 19:55:24.000000000","tz":-480},"subject":"Protect against a deleted node id file","message":"Protect against a deleted node id file\n\nIf we are starting up for the first time, we expect to generate and\nwrite a node_uuid file if one does not exist. If we are upgrading,\nwe expect to do the same. However, if we are starting up not after an\nupgrade and not for the first time, a missing compute_id file is an\nerror, and we should abort.\n\nBecause of the additional check that this adds, which is called from\na variety of places that don\u0027t have the stable compute node singleton\nstubbed to make it happy, this mocks the check for any test that does\nnot specifically aim to exercise it.\n\nRelated to blueprint stable-compute-uuid\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a87b98c1cca6936ea8c4cbe67a5d25761e55c86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a87b98c1cca6936ea8c4cbe67a5d25761e55c86e"}]},"branch":"refs/heads/master"},"82fad1f7c5d83f6b991876973d62103e15d3c9a6":{"kind":"REWORK","_number":5,"created":"2023-01-31 20:11:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/5"}}},"commit":{"parents":[{"commit":"d226b6409a3ec8dc5de5fcd6e927890fed06cc3c","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d226b6409a3ec8dc5de5fcd6e927890fed06cc3c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:15:40.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 20:11:18.000000000","tz":-480},"subject":"Protect against a deleted node id file","message":"Protect against a deleted node id file\n\nIf we are starting up for the first time, we expect to generate and\nwrite a node_uuid file if one does not exist. If we are upgrading,\nwe expect to do the same. However, if we are starting up not after an\nupgrade and not for the first time, a missing compute_id file is an\nerror, and we should abort.\n\nBecause of the additional check that this adds, which is called from\na variety of places that don\u0027t have the stable compute node singleton\nstubbed to make it happy, this mocks the check for any test that does\nnot specifically aim to exercise it.\n\nRelated to blueprint stable-compute-uuid\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82fad1f7c5d83f6b991876973d62103e15d3c9a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82fad1f7c5d83f6b991876973d62103e15d3c9a6"}]},"branch":"refs/heads/master"},"72f478065a6ead0068908819a40f4439b7eb8c1a":{"kind":"REWORK","_number":6,"created":"2023-02-01 16:47:06.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/6"}}},"commit":{"parents":[{"commit":"d226b6409a3ec8dc5de5fcd6e927890fed06cc3c","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d226b6409a3ec8dc5de5fcd6e927890fed06cc3c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:15:40.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-02-01 16:31:09.000000000","tz":-480},"subject":"Protect against a deleted node id file","message":"Protect against a deleted node id file\n\nIf we are starting up for the first time, we expect to generate and\nwrite a node_uuid file if one does not exist. If we are upgrading,\nwe expect to do the same. However, if we are starting up not after an\nupgrade and not for the first time, a missing compute_id file is an\nerror, and we should abort.\n\nBecause of the additional check that this adds, which is called from\na variety of places that don\u0027t have the stable compute node singleton\nstubbed to make it happy, this mocks the check for any test that does\nnot specifically aim to exercise it.\n\nRelated to blueprint stable-compute-uuid\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72f478065a6ead0068908819a40f4439b7eb8c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72f478065a6ead0068908819a40f4439b7eb8c1a"}]},"branch":"refs/heads/master"},"d892905904fd8ed7b5bb916ca4a119997ea0c9a7":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-01 17:23:42.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/872204/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872204/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872204/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872204/7"}}},"commit":{"parents":[{"commit":"72370a188c0755bc9c864b5a5e4a972077cb8dd6","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72370a188c0755bc9c864b5a5e4a972077cb8dd6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-01-31 15:15:40.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-02-01 17:20:59.000000000","tz":-480},"subject":"Protect against a deleted node id file","message":"Protect against a deleted node id file\n\nIf we are starting up for the first time, we expect to generate and\nwrite a node_uuid file if one does not exist. If we are upgrading,\nwe expect to do the same. However, if we are starting up not after an\nupgrade and not for the first time, a missing compute_id file is an\nerror, and we should abort.\n\nBecause of the additional check that this adds, which is called from\na variety of places that don\u0027t have the stable compute node singleton\nstubbed to make it happy, this mocks the check for any test that does\nnot specifically aim to exercise it.\n\nRelated to blueprint stable-compute-uuid\nChange-Id: If83ce14b96e7d84ae38eba9d798754557d5abdfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d892905904fd8ed7b5bb916ca4a119997ea0c9a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d892905904fd8ed7b5bb916ca4a119997ea0c9a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
