)]}'
{"id":"openstack%2Fnova~872804","triplet_id":"openstack%2Fnova~master~I0a1e47baf9f8115cf1d02d57659a88b26eae89d0","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-02-07 11:06:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a1e47baf9f8115cf1d02d57659a88b26eae89d0","subject":"Fix 6.2 compute RPC version alias","status":"MERGED","created":"2023-02-06 21:26:39.000000000","updated":"2023-02-28 08:18:34.000000000","submitted":"2023-02-07 11:06:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872804","meta_rev_id":"56a963f001914a1700e02b721088187d08320751","_number":872804,"virtual_id_number":872804,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 11:06:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2023-02-06 21:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-07 09:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-07 09:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-06 21:47:32.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2023-02-06 22:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-06 23:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-06 23:22:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-07 00:22:03.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-07 00:30:04.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-07 09:02:51.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2023-02-28 08:18:34.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"5154800577feb5e5871e4a8bea1117bace4d4872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-06 21:26:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3eb00534af3d527437d468e4473c36b4e867e869","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-02-06 21:47:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3efa683d1a24a7acdcd52c2812b5dbb490577919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 22:02:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b0953916f524ff590db48aea9273f43\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12f056b875ab4b4089c392409b2d303b : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08b42a4fb45146dbaa8fe211b958355c : SUCCESS in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"323c600891db0ecb7f2aa675ff56c3ae87e748d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 23:22:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/584f0833fc404a458e76ec0909492408\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/be18d181b7b64d30ac73749600d60c93 : SUCCESS in 48m 58s (non-voting)\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/59cd167375874c4baa6c54048daa7d33 : SUCCESS in 1h 24m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3e4ae670036477583380209efc92bcb : SUCCESS in 1h 47m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eec15988fb3c493dac7574039f08a7f2 : SUCCESS in 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b2698aca6b426fa3158cebcd6342ca : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e8491e95d114df9866d14ef907bf804 : SUCCESS in 11m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1580b8cf433148beb5231ffb3430e896 : SUCCESS in 22m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1087ca68f984cfdbfc9bef72cfd29fe : SUCCESS in 11m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/321ec635f44b4895b64cb45feff8282f : SUCCESS in 1h 22m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4ed41e6d7364e68b41c46b57d13b91a : SUCCESS in 54m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dcab4a1530da44cfb439fcf1e67d1775 : SUCCESS in 1h 11m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9decb445576442eabb94f887872e5b3f : SUCCESS in 41m 04s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/273f995af6784d83884d441ca9ba6e26 : SUCCESS in 1h 44m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/0b754bdf01bd430faf0eddc0676cd7ab : SUCCESS in 1h 20m 30s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2a7e928d68d24325a3b90f14b4d97290 : SUCCESS in 50m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a2c35df7b1c14760bdf352dda4ebc186 : SUCCESS in 4m 35s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/adf2a2c5284e4d9f8a67dbb80dd21d32 : SUCCESS in 22m 55s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec39409238224c30b111b35e43e64569 : SUCCESS in 21m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/766c8ea1c2e0496b8e5a29365eda3e98 : SUCCESS in 23m 10s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a4f003dbe78494a80414a48253bc5eb : SUCCESS in 1h 26m 36s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cad7fb2234bb43b38fc246ef9a686c50 : SUCCESS in 1h 23m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5625abc1305474a87786934bbcb2dad : SUCCESS in 1h 13m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1164f5d094274019ac001360d40c2b11 : SUCCESS in 45m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e4810252221f4d0c85975b853396f87c : SUCCESS in 35m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c2b2b204b2f2479bafc8f17c77d4dfd5 : SUCCESS in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97f1de337ab4ce5569826879ee63031679a4232e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-02-06 23:22:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/04/872804/1/check/EMC_PowerFlex_NOVA/fcfb5da/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a52c9a586d58aa5759982b1bb82aeba369900275","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-02-07 00:22:03.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/73a8ac52fc1641fdae83507f4efc98fa : SUCCESS in 2h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"20ebe95880941d88cbd3393f7408de5fd12370ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-02-07 00:30:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/04/872804/1/openstack-sriov-nova-master-macvtap : ABORTED\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/04/872804/1/openstack-sriov-nova-master-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"0323806f720e0e38fdb3d8f2a4df25bcd968159b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-02-07 00:37:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/04/872804/1/openstack-sriov-nova-master-macvtap : ABORTED\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/04/872804/1/openstack-sriov-nova-master-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"06fd88d849d408f1895f6a348aeea489edc444c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-02-07 00:38:25.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/04/872804/1/openstack-sriov-nova-master-macvtap : FAILURE\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/04/872804/1/openstack-sriov-nova-master-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ff6db453e03a31e510780fc37740940a73e252d2","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-07 09:02:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2300a16f1699a11258fa6d70d4c70b4e3f126edf","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-07 09:03:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b6bc52b7e93023633f71b135a3fce35c5958206","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 09:03:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c210296c8cb920e5d18b70e18049d49c7e4cde6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:06:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/076c197ff3cb410ab545070a67aa17df\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2181b277639248e5b71d2e29c898896f : SUCCESS in 1h 26m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4017af87f04f424aad5e429319bf023c : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cdaeace362148cea8731da7c72179c0 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97a5c6408b2e473b81063786133244d3 : SUCCESS in 12m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c006012fa10f4a72b91c7df0af658579 : SUCCESS in 15m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1de9e7fc3048949467b1b21f80e6d1 : SUCCESS in 10m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/90788b7419f440a8be7c9c1e8f53c83b : SUCCESS in 36m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/29f5ce768e3a4cd48935ee9d785968d4 : SUCCESS in 48m 47s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8d9672b30ea4cd590c5b2825fc2d35e : SUCCESS in 20m 37s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6fb065dc38445f99ad1352dabbd4bc5 : SUCCESS in 26m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/24198209e0194c89af85b165e6f4ac17 : SUCCESS in 1h 35m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/eb6865cfbed94830a4da34be3f9fbf8a : SUCCESS in 1h 57m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1d2945c1b82445c9b2bbfc5f6dc7789d : SUCCESS in 5m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b15cb784a69e4990bcbcfba09f445f8f : SUCCESS in 1h 06m 30s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b21bfdbf6fde41e6beb57f3db8320538 : SUCCESS in 1h 28m 34s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e26afe2b6ca64076afcdbaab14d7703d : SUCCESS in 1h 26m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e00ab16a77c4ba5a5b89b841d4b8542 : SUCCESS in 1h 03m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80b0a52fbeb14e86a952916166242fb9 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"869867a9a539321348a982d41db2655c0e5eb743","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:06:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ccd05ebc4c7e948850d8a6b3024e71a3d3cca51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 11:08:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383e654b79e1457ebe1335ff99d1934a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/041fc79b0bbf4c3f81ea89ec4ef9aaaf : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1},{"id":"56a963f001914a1700e02b721088187d08320751","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-02-28 08:18:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56b8bb3fb97b45986d468ffdd7574ff441ad8b11","revisions":{"56b8bb3fb97b45986d468ffdd7574ff441ad8b11":{"kind":"REWORK","_number":1,"created":"2023-02-06 21:26:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/04/872804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/872804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/872804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/872804/1"}}},"commit":{"parents":[{"commit":"72370a188c0755bc9c864b5a5e4a972077cb8dd6","subject":"Check our nodes for hypervisor_hostname changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72370a188c0755bc9c864b5a5e4a972077cb8dd6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-02-06 21:26:02.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-02-06 21:26:02.000000000","tz":60},"subject":"Fix 6.2 compute RPC version alias","message":"Fix 6.2 compute RPC version alias\n\nThe alias for the 6.2 compute RPC version says\nantilope but that is spelled incorrectly and\nshould be antelope.\n\nChange-Id: I0a1e47baf9f8115cf1d02d57659a88b26eae89d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56b8bb3fb97b45986d468ffdd7574ff441ad8b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56b8bb3fb97b45986d468ffdd7574ff441ad8b11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
