)]}'
{"id":"openstack%2Fnova~883411","triplet_id":"openstack%2Fnova~master~I60971b58cf1f24bdb5d40f668e380ebfee2ac495","project":"openstack/nova","branch":"master","topic":"ironic-fix","attention_set":{},"removed_from_attention_set":{"37620":{"account":{"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},"last_update":"2025-01-30 00:31:34.000000000","reason":"Change was submitted"},"29268":{"account":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"last_update":"2025-01-30 00:31:34.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-29 11:59:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-01-30 00:31:34.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-07-24 15:33:10.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-01-30 00:31:34.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2025-01-30 00:31:34.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2025-01-30 00:31:34.000000000","reason":"Change was submitted"},"6799":{"account":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"last_update":"2025-01-29 23:57:34.000000000","reason":"\u003cGERRIT_ACCOUNT_6799\u003e replied on the change","reason_account":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}}},"hashtags":[],"change_id":"I60971b58cf1f24bdb5d40f668e380ebfee2ac495","subject":"ironic: Fix ConflictException when deleting server","status":"MERGED","created":"2023-05-17 15:40:32.000000000","updated":"2025-02-06 02:29:34.000000000","submitted":"2025-01-30 00:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"883411-ironic-fix","meta_rev_id":"84b5240a2e055f73ee399c60cfefad098d324164","_number":883411,"virtual_id_number":883411,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},{"value":0,"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-30 00:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2025-01-29 22:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-01-29 11:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},{"value":0,"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-29 23:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2025-01-29 22:24:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},{"value":0,"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},{"value":0,"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},{"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"}],"CC":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 15:41:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-05-17 15:41:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2023-05-17 15:41:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2023-05-17 15:46:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-17 18:09:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 10:41:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2023-05-18 11:31:44.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-05 18:32:49.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-07-24 09:13:31.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-12-19 20:19:50.000000000","updated_by":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"reviewer":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"state":"CC"},{"updated":"2025-01-21 18:32:25.000000000","updated_by":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"reviewer":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"state":"REVIEWER"},{"updated":"2025-01-23 01:39:16.000000000","updated_by":{"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},"reviewer":{"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},"state":"REVIEWER"},{"updated":"2025-01-23 12:10:11.000000000","updated_by":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"reviewer":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"state":"REVIEWER"},{"updated":"2025-01-29 22:24:00.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-01-30 13:53:26.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-06 02:29:34.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"CC"}],"messages":[{"id":"b9ef7b1e92e6712545dde8b60705cca084c6f49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-17 15:40:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e66e812ec7cf5bec0c26bd47a360e7af893b8cdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 15:46:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/733bb0e054d242028c0951133201937e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081dd5d773ef42e1b74f634ec00cce8a : NODE_FAILURE Node request 300-0021235126 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c79e60562546462cb84b2fc3ae9a2850 : NODE_FAILURE Node request 300-0021235127 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef9035641a824d1f838b5ff0bc74cfd0 : NODE_FAILURE Node request 300-0021235128 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4ff969ca2ae514cd858de6481c4eae52d0d9f66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 18:09:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d5fdcd0f5fc4436bbe3c98cf9d7a15d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a33e050cc0747a092da2addecab5e32 : SUCCESS in 58m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/239a61541233446d837e01e9d2d7d2a0 : SUCCESS in 1h 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/697245ff3fee444890eb0b9a98182989 : SUCCESS in 55m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9ab8ca9bb6e49038c9acdaa99e0aae5 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f36b71a67b8419f93717fcfe71d3a6c : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bbfc0776a4b47aea77183b365bae1a0 : SUCCESS in 16m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4bd092bdb8d43c9b1e3bc82d75fa14d : SUCCESS in 14m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44f08581279e4612832b5ba0f2b80734 : SUCCESS in 14m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3efc8770b44e459db9f796c78c7420f9 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eb5783665cd432998de396a52341fec : SUCCESS in 16m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f9598b96d5334cd0a52e24f7a7bcaa7f : SUCCESS in 47m 57s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dccfd5dc9ba74f57861d672be9063669 : SUCCESS in 1h 51m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6b2ce9307011443aa22637a400d51cd4 : SUCCESS in 1h 02m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dbb69a51f52f4f3688dc5bcb1bef1e99 : FAILURE in 1h 41m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/39e739b7099247a587b3955b1f40fe0a : SUCCESS in 2h 17m 34s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9693a41426cf43d2abfbcfe8069e5a7d : SUCCESS in 59m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c13c59cd28704d7db097f58210d5d90b : FAILURE in 4m 29s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6bf1bf9fc9df4b2085d1fd4227dee88a : SUCCESS in 23m 13s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2bb188bdd991430b8dcbf17cb994b951 : SUCCESS in 26m 53s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2eec90e97525448184e7e07a7815a482 : SUCCESS in 22m 57s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f8d8722b395c48759a98c3c8b9d66bed : SUCCESS in 1h 51m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cf21b391070472fbe837f452ec9bb1e : SUCCESS in 1h 12m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9407647c32f94b5fbb95474dea4e07b0 : SUCCESS in 42m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/05b825c5f14e4765a625f6b59d5d1222 : SUCCESS in 23m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5427d3afc41e4ea9adf879f8307fa0e1 : SUCCESS in 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"189f7f0ef94b9dd870a9f83e13bb474bdee7bf55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-18 08:22:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1221eb7c01a2621c3800680110b38068e0aa3547","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:00:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a2840c768d84650b1c18d2bff2c7555\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/117fbb81ce9043cbbab5b6c3f1e2ab48 : SUCCESS in 31m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15cfbecb1ce44a8ea73ed36b70f705d6 : SUCCESS in 30m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49f3ab094319448f9e5ce42b0964a2db : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae29b19d38eb42c5040cffc0fce7025aefc29153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8663d923c46540e3b58e3105a54ce4e4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f595f6e9a72440286b9eb7f51129458 : SUCCESS in 1h 05m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9e532b622edf406ab2bf7ad2686c3da8 : SUCCESS in 1h 32m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f76112d3d22471d8b0f9b01687ca2f1 : SUCCESS in 1h 01m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7b26d906334a52ac4b9ceb1613ea39 : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/924cc209289b4735b3544f4448bc1d11 : SUCCESS in 12m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd96321971c24957aeac846f1f18e021 : SUCCESS in 13m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29dea0cb35e54f1e9528c71f35cf7141 : SUCCESS in 13m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b096307aa0e94607a8bbe5b32ed821a9 : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593570d9bdfd4c72b59a8dffecebd4d6 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06d66d3b23764041ba3950b4bdb16b36 : SUCCESS in 28m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f65f670818a841c5b1aed47b29e28bde : SUCCESS in 1h 04m 15s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/763bca7ff1b94ce299f6dcfedc524d47 : SUCCESS in 1h 41m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bde98703ca574f9b9c1fb75e57b57a2e : FAILURE in 51m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2c7f8bb6a93a4d1e87a06cbad827c32e : SUCCESS in 1h 41m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/b56f120e51614dd79f9d1b697c8e4163 : SUCCESS in 1h 41m 58s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/0b0e53cff1c94bb2a1b6827720d13a73 : SUCCESS in 46m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dd99fba171204f578b07eb0f13c66a8c : FAILURE in 5m 30s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b7df290af4b4006b7a5102b41f515cf : SUCCESS in 24m 59s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1321ae7c66a449519a43568b9cfcd7d8 : SUCCESS in 25m 21s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d36b528ed40c42ca923ea0c512d9f16e : SUCCESS in 23m 32s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c4975fb829a4ebc84f47d24db6289ea : SUCCESS in 1h 19m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/241c16e2ef2b4c8ea857d814461cd192 : SUCCESS in 1h 05m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e98085460e2453abb89786485587552 : SUCCESS in 46m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/23a60751435f49d2bf92b35069e9005b : SUCCESS in 27m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9fa47518d0414c28be2e27714f93c86b : SUCCESS in 31m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a35608ac945fb3ed93aa813745d8a340ebbcc2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2023-05-18 10:41:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"4c41884fd9fa967a4a8c036399b6ba35ff951d0b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2023-05-18 11:31:44.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/6e83414fa39b4422b64ceae4b80f0927 : RETRY_LIMIT in 1m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a04a8ec4cb2f79f7472351250b48f3b52e2664d2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-06-05 17:10:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20cffe307547ce3a7fc3b0a2c5ae749eb0b4c692","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-06-05 18:32:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f5bf25548361da1aae050911b1c3a2728abae683","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-04 18:39:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00540de69a541a014b37ccbc89ee89966fcefe02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:39:58.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 883411,1","accounts_in_message":[],"_revision_number":1},{"id":"2e4ce6ab8f791000189e0d94011929f747220ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:39:59.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 883411,1","accounts_in_message":[],"_revision_number":1},{"id":"13b00fed82afc283a9b283758a90f41665538999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-24 09:02:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8c217426770c3e1a895b61c76a7d2f3f3a8ed95b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-24 09:02:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93b5fabac34f7f57eb502fa2b6316f2ce97cde03","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-24 09:13:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4e105b531e609747d2a50309b01c8e340f45aad4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-24 09:18:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3a5f5e839cd6261e59a37c3301203d4edb1f1ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:38:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce6546a6aaf4574931c963224386006\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7da97580b6b94b11ae5a666140b7e361 : SUCCESS in 33m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a39b9e6a1e04a76b95c3af46672ef6f : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b453c10cdda4524696fc94d2d44ed1a901920954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:04:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d6262e51774530b2ad64299bc1b8ec\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c1d5468d65ee413fa9e4d967063d6418 : SUCCESS in 1h 03m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/71fc91d46679446a8250fe3929a039bf : SUCCESS in 1h 39m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd9a375b00a5412eb157bccb6a472004 : SUCCESS in 59m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f14475e6e48d4384ad089a710321912b : SUCCESS in 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218bfa0f3a334f318c931afe79821050 : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac2c1a965a4c46069941bd5f53d9bb46 : SUCCESS in 13m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c859c89365204efabc49778a27337c74 : SUCCESS in 11m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5acdee4752d0457ea3ab0b982f92f1a9 : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e7b9bfcf2742798af1ec85ae08660f : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba33ee2431e4407bb15b851cf91cbc61 : SUCCESS in 33m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fdcc957342e646aa940765ddce7621f5 : FAILURE in 18m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1779482eb9034e56b1e229fa8c54fcb9 : SUCCESS in 1h 49m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/961a3624a6c244dda8e96572dcb31af1 : SUCCESS in 39m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6031138f18754e9eab66499827709f5a : SUCCESS in 1h 53m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/4629ae0ab83f4d84ad969c8989f6bd99 : SUCCESS in 1h 28m 01s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6fd4382de5fe49e6adf0527f59b123b8 : SUCCESS in 55m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/088e5ef325ec4823a4b6caad550154f3 : FAILURE in 5m 25s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dae0c45a4ee840309e3c0b5cde3b6919 : SUCCESS in 25m 17s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c2995541bab45fb9f5205f69c3bb097 : SUCCESS in 24m 37s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/106a386d9ac0419a8599a7719772acf2 : SUCCESS in 35m 40s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/709d22b10a8c43e5be814d544d4948ab : SUCCESS in 1h 49m 12s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/7a97c06f316d4a71b2d2da9573bb322e : SUCCESS in 1h 53m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/883f1c29c7504952b657ad8a22aa9d85 : SUCCESS in 1h 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6dd57e4abc14493a48c8175667756f2 : SUCCESS in 1h 04m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/74191f1f2b3b4405aa4cfb3ac66f39a5 : SUCCESS in 22m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e5507c4eac024fc4aaa3cd43d7a26396 : SUCCESS in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4b62a8a485bf7e46c3a72fe78bc6de62574cc67","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-07-24 11:10:21.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"4c3db080ea6838692ad1fe8b0e617e9b825096e1","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-24 15:33:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"71982408cc55ffa3901c72c1f81a5eee954207e9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-07-24 18:38:28.000000000","message":"Patch Set 2: -Code-Review\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ba57e604913172ede9b3d19289b3c8eafc44cae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"real_author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-21 18:20:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_14826\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"_revision_number":3},{"id":"1060bcde91b9bf2f95a191249c159773810667d7","author":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"date":"2025-01-21 18:32:25.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2294e9a35e2c52bc1bd6e2b460ba4c8be36e6784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 21:02:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4542c70c06fa4402b33b5ce49ace8d60\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b87c37bf2983403e986889f74e15f067 : SUCCESS in 1h 14m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5844357dac7d4d318471d07158e74758 : SUCCESS in 1h 56m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a117f37d7a44790916933e852ee2176 : SUCCESS in 1h 11m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee63848a1d7e4fc7875efd7911e556a8 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72949fd85114c0ca45ffc80d7d7498d : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f6cc7c67ee34c33bdf9110b4f00304d : SUCCESS in 12m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/746bdb493ded4498b8cb679aa64068c4 : SUCCESS in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/191a675152b04ca3b257d803be1d3a12 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86062105f6c449b9a905aae82c5886aa : SUCCESS in 25m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/040575317ed94f3aa5a1530412cfcb2f : SUCCESS in 36m 13s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2117a49e23e540bc9620309622428b5e : SUCCESS in 1h 50m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9e8057bb74a14788909d1f29903c3991 : SUCCESS in 33m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/27fc7485daff45a29b5ee5d86e1de1a8 : SUCCESS in 2h 00m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/ba65f1e532634877ae3a41cc3bb1a60a : SUCCESS in 2h 34m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/62b1bada00a54ad680d82bd845a30a53 : SUCCESS in 55m 45s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b17135732e6d41e998493d51a33e566d : FAILURE in 5m 03s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/944b3e6dc6e141deaa2707251aa30cca : SUCCESS in 26m 55s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/576244d27f52498ebc669c04db5d1b07 : SUCCESS in 33m 16s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2a188767d6434783b337150d02a688f8 : SUCCESS in 29m 32s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd501f9362a9444ba9d1fd47ebf5b94f : SUCCESS in 1h 46m 57s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/13f474a51539483499bde846655351a4 : SUCCESS in 2h 03m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e44d43e790b44ab977c3534c38146cc : SUCCESS in 1h 23m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3364defa3f44e0b8a62c354e2b88761 : SUCCESS in 58m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1527d4ad310d4713a18ca172d8ecabad : SUCCESS in 32m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1afc46c99db542689b7a80f70622d0e5 : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58da25b2b0c0d95eb6702958b0e897e7df115dbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 21:04:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4376759095a1495c8e1207ce8e41572b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fbb094a3c424bb1a457b872852f8c10 : SUCCESS in 34m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d32742b048934444a5e8454b417f5921 : SUCCESS in 31m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dbe74a9c048ae2820387482a6dd55510259965cd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-21 21:07:01.000000000","message":"Patch Set 3:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"ab729c703be531456eabf934acb75379bdbe8d81","author":{"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},"date":"2025-01-23 01:39:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"329bd4e3c725c206692ace4e8c6cb3c892559e6d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-23 10:07:52.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f611b04625cfb9d1eb4db38d02674ffe7e188523","author":{"_account_id":29268,"name":"Matthew Heler","email":"matthew.heler@hotmail.com","username":"mheler"},"date":"2025-01-23 12:10:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c6c633944a211a13f6a937d4056aa4bb90e3f57e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"real_author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-01-29 09:57:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_14826\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"_revision_number":4},{"id":"3cc4612e1349268e345046b10e4779cbf92e497c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2025-01-29 11:11:56.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"bc2a77ee363f14d1e7e543cec8f589d335163d42","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-29 11:59:29.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d24adc26d994911778494382a4e702f7fcb3b9fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 12:56:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf2888938214aef8ca511a64c1cc71a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4bbcd8a6f9046c4b49ca1904bddffd2 : FAILURE in 39m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b5a57366e9d4e04abd3ae2dd9e424bf : FAILURE in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"647657c90c6abf8db05ab3bd6007dfe1de293a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 13:46:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33405f460fe04441813077a20a1b82c8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be47cd47833446ad9f327ad244a0e32b : SUCCESS in 59m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a75b177407d046a8ad4df9c75ad41c44 : SUCCESS in 1h 54m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c695d3c9d01f4b68b9ce098b6441db59 : SUCCESS in 1h 12m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3d9461ca13c4b2c937b6ec7819e130c : SUCCESS in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee017815f56456bbeb30a97d6ee26f1 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb497199aec1419ca13bd36ebedaff9d : SUCCESS in 14m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/700e6f710ca147148c529d4d9158d7c6 : SUCCESS in 13m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7a4fd0f10a4f81a0414fb2058b96dd : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4971355339c40cdb5ac6e3f7539c9c0 : SUCCESS in 29m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/78499089d25c44ca8dabb235bc3caf5a : SUCCESS in 52m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/706b43d6c50846b08065e05d1c7b7bef : SUCCESS in 1h 41m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1f7cf7105d8442f09708e86ad057e5cb : SUCCESS in 46m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/831566e6b24a40d4b74fb395bc9ffb33 : SUCCESS in 1h 51m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/210c06d01a8c45e29c073f8461590d1a : SUCCESS in 2h 18m 00s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b60b95ffeb5a42e5af1deac32c477b92 : SUCCESS in 42m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bfbdd8f27a4d4fc2a5435f7f7b07e6ef : SUCCESS in 5m 31s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e8e361a64a548c7b1c9d0ee0eac7f13 : SUCCESS in 27m 01s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f9d2f007a5aa47b7997977b1eb37d3f5 : SUCCESS in 28m 45s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c04c7d4f3cf4915bb7ec766962cd321 : SUCCESS in 34m 52s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a196452d82724d1dbb506c26550ac5dd : SUCCESS in 1h 20m 57s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/857e08b8ea30489492ebef5c0e121d69 : SUCCESS in 1h 55m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a24cab4da414c32a7aabc34c9ef6642 : SUCCESS in 1h 24m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9042043c31414ab8a97725eec519f310 : SUCCESS in 42m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/369c63a3936f4acd940ee0e918c192a9 : SUCCESS in 34m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/28b5297867cf4b5ca6a88f53e0c5713d : SUCCESS in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e619700b2390591566c77eb4250f65c393182d43","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-29 13:51:27.000000000","message":"Patch Set 5:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"7e4a5e06420d19f2a30494a4e9623e834d2c604a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-29 22:24:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"261f5ef7637820e03a3d381e4cb8bf53adff59a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 22:24:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9d9cbd8b9b8557003b41238734de420a53060d95","author":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"},"date":"2025-01-29 23:57:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7f3471b644ccb85bf99ae3da0964ac9d8855ff13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 00:31:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195c1b9b89eb49f39dc21c3b07516cb0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d1370687bb14582bccab2127d26cef2 : SUCCESS in 1h 04m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/317e70e8ebfc46afaf36fb86195b6dfb : SUCCESS in 2h 02m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1345e6b2e0b448eb4b07dbd997c771d : SUCCESS in 1h 03m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07675291e7bf45c09ef62461868d5604 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c79b9df6df148b7a07ba821e3780a4b : SUCCESS in 13m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f90a41dc45b5430aad9dc1e6f825c625 : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da8a64847934e03b42d0fbfb7fe561d : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2389d4bc36ea466580133ccfac7ed185 : SUCCESS in 26m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21f0a07d57104f36a1b0710560327e35 : SUCCESS in 26m 01s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3094a1c90b4249ba92fd2f8d78ba939f : SUCCESS in 21m 30s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fc4a0de5b5634f4882920ad9e354b1b4 : SUCCESS in 23m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/46256f76f47d47b081af114addb791b0 : SUCCESS in 1h 33m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/4713f4c4af234e9ebbee41146baa8ebf : SUCCESS in 2h 04m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/199c0efce0294f559bfb27cbb787b3b3 : SUCCESS in 5m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5f65b33f76a493fa5106560c084507d : SUCCESS in 1h 23m 53s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3cbe474fb4424b80b0f2e8da760e6a5a : SUCCESS in 1h 57m 59s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/6fd5c18cc6f74e72bc7cb65c595b58ff : SUCCESS in 2h 00m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/239ac658f3044bd69862163b8f7e0d13 : SUCCESS in 1h 15m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3392a05f79d42ddb79d4024c37b1965 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":5},{"id":"51511f236637ac580c87d0ace8587d343b49a55f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 00:31:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"aac91d839aefedd255bee66e9bab23ece1e9d7a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 00:32:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a3bfd3fed964f12971916467191f4d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e328ccc144874a498c53f886fa39e307 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2544a69bf4b3427789709d99143115ee : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5},{"id":"c9b03568e82984dddc947f3b7952e310956ce8a3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-30 13:53:26.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-883411-p2hdm/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"410bad19d34a5f683ce3ac39d0347c63df633c52","author":{"_account_id":37620,"name":"suiong ng","display_name":"Suiong Ng","email":"YoungN@supermicro.com","username":"suiong_ng"},"date":"2025-02-03 05:52:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"84b5240a2e055f73ee399c60cfefad098d324164","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-06 02:29:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6ebd8a56d1c79fc15f385f232ce0e22937b2fdac","revisions":{"2bf0f888180e3e7c617f9d4054f2306f0833cffc":{"kind":"REWORK","_number":1,"created":"2023-05-17 15:40:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/883411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/883411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/883411/1"}}},"commit":{"parents":[{"commit":"b3fdd7ccf01bafb68e37a457f703b79119dbfa86","subject":"Merge \"doc: Update version info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3fdd7ccf01bafb68e37a457f703b79119dbfa86"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-17 13:20:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-17 15:40:09.000000000","tz":60},"subject":"ironic: Let Ironic handle deployment cleanup actions during destroy","message":"ironic: Let Ironic handle deployment cleanup actions during destroy\n\nFor bare metal (Ironic) instances, we previously performed some cleanup\noperations on the Ironic node after deprovisioning the node, including\nunplugging VIFs and removing volume targets. Prior to\nIea337f73c231db2cb9d9f639b92475daaede6793, we also reset the\ninstance_info and instance_uuid fields.\n\nWhen Ironic automated cleaning is enabled, these cleanup actions race\nwith Ironic, and may result in an HTTP 409 Conflict and a number of\nretries, due to Ironic holding the node\u0027s lock or the node being in a\nstate transition. In the extreme case, if retries are exhausted, the\ncleanup operation may fail, leading to the instance moving to an ERROR\nstate.\n\nThe VIF unplug and volume target delete cleanup operations are actually\nunnecessary once a node has been deprovisioned, since they are performed\nas part of the node\u0027s tear_down routine [1][2]. This change removes\nthese cleanup operations in that case, retaining them only when the\nprovision state of the node was never changed (i.e. spawn failed early).\n\n[1] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/conductor/manager.py#L1117\n[2] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/drivers/modules/agent_base.py#L685\n\nCloses-Bug: #2019977\nRelated-Bug: #1815799\nChange-Id: I60971b58cf1f24bdb5d40f668e380ebfee2ac495\n(cherry picked from commit 010f5aafbf9375c7740fe239a9a76df4a3264ca2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bf0f888180e3e7c617f9d4054f2306f0833cffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bf0f888180e3e7c617f9d4054f2306f0833cffc"}]},"branch":"refs/heads/master"},"d56b7e639866246b6a2f3b03804c1956e4e9175a":{"kind":"REWORK","_number":2,"created":"2024-07-24 09:02:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/883411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/883411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/883411/2"}}},"commit":{"parents":[{"commit":"8c4a2e1b4759b9932bcad10810143e61ec9b4c36","subject":"Merge \"tests: Ensure all APIs have a request query schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c4a2e1b4759b9932bcad10810143e61ec9b4c36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-17 13:20:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-07-24 09:00:44.000000000","tz":60},"subject":"ironic: Let Ironic handle deployment cleanup actions during destroy","message":"ironic: Let Ironic handle deployment cleanup actions during destroy\n\nFor bare metal (Ironic) instances, we previously performed some cleanup\noperations on the Ironic node after deprovisioning the node, including\nunplugging VIFs and removing volume targets. Prior to\nIea337f73c231db2cb9d9f639b92475daaede6793, we also reset the\ninstance_info and instance_uuid fields.\n\nWhen Ironic automated cleaning is enabled, these cleanup actions race\nwith Ironic, and may result in an HTTP 409 Conflict and a number of\nretries, due to Ironic holding the node\u0027s lock or the node being in a\nstate transition. In the extreme case, if retries are exhausted, the\ncleanup operation may fail, leading to the instance moving to an ERROR\nstate.\n\nThe VIF unplug and volume target delete cleanup operations are actually\nunnecessary once a node has been deprovisioned, since they are performed\nas part of the node\u0027s tear_down routine [1][2]. This change removes\nthese cleanup operations in that case, retaining them only when the\nprovision state of the node was never changed (i.e. spawn failed early).\n\n[1] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/conductor/manager.py#L1117\n[2] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/drivers/modules/agent_base.py#L685\n\nCloses-Bug: #2019977\nRelated-Bug: #1815799\nChange-Id: I60971b58cf1f24bdb5d40f668e380ebfee2ac495\n(cherry picked from commit 010f5aafbf9375c7740fe239a9a76df4a3264ca2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d56b7e639866246b6a2f3b03804c1956e4e9175a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d56b7e639866246b6a2f3b03804c1956e4e9175a"}]},"branch":"refs/heads/master"},"9acb7182bd2b2cc7aaa4093f8230f280d4f94a9e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-01-21 18:20:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"real_uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/11/883411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/883411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/883411/3"}}},"commit":{"parents":[{"commit":"3d21445b73c2cbe044d6086c2e9d8f276c2c736d","subject":"Merge \"Add whitebox-devstack-multinode job to periodic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d21445b73c2cbe044d6086c2e9d8f276c2c736d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-17 13:20:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2025-01-21 18:20:23.000000000","tz":0},"subject":"ironic: Let Ironic handle deployment cleanup actions during destroy","message":"ironic: Let Ironic handle deployment cleanup actions during destroy\n\nFor bare metal (Ironic) instances, we previously performed some cleanup\noperations on the Ironic node after deprovisioning the node, including\nunplugging VIFs and removing volume targets. Prior to\nIea337f73c231db2cb9d9f639b92475daaede6793, we also reset the\ninstance_info and instance_uuid fields.\n\nWhen Ironic automated cleaning is enabled, these cleanup actions race\nwith Ironic, and may result in an HTTP 409 Conflict and a number of\nretries, due to Ironic holding the node\u0027s lock or the node being in a\nstate transition. In the extreme case, if retries are exhausted, the\ncleanup operation may fail, leading to the instance moving to an ERROR\nstate.\n\nThe VIF unplug and volume target delete cleanup operations are actually\nunnecessary once a node has been deprovisioned, since they are performed\nas part of the node\u0027s tear_down routine [1][2]. This change removes\nthese cleanup operations in that case, retaining them only when the\nprovision state of the node was never changed (i.e. spawn failed early).\n\n[1] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/conductor/manager.py#L1117\n[2] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/drivers/modules/agent_base.py#L685\n\nCloses-Bug: #2019977\nRelated-Bug: #1815799\nChange-Id: I60971b58cf1f24bdb5d40f668e380ebfee2ac495\n(cherry picked from commit 010f5aafbf9375c7740fe239a9a76df4a3264ca2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9acb7182bd2b2cc7aaa4093f8230f280d4f94a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9acb7182bd2b2cc7aaa4093f8230f280d4f94a9e"}]},"branch":"refs/heads/master","description":"Rebase"},"48097b8455710edcd2f0680ff8f41162f5755be4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-29 09:57:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"real_uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/11/883411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/883411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/883411/4"}}},"commit":{"parents":[{"commit":"8c953d4d257b05009111fdd048ea7c05f85c0247","subject":"Merge \"zuul: Add missing context comments for nova-next\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c953d4d257b05009111fdd048ea7c05f85c0247"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-17 13:20:54.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2025-01-29 09:57:36.000000000","tz":0},"subject":"ironic: Let Ironic handle deployment cleanup actions during destroy","message":"ironic: Let Ironic handle deployment cleanup actions during destroy\n\nFor bare metal (Ironic) instances, we previously performed some cleanup\noperations on the Ironic node after deprovisioning the node, including\nunplugging VIFs and removing volume targets. Prior to\nIea337f73c231db2cb9d9f639b92475daaede6793, we also reset the\ninstance_info and instance_uuid fields.\n\nWhen Ironic automated cleaning is enabled, these cleanup actions race\nwith Ironic, and may result in an HTTP 409 Conflict and a number of\nretries, due to Ironic holding the node\u0027s lock or the node being in a\nstate transition. In the extreme case, if retries are exhausted, the\ncleanup operation may fail, leading to the instance moving to an ERROR\nstate.\n\nThe VIF unplug and volume target delete cleanup operations are actually\nunnecessary once a node has been deprovisioned, since they are performed\nas part of the node\u0027s tear_down routine [1][2]. This change removes\nthese cleanup operations in that case, retaining them only when the\nprovision state of the node was never changed (i.e. spawn failed early).\n\n[1] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/conductor/manager.py#L1117\n[2] https://opendev.org/openstack/ironic/src/commit/912dcbbdc9c0859168ac4ea6fa6b6f7ad4358e84/ironic/drivers/modules/agent_base.py#L685\n\nCloses-Bug: #2019977\nRelated-Bug: #1815799\nChange-Id: I60971b58cf1f24bdb5d40f668e380ebfee2ac495\n(cherry picked from commit 010f5aafbf9375c7740fe239a9a76df4a3264ca2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48097b8455710edcd2f0680ff8f41162f5755be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48097b8455710edcd2f0680ff8f41162f5755be4"}]},"branch":"refs/heads/master","description":"Rebase"},"6ebd8a56d1c79fc15f385f232ce0e22937b2fdac":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-01-29 11:11:56.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/11/883411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/883411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/883411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/883411/5"}}},"commit":{"parents":[{"commit":"8c953d4d257b05009111fdd048ea7c05f85c0247","subject":"Merge \"zuul: Add missing context comments for nova-next\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c953d4d257b05009111fdd048ea7c05f85c0247"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-17 13:20:54.000000000","tz":60},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2025-01-29 09:59:58.000000000","tz":0},"subject":"ironic: Fix ConflictException when deleting server","message":"ironic: Fix ConflictException when deleting server\n\nWhen unprovision works via Ironic, all operations in _cleanup_deploy\nhave already been completed. Previous to this patch, we continue\nattempting all the clean up steps, which eventually errors out with\nBadRequest, or similar, and we complete the delete.\n\nHowerver, if cleaning has started, we hit a conflict exception,\nso we don\u0027t hit the expected error above.\n\nPrior to moving to the SDK, that landed in Caracal,\nwe would retry on conflict errors. You can tweak the\nconfig to keep retrying for the length of time cleaning\nusually takes in your enviroment.\n\nAfter this patch:\nIeda5636a5e80ea4af25db2e90be241869759e30c\n\nWe now hard fail with this error:\n\nopenstack.exceptions.ConflictException:\nFailed to detach VIF xxx from bare metal node yyy\n...\nNode yyy is locked by host zzz,\nplease retry after the current operation is completed.\n\nThis change simply skips calling the operations that\nwill always error out, avoiding the need to wait for\ncleainging to complete before getting the expected\nerror message.\n\nCloses-Bug: #2019977\nRelated-Bug: #1815799\nChange-Id: I60971b58cf1f24bdb5d40f668e380ebfee2ac495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ebd8a56d1c79fc15f385f232ce0e22937b2fdac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ebd8a56d1c79fc15f385f232ce0e22937b2fdac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6799,"name":"Nicholas Kuechler","email":"nkuechler@gmail.com","username":"nicholaskuechler"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
