)]}'
{"id":"openstack%2Fnova~883904","triplet_id":"openstack%2Fnova~stable%2Fvictoria~Ic6566272b8c5af57cd1c73c73b3cba6cd265bebe","project":"openstack/nova","branch":"stable/victoria","topic":"bug/1941005","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-03-05 18:56:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic6566272b8c5af57cd1c73c73b3cba6cd265bebe","subject":"Reproducer for bug 1983753","status":"ABANDONED","created":"2023-05-23 11:45:33.000000000","updated":"2024-03-05 18:56:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7cc0d63329bbb646cd85d468f0fbddb717f9ddeb","_number":883904,"virtual_id_number":883904,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-23 12:13:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-23 12:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 18:56:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"57bf18a63bb2077ef8919a9365ee2b1d9c698f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-05-23 11:45:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2de5387dea226ce87db1b713b6115957be9e1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 12:13:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c902b1ee554e3e8a134f73efed3197\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45947e3f53c7456fb759d1009ae7f1f8 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752d07c0a8694ad486e130e34519aa0f : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e48960797eeb453b8d63302240a2bb56 : SUCCESS in 14m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c0927ba978e4880b45d62f35470c2fe : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6340d4ae2e44b798bca2b969e9296e : SUCCESS in 12m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f2c354d26dc44404a66491b2b2a2e51e : FAILURE in 6m 35s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab4cc8c2872a4621bc79b907650cc55c : SUCCESS in 20m 42s","accounts_in_message":[],"_revision_number":1},{"id":"7cc0d63329bbb646cd85d468f0fbddb717f9ddeb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 18:56:41.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/nova is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f97c65b8c56facedc0fe69d150c22d7d9218b52","revisions":{"7f97c65b8c56facedc0fe69d150c22d7d9218b52":{"kind":"REWORK","_number":1,"created":"2023-05-23 11:45:33.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/04/883904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/883904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/883904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/883904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/883904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/883904/1"}}},"commit":{"parents":[{"commit":"cdccf4ed2ad5d73756b76bb2740e93010f8a33af","subject":"Merge \"[stable-only][cve] Check VMDK create-type against an allowed list\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdccf4ed2ad5d73756b76bb2740e93010f8a33af"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2022-08-06 12:48:52.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2023-05-23 11:42:39.000000000","tz":120},"subject":"Reproducer for bug 1983753","message":"Reproducer for bug 1983753\n\nThere is a stable/victoria specific change in test_pci_sriov_servers.py\nas the assertPCIDeviceCounts was not available in the base test class in\nvictoria. So it is moved there by this patch\n\nRelated-Bug: #1983753\nChange-Id: Ic6566272b8c5af57cd1c73c73b3cba6cd265bebe\n(cherry picked from commit 6d602c6b734c2e360fa319ba22c2fced02ad3d29)\n(cherry picked from commit bdc32226e2d9ce4a073306f8f6267fbe305e64e4)\n(cherry picked from commit 87a8a7df9800f3c81dba2e2c0520168115ad1b28)\n(cherry picked from commit 81741d7ce8031367ca4055227ecd55014d0aced8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f97c65b8c56facedc0fe69d150c22d7d9218b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f97c65b8c56facedc0fe69d150c22d7d9218b52"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
