)]}'
{"id":"openstack%2Fnova~888206","triplet_id":"openstack%2Fnova~stable%2Fussuri~I4ab97626c10052c7af9934a80ff8db9ddab82738","project":"openstack/nova","branch":"stable/ussuri","topic":"bug/2003991","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-02-07 12:38:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4ab97626c10052c7af9934a80ff8db9ddab82738","subject":"Enforce quota usage from placement when unshelving","status":"ABANDONED","created":"2023-07-12 02:41:03.000000000","updated":"2024-02-07 12:38:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1666579df985d76a37a3c97a9230d3027c3615e7","_number":888206,"virtual_id_number":888206,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-13 23:44:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-12 05:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 12:38:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"43ddaf4d5197ff138554d7c6082e869d3f3bceb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-07-12 02:41:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60025c6d20e033668cb240de22c7381fd0b955f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 05:19:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9da69a71594461a8759730eea8fe4bb\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/857fee09a54f4e0ea1d44575e2c794ca : SUCCESS in 1h 05m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3124fe5957224474a7321b392d5c7357 : SUCCESS in 54m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3c683436b6843338d9b465bcb84d9cd : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/041915f6280a4695ae47cd1141363776 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67afcd1fc91d478ab0a1650725f9849e : SUCCESS in 15m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eef0bd591cdc43cda296ed7811d18128 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e28cde13594f569901125cf0c93a14 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/861a806c344244ac9d96249887e2af40 : SUCCESS in 26m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fa195a2cfd9e4defa92c11f1953e8ffd : FAILURE in 41m 25s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bdc0448bde4147e7816625cb64f98f6b : SUCCESS in 1h 27m 22s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9e50dbe466404356a3e2effcae2f3066 : SUCCESS in 41m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3a94f00860104570a0e3daf14000fd21 : SUCCESS in 1h 39m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/40270fa4dc084fb28b7f44b8121e3a78 : SUCCESS in 1h 43m 13s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8dcef2969ba642a8a4cf61c87e87835a : FAILURE in 15m 48s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/23ce9aeecc0c4616b6ca8985fd980755 : FAILURE in 5m 33s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/096bb865184c4ec794308eb8773acc3b : FAILURE in 1h 49m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9706192c520548acae2ec9b9f2865887 : SUCCESS in 47m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/db4949f031b44df0867396920b29f80c : FAILURE in 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"564386400f8d6257b51a369529431f009b210ee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-07-13 22:25:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c52608da82a50ebb3189001d49dbd26cfab4c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 23:44:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5750033b6234e719404ad57da0d5eb8\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fae9f79de207497986e2796897897943 : SUCCESS in 1h 06m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b39f41e74c194811be06f6cb6aeaf68a : SUCCESS in 41m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39fa5031f4a64d59a9f902b9bacd4628 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ce90e166db41fc933b227a93b08bd3 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59f0773eeae944be81fb5f27b206f395 : SUCCESS in 10m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc34317cb59140bd905f5788d2dedca1 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574a5e2130ca4d6a9562f19e1f3ff733 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/605290461d9749c790b30ab5dd450483 : SUCCESS in 24m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/743627f942d14a47959c8678cf542435 : SUCCESS in 50m 41s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5c2edbe42a764e1f9599610c70921761 : SUCCESS in 58m 18s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ff5d5d29a1d6455c8685766c94233161 : SUCCESS in 32m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/60f7b8a9c1a144c29a4d9f47bd57e57f : SUCCESS in 1h 06m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/f2db600caf654ce1a0baf48a5f714e25 : SUCCESS in 1h 09m 22s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ab195b40c634bdba357142b3cbb462f : SUCCESS in 15m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d902e0726d504f7285c622074c0449f1 : FAILURE in 5m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7fc2f1ebf0c42da8a8cc664d5005c3d : FAILURE in 45m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64c41707da134bbdb17536afbf92aed7 : SUCCESS in 49m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b3cba83531404d2ca3bd94790e933d25 : FAILURE in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1666579df985d76a37a3c97a9230d3027c3615e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 12:38:09.000000000","message":"Abandoned\n\nstable/ussuri branch of openstack/nova transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5c9b39f6e5ccdefbc3e47645e6fc9e0b1c7e000","revisions":{"e69f6deb086e624961b2e44b1a774baeee4ec38e":{"kind":"REWORK","_number":1,"created":"2023-07-12 02:41:03.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/06/888206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/888206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/888206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/888206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/888206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/888206/1"}}},"commit":{"parents":[{"commit":"82fe224835c3c841e0a83a14ea6021c9093cb456","subject":"Reproducer for bug 2003991 unshelving offloaded instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82fe224835c3c841e0a83a14ea6021c9093cb456"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-01-27 03:08:09.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-07-12 02:40:35.000000000","tz":0},"subject":"Enforce quota usage from placement when unshelving","message":"Enforce quota usage from placement when unshelving\n\nWhen [quota]count_usage_from_placement \u003d true or\n[quota]driver \u003d nova.quota.UnifiedLimitsDriver, cores and ram quota\nusage are counted from placement. When an instance is SHELVED_OFFLOADED,\nit will not have allocations in placement, so its cores and ram should\nnot count against quota during that time.\n\nThis means however that when an instance is unshelved, there is a\npossibility of going over quota if the cores and ram it needs were\nallocated by some other instance(s) while it was SHELVED_OFFLOADED.\n\nThis fixes a bug where quota was not being properly enforced during\nunshelve of a SHELVED_OFFLOADED instance when quota usage is counted\nfrom placement. Test coverage is also added for the \"recheck\" quota\ncases.\n\nCloses-Bug: #2003991\n\nChange-Id: I4ab97626c10052c7af9934a80ff8db9ddab82738\n(cherry picked from commit 6f79d6321e7c3edaab2eb911198b7b7f851371b3)\n(cherry picked from commit c67e69c0e35d837fcefd7e4ea010a956569c3a55)\n(cherry picked from commit 490d5de6bb9ef3123cbbcea50e08ad0ad2edcbbe)\n(cherry picked from commit da4b7d1e70486402635a009724b263981cd22861)\n(cherry picked from commit 746bc7222276a8c6bb4f0f9317802e332d25b756)\n(cherry picked from commit c13c26f88edea2d762aef97013662859496dd3e6)\n(cherry picked from commit da5e34e33d8ad0f85da1dd83c0e020efb6fc3978)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e69f6deb086e624961b2e44b1a774baeee4ec38e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e69f6deb086e624961b2e44b1a774baeee4ec38e"}]},"branch":"refs/heads/stable/ussuri"},"d5c9b39f6e5ccdefbc3e47645e6fc9e0b1c7e000":{"kind":"REWORK","_number":2,"created":"2023-07-13 22:25:42.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/06/888206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/888206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/888206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/888206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/888206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/888206/2"}}},"commit":{"parents":[{"commit":"882f0a038ea6b2f4cd75c8d761fe3effebf67cf9","subject":"Reproducer for bug 2003991 unshelving offloaded instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/882f0a038ea6b2f4cd75c8d761fe3effebf67cf9"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-01-27 03:08:09.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-07-13 22:24:09.000000000","tz":0},"subject":"Enforce quota usage from placement when unshelving","message":"Enforce quota usage from placement when unshelving\n\nWhen [quota]count_usage_from_placement \u003d true or\n[quota]driver \u003d nova.quota.UnifiedLimitsDriver, cores and ram quota\nusage are counted from placement. When an instance is SHELVED_OFFLOADED,\nit will not have allocations in placement, so its cores and ram should\nnot count against quota during that time.\n\nThis means however that when an instance is unshelved, there is a\npossibility of going over quota if the cores and ram it needs were\nallocated by some other instance(s) while it was SHELVED_OFFLOADED.\n\nThis fixes a bug where quota was not being properly enforced during\nunshelve of a SHELVED_OFFLOADED instance when quota usage is counted\nfrom placement. Test coverage is also added for the \"recheck\" quota\ncases.\n\nNOTE(melwitt): The difference from the cherry picked change is\nbecause change I5d6db24a440397e588ba69f98a9cd2b8a846adc2 (tests: Add\nhelpers for rebuild, cold migrate, and shelve/unshelve) is not in\nUssuri.\n\nCloses-Bug: #2003991\n\nChange-Id: I4ab97626c10052c7af9934a80ff8db9ddab82738\n(cherry picked from commit 6f79d6321e7c3edaab2eb911198b7b7f851371b3)\n(cherry picked from commit c67e69c0e35d837fcefd7e4ea010a956569c3a55)\n(cherry picked from commit 490d5de6bb9ef3123cbbcea50e08ad0ad2edcbbe)\n(cherry picked from commit da4b7d1e70486402635a009724b263981cd22861)\n(cherry picked from commit 746bc7222276a8c6bb4f0f9317802e332d25b756)\n(cherry picked from commit c13c26f88edea2d762aef97013662859496dd3e6)\n(cherry picked from commit da5e34e33d8ad0f85da1dd83c0e020efb6fc3978)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5c9b39f6e5ccdefbc3e47645e6fc9e0b1c7e000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5c9b39f6e5ccdefbc3e47645e6fc9e0b1c7e000"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
