)]}'
{"id":"openstack%2Fnova~908687","triplet_id":"openstack%2Fnova~stable%2F2023.1~I47df4d147d9dd4c901af2c55c97e111e603cd806","project":"openstack/nova","branch":"stable/2023.1","topic":"bp/2048184-stable/2023.1","attention_set":{},"removed_from_attention_set":{"9373":{"account":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"last_update":"2024-02-12 20:19:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I47df4d147d9dd4c901af2c55c97e111e603cd806","subject":"Updates glance fixture for create image","status":"ABANDONED","created":"2024-02-12 10:30:12.000000000","updated":"2024-02-12 20:19:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":906088,"cherry_pick_of_patch_set":1,"meta_rev_id":"9c64282623d93674f76cc0eb1973c5b1d6143f9d","_number":908687,"virtual_id_number":908687,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-12 10:59:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-1,"date":"2024-02-12 17:50:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 10:30:12.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"},{"updated":"2024-02-12 10:45:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-12 10:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-12 17:50:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e41807c52f2fc9faa68c9630fadd0301fce2cbb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2024-02-12 10:30:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"04135193e9e1ad8fe81cf69d0edfc78c4f8d316a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 10:45:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f06d9eb211746cd9a034d074a8140e4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ba11efb65af40c3ba2e40c1d56babf8 : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70f5af64a0564585aab52436b4470858 : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b908efa1c7aba16e12fa5d34be0513e864d2db7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 10:59:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aca2b77af0443e7b6e019d883c02b8c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff5bfe6400c8478f8d82d09d803dee5a : SUCCESS in 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3654bcf563d5426d97b393a993267027 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31d0064ce61a43a5a6339843f2895ebf : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d314197e9b1b4f5589bbed9d1437c3b2 : SUCCESS in 9m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/149e016a8fb04d11aaff4a707b091686 : SUCCESS in 5m 46s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d4bb759e6c0442189a8ba4293ab2140 : SUCCESS in 22m 46s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40e384fb6e824d36b90d6788a9e4550f : SUCCESS in 24m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3dc17af05974430b96e1294189dc965f : SUCCESS in 23m 03s","accounts_in_message":[],"_revision_number":1},{"id":"16c4102c1b4d73d2a649b13731cc7ef810307049","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-12 17:50:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c64282623d93674f76cc0eb1973c5b1d6143f9d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2024-02-12 20:19:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77ac757f383b8956b90f1c4197a525ab237ea5fb","revisions":{"77ac757f383b8956b90f1c4197a525ab237ea5fb":{"kind":"REWORK","_number":1,"created":"2024-02-12 10:30:12.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/87/908687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/908687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/908687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/908687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/908687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/908687/1"}}},"commit":{"parents":[{"commit":"5c33bd4dfc8c1fcf3eecd656130408fb0a5a3bc5","subject":"cpu: make governors to be optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c33bd4dfc8c1fcf3eecd656130408fb0a5a3bc5"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-01-16 05:55:12.000000000","tz":0},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2024-02-12 10:30:12.000000000","tz":0},"subject":"Updates glance fixture for create image","message":"Updates glance fixture for create image\n\nglance client takes image_min_disk as string and return as int,\nwhile glance fixture is returning as it is.\n\nChange-Id: I47df4d147d9dd4c901af2c55c97e111e603cd806\n(cherry picked from commit 12ca930e459ce2b5487f8d2f9854069ed9d95cf1)\n(cherry picked from commit 19a4402fc20f3083d468aeaec072afd4dba6ff73)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/77ac757f383b8956b90f1c4197a525ab237ea5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/77ac757f383b8956b90f1c4197a525ab237ea5fb"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
