)]}'
{"id":"openstack%2Fnova~908806","triplet_id":"openstack%2Fnova~stable%2F2023.1~Ia916310522b007061660172fa4d63d0fde9a55ac","project":"openstack/nova","branch":"stable/2023.1","topic":"bug/1946339","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-02-13 17:05:57.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"Ia916310522b007061660172fa4d63d0fde9a55ac","subject":"tests: Use GreenThreadPoolExecutor.shutdown(wait\u003dTrue)","status":"MERGED","created":"2024-02-12 16:37:44.000000000","updated":"2024-02-13 19:49:12.000000000","submitted":"2024-02-13 19:47:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908806-bug/1946339","cherry_pick_of_change":883317,"cherry_pick_of_patch_set":2,"meta_rev_id":"c41e4f7c8c33737ba0814282f364b5b55215bc6b","_number":908806,"virtual_id_number":908806,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-13 19:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2024-02-12 17:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2024-02-12 17:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2024-02-12 17:39:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-12 16:37:44.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2024-02-12 17:12:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-12 17:26:37.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-02-12 17:39:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1934fd974ee9b144801f7227b95bfee827b9ec6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:37:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* nova/test.py","accounts_in_message":[],"_revision_number":1},{"id":"913f25dd7f7b2bb607d694b53174f4a175c1c44c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:38:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"98221665fd94f611b1f1139273d724b0884a9a46","tag":"autogenerated:gerrit:restore","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:43:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"8a14786f5fa0d98f515875402c384e8f69f679cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:43:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18c372d1f19b6ae91d26b9d85af54eb910493de8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:43:33.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"5e5b49960982a63e0cc9221ec353edf405e3d35f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:44:03.000000000","message":"Topic bug/1946339-stable/2023.1 removed","accounts_in_message":[],"_revision_number":2},{"id":"2d1a539bdb984171778dde0fb8918af474ae3a3e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-12 16:44:06.000000000","message":"Topic set to bug/1946339","accounts_in_message":[],"_revision_number":2},{"id":"6c5c67d1e073b43a009ae18dabccfe34317ae7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 17:12:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb35076a5654435b7620767ec8861a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc830e376cd147e2b200c266742934d0 : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d98ddafcaf4522a4e41afdca693e8f : SUCCESS in 12m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1182c91b64774b238fd0ec9d5662d361 : SUCCESS in 14m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc23a8bdb4c40f282b4bd0ed4f620f2 : SUCCESS in 13m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0c79e8bc0a5f41fdaf1f717418971114 : SUCCESS in 5m 20s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9540f589c1e4821989d17f1988cc9c6 : SUCCESS in 25m 30s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2dcce7d3fc9e46e3ad8b426dd4d4c943 : SUCCESS in 27m 08s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08813e06e2474e3e80b2bf79ef55e5ca : SUCCESS in 22m 36s","accounts_in_message":[],"_revision_number":2},{"id":"dd4e453023288ce5f0b155da0120077aac3f0d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 17:16:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ac10c1268743b4bbe09d5fc032b238\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60658ad7156e475bbfa0f6fceb90afa1 : SUCCESS in 10m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d4a9f8db5da47b4aef761fe1be50887 : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32251af466997da78d7dbd377bdad823fc33bfda","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-02-12 17:26:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9babdcaa5222b1978dac4bc66bc79c45f5a28f2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-12 17:39:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2361a100a57d65f2dfc177063ccc22c63d5c4bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 18:34:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9edaf8c8eb8dd937dbfda5e4996f84a623ab7290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 21:03:19.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/908839 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"c21966f1dc07c52f9e2a6d7bde4584599f4fd61f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-13 07:31:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b447e5f7ec631ee0475a75f1cfd07e2ec48ee8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 07:43:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f40c0e634bbc4d85ad5ec0d137add10d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82ab136a2a054c2281d7bedec90a9b81 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/87c0cf90a97144adb30b898b18756d58 : SUCCESS in 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6464e7042f4eee0950a1e20ded6c53e546d789fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:26:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c53d8f0965e04e5480519ec9450d9d0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98b40f1f814e47e2bb805f9a054a507f : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87e049adc204805a8e501fc47f310a5 : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e03a232f3394edaa4b6565bcbf37468 : SUCCESS in 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb3f655cc794c47b0be571ee357fbd3 : SUCCESS in 9m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e1d96ad503c248948969a4226dcd4419 : SUCCESS in 4m 43s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07b961c90b1f42548b6bde792730b7f7 : SUCCESS in 54m 08s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5940c85e6cbb4ecca97a7001db4c046e : SUCCESS in 24m 42s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98a50af2f5654f7a99fb413e6f1c8325 : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0c5217910844cddf0b118bda75b1df32fa348290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 09:47:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70745afef0a4fa29c8850d5998bf26117613babd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 11:29:02.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/908839 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"0aab74f2b433610cf35a1f8249480875e02a932a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-13 17:05:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77f2c27c92035a1d34dd3264ffbcde275c9f7f2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 17:23:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49396cd0973845098642b0db4c66cfcc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce32d1b4b73c41318c99b8ea3838cf0d : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d394439b62f9429dba50d21573cc9903 : SUCCESS in 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a9e5baccca1d47ba5e88773ae58647639d0ab47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:08:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6f804de5a494ef4bdc6aeae63f40dc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f2f3d3d9a4d4e7e805c042a31dd2e95 : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a7fbcb52f64251b9e42a2e6de09667 : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11d54eb66f8549e5a734f54e42b6b50a : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ad4dd4ae8f43c288475ccd09e3e539 : SUCCESS in 9m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b2f71ebe55a446d79ed9574128729276 : SUCCESS in 4m 23s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/833ef743788942ce8843473fd9f2a181 : SUCCESS in 27m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e2492fee8354027b38d7cf0d774a405 : SUCCESS in 23m 29s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c6cf115169d491fbe1f9405ea0c94e0 : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b39e29030418e6d929abebc9632e0aaa01428d19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:08:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8c591a377e079f0b6d4efdb6d8fce76095ba465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:47:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c1ceabcba8c4c1796d81e448cf519f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a17df9b1cc4e445ab5eb298c0e0fe7a0 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2012a6b59c2d4d5e85a5b17ae1d0d94d : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442803bc8f074b6caac96791415f15b0 : SUCCESS in 11m 00s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/893e7021db954619ae8440e940eb5405 : SUCCESS in 24m 51s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95cd0ac87df14334ae91ae60621a5421 : SUCCESS in 32m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c428a850bc0d452aa53b857d062f0a80 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fec96ca70a2b3cd8adcd527481088bbce69a23e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:47:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c41e4f7c8c33737ba0814282f364b5b55215bc6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:49:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/647874b19e6348758e2f648635b2b2d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274061f0a9354d5a96a0cf35fca7d661 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22cccb9bd9b5e1eb27f7e94855e734ed3863ab8a","revisions":{"887ba1ae8e2ffe87bf31aeb3e608d71013ebfeb9":{"kind":"REWORK","_number":1,"created":"2024-02-12 16:37:44.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/908806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/908806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/908806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/908806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/908806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/908806/1"}}},"commit":{"parents":[{"commit":"5c33bd4dfc8c1fcf3eecd656130408fb0a5a3bc5","subject":"cpu: make governors to be optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c33bd4dfc8c1fcf3eecd656130408fb0a5a3bc5"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-05-11 19:23:52.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2024-02-12 16:37:44.000000000","tz":0},"subject":"tests: Use GreenThreadPoolExecutor.shutdown(wait\u003dTrue)","message":"tests: Use GreenThreadPoolExecutor.shutdown(wait\u003dTrue)\n\nWe are still having some issues in the gate where greenlets from\nprevious tests continue to run while the next test starts, causing\nfalse negative failures in unit or functional test jobs.\n\nThis adds a new fixture that will ensure\nGreenThreadPoolExecutor.shutdown() is called with wait\u003dTrue, to wait\nfor greenlets in the pool to finish running before moving on.\n\nIn local testing, doing this does not appear to adversely affect test\nrun times, which was my primary concern.\n\nAs a baseline, I ran a subset of functional tests in a loop\nuntil failure without the patch and after 11 hours, I got a failure\nreproducing the bug. With the patch, running the same subset of\nfunctional tests in a loop has been running for 24 hours and has not\nfailed yet.\n\nBased on this, I think it may be worth trying this out to see if it\nwill help stability of our unit and functional test jobs. And if it\nends up impacting test run times or causes other issues, we can\nrevert it.\n\nPartial-Bug: #1946339\n\nChange-Id: Ia916310522b007061660172fa4d63d0fde9a55ac\n(cherry picked from commit c095cfe04e2c71efcfbfdd95948af080a98065e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/887ba1ae8e2ffe87bf31aeb3e608d71013ebfeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/887ba1ae8e2ffe87bf31aeb3e608d71013ebfeb9"}]},"branch":"refs/heads/stable/2023.1"},"22cccb9bd9b5e1eb27f7e94855e734ed3863ab8a":{"kind":"REWORK","_number":2,"created":"2024-02-12 16:43:24.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/06/908806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/908806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/908806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/908806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/908806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/908806/2"}}},"commit":{"parents":[{"commit":"d045b6973c96787cd753762e8794ed02e5408a69","subject":"testing: Reset affinity support global variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d045b6973c96787cd753762e8794ed02e5408a69"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-05-11 19:23:52.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-02-12 16:42:30.000000000","tz":60},"subject":"tests: Use GreenThreadPoolExecutor.shutdown(wait\u003dTrue)","message":"tests: Use GreenThreadPoolExecutor.shutdown(wait\u003dTrue)\n\nWe are still having some issues in the gate where greenlets from\nprevious tests continue to run while the next test starts, causing\nfalse negative failures in unit or functional test jobs.\n\nThis adds a new fixture that will ensure\nGreenThreadPoolExecutor.shutdown() is called with wait\u003dTrue, to wait\nfor greenlets in the pool to finish running before moving on.\n\nIn local testing, doing this does not appear to adversely affect test\nrun times, which was my primary concern.\n\nAs a baseline, I ran a subset of functional tests in a loop\nuntil failure without the patch and after 11 hours, I got a failure\nreproducing the bug. With the patch, running the same subset of\nfunctional tests in a loop has been running for 24 hours and has not\nfailed yet.\n\nBased on this, I think it may be worth trying this out to see if it\nwill help stability of our unit and functional test jobs. And if it\nends up impacting test run times or causes other issues, we can\nrevert it.\n\nPartial-Bug: #1946339\n\nChange-Id: Ia916310522b007061660172fa4d63d0fde9a55ac\n(cherry picked from commit c095cfe04e2c71efcfbfdd95948af080a98065e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/22cccb9bd9b5e1eb27f7e94855e734ed3863ab8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/22cccb9bd9b5e1eb27f7e94855e734ed3863ab8a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
