)]}'
{"id":"openstack%2Fnova~909189","triplet_id":"openstack%2Fnova~master~I2367680ca0f39cdfb5e0600ea67de254e744f4c3","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-12 16:04:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-13 12:45:58.000000000","reason":"Change was submitted"},"26250":{"account":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"last_update":"2025-08-13 12:45:58.000000000","reason":"Change was submitted"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2025-08-13 05:47:21.000000000","reason":"\u003cGERRIT_ACCOUNT_20733\u003e replied on the change","reason_account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}}},"hashtags":[],"change_id":"I2367680ca0f39cdfb5e0600ea67de254e744f4c3","subject":"Drop unused parameter type validation method","status":"MERGED","created":"2024-02-16 09:10:34.000000000","updated":"2025-08-13 12:47:18.000000000","submitted":"2025-08-13 12:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909189","meta_rev_id":"853aba71afdf0cf5c7369bf3816095757e9106cc","_number":909189,"virtual_id_number":909189,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-13 12:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-08-12 16:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-13 05:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-08-12 16:04:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 09:10:44.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 09:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 09:24:27.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"REVIEWER"},{"updated":"2024-02-16 11:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-22 08:42:54.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-22 14:52:52.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2025-08-11 13:40:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-08-13 06:07:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"d750d56a0a09eeaa37ccb6d9479f144a098b121a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 09:10:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b80d4b57c522530af0b36d348923a1e484b3d490","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 09:10:44.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42586/","accounts_in_message":[],"_revision_number":1},{"id":"5118065fb6c2079cf7b23784260bd5ff903c1f75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 09:10:47.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42586/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"08316d27b4ec6838b367080a4c6bb96938c943b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 09:23:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba0fedcd0fa4816973c2e8686febeb9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c1e6716e692483aad5d89b78b99fa0a : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8871ac634e234c9bad6c610918a9cd49 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d9e9364545342f9a097fa93115b3cef : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"578b3f75b3f4839564892735b5534c11593145ac","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2024-02-16 09:24:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2b247f1f9a8964fd9a9d93bc484a41e063266346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 11:20:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/166a0ad52fe94144a42939269968fb8d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c4cf2e662114e91a8ddabb2da0cb928 : SUCCESS in 40m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6730a7b09a364eb9a5ed85bf113192a3 : SUCCESS in 1h 25m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dadd9230e784c4dbb70d0d2764eed76 : FAILURE in 37m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f32e0d799d44b65b7b104887fc6fed4 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269d532c56254be38f6f704433683e17 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a36bcd424b8f4516aff14f36dac7fdd6 : SUCCESS in 12m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75f847f7c27a41b296599dcf571de711 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd86f9d2a6a456a97847a93ca856daa : SUCCESS in 9m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a4bba8d798864e6eaea270af7d06f3ef : SUCCESS in 1h 10m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/483d15d593924d318b5ae07fbf9b3974 : SUCCESS in 2h 00m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0bfbbec53e9c469abf8432d0512a2045 : SUCCESS in 35m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0d91c30feb824165b47a0c116f8934ac : SUCCESS in 50m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1513799140cb4c0390c2f9c65765e75d : SUCCESS in 2h 03m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/f1a70990956c4971a1debc0b54cc4235 : SUCCESS in 1h 48m 26s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/447cdd4a17f844cb99419bd29a454e48 : SUCCESS in 52m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/66fe200ee4dd4f5281173a05b83917dc : SUCCESS in 5m 00s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bb4e972e71624a88b860b7c05260f412 : SUCCESS in 13m 04s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d6213d80a704f1dae47c6580e80a58a : SUCCESS in 26m 37s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e3f707b4f434c0c809daafb25e3fa95 : SUCCESS in 20m 52s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3aeed3528dbd4452ae262c3e341f0ae7 : SUCCESS in 22m 00s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5ee93bcca455425197682de0c981e143 : SUCCESS in 1h 37m 53s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/053aee03a8c348d0b855020349104594 : SUCCESS in 1h 34m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4e59b2244b345959b5dc444ee6baa3d : SUCCESS in 1h 12m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0337cea1e12f48c6852ed2558f29ad37 : SUCCESS in 47m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ae7098231c03430faa6fbc5765748c31 : SUCCESS in 23m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d6bba592e45a49d2b3d4d1824537f8dd : FAILURE in 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dbdd7da22fcd9b1de57ecf0f5a677c5c843baa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-22 07:29:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1104b49570119d05d4b90189e200e0a6b41d0f6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 08:13:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b370479157a34c809ad799a1e386f9fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8137abb925104648b674da843a20af14 : FAILURE in 37m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dbf912737d1b4843bdc03c196317c20e : SUCCESS in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01e9c9fade106675d0c63194c6039112659d8ca6","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-22 08:42:54.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909189-78n96/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"b7de5bce56055143941b17586e51067fbe9b54be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 09:21:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d194c7ab2784541959619b33cafc472\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7621d48878b4ae7b2d3de1d5adddaba : SUCCESS in 52m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e67338df4b5d42a086cd8706d1b20af7 : SUCCESS in 1h 34m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57ea88900fb840c3b18c7efd14b97a80 : SUCCESS in 1h 00m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e42372f54264a0ebde3a0f35d0010ab : SUCCESS in 21m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf53903448f444397cf0da8616233fc : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/511898931abe4d5db9404d1dc9ef8988 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba8026edc6954a79a7de50800e2304de : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a372409f634fa2bacc9a34d95b53d0 : SUCCESS in 9m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6cfa63e614bd47efbd8a1377f9f8d363 : SUCCESS in 1h 00m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14f844450e3b419cac7952c36518fd8f : SUCCESS in 1h 22m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f75bcc2a5e364e9d9d5b1f4152714e9e : SUCCESS in 35m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7ceed35aa943465fa2cc740c2b72d8da : SUCCESS in 47m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/112b18d928794e8ebb5ee4f4daa255c0 : SUCCESS in 1h 42m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/8da38c189f134b2ba882afc821b4bf87 : SUCCESS in 1h 51m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/239a03b35f2c4eb48e08b1a92b13c2a2 : SUCCESS in 45m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8c372ee75d01428fa605c5ca793c5730 : SUCCESS in 3m 12s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/683c7d44e62c438a8a831d1c3744f52c : SUCCESS in 24m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/560bd3fe29984487b61fc8bf314b2af0 : SUCCESS in 21m 01s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4e7052deee6f4fcb866491e3e88fbc30 : SUCCESS in 23m 44s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c00aa86bd56467188d823036c909ab3 : SUCCESS in 1h 25m 42s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/7207a76a96de4566b6d86d6e3fdc4c21 : SUCCESS in 1h 13m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/86c530abe2e24be8b9f01a32fe6b3f2a : SUCCESS in 1h 21m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6e0e474312b4b768ad80bcfe8ee73b3 : SUCCESS in 50m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2ce2dd31a65b4809b43fbd184fa0dabd : SUCCESS in 19m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a0dba7ac776f4d3a965927dcbec9458f : SUCCESS in 27m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a90da888fb73e167196a52bebed2133e5181830","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2024-10-22 14:52:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c403c66a6b416a7c600fcddd1630aba50c7edbeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-18 10:53:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cb6797d3fe45f1006908475ed76cd3842ddcc9aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 12:30:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae770db9ca4d4b0090f55408f974cad7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3aa4c802bdc843d3a643b8dcf03660d1 : SUCCESS in 36m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aca7f3cb913543aab6fe725454e76845 : SUCCESS in 32m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb5f0814aa96454eac0250195cfea9db : SUCCESS in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d80ad447df1bcda9294e286310ca331c3f996f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 13:35:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f0ab2a2f7249e1b714495e4e87b240\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf812bcc75304854a715a177e64e7123 : SUCCESS in 51m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d7df48529f9d4b68884fd4df4a4042de : SUCCESS in 2h 07m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/752bc605aad944e59a01c13396c99a9b : SUCCESS in 1h 01m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e393977329164c5a969791bd66671d0f : SUCCESS in 23m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca69636f724424d975d6e0b797bd49d : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22f39a8e6b6e437692be39ce7a0dcfe3 : SUCCESS in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47057dead8084827892003092a0e8410 : SUCCESS in 15m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50da83cb3f0345a5979673c567a2cc6d : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebec31c50df4a659b6c2de66d4d6b8d : SUCCESS in 12m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3f15477f9d254d9fbf52c058a2d46625 : SUCCESS in 1h 03m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/76f8a0270e6446aca07cacf91f58cd2d : SUCCESS in 2h 08m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1bd01b971f124fa3b4788ed375775954 : SUCCESS in 45m 17s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0213103c5a224a52bb9fcb09eeb4fd66 : SUCCESS in 41m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0871a8ea7e7b4ff0aacf1de8eb936aa1 : SUCCESS in 2h 09m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/f9aae95efd2143a5baf3db88171db318 : SUCCESS in 2h 36m 44s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2216935b2aa64b48995bcf3efe55d611 : SUCCESS in 51m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/97918342adde469a88b9e45ca4bbf85b : SUCCESS in 6m 54s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fac1b302678e48e3880235adf876fcb3 : SUCCESS in 25m 07s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0bb66bef07d84422895178909fe7b0b5 : SUCCESS in 33m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/82b111fef41d43459f2903e10e06c731 : SUCCESS in 1h 00m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b8a3fc2c8e340eebed5ee1bc0230bb3 : SUCCESS in 58m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/850999abc5f54e8281d0d0a00cf41501 : SUCCESS in 28m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5726664fbf124e30b6a5491d7ef2122e : SUCCESS in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8ffc65b7209085244dc505d7f3e0cfb4f995c53a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-12 16:04:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b0fea4755f94c2c0137cb55c8207ca6432af3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:04:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1676aa381ad3c004ce3a6f313b015d822fbc66bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 19:31:09.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1656d7d213f4f04bcb3aee35158a983\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7aa1b34fef8c40918e02273759ece963 : SUCCESS in 53m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/eb7dd431d9f14dcea96ad78f9e414fc7 : SUCCESS in 1h 47m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14a641e031844a4380b96ba8fd3920c1 : SUCCESS in 55m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/230adff707bf441b9def1b1c75f6d12d : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f3a1e0b33f54fb29bc9a0a792422023 : SUCCESS in 14m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae239034ac5940128af4d2ea5736fdae : SUCCESS in 14m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1307e41515f44c108aba65b050914df3 : SUCCESS in 11m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/32a3b9d791ef4b34a136179afa6134e2 : SUCCESS in 46m 23s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c68ac427801b4d3eaec9cdfdf0151724 : SUCCESS in 26m 07s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d5094670441d4ab88b5806823f7b6ac6 : SUCCESS in 34m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/83afe6b67e664007bc7dfadc0e587132 : SUCCESS in 1h 51m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/7ef7f8ff66b840649a3f58ca9a37a55d : SUCCESS in 2h 03m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9606ea91a1fc4b5397b65baadea85e4c : SUCCESS in 4m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/286d89505bb84068ae8511c16eb55d54 : FAILURE in 1h 35m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3011115eee4b4966aee6f65d9669b696 : SUCCESS in 1h 11m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a0dda28ac1846338e4775dc896201e0 : SUCCESS in 57m 34s","accounts_in_message":[],"_revision_number":3},{"id":"3fc44ced415048c6536e7baaa901fd3065ccc47a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 01:11:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb7d691bcd4b9023777455da7efa58714bc5eae4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 02:30:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2310eb72ac394a2b96775b86b390a3ed\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a0cfa55801c4d0f9a516f2110c9add2 : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6242eff48d7a43da95be27f1947a6e43 : SUCCESS in 38m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/801a74298b3142a08e663ad7b9a12311 : SUCCESS in 47m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f6112d9d5a5db1bb4399df15df8fa46756e7dd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 03:36:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d72606886d146dab4daa691dfb65452\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0f60e223dfe94b498f7d35cdb90626c2 : SUCCESS in 31m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6a700095461c4e1dbb6100a799120513 : SUCCESS in 1h 56m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81e31fc3013042b7b9a121255744d89a : SUCCESS in 1h 01m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/590f100732924cac8512f9f23abef3b9 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f794a0ccb68421e8e628f54694aea1b : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1959543a74ea446b806914fa0d436877 : SUCCESS in 13m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a474813ce294716af725f8825afec15 : SUCCESS in 13m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a93f20a634404a4db2186484fa39f2ea : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5306c9ec6fd34f598a8e17f4020b633d : SUCCESS in 13m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/121688e24d834da386f29c0a56e66ae7 : SUCCESS in 1h 12m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/df49b8eebd274200b894d0e4d02611ef : SUCCESS in 1h 51m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/06b57584813c4e5e8e921fe8711578eb : SUCCESS in 1h 02m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/89448a2054a2440792b7412c92365aa0 : SUCCESS in 50m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/510ec68007364df196dec3fd883f9834 : SUCCESS in 2h 14m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/cce8b54e31414e1bafa7cc6bdb0fe7d4 : FAILURE in 2h 11m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f802445e1c1741bc9a76820b714074aa : SUCCESS in 51m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/527ffb11e6a642b6b2b8f6f26fbb53e1 : SUCCESS in 2m 52s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da9853b601f8427c9018ab54014b2478 : SUCCESS in 27m 20s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c191412537a54e018923705bd8038c1b : SUCCESS in 23m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa17985c90b041c480395f051e160fce : SUCCESS in 48m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4639b93d4f854e819473e87bbd151a5d : SUCCESS in 1h 03m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2f102ac7c09b47748ff6461ab2892a32 : SUCCESS in 12m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1421ec0e3a0b4a00b88876afa10f569e : SUCCESS in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d1e7af5c89a7d6082e57a8a539f7d65517b8d3f1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 04:11:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c7eee4c3ae356024cba8369b18a3a7120f66adf","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2025-08-13 05:47:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f2352316e016e8daa9a1fbe82d14cb495e25d1b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 05:51:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be08bb3682f499a92b1e44abd02a92e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/585ccf020ba34ea1a28a6854394c4dd0 : FAILURE in 47m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f9205cc9e3f47d69a815bf814aff632 : SUCCESS in 46m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78f19dff064e42d4ba626373fc03e3f8 : SUCCESS in 50m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba378c636eb69839be5559edee3b6f399e8f8feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 06:01:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f1cfdb5cbf042bc9fdec634b688d40b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad244570b619413bb6e88055491151a1 : SUCCESS in 33m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b7d32bfe5b1c46028d5956c1f3851b78 : SUCCESS in 1h 03m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df31947e5f89427b8512c61fb275ca01 : SUCCESS in 1h 03m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d78162185b44b2badf95143fd7f8ecd : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a718e78e5bbb46d696bc696b83966db8 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71bb2abdcda34ecfa23471ccbfbf8a4a : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f302da0cbe745478a323e76c087e29b : SUCCESS in 10m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa9ad6e078624d7583fdbeb7d0cc8170 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1c004858314d569e6b93933e7ae601 : SUCCESS in 9m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d88ea0e4fb24e03b99ec2e4bd37be52 : SUCCESS in 1h 05m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bdb4c3bac4024b7b941cf408618f83d1 : SUCCESS in 1h 37m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a046991d68654d24a070451a637e2bf8 : SUCCESS in 28m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/716b3d22e1de41f1aa0243c2cf14db57 : SUCCESS in 51m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/729b46e23f924fc7a3143940e751c5e6 : SUCCESS in 1h 31m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/c39253ad5b27419fa3d5eb135e2df9d5 : SUCCESS in 1h 49m 00s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3a2f5b159acd4449b2a14673c2403c54 : SUCCESS in 47m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/88a5091ee09b4ee5a1c1f986d498ad01 : SUCCESS in 4m 24s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/323bc0f045974fcdab4c0c4b02baaa56 : SUCCESS in 27m 34s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a6e29d27a8314cb7b041aa75dc52aa84 : SUCCESS in 24m 43s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/109b3cdcb49b48f4a4cd75b5a71b2821 : SUCCESS in 49m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/778be3ea328c4936be15c7f0318580a7 : SUCCESS in 53m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b8870c53d6f048f7862c99604e88ada7 : SUCCESS in 12m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d8917193943c4760a47dcbcdb2e6c280 : SUCCESS in 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f404b8d8d1e28bda3f98aea25c28d08bb3a050ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 06:02:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8a44a3b743c644dd7be0587c6fd1f8872123ae49","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-13 06:07:42.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"4ef9c194ea29bdd1cb74ca1bf6fcc0ba9db2cc04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 08:24:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8883d60cd90a42f9b47e765adeeaacc5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1fa84ad855148cf9f8fddc034d28ee0 : SUCCESS in 59m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f90b9e0025984ab0928d609e5455f1b9 : SUCCESS in 1h 42m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4519a9ab5b6490fba7d0418c2a3a9ec : SUCCESS in 1h 06m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f481a3f57afb410484ba25fb9910225e : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f57f7b78524643ec9a963027161e4604 : SUCCESS in 12m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0980b2422c7642ac91556c04ad2d5743 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91774f86b2b1418e867f0bf5639eb951 : SUCCESS in 12m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0cf4e449179f49ba9ad3abc37baec190 : SUCCESS in 38m 15s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1dbdcd525b3a437e833443f8323b0e48 : SUCCESS in 28m 28s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aaccc341b78a496e845201bfa72d2c4a : SUCCESS in 33m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ecdc6fb4749a4566ad7cbaab34ca11c2 : SUCCESS in 2h 16m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/3522ceb91d1f499fb43cbe7aecec274b : SUCCESS in 2h 16m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/af9f33048d0d474e9df9c29e6409950a : SUCCESS in 4m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d368340b0dc44641a004fa7e330b4a55 : FAILURE in 1h 01m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd2398eeff934b6fa8358c9897e215fa : SUCCESS in 1h 15m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6af7334ddae244da8afae095e771e853 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":3},{"id":"eb6894c7cd30f86611e962392bd6603bca923a30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 08:47:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5447eb7982db1c3d301042459c275edf67f15351","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 10:33:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ba66579037444e948020f0a2ab3408\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39a026d8631c405f9a8e2f0bbf1af5af : SUCCESS in 40m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e646582cc7342f9a0b7db925d5bfcac : SUCCESS in 39m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebb562d6431c48b580d6e58006592273 : SUCCESS in 44m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"790872a9926d39f36c6fdd9c02daaa2a985fe403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 10:53:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce032bc79b514f3b9e1588f74c077f11\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36ef06e117724c58b049dd1fd46aa981 : SUCCESS in 54m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/30b260923880437aa7e7dcd2c0056a73 : SUCCESS in 2h 00m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a1b615d87d54f638a0fb616e248525f : SUCCESS in 1h 02m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f29c14657f824e7b9736fe5d59b10f43 : SUCCESS in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3a02526562441da46d9461adb895e4 : SUCCESS in 9m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffb9cbeb2a9d43e8a54c64453cc39196 : SUCCESS in 12m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7645165fd39a4f34aa7cc3b7a88ee8c8 : SUCCESS in 14m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7df6f07b503e44bca77a1739e9dd51b7 : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e3c6dc95834b319276c4814602cc93 : SUCCESS in 12m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/18f6bbfcf0df49b382a18d0feb0251dd : SUCCESS in 1h 03m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6f8033ec81f74498977ecbf499965cb1 : SUCCESS in 1h 52m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/794df53a6fb243ba82dc7aa7fff6ed00 : SUCCESS in 29m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9108a1dfbab348a392cbbb4860cf317f : SUCCESS in 41m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/95b4cda43dd447c981bc859a4355ea7a : SUCCESS in 1h 54m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/6ebfd4082edd497e8bc02cf385689567 : SUCCESS in 1h 53m 05s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/cb6e67bc414942cc9877cabb30b63511 : SUCCESS in 54m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ae49a399a2724b2bb3b3d702c1bae87f : SUCCESS in 3m 14s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf08e076774748f3b3607a7f917d51d6 : SUCCESS in 26m 23s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dab222529bd04c0eb3490a7dcffc4b4a : SUCCESS in 34m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe20cc63022246509e6d1142ab7f89f6 : SUCCESS in 1h 26m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8eaccf390c7d4cd2aecfc8de6cc6a7ac : SUCCESS in 32m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ee81c979584a4afe873c6444aa824077 : SUCCESS in 13m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d2691dfc7f1445f1896ecde4a90819d7 : SUCCESS in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d6bad2b69e8492af2faea0a2c3a2e35d5f0401e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 10:53:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4f5fac200dbc8f75bcbe54450993e6d704e8c12","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-08-13 10:58:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":3},{"id":"f5f89c8d822577bf7b64206d2c6f2c101d506c58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 12:45:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9ae10c4a5cc4c64b79db32c05ae9f68\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23823fc971ee47c8b6d055c01702dcfe : SUCCESS in 35m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e9c0e84be93848a3ae8a50a88bd6b240 : SUCCESS in 1h 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e262c13848a4e82935ca554f664105c : SUCCESS in 34m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b3983f6df1487f9700629144864db3 : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f57949de0b0f4ecb933d8533304f94bf : SUCCESS in 8m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/736fd9174443436fabaf2c24062515f9 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc43bba86a34236b9d1d410d05c2165 : SUCCESS in 10m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/55f6cf4a69894c41bd95f8d4610657fc : SUCCESS in 30m 44s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1b6f34c162e945469ccb7918664493e1 : SUCCESS in 24m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0c077393a5d3469a874774331a8ef3e6 : SUCCESS in 26m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9baafd2556304669bc116c23cda7362d : SUCCESS in 1h 17m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/6f27c90380084ee8a2a942dd54e521b3 : SUCCESS in 1h 24m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4fdf9d6b4b384de9b585df78f4595f29 : SUCCESS in 5m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a373c93d42d4ea98729d7e0039918ab : SUCCESS in 1h 51m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/948b45bfabdf4ea88ec163f655210656 : SUCCESS in 1h 04m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cbd89842e954ca39671453486527b5c : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":3},{"id":"b406b6aa78112db82f8ff3f85b2e7afb5479c33c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 12:45:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"853aba71afdf0cf5c7369bf3816095757e9106cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 12:47:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd14d03e7ccd4969927a14ed93248a75\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175448255b73405e81049411c122c1f3 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"920344c276c466ec1a6b8d1c2bbd56f9c71e53d1","revisions":{"342b09c7448ffe421c2a0aa380a5b8699ea534ca":{"kind":"REWORK","_number":1,"created":"2024-02-16 09:10:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/909189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/89/909189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/89/909189/1"}}},"commit":{"parents":[{"commit":"84a3e254e88e8cc2797b61c082d9a67f27402682","subject":"Merge \"libvirt: stop enabling hyperv feature reenlightenment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84a3e254e88e8cc2797b61c082d9a67f27402682"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 09:09:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 09:10:19.000000000","tz":540},"subject":"Drop unused parameter type validation method","message":"Drop unused parameter type validation method\n\n... according to the TODO note added almost 5 years ago[1]. The method\nis not used even now.\n\n[1] fb14f24cc38a546f96adfc0446879716ca4c883e\n\nChange-Id: I2367680ca0f39cdfb5e0600ea67de254e744f4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/342b09c7448ffe421c2a0aa380a5b8699ea534ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/342b09c7448ffe421c2a0aa380a5b8699ea534ca"}]},"branch":"refs/heads/master"},"a846c2cfcdab75c89a5b47b6f6ab7804306fbe44":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-22 07:29:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/909189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/89/909189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/89/909189/2"}}},"commit":{"parents":[{"commit":"ed2bf3699ddc360fb646b895560fa9dbcfd6beba","subject":"Merge \"VMware: updates resource provider trait list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed2bf3699ddc360fb646b895560fa9dbcfd6beba"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 09:09:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-22 07:29:05.000000000","tz":0},"subject":"Drop unused parameter type validation method","message":"Drop unused parameter type validation method\n\n... according to the TODO note added almost 5 years ago[1]. The method\nis not used even now.\n\n[1] fb14f24cc38a546f96adfc0446879716ca4c883e\n\nChange-Id: I2367680ca0f39cdfb5e0600ea67de254e744f4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a846c2cfcdab75c89a5b47b6f6ab7804306fbe44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a846c2cfcdab75c89a5b47b6f6ab7804306fbe44"}]},"branch":"refs/heads/master","description":"Rebase"},"920344c276c466ec1a6b8d1c2bbd56f9c71e53d1":{"kind":"REWORK","_number":3,"created":"2025-07-18 10:53:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/909189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/89/909189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/89/909189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/89/909189/3"}}},"commit":{"parents":[{"commit":"54b65d5bf2b23fa8a4612fd3adddc8751192a807","subject":"Merge \"Add spawn_on\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/54b65d5bf2b23fa8a4612fd3adddc8751192a807"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 09:09:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-18 10:52:56.000000000","tz":540},"subject":"Drop unused parameter type validation method","message":"Drop unused parameter type validation method\n\n... according to the TODO note added almost 5 years ago[1]. The method\nis not used even now.\n\n[1] fb14f24cc38a546f96adfc0446879716ca4c883e\n\nChange-Id: I2367680ca0f39cdfb5e0600ea67de254e744f4c3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/920344c276c466ec1a6b8d1c2bbd56f9c71e53d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/920344c276c466ec1a6b8d1c2bbd56f9c71e53d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
