)]}'
{"id":"openstack%2Fnova~909218","triplet_id":"openstack%2Fnova~master~I02e127bd73059b0040a9288ef623b863915de49e","project":"openstack/nova","branch":"master","topic":"service-check-cleanup","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-02-16 16:42:52.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-18 02:31:25.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I02e127bd73059b0040a9288ef623b863915de49e","subject":"Drop logic for compute service version \u003c 33","status":"NEW","created":"2024-02-16 15:18:30.000000000","updated":"2024-03-18 02:31:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6275851350695bacbdb643e23d1798f345e63bf","_number":909218,"virtual_id_number":909218,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-16 19:01:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 15:18:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 15:22:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-02-16 15:40:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-16 19:01:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2455d4303050e226ffe931cb12add3075b7cb7c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 15:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f7e3b13c4f1389fd6555c4ea0bc17c46f7fc939","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 15:18:43.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42596/","accounts_in_message":[],"_revision_number":1},{"id":"8c95db2ec8ee9cb52ab1b7f3adfcba182650a1ee","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 15:18:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42596/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"71b7f76222412fc2c0f0f1dab1eaefd52a8fe786","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-02-16 15:22:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a2fc637a517d05b345a21351237cf41f7265b7df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 15:40:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b682ece1ae44b25bf28c0d261072283\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8f045e18f1148fd8d2504bd55584ae7 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9747c80092e4453a850996a63fee03ed : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3228ae39e663486b98609f8032682f1a : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"501deb89158532ec7ef691b2dc1ab7ac1bcb25e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 16:42:52.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9f571665300e57707efbcf5fd4be6c9965c325f8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 16:43:03.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42609/","accounts_in_message":[],"_revision_number":2},{"id":"d105844cce770e8852a86c26248b7bd5dc63c44a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-16 16:43:06.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42609/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"beb1da24c4e71f1f4b818d26298949b8a69169c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-16 16:47:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af4ec9c464c253647f05198c48ba6d12cea86bff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:10:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4dd8bc7717248e6b1959ad9460196b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc72a8621b3f4f02ac604334d6925155 : FAILURE in 26m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6decfad3d7434e1e9c59c0a4262ca312 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/212774f237164b6b85e43db6029e30b7 : SUCCESS in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed2125f00022c72e9d842d84baaf3f71d5ff0155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:01:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2805ea4c85844f19ab25c5dab451f422\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dda58d66d2be4a4480fe88bbb6de075c : SUCCESS in 53m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a1cab6e075f7423ca3fe00d98e325b02 : SUCCESS in 1h 39m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c415416861b148efaa8a015e7446f296 : SUCCESS in 1h 04m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/612a22b2ee074fe7a6ade87f7c12253e : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43f4d14e8ac4996b70b77f4add689b8 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4911d6d2a0e42ca9c81701d68088f57 : SUCCESS in 12m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9993f00d4ae349d799ead1deb2b72d75 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ad67dd493046fd9d5a0144db66d358 : SUCCESS in 10m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8f7e0ecf2f214541b385a2e149f64d96 : SUCCESS in 51m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b0436d0ff6446ac910f6edab32403b2 : SUCCESS in 1h 56m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fbf370e4ace84aef90bd61093ccd282e : SUCCESS in 50m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d62264411ed2435695546e3ea74835cb : SUCCESS in 44m 17s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3752b0c04e464c5a9576df6838967249 : SUCCESS in 1h 31m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/c836d60c511545c8ac6e791cf587858d : SUCCESS in 1h 30m 11s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a2c9cbf2cbc04f5497c4a73f5fb41083 : SUCCESS in 41m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a3967fdc87d44005b44e7cc6874882d8 : SUCCESS in 5m 08s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a335da06ad2e49479383611863752b04 : SUCCESS in 15m 53s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba0b2f509cd248f0a5e87d7ad35b3bec : SUCCESS in 25m 28s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7582b41e35814c78b63e18d287b4f824 : SUCCESS in 25m 10s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/165e5979d1a9468babc60fa6fc7d5536 : SUCCESS in 22m 54s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4791697b651e47a8a6d0a54609c34cf6 : SUCCESS in 1h 47m 16s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/14289b9f8f1243d78e053480e2aa7141 : TIMED_OUT in 2h 04m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b7b3710b3ec452a869657baa1114b63 : SUCCESS in 1h 24m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7f72a3a75ab4d19b3e0458f1e463ba6 : SUCCESS in 1h 03m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0fb5c20d326c45d1bb96831a64d91cb7 : SUCCESS in 36m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3f93dfedbb88460fb4fd10378f877d52 : FAILURE in 1h 06m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76f14d0bcfe51a5da125d62a7b04429415359bc9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-18 00:21:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56a1d3157f57d4b0d47e0a5e55a938750873ec28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 00:45:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76ce3f316234026a2f6da726118ecfe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86ac9b03a16a40049481509b65533473 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/500010a7958f45d58951bf50383a88b9 : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30e1160a6876400f859350ef2458f518 : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b65e20876a378bfdf5ac7e2a7223f4583491f55e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-18 01:17:54.000000000","message":"Topic set to cmp-svc-ver-cleanup","accounts_in_message":[],"_revision_number":2},{"id":"eafb26bf86b4dc8bf390147a7aa308aed128dcb6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-18 01:18:24.000000000","message":"Topic cmp-svc-ver-cleanup removed","accounts_in_message":[],"_revision_number":2},{"id":"7168ba6361ea8e041755fbbe149474018ddc8850","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-18 01:18:27.000000000","message":"Topic set to service-check-cleanup","accounts_in_message":[],"_revision_number":2},{"id":"d6275851350695bacbdb643e23d1798f345e63bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 02:31:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e5b97db46541fca79d2dabeb100845\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/09d0e6aafbf74eb39151b6534e23be5d : SUCCESS in 57m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d91493de9cef4b04bd74985e61fe9bd6 : SUCCESS in 1h 51m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e57b1cb3365244d98d407f4ad361b262 : SUCCESS in 1h 00m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68b600067fe94eab8dd64d07b4e61996 : SUCCESS in 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b469ba5004443dfb49779e6f83a6c88 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/592c4260a2204183b2f3c99292c5c78c : SUCCESS in 11m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e43d7db62314cfab5a8d281a681ce4a : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c042f2411342f390acd8f4282c3578 : SUCCESS in 10m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a5b25344e0c04768934e62e1bde55b46 : SUCCESS in 1h 01m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a5745fac1a1494b9fee5b2e7034b4bc : FAILURE in 2h 01m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9306772462114396a3af99c7b8514973 : SUCCESS in 34m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2ec78fd32b1f4fdc837edb3be2da4cd9 : SUCCESS in 51m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e7641c9b9a3d4c528a2b776120506017 : SUCCESS in 1h 34m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/f21b632391de4c1aa446b615161a677b : SUCCESS in 1h 15m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8577ac5d7eaa4844b9623702e0a887bf : SUCCESS in 45m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9366823d59c245f0b34a2b2838835f17 : SUCCESS in 4m 43s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/93874a772b7446038c120806a025d57e : SUCCESS in 13m 12s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/130b136887984158afe8752bbd3e03b7 : SUCCESS in 22m 51s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/023aef26e8ca42c38185d717a8885d92 : SUCCESS in 21m 18s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f43fa23c8d8c459e8605cf8f26fb5b6f : SUCCESS in 23m 10s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/31084c2d024c4003b9934504067a768a : SUCCESS in 1h 20m 07s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/f2c4fafb939741969d1d900cd578daee : SUCCESS in 1h 26m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2344936e54b9455abec347b3b33462cd : SUCCESS in 53m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3b16de0fdfa487c82a99a79464e65d3 : FAILURE in 47m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9e574761c26a4b98b8200aee4845a11e : SUCCESS in 25m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dc061bfb27264d16b7efd035d6a464dd : SUCCESS in 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"083e088a295e801fc61fad88364eda51e785d28b","revisions":{"b5a51421f11c12ac37e8224424c5ca2e12dfc6da":{"kind":"REWORK","_number":1,"created":"2024-02-16 15:18:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/909218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/909218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/909218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/909218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/909218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/909218/1"}}},"commit":{"parents":[{"commit":"84a3e254e88e8cc2797b61c082d9a67f27402682","subject":"Merge \"libvirt: stop enabling hyperv feature reenlightenment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84a3e254e88e8cc2797b61c082d9a67f27402682"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 15:16:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 15:16:58.000000000","tz":540},"subject":"Drop logic for compute service version \u003c 33","message":"Drop logic for compute service version \u003c 33\n\nThe version 33 is older than rocky.\n\nChange-Id: I02e127bd73059b0040a9288ef623b863915de49e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5a51421f11c12ac37e8224424c5ca2e12dfc6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b5a51421f11c12ac37e8224424c5ca2e12dfc6da"}]},"branch":"refs/heads/master"},"083e088a295e801fc61fad88364eda51e785d28b":{"kind":"REWORK","_number":2,"created":"2024-02-16 16:42:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/909218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/909218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/909218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/909218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/909218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/909218/2"}}},"commit":{"parents":[{"commit":"84a3e254e88e8cc2797b61c082d9a67f27402682","subject":"Merge \"libvirt: stop enabling hyperv feature reenlightenment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/84a3e254e88e8cc2797b61c082d9a67f27402682"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 15:16:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-16 16:42:52.000000000","tz":0},"subject":"Drop logic for compute service version \u003c 33","message":"Drop logic for compute service version \u003c 33\n\nThe version 33 is older than rocky.\n\nChange-Id: I02e127bd73059b0040a9288ef623b863915de49e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/083e088a295e801fc61fad88364eda51e785d28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/083e088a295e801fc61fad88364eda51e785d28b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
