)]}'
{"id":"openstack%2Fnova~909465","triplet_id":"openstack%2Fnova~master~Ife02eaca633cdc9d2a09f64933ee72e60a05d56c","project":"openstack/nova","branch":"master","topic":"shelved-image-protection","hashtags":[],"change_id":"Ife02eaca633cdc9d2a09f64933ee72e60a05d56c","subject":"Changed: adjust shelving for shelved image protection","status":"NEW","created":"2024-02-19 16:01:50.000000000","updated":"2024-11-15 14:53:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c2278614167dc692bdd9a32347a533ecd81513f","_number":909465,"virtual_id_number":909465,"owner":{"_account_id":36805,"name":"Mateusz Klejn","display_name":"Mateusz Klejn","email":"mateusz.klejn@ovhcloud.com","username":"mateusz.klejn","status":"Employee @ OVH"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-19 18:20:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-19 16:02:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-19 16:29:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-19 18:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 14:53:14.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"CC"}],"messages":[{"id":"092604ed2f4e8a238929baccbfcdb929edf7ec3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36805,"name":"Mateusz Klejn","display_name":"Mateusz Klejn","email":"mateusz.klejn@ovhcloud.com","username":"mateusz.klejn","status":"Employee @ OVH"},"date":"2024-02-19 16:01:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43fc413fc81dd7a97f4ac44a849801a93996cba8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-19 16:02:06.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42627/","accounts_in_message":[],"_revision_number":1},{"id":"82999f4ac166ce561d58cf2625818e0aa6a7e239","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-19 16:02:08.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42627/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"74d23806d2332280dc60c9ab3c902187066a024a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:29:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238393ee9da14c1fa04aa4b820a5f662\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/875dcdfbf32f45d4a8e81cdee68b83ce : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12b8d3581f4548728e4609835b02ba0c : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45e32eef3a2e4853b1ab8c58a8c2e92a : SUCCESS in 19m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd72ac5b25e6cfa308c6cd88aa5d757f6ee3cbb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 18:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f4d32415d0453b9d9d1e2dc770bfc5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2d7f655410846039da51e1c38fc1cb6 : SUCCESS in 1h 11m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c5ce73b1d87e4b73ac5f432568258550 : SUCCESS in 1h 55m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2fc412bc3bf485182b9ae623fa6c85b : SUCCESS in 52m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bae32b2883f64452b6bb970df8d170ab : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a03c13ae05348e29745d8e28d35ee83 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f0b3dda95044436a5472288e921894e : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c71a6aacf2a455788289d1c748ae0c0 : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e59ce111454774a141bb1b69fbc546 : SUCCESS in 10m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/231aad505dd0428b93251562277e08bd : SUCCESS in 1h 05m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/657064e68a484b2c906e04ce4db9a3ab : SUCCESS in 1h 47m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/000418cd6f0a48cc9c41c85591b422e8 : SUCCESS in 49m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d4080dffcdac49fb9163b5b0555dac32 : SUCCESS in 51m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/be86b2ef7c974f5baaf15566a4a794b6 : SUCCESS in 2h 07m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/6200b4cd5cc64957821a811592469863 : SUCCESS in 1h 06m 10s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4ed4d1ed14ab4e14a6c6b42ba8249fda : SUCCESS in 48m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4e4209044cbd488899cd82c2635aacb6 : SUCCESS in 6m 00s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a6f26f232993440b86b55526665fe864 : SUCCESS in 19m 28s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa0b4d86035a4bc091a9cb537c39cb19 : SUCCESS in 21m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7f7c63f9d524555869fec73e38acaa7 : SUCCESS in 26m 28s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d078fd1c59ea4fba8689dc28e05e7810 : SUCCESS in 22m 30s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/237d70db479840e594a5c25c5c9360e8 : SUCCESS in 1h 30m 55s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/e10135f195b74cad91e7835395d4b194 : SUCCESS in 1h 21m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5530ea1b3495413bb9f3a75b48c44068 : SUCCESS in 1h 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a4bc64efe94490d8da8ed52c5116425 : SUCCESS in 45m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6e832e08ac1042619826b443c4bcb996 : SUCCESS in 32m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/826ed378a2164f0fb06a395d05a233d8 : FAILURE in 50m 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c461c40f6a3328df4c7c1fe435ef2b5cb0ec031","revisions":{"1c461c40f6a3328df4c7c1fe435ef2b5cb0ec031":{"kind":"REWORK","_number":1,"created":"2024-02-19 16:01:50.000000000","uploader":{"_account_id":36805,"name":"Mateusz Klejn","display_name":"Mateusz Klejn","email":"mateusz.klejn@ovhcloud.com","username":"mateusz.klejn","status":"Employee @ OVH"},"ref":"refs/changes/65/909465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/909465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/909465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/909465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/909465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/909465/1"}}},"commit":{"parents":[{"commit":"67984cce12edd2a7b377cd90a8ba82075f351769","subject":"Fixed: unit tests for force delete instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/67984cce12edd2a7b377cd90a8ba82075f351769"}]}],"author":{"name":"mklejn","email":"mateusz.klejn@ovhcloud.com","date":"2024-02-19 15:21:16.000000000","tz":60},"committer":{"name":"mklejn","email":"mateusz.klejn@ovhcloud.com","date":"2024-02-19 15:58:53.000000000","tz":60},"subject":"Changed: adjust shelving for shelved image protection","message":"Changed: adjust shelving for shelved image protection\n\nAdded extra property for snapshots when shelving\nChanged operations order so that shelved image is removed after instance\nis in state \u0027ACTIVE\u0027 or \u0027DELETED\u0027\n\nChange-Id: Ife02eaca633cdc9d2a09f64933ee72e60a05d56c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c461c40f6a3328df4c7c1fe435ef2b5cb0ec031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c461c40f6a3328df4c7c1fe435ef2b5cb0ec031"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
