)]}'
{"id":"openstack%2Fnova~909611","triplet_id":"openstack%2Fnova~master~Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e","project":"openstack/nova","branch":"master","topic":"bug/2054446","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-02-25 23:13:29.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-10-08 15:43:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2025-02-26 01:24:45.000000000","reason":"Change was submitted"},"2711":{"account":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"last_update":"2024-06-11 02:10:15.000000000","reason":"\u003cGERRIT_ACCOUNT_2711\u003e replied on the change","reason_account":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-02-26 01:24:45.000000000","reason":"Change was submitted"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-02-26 01:24:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e","subject":"Fix device type when booting from ISO image","status":"MERGED","created":"2024-02-20 19:53:47.000000000","updated":"2025-02-26 01:25:57.000000000","submitted":"2025-02-26 01:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"909611-bug/2054446","meta_rev_id":"300429de558b8509255e5f35f698aef998a6c38c","_number":909611,"virtual_id_number":909611,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-26 01:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2025-02-25 23:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-02-17 11:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"date":"2025-02-17 08:52:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-23 10:48:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2025-02-25 23:13:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},{"_account_id":36909,"name":"Carlos Bravo","display_name":"Carlos Bravo","email":"cbravo@itcservicios.com","username":"cbravo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-20 19:53:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 20:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-20 22:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-29 02:28:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-05-19 02:48:35.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2024-05-22 02:05:00.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-05-22 02:05:00.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-05-24 11:37:08.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-22 10:46:58.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-09-24 13:42:30.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2024-10-05 17:43:33.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-10-08 15:21:22.000000000","updated_by":{"_account_id":36909,"name":"Carlos Bravo","display_name":"Carlos Bravo","email":"cbravo@itcservicios.com","username":"cbravo"},"reviewer":{"_account_id":36909,"name":"Carlos Bravo","display_name":"Carlos Bravo","email":"cbravo@itcservicios.com","username":"cbravo"},"state":"CC"}],"messages":[{"id":"37f3d7c28ec8063f5b383c1740292a1dce5b236a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-20 19:53:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1df12b3419f718e06ffeb1e259d3f3e64e935c64","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-20 19:53:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c5f3312e165dd677a5281375c7e576e9f5e22a67","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 19:53:56.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42644/","accounts_in_message":[],"_revision_number":1},{"id":"a97d343bf8ba0a90084389b730b7a02b68830b7e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-20 19:53:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42644/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"92566463cf791a78448bfbc6f3b8ce376455f39c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:12:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9a64c2b7194ab9b7f52d4b5866cc09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3cd012687e104abaaefc3a89a95940f9 : FAILURE in 10m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4e4518bb4ea44a458500e17b7f25fc5a : FAILURE in 9m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ccb3529c5e14203b8d7fde5be7d6e9c : FAILURE in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7dc6965ec119105b067941ba00817b62422ddd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:03:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/652bb27a04334f64af577ce7750bdec1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ef22d50351a452c8f65a5e9e1b3eeb9 : SUCCESS in 56m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9491423716c841ad83ffb61d3f42773b : SUCCESS in 1h 29m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6843cbb367c44be3aa0367e35c5d697c : SUCCESS in 55m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3fe363760814d9d9ab305b0892bcaf2 : FAILURE in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c46b7c0a8db4cdda7efd95207b9defd : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7e4c938ab8a4ac1b0d7a27b98645d77 : FAILURE in 12m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d5dededa30e43ea89c3c3f179e716fd : FAILURE in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e149d6e1cb0455286912d1af23a2192 : SUCCESS in 10m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1f4b41cf5732487aba9a6438d08ca668 : SUCCESS in 57m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/693e83e8135346169574ca6f732b8e30 : SUCCESS in 1h 19m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/af70d676ba0d4703bcce52386a397ba2 : SUCCESS in 38m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c57726fdc771468fabeb2714d01a19da : SUCCESS in 1h 01m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1bc4178232584a0581b73281fb3cefa8 : SUCCESS in 1h 04m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8c00cb640d4a4df380bdb4189c3091a1 : SUCCESS in 1h 58m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/a21cc2be67df4c678fc9e82f44b989af : SUCCESS in 1h 16m 12s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a7773619855e4a2fb4d029799cd4bfe5 : SUCCESS in 35m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9221048a476d492bb214386e9763bdf9 : SUCCESS in 4m 25s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9f4f101c1e7f43b7a3f48513fbde54e2 : FAILURE in 16m 18s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7fc683bfcfc14f308fa4e4db82b2110c : SUCCESS in 24m 15s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08331cf6a5a54b218a13bd8b6deec3f8 : SUCCESS in 24m 00s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10dddbd5a4f64e4194c85e6adae91f0d : SUCCESS in 26m 06s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85759656e45741c5b4061b646881fe9a : SUCCESS in 1h 23m 14s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/1affc6f8ccc84647bc915d66c3530039 : SUCCESS in 1h 47m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/08c70fa5f72c40978a28d4ab16b5d521 : SUCCESS in 1h 26m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d138171d846041419b29f7bcef27eb98 : SUCCESS in 59m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/af045d5e6bb8495aa1a4a87e53967b0a : SUCCESS in 22m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3b04e6b61f7a4abb83a49c4ed9f806bf : FAILURE in 52m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"faa78a81451ce1dce2c92361b1d8c77de974b5bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-21 21:09:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9c29e076eb7b36b4d339c6812892b086aa28bd68","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-21 21:09:56.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42653/","accounts_in_message":[],"_revision_number":2},{"id":"cc20b27cffa0407454228ebf301c0afe01d2026c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-21 21:09:58.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42653/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"c67d02f618f288768ff596b112e3c7b805a1037e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 21:22:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dec79f7317f848f59c06ea9b71651ac5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc22551856ef459586b132b2e44cdb0e : SUCCESS in 11m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/daa4e793ecd9493bbb6af8efc7cb3d4f : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7d6f1c5d12de45d4b82177eea6177934 : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db7c234045a4d856676385dc8352fe284800cc62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 23:03:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3084d2f669ff4258954a233080c56892\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a91dfddf8f9d44c1a51a94582f25e890 : SUCCESS in 1h 01m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3093b275c1db460c90f696dbcf92eb96 : SUCCESS in 1h 35m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f92622371b425385b4eac2577cc0b8 : SUCCESS in 44m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6578125e2d2748c492065c507c492d24 : SUCCESS in 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64220d3d861f458b8bccb4c9d4da9793 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d4e301f689440febc47cfade1e7b807 : SUCCESS in 12m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/793917fdee634b3192a9894f49350a26 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04d4691661674e4ca682b7559adc3863 : SUCCESS in 10m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9bfbd1bc146c40a7ba24fb5a1aa2576e : SUCCESS in 1h 08m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0633c7ad490d4ced846056442d49dae7 : SUCCESS in 1h 43m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/125d3586e82c4164a8bea7837049aad3 : SUCCESS in 37m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/072d8949a5b2492a84e46d8538e32e52 : SUCCESS in 1h 01m 52s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/20b23f04db7147619feed51d6ec4544a : SUCCESS in 1h 07m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ff8fd8518ce64564b68254d38bcdf3cb : SUCCESS in 1h 44m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/35770a33177f41989e7524c5ca4b9fc7 : POST_FAILURE in 1h 15m 20s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/53f1bc93a6aa48f18fd420e2a5fda25b : SUCCESS in 44m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/eef9e5cff5cd49618879b0dd60afce6f : SUCCESS in 4m 52s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6c58fdf5d6e84906ba130ecf124dbe5c : SUCCESS in 13m 05s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/745c126d80be4178862ee461e484edf3 : SUCCESS in 23m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa2bbb95d6f841adb297c4a3b17073bf : SUCCESS in 23m 58s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2acb6a3cd388431f9b2231b85504bdef : SUCCESS in 19m 43s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f1e6f73e9f0e4a358d28159a8ba5e417 : SUCCESS in 1h 38m 03s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/724c3f5bcb344babbe006118a10e68a7 : SUCCESS in 1h 40m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/406d84acabb3492b856fdf5fcf25741d : FAILURE in 1h 02m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54aff3286e2a44e885cb787286e53664 : SUCCESS in 56m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/674f388fad234db798b01ae41edf2e5a : SUCCESS in 18m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5f14b2891cc54c63afffb5d0e98ad55b : FAILURE in 1h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b8bfc07b150819d91545dc2d7821504f1dee4dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-21 23:08:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a9319c5c8c6dd71b9107f3eb4b4362a6c1858fad","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-21 23:08:24.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42654/","accounts_in_message":[],"_revision_number":3},{"id":"6d29713d13b4c86d98606a75b6ebb5de90a3d75b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-21 23:08:27.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42654/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"40feeb298c20591296cf38d404dcb2d08bab143c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 23:39:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9821ee7d7df249d8a5d280602669c571\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7658e099e88c427db5e29736fec71143 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b38f1272806743f88276c2cab4c909e8 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8732ad6fe4f2428bb78ec9822c40dfbb : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2ccc509fe6aaeb01bb3bb8454ef3cfe19658f370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:04:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6b1e24601b645f78836d8acc1a53bd1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c86c8197bc94b329ca08182732ce03e : SUCCESS in 53m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e70c3475b78c45c482ac7acccc09b6fd : SUCCESS in 1h 23m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d862a3608d94bcbaf5f04664f2fd26f : SUCCESS in 1h 05m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3fbfea8ce3743a0b16b6c22e1f72861 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1230e89e22a846d8896a94cd34b39a72 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b90ca722ff24dda8ac75ec831c4e8e8 : SUCCESS in 11m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e782a58842404509aac76e72bccc3c85 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2beb9ce5594733bcdcb01b5b2018e1 : SUCCESS in 10m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8af91af0c56642a5846c831328157d0b : SUCCESS in 1h 06m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d011e1c332924422822fac3feefa7eb3 : SUCCESS in 1h 30m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3de4702931874cf8921dc9a10007de9a : SUCCESS in 50m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/768b99fa196f40759a5343fc51bebcb0 : SUCCESS in 51m 29s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e619468422604da2aa4f601959b55d96 : SUCCESS in 1h 00m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f6aecb1470554d2db6c6ce27009b2493 : SUCCESS in 1h 49m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/aa11a9c644a7438e8700c75dde91ecc1 : POST_FAILURE in 1h 07m 53s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/621b711cde964d6f9351a9606a5a0336 : SUCCESS in 33m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/20c010decac14e2cb19e51a2b8b9c61d : SUCCESS in 4m 55s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cd49804efe574894ac7d18c085391ee9 : SUCCESS in 15m 35s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8e79b7a08bd341c1879901f187b7fcae : SUCCESS in 27m 19s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97ac1736496b4524aaafb38ee81aa47d : SUCCESS in 22m 34s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ded945f19c7948328998ca9bd609ed4d : SUCCESS in 25m 02s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a935f984bb3c407cb53f1d9ffccfd811 : SUCCESS in 1h 38m 35s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/e9a2e5790d7f4eb9a2c8049fc9b1b715 : SUCCESS in 1h 28m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f1d4face04c46ceb5a62c3c22c0e51e : SUCCESS in 1h 33m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a095cc791868400cbe62d6afb4181a0a : SUCCESS in 42m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9b29a4e7e05941d2b19b3458c449c319 : SUCCESS in 36m 29s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7381e1961b3249e29cc75f3aa338502e : FAILURE in 52m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fcb9a5cfbe75871b66853057bedda5365f7e19e1","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-22 11:49:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad88555a033246c78ab0079e5503ef46d7421e98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 12:03:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/107ed7f431a748f08724711333b28566\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d5ff502c2254fcc887c169d9e9dc472 : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a84d4cc2a65a4925b906467fd11480ca : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f6ab5c91f524331b63dbbb8d5f28fe0 : SUCCESS in 12m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb265150506c8e4ad7d8db3feba3c587cc5a1180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:46:59.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e91bcde834fc4d68a5ab85c67038668a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c06ad71ce47840e5b3f53c19f46a7a43 : SUCCESS in 55m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/46de1705c0fa4250aea34de6f7ab604a : SUCCESS in 1h 32m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51f67616d623437d99fd58561facbd7c : SUCCESS in 1h 03m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a30255ff3ba841c6a2307e847fa42c63 : SUCCESS in 24m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50343d4efd95424e9344f74129f5ef68 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cf67d0abf2d4dc5b6fc6bb2d93cf7cb : SUCCESS in 17m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a4ee22fd21149649c607a358dfdf427 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c8851ed96b4dd6bf56436ce722c917 : SUCCESS in 12m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c5215f0a17e04567bc7cb2a1eb56cbe3 : SUCCESS in 49m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/481a7d25979446338e733007e34cde9d : FAILURE in 1h 50m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8d1f19712e254efb9967c95a390c3503 : SUCCESS in 47m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7f1cbbb13db84d0cb092793f21211975 : SUCCESS in 56m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/93ad4cd027374edea0fd64b5a7632e4c : FAILURE in 1h 16m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/65c16116834341398577c04eeadef8e5 : SUCCESS in 1h 48m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/e790a5474fe7456ca37b9e2434c33f36 : POST_FAILURE in 1h 32m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/448998ee064147a2a8f45301bf3e056a : SUCCESS in 48m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f8fb7c200d774ccfac5ea257cf089813 : SUCCESS in 5m 33s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/625c5540b5034922b6556f73432a7813 : SUCCESS in 17m 42s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/700cd9eb2d7b47719f8b5781fcedae0b : SUCCESS in 29m 05s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35b468e2fc7641a3af76801890eec3d4 : SUCCESS in 21m 54s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37fab533555547eca60119db3cba5ad5 : SUCCESS in 24m 36s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/62ed63aa95884e3fb196c566d916fa0b : SUCCESS in 1h 47m 52s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/0da364b242b64d1daeb13b8bb3db684e : SUCCESS in 1h 31m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dfcdce836ad4671a4bbf9cba4879d9f : SUCCESS in 58m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7043f578cd0b44f69a3322f05edca17b : SUCCESS in 59m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c038d536c7724c5ca48830d60945d844 : SUCCESS in 31m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/43a6dace2b0043efb01cc89d7bc86ec1 : FAILURE in 1h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"38c2cccf7427988025cdffacb994bfefe8dc3edc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-26 11:04:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f377b4d4b7ceaed6bdacaac3f59d87f5cb83e6b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:20:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9798cd095524ee1abf3b83b9de62e6a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df99bba44b37472e9bfcda9a246b01af : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/773f961f7c0049fab92a09b19dda5cb3 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76310e4ab67242ceb5489f497406aee8 : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b319cb52d25277c968ed04a980c72c227313e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 13:06:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2d6ba8b265449281afebe398da5487\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/edbe8f98c8a14f1d80ab04e273ac1979 : SUCCESS in 1h 13m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6ad6f38937164104aee6c569c3b84665 : SUCCESS in 1h 45m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00d5e020bfb9443f8a738fa40376b52f : SUCCESS in 1h 01m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/718f2a82595741c8b3766db7afb05adc : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f7defe3e384bc5abe1bc0219e0291c : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac13f94797b64792bb6bca82205d6954 : SUCCESS in 14m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f26af4b82b744d5da8d83040a8a90ac2 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8940db40c034d6887f365167d5af98a : SUCCESS in 10m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/98f484103f3d4e2492aa42fa001219b1 : SUCCESS in 1h 09m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/99ebefb88a714487a96ad6a7feb3aabe : SUCCESS in 2h 01m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9a2aec579d944c0f911a147f2f52ce55 : SUCCESS in 45m 12s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/51f04c7a9fbc457a84a3b51954707367 : SUCCESS in 42m 15s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a978ac31ffaa419c9342a36148a1a0de : SUCCESS in 59m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/30a969f2ead147d094ed5fabc7676372 : SUCCESS in 1h 49m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/70dbb4e5530d4331b75eafbfd0b4779d : SUCCESS in 1h 39m 32s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/fbc455f8911946e4aa8b340a3650926f : SUCCESS in 50m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a5e4f07203e44fd09f6a99955d8a1133 : SUCCESS in 4m 47s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d8cf57d1c962492397e97c3a09d5f734 : SUCCESS in 13m 17s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83d96f0dfa814468948b0fc397397076 : SUCCESS in 24m 08s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f6c779d3a714d0c98b3d56d5753292c : SUCCESS in 20m 19s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28a97c91f54146e7855ca8c55e55c6fa : SUCCESS in 22m 43s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/888e8947848f4b818e4ba61a1068acbd : SUCCESS in 1h 47m 06s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/233fc53a6c9a457bbfd115f8c9bfa65b : SUCCESS in 1h 54m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/85d1bd58fb034f1db4128493459a6dd6 : SUCCESS in 1h 14m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a23e7792aa81457a9a77e2bba0821034 : SUCCESS in 59m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f74f3046026b4d9d9d059c698261d2c0 : SUCCESS in 33m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d6a329830a224ad68380a88d3130a1f7 : FAILURE in 1h 04m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ef9efa7593e4f52aa9df91b9d85179ac876e916a","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-05-19 02:48:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"01f30d5aec556d43f6c5df7e7a48db149ae69944","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-05-21 20:45:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8dcd54e0efc96f190220a6a4d079173f6f6ace3","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-05-22 03:13:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c58e93380768e71b94b5aeed80dfb74f5d80fd7","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-05-22 03:21:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6243b23fcab69ab699b5eda1c2f1db35b2f367fa","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-05-22 03:22:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f0f81b872fcaac324220047673ebd6b6772c30ed","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-05-23 02:53:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"14bce31749d4d34a054114195985699ac7b5c444","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-05-23 03:14:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60b59cd82da13715e862f60763991179a14874ba","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-05-23 17:39:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd3b4176117041ccd4b08c95c4ab9b357bd79405","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-05-24 08:06:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"59e37dc72df973b6e719a725251ffb84759bd89c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-05-24 11:22:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eef94b938993258efbb57e27ae6e671dbe9a789c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-05-24 11:22:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0183ab28e1851a8b3b120ce493e0e331ad110138","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-05-24 11:22:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"16f601766b68ad8f186afd046789ad4fa8dbd09f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 11:34:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173c3e354eeb47afb25a25fa8188f7e6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c46185691944eadb28ab9d162820e48 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7ad1acff5d2498689a0daf247f2fe37 : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c29e11d6059304ec8a09d69979af113c9fd8a299","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-05-24 11:37:08.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-w66md/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"2b0e5464283b423f078d94b8c2fe039720dd2157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 13:18:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e30116d9a9f4bdda9843f9eb1cc856a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cde99ee74acb454889630217cc8d26d4 : SUCCESS in 59m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/55c9eab1b7af44f0afff45f45884e164 : SUCCESS in 1h 20m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/031284d8904941319bf6ac9c8fe1a736 : SUCCESS in 55m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36b0f763829b42f18d5d648763f8c5ea : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e62aa0a947b4a0fa016e0f49a9d3166 : SUCCESS in 12m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7016283359184933a556cb31f71dce8d : SUCCESS in 12m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b63114ddcc2740cb92db59afa37868d5 : SUCCESS in 8m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b938f5d16fa456da02e538459223d0d : SUCCESS in 18m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612dbf2ae6424861a57f48f2b5ca1abf : SUCCESS in 9m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/328eeb3c27bf4703b8b62b098835b88d : SUCCESS in 1h 05m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6999493a7a7042e88c9521cc4ba74cd8 : FAILURE in 1h 49m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5751f63d3f244f1ab01d530d022d986f : SUCCESS in 31m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1a4e5d767fbf47d7892f3d1dac38cf67 : SUCCESS in 43m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/56a5f5bd28894c12904d3a6380c4ab95 : SUCCESS in 45m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b6f966c4404d4ba698d5f97e8ef70256 : SUCCESS in 1h 10m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/e375034e79bd42d1b52c29ffc72e1fbb : SUCCESS in 1h 44m 27s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/72a92a84a379431f95d542c2675824c9 : SUCCESS in 41m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a88545ee9c8646f581ff8cfb99a9dc1b : SUCCESS in 7m 06s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9c63c1cb46641ef8ffb191d89a8bc22 : SUCCESS in 29m 27s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1be911c69974e9e9ff43357ec5ac3c1 : SUCCESS in 24m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c338b717a0584e459cc9612238cf4ff9 : SUCCESS in 34m 40s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/545406ef4a5d4089807e78ffb56265da : SUCCESS in 1h 44m 04s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/26aa75a143c0497a98a3ae4af303104a : SUCCESS in 1h 34m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa4e624f17204cb7b586608bb89ca18f : SUCCESS in 1h 00m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7946ccf0c4e048ac82c5bbfe1aa1e1e6 : SUCCESS in 40m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d7e496b15d76468bae36df49c17ee70c : SUCCESS in 34m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2fe4b0af713c4b1cb8bcd6fd046c582f : SUCCESS in 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"08d83d974f680d6afb3c336a0a41d90fe10226bd","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-05-24 13:21:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eabb4e3e2fdb84f78c5ff355b8e05c85bcfae87c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 13:33:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82bbe04a5cf4202b18c000a607d6f5b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa98803048d844bb973e1fe17dd99406 : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e5ae1e31bcc48b5ba7b05921ffb3be6 : SUCCESS in 9m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cb6b9577ce88a37fdc1271dfcde47bf8ad7a7a8a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-05-24 13:40:12.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-9c22m/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"3df3b62e382bd7b3b2da418b4c977b451d9d1d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 15:29:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ce1c45820d141d1bc70453360ce5b1f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67c54e7796eb42e18dbde6ac9944658f : SUCCESS in 45m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/513fefa2e88844999a8e58c1a272809f : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ad6bdbfa1a49efb4fe99312341a3b3 : SUCCESS in 52m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042eebae81d54678ad99eae225a54472 : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b782a4defc874223ac57524fe81f5e20 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1b0d858113841d9b6efaa2305a68333 : SUCCESS in 12m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d1b30ea326441d0928899993b98970a : SUCCESS in 12m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f18c9a207ba34fefac53eafa34e33389 : SUCCESS in 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c78e007de84f60988751c83fc66557 : SUCCESS in 10m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/548b31dabd41478dabb1d0e6fcfe7817 : FAILURE in 1h 03m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89925adcc91c4a87ba40c5216dd38a74 : SUCCESS in 1h 34m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9d72411dced44a4fbc2abe744445194d : SUCCESS in 41m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/28c6b629774844918209ff788cd5601c : SUCCESS in 38m 33s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/990399c5be89492598fc757b062f61a7 : FAILURE in 1h 04m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fafd84f03d1e458bb3b9c7e624c25c28 : SUCCESS in 2h 00m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/7404ea07d3664763b483b713a47a0793 : SUCCESS in 1h 28m 08s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4d47999dc03b48d3aa2460a01d478d03 : SUCCESS in 50m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3cedaee6f6764d778704f7062260d4eb : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/746fd453f5f343ba8581a61edeaf5128 : SUCCESS in 21m 06s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/671d367b7dd142da872526e70de6fec0 : SUCCESS in 23m 31s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/be308927c3fe4143b9cd45ed5bf30fca : SUCCESS in 29m 23s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4c23e907182b4bc3811bc5dd690dbff1 : SUCCESS in 1h 57m 40s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/ea92b328f6c642d99d2fb85308affe63 : SUCCESS in 1h 30m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f18492893bb45a1b6e837ab0590d917 : SUCCESS in 1h 04m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/027b10625929437ba3923390bece7145 : SUCCESS in 1h 12m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9e389b4370904b36acd8ce8155a6f930 : SUCCESS in 30m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/63664bdfbb0f4fdead4f578bfb7240ff : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b9468ee38257d5eeba2d3d20a62e87bcab862d1d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-05-28 12:09:22.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-sml6k/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"b4546e72a0d550a0f646560445747ce4f39b5a3c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-06-03 15:52:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"639d82d1d23000faee86990624bdf9c58287aa20","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-06-03 23:38:50.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-gq9zx/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"0b3f44749903ecbae949f1b74d971fa701532c8f","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-06-11 02:10:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"46fc23fd003d719999dc3ce9c6b5be01505f6b61","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-06-11 02:28:19.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-6qqcg/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"48475c8298dc85c48d6d95599c7610ecbab74ea7","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-06-28 03:23:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b48cfc481740e942123ed5ea61bac8535f1c6a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-07-05 18:51:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4ec5e4edfdbee99b54501d00863ddb615b214c45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 19:04:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b925dc9064846f5ad7af7d5a0575abc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0af5c3971f7d475e8bb671cffef6e52f : FAILURE in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a0fb5a32576472eb9728ce09c3ceee3 : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2354c06e4d0103641fcfa653508ec45e8332cb39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 20:51:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/439b7348d18549eba57731f2df33a8bb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f8b50011f554c8fbc2d32df0d3fc9ae : SUCCESS in 59m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/09c420217f104e9ab1b7acd68dddb709 : SUCCESS in 1h 59m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ebda299d80042e3bc20227eb2d5871f : SUCCESS in 1h 01m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cdf987117354a7ba0105093eac2ebf6 : SUCCESS in 22m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076a2183739843aabffc80573f1ef64a : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab222b9b8d7d4d36b02e985140c11cb4 : SUCCESS in 12m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfe87b803afa4ffaaed663f2e5702f89 : SUCCESS in 11m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29837909701a48fa823d34e1f595e73f : SUCCESS in 17m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefe2768b0c04c27ab8430839a9aad6b : SUCCESS in 10m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66d785f048074e56b108b0be24926af2 : SUCCESS in 1h 04m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f61763f81ca8413f872d02859ff44223 : SUCCESS in 1h 33m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/11bffd09b26a423aba2986a28975f695 : SUCCESS in 43m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a1b34bceed4e4029a54c5fadf12adac8 : SUCCESS in 47m 17s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e9093faf2d8d49988b775ae11edcc3be : SUCCESS in 55m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2d490eaff85a421fada7f98bbdb0f089 : SUCCESS in 1h 38m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/4153fb9ddc814b388afd2b9a2f52c96b : FAILURE in 1h 34m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e21818776a854109b2a6caff82b19beb : SUCCESS in 50m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/331b7ae46e7c4b6dba0169bc042fb52a : SUCCESS in 4m 41s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5732e8eaf4e746579e2b7ba73288f864 : SUCCESS in 22m 44s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98b3acd395254402b3a579d7efbfd2cf : SUCCESS in 24m 52s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3c8e36f338424d4b9f671e957b44d4c0 : SUCCESS in 29m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d32608dd7b8433592f5bab18fc13439 : SUCCESS in 1h 51m 46s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/ae252f004abe4f5c8b67fa95c5d939cb : SUCCESS in 1h 53m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b1dfa2947d84d76b4f3534556af0894 : SUCCESS in 1h 15m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d7df87c2b2e4332b2a2d25f1a9d520f : SUCCESS in 1h 02m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2ddea19f58264f078925a534c7a5a9fa : SUCCESS in 42m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e5b0b5153b98485e804387a49cc9bc9a : SUCCESS in 33m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0b3e4214f06b0980f0823b20d6cefc0405030595","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-08-21 03:52:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e550298dc95c0f0337a09bdb2213a166a94e496e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-21 12:20:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bb52cdd2b5ba0d8a0846d2183fc019e53188400c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-08-21 16:37:09.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1bec0d2e2a0a1c87ede7a9fdb07bbcfda2deb933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-08-21 16:38:36.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5ab94491f45d45dd08aab70d0ebeaf962f156937","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-21 16:55:17.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5a2327c26c9c88545f148d881e79c25fd39e30b3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-21 16:57:07.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-gljks/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"75f09f2d935d552b9cfc378acaab45f0645e8be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 18:47:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cec947e9e946471dba33dcfa1b461291\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a45b0409b20b4c308b08df49d4481b1c : SUCCESS in 43m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f7edbfc0db2e467ca74db034c30d669e : SUCCESS in 1h 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c5ce93c0294e19b563871a32250e4a : SUCCESS in 46m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/037bd0bf283241cb88f5079376834a97 : FAILURE in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17343ec9f7b4478bbfcb998565ccc785 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ec080d2650e4a60988caf04e3239503 : FAILURE in 13m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c28d2e807b1549d690381fcf12786290 : FAILURE in 9m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3d5a45b58bf4f06b3c63823810af554 : FAILURE in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a678550058e4f45a30e3d97402598d1 : SUCCESS in 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4004d2b9e6b14306a6e231108569b829 : SUCCESS in 59m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5810ad349f9d4534b45edd9c5fc23daf : SUCCESS in 1h 37m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e38021dab2aa4197a65b517b0e6af7d7 : SUCCESS in 48m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f1732cfb63f74e8b90f5b1e5f8d16a11 : SUCCESS in 50m 21s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/90c15baee2594f3bae5e8d5ad452c948 : SUCCESS in 1h 06m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/de7d80f182a94200a1113c8b8654e9c0 : SUCCESS in 2h 00m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/c25945b2ae634fe59af69459925e5f61 : SUCCESS in 1h 54m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5fd7a389667a48618cf7f9e6a909bb03 : SUCCESS in 52m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c587d0c9f105413197b9f36395df7cac : SUCCESS in 4m 54s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37c3bb1937ce414a89ef17dba12dc99d : SUCCESS in 24m 29s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3acf28d997342088d2835377ccb6f44 : SUCCESS in 23m 14s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/583571533dac49059412685450a6a972 : SUCCESS in 30m 00s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/153724f20afd4b67ab9bdcd19ca6a4be : SUCCESS in 1h 45m 24s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/5bdb22f59e4840d19207637018a4ea9e : SUCCESS in 1h 46m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b9ae671ff254cefa59a9b4248c48dea : SUCCESS in 1h 10m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e8d7101def247d590efbf67f044ebba : SUCCESS in 1h 10m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/effe8b8d5ae24ad0afb620012cc72116 : SUCCESS in 28m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a86893e71904490ab92a50f10204cbd4 : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9e487658418fdec75139dd9199b8b61619f9b0be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 21:17:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/193ad9e5e21640a58a9461329640d5b2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6d5c75b2b824991b4151bad19163086 : FAILURE in 45m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bb0d01d671b42c8bb0d37c435dfb4f2 : FAILURE in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5ed61b9017d8fb8c42583d9da2ff022de39f05f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-08-22 08:38:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8718dd0417c5e8f171ce3bd393f3648da38a3d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 10:41:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce44891f35074dd6850762fa1f7f9eaa\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/755be121f0c64ef6a11ced8e54fdee6f : SUCCESS in 57m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f7f652189a9b4be79b5594b231832530 : SUCCESS in 1h 43m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/750f934a2a214bf5afcfc56a105e750c : SUCCESS in 56m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af69e033663c4dc79a11d091d31656b3 : SUCCESS in 33m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69de5855d34147c4a71a68a318bc2ae5 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85bf30a8689d495e93e6d6cce35dd8e1 : SUCCESS in 12m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7dfb09f8d37c42499b3f56cdbdb7ce99 : SUCCESS in 11m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec6525983fc942449c47ecdea2d2f57c : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a869b9c4ff4c77a57d35be3d7cbff0 : SUCCESS in 10m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/48ba535d48fc47a186d36ef41d11fa83 : SUCCESS in 1h 02m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/144e5002c5334e47b2bcb6bdde2bc2c5 : SUCCESS in 1h 46m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5bbb20bbbc20433aba0c78a8ee6df0a6 : SUCCESS in 53m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ebbfd6a9682b4a6f963f0568e5716f0a : SUCCESS in 50m 26s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f4ebd097e1ea4ed2aae7661320e6a138 : SUCCESS in 1h 00m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c26dd386219d463fb8bb3953ffdcac04 : SUCCESS in 1h 46m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/f66adb039d8b49968c6b05a548fb1d57 : SUCCESS in 1h 39m 16s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/567b6bd68779458ca0c22431c2f936af : SUCCESS in 47m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d51e5f2821344dab9d5ad7b8484383e2 : SUCCESS in 4m 37s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a7358d21e2a4e29a37e52a91ad811b1 : SUCCESS in 22m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f5cfac1a713410eb5ab5d586f88d7e4 : SUCCESS in 24m 23s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/72e930ec057e4ffa906d5508ef36be06 : SUCCESS in 28m 31s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0905b97196884faf84037a8671f98d3f : SUCCESS in 1h 48m 13s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/d90f2c53cded48929adf8186c6fb9eb7 : SUCCESS in 1h 47m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/163adab2afca4569a22cef80b44e2647 : SUCCESS in 1h 22m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a1c77a7051e4decb68143a915b2f619 : SUCCESS in 1h 04m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1f1e3ef4971847f19e3eca48abcbd31f : SUCCESS in 27m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/355d2eaf5ceb4ee7b91a0c4df4458379 : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f1518cfcabbdb5ac66f5048cc4b579c044a83c77","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-22 10:46:58.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"1cf31107d8d8b982df4dccb698dfd364b7a5c1be","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-22 11:30:27.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-278f4/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"9854775ed0938d6925a207fb16691d63ad948e82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:49:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c6e25612484fd4b1f1cca5f87f0cfe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0b18ec1d35a4352a53955f62da071b3 : SUCCESS in 50m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b2628ae460c418bac55bc359de378dd : FAILURE in 28m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ab139220f3274aa08c7b7ab3b82996224e221f38","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-09-19 05:29:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a954cc2aaaf0121dcb37235937eb169736457391","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-24 11:46:49.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1b07d0f5069b8b590366979fdacbc965ab920bce","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2024-09-24 13:42:30.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"66ded7d44e80a199cfa47eae2008a9edfd147be2","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-10-05 17:02:21.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f665399671e0012e24eaa9cbcfb9437d4382b2b0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-05 17:43:33.000000000","message":"Patch Set 8: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2d6969f9050347652d195b7dc795fc00f8d0c25d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-10-07 13:39:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c035c9c1f3a7a4bc36dabda4c7a30b7e0429c0ef","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-07 14:39:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4ff1e37bddf52db8f85135a04f3650855ebe2502","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-10-07 14:59:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d0daf7f81d9f1791e3544adffd2773718e8db085","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-10-08 10:52:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"57255b9152de82751bfc74df8e6384cba6b857c1","author":{"_account_id":36909,"name":"Carlos Bravo","display_name":"Carlos Bravo","email":"cbravo@itcservicios.com","username":"cbravo"},"date":"2024-10-08 15:21:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0dd2229712e5a8a10e93851bd607138c084227b7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-08 15:43:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fc2a59cd064050d97ca1e98bd5000965b5cdab8d","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-10-17 09:02:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"078095f4eaa7e8b22b9527a07ee005ae2faaaf44","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2025-01-23 07:05:39.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7c707e8a8df8f2ee5c9e3dfd9389f7eb1b26bb43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-01-23 10:48:56.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"873287725b104f2b562589f79f3e41fe9452755d","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-01-23 10:49:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7d713a562de787d1ecf8d7db2af126c6f8c4b290","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-23 11:35:18.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-5b7cc/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"0c864ac0e68101d5a41d9d1f221ce6a6fe2ceb07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 11:49:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f610a03a6b744d80bb360e1a11cad6a2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2fe4bc5af9454f9a929b8b19479e0cf4 : SUCCESS in 39m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3dedfd3dfd44094930d2c04b770caee : FAILURE in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"65b886b715283e1ddffb31526d889b00b614db22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:12:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfad7f67862c45fa9907aec4382b955c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb7770cc3e4542f8a5dd2a38de81f5bd : SUCCESS in 48m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6478a3cb2271446688be360c2e06aca3 : SUCCESS in 1h 47m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bf417ef3e334a83b65364dc477dd3ff : SUCCESS in 1h 05m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6c395e772b54887b4a7cbc7fd544aa4 : SUCCESS in 23m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cee06315d804713ae41f112a836ea13 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f5dfca39a82475e8a1c2ad75a97ed6a : SUCCESS in 14m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4e66f52600245f685d5c83e13e416ad : SUCCESS in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fceec26194794cff955653a3f3a47fe5 : SUCCESS in 11m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4a57eb27e51c4660ae4c1140437dcd20 : SUCCESS in 57m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac35b9d532034c0e910effdc4a3cce98 : SUCCESS in 1h 13m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/14f2141e12d84baa9afcd0006a5b3c51 : SUCCESS in 1h 00m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/48dc7f5463814744b132e01b16e2631e : SUCCESS in 46m 34s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/28532baa4a01483094479fa8054f5452 : SUCCESS in 57m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2ad46d72a7f74b5bade86b52c892d67b : SUCCESS in 2h 00m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/069c1d492d964a2f8ec6e33d53032e07 : SUCCESS in 2h 09m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6b632aa31f4c4bb8864835d5edcc0950 : FAILURE in 59m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/41ed298e713f42c18b5ff53ba8b83ed6 : SUCCESS in 3m 17s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ea76170cd424b9fa0a2bb5c0ef211dd : SUCCESS in 27m 25s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a8f67cc83a34b8c98d76669872924b4 : SUCCESS in 23m 47s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a9ceaa51ad844816856a1beaee1cfc0c : SUCCESS in 31m 54s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c89193c1caef415dae5f3f0b8127155c : SUCCESS in 1h 09m 02s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/4fe3be291ab54de3affeabdd64178d71 : SUCCESS in 1h 42m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3465e83a2f6643b8bd723ff4c057e3a9 : SUCCESS in 1h 28m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a22908adcf44ea89c55f2dd8bf85fe8 : SUCCESS in 1h 02m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/562030f5c8ca4eabab6feac72601f15e : SUCCESS in 43m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f7585d55b5ee43459143b37791f9fe51 : SUCCESS in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"332747d84d816e916c87d53b93b882b6b340d57b","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2025-02-17 08:51:37.000000000","message":"Patch Set 9: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"da34cee20eea2548be0702cdcc2ce94b2c9cae87","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2025-02-17 08:52:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2b4516fe13ff1bf0b5650b7a053dc15f94ccb135","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 08:52:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383284505eb641b7886421b011d4e6fa\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa545ea4fbe54e2e980cc0e9a5ee994d : NODE_FAILURE Node(set) request 300-0026355264 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a63e7c422fc4ffab7a18f3f6097731e : NODE_FAILURE Node(set) request 300-0026355265 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d991164a7c930745d277d34812968ac6c7f4214e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-17 09:14:54.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909611-dvg8z/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"8f3d9a15df182bcd4b155a85df8ed46103acbab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 11:00:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c94e79b96302423faa050675871b668f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/28323d03c4be43499644a69baa090008 : SUCCESS in 36m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/69db06724ac24ede8c488726504e09c3 : SUCCESS in 1h 55m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a6ee763ebdc494abda4bb8d37cd1987 : SUCCESS in 1h 12m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a4b81cfae974334a9fef797694f3dae : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94057189a8da43a2b608e392084b2911 : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f97cbc1aec144a0eb674ff7e17534b50 : SUCCESS in 14m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bbbdebb535d4236b70cc73bfb9aedd5 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07fa61a41f44f9692e4e442a3ff3aa3 : SUCCESS in 6m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/78428c3681124574907737afc7b645dd : SUCCESS in 59m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f496e4ee582846a4b01419da83a7481d : SUCCESS in 1h 15m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3c0fd2c2cfa9462991a82273102cee65 : SUCCESS in 54m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/63878f95e67341dd8187a8f581624ccd : SUCCESS in 55m 52s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d0b9dd2ad0214a07b846ecb340d65de7 : SUCCESS in 47m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/205e0e88e8dd4b93be982a588d88c2bd : SUCCESS in 2h 02m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/bcf9ea7c98484d84ba39244efead30c3 : SUCCESS in 2h 06m 51s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/41e5cc1a976f4b739a07f28c5a5c33bb : SUCCESS in 52m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7d9e5355e33a4e0485c208bce592ac73 : SUCCESS in 3m 17s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e89b386bf1f94a9dbea52dfd2bb7a6a0 : SUCCESS in 25m 33s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5ab1e3ec32947ee94d65de760cb5f18 : SUCCESS in 20m 18s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2137d7f20d3d4ac695a5119bf6afcee2 : SUCCESS in 26m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1503d61f4c394cfbaa86fc01fd354f89 : SUCCESS in 1h 16m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63f68a543bc8468b8e27f118e6a06039 : SUCCESS in 39m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6ae9ce11da2047219a3d6210f6d0a2d8 : SUCCESS in 38m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b997f21fac0b463aa0ea47c0a96cbeb4 : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"064d1c4fcbe378a1525957349c09514a419fe560","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-02-17 11:02:11.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"afbfd2d87ae9a24362d8ebf5a187690a5b573500","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-17 11:06:32.000000000","message":"Patch Set 9:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":9},{"id":"1b340aa2f1b046b3fdd7a7cae7c259bd01e923bc","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-02-25 23:13:29.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5525673d1b0f3c6eecd214771ebb5acdb362e77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 23:14:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5c31e235c558317d2d2b34868981525badb9a192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 01:24:45.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1a69bccd1c4fb9acb38a97b6987c62\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1712c8760ee431990df2146e4333b82 : SUCCESS in 54m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3d98c8caca44446eb89e3bd7e4979174 : SUCCESS in 1h 56m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93ea75e6912044d99372f3c503dcd659 : SUCCESS in 53m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f359989a3efd45d7a15f66e4e30724c4 : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1595e7c9fe1941b09cac1eebc0b34e16 : SUCCESS in 13m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54b7eeb164ca4f6ca38060dfb8a97c21 : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870a213072ee45eab512d1bec02a5ed2 : SUCCESS in 8m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a938234eb7b345c0a8efb5a10e770d65 : SUCCESS in 39m 10s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1802196aee8e48b6b68904f139ad7a82 : SUCCESS in 22m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce02e976f58949288de56860c26b0125 : SUCCESS in 23m 12s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2ea79e9c075448cea6aeff6eed6eb5e5 : SUCCESS in 28m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1ffa7feba6a04331904959d21c9ce922 : SUCCESS in 1h 41m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/f66d084ede6548c19176195ced1da281 : SUCCESS in 1h 51m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/07031dbf361a4a4fa899cf6534685f73 : SUCCESS in 4m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb500b47869a4c56a08b3a55b2254a0f : SUCCESS in 1h 03m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/14d472bab6404932a318ce2789e114cb : SUCCESS in 1h 37m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8b6be02996f4b42b8528084b9c8897b : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":9},{"id":"a046d7fef50c1846dd7c5dcface0681626c7fc43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 01:24:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"300429de558b8509255e5f35f698aef998a6c38c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 01:25:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dbb685c845e4da7a7154e16e00aaf58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf43a272c114063a579b6c711846eae : SUCCESS in 53s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"96a5c21f24f6a89411d12677ae5fa75adf26ed61","revisions":{"bf84caba2585742b03771dbacee284aed20fda9e":{"kind":"REWORK","_number":1,"created":"2024-02-20 19:53:47.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/1"}}},"commit":{"parents":[{"commit":"6e510eb62e00c34e98a5245a6de2dd2955ffb57a","subject":"Merge \"HyperV: Remove RDP console connection information API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e510eb62e00c34e98a5245a6de2dd2955ffb57a"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf84caba2585742b03771dbacee284aed20fda9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf84caba2585742b03771dbacee284aed20fda9e"}]},"branch":"refs/heads/master"},"912354be51e14de47bc10a74c5aee1aaeb913072":{"kind":"REWORK","_number":2,"created":"2024-02-21 21:09:46.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/2"}}},"commit":{"parents":[{"commit":"6e510eb62e00c34e98a5245a6de2dd2955ffb57a","subject":"Merge \"HyperV: Remove RDP console connection information API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e510eb62e00c34e98a5245a6de2dd2955ffb57a"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-21 21:08:15.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/912354be51e14de47bc10a74c5aee1aaeb913072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/912354be51e14de47bc10a74c5aee1aaeb913072"}]},"branch":"refs/heads/master"},"564780c1747582bac220db8e67b678da2483da31":{"kind":"REWORK","_number":3,"created":"2024-02-21 23:08:12.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/3"}}},"commit":{"parents":[{"commit":"3209f6551652cff7bef0b9d9719ab940dd05a0f8","subject":"Merge \"HyperV: Remove RDP console API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3209f6551652cff7bef0b9d9719ab940dd05a0f8"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-21 23:06:41.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/564780c1747582bac220db8e67b678da2483da31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/564780c1747582bac220db8e67b678da2483da31"}]},"branch":"refs/heads/master"},"8b915e058122448698a0d4747ac93dd0581dd4cd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-26 11:04:00.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/4"}}},"commit":{"parents":[{"commit":"065f6e1b51ffc8f15953270c98d225d9f5d15644","subject":"Merge \"trivial doc fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/065f6e1b51ffc8f15953270c98d225d9f5d15644"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2024-02-26 11:04:00.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b915e058122448698a0d4747ac93dd0581dd4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b915e058122448698a0d4747ac93dd0581dd4cd"}]},"branch":"refs/heads/master","description":"Rebase"},"ef5184dc30ec0d1bf371a1aede5e42e8294eea3e":{"kind":"REWORK","_number":5,"created":"2024-05-24 11:22:38.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/5"}}},"commit":{"parents":[{"commit":"9ca0ba58dad1db459a992dba0442c4af3086daec","subject":"BlockDeviceMapping object has no attribute copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ca0ba58dad1db459a992dba0442c4af3086daec"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-05-24 09:08:44.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef5184dc30ec0d1bf371a1aede5e42e8294eea3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef5184dc30ec0d1bf371a1aede5e42e8294eea3e"}]},"branch":"refs/heads/master"},"3746a9bfa9850030015878604bab9059f7642bf5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-07-05 18:51:29.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/6"}}},"commit":{"parents":[{"commit":"a3b15ef88d1be683060f7181476e53b5ed37ea0a","subject":"BlockDeviceMapping object has no attribute copy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3b15ef88d1be683060f7181476e53b5ed37ea0a"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-07-05 18:51:29.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3746a9bfa9850030015878604bab9059f7642bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3746a9bfa9850030015878604bab9059f7642bf5"}]},"branch":"refs/heads/master","description":"Rebase"},"4ca2e75db9eb4d080390379c994d52ffa95c83fb":{"kind":"REWORK","_number":7,"created":"2024-08-21 16:38:36.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/7"}}},"commit":{"parents":[{"commit":"a2cc66a3775403edba369d8f822ab93592bdde21","subject":"Merge \"Fix deepcopy usage for BlockDeviceMapping in get_root_info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2cc66a3775403edba369d8f822ab93592bdde21"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-08-21 16:27:10.000000000","tz":180},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ca2e75db9eb4d080390379c994d52ffa95c83fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ca2e75db9eb4d080390379c994d52ffa95c83fb"}]},"branch":"refs/heads/master"},"a4632d0b7277ce6165559eac65272a9a74a469a2":{"kind":"REWORK","_number":8,"created":"2024-08-22 08:38:41.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/8"}}},"commit":{"parents":[{"commit":"a2cc66a3775403edba369d8f822ab93592bdde21","subject":"Merge \"Fix deepcopy usage for BlockDeviceMapping in get_root_info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2cc66a3775403edba369d8f822ab93592bdde21"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-08-22 08:30:30.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4632d0b7277ce6165559eac65272a9a74a469a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4632d0b7277ce6165559eac65272a9a74a469a2"}]},"branch":"refs/heads/master"},"96a5c21f24f6a89411d12677ae5fa75adf26ed61":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-01-23 10:48:56.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/11/909611/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/909611/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/909611/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/909611/9"}}},"commit":{"parents":[{"commit":"932866d078cdec51ad654aa0626a635e65975b7f","subject":"Merge \"Run nova-next without periodic cache healing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/932866d078cdec51ad654aa0626a635e65975b7f"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-20 19:47:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-01-23 10:48:56.000000000","tz":0},"subject":"Fix device type when booting from ISO image","message":"Fix device type when booting from ISO image\n\nCloses-Bug: #2054446\nChange-Id: Ib4cc34c30a8acf09a1ab6be89d0faa76fb6e705e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96a5c21f24f6a89411d12677ae5fa75adf26ed61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96a5c21f24f6a89411d12677ae5fa75adf26ed61"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
