)]}'
{"id":"openstack%2Fnova~909806","triplet_id":"openstack%2Fnova~master~I787bd7807950370cd865f29b95989d489d4826d0","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-03-14 00:06:49.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-03-13 19:41:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-03-13 15:07:03.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"I787bd7807950370cd865f29b95989d489d4826d0","subject":"pwr mgmt: handle live migrations correctly","status":"MERGED","created":"2024-02-22 01:02:02.000000000","updated":"2024-03-14 00:07:56.000000000","submitted":"2024-03-14 00:06:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909806","meta_rev_id":"66345aa3adb607b56f826e045bc2413d350f5e64","_number":909806,"virtual_id_number":909806,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 00:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2024-03-13 15:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-13 16:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-13 16:36:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 01:02:12.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 01:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 03:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 08:31:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2024-02-22 10:38:37.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2024-03-11 12:55:53.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-03-13 16:36:00.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-03-13 21:01:51.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"96a739fa51a051ede51f013bff4180e4eb0f20ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:02:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ea8fad92dd42d60e5a011f53f8dbcae732ae55","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 01:02:12.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42655/","accounts_in_message":[],"_revision_number":1},{"id":"6ebfee6fdaa3a8082e594513369a3dcec9f0886b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 01:02:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42655/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"e834b866bd3f98afc71d1e8bba136d9313d1543b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:22:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c42b1629c388a052f8f72660a5d49faa2831d9f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 01:22:41.000000000","message":"Patch Set 2:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42656/","accounts_in_message":[],"_revision_number":2},{"id":"8a567680b99156e88f6f0f1aac28ca753ce6fdb9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 01:22:44.000000000","message":"Patch Set 2:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42656/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"fddfb03e233d8981c26428606ac18e1523eccd23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 01:24:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bb7f7352445dc23ea835fc488bdb3cc4d98ab5de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 01:24:26.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42657/","accounts_in_message":[],"_revision_number":3},{"id":"c995d19005626093946d6804f398bb07059f45fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 01:24:29.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42657/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"16103b6b335e0101dea240cd8b5c1bf3e2661412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:51:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/005afe9e423844c0aea63625b2b52c98\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7be2665c7d0d496aab8b727ae092e9d8 : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efcb94b0a586458d9b7ea2c438442d30 : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c0176c7e4bbb4e97929e132eda6a87f4 : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e794307a6f44ddeafd7a3f369d8bad66dd4da3f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 03:42:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac36aa84d7644cb6b87721adb706dda5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d43d8dec2e83467aa9523c43f2d9fc93 : SUCCESS in 42m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/76389c611f8b4589b856a5a6c12d6586 : SUCCESS in 1h 34m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d37b01eaae664b7da1159fe48406e9b3 : SUCCESS in 1h 02m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3f62e8df3114501b69993d89e121087 : SUCCESS in 20m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e88d6b42d414ccca8c84f4f159dc4b2 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17da896cb0c443ba8021979ef357f2c9 : SUCCESS in 10m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c62d1b33d9b34f4c817f30d72d824fe0 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764a62b83b89421dad918a62cad3a6c9 : SUCCESS in 10m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/800da72d7fa842ff89922c52b7d14995 : SUCCESS in 59m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f0b94a799ad48cc9a6dbfc512b04fee : SUCCESS in 1h 54m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a7b7abeed8cf4254af72e556bb1bcf47 : SUCCESS in 37m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/22516af60971434eb2cefc4b7ed664d7 : SUCCESS in 53m 56s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1106c05de40f484495cc647799e5aaf9 : FAILURE in 1h 34m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e956fe889910473482661998b9e1ac8f : SUCCESS in 2h 11m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/3b740140f12a4443824e42c17f0786d0 : POST_FAILURE in 1h 25m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/174c757af6294da7b699dc2795a0280a : SUCCESS in 33m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/52c726528f1749dc9e965a0a44f3454b : SUCCESS in 4m 14s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5ad5423d24b34d408a6c5a0eab148690 : SUCCESS in 12m 21s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/847f1d4c60a041a1829d935e0184597f : SUCCESS in 25m 03s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f2cc904984545518f61910a3f5534f3 : SUCCESS in 18m 20s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b729af9fff74a36bae6c1de0d7503ca : SUCCESS in 20m 47s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/12f34f5cd4ea4107bfe91db7fc1aa389 : SUCCESS in 1h 28m 16s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/dd0e55a9e27243b0a7936356aec0fd0e : SUCCESS in 1h 39m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8fe40e577224598b87a699117be1d55 : SUCCESS in 1h 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b416f0552dda44c7a30ea0301b4369bc : SUCCESS in 36m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/13e1b43741e44381ac921d072a1172ad : SUCCESS in 17m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9d6309a940c54995b10ef28e56eb8edf : FAILURE in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4eb457d3315dfdbf51ba72ce8bf79eeffd121979","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-22 08:31:07.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b546293c71016e707e56d1bebc9483928e292246","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-22 10:38:37.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4971c7551171d87c659cc630d17f22881c733677","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-02-22 13:12:21.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4ba4ce1fafba298d3af9648946ee9f3d112738bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-22 15:47:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e53472c779c93bb94f6f7672abfbef350316ef80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 15:47:33.000000000","message":"Patch Set 4:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42675/","accounts_in_message":[],"_revision_number":4},{"id":"6fbdfe42c9e6f935348ed92b79c99451914be5d9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 15:47:36.000000000","message":"Patch Set 4:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42675/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"c3346925bd9c8092a19450d59b96c036a3807413","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:16:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cc2e0ee3b894fb5b159a81b95ee5592\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8874d73fb90645bebe85515aba19ce2a : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89d74b45438a4f47baffb20d8773f27f : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1fc093ffbed64e78b9a7c3026c432d7d : SUCCESS in 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e19aecbbe819f09fc65f9fd28eb4619d17fc26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:40:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bab20059955e4a5bb9c5b16aa357de42\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a0f2f72c388249e684154e452d0ec4e6 : SUCCESS in 50m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dccd0f3cbdbe4e1fb3bcaa1bfc010b7f : SUCCESS in 1h 43m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2339b0de5b364a3a85b81c53d4b61391 : SUCCESS in 57m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ecf573d42e04cf4bb022e1fa7456467 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca9fda6d3094a259eeb8d801d1cd276 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91793b251751452e88fc26e680615b27 : SUCCESS in 13m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c302c2c3c6b84077b9694d74ab15a1fa : SUCCESS in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675a5f55ff1548edb9ff4aea5f1de5f7 : SUCCESS in 9m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ead858f9400a4d3db4ad1ce32e5e785d : FAILURE in 1h 18m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a5eba968d4f43a58a7296c8460bebff : FAILURE in 26m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/028d1aca3fe94e5ba971c4fa4c16d346 : SUCCESS in 46m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/73fe40fac80b4104a073ece3439af54f : SUCCESS in 46m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8fbbaa11456f49109813ebd2b3592f50 : FAILURE in 1h 16m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3a272a2517a7443b9e3b04bdace21d4f : SUCCESS in 1h 27m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/13a5783d2ae3417cac90d62705dc5709 : POST_FAILURE in 1h 31m 01s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/36f3c5b12b0d405baf463045db930a6d : SUCCESS in 42m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/09eb52dda4c24feba75f59653639acd5 : SUCCESS in 4m 50s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/52e5aa33fba042ee8ce2920931a68c48 : SUCCESS in 15m 57s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a66e49f9f25f4af0a2ff6b456a203f08 : SUCCESS in 22m 24s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6942f5f613194a5184cbbc7e03d1356b : SUCCESS in 22m 10s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b405ad8b46484bf683cffbb41b65f129 : SUCCESS in 24m 42s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/437c7e09dbb943978ae182ee49ea5df2 : SUCCESS in 1h 45m 52s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/cac5191d396c4be9b93f419cfc0fecf6 : SUCCESS in 1h 39m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/917e1c8e70824fe0a7d9152c7fb1d6e6 : SUCCESS in 1h 09m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20451f25ea264ed48822c49da87061d8 : SUCCESS in 1h 01m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/140575ed512543289ba86e1efb5b51c9 : SUCCESS in 33m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7a3b50b52bc4477081d5e73987f7a353 : FAILURE in 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"892745d31f4bd2e82d6bcccc5773e26c18e4657e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-02-23 18:14:53.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8173ec5dd5fcf6ed2e043ca9ae1a0984dd968807","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 19:03:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6d63e67f45044c79d115cfb2eaa0f3a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32a484ab0b5b4382aaec4fde46f70d67 : FAILURE in 44m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b82ff6446ad64f18bd943d99fe093388 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bfcd46da5ea242f3a5629318229c8612 : FAILURE in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a3b993b2665a8112acc8cd24adf14cdfb156bf25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 20:01:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b19dcad9e4834c0ab0da0acc993d158a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/03fd7e5cb46644f5a9bfca4b4d72fed5 : SUCCESS in 1h 00m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/84bbdc21f296452883eefa07eaaa3aab : SUCCESS in 1h 35m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba60e8e950ad405daeb806fc83833b2f : SUCCESS in 1h 20m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20358ecdb31d46858bcb243c7186cd39 : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a4fe2789204c119332e949648abbcb : FAILURE in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed8591ce311b44d08286b5788bfaff46 : SUCCESS in 12m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f3ec0d2eda9460f9594932d9164aea2 : SUCCESS in 15m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc080108c0d404e9de5deaaf621f57a : SUCCESS in 10m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cc271e54637c4bffb96feed7b4377f4b : SUCCESS in 50m 13s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c689cf47d4a24648b053c239ad207d8e : SUCCESS in 1h 32m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/78746ec24af448839d33b13bc802d910 : SUCCESS in 42m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a9c511b066f6407db4dae3290e294b1e : SUCCESS in 42m 50s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7fa7f49183e448f4a4f2a3389f24dc8d : SUCCESS in 48m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/031c0e2044984c9b9b6414826d2b8a74 : SUCCESS in 1h 39m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/d63ed46bed0f46b7ae948a448bea98ce : SUCCESS in 1h 12m 01s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e7c8dc62ebb1490e8c5e1a1c3d4ba52e : SUCCESS in 55m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4737ff8139284b0a8c397563f3115a00 : SUCCESS in 5m 10s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e7559d52e7449d9b4ae008579414ba7 : SUCCESS in 19m 13s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dbafa13ed83d4c94bf3159e6ed9639fa : FAILURE in 26m 15s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a2d29dd80fd469aafb56e4616c84834 : FAILURE in 23m 26s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b35b36295aa641f59c188bd75dd137af : FAILURE in 26m 34s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/70414049765540c89647612dd2051b69 : SUCCESS in 1h 32m 31s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/74e06394f1d244859d0e6afd76af4beb : SUCCESS in 1h 06m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6333eb2f938d4529b76513df5991a38a : SUCCESS in 1h 02m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afdddef8fc7945bab1d0132188c8a529 : SUCCESS in 1h 02m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/918e1afac5c94731b99fcc95b65b83ef : SUCCESS in 33m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4ffc7cae9ef745e8afcc1546d5f38f6d : FAILURE in 1h 11m 11s (non-voting)\n\nWarning:\n  Comments left for invalid file nova/tests/functional/libvirt/test_power_manage.py","accounts_in_message":[],"_revision_number":5},{"id":"3ab48219f9feaead9fd46515994e7f024e00f1e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-08 23:05:57.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1f31a1e88981577a67538148e8048f46f945a4a5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-08 23:06:19.000000000","message":"Patch Set 6:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42798/","accounts_in_message":[],"_revision_number":6},{"id":"35bb0095d28326a8bbdd342d01d08419ed3de035","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-08 23:06:24.000000000","message":"Patch Set 6:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42798/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"b1403721ffcf75ddae4c5e38c45ff1a170200dad","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-08 23:10:46.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0cf70ac7273e9ea7df6c4cf2fd4a41dffd13c361","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 23:48:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/378d357627d8434a876ff1319cd22252\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1595c58f03ed43fe84ffe896904d2155 : FAILURE in 40m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/58803a67d20347a3beb25b0253b764cb : FAILURE in 11m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0da60ab89afc47b480d3130fc716b17e : FAILURE in 26m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"acad8e2cf8406065f792b07496119469c3c9a728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 01:08:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1787b43fd2634feb809c6d1ec27f3e59\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc071198ab244173ba705fd5f12c7253 : SUCCESS in 47m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2c6c6c1afcfb4413a094fb3ea9883766 : SUCCESS in 1h 37m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c54ae53bc4d4451f83bc6af1790ebc39 : SUCCESS in 1h 00m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cc978e7a7614b72bf2fb3f4097ccd84 : FAILURE in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04598edbcd984ae08397e0dda92a3590 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84ffbee329204b8cb5ae81e116a2c1f1 : FAILURE in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f08524542f2a4714a4e420cff5945934 : FAILURE in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d5c1c040994caa888cf4c05dbad47d : SUCCESS in 8m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66912a138ac34304800807de06928598 : SUCCESS in 52m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f9de29f2c2ca4140a33a755b29f59edf : SUCCESS in 1h 55m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a43c25aebe7f47679943f283e2111371 : SUCCESS in 53m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8eb79fee48854bfab54d18f41afc747c : SUCCESS in 38m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/cd19e01dd45244f595d803910d5a23a4 : SUCCESS in 55m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/99b768b87c744780a1548693d9444248 : SUCCESS in 1h 39m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/35224adff1354dd3b8ec8620d2b7cb8c : SUCCESS in 1h 20m 16s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/db7dc128401e45528dcd8705d8d98b36 : SUCCESS in 49m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b0777c74e8de41a788206fdc0386c95e : SUCCESS in 5m 19s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fc7d732db866438e974a0b0f501e9ec4 : FAILURE in 16m 27s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ee1e55efd104a25a65260cf2e91873c : FAILURE in 27m 20s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e188d88534fa4637836314121e2ad722 : FAILURE in 26m 30s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57f30bbac1654d85b193d1473ea4e2ea : FAILURE in 25m 31s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d914af2c1daa46eaab8fe4d2ebed28ad : SUCCESS in 1h 44m 58s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/36e4f5f18b5b48d9a68a0d680f6950f2 : SUCCESS in 1h 49m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/809ea5c3ab4f49c78a0b7c3c097326b3 : SUCCESS in 1h 22m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38bdce24063b4dff90729af6644eeeb2 : SUCCESS in 36m 32s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/03be4edb8076426c8e02480229a142f5 : SUCCESS in 26m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a28a4215ebd4699847b984e8e3822e6 : SUCCESS in 40m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8f84a1523673c4f10421fee22e7814c176af7bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-09 01:39:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9d330cde61c6ed899e05e9840d7b829f499e15de","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-09 01:40:25.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42805/","accounts_in_message":[],"_revision_number":7},{"id":"50082a6ca0e29a0ac2f47b6445125304b5b860da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-09 01:40:28.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42805/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"52389129f55f4a453e8a226b94fd40702e52f892","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 02:09:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/336742c0c7064149a9aa44e59f23e15e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd2e6262ae44485fa43ef822c485c312 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6b0e25e0aa14589bd51ff6d8be8fe5c : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/689f093d6170467a85122af7ad2fea94 : SUCCESS in 26m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ef61ffe8a6c8cb5da8cdd2d3db9e2a332873a253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 03:59:06.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a63804abd97b42bd82c19c89ca6cff63\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a06d8bc7f56d4be8a61f078e71f0aa18 : SUCCESS in 1h 05m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bff1e40f2ef94e90ae405b3873fd9ae3 : FAILURE in 1h 17m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07393095d5de41a984bdddde9e3356c8 : SUCCESS in 1h 05m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3856d60605bd4198a1639177e47828fe : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f35dffa6c98a4a7d95809eefed586e05 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9929bbbf6e35494987285a99bd7228a7 : SUCCESS in 12m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/142034d092f043dc8e6cc1b5ef85f6c5 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0bf80918a0d40eba7fbb31ff7b14191 : SUCCESS in 12m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c2baf7496dc244d998f25d8d30deab54 : SUCCESS in 1h 00m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/32615514f7514e94839e0242211af455 : SUCCESS in 2h 09m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8323ede76d204c678ea63b307d4ac757 : SUCCESS in 52m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d24a9f8e8ce8460d8f1950f7c10f87e4 : SUCCESS in 59m 53s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5ebb3bb6d46045549b35599566657694 : SUCCESS in 58m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/79a181b3e00a4c4586c49d116e85a5e3 : SUCCESS in 1h 58m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/70dd213a040d455097d1615fb8cdf6cf : SUCCESS in 1h 41m 34s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a65096de0d1446159fefd10bcab70e54 : SUCCESS in 43m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2f633b833a5e40978b7dba49ecfb1797 : SUCCESS in 4m 37s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/199cdcee0aeb4d028075968a07d5d779 : SUCCESS in 20m 53s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8af1c00f0bfc48ab9dad2623521df3c7 : SUCCESS in 27m 36s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3c7b90f10fb4cd48e05cefdbfcf4a5d : SUCCESS in 25m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b71b902945a401eb2ebf9c51116e464 : SUCCESS in 25m 38s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dfc07f56ffdc487ca54e3501af9aa52e : SUCCESS in 1h 43m 49s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/b1e1eb3b179545d9b29898e28ebea20e : SUCCESS in 1h 35m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cfeded1c1a547ad9f2bd4ec2f3fa394 : SUCCESS in 1h 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a20b889e68d47119210f17842419380 : SUCCESS in 49m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d2b14ec578084c9780ef14b7f100b23b : SUCCESS in 23m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6c7a5a29442141288ecd5c6b94662632 : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f2ba2a79b05f41ade7cf886c01bdc0891c909178","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-03-11 12:55:48.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"52e01e88940a8c48cdd989e8ac06ed4106cfe0bd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-03-11 12:55:53.000000000","message":"Patch Set 7: Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"282546d614b1e1389c8a2fbca9aa6927c6612932","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-11 14:13:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"73088be9edcc2bd984fbbab3a612ce9f30679299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-11 18:24:35.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d3d1f47ea03b5c8b7e6fd570a3afac54bb9fbf3a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-11 18:24:49.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42812/","accounts_in_message":[],"_revision_number":8},{"id":"f0e4643b4be5fda822d9d3892f6644cbc838ff00","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-03-11 18:24:50.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42812/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4ff06b15dc77e4f8f0955cae395933b63a36122f","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-11 18:59:37.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"026056bd88aebe0dbe5ad82b59279bab09f77961","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 18:59:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b41bc504354bfd8e5e01bb92c83456\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3d4ba8bdbbe47878330cbef295e8ef0 : SUCCESS in 32m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38953d4ecd5f4048948dc1c1e274a038 : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1709455ce1734eb3995a91f07b18eef7 : SUCCESS in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"90b4ff23b2db2a3e627d7d27a255c1ab3fd0f4dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 20:20:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4425c62375454f42b58db6f40882768a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0bd103a89f04fcd9169916c2c434dd4 : SUCCESS in 1h 04m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/eb9e09a8025f4b12af5cb24f4e32b2f5 : SUCCESS in 1h 38m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c4a36bfa3d34222ad93ad0a46285753 : SUCCESS in 1h 03m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c64a0a7c5fce48ef95c067f3930463ef : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4a44660302434aad7dae1c26f1b59b : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92f47361b1ad43f690979985cfaccff3 : SUCCESS in 13m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c773ebbb63a4094a94fc141e996b3bf : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874e076729184033a7a7b8ddf8408089 : SUCCESS in 10m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cccdf2e16b4f4ce3b4fe5e4c363b6d93 : SUCCESS in 53m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9607e1f352cb4028b9266019c432532b : SUCCESS in 1h 38m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4a05226656d444ddb115c2505b0a6147 : SUCCESS in 1h 08m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b6f1d11e706e460982af458463569f8e : SUCCESS in 55m 58s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/62ae446f26f14965b3649ec15ece4a05 : FAILURE in 1h 08m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/412b521e70bb42a78af2f35b37d02b59 : SUCCESS in 1h 32m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/c8a913db1ab94ca2ba5b77362456a920 : SUCCESS in 1h 29m 59s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9fbac5c650b642cbb2d75230f63c82ae : SUCCESS in 48m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/582a92ba4134473d973059aa2d07b59e : SUCCESS in 4m 59s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/badea925bb9b47a3b638b1572ae4a4bd : SUCCESS in 17m 22s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/339f7354b0c045e2ae5adc968b9f7465 : SUCCESS in 22m 16s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5df3fbce13d4ebba37ecd20a141b30a : SUCCESS in 24m 40s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88967a24b4f14d5798162eb5f620c445 : SUCCESS in 24m 56s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d6b61b13c9b84a59abbf0aa43f7eb4be : SUCCESS in 1h 19m 46s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/e274a8ee77ee40a3a607d6fe714f3a3b : SUCCESS in 1h 49m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d978d9c0cf2a458f94df980909dd2b11 : SUCCESS in 1h 22m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa42a5446874431db00a152dfd0f90cb : SUCCESS in 1h 07m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3b794f9a9bd549d498a5fedc8b895db5 : SUCCESS in 21m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/512e2fc7dfe847998429dd9817d19c52 : SUCCESS in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ec5070c8440b000f11f3e248aa43eed025c367e3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-03-13 15:07:03.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"b733d1ae2e102a888d884277dc8416b21839debf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-03-13 15:13:53.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0c2b7db7126dccf24e71d78e4cbc2534d89c3d88","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-03-13 16:36:00.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f234193a5ec1e835f5ed2200e2eb67c50b76190e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 18:47:59.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d22666f6fc2644bc99fe57c28ac9bde6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11872d8660b7471b9b86ef33f4afc9cc : SUCCESS in 50m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c457fc80aa5a4e3cac71d21405cd40f6 : SUCCESS in 1h 37m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aabe7a01a7a1402eb6715a7b6e977056 : SUCCESS in 1h 07m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4370fcd930eb43eea76ac526cc3c97e8 : SUCCESS in 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b459df2fa44178a909a552ddf521a7 : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e426baafb0d4bf59ffad65f465569b1 : SUCCESS in 14m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0050ba7b414f4efe80fc8303c26451fa : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342fdd5c50584abeab7e841865547404 : SUCCESS in 12m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4f9d68f5b8cb41a587c1f077eefa55e3 : SUCCESS in 1h 02m 24s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd32b5f3b50844a2b5f979ed42b47b45 : SUCCESS in 1h 51m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5296109f50054245a390def7bbac6344 : SUCCESS in 51m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4f0986d9f7f7496b9f3bc0d4bd2870b9 : SUCCESS in 51m 56s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2ce99edb13024549bee47f31dc4d94ae : SUCCESS in 1h 01m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7458ca8c23f448deb71896921a9df763 : SUCCESS in 1h 33m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/9adfa66e784a4926a8b14a17bc029f19 : SUCCESS in 1h 35m 39s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ce5b3c494ed74970b8188da642098452 : SUCCESS in 46m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7c777badc3284e45b81fe8e537c7b735 : SUCCESS in 6m 18s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cf1d2bd68f604874bcd26dda8ff0030b : SUCCESS in 16m 15s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a745408d805477eb889331e57e9cba4 : SUCCESS in 28m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bd6389bc9544650bdd7e1dc33f4869f : SUCCESS in 36m 47s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e59285b9b1143e08f1e568541c15cb1 : SUCCESS in 27m 01s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/31542db39b0d44acbe43e32d49d381d2 : SUCCESS in 1h 47m 14s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/6d52756ac1794cb6b9ffa6a4d08b7a6e : SUCCESS in 1h 22m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/170531a0a8a6446bb1ab7f2530c93aaa : FAILURE in 1h 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/237e8c64b9f04bd6821a705123fa2543 : SUCCESS in 1h 01m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c311508249854643be20a63dfe72be70 : SUCCESS in 31m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0e8f77ab876b423daad651079dd536e8 : SUCCESS in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"13afeca971275e4be07ddd5e05061a356885dfed","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-03-13 19:41:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0e952c375362e1a35e5eb0363efca652fb4ff5e4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-03-13 19:59:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9b0fa414c271011aa4dea683dc87e07c5271cac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 20:23:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/507d8f1861284c019282e012327db522\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83fcc279e78546e9ab7f567bef929fda : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0a26413d4f044a79a94c93156ee951d : FAILURE in 41m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76fedcd8e9b545ac9e49a9024a2d2fc2 : SUCCESS in 30m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4ae2d735fced1ae5a84cb6d734fc32f7f90c0006","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-03-13 21:01:51.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-909806-mwgmv/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"4efb52d0c8013150ff8fe2cab4f12d0e08da97c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 21:35:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee030557bb34939a1add83014590cc7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b447a7b7f0540589e851ecd23807e6a : SUCCESS in 1h 02m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d0932535a2ca4193bea99f496f3c7f2f : SUCCESS in 1h 44m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe382bc344434e0b990d26d180b69136 : SUCCESS in 1h 09m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3eb991a319714164aad20b9ab6513172 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b34fdd69b84b87a54f3d5a18c7a63e : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ac3a50d2824f12a218d6fa7fb11214 : SUCCESS in 15m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69e1ab1d936a4d7cbbcf76190e11b5b5 : SUCCESS in 13m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160ab7e342bd41e49911d2430680001f : SUCCESS in 11m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/334500ef619d4ffb9ab0c5d72bc70dcc : SUCCESS in 1h 05m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd72a4a4849c4dfa88a33555f6db0ae1 : SUCCESS in 1h 41m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4698bc3a7f7340f096501af1bfa88eb7 : SUCCESS in 57m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/593ccca3d7414771b2b7ff3c40d6aba5 : SUCCESS in 50m 19s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3bfacb4212be41eda420371589af4c4d : SUCCESS in 1h 15m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c177036f48254532afca7846aadd5425 : SUCCESS in 1h 38m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/85e74762bf7a43eaab4d5bb5002f53d3 : SUCCESS in 1h 18m 38s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c443dffb63864d3cb8538087f2099ade : SUCCESS in 46m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/aadaa288f45a4ab09d72929106ff44b4 : SUCCESS in 4m 32s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/61f4d0b33d66465daf8556f3573e15ac : SUCCESS in 16m 26s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9392023fadd445b8f26e208485e5d34 : SUCCESS in 26m 03s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2fcf1ddedd3434c8efbbc0575a74dc7 : SUCCESS in 24m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dad1a83790e14a67bb1128559467c49a : SUCCESS in 20m 56s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6ad876032e774884bf484b61155ba5c9 : SUCCESS in 1h 13m 21s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/a6cae91c4e3f4ab09a88416ef41a913b : SUCCESS in 1h 34m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a638b214d51b4d7fa406d13b662f705a : SUCCESS in 1h 00m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50775239a8a545328001a719216dcd23 : SUCCESS in 54m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/137fc02b45844163995dcca9a5630e22 : SUCCESS in 33m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5316ff9dd4684d2eb48a6881209ebcbd : SUCCESS in 41m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"849f117eb58a979911a9464fa58237f303d8942c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 21:48:27.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fd62453ea3993e4cce1c1eb465019078403852bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 00:06:49.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d16597c0fa4e698ffe1962e8b097c6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a166318e827445cbb3dfeba27bfc4ce1 : SUCCESS in 54m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b35f32b0bc9b45219a8968228c8820f5 : SUCCESS in 1h 28m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de92d5adc14d4744aeb572862c048b82 : SUCCESS in 58m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7897efe89b774441be301c635ea972c0 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f131acb7cb9748e2a66328d615f81723 : SUCCESS in 12m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c29d55a7d4d448ec95a57a22c09c167a : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832777af8f81452c89ed75c379f3010e : SUCCESS in 9m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dd66f1cc0ff547a0a34578362f4d753e : SUCCESS in 43m 33s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba572c10f6ab4fa4b792728d5c8e4813 : SUCCESS in 25m 48s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3eccc4fcf79e4264a35df0f6ea281f54 : SUCCESS in 24m 55s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34ddb29d8aca4c199085cc35348bfbe5 : SUCCESS in 23m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3c6244692e734bb78d39fe9435cb9597 : SUCCESS in 2h 09m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f4661f8a8fc43978e7d2719470c0256 : SUCCESS in 1h 38m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/14c7cbcececa4164bf3e0761d784b616 : SUCCESS in 4m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c16e83abafa401a818d9e7d3203a12f : SUCCESS in 2h 03m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9c49fa0e262d4c9cb9a7157ce6906f2d : SUCCESS in 1h 35m 37s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/dc66c4a47fcf4c22a619e6e1539052c3 : SUCCESS in 1h 37m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8105c50b75c144a4b132e28ac86360b4 : SUCCESS in 1h 23m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a412c2ab2da4d249c270c3fedc4b786 : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":8},{"id":"384fe1e3601a5d45d253cebe1ed7fc6f8b9064b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 00:06:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"66345aa3adb607b56f826e045bc2413d350f5e64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 00:07:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ff1e1e50104c4b82981cbf7fa35ff6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2603299e21a749d3bf9a7015962ef92f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c1ccc1a3165ec1556c605b3b036274e992b0a09d","revisions":{"5365923984365f7621b078f702af0b3d9343f510":{"kind":"REWORK","_number":1,"created":"2024-02-22 01:02:02.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/1"}}},"commit":{"parents":[{"commit":"04b59b73af87e5d01964e586542b13bd1c3648da","subject":"POC: online dedicated emulator threads as well","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04b59b73af87e5d01964e586542b13bd1c3648da"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"subject":"POC: power up dedicated cores during pre_live_migration","message":"POC: power up dedicated cores during pre_live_migration\n\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5365923984365f7621b078f702af0b3d9343f510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5365923984365f7621b078f702af0b3d9343f510"}]},"branch":"refs/heads/master"},"d4449e6a1a70563761e4291b0d506d7757f75224":{"kind":"REWORK","_number":2,"created":"2024-02-22 01:22:33.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/2"}}},"commit":{"parents":[{"commit":"04b59b73af87e5d01964e586542b13bd1c3648da","subject":"POC: online dedicated emulator threads as well","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04b59b73af87e5d01964e586542b13bd1c3648da"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:22:24.000000000","tz":-300},"subject":"POC: power up dedicated cores during pre_live_migration","message":"POC: power up dedicated cores during pre_live_migration\n\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4449e6a1a70563761e4291b0d506d7757f75224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d4449e6a1a70563761e4291b0d506d7757f75224"}]},"branch":"refs/heads/master"},"a5dcd89592a7a6dbc29075c410692fc04342dfbe":{"kind":"REWORK","_number":3,"created":"2024-02-22 01:24:15.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/3"}}},"commit":{"parents":[{"commit":"04b59b73af87e5d01964e586542b13bd1c3648da","subject":"POC: online dedicated emulator threads as well","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04b59b73af87e5d01964e586542b13bd1c3648da"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 01:24:03.000000000","tz":-300},"subject":"POC: power up dedicated cores during pre_live_migration","message":"POC: power up dedicated cores during pre_live_migration\n\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a5dcd89592a7a6dbc29075c410692fc04342dfbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a5dcd89592a7a6dbc29075c410692fc04342dfbe"}]},"branch":"refs/heads/master"},"d970109327064cbf9154b2c3aac64618dfde0e5a":{"kind":"REWORK","_number":4,"created":"2024-02-22 15:47:24.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/4"}}},"commit":{"parents":[{"commit":"04b59b73af87e5d01964e586542b13bd1c3648da","subject":"POC: online dedicated emulator threads as well","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04b59b73af87e5d01964e586542b13bd1c3648da"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 15:47:18.000000000","tz":-300},"subject":"POC: power up dedicated cores during pre_live_migration","message":"POC: power up dedicated cores during pre_live_migration\n\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d970109327064cbf9154b2c3aac64618dfde0e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d970109327064cbf9154b2c3aac64618dfde0e5a"}]},"branch":"refs/heads/master"},"926d9e127903b85d89fa8c4bdcf8714f02e12a58":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-23 18:14:53.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/5"}}},"commit":{"parents":[{"commit":"c23cdd77e12e0ba55dacc333c502fa9e7d5ce380","subject":"WIP: Reproducer tests for live migration with power management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c23cdd77e12e0ba55dacc333c502fa9e7d5ce380"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-23 18:14:37.000000000","tz":-300},"subject":"POC: power up dedicated cores during pre_live_migration","message":"POC: power up dedicated cores during pre_live_migration\n\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/926d9e127903b85d89fa8c4bdcf8714f02e12a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/926d9e127903b85d89fa8c4bdcf8714f02e12a58"}]},"branch":"refs/heads/master"},"4a06f5a2440a4e9f57158272a0fb002e6f3bbf1f":{"kind":"REWORK","_number":6,"created":"2024-03-08 23:05:57.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/6"}}},"commit":{"parents":[{"commit":"e5bc9f54c8e95b519cbd0aa3e33256b5ce6f8aa5","subject":"Reproducer test for live migration with power management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e5bc9f54c8e95b519cbd0aa3e33256b5ce6f8aa5"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-03-08 22:54:22.000000000","tz":-300},"subject":"pwr mgmt: handle live migrations correctly","message":"pwr mgmt: handle live migrations correctly\n\nPreviously, live migrations completely ignored CPU power management.\nThis patch makes sure that we correctly:\n\n* Power up the cores on the destination during pre_live_migration, as\n  we need them powered up before the instance starts on the\n  destination.\n* If the live migration is successful, power down the vacated cores on\n  the source.\n* In case of a rollback, power down the cores previously powered up on\n  pre_live_migration.\n\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a06f5a2440a4e9f57158272a0fb002e6f3bbf1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a06f5a2440a4e9f57158272a0fb002e6f3bbf1f"}]},"branch":"refs/heads/master"},"01141881defdd73810ebace61226e9a544e8f246":{"kind":"REWORK","_number":7,"created":"2024-03-09 01:39:59.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/7"}}},"commit":{"parents":[{"commit":"95b5b0c0c791d6a449fc82b10c0bcc0e331c54b9","subject":"Reproducer test for live migration with power management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95b5b0c0c791d6a449fc82b10c0bcc0e331c54b9"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-03-09 01:39:47.000000000","tz":-300},"subject":"pwr mgmt: handle live migrations correctly","message":"pwr mgmt: handle live migrations correctly\n\nPreviously, live migrations completely ignored CPU power management.\nThis patch makes sure that we correctly:\n\n* Power up the cores on the destination during pre_live_migration, as\n  we need them powered up before the instance starts on the\n  destination.\n* If the live migration is successful, power down the vacated cores on\n  the source.\n* In case of a rollback, power down the cores previously powered up on\n  pre_live_migration.\n\nCloses-bug: 2056613\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01141881defdd73810ebace61226e9a544e8f246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01141881defdd73810ebace61226e9a544e8f246"}]},"branch":"refs/heads/master"},"c1ccc1a3165ec1556c605b3b036274e992b0a09d":{"kind":"REWORK","_number":8,"created":"2024-03-11 18:24:35.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/06/909806/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/909806/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/909806/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/909806/8"}}},"commit":{"parents":[{"commit":"1f5e3421ec1e15a58b9d9bdb9fc4312373ec4408","subject":"Reproducer test for live migration with power management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f5e3421ec1e15a58b9d9bdb9fc4312373ec4408"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-22 00:58:32.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-03-11 18:21:27.000000000","tz":-240},"subject":"pwr mgmt: handle live migrations correctly","message":"pwr mgmt: handle live migrations correctly\n\nPreviously, live migrations completely ignored CPU power management.\nThis patch makes sure that we correctly:\n\n* Power up the cores on the destination during pre_live_migration, as\n  we need them powered up before the instance starts on the\n  destination.\n* If the live migration is successful, power down the vacated cores on\n  the source.\n* In case of a rollback, power down the cores previously powered up on\n  pre_live_migration.\n\nCloses-bug: 2056613\nChange-Id: I787bd7807950370cd865f29b95989d489d4826d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1ccc1a3165ec1556c605b3b036274e992b0a09d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1ccc1a3165ec1556c605b3b036274e992b0a09d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
