)]}'
{"id":"openstack%2Fnova~909880","triplet_id":"openstack%2Fnova~master~Ia9d230cec461b47e56bb07ac45c5417e1747dfaf","project":"openstack/nova","branch":"master","topic":"auto-pthread-heartbeat","attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2024-02-22 16:56:38.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia9d230cec461b47e56bb07ac45c5417e1747dfaf","subject":"Auto set heartbeat_in_pthread for wsgi services","status":"NEW","created":"2024-02-22 14:53:38.000000000","updated":"2024-02-29 02:24:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"520561cd8796c73b4fc22c691d595e1bfb200155","_number":909880,"virtual_id_number":909880,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-28 16:22:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2024-02-29 02:13:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 14:53:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 15:15:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-22 16:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-29 02:13:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"5f6c41ee29ba574fe9f6b7355128150c1e6d86d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-22 14:53:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17c813091c88867abcbc318497cddf2538e22987","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 14:53:47.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42674/","accounts_in_message":[],"_revision_number":1},{"id":"c19895be8f74c0e4ed2825fd1598319223878254","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-22 14:53:50.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42674/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"6758fc03cd6a2897f166fdfc436eff28f258dc08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 15:15:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407cc90d71aa4d038847ff2c0225e826\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6e352b1cba049439646b6f7f1fb49a3 : FAILURE in 18m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fdb006431394fcb82442e3356dea370 : FAILURE in 9m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bdff969a5f74e00bdf024a4e93dbdc5 : FAILURE in 19m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c5cfeabac353a35d3ad3eca7a2b733767b12155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 16:56:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d661045d40643e69a742d262af6ea51\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9194466548245ef83758a02990f4931 : SUCCESS in 55m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/29baf8cebbce4fec981e0f8bda2c042f : SUCCESS in 1h 37m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90c096ff3f6445959b39ae2e9116b2e6 : SUCCESS in 1h 03m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68e1c718d3784fafbba030a0edbc229b : FAILURE in 20m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5445b195340c47bcbf59c11977e9fa3d : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c82891aae1094d02bf066a9e283a6334 : FAILURE in 11m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d55435ec0b5c4021970c4ef144db62ce : FAILURE in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b979cdbd50ec4deab723c229534337bb : SUCCESS in 10m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c3fab734dd664fdfa2b931999b000b7c : SUCCESS in 1h 12m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/991032fdc02a4e15a93feaa24b81c902 : FAILURE in 1h 53m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8168fdc63e584b6fb802dff03a5d3efa : SUCCESS in 1h 02m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4adec31aae2941b7a5491da16e53ca80 : SUCCESS in 59m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8cb94a1c692b436294e83eacc399cbcf : SUCCESS in 1h 39m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/8c521c575a394f77bf9b8534c1680fdd : POST_FAILURE in 1h 04m 48s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f00684e2eca44be393953f00ca931d20 : SUCCESS in 45m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/76a1f007e5ad4e898f5619a14d811b07 : SUCCESS in 5m 32s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/10aa476af82e47e9827efce5e2e8d762 : FAILURE in 17m 05s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7b2958a5f5904a30ac9f223ff403315f : SUCCESS in 27m 32s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bd307442c1d4d3789d09036f44dcaba : SUCCESS in 26m 33s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78afb10577c64b6ea04ffaca91563c64 : SUCCESS in 23m 32s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/999e1115413e44a0ad2522d422e7d9bf : SUCCESS in 1h 47m 48s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/aa0eb547b9bf4bc39cf3b7052bf4982e : SUCCESS in 1h 37m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c532341cd93a49768c6ec731a01f8323 : SUCCESS in 1h 13m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e61e1c1443042f2a4c393708763e6a0 : SUCCESS in 47m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f36bc067089044b3aa6cc387e64472f2 : SUCCESS in 27m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7ce4dcca37ed43ec850db49c3d31e264 : FAILURE in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f53496e0f59f0f92b488d71ba5f8eebdb8692fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-28 12:35:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5105f5bdf06fe826af86c72164fd09ef0585e836","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 13:05:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed915421322b48e7a065eae9a94cf0bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e01d4827c494dbaa5fcf24f8f169a5e : FAILURE in 28m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8c4249ec1ca4dc0b02fba67a98a56f4 : FAILURE in 24m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/003fc8e51fcf4b798aac351111b7defd : FAILURE in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30debf06b13c0842650de2066fcf8a326db0cba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2024-02-28 13:39:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f8e2ed4f421aada33e8404d62f86001a0b4aaacc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 13:40:02.000000000","message":"Patch Set 3:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42743/","accounts_in_message":[],"_revision_number":3},{"id":"6de9d48c05e124ad9df0e4883d32e1a6617b110d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2024-02-28 13:40:04.000000000","message":"Patch Set 3:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/42743/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"b8f2832e0bcfe8833a806890f9f0612325c74690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:15:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ae8a6e1ec54ebabe0dbbb22281879c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/688541458f87489ca5bb2118d2464524 : SUCCESS in 20m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fdfd49aaa9af42119320b2ec863ef1f8 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15ba507edefe49e4a53319fc6204f576 : SUCCESS in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d4389d0601203bc36d446442fd1eec5aff145a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:22:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69d0dba4e9ff4ad197a4de3492a8de4a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b841374778a4b0c9e43f4e7d9b3f002 : SUCCESS in 1h 13m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/88b6603667ec4e27877aab237ac0ba74 : SUCCESS in 1h 59m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5600a4340af4716a1ddc8e96985a768 : SUCCESS in 52m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40b3b17ebf8547c3ba6d8e2db0a8cee4 : SUCCESS in 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfdf585393c943ab87c9638cfbd28c1f : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/359e5a44966f446b9137c7e56ba49840 : SUCCESS in 16m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e114416191124274abf906c5748c5835 : SUCCESS in 13m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5898fb82014cc0acd54323ce43e8a4 : SUCCESS in 10m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/edb00a01b77449418197ca73a01777c9 : SUCCESS in 58m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15c15e53c0824b2ca407aeee0e8a8277 : SUCCESS in 1h 57m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3879540374cb4fb49922482ff6ab85fa : SUCCESS in 49m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f6ddf031faa347c5a4c9b50411d33122 : SUCCESS in 51m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/31b35dfb88e443e2bd5f7a1c506dc1ef : SUCCESS in 2h 15m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/a454c88abc974023a7405d0e228deee5 : SUCCESS in 1h 11m 05s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/fe03f696bac542159b66dd6bd9318938 : SUCCESS in 45m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/36a7459035bc496685145fd65b42c46a : SUCCESS in 5m 05s (non-voting)\n- nova-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cd74348ad296415386b4cc5646b65d9a : SUCCESS in 15m 23s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d44c04dafb5543328f1c9ed08ad294d6 : SUCCESS in 22m 36s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c59be7d8ee04b6f8631f59f0a9650f1 : SUCCESS in 24m 07s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89c348e337f14fa6841bca8475b0c4ff : SUCCESS in 29m 13s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/334dc4b208b14bb782129fd0aa707d12 : SUCCESS in 1h 59m 18s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/fa53ded64b9a45ba8ebec4b5b7dabb69 : SUCCESS in 1h 46m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/22468730ccad44798ad16522d773fa3d : SUCCESS in 1h 10m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cea389e60fd4e24ae4a7aa1d12c0dda : SUCCESS in 1h 05m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/14ba46f722994c92bb23c0fea6660acd : SUCCESS in 27m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8f893503522840d3a0f28a6d5b74edaf : FAILURE in 51m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a253f814170c2b0be4e116e0c2e75cd7fda938d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-29 02:13:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"520561cd8796c73b4fc22c691d595e1bfb200155","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-02-29 02:24:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4cbb27854bba8655f1d8b32aeb9f41a9f1970b82","revisions":{"fe47812911eed11e984c6a4935b140999fc9c29b":{"kind":"REWORK","_number":1,"created":"2024-02-22 14:53:38.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/80/909880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/909880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/909880/1"}}},"commit":{"parents":[{"commit":"3209f6551652cff7bef0b9d9719ab940dd05a0f8","subject":"Merge \"HyperV: Remove RDP console API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3209f6551652cff7bef0b9d9719ab940dd05a0f8"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-19 20:49:46.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-22 14:51:39.000000000","tz":0},"subject":"Auto set heartbeat_in_pthread for wsgi services","message":"Auto set heartbeat_in_pthread for wsgi services\n\nRelated-Bug: #1825584\nRelated-Bug: #1934937\nRelated-Bug: #1961402\nChange-Id: Ia9d230cec461b47e56bb07ac45c5417e1747dfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe47812911eed11e984c6a4935b140999fc9c29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe47812911eed11e984c6a4935b140999fc9c29b"}]},"branch":"refs/heads/master"},"43b8c51458583dc43b3696baef2f1f9982169885":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-28 12:35:02.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/80/909880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/909880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/909880/2"}}},"commit":{"parents":[{"commit":"7fa1859576ac577da89cb817f93b1c54d3a937ce","subject":"Merge \"libvirt: Support maxphysaddr.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7fa1859576ac577da89cb817f93b1c54d3a937ce"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-19 20:49:46.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2024-02-28 12:35:02.000000000","tz":0},"subject":"Auto set heartbeat_in_pthread for wsgi services","message":"Auto set heartbeat_in_pthread for wsgi services\n\nRelated-Bug: #1825584\nRelated-Bug: #1934937\nRelated-Bug: #1961402\nChange-Id: Ia9d230cec461b47e56bb07ac45c5417e1747dfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43b8c51458583dc43b3696baef2f1f9982169885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43b8c51458583dc43b3696baef2f1f9982169885"}]},"branch":"refs/heads/master","description":"Rebase"},"4cbb27854bba8655f1d8b32aeb9f41a9f1970b82":{"kind":"REWORK","_number":3,"created":"2024-02-28 13:39:51.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/80/909880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/909880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/909880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/909880/3"}}},"commit":{"parents":[{"commit":"7fa1859576ac577da89cb817f93b1c54d3a937ce","subject":"Merge \"libvirt: Support maxphysaddr.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7fa1859576ac577da89cb817f93b1c54d3a937ce"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-19 20:49:46.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2024-02-28 13:39:26.000000000","tz":0},"subject":"Auto set heartbeat_in_pthread for wsgi services","message":"Auto set heartbeat_in_pthread for wsgi services\n\nRelated-Bug: #1825584\nRelated-Bug: #1934937\nRelated-Bug: #1961402\nChange-Id: Ia9d230cec461b47e56bb07ac45c5417e1747dfaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4cbb27854bba8655f1d8b32aeb9f41a9f1970b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4cbb27854bba8655f1d8b32aeb9f41a9f1970b82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
