)]}'
{"id":"openstack%2Fnova~913223","triplet_id":"openstack%2Fnova~stable%2F2023.1~Icd586cdd015143b2e113fd14904f40410809d247","project":"openstack/nova","branch":"stable/2023.1","topic":"bug/2056612","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2024-04-05 22:09:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd586cdd015143b2e113fd14904f40410809d247","subject":"Reproducer for not powering on isolated emulator threads cores","status":"MERGED","created":"2024-03-14 15:08:10.000000000","updated":"2024-04-05 22:11:53.000000000","submitted":"2024-04-05 22:09:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913223-bug/2056612","meta_rev_id":"0197199a92663059e331d073a70777f90f9dc2e6","_number":913223,"virtual_id_number":913223,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-05 22:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-04-04 09:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-04-03 08:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-04-04 09:25:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 15:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-14 15:40:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 21:30:25.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-03 08:06:01.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-04-04 09:25:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"55dd7280b5eba8b7ca01633046d3083c73b3e91f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2024-03-14 15:08:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ae2d25fd574b3f0f87573f980a8f0b3380bf3c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 15:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/390099015313417eb6ce7fd3b3e0d5f0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/374ddaf3afbf4289b8684750f770698c : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d0683848ac549ef80cebaa7f3eb1b78 : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a924d1dc950e12e0556637c1e777178ee1c38d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 15:40:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60790f9bc66c40dcba10b5a7abe30677\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/021e9a433dc642b1966c19fd0f05f4e2 : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9bc989115624ce49d7ff96e8146a96e : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8a87c08cf8140d38e729c2f7872a76d : SUCCESS in 16m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd4433fac7e4cb391105ef346dad54d : SUCCESS in 13m 08s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4500167f57c94236bb674c89c09a4547 : FAILURE in 4m 56s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3fae2b831e8044d783864ee2126bb6b8 : SUCCESS in 23m 55s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/006fe2d4d2d546659bf584297c9d52ac : SUCCESS in 29m 26s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b062285b3974fcd852e7b0035a01a6b : SUCCESS in 24m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5450d5b56707b0bdc8884d9b0ac1872293fd58ca","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-03-14 21:30:25.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-913223-hhz2z/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"0cfde452ab6b5e82d3544b224d3bba6be7743e43","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-18 16:01:14.000000000","message":"Topic set to bug/2056612","accounts_in_message":[],"_revision_number":1},{"id":"c44b5ab5bbc211d0d84a6bb12e30c430812d1713","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-04-03 08:06:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0d397f3978b123aa15921403910e970f02017bb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-04 09:25:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37009392673c6234c54f8621e6aa627332bf1bba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 09:31:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2260402a695e50d044503c3e5579bcc90d423fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 11:43:47.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/913222, https://review.opendev.org/913221 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"07729f9a329871b7761a177ac793126cd2edb96f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-04-05 17:56:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e7905ebcfba80fca7d614ca9d50683c67a66712","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 18:47:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1411a73d9cde42d38f4ca79948361af5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd86f04da9f64313aaf76ad28fbb8ebc : SUCCESS in 28m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"968294806830f3bd4784261b0c786d8ee1e3d493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 19:27:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fe9a9e0c82e4a5da0aec325f37cd0d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c47129809885451387a69d1977cb65fa : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7dbdaf1a7e454aa1ae7216c1bbe989 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9048337c8a804c26a8a60730622ccb6e : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf2e8979bd4400597b62d71489a0a9a : SUCCESS in 11m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/14e9788dac0a4ff099b8ebfecf49f3ab : SUCCESS in 4m 46s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d0924df44f74ea1a69a9213953f6673 : SUCCESS in 25m 47s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a3127dcc53944d5ba500a484b30fdb4 : SUCCESS in 26m 20s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50944eb9bd254def8dd1a8f897908c5d : SUCCESS in 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a38978bb785319ad4945c3f61d3ffa73cb7fc58e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 20:25:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5b9cdc41f69da962d7fd22f279bb7e919b75862","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 22:09:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ce18f672e064fca8827607745621aa4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3539ec53c8d4200bcdc93af5c2de8ff : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/101f8d3e9ed342fa92dc74eae46b81a7 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a7055333f446e2b86437673ee7e110 : SUCCESS in 10m 31s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/985bcb05260648e59fd23d0ef6015a7f : SUCCESS in 28m 34s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0786842baab43c1b9d4988aa50aec77 : SUCCESS in 25m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/47980a6115c2498cb4032dd8b7b9ceb0 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"722bb96c83bca9dcaa6a5a7b925689664f3bc04e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 22:09:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0197199a92663059e331d073a70777f90f9dc2e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 22:11:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85e5573705f7476d909f2539f102c80e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3032488d1a26498f961269974e7c88e2 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ea5c691f3c3cc6aa3588b442f084e4737ae3079","revisions":{"6ea5c691f3c3cc6aa3588b442f084e4737ae3079":{"kind":"REWORK","_number":1,"created":"2024-03-14 15:08:10.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/23/913223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/913223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/913223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/913223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/913223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/913223/1"}}},"commit":{"parents":[{"commit":"b60ff8ca5ada95b562279da36849733d0d2d964f","subject":"Add cpuset_reserved helper to instance NUMA topology","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b60ff8ca5ada95b562279da36849733d0d2d964f"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-02-23 15:19:44.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2024-03-14 15:04:30.000000000","tz":-240},"subject":"Reproducer for not powering on isolated emulator threads cores","message":"Reproducer for not powering on isolated emulator threads cores\n\nRelated-bug: 2056612\nChange-Id: Icd586cdd015143b2e113fd14904f40410809d247\n(cherry picked from commit 521af2620958fb1a466bfb1dbf2da566634eb992)\n(cherry picked from commit 3dde97216cfc8f30aaa9f3e2f5a67b1e252c85a0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ea5c691f3c3cc6aa3588b442f084e4737ae3079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ea5c691f3c3cc6aa3588b442f084e4737ae3079"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
