)]}'
{"id":"openstack%2Fnova~91700","triplet_id":"openstack%2Fnova~master~Idd888cc45533140d500fbe0bca1fed36d6fb0274","project":"openstack/nova","branch":"master","topic":"noquotaclass","hashtags":[],"change_id":"Idd888cc45533140d500fbe0bca1fed36d6fb0274","subject":"Remove quota_class db API calls","status":"MERGED","created":"2014-05-01 23:18:15.000000000","updated":"2014-05-23 03:44:52.000000000","submitted":"2014-05-17 22:53:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0da390bbfe45337bb4a19a730c39a0b2e408313a","_number":91700,"virtual_id_number":91700,"owner":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-05-17 22:53:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2014-05-16 20:35:43.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":1,"date":"2014-05-16 15:20:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"date":"2014-05-23 03:44:52.000000000","post_submit":true,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2014-05-16 16:05:04.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"date":"2014-05-16 16:18:35.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2014-05-16 14:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2014-05-17 14:13:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":1,"date":"2014-05-17 16:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},{"value":2,"date":"2014-05-16 14:55:26.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":1,"date":"2014-05-16 14:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},{"value":2,"date":"2014-05-17 20:14:16.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},{"value":1,"date":"2014-05-17 20:14:16.000000000","_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},{"value":0,"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-05-02 16:55:10.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2014-05-09 16:15:12.000000000","updated_by":{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},"reviewer":{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},"state":"REVIEWER"},{"updated":"2014-05-14 20:55:38.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2014-05-16 15:20:17.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-05-16 16:05:04.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-05-16 16:18:35.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-05-16 20:35:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-05-17 14:13:07.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2014-05-17 16:01:09.000000000","updated_by":{"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},"reviewer":{"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},"state":"REVIEWER"},{"updated":"2014-05-17 20:14:16.000000000","updated_by":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"reviewer":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"state":"REVIEWER"},{"updated":"2014-05-17 22:53:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-05-23 03:44:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdd15deb7a3f5341365e45cd8f50145e29d5e746","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-05-01 23:18:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f9019815b52c6d781ed7f6927e6e1e0dae68c6d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-01 23:18:19.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"103f9d747d7a8565d9ad96c90bc20037c2c75352","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-05-01 23:43:19.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/2002828/131007_devstack_export.log : SUCCESS in 9m 28s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_mysql_user_001/b6f8699/user_001.log : SUCCESS in 18m 40s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/0e576f8/131007_devstack_export.log : SUCCESS in 10m 47s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_percona_user_001/b010f1a/user_001.log : SUCCESS in 20m 20s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_mysql_user_002/cda47a3/user_002.log : SUCCESS in 13m 39s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_percona_user_002/0018219/user_002.log : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b2cff732c723d381fcc068c812db3fd7306ee102","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-05-01 23:56:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- gate-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/91700/1/gate-ibm-tempest-dsvm-full/321b46e : SUCCESS in 36m 05s\n- gate-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/91700/1/gate-ibm-tempest-dsvm-postgres-full/480d603 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c97ea273f5420c9f7607b5a55d436f87eabee4fb","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-02 00:25:23.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/91700/1/6574/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":1},{"id":"6dddd4e4595b80adc6ef029cffb492ec6f1ce36c","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-05-02 16:55:10.000000000","message":"Patch Set 1: Code-Review+1\n\nthx Joe.","accounts_in_message":[],"_revision_number":1},{"id":"cf68f27653194aed59edf54d81a842f7d76af722","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-05-02 21:10:18.000000000","message":"Patch Set 1:\n\nBuild failed. Logs: http://208.91.1.172/logs/91700/1 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":1},{"id":"4683809ea08821b7e078f1f859a59d4eac9e1515","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-05-07 18:26:43.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"f51ac318ace5e371932ec08a7c4ae4153fc6a148","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-07 18:26:48.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"59ac17edf7a80ce251897c205425779b5f9f3b5a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-05-07 18:47:30.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/3c65720/131007_devstack_export.log : SUCCESS in 10m 28s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_mysql_user_001/ffeb5da/user_001.log : SUCCESS in 18m 43s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/5ce5ae6/131007_devstack_export.log : SUCCESS in 11m 16s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_percona_user_001/0f4bf5a/user_001.log : SUCCESS in 19m 57s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_mysql_user_002/e0bd03f/user_002.log : SUCCESS in 14m 57s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/1/check/gate-real-db-upgrade_nova_percona_user_002/0750185/user_002.log : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b4d30da397ec6317faffad4d4493ea5aa4fa436a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-05-07 18:54:13.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/91700/1/ : Test run failed. in 27m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8045eab6b2cd7df162f381469b9e3d7e00957bf3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-07 19:30:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/00/91700/1/check/gate-nova-pep8/1516264 : SUCCESS in 9m 08s\n- gate-nova-docs http://docs-draft.openstack.org/00/91700/1/check/gate-nova-docs/3e4ab7e/doc/build/html/ : SUCCESS in 8m 02s\n- gate-nova-python26 http://logs.openstack.org/00/91700/1/check/gate-nova-python26/e1965f4 : SUCCESS in 15m 23s\n- gate-nova-python27 http://logs.openstack.org/00/91700/1/check/gate-nova-python27/0fe8886 : SUCCESS in 22m 00s\n- check-tempest-dsvm-full http://logs.openstack.org/00/91700/1/check/check-tempest-dsvm-full/ae58ce9 : SUCCESS in 43m 08s\n- check-tempest-dsvm-neutron http://logs.openstack.org/00/91700/1/check/check-tempest-dsvm-neutron/2f4a5f2 : SUCCESS in 1h 02m 38s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/00/91700/1/check/check-tempest-dsvm-neutron-heat-slow/627907c : FAILURE in 29m 58s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/00/91700/1/check/check-grenade-dsvm/1597caf : SUCCESS in 31m 41s\n- check-grenade-dsvm-neutron http://logs.openstack.org/00/91700/1/check/check-grenade-dsvm-neutron/38f7d90 : SUCCESS in 1h 03m 38s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/00/91700/1/check/check-devstack-dsvm-cells/3a5f69c : SUCCESS in 18m 16s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/00/91700/1/check/check-tempest-dsvm-postgres-full/9724093 : SUCCESS in 56m 03s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/91700/1/check/gate-tempest-dsvm-large-ops/efa82c7 : SUCCESS in 26m 51s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/91700/1/check/gate-tempest-dsvm-neutron-large-ops/d4374ac : SUCCESS in 21m 35s\n- gate-nova-pylint http://logs.openstack.org/00/91700/1/check/gate-nova-pylint/bc9ef05 : SUCCESS in 16m 11s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/00/91700/1/check/check-grenade-dsvm-partial-ncpu/66ffbae : SUCCESS in 34m 05s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/00/91700/1/check/check-tempest-dsvm-virtual-ironic-nv/01718f2 : FAILURE in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce06d9532180e0b181afd1dbac47410614b3fe05","author":{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},"date":"2014-05-07 19:41:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb06e21bab3747a3791070acdb6b06701a990f05","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-05-07 23:28:14.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nGood question, see inline.","accounts_in_message":[],"_revision_number":1},{"id":"732cac9eaa74d0ea9d32a6fa588668a22c6150d3","author":{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},"date":"2014-05-08 16:34:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3590252eff7431d684dde416f38b145731ad6f6c","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-05-08 21:04:34.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nsee inline.","accounts_in_message":[],"_revision_number":1},{"id":"1f0f9d0210d3520c4f7feae0915154c491b38f3c","author":{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},"date":"2014-05-08 22:24:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad3f862e1d5fa39d5609e17234da75a082fe2b7d","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-05-09 00:10:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18899c84ef9c694e9fabab8da3dbd8356410880f","author":{"_account_id":9214,"name":"Sergio Cazzolato","email":"sergiocazzolato@gmail.com","username":"sjcazzol"},"date":"2014-05-09 16:15:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b06816001472651ca42db70efd228778545a7d9c","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2014-05-09 22:03:34.000000000","message":"Patch Set 1:\n\nthis patch series should remove about 9 DB calls from a nova boot according to boris-42\u0027s rally project  (http://pavlovic.me/rally/profiler/)","accounts_in_message":[],"_revision_number":1},{"id":"8a4bc0ea2dc40a3a7e81deb1ab0bc422656edb6e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-05-14 20:48:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bfa9097f6121cb25084bfad512eb8ff12d4c957","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-14 20:48:49.000000000","message":"Patch Set 1: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"77645d310ab6df7ec89b9a8f8fcdedb6f772c169","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-05-14 20:49:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab581e8d88b293f60c699df4a219b981ffefe02d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-05-14 20:54:42.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"d26d6e4b9f839e7a5acb6a3c34e063c497102c15","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-14 20:54:45.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"d062f9b8bf1e6a54aa672de63f7ca49e6c5eb74b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-05-14 20:55:38.000000000","message":"Patch Set 2: Code-Review+2\n\nI pulled this series down and the only thing left is the data model and table which as pointed out will be removed in another patch with a migration, so LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"6463247cc40be9153b9117ffd84b5eaf42221e24","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-05-14 21:16:32.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/2/check/gate-real-db-upgrade_nova_mysql_devstack_131007/ea4c887/131007_devstack_export.log : SUCCESS in 9m 40s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/2/check/gate-real-db-upgrade_nova_mysql_user_001/f21edde/user_001.log : SUCCESS in 18m 38s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/2/check/gate-real-db-upgrade_nova_percona_devstack_131007/ee393da/131007_devstack_export.log : SUCCESS in 10m 39s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/2/check/gate-real-db-upgrade_nova_percona_user_001/4d45164/user_001.log : SUCCESS in 21m 01s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/2/check/gate-real-db-upgrade_nova_mysql_user_002/ca44ccd/user_002.log : SUCCESS in 14m 33s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/2/check/gate-real-db-upgrade_nova_percona_user_002/6a8de7b/user_002.log : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ef02ae4cc3eb9b50f1f415dab6b9261eb334236d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-05-14 21:18:42.000000000","message":"Patch Set 2:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/91700/2/ : Test run failed. in 23m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f3542cd9e6aa5a93620d891b8a6257662375d673","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-14 21:53:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/00/91700/2/check/gate-nova-pep8/cbcfb5a : SUCCESS in 8m 52s\n- gate-nova-docs http://docs-draft.openstack.org/00/91700/2/check/gate-nova-docs/ec0225f/doc/build/html/ : SUCCESS in 9m 06s\n- gate-nova-python26 http://logs.openstack.org/00/91700/2/check/gate-nova-python26/021ea38 : SUCCESS in 16m 10s\n- gate-nova-python27 http://logs.openstack.org/00/91700/2/check/gate-nova-python27/5099563 : SUCCESS in 19m 32s\n- check-tempest-dsvm-full http://logs.openstack.org/00/91700/2/check/check-tempest-dsvm-full/4dcea4b : SUCCESS in 47m 14s\n- check-tempest-dsvm-neutron http://logs.openstack.org/00/91700/2/check/check-tempest-dsvm-neutron/63d93a4 : SUCCESS in 58m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/00/91700/2/check/check-tempest-dsvm-neutron-heat-slow/d92b3a1 : FAILURE in 26m 18s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/00/91700/2/check/check-grenade-dsvm/625b604 : SUCCESS in 43m 59s\n- check-grenade-dsvm-neutron http://logs.openstack.org/00/91700/2/check/check-grenade-dsvm-neutron/db1a89c : FAILURE in 50m 22s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/00/91700/2/check/check-devstack-dsvm-cells/a07a232 : SUCCESS in 13m 56s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/00/91700/2/check/check-tempest-dsvm-postgres-full/c6354c6 : SUCCESS in 57m 23s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/91700/2/check/gate-tempest-dsvm-large-ops/769e391 : SUCCESS in 19m 59s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/91700/2/check/gate-tempest-dsvm-neutron-large-ops/85dcf57 : SUCCESS in 27m 46s\n- gate-nova-pylint http://logs.openstack.org/00/91700/2/check/gate-nova-pylint/f3190fc : SUCCESS in 16m 14s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/00/91700/2/check/check-grenade-dsvm-partial-ncpu/d60fe87 : SUCCESS in 41m 03s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/00/91700/2/check/check-tempest-dsvm-virtual-ironic-nv/c7cd8e4 : SUCCESS in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"335ca24a3407537be2b381a8a5a1424ffffdb9c6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-14 22:05:16.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/91700/2/7843/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":2},{"id":"fd7487c94d63de78790c1665bf39ae1e20dacee0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-05-15 00:24:55.000000000","message":"Patch Set 2:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- gate-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/91700/2/gate-ibm-tempest-dsvm-full/66e5db3 : FAILURE in 3h 20m 02s\n- gate-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/91700/2/gate-ibm-tempest-dsvm-postgres-full/59ded3d : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":2},{"id":"586885e1694c9ac11116a070bd84b653d5d5027e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-05-15 02:47:54.000000000","message":"Patch Set 2:\n\nBuild failed. Logs: http://208.91.1.172/logs/91700/2 For information on how to proceed, see https://wiki.openstack.org/wiki/NovaVMware/Minesweeper#What_to_do_when_a_build_fails","accounts_in_message":[],"_revision_number":2},{"id":"f9833d73184ce48de6538f3ac068804d69a5ff96","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2014-05-16 14:55:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e352aad6b1486a97721d0b549022edd8d9b0fca9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-16 14:55:29.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"9298d8b01b473f4a13a17979e872c7cb1bf7d7f1","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-05-16 15:20:17.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/3/check/gate-real-db-upgrade_nova_mysql_devstack_131007/25ee333/131007_devstack_export.log : SUCCESS in 9m 49s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/3/check/gate-real-db-upgrade_nova_mysql_user_001/fe4c306/user_001.log : SUCCESS in 23m 31s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/91/91700/3/check/gate-real-db-upgrade_nova_percona_devstack_131007/3fd0655/131007_devstack_export.log : SUCCESS in 9m 46s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/91/91700/3/check/gate-real-db-upgrade_nova_percona_user_001/d734e03/user_001.log : SUCCESS in 20m 59s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/3/check/gate-real-db-upgrade_nova_mysql_user_002/809d52d/user_002.log : SUCCESS in 14m 59s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/91/91700/3/check/gate-real-db-upgrade_nova_percona_user_002/9252d5f/user_002.log : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":3},{"id":"9211c2a24b9139aedee44921ac534aee8ea8e6d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-16 15:54:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/00/91700/3/check/gate-nova-pep8/d59439f : SUCCESS in 10m 56s\n- gate-nova-docs http://docs-draft.openstack.org/00/91700/3/check/gate-nova-docs/811cd79/doc/build/html/ : SUCCESS in 10m 38s\n- gate-nova-python26 http://logs.openstack.org/00/91700/3/check/gate-nova-python26/dc64f80 : SUCCESS in 21m 04s\n- gate-nova-python27 http://logs.openstack.org/00/91700/3/check/gate-nova-python27/a9e75b9 : SUCCESS in 21m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-full/a3dfedd : SUCCESS in 58m 44s\n- check-tempest-dsvm-neutron http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-neutron/2a54683 : SUCCESS in 55m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-neutron-heat-slow/e35b06b : FAILURE in 18m 24s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/00/91700/3/check/check-grenade-dsvm/ad20624 : SUCCESS in 36m 48s\n- check-grenade-dsvm-neutron http://logs.openstack.org/00/91700/3/check/check-grenade-dsvm-neutron/b19e6f2 : FAILURE in 29m 07s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/00/91700/3/check/check-devstack-dsvm-cells/0aa851c : SUCCESS in 10m 51s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-postgres-full/0caf80d : SUCCESS in 47m 17s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/91700/3/check/gate-tempest-dsvm-large-ops/5bd04d1 : SUCCESS in 24m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/91700/3/check/gate-tempest-dsvm-neutron-large-ops/93da4e6 : SUCCESS in 23m 05s\n- gate-nova-pylint http://logs.openstack.org/00/91700/3/check/gate-nova-pylint/8d3ca57 : SUCCESS in 22m 02s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/00/91700/3/check/check-grenade-dsvm-partial-ncpu/32e4d97 : SUCCESS in 37m 03s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-virtual-ironic-nv/fabad54 : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d30ca44dd59f0f65470233e9b00cf8593258b743","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-16 16:05:04.000000000","message":"Patch Set 3: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/00/91700/3/8017/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":3},{"id":"26b37a6c20cfbbedb815483101eb18abae398d65","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-05-16 16:18:35.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/91700/3/ : Test run failed. in 1h 22m 32s","accounts_in_message":[],"_revision_number":3},{"id":"179f20919b9c76c54592a23501fe2e8c3ed85cb4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-05-16 20:35:43.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- gate-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/91700/3/gate-ibm-tempest-dsvm-full/e4c0682 : FAILURE in 1h 24m 24s\n- gate-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/91700/3/gate-ibm-tempest-dsvm-postgres-full/6d25420 : FAILURE in 3h 05m 02s","accounts_in_message":[],"_revision_number":3},{"id":"ddefb0943f20a0a412442038c64593e676010ca1","author":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"date":"2014-05-17 14:13:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4604c2857402c90beb9b6dfa13390c46e3d32bf1","author":{"_account_id":9847,"name":"ricky.bo","email":"boh.ricky@gmail.com","username":"boh.ricky"},"date":"2014-05-17 16:01:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fbe5e52aaff8bc5d6df2887874498836ad593e03","author":{"_account_id":2271,"name":"Michael Still","email":"mikal@stillhq.com","username":"mikalstill"},"date":"2014-05-17 20:14:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f6f9f86b6037038f88612ed478da412789c4fb91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-17 20:14:17.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"5cbfb06f45af045db6f2ece1a38261aa37cdb07d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-17 21:13:03.000000000","message":"Patch Set 3:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"e949be9ca3f2ff93b42afef76b80e3fa906cfb8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-17 21:13:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/00/91700/3/check/gate-nova-pep8/3160467 : SUCCESS in 9m 26s\n- gate-nova-docs http://docs-draft.openstack.org/00/91700/3/check/gate-nova-docs/47ef7bd/doc/build/html/ : SUCCESS in 10m 47s\n- gate-nova-python26 http://logs.openstack.org/00/91700/3/check/gate-nova-python26/50c35df : SUCCESS in 18m 08s\n- gate-nova-python27 http://logs.openstack.org/00/91700/3/check/gate-nova-python27/1c8625f : SUCCESS in 21m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-full/bf81c90 : SUCCESS in 58m 40s\n- check-tempest-dsvm-neutron http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-neutron/f04e297 : SUCCESS in 56m 56s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-neutron-heat-slow/c44877c : FAILURE in 19m 39s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/00/91700/3/check/check-grenade-dsvm/5de382c : SUCCESS in 39m 36s\n- check-grenade-dsvm-neutron http://logs.openstack.org/00/91700/3/check/check-grenade-dsvm-neutron/105aa9b : FAILURE in 49m 53s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/00/91700/3/check/check-devstack-dsvm-cells/013df44 : SUCCESS in 12m 32s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-postgres-full/06f3a61 : SUCCESS in 46m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/91700/3/check/gate-tempest-dsvm-large-ops/87dc108 : SUCCESS in 20m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/91700/3/check/gate-tempest-dsvm-neutron-large-ops/81c3e11 : SUCCESS in 22m 37s\n- gate-nova-pylint http://logs.openstack.org/00/91700/3/check/gate-nova-pylint/0bc11b1 : SUCCESS in 17m 19s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/00/91700/3/check/check-grenade-dsvm-partial-ncpu/b7676ed : SUCCESS in 36m 18s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/00/91700/3/check/check-tempest-dsvm-virtual-ironic-nv/42ccd2e : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"def52d0d7fbd675c9945e822f33957dcaeba90df","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-17 22:53:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded.\n\n- gate-nova-docs http://docs-draft.openstack.org/00/91700/3/gate/gate-nova-docs/218b81d/doc/build/html/ : SUCCESS in 7m 52s\n- gate-nova-pep8 http://logs.openstack.org/00/91700/3/gate/gate-nova-pep8/2ee31cc : SUCCESS in 9m 14s\n- gate-nova-python26 http://logs.openstack.org/00/91700/3/gate/gate-nova-python26/b102122 : SUCCESS in 15m 08s\n- gate-nova-python27 http://logs.openstack.org/00/91700/3/gate/gate-nova-python27/6d67021 : SUCCESS in 18m 03s\n- gate-tempest-dsvm-full http://logs.openstack.org/00/91700/3/gate/gate-tempest-dsvm-full/9bdfb3e : SUCCESS in 51m 01s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/00/91700/3/gate/gate-tempest-dsvm-neutron/45d7b71 : SUCCESS in 58m 32s\n- gate-grenade-dsvm http://logs.openstack.org/00/91700/3/gate/gate-grenade-dsvm/139f66e : SUCCESS in 46m 41s\n- gate-devstack-dsvm-cells http://logs.openstack.org/00/91700/3/gate/gate-devstack-dsvm-cells/bc566d0 : SUCCESS in 11m 56s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/00/91700/3/gate/gate-tempest-dsvm-postgres-full/3fa0d6c : SUCCESS in 53m 25s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/00/91700/3/gate/gate-tempest-dsvm-large-ops/270fe19 : SUCCESS in 20m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/00/91700/3/gate/gate-tempest-dsvm-neutron-large-ops/b14e61b : SUCCESS in 19m 34s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/00/91700/3/gate/gate-grenade-dsvm-partial-ncpu/8564b33 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":3},{"id":"2813893c4b199f95937b1c81ad938814703a04f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-17 22:53:22.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3},{"id":"16ceda20cf2dcd4cdde670304b6155e61bdedc0f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-05-23 03:44:52.000000000","message":"Patch Set 3:\n\nBuild successful. Logs: http://208.91.1.172/logs/91700/3","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9c285c4bc777141edc80c6a40a3e1ff491aae217","revisions":{"ae0345001752bdfcebce180534ab297968f6a3be":{"kind":"REWORK","_number":1,"created":"2014-05-01 23:18:15.000000000","uploader":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"ref":"refs/changes/00/91700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/91700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/91700/1"}}},"commit":{"parents":[{"commit":"12b616d0ded1c55ff1524a7c5a8a208b5f11f121","subject":"Remove quota_class params from rest of nova.quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12b616d0ded1c55ff1524a7c5a8a208b5f11f121"}]}],"author":{"name":"Joe Gordon","email":"joe.gordon0@gmail.com","date":"2014-05-01 23:13:54.000000000","tz":-420},"committer":{"name":"Joe Gordon","email":"joe.gordon0@gmail.com","date":"2014-05-01 23:13:54.000000000","tz":-420},"subject":"Remove quota_class db API calls","message":"Remove quota_class db API calls\n\nPrevious patches:\n* Remove quota_class API 1110022d6f628d03aaf363da707f2d2ef1600437\n* Removed the quota_class from nova.context in I85dd24121a434b95dc6f0af180444d4258b50765\n* Removed quota_class from nova.quotas in I1e788414f2f86d846612d3c4e909caae46273f4e\n\nNow that we aren\u0027t using quota_class anywhere, remove the DB API. A\nsubsequent patch will add the database migration to drop the table.\n\nChange-Id: Idd888cc45533140d500fbe0bca1fed36d6fb0274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae0345001752bdfcebce180534ab297968f6a3be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae0345001752bdfcebce180534ab297968f6a3be"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12b616d0ded1c55ff1524a7c5a8a208b5f11f121","is_merged_in_target_branch":false,"change_id":"I1e788414f2f86d846612d3c4e909caae46273f4e","change_number":91699,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b1a66bf7b5f65939b6c5acd0516c578b54be8fa1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2014-05-14 20:54:42.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/00/91700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/91700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/91700/2"}}},"commit":{"parents":[{"commit":"12b616d0ded1c55ff1524a7c5a8a208b5f11f121","subject":"Remove quota_class params from rest of nova.quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/12b616d0ded1c55ff1524a7c5a8a208b5f11f121"}]}],"author":{"name":"Joe Gordon","email":"joe.gordon0@gmail.com","date":"2014-05-01 23:13:54.000000000","tz":-420},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2014-05-14 20:54:42.000000000","tz":0},"subject":"Remove quota_class db API calls","message":"Remove quota_class db API calls\n\nPrevious patches:\n* Remove quota_class API I1110022d6f628d03aaf363da707f2d2ef1600437\n* Removed the quota_class from nova.context in I85dd24121a434b95dc6f0af180444d4258b50765\n* Removed quota_class from nova.quotas in I1e788414f2f86d846612d3c4e909caae46273f4e\n\nNow that we aren\u0027t using quota_class anywhere, remove the DB API. A\nsubsequent patch will add the database migration to drop the table.\n\nChange-Id: Idd888cc45533140d500fbe0bca1fed36d6fb0274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1a66bf7b5f65939b6c5acd0516c578b54be8fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1a66bf7b5f65939b6c5acd0516c578b54be8fa1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"12b616d0ded1c55ff1524a7c5a8a208b5f11f121","is_merged_in_target_branch":false,"change_id":"I1e788414f2f86d846612d3c4e909caae46273f4e","change_number":91699,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"9c285c4bc777141edc80c6a40a3e1ff491aae217":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2014-05-16 14:55:26.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/00/91700/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/91700/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/91700/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/91700/3"}}},"commit":{"parents":[{"commit":"d1442a6aec5e1e20cb35627b77739dbe92770e7a","subject":"Remove quota_class params from rest of nova.quota","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1442a6aec5e1e20cb35627b77739dbe92770e7a"}]}],"author":{"name":"Joe Gordon","email":"joe.gordon0@gmail.com","date":"2014-05-01 23:13:54.000000000","tz":-420},"committer":{"name":"Matt Riedemann","email":"mriedem@us.ibm.com","date":"2014-05-16 14:54:55.000000000","tz":-420},"subject":"Remove quota_class db API calls","message":"Remove quota_class db API calls\n\nPrevious patches:\n* Remove quota_class API I1110022d6f628d03aaf363da707f2d2ef1600437\n* Removed the quota_class from nova.context in I85dd24121a434b95dc6f0af180444d4258b50765\n* Removed quota_class from nova.quotas in I1e788414f2f86d846612d3c4e909caae46273f4e\n\nNow that we aren\u0027t using quota_class anywhere, remove the DB API. A\nsubsequent patch will add the database migration to drop the table.\n\nChange-Id: Idd888cc45533140d500fbe0bca1fed36d6fb0274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9c285c4bc777141edc80c6a40a3e1ff491aae217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9c285c4bc777141edc80c6a40a3e1ff491aae217"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d1442a6aec5e1e20cb35627b77739dbe92770e7a","is_merged_in_target_branch":true,"change_id":"I1e788414f2f86d846612d3c4e909caae46273f4e","change_number":91699,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
