)]}'
{"id":"openstack%2Fnova~922073","triplet_id":"openstack%2Fnova~master~I585362995845f9a4299290aec67c085db55f933a","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-25 11:50:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-27 06:27:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I585362995845f9a4299290aec67c085db55f933a","subject":"Remove unused default argument","status":"MERGED","created":"2024-06-17 07:11:41.000000000","updated":"2026-08-28 06:29:56.000000000","submitted":"2026-08-27 06:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922073","meta_rev_id":"86fd62c9cb32a50d0c734d242005ff72c6106777","_number":922073,"virtual_id_number":922073,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 06:27:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-25 11:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-08-25 22:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-25 11:50:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 07:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 09:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-11 02:53:27.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-11 12:19:24.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-11 18:27:09.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-24 22:17:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-25 22:39:21.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"143fc3335c6b8b0e92f0389f50b0bba225eec23b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-17 07:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e380b69725c63a16d37d9ffbbc4de9e585f06b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 07:27:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/381c72a626cd435da3427bba22c51356\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08956e8e466d471294eb5d0f38b26dd3 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8c367c1c20b4d379296d18a3a326133 : SUCCESS in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b49982650e1831b438eb7f1c72302b6aeb87fdfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 09:30:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aec60c65374d4f1cbea1d721b16b52ff\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fca8b51883e04f609aa8e6a1e42861ca : SUCCESS in 1h 00m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/35693986ae124460b7190465b0044c11 : SUCCESS in 1h 48m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6ec1d3820884e128722d0d19647e855 : SUCCESS in 1h 05m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1bed9e2362245cdb21f232a5027b8d4 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf042f15dbe4ec6874ca64372b31249 : FAILURE in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c5edd33906144fd89f118d0bf9a6120 : SUCCESS in 12m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03855b07392d4cf3b431b263055a54f7 : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9b938e3cee848538739bb07d171a8b6 : SUCCESS in 19m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf66934f63d40849df189f757df6666 : SUCCESS in 11m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe584d50cacb4b77b49c5a41581c4c07 : SUCCESS in 1h 07m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d18c9c8544e049748066bd9ece175351 : SUCCESS in 1h 59m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5a77daaee2a64a9abbeb1b3d45a0bd9e : SUCCESS in 52m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/94ff2e513a2f4367bb0ca8a3f437aa94 : SUCCESS in 54m 42s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/dd6737bdfd8246a582f02392963e6c70 : FAILURE in 1h 08m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c2da68db2960406da507b9854b7e47e0 : SUCCESS in 2h 05m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/5221148436e0471dba17ef54fc97092b : SUCCESS in 1h 28m 23s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9da77df0c46e4a7e9cf93990b6e9de37 : SUCCESS in 47m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7d71321a8e774c11b2fc2f76f526b435 : SUCCESS in 4m 42s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7679fabe63b4c779f43cc917f9a801d : SUCCESS in 27m 10s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2216503221448018b2ae77e3aeb7809 : SUCCESS in 25m 01s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/176f53d781194c09b0cd5fdcaab3a658 : SUCCESS in 27m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3ee3f54ce46d48528c2a9dd2c2d74139 : FAILURE in 1h 37m 00s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/2e6b22cd0d474ff9a4525e5e92785d85 : SUCCESS in 1h 44m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/28e64af205cd4f9d9bb6508c14999d07 : SUCCESS in 1h 20m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8906e0b11a234ae88b47bddf6d48a738 : SUCCESS in 51m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e91178d5d34b442d86bef096d39d34fc : SUCCESS in 24m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/27b318c3ca6c42fe927c3f512c59ab2c : SUCCESS in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9459bf54c9c33076c5313b4f7bb1018cced7dcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 01:52:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b3e302e2df087be2082f4ae5e66353519833d70b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 02:38:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b755920a9064660aaeafde749ff1311\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d77a45ec5cf945ec9fb1658924f883a6 : SUCCESS in 23m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d1cb93a993104bad844c0385bf54d8fd : SUCCESS in 25m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e465925c6e7340fcba6dcae5bc054115 : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95d453af2deb4bf01a3d52cf6b6389f6ffc55bde","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-11 02:53:27.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I585362995845f9a4299290aec67c085db55f933a .","accounts_in_message":[],"_revision_number":2},{"id":"49298cc0c5dce68da9ea65a6a886555d2b26c843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 04:20:04.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4527df78409247a685d5a0b5b1d3bdd1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6200775d10b7498b852bd8505281b808 : SUCCESS in 1h 03m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4bc09fba30a245d6ba7eebfb63015854 : FAILURE in 1h 29m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d404b6f104c74c4694c6ca777834616d : SUCCESS in 1h 15m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcf13b313fa747aa83c50e703de1e322 : SUCCESS in 1h 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73aa461592a40b89608c8513c975049 : FAILURE in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/806047caebd244b6a247269a5e80e58e : SUCCESS in 15m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fb90c67c9e24f57b80b00685e40921c : SUCCESS in 15m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64cba87629cc43338bf02b3b51947768 : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/041fa102a3e04773b537f0026449725d : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31e6c05c01084183b44eec9277c78546 : SUCCESS in 9m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/604e5e4d9b604b229d3afe3cec0df104 : SUCCESS in 1h 00m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d48d9ef02b5949a4a314411b3fcc04c4 : SUCCESS in 1h 43m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f3f304c6c8424377903e5d2d3b12557e : SUCCESS in 58m 12s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a315919a2a8d4df3aeddfb79d21ac706 : SUCCESS in 39m 49s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f2374fc228434377ade1f7ccd9a56510 : SUCCESS in 1h 05m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6d9039a651994839a8e484dbd7669bc7 : SUCCESS in 45m 44s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/50ad229e54b248549da1c6c6c4cc952b : SUCCESS in 58m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cacbb03755774731bb0bf84ed13e386d : SUCCESS in 2h 20m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/a54f7e718a2a4e6281f911c94d5af087 : SUCCESS in 1h 58m 00s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/65697b0384d04ba8a43128efc21a0ef9 : SUCCESS in 55m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/86f29da9bc914edcbed4d1b855f14705 : SUCCESS in 6m 33s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b1f7748bde14a179a2bf2ed6afea5af : SUCCESS in 25m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/45450918781d4b999508940eeb0fc5e4 : SUCCESS in 26m 23s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a51befaace7741709cbe8fcf5d7ad7ba : FAILURE in 49m 16s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/03e10672b64b404d92ff0c56c8d77b2b : SUCCESS in 1h 33m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ede2d08c68c414bb25a46f6d3a9fefb : SUCCESS in 51m 16s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/77f73a242d0845528b9572fa5925c992 : SUCCESS in 38m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/90c285e471f34055b1c4c81bf44141b0 : SUCCESS in 26m 19s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0e86c6583b3941b0b583934ca0bd4400 : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d21e72401eef148dcd76f7ac615e077d0cb77a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-11 10:25:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1443292f190a8a1984a60ab8d1395ca88b4455d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 11:22:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/551af741acd44a5db37a94fc8c9002f7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6917647b2ae54db29e5308fcaab3e738 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47a11bc91a064e368519c6d5f531dd54 : SUCCESS in 26m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1cc76ba02b644fbae4138830cf5dde4 : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d5378446f15e3e6c54ea84945e82cb83ba2338f8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-11 11:26:21.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I585362995845f9a4299290aec67c085db55f933a .","accounts_in_message":[],"_revision_number":3},{"id":"7e3fba0ba99c38b571c94b19a9afc950b43b92ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-11 12:19:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73e2de1c21fc4e178f35c6f82af71149\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f14b3f5c48214985b815c9a6a251965f : SUCCESS in 1h 04m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f8cf60ee62724cc380dd5f85d78624d4 : SUCCESS in 1h 53m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b99acf3d4c54453a52e08468c995c60 : SUCCESS in 1h 09m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3fc0a9297834f538d00085f0a926c03 : SUCCESS in 53m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4471151f329848ff9376edc3208c77bf : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/555f25e18f084ba4b9efd0c1f749ef50 : SUCCESS in 9m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/009d7c0b909f42b799229ff45352b9a4 : SUCCESS in 12m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0b69000fd1654181ad9e3c9341d96f7b : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b354176d831047e39108f0ad8f591ddc : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c1a9c679594427e917b9f80db628daa : SUCCESS in 6m 29s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bfab85f3225843069566d61ccf8be509 : SUCCESS in 58m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/231b83cfa74843d4b509df8233a66f69 : SUCCESS in 1h 41m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/043bc656755747b48166011b5aa743ef : SUCCESS in 46m 11s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ae2479f745b345a28bcdc969d6e180ae : SUCCESS in 41m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3958a33b68ab4ebbb5bca2825d8148bc : SUCCESS in 1h 03m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9ac69eba25b0436bab801c1741efd00b : SUCCESS in 53m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/24e3856562be4a6c8ab42610d42c9866 : SUCCESS in 1h 00m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5b0cd66652b249ef8d062e968dfbd192 : SUCCESS in 1h 29m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/4260d88d66c240da8aef43787c130460 : SUCCESS in 1h 48m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5a367e91d06d4ccfb58b32ab4b647736 : SUCCESS in 42m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6c457668f64741028647b3c788fe6040 : SUCCESS in 4m 31s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9b614af0b5d24ec6a3bf5fe719ab9487 : SUCCESS in 26m 39s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ad27baa40d4f41d99078a270bb2cca7d : SUCCESS in 27m 20s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/756b8365a2ef4e25b6773d00cc7797a8 : SUCCESS in 50m 32s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/63d9c17a850b48afb4d2ae604b8834ad : SUCCESS in 1h 32m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a215c8675c4431b9b8386acdc45ecfe : SUCCESS in 58m 13s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1f9e9348ac7e46d99f0257c6d580d50d : SUCCESS in 35m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5d572692a6414635b5e08ad7942d1899 : SUCCESS in 33m 14s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d2723ee98dc64f8d9fdd19d283beb03f : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"04da0d03686d78c2a3a25fb5e0ae2fbb5d1aed67","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-11 12:29:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"3b7fc425412a9386c6ddc8ab7b5cb81cec037cb3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-11 18:27:09.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-nl64r/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"7f46be38e2267bd86f50b4a92968accca17387e7","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-12 04:58:48.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-7cdql/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"a157bec44b0e57a293c0379f2c480b03c4991331","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-16 02:26:16.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-9z46b/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"5cbcd85e343d9b010692631969830454888f8fc9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 22:17:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"54632f663153163a35e2f3d5d9732dd87ef62228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-25 10:07:39.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a7abeeb6a56b75597acdae3509c6961b08690736","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-25 10:08:02.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"73708c4d8169f005c4803840fe6868faaad4de7a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 11:05:05.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I585362995845f9a4299290aec67c085db55f933a .","accounts_in_message":[],"_revision_number":4},{"id":"bfb274d8c947acdfb4daabb1ed12b65be5c77604","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 11:50:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"63882b7c9bab4813f4d6fccfcbec912e4f5b631a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:34:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd1da433bae04ab0914f646319eeb81a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67d8828002e84c5fa5d48a46c80777ec : SUCCESS in 1h 08m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/46ecd88d39714224bcd69ca0042aca9a : SUCCESS in 1h 49m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a11530f04e0b4a4fb56f584f8e056288 : SUCCESS in 42m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98cbf6e2e81f4d9f994f7d06951ccc95 : SUCCESS in 1h 40m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1a43c69db5414896c917756550bdbf : SUCCESS in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65c2d8902f4e429dada18e8b694694cc : SUCCESS in 11m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75d5dd67d56149e8b9ebcf06d8047437 : SUCCESS in 12m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1fa94ec375ce451594fe0fa79e0bdaa8 : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef664bd699024607be26df6409fbf6a4 : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22bcbf5134d64306b63a3b89bc0eabf5 : SUCCESS in 9m 00s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/22cd784e081844478dfb778f2c57b301 : SUCCESS in 1h 05m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59874668dcf74e59b504ccdc6050ae61 : SUCCESS in 1h 38m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/260c5ad61fb04507a5d79f966e89f648 : SUCCESS in 48m 40s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/47ca00f6894d45578cc037d7ce29b7f6 : FAILURE in 39m 04s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/77f461af52f646a5bdb4ce1178084997 : SUCCESS in 1h 01m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f306cb53811f4327adc2e94ce5fcf422 : SUCCESS in 52m 01s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/39b1759ce4284f13bac637b64231a52b : SUCCESS in 1h 07m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0e80ac0f90294afbb0c8ac47223f49aa : SUCCESS in 1h 59m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/7fe932b2970a49d0b74c0a83c378b358 : SUCCESS in 2h 18m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d438069f43f444b5a83eb4c00ab1c695 : SUCCESS in 52m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ac23d8b00cd64e57bdfb4d993c542c37 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9a557e2aaa51452c9e2b8ab6537b491c : SUCCESS in 23m 18s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e68d432a6c044eb1affe389b615d245c : SUCCESS in 35m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4fc3152da87d41f2ac4e24ea2fe6cc64 : SUCCESS in 33m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c38fdba533854d02902fa5f256a023d5 : SUCCESS in 1h 46m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18a9a9cfdae443cf9274f96caf321fa9 : SUCCESS in 59m 52s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0d6052dfbc164701b364b44793ef9710 : SUCCESS in 37m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/21f1a690a32e4bb493a2a6b0069b06a1 : SUCCESS in 24m 01s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1e90952d202747829252bfe05cc3ea4e : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":4},{"id":"d847cd2e67cbe0e1bfd73ffc9269d22e8ee2719f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 12:53:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/093181ec74564d6bb73840839bd34c85\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/24e4ba6edb1648ba9b95591f8356701f : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eb788275743446138531e26641ee8c31 : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e37a052cca8f4c4ebfef9cb12bfd528e : SUCCESS in 22m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"54e39cdd995ee12aa351448dd460fa55f8eca1eb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 22:34:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36a9f685ee85f8b23a4ff8934658fd9fb6bd0019","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 22:39:21.000000000","message":"Patch Set 4: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2e7d260a477647d98dbdca78f9385af2\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/0f6f42ab83c24d68998c6206d8a819cf : FAILURE in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e273ab0d76958e30d688bfa5c3076b2f640846d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 23:13:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c268b41d6cf439894a0b1f78fbb1b15\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3e72e38f48c4fb4ae74ace0c003501b : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1f9b2b811d5c4b75a85da899baad0a49 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/12adb1c75c9846308095b97ef18c345c : SUCCESS in 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c0ce63ad4d7d2781ba2c086434f79f0af5c473a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:21:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fd2c6b734d4b3da435026a87892fee\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ca8ba0fd97e4bb5850ddd1d3c77becc : SUCCESS in 1h 06m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ba00a2cce30c4f12bbd74807f305c956 : SUCCESS in 1h 06m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a63f0d88a3244327adc9dd8dbba19820 : SUCCESS in 1h 01m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8d7c4df1ab94372a8445fcd7e64a968 : SUCCESS in 1h 04m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a327b73bff4f41eb94f2a7ac5aa9aa31 : SUCCESS in 9m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46a157ce3da54b0592db00598d673d7f : SUCCESS in 12m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a4521247f42454ba09243e18465d1be : SUCCESS in 13m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a702e73df9214e4a9603290dd7c41585 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52ec7b2341e44b18ac0eef68ffca192 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0558c6180d214f729c2fded7ef9c40db : SUCCESS in 8m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/cc5c67448095401dad6a662b6f20798d : SUCCESS in 56m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/994d09a195b7402c8856e3a0f1724951 : SUCCESS in 1h 37m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f76210e97d0b4f56b4d2323bfc66a528 : SUCCESS in 25m 47s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d6849207862e4fa6a71632ef37f7317d : SUCCESS in 35m 42s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0dc69cc2d09f4a0dafa15d827ebaa1b1 : SUCCESS in 1h 01m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a67136b160ac475bac791578bccdcd3c : SUCCESS in 30m 06s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/781aa6deb67a4dd7931e21e6803a7bec : SUCCESS in 1h 08m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/befecc3d12f14811bf34237f6b8bfcd3 : SUCCESS in 1h 42m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/89d9cc8507c447daafa4d9ced185fa1f : SUCCESS in 54m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d42ff59086cb4fd7a4c5f173e6e76ac6 : SUCCESS in 50m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/58e0ff6ae43d474ea08f514c1f3d2860 : SUCCESS in 3m 54s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8c5458c572d4430f9eef577f44ed637e : SUCCESS in 28m 09s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/25f58814b7eb4289b8c1ca3da778993b : SUCCESS in 30m 25s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/edfbbce2bf0646d7b00d1b0a7812896f : SUCCESS in 45m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/43e07c71a0e9443f885e74a93f9a51ed : SUCCESS in 1h 29m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b8ca3a899e04806a0ef8f23353e5da1 : SUCCESS in 40m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f1f5ee3a078f4d88900c24c7593862f6 : SUCCESS in 26m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4db4013b0a8740b98eb2736631ddfe61 : SUCCESS in 32m 33s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/53fcb206a82e4331850886b65003d2a6 : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":4},{"id":"317668ca9da8eaa0eb3bcaceae3d2990ccb8b36d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:22:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"de257198aacddb9e4e7a5793cf1043629829b65d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 00:32:05.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"5b8a20938420fdd268bb0382973761c3f6472cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:17:56.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/672234d9c7144ef58463c2c0f3c2c5e9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a16b45082f624070b8fd89f1fbae8510 : SUCCESS in 1h 03m 05s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6fd2935e0c3d42cbb5f37dc5be5fa5f2 : SUCCESS in 1h 49m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a45eb292645a47e198d54cf24639a160 : SUCCESS in 1h 02m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5581763bc34b1f919028195a755328 : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd1001ca64d84b3fb11b655f05b2ede3 : SUCCESS in 16m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c9746e5b54f4175b1a6fedbd1f9fd79 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26bac5c1477041a28d3ebc242f26608a : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3263e7e9b4114d2eaf20fb1d20df9835 : SUCCESS in 9m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d58d5955ed074d1186230e5c55da2c42 : SUCCESS in 51m 49s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9db04d91532f4667b2fc3e70f07e61e9 : SUCCESS in 27m 35s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6aa64620bc1e4a36885d705b2c5c9e8e : SUCCESS in 26m 25s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2ebbff5f477044d0854d3b559e92f351 : FAILURE in 31m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7cc81f33cf894de8b9f99642d4add9f0 : SUCCESS in 1h 45m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/1f1356843d9548498a37ea25bb392b32 : SUCCESS in 1h 45m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3036ab0db0954b5f92c5d3826cd85e78 : SUCCESS in 7m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f436e427aee7463ea7338e85751d48e5 : SUCCESS in 47m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4f5eca638e84940bdf7e52182ce7755 : SUCCESS in 1h 34m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43978a1729244149ae82720470eac441 : SUCCESS in 1h 00m 52s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a22cee565c554d1c8b0e6cbc613b12a8 : SUCCESS in 13m 22s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a61a48c8fea0478fb33866699875c400 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":4},{"id":"8690d808eb9c0a256dda36bf015fd04050415b17","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 06:32:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5530d05bc7acaf1b676179c4f2b3a50a95cb308f","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 06:39:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/deaa8425eb9c4b4aa0ebbf0ffc5ccbbe\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/33fe3bf8445f469a903dbf8a171689e3 : SUCCESS in 6m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1671180ead22211cc7c589aaf42cb8d19fdb1041","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:15:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d9297dc7a0413bbfde0d90e90ee0c2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4c51b9fa8d4946da8e97a974794286ee : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b96cfa43ee4242179b468e12ca16504b : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7434fedabdec45129b76929477c7c9a0 : SUCCESS in 20m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c9a2ea76c1842371d5db7a9da75c8b33ac47e2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:34:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc08173fa62443048c7001c9c0fea712\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f16e20499f014f6382a19bda81c0bc6c : SUCCESS in 37m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/493daf1f020b465b848288235811a260 : SUCCESS in 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a7916bdc1a242c5be93cb9ebee4928d : SUCCESS in 28m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d531e19539ec4793ada60017f420ec22 : SUCCESS in 1h 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdfa5841d3e740af83c862b07f4182b0 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f7b873f9a4245f4af772d9daf22cf03 : SUCCESS in 12m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e9f404a65fa4c31b0f8eeec1e9247b6 : SUCCESS in 16m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c649112746a4ba9948daa61a707a16d : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487a130ad052462c93d13a123eabb7f5 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86b718e294474963b7fb286e6d87efe1 : SUCCESS in 8m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ace14cc76c3a44b88a20d3756185036d : SUCCESS in 27m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbb8e410909745bab0fbc6d5ab7ab7c9 : SUCCESS in 1h 13m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/12fef3ec1b2f413394457be64b862088 : SUCCESS in 41m 49s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1857017c8e0c4f26a6708fcb6be39442 : SUCCESS in 35m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2b01f4280f2a42fba999b5dce1634e3d : SUCCESS in 49m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/685118c4f5bd46d7b79edea247a89c23 : SUCCESS in 29m 53s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a577cba2ba5740bf9d8ee74849c6c886 : SUCCESS in 43m 26s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/aaef9776f5ca4497825d6c9667349ffb : SUCCESS in 1h 48m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/deebd0b1489644189058f98da1503c66 : SUCCESS in 1h 53m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e3605e019b994ac28492e4da71a1407b : SUCCESS in 38m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7de48a5969e743f596caa172ee80e06b : SUCCESS in 5m 33s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/74e5516f1982489ba26ff17ee7f549dd : SUCCESS in 24m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b6ea1f23bc54441ab74fbb56077e933b : SUCCESS in 31m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/146a9539cebd45af816e3cebe28a38ad : FAILURE in 36m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8751f4678e6d4a1aba0dc5a200830a55 : SUCCESS in 1h 23m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e98365e3c81043daa3b5b3be3aa925a1 : SUCCESS in 1h 01m 44s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4c150d716b9a4c09b7600a5277514057 : SUCCESS in 54m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bce5de37db774244a4feaa747500abd5 : SUCCESS in 16m 59s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c5f2563a270b4a80b540f67452433c00 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":4},{"id":"bcad9b34e50993a74631068d3db2ea92c30788b9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 16:30:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a8daa1bf27410fed8ab6aa05da967749991f0b93","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 16:39:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b6b8ce5668f744b08cc7ce673627d814\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c4b3c15fa90c4d4ebc9e6c54880c3035 : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d32d245c1502b41b9346403bf7861b8aee258e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:49:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d12fb764231445ba60268db668ecf16\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e0d318beb084f7f9cc66bae678fb39d : SUCCESS in 1h 08m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bacd05ef08bd427786178b8bf7ad4ab9 : SUCCESS in 1h 09m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa645d764e664a5ebd2af8d4fd60aa27 : SUCCESS in 1h 06m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd8312f3218c4ff9bde8094cf0aae47e : SUCCESS in 1h 38m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f95d9d433e748ff84c0a133c740abbe : SUCCESS in 10m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db0bdd98a7b74740bf2a09282eed1b28 : SUCCESS in 13m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f8434463822444a8949f8f93187e1f5 : SUCCESS in 11m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c3c3491edf14bae968c61b9e750ff61 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1dba0c6b814ded86c2e219b3a9688b : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/504d0153614041909bedc825af882a06 : SUCCESS in 8m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/8883afac50db4cfa81cdd644ceb17e75 : SUCCESS in 1h 03m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56425cad599645dbae45112364222942 : SUCCESS in 1h 52m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0f3620fbf37946c6a6d572fa41e58ea2 : SUCCESS in 50m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/7528741c077045b4a1ca9b9ce483e205 : SUCCESS in 40m 04s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6b6269e9c3324ff9ae0901f12a9515e1 : SUCCESS in 39m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/96e17007257a4b4a90e50fa6593bc25b : SUCCESS in 51m 45s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6749c235d39c400db6eea02e529389e9 : SUCCESS in 1h 15m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c47ae0d29c0f434a85060898d645ea3a : SUCCESS in 1h 42m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/73be1f4b3eec4d31b37f148379615ff3 : SUCCESS in 1h 34m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/97a4cdb9be484742a0bcf6d70f121f7f : SUCCESS in 50m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/792e683ddd444b128e2017c41e1563fb : SUCCESS in 5m 09s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6bf8bb708a874d4cbc73b72146e11cee : SUCCESS in 27m 39s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b292758dbd6b4a288390200cbb9bece6 : SUCCESS in 27m 12s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/70996a37f4d24056b3cd887aad134516 : SUCCESS in 29m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2952d909f84c42648b5e50d768b37af9 : SUCCESS in 1h 29m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff62f92e88254aeabc997afbea68ee26 : SUCCESS in 1h 14m 11s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2e3faaef920f435481cfd3852c3e4dd7 : SUCCESS in 52m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/50cd0fe06e5e48d88d03ddfc3d8f0d34 : SUCCESS in 32m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/75ecf7f61c1f4320b8cde9cb490501e3 : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":4},{"id":"c7d06f999dd43fadfd6ebeba7ff040270cf017e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:50:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f4ca3406a4fbe5d4e57cfdb626c2e5dfd98dfdd5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 19:00:01.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"5f4daf2b2af62fff442be7c3ca122f2608d60310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 21:03:37.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d418301d48c4c65aa8e1ea6058544e1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/653b6bcbba38451bb6275b72569f8c5d : SUCCESS in 42m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c2158d25c8a74a8f900e8932620ce108 : SUCCESS in 1h 39m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd006a7dd6c24a9bac57a19ddeb4a7ea : SUCCESS in 40m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9382cd621a44224855007f6e5e7286f : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b777431acacc4a59ae02d05ea7a4ea65 : SUCCESS in 14m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67692684b2d840b7bd39bfcd163525c0 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a1f79367a4471692b3de123f6e0273 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7d0935987fc4d36ace2488144960d18 : SUCCESS in 7m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c1ede122474b46dfb07311fc28ee0465 : SUCCESS in 28m 53s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/04993a66e8374f3cba7ce5db376803d8 : SUCCESS in 30m 38s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/67ef348020354b11a28fbc71ea05cbde : SUCCESS in 28m 32s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/aec5c11e8cc44251a31353e0eac5c781 : FAILURE in 40m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3ff0f93994e04dedaca46a26cd1bbb7b : SUCCESS in 1h 47m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/034bed4dd1af4676b5a2a575e7163363 : SUCCESS in 2h 08m 36s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3abdfdb69d394c5191da121e7adedef6 : SUCCESS in 5m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8087361f0b7047179d0201503b4c36a4 : SUCCESS in 44m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/27136efcc3d243e29e691ccf9ba6406d : SUCCESS in 1h 12m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40f267dda3d6461b832c6526feee0cb0 : SUCCESS in 59m 13s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/912cefe151c84e3e9738c7fd7ddb8c6f : SUCCESS in 12m 59s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5ce4c6e7a3454f93ba749ceadedb18e0 : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":4},{"id":"240559eeb7faf305106e9b7565c26d20307d1d17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 21:44:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44026299aa09446bb3803b641f0255ee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e81723e1d8064bb29d920e9f39fafde1 : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f5ac4ec7a3904049bb4e5ee255589ad8 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1a78438abb704de4a5ab172e76477a73 : NODE_FAILURE Node(set) request 0b6a5af8d0db420eb8398fdc2fdafd6d failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7d6b4c4c239dc54917872487a3451418269c1bc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-27 01:03:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e8fb0e392ce1324ca4213836ed5266f13b18d86","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 01:10:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b2b04883feee45f7b76c2f2af3bd38e7\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1e04d4e277244cbdb743b908d41e74a9 : SUCCESS in 6m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9e3ef269512117aa73b25b33546e5cc3f74e3bce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 02:16:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f9e9ba227644d5a0030806fa4482c7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e8f3efa551b4accb848213f54f95563 : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83eead8cd33a489c8ee2177aaef622cf : NODE_FAILURE Node(set) request 79b8a48b35d24805906a9c428580c7b3 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c341474f27b94fabae836f03182a73cd : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ecb6813b54a372d17c62de58fe994078c4df2a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 03:13:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da9bfedaaf834522a8652c08f79c5143\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6acbc65ec7bd4a20afc642ae167aabd3 : SUCCESS in 59m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c047f1ffeada4d84bab22abaecb8e7e9 : SUCCESS in 1h 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/900244e2374e4c5aa36c7d5d3db37b22 : SUCCESS in 57m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9c7dfaea6ab461e9fe269256a38e52f : SUCCESS in 52m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a476f3a83f34b90983fbe8814c51bc5 : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cabebaf8565046e3bf2df5429f9dfdbe : SUCCESS in 15m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e65920843c9a430484bd8b63c84d454e : SUCCESS in 9m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ac4350ed4c7405a9c0029f16db86ad1 : SUCCESS in 14m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a52e1cfeadd4efe80ad1b97a026a26e : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5d5341e418f4140933a86b8e05fb8c0 : SUCCESS in 7m 40s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1f34649933c047ee93e19981aa7ac570 : SUCCESS in 1h 01m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/401ed9fa56d04b549ecc1a54b13fbec8 : SUCCESS in 1h 26m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cc2c32ee8e7946e48f3a8b92ffaa94ed : SUCCESS in 52m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/3865e8f5d1cf4c0688270d1c6e159527 : SUCCESS in 37m 23s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/20779b5b55bc4d969a353f6c1ffcb427 : SUCCESS in 1h 00m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5b78e2cd53b6491a85e182f6fac05055 : SUCCESS in 54m 03s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1d5d5cc9328640a0b515c13290c291a9 : SUCCESS in 38m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d03b6c9889b94931a5abcba42d398ea7 : SUCCESS in 1h 49m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/01795540503e49e6935653e38e9b73b2 : SUCCESS in 2h 04m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ca9b6312d23749d78eb123a43b0b7d5d : SUCCESS in 38m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/db1f0a7761b1416f85285f9d40019369 : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8d093455ca2e45dcaf20179d2f4e798a : SUCCESS in 27m 03s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/de1288f59eab454284b8d7b480db9ae9 : SUCCESS in 28m 17s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c7ab729c816e4776bac6c429ce408333 : SUCCESS in 28m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/684db17d25f148c3b983b67b0a01f33a : SUCCESS in 54m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e317badc59034f67b9bdeb4204aa46a4 : SUCCESS in 30m 27s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/66d26bbf14f643ec92ea64b283c6f662 : SUCCESS in 52m 15s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9dd97aebb4644dd684cc0a2fa2b2bb05 : SUCCESS in 32m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c7e41d440d2447eea1454f589d22b0d8 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":4},{"id":"cbc1b6a510547f81139c2b8d898c2d6ddd61af39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 03:15:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"496181cd43e7ed349b423efec44895af2f79981c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 03:23:59.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"482f02fb4e65d93e13e0ffb45f210a6b455eea9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 06:27:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22343988bad14973b642d8185df4df9d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67352fe9ef6249d39a70231ed04e0872 : SUCCESS in 59m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0bc596df624c47e7b8b939b5176c061d : SUCCESS in 1h 31m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bddaf308a514369836017babd895424 : SUCCESS in 54m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff04eae243a94598a0a5e96ee138dbae : SUCCESS in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9fe4ec008df446b8a631843278a4eef : SUCCESS in 15m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b082d2a4ac244efd8d8d8f4218e408dc : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb12949210a496e9decfc4d4153aa58 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06725940b46541bd98c9b2aa6059b877 : SUCCESS in 4m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/88f31e9af2614a888cd35d63d2338b4d : SUCCESS in 49m 52s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7082c67a4f76463a96b4e1b1b1cc1889 : SUCCESS in 27m 35s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1d6897d9570042059cad559b31068ad4 : SUCCESS in 26m 51s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dceab708511d4904bcea3f5c551f59dc : SUCCESS in 29m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a0891e8a6c5d4bc180c7011a0ab32428 : SUCCESS in 1h 47m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/73087223742c4f2491e1cd7cc90adf05 : SUCCESS in 1h 56m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/44d4c37057d94e4f9a3c865c07dad1bb : SUCCESS in 7m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3004b38b53684d28a16987e355926600 : SUCCESS in 47m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b341aeee7c04631b109f6d4c6f6549d : SUCCESS in 1h 27m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c08404fc03d4df3b4b514e4c55a73a0 : SUCCESS in 1h 04m 30s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0d3132f838ad4bee9872bacea88d65fe : SUCCESS in 11m 48s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d9089d7c54bc4fd99b2d48197ca52b82 : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":4},{"id":"6f2b31416eb5b66956d85be20828270cfb534229","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 06:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2732968d1cd4655bfe6f5ebee62a06aca3bad217","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 06:29:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bfd62016bd14d0ab10bbffae1373598\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08abecf303c24d62b97d8f41a0a3b9ea : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4},{"id":"fb14cc0492457c6efaae75ad0d83eeab339b8303","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 18:50:14.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-qv6wj/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"413cf18a6043e360a4d07489928cfc7b4579e934","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 19:22:31.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-hgr2h/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"4c81eadabea5ffcc14b6d10019754a248e4824ca","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 19:57:01.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-rvkp6/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"4f06747618cf0bfd8dbd63c9190e20e83db03762","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 20:59:17.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-hpjdv/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"86fd62c9cb32a50d0c734d242005ff72c6106777","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 06:29:56.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-922073-qlnsd/index.html\u003e","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"24378cf0beea915451355a92d8edb45866b78d6a","revisions":{"e7e91c89a57127b9c8712addaabf3c8262844cb4":{"kind":"REWORK","_number":1,"created":"2024-06-17 07:11:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/922073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/922073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/922073/1"}}},"commit":{"parents":[{"commit":"428990e7b7908627e3f5f14e8b013b2325a0afdd","subject":"Merge \"Do not close returned image-chunk iterator \u0026 get_verifier early\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/428990e7b7908627e3f5f14e8b013b2325a0afdd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-17 07:10:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-17 07:11:09.000000000","tz":540},"subject":"Remove unused default argument","message":"Remove unused default argument\n\nThe _discover_vpmems method is always called with the vpmem_conf\nargument so the default is not required. Drop the unused default and\nmake the interface more strict to detect unexpected usage.\n\nChange-Id: I585362995845f9a4299290aec67c085db55f933a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7e91c89a57127b9c8712addaabf3c8262844cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7e91c89a57127b9c8712addaabf3c8262844cb4"}]},"branch":"refs/heads/master"},"db26ec2ba3949d57c796262b2eef399760ebf62e":{"kind":"REWORK","_number":2,"created":"2026-07-11 01:52:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/922073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/922073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/922073/2"}}},"commit":{"parents":[{"commit":"334190981e66173ee2697dd03d3aa94fb485c0c2","subject":"Correct release name in concurrency_backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/334190981e66173ee2697dd03d3aa94fb485c0c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-17 07:10:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 01:52:20.000000000","tz":540},"subject":"Remove unused default argument","message":"Remove unused default argument\n\nThe _discover_vpmems method is always called with the vpmem_conf\nargument so the default is not required. Drop the unused default and\nmake the interface more strict to detect unexpected usage.\n\nChange-Id: I585362995845f9a4299290aec67c085db55f933a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/db26ec2ba3949d57c796262b2eef399760ebf62e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/db26ec2ba3949d57c796262b2eef399760ebf62e"}]},"branch":"refs/heads/master"},"c581ff46ebfe0336c9ec337670c8d4b18618e7b9":{"kind":"REWORK","_number":3,"created":"2026-07-11 10:25:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/922073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/922073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/922073/3"}}},"commit":{"parents":[{"commit":"334190981e66173ee2697dd03d3aa94fb485c0c2","subject":"Correct release name in concurrency_backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/334190981e66173ee2697dd03d3aa94fb485c0c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-17 07:10:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-11 10:21:58.000000000","tz":540},"subject":"Remove unused default argument","message":"Remove unused default argument\n\nThe _discover_vpmems method is always called with the vpmem_conf\nargument so the default is not required. Drop the unused default and\nmake the interface more strict to detect unexpected usage.\n\nAlso improve type annotation coverage about the relevant methods.\n\nChange-Id: I585362995845f9a4299290aec67c085db55f933a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c581ff46ebfe0336c9ec337670c8d4b18618e7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c581ff46ebfe0336c9ec337670c8d4b18618e7b9"}]},"branch":"refs/heads/master"},"24378cf0beea915451355a92d8edb45866b78d6a":{"kind":"REWORK","_number":4,"created":"2026-08-25 10:07:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/922073/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/922073/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/922073/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/922073/4"}}},"commit":{"parents":[{"commit":"334190981e66173ee2697dd03d3aa94fb485c0c2","subject":"Correct release name in concurrency_backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/334190981e66173ee2697dd03d3aa94fb485c0c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-17 07:10:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-25 10:07:39.000000000","tz":0},"subject":"Remove unused default argument","message":"Remove unused default argument\n\nThe _discover_vpmems method is always called with the vpmem_conf\nargument so the default is not required. Drop the unused default and\nmake the interface more strict to detect unexpected usage.\n\nAlso improve type annotation coverage about the relevant methods.\n\nChange-Id: I585362995845f9a4299290aec67c085db55f933a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24378cf0beea915451355a92d8edb45866b78d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/24378cf0beea915451355a92d8edb45866b78d6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
