)]}'
{"id":"openstack%2Fnova~923275","triplet_id":"openstack%2Fnova~stable%2F2024.1~I4881c8cbceb30c1ff2d2b859c554e0d02043f1f5","project":"openstack/nova","branch":"stable/2024.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-04 18:00:55.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-07-04 21:35:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4881c8cbceb30c1ff2d2b859c554e0d02043f1f5","subject":"Additional qemu safety checking on base images","status":"MERGED","created":"2024-07-02 14:10:34.000000000","updated":"2024-07-04 21:35:48.000000000","submitted":"2024-07-04 21:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923275-bug/2059809","meta_rev_id":"a3228bba532832d014973751c5bef8fed219f51a","_number":923275,"virtual_id_number":923275,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 21:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2024-07-04 18:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2024-07-04 18:00:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 15:04:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 16:31:29.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a43b06473953d0424f1683e5cc1fadc08e30213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-02 14:10:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36bebc62788ce217ee2af85dd174432d75abe305","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 15:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fc752c8c6754250b871f1582643ab2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eb98784c9fd434db036f9ca677a0ab8 : SUCCESS in 32m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18a87f6de28045d681e3d10694f9b702 : FAILURE in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9adf5a05dc0c6b93f66d1b5fb2bbd412a441997","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-07-02 16:31:29.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-923275-p5ssv/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"8008737b084158562f34ef5e9dd8e5a90d3a45a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:13:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41ca5dee72eb4475860899b2a0a94bcf\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/037ab4b5026f4e668629d89e285253c6 : SUCCESS in 1h 04m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0a21319235a74bf9a1917e8280324690 : SUCCESS in 1h 44m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/643b663ead1544c2af21dad05af3b2e8 : SUCCESS in 1h 02m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a8547f785b54bf890e2470ba78f10c7 : SUCCESS in 26m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02a606d58c244e882539047ba93072a : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd9a8a1296c9424a9de12465cb769abd : SUCCESS in 15m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8b9e7e26c0c40279baf78f30abb36bd : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20e281eb5f94c7daf89de888d99c165 : SUCCESS in 9m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b250b6b611274330af55fa8194647378 : SUCCESS in 1h 06m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/876f06fce600440a80e8160d060c4503 : SUCCESS in 2h 05m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/784d390ecc75459fb07a1357a0bd06dd : FAILURE in 47m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3f033d93a5c54f05b04261656566a1c9 : FAILURE in 50m 10s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/39e3893f11d34e39a2f4dca1c3c5715d : FAILURE in 59m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/04dabf8464e34f49bbba37702aa24c0e : SUCCESS in 1h 58m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/e0eaf0fddfc44c03bdf1c1cfcdc9e26b : SUCCESS in 1h 29m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e885694a53c74bc7b332bc9476017ca8 : FAILURE in 46m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ecf585d2ef984d60bb2ecca09825618e : FAILURE in 4m 34s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc0e5a9056b54db88dde9c3e7b63de78 : SUCCESS in 25m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48d5f9ebdeb04d438fca121d53fa7d9f : SUCCESS in 25m 42s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb9f97cb48644d49955877e5630a2ee8 : SUCCESS in 24m 15s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/950fdbaf6778445f9abd2030aaa878ca : SUCCESS in 1h 47m 24s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/d20602b69ade4fb59d146cb68a52381e : SUCCESS in 1h 31m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1a653744f5a4b229ba2a065d2877c04 : SUCCESS in 1h 18m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/189bad429187483ab5486357be712615 : SUCCESS in 1h 08m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c1f43ae465bf44ecb70ec0d169be80c0 : SUCCESS in 25m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7b5d7eff2f7f45cda09c637e8d0d5cfa : FAILURE in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58b57e5d8510c22c1c613d9a31254df75cd09e9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-07-04 15:22:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"20c1d236a1dd76c3c09d7c839f7293cdfa5c577c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-07-04 15:39:41.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-923275-2kf6n/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"40f2dee344dc2e9551661c3e98286220e0623f64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:01:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f24fe257d9457badd949402bd67aa2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64ce2afca6b4407e921fa20e4ec6f672 : SUCCESS in 37m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/690c1b026a514710b276715361405b82 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05fbdffe77961b56c5e2c0ab84a7a474e2a73267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:29:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8510dfff117b495eaed34c25772667bb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d80fef7e35b448a81cbc5a8e3dc1499 : SUCCESS in 57m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9d3da3a2bcbe4a8f846fba8993fc0187 : SUCCESS in 1h 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18704c51ff4847f09c9abbe5f352a9bb : SUCCESS in 1h 03m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ada3257db03448d8f485c08885033d0 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe477b3de40478aa1d724be3ea1ffe7 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dea04b0dc52640449c305db7b14df36e : SUCCESS in 12m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc9e7a2707c54a12a0dffe29c1992f20 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a7321eea324c29b97da48552971bd3 : SUCCESS in 11m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/53a9d406c5de454f9aa4a2d98b8e7f13 : SUCCESS in 1h 00m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2487ba326d9c4712b7c68231323c7a8b : SUCCESS in 1h 40m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/13ad1c0dc1c44bc4981a3b028120a161 : SUCCESS in 41m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0797d29d32d44884bd33a853e36c3651 : SUCCESS in 55m 24s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7591cf00cbf540f9aca24a8f492d979f : SUCCESS in 1h 14m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9badd30b0b6e4cbd939a5a4a7da9e7de : SUCCESS in 1h 37m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/1b98fb0b7db84369bc2c412166f0f337 : SUCCESS in 1h 19m 26s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/19b5b916592c47efbd3398942cc1a2ca : SUCCESS in 45m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4cd35d4315144a52841173cf6b965171 : SUCCESS in 6m 03s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58746e063b3f4365aa101c3d93507d4b : SUCCESS in 26m 43s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2e7d1fbee7544780b78018e6ea2d37b5 : SUCCESS in 25m 50s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c58cc32d52248fd8643d9378e5c9fb5 : SUCCESS in 22m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/201e5464840048cea8265d410e463e15 : SUCCESS in 1h 59m 50s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/32318cd28ad744498b5acf04e7668a2e : SUCCESS in 1h 56m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d931bbd1dc084f9cb85e4458e2102019 : SUCCESS in 1h 25m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b096f700586b46fb97a7f7b932c5a866 : SUCCESS in 1h 03m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/91797e7f54a34276bb6f007ab6841bfe : SUCCESS in 28m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/07d4662645f44988b3d62192989668a3 : FAILURE in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d809ee9e429625d32d54a307d7456efa57c7c0f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-04 18:00:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc05464a503421f379b598cbc397ff3419df6c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:08:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"487bcf6be134778f15cee1836540c0a5e5a19899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:35:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1928a131d1fb451e957dd6669db6808d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24d83970874a489eb79e69c53450fcee : SUCCESS in 58m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/16fb0b436b4c49459a9b5d264d2c088a : SUCCESS in 1h 34m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f3e1b58ab4147a59d8d8b291d76465c : SUCCESS in 1h 01m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f95efda1da24354a307852b32b43ddd : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91c7dd915a8141e3a8d2e6474d895ccf : SUCCESS in 13m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8357de8b445a47c78cef96df324fd767 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57f5e571372407ab56bc4a2b434934a : SUCCESS in 9m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4a2329ced20a47d68e26f9ed402f959d : SUCCESS in 47m 52s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/622b5d9ba5b44caaa3ef647ce08995b5 : SUCCESS in 27m 04s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fa5fe45621ca4646b13eae23405696a6 : SUCCESS in 30m 22s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0eb4095340e4838ac17290cb8060c96 : SUCCESS in 25m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d329103e858944bb8970b149aa0ead7d : SUCCESS in 1h 40m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/d26e81c88c514b0da919878282b825b7 : SUCCESS in 1h 16m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2adde207d6ec4eaaa86069b18a73398f : SUCCESS in 7m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/148dbfddc51a429385fb334ce67dff6c : SUCCESS in 2h 12m 28s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/626037f1f8504c39b20e51cccc534ccf : SUCCESS in 1h 45m 45s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/5ac0ba050cbf448eb9661e99eb8af1ec : SUCCESS in 1h 38m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0375da9fd8604db29626844f3ed1ed55 : SUCCESS in 1h 14m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6c386738a8945279e69617038f264cb : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a3228bba532832d014973751c5bef8fed219f51a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:35:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a0d5f2afaf40c4554419a0b2488ce092eda7a1a","revisions":{"37b1ab69ad9623876fd138374a556392a8fe36dc":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/75/923275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/923275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/923275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/923275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/923275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/923275/1"}}},"commit":{"parents":[{"commit":"cb9806d8c440e6003ad48ba80d42e0b80700eac7","subject":"Check images with format_inspector for safety","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb9806d8c440e6003ad48ba80d42e0b80700eac7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-24 16:09:36.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-07-01 22:14:34.000000000","tz":-420},"subject":"Additional qemu safety checking on base images","message":"Additional qemu safety checking on base images\n\nThere is an additional way we can be fooled into using a qcow2 file\nwith a data-file, which is uploading it as raw to glance and then\nbooting an instance from it. Because when we go to create the\nephemeral disk from a cached base image, we\u0027ve lost the information\nabout the original source\u0027s format, we probe the image\u0027s file type\nwithout a strict format specified. If a qcow2 file is listed in\nglance as a raw, we won\u0027t notice it until it is too late.\n\nThis brings over another piece of code (proposed against) glance\u0027s\nformat inspector which provides a safe format detection routine. This\npatch uses that to detect the format of and run a safety check on the\nbase image each time we go to use it to create an ephemeral disk\nimage from it.\n\nThis also detects QED files and always marks them as unsafe as we do\nnot support that format at all. Since we could be fooled into\ndownloading one and passing it to qemu-img if we don\u0027t recognize it,\nwe need to detect and reject it as unsafe.\n\nChange-Id: I4881c8cbceb30c1ff2d2b859c554e0d02043f1f5\n(cherry picked from commit 0b0edae94fbc1057d94fe64745125270659e4ca1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37b1ab69ad9623876fd138374a556392a8fe36dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37b1ab69ad9623876fd138374a556392a8fe36dc"}]},"branch":"refs/heads/stable/2024.1"},"8a0d5f2afaf40c4554419a0b2488ce092eda7a1a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-07-04 15:22:07.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/923275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/923275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/923275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/923275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/923275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/923275/2"}}},"commit":{"parents":[{"commit":"f07fa55fd86726eeafcd4c0c687bc49dd4df9f4c","subject":"Check images with format_inspector for safety","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f07fa55fd86726eeafcd4c0c687bc49dd4df9f4c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-24 16:09:36.000000000","tz":-420},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-07-04 15:11:12.000000000","tz":120},"subject":"Additional qemu safety checking on base images","message":"Additional qemu safety checking on base images\n\nThere is an additional way we can be fooled into using a qcow2 file\nwith a data-file, which is uploading it as raw to glance and then\nbooting an instance from it. Because when we go to create the\nephemeral disk from a cached base image, we\u0027ve lost the information\nabout the original source\u0027s format, we probe the image\u0027s file type\nwithout a strict format specified. If a qcow2 file is listed in\nglance as a raw, we won\u0027t notice it until it is too late.\n\nThis brings over another piece of code (proposed against) glance\u0027s\nformat inspector which provides a safe format detection routine. This\npatch uses that to detect the format of and run a safety check on the\nbase image each time we go to use it to create an ephemeral disk\nimage from it.\n\nThis also detects QED files and always marks them as unsafe as we do\nnot support that format at all. Since we could be fooled into\ndownloading one and passing it to qemu-img if we don\u0027t recognize it,\nwe need to detect and reject it as unsafe.\n\nChange-Id: I4881c8cbceb30c1ff2d2b859c554e0d02043f1f5\n(cherry picked from commit b1b88bf001757546fbbea959f4b73cb344407dfb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a0d5f2afaf40c4554419a0b2488ce092eda7a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8a0d5f2afaf40c4554419a0b2488ce092eda7a1a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
