)]}'
{"id":"openstack%2Fnova~923298","triplet_id":"openstack%2Fnova~unmaintained%2Fzed~Iaefbe41b4c4bf0cf95d8f621653fdf65062aaa59","project":"openstack/nova","branch":"unmaintained/zed","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-07-10 20:48:50.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2024-07-11 00:53:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaefbe41b4c4bf0cf95d8f621653fdf65062aaa59","subject":"Check images with format_inspector for safety","status":"MERGED","created":"2024-07-02 14:36:51.000000000","updated":"2024-07-11 00:54:05.000000000","submitted":"2024-07-11 00:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923298-bug/2059809","meta_rev_id":"44a76b7d3bd9f859aa84cd44ed4d30c5f471e684","_number":923298,"virtual_id_number":923298,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-11 00:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-07-08 16:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-08 16:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-07-09 07:33:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:36:51.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 15:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 18:39:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-06 12:25:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-07-08 16:29:20.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-07-09 07:33:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e9a08d15f063f2473fcce42fb77335514d0790b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-02 14:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fa63c559a92066b29e4ca027ad5382ed4334126","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 15:34:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c879a97376034fbd8b2fcf33daa9a202\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18de36582e2d418d99ba30890e3d821c : FAILURE in 44m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72266f0bbdc44e0b8771bab09a826f12 : SUCCESS in 46m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"077ec14ab547700a24837d54d4b79724bcd64fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 18:39:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98490b3bc03a417cabbf739b70f8ed9b\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8bae49e1ea7e4baaa9311082b2aa69cc : SUCCESS in 1h 20m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b637c4728714379ac5ace8510f26e11 : SUCCESS in 46m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476cd2f0addf4fc6b5c579d89903ff78 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572bcb9f48fa4298a4c26611901ba3d5 : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01d499702f61457997c50abf5fec76ce : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2be7df3d1a954f4dafbd36c54f21874a : SUCCESS in 12m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/600337a54c5445b688499fa6875dc4b0 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd3049314a14f18a6602c27d959738c : SUCCESS in 8m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c6df3850e1514b539f3c47cddd585b5d : SUCCESS in 1h 03m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a293a9a2cf44beb9a9f489c6f00713d : FAILURE in 1h 31m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b256865c465e423f9ed52c3ff04a49cc : SUCCESS in 46m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5ea73c2c6e3b4e33a8547aee3bfd17b1 : SUCCESS in 46m 24s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c98e8697f5f345c4b49bc8e700bb4d3b : SUCCESS in 1h 21m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/1de4b78e7ea94f5c9dc992ff4d11879a : FAILURE in 1h 34m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/783c7402e11145938895011ddb8f5422 : SUCCESS in 47m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ba1b65e1793b47129ce7f6035ff609bf : FAILURE in 4m 49s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f021da1e9fb7470496d2c5fbfa2f912c : SUCCESS in 22m 41s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da932c530b3a40ca8207dcc8337ec1f6 : SUCCESS in 22m 33s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e63a019c3854f0da24b0029be3aa37a : SUCCESS in 28m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/51f43392a7814217bb2fb906a2bd4eaf : SUCCESS in 1h 07m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52ac03b4bbf34405b0417550b7e163c6 : SUCCESS in 43m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/233a19131cd84ee8adf82591390ce48c : SUCCESS in 34m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/60cc1350273c49dbbbb65d4b7dbdf518 : SUCCESS in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1cefd8e02e6898df0511454d52f0bf5a9499e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-06 10:27:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"696572c476c87696570dae5e33dbe642db7c5775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 10:33:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc9a29441f247c28f4fa638631b88a9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba81152bab1040b2add53502bfc9d6b3 : FAILURE in 4m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7ecdaf280764fff87531d12f9d114ef : FAILURE in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef0c45b27bc58c985aac0e2596284a329299d4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 12:07:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/317c6c8772f4459da0c73e938d15a3c8\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/631fdb857170445cb5495a906623950a : SUCCESS in 1h 33m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf35db788e344d658aa2993e433d75b0 : SUCCESS in 53m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4209d111efc34a718a095f1960178bd1 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7732fb9cb046d3b2d4153c7b7faf53 : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d15882d48714de7abe65b4c8c41380f : SUCCESS in 13m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9223138925a44af8975be90e35896e2d : SUCCESS in 11m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/402bf60562214f7d8930d913218dde39 : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2e02759bb048379f4527d206c1ba5d : SUCCESS in 9m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7574e384ae9d4408b7dc7491d0edf099 : SUCCESS in 55m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b654e1911d94293b6a5cff1c9318f0c : SUCCESS in 1h 26m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/33d34122d6ee469a89d2acaf51396dab : SUCCESS in 55m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ea1e940295604ea2866291c5fae4ee45 : SUCCESS in 51m 03s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/05e33de0a50541a6b9bb3e6a5f6897c7 : SUCCESS in 1h 12m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/a44eb4ba9e0e47e3b713e8e88133d30a : FAILURE in 1h 24m 10s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e89717ff93da4064b29b6d112371a4f4 : SUCCESS in 52m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8369477aa2c744459cf7729d8db66306 : SUCCESS in 5m 00s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6175b00802b645a3b2363693dfe50b21 : SUCCESS in 24m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/493712dd68624bfbacb5554d522eb24f : SUCCESS in 25m 53s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3fec0c4762524e0c81aea1c41080f019 : SUCCESS in 24m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea174c92f9bc485d920234edff862e8f : SUCCESS in 1h 07m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53624cca9e434083bee3f9f8f6ad0aea : SUCCESS in 42m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1123c744991e428080f410faa663906c : SUCCESS in 26m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a5e1977f171f4cb4a218292e4ccdb521 : SUCCESS in 41m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b37356f7c47e76d8e685fa9e209c6389aece861","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-06 12:25:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d375d648bba8b9d7983945cae883c917535c271","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 12:30:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46a79af64404bfba6ba71ee65b39154\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6373a1822efb45579246ec82063a1685 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fccf02a4405455d90ec6b57c92a4eee : FAILURE in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e43dba6d839df66d1c69ccbff0cfb69235e453e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 14:09:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac66da43da32488f943985816c55ae63\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dba8e45293f14dde8957de979fd1db7f : SUCCESS in 1h 22m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf09935140e24c7b921d6cf525be29ea : SUCCESS in 49m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc9aef4d0388438aabe97522ed49f552 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa5e2047c6944639513c53380f683d0 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abeee125bb0a4f1aa1837854e257c399 : SUCCESS in 11m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3713f1c1877f4c439fc842997bc4936c : SUCCESS in 11m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7752fcb9ecc24612a7bc91e9131369b5 : SUCCESS in 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d56692ce4947fb91a6da068b26ce76 : SUCCESS in 8m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ae32714eeeb4ca99539d0ea137a54a4 : SUCCESS in 1h 02m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3966fcfe6a524d7e9e6714517ab68894 : SUCCESS in 1h 31m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3d1fc966b64c442582be0c4ac4d19c96 : SUCCESS in 50m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e6e2e364729a4fe8836248b8e9f8691a : SUCCESS in 34m 19s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/76e0d93f054344c69acef589d8c946d6 : SUCCESS in 1h 32m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/f8072088c21c4295aaca19ee3ec36d09 : FAILURE in 1h 08m 57s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/70d31f47dc0148d4ad41e6678939c6f5 : SUCCESS in 40m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/775791c240a54637838502af5828157f : SUCCESS in 7m 23s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d390e3ed2954007a5a3d4f470dd5ce2 : SUCCESS in 19m 36s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/14459274f01b427b9a50fe360be871ba : SUCCESS in 18m 26s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a34a975717e42cdab25c96135736b8f : SUCCESS in 20m 49s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d1032b5d06341ffb7c0ff5b6b2371d7 : SUCCESS in 1h 00m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a100ee43f3d34095b3dde422f5f31970 : SUCCESS in 41m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8ff6a6895d774810b447f6c3e52d4d44 : SUCCESS in 26m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d567d10429c44aaa93df26dca49b9c61 : SUCCESS in 43m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91206b43e0263c680e629d5d84febf3b1967f3c1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-07 12:52:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5518d576415d6b297ea7063392d6113fe335292","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-07 12:54:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0d6768ac0facb7ea4aa58c85c0505d5707ad21e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 13:31:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b872f41f9998416cbd64927be47b1fb0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c38597797ce4e3495006015d9119c6f : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbcdcbacf96c414c885e0e1496fe2d7c : SUCCESS in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"263f38a9fa70c9b1e6afa5c373c6b249f55ffd04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 14:37:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a7fbcdc5f3a42319058600719a00a98\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aafe349f6b954e378e5b0124d3bec65e : SUCCESS in 1h 22m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4417e0e27128432591574509ad574599 : SUCCESS in 51m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4629272dcdec4922bc6100ad846dec42 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d144c00aa2c241ccb5374bfec1130c94 : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a5713e988e4ed5ae56fdc35beabcce : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a764200bcd14b889c49eed5956eb60a : SUCCESS in 13m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8d2908672764896a469331ca2027bdf : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2603c1c8484f488eb3354a5faf2797ff : SUCCESS in 8m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3a1ec946867445479d9cc49f388e8447 : SUCCESS in 1h 06m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d39cfe7abfdf4c92b22ae26726974558 : SUCCESS in 1h 28m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ec31031ae73547b5b0eb0e03ba7afb73 : SUCCESS in 50m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a6e7b861fd914f37923bf3a020f67407 : SUCCESS in 49m 38s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/03ae4fbe55674d6bbbc0a768fc1964c4 : SUCCESS in 1h 38m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/bc06a339675148ad9a772d819e09cce4 : SUCCESS in 1h 35m 12s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/cd607195c7e4488793dcf8f596103147 : SUCCESS in 53m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/72dbcece5fd84ba2b209b319a46a0fdf : SUCCESS in 4m 53s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8139ff2d16c24809a60b5b2dcf1b442c : SUCCESS in 21m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aeb8e5e022374e53b0348ffe35fcae20 : SUCCESS in 24m 55s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40df9cb9171c4693b8c73cdc98d5856d : SUCCESS in 18m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7911acd7940b40ee890f716ea182bf6d : SUCCESS in 47m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8525b43d9df14f6cbd1c7304b4ccb739 : SUCCESS in 38m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/028db596c587485e9b081f84571b2ef3 : SUCCESS in 21m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3e5a4d4e76f247ea907fa78a72d0c2dc : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9801f75cb6319279a5e19c0d05e15646497afae7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-08 16:10:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d69c16aea65e1a44a5e51dadd5b2d973ebb3e2a3","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-08 16:29:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a516225a396cd371db2454b41d8cd39bf997ee88","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-09 07:33:55.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83a9b38652e06845184a716a584951ae767b86ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 07:34:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7527bbca0672cf862cb6771d2608f59d647e0928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:27:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/765da6d454ea4bf49c8e91b0fa1a3a44\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/71bec8fbf20b4cfaa7575de292a155bb : SUCCESS in 1h 16m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/702d5ee0991045e48d9bec6f3732ebc3 : SUCCESS in 55m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892f8988c6914e3286451729c481b340 : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2961219064471092597bf50fd29e6c : SUCCESS in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4eadc9de931454eab9a82ce3023812a : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2efbe7c636e4609aecca9972ddcdfe2 : SUCCESS in 9m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3a0358ca35594398b4f42c06b1e6cdd7 : SUCCESS in 54m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8a55759125f34e36b489bea3a84cd1f8 : SUCCESS in 46m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/75b0de0afa7942c183835470de395b12 : SUCCESS in 21m 04s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ab554dfcb664618bf8af92eb18bd01d : SUCCESS in 21m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/de3079056310482c936e27453a768653 : SUCCESS in 1h 11m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/47f6b6a82d90469797dca0b2f73f08f4 : FAILURE in 1h 36m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d9724877482a4fdfbddf42ffebd40c8b : SUCCESS in 5m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/50efe266e8a74421b947fdd5a2cd08a8 : SUCCESS in 1h 28m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3a770995fcf40268dc080c663b1389f : SUCCESS in 1h 01m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/797b4e91003c46458b66cfff5694303c : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c191a5c05ad1d33a7767cac75be96c298e5b5aee","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-09 09:49:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"117b97c88b0367dfd7147b238162be62a4d39032","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 10:29:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b2bc333aa3f4ed09ab3f4dc4e76edd3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/064cd16737ba485ca5cbfa0dcc031d5f : SUCCESS in 30m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7aa2614285534c09a1236b675d493752 : FAILURE in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9807f5f07a7433be4727e32fe3306c7340663133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 12:06:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41b5fa0072784703a51eed304c8df0ae\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/87703a4678614ca6b9b62f92b7dcf7ab : SUCCESS in 1h 24m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd66cac431e54125bfae99bdee3d636b : SUCCESS in 51m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a249e07e969f433c98d92d6288193d76 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9763bb016624807ad919220863ce6a8 : SUCCESS in 12m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46694853d33145cea8a98e31353a51a2 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ab8ba4739c442228489a37e5e8a2a8b : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cb32106db494bf3bd53b8cbd045c011 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2473814dee74faf9bdf391d7da3b1ec : SUCCESS in 9m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/46ae5fdfbc1d47a3aa765c7bcd6b78a3 : SUCCESS in 1h 02m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0d25c31916684989b940355a79d9e3af : SUCCESS in 1h 50m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b1b9f6d1ddb044b8a95bff3e61038763 : SUCCESS in 38m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cfb7e45720b1457f9342b7287f0a0f8e : SUCCESS in 42m 37s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3bd3c2348bee49fd8b3b8cc21c2653fe : SUCCESS in 1h 24m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/b1757bcc9a284484874b2f8cc90835d0 : SUCCESS in 1h 24m 23s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1c50a2ab08b246b1875194f9b67cfac7 : SUCCESS in 43m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/171982458da34371b4379b3d62c925b8 : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a092cf03b4854792bba38f048639bc5c : SUCCESS in 27m 17s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/787cce0b203949dd8b36969484345e98 : SUCCESS in 27m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/67539c14a0844dce8e2549cacb629236 : SUCCESS in 28m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6071d199ddad4ab8908ea7ba7904d133 : SUCCESS in 54m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/946443030df347c7a9a394b89c920268 : SUCCESS in 57m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ddd76242dc734066bb1139c6ff5f198c : SUCCESS in 34m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a24a87eb3e6462da8090f7059292585 : SUCCESS in 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"261c85a25c0f346fb18b0cfec4930d36fe67af61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 12:07:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"958d252e9c9ea1f35d7504568f75a86161a220c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 15:42:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14b83e1a3d6241618f42c74886bf9ace\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d31a043939db4d1e839ea87a76d7993b : SUCCESS in 1h 31m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a9234b989064f348f4826e9edfbfaab : SUCCESS in 50m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de329260fe314d9fb76e2334a6f410e5 : SUCCESS in 13m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6117d9ba304f4c56aadaa2da56586d96 : SUCCESS in 13m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a52020f3de6453f82c0e48a5c016f50 : SUCCESS in 13m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21882ef157c4c78b5d5dbff926f378b : SUCCESS in 11m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/89ca4f901a374b59bbdbf1c6c947c231 : SUCCESS in 42m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4a861c40e7454db1aaaff66a13fc446d : FAILURE in 53m 34s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50822a19479b46e79702a46d3d316698 : SUCCESS in 22m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/916b1a7b1bb44be2802305c2db5d2d5d : SUCCESS in 21m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f0601a870ba54d40a32c1b1ab42a4415 : FAILURE in 1h 37m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/b3f41bc5ec804f4898d4fa5a0a14221a : FAILURE in 1h 40m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/731c6f2aa6964e54a464fdda34eeaf8b : SUCCESS in 4m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3a8e0ea3edf44b2a844e5318bbc61f2d : SUCCESS in 1h 41m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/27e9fa70f5044ae8903372e2b45ea4f9 : SUCCESS in 1h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3edc8351092049be94f9992ab2a914b1 : SUCCESS in 40m 21s","accounts_in_message":[],"_revision_number":2},{"id":"574173709e9f7f225f0cc10881072664c645e820","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 05:41:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3a3e7152e5d901ef7e6f4dffb49fce91854295a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 06:24:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5650b35fceff45dab82bdcc7d0ace47d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e8a7edf8ea94a908a40ac22ca9f0ab4 : FAILURE in 41m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db3f6399735b4552a3e7a5d317d86e67 : SUCCESS in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae622ead4110bd9bb2b8d7ef05aae221d4b4dc3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 07:21:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a839f1daefd4cef90b68238564c6f39\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/05a44ca15658483bb18d860f97dd30a8 : SUCCESS in 1h 30m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37d16685e2304bf0978dd29b3a9480da : SUCCESS in 55m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e8034ebba6343318631ed38be339229 : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d88ebdb30f4f579a55ac4df209e3a8 : SUCCESS in 11m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/148310aa90c44fdca1079df62ca1af9d : SUCCESS in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dae8282ab394eb285cea85f2d75a63a : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/299e8b430e2a4cba8a35d608da31ce8c : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55d0cab2ab241dc81d15794dfcdc3a3 : SUCCESS in 9m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a91741aafbbc4528830cb70dce81e570 : SUCCESS in 52m 37s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b30985ea8964cfc999fee4d64219279 : SUCCESS in 1h 33m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4547b0efcf3546a6a67135b3b4767e09 : SUCCESS in 40m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/213343ae5a7745d7aa74bd1fa66a6cfe : SUCCESS in 42m 51s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c68aecd32b554608b46c360c18f2fd70 : SUCCESS in 1h 14m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/50e145635b104935acfd9930c4432b9c : SUCCESS in 1h 33m 19s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3581874a512b44599b114d065650622f : SUCCESS in 55m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6d7516748dc44475aa29b9e516819aa0 : SUCCESS in 4m 17s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0cbdae498ddd48819bd3091b1228541f : SUCCESS in 18m 55s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ebefad078b84bcca49763d4eeaf311e : SUCCESS in 22m 38s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c80b2218eca9456bba82ad981d650432 : SUCCESS in 22m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4db9b2e615f45d9ba541a37b0c29ec0 : SUCCESS in 1h 00m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97a48c118cd24c4bb9ed325683a68e5c : SUCCESS in 53m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2c88199808534df4b83cddf078c05f16 : SUCCESS in 35m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4f749cef2c0b42eeacf7f8c17803719b : SUCCESS in 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a617b02443225b2cc810728751fb5f8bac6cf712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 07:21:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"369a1200d1789687d47d6b49b0efed55cc8be6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 09:11:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90aead45a72847f7a6279e31bd4e07cc\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e5b456942efa4a948c9e0ff919dcb388 : SUCCESS in 1h 20m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c07f127f77484d7db1adaf98d8691efd : SUCCESS in 51m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452efb69a8d7428089681f95f5a38468 : SUCCESS in 12m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b73674d4b94709b7ac2ef4a17dc300 : SUCCESS in 11m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0edfc97b8a54c6f8127e815cd6c59a8 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0838d2096e384de4b78525fe14b65e1d : SUCCESS in 9m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ee47c7d7b95340039737fb2b728578e6 : SUCCESS in 40m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/47282517db3c4108bd3a5bd3cb79095c : SUCCESS in 1h 00m 52s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d235022d89f4a0498003ccf2048af85 : SUCCESS in 26m 49s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0cf21b885e114baa9654db43108c0cce : SUCCESS in 21m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/242eff79ca3c41ffa304b040e0569b01 : SUCCESS in 1h 28m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/b98c74575eac4c108bf2dcac56ef3ce8 : FAILURE in 1h 34m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1aa148bcc1114a0daa7511a35c72e4aa : SUCCESS in 4m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be7714dbd2fb44758688d7daa40f086e : SUCCESS in 1h 43m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6207f63ccff1420eb7995b47055e0e5f : SUCCESS in 57m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136d361223944df0bc66a5748c9a1643 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":2},{"id":"008ed5bd538fedc5188b686d53105fb1f3bc63b8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 09:26:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"527852709e44a6d18f37db6e1c6c6862e1c81242","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 09:41:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b263c97a518546afab5a35b9ad6ce316\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c058eb6db064f02a483ad54720f489b : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/179442ae1e2c43bd9bc3106e04d78aad : FAILURE in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38f43308fd9a8c87bab0549f8e38fcd7a1774724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 11:01:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9658602f28834ec99762f88e9fe8578f\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/73ffa5329d5d47dbb66aca6d15a1c12b : SUCCESS in 1h 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f56cee67ab50406693b66d2d45dfd9dd : SUCCESS in 49m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e39504102584d5fb3f29c6e4f2fb7ce : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51e5fc12d944035a43456586f98a269 : SUCCESS in 15m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fc98206fa5243f9bde19dac2f193eaf : SUCCESS in 13m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85292e19353848b0afaafa3fe66399dd : SUCCESS in 14m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfe991c29b6d4b3fa918d8c092d11527 : SUCCESS in 13m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11df11a889a840aeb76180af965497e5 : SUCCESS in 9m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/48736874be634689be760d2482fd6a29 : SUCCESS in 1h 02m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89fa119729ec4b62b4884a3f7bd4d987 : SUCCESS in 1h 29m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d312a495310e4638a0098ac220e5a005 : SUCCESS in 51m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/af861e7c22cc479f9430675505fdb44e : SUCCESS in 40m 53s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3afcc1f69aae4f3f948244e0d843a80e : SUCCESS in 1h 13m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/0ae2fa184606493a9ae41231c7cb2ca2 : FAILURE in 1h 23m 09s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/095ef43640e4499abd16b59a11841505 : SUCCESS in 47m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/afdac8a511194303b4355d737efa7016 : SUCCESS in 6m 42s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d88cd9fe7cd0480898e729918eeaad64 : SUCCESS in 27m 58s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8cb2bde09f24cc4aa72afc56d671f12 : SUCCESS in 27m 16s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d84db7ffa4e4d608879f53254fc4f7a : SUCCESS in 23m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/06d038c60ec24b99b28b7dc66b5b905b : SUCCESS in 1h 05m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a659820b53054349b801b050b6b220b4 : SUCCESS in 57m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3a45fe8c539c44efa809a02626da0d31 : SUCCESS in 38m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/55fd3ef451e246009211a57e264e751d : SUCCESS in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eadce553be4027ac2761171ea31f87c10c5c3f16","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 11:15:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25723541bca712c4bdbcd106faef100f32789e19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 11:22:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9e7cd2ea434ac7b0422dd6b52f4c5d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edd7cee6cbc243c38e31f967f5573c03 : FAILURE in 3m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f9f134a4fce44f3b6eba809c7cd2e17 : FAILURE in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32481b2b009da9307aaae7bc0d6b041d567aa6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 12:54:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e35a61db4c94a9c9a8b83700af16376\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/67a62550276849ca8fc53fc3d2a0ae8f : SUCCESS in 1h 19m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b12ee2e5ccab422483c0eae2d8e36520 : SUCCESS in 49m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c289859a2b9b403189a6085cad6d7670 : SUCCESS in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d74ce009ebd4b60bec2c324261ddc2b : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7b8a9514e7d43d8a145d7a494da94de : SUCCESS in 13m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7d6eb90cb343bc9a38016e331324f8 : SUCCESS in 13m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7756e7e43acd4b509714f12147188069 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a321923f644435aa7381626d7e2af9 : SUCCESS in 9m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a6762691f33c4669a57cfef5adfc66f6 : SUCCESS in 1h 08m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ac173ba57ff4f67a37bd96e16b67b30 : SUCCESS in 1h 22m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/81eb2a73a7b544d582f453b6a9e1250d : SUCCESS in 39m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/850fdcf3ecca41068e79b88abd8263b0 : SUCCESS in 54m 04s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b8bebc9281d14873abb0d86435797c21 : SUCCESS in 1h 26m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/c1412ddb4a7e44d2a6ed833b3981188a : FAILURE in 1h 31m 11s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a69667348f0a44bcbb0f0b21a04a5a05 : SUCCESS in 54m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/710a6e5adc4e4aedbc7dfbdb0509c0b4 : SUCCESS in 5m 16s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e236bf13bda4e3dbaa7212a39cc85f9 : SUCCESS in 22m 40s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07d7bfb3142145a1bc1177e3c10f6419 : SUCCESS in 23m 50s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96058b9bad2d45bc88260d06d76b40eb : SUCCESS in 23m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/89cb35d3cd554edd960f23041fe76113 : SUCCESS in 1h 13m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6de17c7afebf43a19f0e8cfc1a41fb3e : SUCCESS in 57m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dedf8c48fcd8470da9efa5501cc92a4c : SUCCESS in 23m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/542ca631bc704582be36b21117cc8854 : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85c0b60ab258fff5818c86fbe5377e4be5fa78b9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 20:48:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66b8ab1bb7a81d95c59a408a6c762f6a44ee0916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 21:25:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b98a764e1904c9e90e296904f18a886\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60e9dac08a224fd89376e836270a390c : SUCCESS in 26m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85f3519c46a54a3086135231bd00f1fa : SUCCESS in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bcb9fced5590d9cd5e96ffbfbdd72a435c0a18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 22:40:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38cb4b9a52624fdcb7edb69725c647bc\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5aea5e6fe17f446a81bc465c02c03dc2 : SUCCESS in 1h 24m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd4264e8e0dc4524b7835ec7e8cd3b0c : SUCCESS in 55m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b697da7db57244aab2f19660f32e7b7e : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121f660096e94d57992d52c694c1c166 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ad706e5f3744fbf9b2ae23281491181 : SUCCESS in 14m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1307930cbde54efb9874566a8a9a9eb7 : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb4131149ef241e9ae4085fbbe2bc715 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26977fd04cff421da8072f4efad3de42 : SUCCESS in 9m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0a1fbfd4a8cb4cc9b873fa96bacf1298 : SUCCESS in 1h 04m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/95382b4359e04612a09342ee6cb13d56 : SUCCESS in 1h 27m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b70ea1aaeb3442e98777266c66f995eb : SUCCESS in 43m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9f3821b0d1344b92994a651b8c4e52fb : SUCCESS in 52m 22s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9eb2f62593a54af08ae1f50cd7e73621 : SUCCESS in 1h 43m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/14861b691b704018b3720251e8a8c6c8 : SUCCESS in 1h 37m 58s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9e45b40910b54275bac4f8995c28c4c8 : SUCCESS in 50m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3b96acc4aaa44d258f149bf3b73ad3f5 : SUCCESS in 5m 26s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a97b56392dc2433984029e4ef1a56d91 : SUCCESS in 24m 45s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c209f9a20434e4a888034ec77e29c49 : SUCCESS in 19m 53s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ea063d572eb49cba1d5d177dee0f367 : SUCCESS in 26m 43s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbcd88db57664156a9276795ebbe164e : SUCCESS in 1h 13m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a69c655a4d145a6b96af3786d6ce267 : SUCCESS in 55m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e08b7a92b6cb4b2cbd9cd93353a15126 : SUCCESS in 26m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a485fd3b232e4f568bd48c5190a82db9 : SUCCESS in 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54b113defc8223f218c2f7412e75c00dfeb319a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 22:41:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b686ffe46d8624fe6b461b60e3b53811eb4bbcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 00:53:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb7fdbefc2a4d2c8608f10d9f673bbf\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d2834347447e40769e09aa6def82e63e : SUCCESS in 1h 28m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc7ffcd2394244a99ec0cb9c6f87050f : SUCCESS in 53m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2548a6177b345adace09ab02a3657f0 : SUCCESS in 13m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4570e0e225d46f68d666b2692a3bc35 : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c1e0cace27346af9a2538a1e1f4a167 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3f4815a4724a8aa601c4e6a60eb9d3 : SUCCESS in 8m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b053b5e9b76642a297637e4e95faa58d : SUCCESS in 44m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a483ed2f459442e384c4bbe4dc19fe11 : SUCCESS in 35m 48s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6c66d1e41eb7465c8356afb19fdadc67 : SUCCESS in 29m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d2d19f342a44053850c8c7f08e9300a : SUCCESS in 19m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/86d268d3be3446a6a9849668065856c5 : SUCCESS in 1h 20m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/a818178a7df148bbb304d8b4d129cdde : SUCCESS in 1h 35m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b85ebed966e24488ad8219916a2be962 : SUCCESS in 4m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/602e46fd70c047b98f9b278e52158ccb : SUCCESS in 1h 16m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aef11b58895a4ea5b50401bd2f325949 : SUCCESS in 47m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b28049ceb71409aac04a5a78d0669fa : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"759547dd590a8a969e193f0dc059834776923190","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 00:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"44a76b7d3bd9f859aa84cd44ed4d30c5f471e684","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 00:54:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87a85bdd9664c3da7d22c12d7532dd2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446f34cd4c3d4098ac8b61ec20d08f51 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da352edceb74dbd715268f94516503042b48cc90","revisions":{"fc79285b12f53ede5c6a20f0ab6b94514a80a0d0":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:36:51.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/98/923298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/923298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/923298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/923298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/923298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/923298/1"}}},"commit":{"parents":[{"commit":"dd6252dfffc19d260493fe7541fe3818ec964535","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd6252dfffc19d260493fe7541fe3818ec964535"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-17 14:06:13.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-02 14:28:22.000000000","tz":-240},"subject":"Check images with format_inspector for safety","message":"Check images with format_inspector for safety\n\nIt has been asserted that we should not be calling qemu-img info\non untrusted files. That means we need to know if they have a\nbacking_file, data_file or other unsafe configuration *before* we use\nqemu-img to probe or convert them.\n\nThis grafts glance\u0027s format_inspector module into nova/images so we\ncan use it to check the file early for safety. The expectation is that\nthis will be moved to oslo.utils (or something) later and thus we will\njust delete the file from nova and change our import when that happens.\n\nNOTE: This includes whitespace changes from the glance version of\nformat_inspector.py because of autopep8 demands.\n\nChange-Id: Iaefbe41b4c4bf0cf95d8f621653fdf65062aaa59\nCloses-Bug: #2059809\n(cherry picked from commit 182ba1067bcc4b5092d282e57317a652ac152886)\n(cherry picked from commit cb9806d8c440e6003ad48ba80d42e0b80700eac7)\n(cherry picked from commit a8aed2401689b3f0e8be1928d36a5126b54d15dd)\n(cherry picked from commit aee628be41dc0d6cb3d86ead1231d8db4433c5eb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc79285b12f53ede5c6a20f0ab6b94514a80a0d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc79285b12f53ede5c6a20f0ab6b94514a80a0d0"}]},"branch":"refs/heads/unmaintained/zed"},"da352edceb74dbd715268f94516503042b48cc90":{"kind":"REWORK","_number":2,"created":"2024-07-06 10:27:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/98/923298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/923298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/923298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/923298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/923298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/923298/2"}}},"commit":{"parents":[{"commit":"d69d441cf5d82f69d8ed7d555a6af73624866400","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d69d441cf5d82f69d8ed7d555a6af73624866400"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-17 14:06:13.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-07-06 09:53:39.000000000","tz":120},"subject":"Check images with format_inspector for safety","message":"Check images with format_inspector for safety\n\nIt has been asserted that we should not be calling qemu-img info\non untrusted files. That means we need to know if they have a\nbacking_file, data_file or other unsafe configuration *before* we use\nqemu-img to probe or convert them.\n\nThis grafts glance\u0027s format_inspector module into nova/images so we\ncan use it to check the file early for safety. The expectation is that\nthis will be moved to oslo.utils (or something) later and thus we will\njust delete the file from nova and change our import when that happens.\n\nNOTE: This includes whitespace changes from the glance version of\nformat_inspector.py because of autopep8 demands.\n\nChange-Id: Iaefbe41b4c4bf0cf95d8f621653fdf65062aaa59\nCloses-Bug: #2059809\n(cherry picked from commit 9cdce715945619fc851ab3f43c97fab4bae4e35a)\n(cherry picked from commit f07fa55fd86726eeafcd4c0c687bc49dd4df9f4c)\n(cherry picked from commit 0acf5ee7b5dfb6ff0f9a9745f5ad2a0ed2bf65bf)\n(cherry picked from commit 67e5376dd64407f5aaf1ea5f8c896e356064a2c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da352edceb74dbd715268f94516503042b48cc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da352edceb74dbd715268f94516503042b48cc90"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
