)]}'
{"id":"openstack%2Fnova~92443","triplet_id":"openstack%2Fnova~master~Ic280e569d963c25f424d5dbc329bcb08fdaba04c","project":"openstack/nova","branch":"master","topic":"bp/lvm-driver-for-shared-storage","hashtags":[],"change_id":"Ic280e569d963c25f424d5dbc329bcb08fdaba04c","subject":"LVM: Support a volume-group on shared storage","status":"ABANDONED","created":"2014-05-06 19:26:35.000000000","updated":"2015-05-13 17:19:29.000000000","total_comment_count":72,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"45807d78983bc726c3d50623306f0b2f0ed2ba7c","_number":92443,"virtual_id_number":92443,"owner":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":-1,"date":"2015-01-11 08:46:02.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"date":"2014-12-12 21:48:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2014-12-12 22:13:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2014-12-16 00:05:00.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2014-12-12 22:37:28.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"date":"2014-12-13 00:35:23.000000000","_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},{"date":"2015-05-13 17:19:29.000000000","_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},{"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2014-12-17 18:55:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":-2,"date":"2014-12-12 21:47:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2015-03-11 10:06:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":-1,"date":"2015-03-11 06:54:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"all":[{"value":-1,"date":"2015-01-13 21:42:05.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},{"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-05-06 22:40:41.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2014-05-08 18:19:54.000000000","updated_by":{"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"},"reviewer":{"_account_id":4159,"name":"Cory Stone","email":"corystone@gmail.com","username":"corystone"},"state":"REVIEWER"},{"updated":"2014-06-20 20:04:23.000000000","updated_by":{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},"reviewer":{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},"state":"REVIEWER"},{"updated":"2014-07-01 11:54:55.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2014-07-16 16:47:16.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2014-11-14 02:00:05.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-12-12 21:48:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-12 22:13:08.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-12 22:37:28.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-13 00:35:23.000000000","updated_by":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-16 00:05:00.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-12-17 18:55:08.000000000","updated_by":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"reviewer":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"state":"REVIEWER"},{"updated":"2015-01-11 08:46:02.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2015-03-11 06:54:44.000000000","updated_by":{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},"reviewer":{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},"state":"REVIEWER"},{"updated":"2015-03-11 10:06:58.000000000","updated_by":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"reviewer":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"state":"REVIEWER"},{"updated":"2015-05-13 17:19:29.000000000","updated_by":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"reviewer":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"state":"REVIEWER"}],"messages":[{"id":"e42b44b1f5e132d90ce1d27970684af3e5dce500","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-05-06 19:26:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7a7acc21438e7fdca7957dc874297d9d61bd91d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-06 19:26:37.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"d5ce003db2c9f2fbb588ebd7b5e6143b0e6c4d07","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-05-06 19:56:23.000000000","message":"Patch Set 1: Verified-1\n\nDatabase migration testing failed either due to migrations unable to be applied correctly or taking too long.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/cfd958a/131007_devstack_export.log : SUCCESS in 10m 15s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_mysql_user_001/1b7c475/user_001.log : WARNING - Migration 204-\u003e205 took too long in 29m 01s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/f6d289c/131007_devstack_export.log : SUCCESS in 10m 12s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_percona_user_001/4f3a0af/user_001.log : SUCCESS in 20m 57s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_mysql_user_002/3f57a3e/user_002.log : SUCCESS in 15m 05s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_percona_user_002/95c0523/user_002.log : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8ce81e2743042ada964f7ab706ad1eb70f658e07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-06 20:30:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/1/check/gate-nova-pep8/73a0104 : SUCCESS in 10m 07s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/1/check/gate-nova-docs/f4afd54/doc/build/html/ : SUCCESS in 9m 41s\n- gate-nova-python26 http://logs.openstack.org/43/92443/1/check/gate-nova-python26/d51c675 : SUCCESS in 22m 13s\n- gate-nova-python27 http://logs.openstack.org/43/92443/1/check/gate-nova-python27/c25ef11 : SUCCESS in 22m 45s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/1/check/check-tempest-dsvm-full/7ba33ef : SUCCESS in 59m 17s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/1/check/check-tempest-dsvm-neutron/f85eb32 : SUCCESS in 1h 03m 48s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/1/check/check-tempest-dsvm-neutron-heat-slow/8bf8bb3 : SUCCESS in 22m 55s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/1/check/check-grenade-dsvm/c560453 : SUCCESS in 36m 41s\n- check-grenade-dsvm-neutron http://logs.openstack.org/43/92443/1/check/check-grenade-dsvm-neutron/0042001 : FAILURE in 40m 13s (non-voting)\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/1/check/check-devstack-dsvm-cells/536aeaa : SUCCESS in 15m 06s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/1/check/check-tempest-dsvm-postgres-full/ef2d457 : SUCCESS in 1h 00m 00s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/1/check/gate-tempest-dsvm-large-ops/c5ea5fc : SUCCESS in 21m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/1/check/gate-tempest-dsvm-neutron-large-ops/a89ef21 : SUCCESS in 22m 16s\n- gate-nova-pylint http://logs.openstack.org/43/92443/1/check/gate-nova-pylint/97aa285 : SUCCESS in 16m 38s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/1/check/check-grenade-dsvm-partial-ncpu/cd668d0 : SUCCESS in 37m 17s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/1/check/check-tempest-dsvm-virtual-ironic-nv/52d6a59 : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59c80befce8868ef52b8ee9e7a4bad41f1d534ce","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-06 20:50:09.000000000","message":"Patch Set 1: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/1/6878/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":1},{"id":"451f0bd32176150eb8eea2f05358d9d718b9a87d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-05-06 21:01:10.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- gate-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/1/gate-ibm-tempest-dsvm-full/850ffa6 : FAILURE in 42m 49s\n- gate-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/1/gate-ibm-tempest-dsvm-postgres-full/99e4bfb : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d2fc78895c5b34d1363c5efe2e691b0deceba8e5","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-05-06 21:48:55.000000000","message":"Patch Set 1:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/92443/1/ : Test run failed. in 2h 21m 47s","accounts_in_message":[],"_revision_number":1},{"id":"65824f79f2f177b0ce91506ed923be134ce86649","author":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"date":"2014-05-06 22:40:41.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m kinda confused by this.  I had some questions/comments in the Cinder version of this as well (https://review.openstack.org/#/c/92479/).  There\u0027s the question of overlap with some of the work being done already and why not just use the FC connector etc.\n\nThe bigger concern here is I\u0027m not sure how you keep consistency here?  I suppose you could just try and rely on Nova not attaching the same LV to more than one instance at a time but I\u0027m not sure I see how that can be ensured here.  Ideally sure, you\u0027d want to have multi-attach but there needs to be some awareness built in to make that work properly.\n\nIt seems like leveraging Walt\u0027s work on multi-attach and adding an FC connector for LVM might be a better way to go here.  This implementation seems to just rely on all of the plumbing and connectivity being done by the admin outside of OpenStack and then *used*.  Maybe I\u0027m missing how this goes together.","accounts_in_message":[],"_revision_number":1},{"id":"3c89dff6d086e72baa070e8e7a58f469a357e513","author":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"date":"2014-05-06 23:36:35.000000000","message":"Patch Set 1:\n\nrecheck migrations","accounts_in_message":[],"_revision_number":1},{"id":"517acd43de97e7fc4c859e2cad4d95ddaaa4b991","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-05-07 00:24:21.000000000","message":"Patch Set 1: Verified+1\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/a2a8e48/131007_devstack_export.log : SUCCESS in 10m 38s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_mysql_user_001/ee264eb/user_001.log : SUCCESS in 22m 05s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/4a97624/131007_devstack_export.log : SUCCESS in 11m 06s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_percona_user_001/17c0e4c/user_001.log : SUCCESS in 20m 03s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_mysql_user_002/e6c6096/user_002.log : SUCCESS in 14m 03s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/92/92443/1/check/gate-real-db-upgrade_nova_percona_user_002/9d2eabd/user_002.log : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"148db6d10fd814c967a64bf58382f8c25b8a3a2b","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-05-07 19:14:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(8 comments)\n\nIrrespective of John G\u0027s comments about whether this is a tenable solution compared to the other possible solutions, I\u0027ve left some notes inline on things that should be cleaned up if this moves forward.\n\nBest,\n\n-jay","accounts_in_message":[],"_revision_number":1},{"id":"3f2447a7bc7227c54ba4792fb73346bf372b6319","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-05-07 22:41:34.000000000","message":"Patch Set 1:\n\nHi John Griffith-san,\n\nThank you for your comment. I commented inline.\n\n\u003e\u003eThis implementation seems to just rely on all of the plumbing and connectivity being done by \n\u003e\u003ethe admin outside of OpenStack and then *used*.\n\nYes. This is right.\nI think this comment is an important point of my suggestion and also the point\nis an advantage as well as a weak point.\n\nCurrently cinder drivers support limited model of storages, so if we don\u0027t have\nspecific vendor cinder driver(ex. HDS cinder driver), we can not use the storage\nfor cinder back-end.\nBut using my suggested cinder driver, cinder will be able to support any kinds\nof storages which can export shared volume without a vendor cinder driver because\nthe my suggested driver uses pre-configured VG. This is an one of benefit.\n\nAs for weak point, as you mentioned above, my suggested cinder driver needs\nadmin preparation of the plumbing and connectivity of storage volume\nbeforehand.\n\nCould you give me a comment for this point?\n\n\u003e\u003eI\u0027m kinda confused by this. I had some questions/comments in the Cinder version of this as\n\u003e\u003ewell (https://review.openstack.org/#/c/92479/). There\u0027s the question of overlap with some of\n\u003e. the work being done already and why not just use the FC connector etc.\n\nPlease see comments below.\n\n\u003e\u003eIdeally sure, you\u0027d want to have multi-attach but there needs to be some awareness built in \n\u003e\u003eto make that work properly.\n\nI\u0027m confusing about this comment.\nThe multi-attach support is not a target of my suggestion.\nMy suggested driver only attaches a LV to one instance.\nThis design is same as other cinder drivers.\n\n\u003e\u003eThe bigger concern here is I\u0027m not sure how you keep consistency here? I suppose you could\n\u003e\u003ejust try and rely on Nova not attaching the same LV to more than one instance at a time but\n\u003e\u003eI\u0027m not sure I see how that can be ensured here.\n\nLet me confirm a background of this question.\n\nDo you have a concern about a consistency of filesystem on LV(logical volume)?\nIf the LV attached to multiple instance at the same time and these instances\naccess to the filesystem(ext4, etc) on the LV, data on the filesystem becomes\ninconsistent.(To avoid this thing, we need to use shared filesystem such as NFS)\n\nBut as you mentioned above, current Nova implementation does not permit to\nattach the same LV to more than one instance. Also my suggested driver only\nattaches a LV to one instance.\n\nSo I suppose there is no consistency problem.\nIf I miss to understand your question, please let me know.\n\n\u003e\u003eIt seems like leveraging Walt\u0027s work on multi-attach and adding an FC connector\n\u003e\u003e for LVM might be a better way to go here. \n\nI may not understand correctly, but what your suggestion is, for example,\nStep1 Create a cinder volume from back-end storage and attaches the volume as \n      sharable to multiple computes.\nStep2 Create a VG on top of the cinder volume\nStep3 Create a LV from the VG(cinder volume on cinder volume)\nStep4 Attach a LV to instance using connector(FC?) for LVM\n\nI seems this requires new mechanism for cinder.\n\nRegards,\nMitsuhiro Tanino","accounts_in_message":[],"_revision_number":1},{"id":"e4382aab86be7fd0a68792a081b61367933ec68b","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-03 20:34:24.000000000","message":"Topic changed from bp/s to bp/lvm-driver-for-shared-storage","accounts_in_message":[],"_revision_number":1},{"id":"6b4364cc37668257c60b06fc3ab88aea7958bfd3","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-09 15:12:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"83e0e8ba0126c4d4651c1ede06d518488874fef2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-09 15:12:33.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"946c4e19c19a4ff1e02e628d209f5b7a2ccd839f","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-09 15:35:48.000000000","message":"Patch Set 2:\n\nHi Jay,\n\nThank you for your review. I commented inline.","accounts_in_message":[],"_revision_number":2},{"id":"35891ae0ec36c781247dbfc0ac6ae46307c85ef4","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-09 15:36:34.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e57979fb15ce06e8aeb763107e844bb4b8b4a505","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-06-09 16:19:29.000000000","message":"Patch Set 2: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/2/11158/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":2},{"id":"1371b3d22607378d59b56a168c4638533fb8834a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-09 16:49:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/2/check/gate-nova-pep8/10c0e8a : SUCCESS in 9m 48s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/2/check/gate-nova-docs/23fb8dc/doc/build/html/ : SUCCESS in 11m 47s\n- gate-nova-python26 http://logs.openstack.org/43/92443/2/check/gate-nova-python26/8d89578 : FAILURE in 23m 02s\n- gate-nova-python27 http://logs.openstack.org/43/92443/2/check/gate-nova-python27/7f67f54 : FAILURE in 13m 33s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/2/check/check-tempest-dsvm-full/0c5e556 : SUCCESS in 54m 36s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/2/check/check-tempest-dsvm-postgres-full/ee56363 : SUCCESS in 55m 13s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/2/check/check-tempest-dsvm-neutron/64dde7b : SUCCESS in 1h 01m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/2/check/check-tempest-dsvm-neutron-heat-slow/18e5398 : SUCCESS in 22m 00s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/2/check/check-grenade-dsvm/c64808d : SUCCESS in 46m 31s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/2/check/check-devstack-dsvm-cells/c69e31e : SUCCESS in 13m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/2/check/gate-tempest-dsvm-large-ops/31ccac5 : SUCCESS in 27m 10s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/2/check/gate-tempest-dsvm-neutron-large-ops/33856c7 : SUCCESS in 20m 31s\n- gate-nova-pylint http://logs.openstack.org/43/92443/2/check/gate-nova-pylint/dbb846b : SUCCESS in 17m 25s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/2/check/check-grenade-dsvm-partial-ncpu/748fb1b : SUCCESS in 45m 52s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/2/check/check-tempest-dsvm-virtual-ironic-nv/f052e9a : SUCCESS in 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"784b5ef16c2858321097990592a59a19d66409ab","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-06-09 16:53:58.000000000","message":"Patch Set 2: Verified+1\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/92/92443/2/check/gate-real-db-upgrade_nova_mysql_devstack_131007/d492d13/131007_devstack_export.log : SUCCESS in 11m 07s\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/92/92443/2/check/gate-real-db-upgrade_nova_mysql_user_001/ea0d8f6/user_001.log : SUCCESS in 20m 53s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/92/92443/2/check/gate-real-db-upgrade_nova_percona_devstack_131007/50439ad/131007_devstack_export.log : SUCCESS in 10m 43s\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/92/92443/2/check/gate-real-db-upgrade_nova_percona_user_001/76d47b0/user_001.log : SUCCESS in 19m 55s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/92/92443/2/check/gate-real-db-upgrade_nova_mysql_user_002/9578f67/user_002.log : SUCCESS in 15m 10s\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/92/92443/2/check/gate-real-db-upgrade_nova_percona_user_002/00ff956/user_002.log : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":2},{"id":"702db536e38c7356ed26c2da41cbea87627858c7","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-06-09 17:34:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nFew things more to address inline, thx for your patience!\n\nBest,\n\n-jay","accounts_in_message":[],"_revision_number":2},{"id":"7c5b86284a61400bb4e7291d2e5799eb4763919d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-06-09 18:03:25.000000000","message":"Patch Set 2:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/2/check-ibm-tempest-dsvm-full/10e709a : FAILURE in 46m 51s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/2/check-ibm-tempest-dsvm-postgres-full/4d6e21c : FAILURE in 1h 48m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c63fe711ef3d21f4770f09df3123189ba1cf2557","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-06-09 18:20:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/2/ : Looks good to me in 1h 43m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f850821b375fbec9dbad2b573299e037cc07cf93","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-06-09 19:59:33.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-python27: https://bugs.launchpad.net/bugs/1288178\n- gate-nova-python26: https://bugs.launchpad.net/bugs/1288178\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1288178\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"a8fd7fc57fb20b8f4ee72617f891ad9a43eae106","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-06-11 13:50:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3233d1e284bb7941e56e505002970db684b3bcc8","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-11 13:55:57.000000000","message":"Patch Set 2:\n\nHello, Jay, Daniel,\n\nThank you for your review. I will fix your indications.","accounts_in_message":[],"_revision_number":2},{"id":"13d6c94b28ab1d51d395d17cb9b0ae83f878810f","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-16 21:56:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b27e76643ab5611f9b242802aa2c8e59fdfb4326","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 21:56:11.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"76f4f5261530b12bf3f4f4ba73e4945f3473f107","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-16 21:59:16.000000000","message":"Topic changed from bp/s to bp/lvm-driver-for-shared-storage","accounts_in_message":[],"_revision_number":3},{"id":"4874652da0a51b829545c6a89c8fd4092fbae556","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-16 22:14:45.000000000","message":"Patch Set 2:\n\n(7 comments)\n\nHello, Jay, Daniel,\n\nThank you for your review. I commented inline and also posted v3 patch.","accounts_in_message":[],"_revision_number":2},{"id":"302772fa7c69fc4ca9f8d616786caef953523afa","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-06-16 22:40:06.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/3/check-ibm-tempest-dsvm-full/d8591e5 : FAILURE in 34m 16s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/3/check-ibm-tempest-dsvm-postgres-full/9253c08 : FAILURE in 26m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5e226e20943b59d7052fdea3b62208cdecf69187","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-06-16 23:06:00.000000000","message":"Patch Set 3: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/3/12497/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":3},{"id":"5d98f413ecbb2b3740fcce636e1b33e388e8e8cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 23:28:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/3/check/gate-nova-pep8/4ec325d : FAILURE in 12m 17s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/3/check/gate-nova-docs/e7aad24/doc/build/html/ : SUCCESS in 12m 41s\n- gate-nova-python26 http://logs.openstack.org/43/92443/3/check/gate-nova-python26/3a4a0de : FAILURE in 11m 57s\n- gate-nova-python27 http://logs.openstack.org/43/92443/3/check/gate-nova-python27/115b638 : FAILURE in 11m 05s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/3/check/check-tempest-dsvm-full/1e3678b : FAILURE in 27m 35s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/3/check/check-tempest-dsvm-postgres-full/6f4fb4e : FAILURE in 26m 52s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/3/check/check-tempest-dsvm-neutron/3aca62e : FAILURE in 30m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/3/check/check-tempest-dsvm-neutron-heat-slow/d1d2b26 : FAILURE in 15m 11s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/3/check/check-grenade-dsvm/263eb7d : FAILURE in 30m 43s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/3/check/check-devstack-dsvm-cells/80a6789 : FAILURE in 13m 34s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/3/check/gate-tempest-dsvm-large-ops/597130e : FAILURE in 32m 08s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/3/check/gate-tempest-dsvm-neutron-large-ops/4f623e3 : FAILURE in 24m 32s\n- gate-nova-pylint http://logs.openstack.org/43/92443/3/check/gate-nova-pylint/d4c7cda : SUCCESS in 16m 41s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/3/check/check-grenade-dsvm-partial-ncpu/2740d24 : SUCCESS in 40m 37s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/3/check/check-tempest-dsvm-virtual-ironic-nv/d46becf : FAILURE in 27m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"da96b9ddb60dfa019cf16dc042c8ac21fa1bc795","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-06-16 23:57:35.000000000","message":"Patch Set 3:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-nova-pep8: unrecognized error\n- check-tempest-dsvm-full: unrecognized error\n- gate-nova-python27: unrecognized error\n- check-devstack-dsvm-cells: https://bugs.launchpad.net/bugs/1101147\n- check-tempest-dsvm-postgres-full: unrecognized error\n- gate-nova-python26: unrecognized error\n- gate-tempest-dsvm-large-ops: https://bugs.launchpad.net/bugs/1324793\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1285323\n- gate-tempest-dsvm-neutron-large-ops: https://bugs.launchpad.net/bugs/1324793\n- check-tempest-dsvm-neutron: unrecognized error\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":3},{"id":"5a209d50ebcded1c4af28fde0e5d366252203c77","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-06-17 00:57:02.000000000","message":"Patch Set 3:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/92443/3/ : Test run failed. in 1h 51m 22s","accounts_in_message":[],"_revision_number":3},{"id":"80b46aeb4a0aa3463165e24399152ccad150b48c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-06-17 11:23:54.000000000","message":"Patch Set 3: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/3/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/dd4639f07e9f4be49f70a4ad561f68b8/131007_devstack_export.log : SUCCESS in 10m 59s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/3/check/real-db-upgrade_nova_mysql_user_001:th-mysql/294bb9d896bd4fda8fdd9c3a86fc6dd1/user_001.log : SUCCESS in 19m 59s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/3/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/d6a720dfc9054edd99d69831ff77b480/131007_devstack_export.log : SUCCESS in 11m 00s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/3/check/real-db-upgrade_nova_percona_user_001:th-percona/f3d7fe86f338426ba58bbeb8394ac3a6/user_001.log : SUCCESS in 18m 33s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/3/check/real-db-upgrade_nova_mysql_user_002:th-mysql/7bede1a7f1fb4557bf1850afc6355dd8/user_002.log : SUCCESS in 13m 49s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/3/check/real-db-upgrade_nova_percona_user_002:th-percona/0ceafb0dbd464dd18dff5510013b8f29/user_002.log : SUCCESS in 16m 32s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":3},{"id":"3980011be3d553af9da95a4b492a661f31ff16f5","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-18 22:28:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"caea0632a4886ad4fd06f3c5d8b43eb548f62d41","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-18 22:28:19.000000000","message":"Patch Set 4:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"bcc4a2b87f26e9da0f18316fd3547b1beb4f0e18","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-18 23:21:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/4/check/gate-nova-pep8/d69c7af : SUCCESS in 12m 14s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/4/check/gate-nova-docs/a9204e5/doc/build/html/ : SUCCESS in 10m 56s\n- gate-nova-python26 http://logs.openstack.org/43/92443/4/check/gate-nova-python26/16ca8ab : SUCCESS in 15m 46s\n- gate-nova-python27 http://logs.openstack.org/43/92443/4/check/gate-nova-python27/5ef72ea : SUCCESS in 21m 03s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/4/check/check-tempest-dsvm-full/6f222bd : SUCCESS in 41m 39s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/4/check/check-tempest-dsvm-postgres-full/5684bb0 : SUCCESS in 43m 30s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/4/check/check-tempest-dsvm-neutron/d642426 : SUCCESS in 52m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/4/check/check-tempest-dsvm-neutron-heat-slow/2cfca2a : SUCCESS in 21m 14s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/4/check/check-grenade-dsvm/c819afa : SUCCESS in 35m 58s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/4/check/check-devstack-dsvm-cells/a65cf15 : SUCCESS in 14m 08s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/4/check/gate-tempest-dsvm-large-ops/70928cb : SUCCESS in 27m 50s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/4/check/gate-tempest-dsvm-neutron-large-ops/0f81ba4 : SUCCESS in 25m 47s\n- gate-nova-pylint http://logs.openstack.org/43/92443/4/check/gate-nova-pylint/ab46493 : SUCCESS in 22m 55s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/4/check/check-grenade-dsvm-partial-ncpu/b9202c4 : SUCCESS in 36m 53s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/4/check/check-tempest-dsvm-virtual-ironic-nv/bd57c3c : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8bae78b4d4832e63d88ee7a21e738e80e71d6bc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-06-18 23:38:34.000000000","message":"Patch Set 4: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/4/13062/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":4},{"id":"c622dfe7a9db3c3d8ec8ee748f20e4c035d51d4a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-06-19 00:13:23.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/4/ : Looks good to me in 1h 44m 46s","accounts_in_message":[],"_revision_number":4},{"id":"5c29d4963626e15758be3f34e7fdb9fdbe0f6847","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-06-19 04:17:24.000000000","message":"Patch Set 4:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/4/check-ibm-tempest-dsvm-full/d8717a4 : SUCCESS in 2h 53m 59s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/4/check-ibm-tempest-dsvm-postgres-full/42cd0f2 : FAILURE in 2h 35m 38s","accounts_in_message":[],"_revision_number":4},{"id":"8d3228bace0d71ff66adf733f092038b8ac4176c","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-06-19 07:29:38.000000000","message":"Patch Set 4: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/4/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/11b7d911681f44158645a22330878b04/131007_devstack_export.log : SUCCESS in 12m 44s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/4/check/real-db-upgrade_nova_mysql_user_001:th-mysql/380b96d5d00244ca93c0d6d64ef2e84b/user_001.log : SUCCESS in 16m 46s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/4/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/6d89add5a9ea46d2b57b0ad6e2ded9a8/131007_devstack_export.log : SUCCESS in 11m 52s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/4/check/real-db-upgrade_nova_percona_user_001:th-percona/72ea644e28814c1a86e48efe68dfd576/user_001.log : SUCCESS in 18m 38s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/4/check/real-db-upgrade_nova_mysql_user_002:th-mysql/a45bfd59bede4cb19abfdcad609a2f58/user_002.log : SUCCESS in 14m 35s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/4/check/real-db-upgrade_nova_percona_user_002:th-percona/297b742e54fc40b8948ce31610def2c9/user_002.log : SUCCESS in 13m 52s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":4},{"id":"e077942c6dd3e482c31ebe3129445fc58b54cf84","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-06-19 10:04:36.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"850c65666bc1d389b423e9575b767dbd2ee6062f","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-19 23:44:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"089851a00e443847f9a5e85b4a1a2f02c387659c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-19 23:44:46.000000000","message":"Patch Set 5:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"6cb10c7ceee60b58b2eb348ee45cfb1088e944b9","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-19 23:53:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c730a02e6b4ef4264581d05c0e3db7ccd06ecfa0","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-06-20 00:05:33.000000000","message":"Patch Set 5: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b24ed56111604274a710eb717e431299/131007_devstack_export.log : SUCCESS in 11m 25s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/bb18935b39e04006abaaacedb0ea9754/user_001.log : SUCCESS in 20m 04s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/176a37129b7149078d659efdfbdca58f/131007_devstack_export.log : SUCCESS in 11m 22s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_percona_user_001:th-percona/351e25b64d374791a023b8ee85983a90/user_001.log : SUCCESS in 18m 19s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/4ea3de7acbcf458d91945227c462bc3a/user_002.log : SUCCESS in 14m 06s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_percona_user_002:th-percona/72ec331c524a4ffdaa5e912bc06e6cd1/user_002.log : SUCCESS in 15m 51s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":5},{"id":"cee86a7ccc50c2b8acf4d7502fa2df8d87676a9b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-20 00:37:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/5/check/gate-nova-pep8/24300cc : SUCCESS in 8m 41s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/5/check/gate-nova-docs/17252b9/doc/build/html/ : SUCCESS in 9m 08s\n- gate-nova-python26 http://logs.openstack.org/43/92443/5/check/gate-nova-python26/1141da2 : SUCCESS in 22m 00s\n- gate-nova-python27 http://logs.openstack.org/43/92443/5/check/gate-nova-python27/644da51 : SUCCESS in 19m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-full/94f4ea5 : FAILURE in 6m 28s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-postgres-full/67c335f : SUCCESS in 45m 41s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-neutron/01b8c71 : SUCCESS in 52m 35s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-neutron-heat-slow/de29c99 : SUCCESS in 21m 48s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/5/check/check-grenade-dsvm/d99e2af : SUCCESS in 41m 36s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/5/check/check-devstack-dsvm-cells/ced7473 : SUCCESS in 12m 28s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/5/check/gate-tempest-dsvm-large-ops/a2a2736 : SUCCESS in 29m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/5/check/gate-tempest-dsvm-neutron-large-ops/31fa804 : SUCCESS in 24m 49s\n- gate-nova-pylint http://logs.openstack.org/43/92443/5/check/gate-nova-pylint/e578807 : FAILURE in 22m 12s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/5/check/check-grenade-dsvm-partial-ncpu/c14d9ef : SUCCESS in 32m 09s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-virtual-ironic-nv/c2abe88 : SUCCESS in 25m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0e211819c13bc66fafa3c405352c7a61503190b8","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-06-20 00:38:01.000000000","message":"Patch Set 5:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1286635\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1286635\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":5},{"id":"fe35c1461be5533ddcf29e0565338b84fa84599f","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-06-20 00:43:51.000000000","message":"Patch Set 5: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/5/13265/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"9c69f6c6f44dc7a555fa9ad414cd5c9998ea74f7","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-06-20 01:00:29.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/5/ : Looks good to me in 1h 15m 24s","accounts_in_message":[],"_revision_number":5},{"id":"bba359e557125e471be6a4002f6bf80bc50bbf05","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-06-20 01:44:32.000000000","message":"Patch Set 5:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/5/check-ibm-tempest-dsvm-full/2cb13c0 : SUCCESS in 1h 25m 25s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/5/check-ibm-tempest-dsvm-postgres-full/efc54a8 : FAILURE in 1h 59m 21s","accounts_in_message":[],"_revision_number":5},{"id":"bc981431b1bba3f0c1d4aeb133b204e96256baea","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-20 03:38:06.000000000","message":"Patch Set 5:\n\nrecheck bug 1286635","accounts_in_message":[],"_revision_number":5},{"id":"6239cf47f1d4575ec9457cf8090ac176721a6909","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-20 03:38:08.000000000","message":"Patch Set 5: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"604e3ea209f854d8a081aab4842f2e91c561aebd","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-06-20 03:59:52.000000000","message":"Patch Set 5:\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/3931444ab99a432a9a1bbb390cd05e75/131007_devstack_export.log : SUCCESS in 11m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_mysql_user_001:th-mysql/eb7c1781a2e84ff984628ecc31e8e5a9/user_001.log : SUCCESS in 17m 39s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/a9e0cf7736834247b0cc583398de2ba5/131007_devstack_export.log : SUCCESS in 11m 52s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_percona_user_001:th-percona/ce67aa85fecb455788e6747918685d10/user_001.log : SUCCESS in 21m 05s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_mysql_user_002:th-mysql/6d470ef3d7634d0d8a92f3083d1e6b77/user_002.log : SUCCESS in 15m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/5/check/real-db-upgrade_nova_percona_user_002:th-percona/acb5041e709a45b0b2d8e1f111e38505/user_002.log : SUCCESS in 14m 01s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":5},{"id":"a077a22a044ecba2b16cc7da4008202c7502255e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-20 04:25:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/5/check/gate-nova-pep8/22404b1 : SUCCESS in 8m 46s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/5/check/gate-nova-docs/4d392ae/doc/build/html/ : SUCCESS in 12m 12s\n- gate-nova-python26 http://logs.openstack.org/43/92443/5/check/gate-nova-python26/7cb81de : SUCCESS in 16m 58s\n- gate-nova-python27 http://logs.openstack.org/43/92443/5/check/gate-nova-python27/d3f5585 : SUCCESS in 20m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-full/0546e5d : SUCCESS in 44m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-postgres-full/0c88ba9 : SUCCESS in 46m 50s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-neutron/6a0c9da : SUCCESS in 40m 00s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-neutron-heat-slow/b7542a2 : SUCCESS in 17m 44s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/5/check/check-grenade-dsvm/b0565ad : SUCCESS in 41m 00s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/5/check/check-devstack-dsvm-cells/d3f86a7 : SUCCESS in 12m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/5/check/gate-tempest-dsvm-large-ops/d571008 : SUCCESS in 30m 17s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/5/check/gate-tempest-dsvm-neutron-large-ops/e77044f : SUCCESS in 25m 27s\n- gate-nova-pylint http://logs.openstack.org/43/92443/5/check/gate-nova-pylint/6bf2fa2 : FAILURE in 16m 06s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/5/check/check-grenade-dsvm-partial-ncpu/cc14940 : SUCCESS in 31m 31s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/5/check/check-tempest-dsvm-virtual-ironic-nv/b84b99e : SUCCESS in 21m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"72ebd157648da7ae599daf4e1afb9a017b72a945","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-06-20 04:34:49.000000000","message":"Patch Set 5:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/5/13288/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":5},{"id":"c4133890435abe3aa50cb13a883c5c5e89071d5b","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-06-20 04:53:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/5/ : Looks good to me in 1h 14m 41s","accounts_in_message":[],"_revision_number":5},{"id":"44724889df8e6360513b80c2de66ed00b935e523","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-06-20 09:43:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"708e06f41f1f0edc3a48f1d84e8f13f4a31221eb","author":{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},"date":"2014-06-20 19:08:55.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)\n\nI found wrong exceptions used in several places in lvm.py. Please see inline comments.","accounts_in_message":[],"_revision_number":5},{"id":"db79686748badf48612714c91e414bae060d7a64","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-20 19:16:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3f0d37b06cc0d921e423f61e8e311671d3d198cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-20 19:16:59.000000000","message":"Patch Set 6:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"20e89200eb3380439e008472cc66f60de7b88784","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-20 19:29:34.000000000","message":"Patch Set 5:\n\n(6 comments)\n\nHello Tomoki, Daniel\n\nI got some indications about exception call at LVM APIs.\nI fixed these and also I added assertRaises() to test these exception automatically at test_lvm.py. Thank you for your reviewing.","accounts_in_message":[],"_revision_number":5},{"id":"71e84234448b9066ddc44968a4c1f2eea0e5f8e6","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-06-20 19:38:05.000000000","message":"Patch Set 6: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/6/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/88fc601234ec4a1eb486f4b511ed22f9/131007_devstack_export.log : SUCCESS in 12m 43s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/6/check/real-db-upgrade_nova_mysql_user_001:th-mysql/e511b18c596e4d18adc045784924ed19/user_001.log : SUCCESS in 17m 03s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/6/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/81e44c46d6b24548973f74c24b8f4834/131007_devstack_export.log : SUCCESS in 11m 19s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/6/check/real-db-upgrade_nova_percona_user_001:th-percona/c7da15f69a164c0093ebe5a75d916724/user_001.log : SUCCESS in 20m 22s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/6/check/real-db-upgrade_nova_mysql_user_002:th-mysql/8c3b8580e7a442a19f202e1f459b7e1f/user_002.log : SUCCESS in 13m 41s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/6/check/real-db-upgrade_nova_percona_user_002:th-percona/ca8eb9f262d540c2ad39588bef4ef286/user_002.log : SUCCESS in 13m 49s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":6},{"id":"42ff0cd25871ecdb3b835ad65941f06a2dd6bc82","author":{"_account_id":9176,"name":"Tomoki Sekiyama","email":"tomoki.sekiyama@gmail.com","username":"tsekiyama"},"date":"2014-06-20 20:04:23.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a67cdc10835f1990843bb8168f164425f515dc2d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-20 20:10:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/6/check/gate-nova-pep8/b3c6bcc : SUCCESS in 9m 20s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/6/check/gate-nova-docs/d6c49b3/doc/build/html/ : SUCCESS in 11m 32s\n- gate-nova-python26 http://logs.openstack.org/43/92443/6/check/gate-nova-python26/379071e : SUCCESS in 14m 05s\n- gate-nova-python27 http://logs.openstack.org/43/92443/6/check/gate-nova-python27/ef93ae5 : SUCCESS in 19m 21s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-full/ba7aea6 : SUCCESS in 37m 42s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-postgres-full/beffcb6 : SUCCESS in 44m 49s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-neutron/087e254 : SUCCESS in 52m 07s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-neutron-heat-slow/df3b687 : SUCCESS in 20m 29s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/43/92443/6/check/check-grenade-dsvm/9fb87fa : SUCCESS in 38m 07s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/6/check/check-devstack-dsvm-cells/5969a21 : SUCCESS in 14m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/6/check/gate-tempest-dsvm-large-ops/f025b33 : SUCCESS in 20m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/6/check/gate-tempest-dsvm-neutron-large-ops/04ad186 : SUCCESS in 24m 11s\n- gate-nova-pylint http://logs.openstack.org/43/92443/6/check/gate-nova-pylint/493aaca : SUCCESS in 14m 36s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/6/check/check-grenade-dsvm-partial-ncpu/3913ee7 : SUCCESS in 42m 21s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-virtual-ironic-nv/6f73b90 : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7a74a129eae0fa33ef7aad9fbe47698902315abc","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-06-20 20:17:12.000000000","message":"Patch Set 6: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/6/13400/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":6},{"id":"3a17e9683ac4891deeec4bd6e140f17674cf566f","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-06-20 20:37:54.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/6/ : Looks good to me in 1h 20m 36s","accounts_in_message":[],"_revision_number":6},{"id":"8366421b42fe5c99fe0516b53a217ec10a366ee2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-06-20 21:13:46.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/6/check-ibm-tempest-dsvm-full/af47b84 : SUCCESS in 1h 56m 08s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/6/check-ibm-tempest-dsvm-postgres-full/8fb2369 : SUCCESS in 1h 44m 53s","accounts_in_message":[],"_revision_number":6},{"id":"4d77623094315bbefebda58eee52819e56ab5a23","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-06-25 23:00:33.000000000","message":"Topic changed from bp/s to bp/lvm-driver-for-shared-storage","accounts_in_message":[],"_revision_number":6},{"id":"a3c74822611149895144f48492a80a2bf8bf92a2","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-07-01 04:10:46.000000000","message":"Patch Set 6:\n\nHi Daniel,\n\nIf you have a chance, could you review the code again?\n\nAlso here is a nova-spec for this proposal. I would appreciate if you would also review this. https://review.openstack.org/#/c/97602/3 \n\nRegards,\nMitsuhiro Tanino","accounts_in_message":[],"_revision_number":6},{"id":"4ff45d2d77377ede641e36c0d8e3584566bfd523","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-01 04:10:48.000000000","message":"Patch Set 6: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"24b96d8c7b508b4c1cfaecfec5d7d16881fedcb8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-01 05:20:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/6/check/gate-nova-pep8/9e587fa : SUCCESS in 13m 03s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/6/check/gate-nova-docs/dea4bc2/doc/build/html/ : SUCCESS in 12m 49s\n- gate-nova-python26 http://logs.openstack.org/43/92443/6/check/gate-nova-python26/37f1609 : SUCCESS in 21m 57s\n- gate-nova-python27 http://logs.openstack.org/43/92443/6/check/gate-nova-python27/2d4030f : SUCCESS in 18m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-full/62ce1b4 : SUCCESS in 43m 39s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-postgres-full/f5955db : SUCCESS in 44m 53s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-neutron/d70a177 : SUCCESS in 1h 09m 43s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-neutron-heat-slow/6e87a0c : SUCCESS in 19m 44s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/6/check/check-grenade-dsvm/a612b9e : SUCCESS in 43m 24s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/6/check/check-devstack-dsvm-cells/2754d1a : SUCCESS in 11m 38s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/6/check/gate-tempest-dsvm-large-ops/5f28c5b : SUCCESS in 30m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/6/check/gate-tempest-dsvm-neutron-large-ops/0f7dc93 : SUCCESS in 23m 29s\n- gate-nova-pylint http://logs.openstack.org/43/92443/6/check/gate-nova-pylint/acff362 : SUCCESS in 28m 50s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/6/check/check-grenade-dsvm-partial-ncpu/4ecf20a : SUCCESS in 36m 09s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-virtual-ironic-nv/3ed8196 : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a0bc6a2ab20f79955daa14dd50495557a6931f1a","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2014-07-03 11:29:57.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"92fcb4b2e49a03b8a5e23b35959d7f4d60f51d0d","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-07-03 17:41:44.000000000","message":"Patch Set 6: Code-Review-1\n\n(10 comments)\n\nMitsuhiro-san, please see some inline requests.\n\nThanks!\n\n-jay","accounts_in_message":[],"_revision_number":6},{"id":"203e0c98fedff8ac0fbc3c3d5a67c6d3ce12830b","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-07-08 21:35:23.000000000","message":"Patch Set 6:\n\nHi Jay-san,\n\nThank you for your review. I will fix your indications.","accounts_in_message":[],"_revision_number":6},{"id":"18588b2531c0cc156894c91e0d33628eed04a483","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 21:35:26.000000000","message":"Patch Set 6: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"df3815e717986d0bc2ebd7921ecbfa1d20c6cc08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-07-08 22:25:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/6/check/gate-nova-pep8/381ab7b : SUCCESS in 11m 46s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/6/check/gate-nova-docs/3c23b64/doc/build/html/ : SUCCESS in 11m 00s\n- gate-nova-python26 http://logs.openstack.org/43/92443/6/check/gate-nova-python26/0f65cc3 : SUCCESS in 22m 55s\n- gate-nova-python27 http://logs.openstack.org/43/92443/6/check/gate-nova-python27/e43f91a : SUCCESS in 14m 53s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-full/686fe04 : FAILURE in 37m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-postgres-full/43ba88e : SUCCESS in 37m 24s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-neutron/d538358 : SUCCESS in 45m 52s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-neutron-heat-slow/7697af7 : SUCCESS in 26m 36s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/6/check/check-grenade-dsvm/3ecd9b8 : SUCCESS in 41m 39s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/6/check/check-devstack-dsvm-cells/dbca95a : SUCCESS in 18m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/6/check/gate-tempest-dsvm-large-ops/020fd49 : SUCCESS in 22m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/6/check/gate-tempest-dsvm-neutron-large-ops/94dc70e : SUCCESS in 24m 13s\n- gate-nova-pylint http://logs.openstack.org/43/92443/6/check/gate-nova-pylint/b4efd5e : SUCCESS in 21m 59s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/6/check/check-grenade-dsvm-partial-ncpu/b6b01f3 : SUCCESS in 39m 53s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/6/check/check-tempest-dsvm-virtual-ironic-nv/3abae73 : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"809d2445476886173efbed2a7bd92cc8f2287a86","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-07-16 16:47:16.000000000","message":"Patch Set 6: Code-Review-2\n\nPlease work on getting the blueprint approved. The blueprint has to be approved before we can merge this code.\n\nHowever we have now closed blueprint submissions for Juno, so please resubmit for K, or work on getting an exception for K.","accounts_in_message":[],"_revision_number":6},{"id":"4854e8cedf514bee9a57c3068ee71fe49bd87c4a","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-07-16 16:54:28.000000000","message":"Patch Set 6:\n\nHi John,\n\nI\u0027m working on getting the blueprint approved again.\nSo I would like to get an exception for Juno release.\nI will join nova meeting tomorrow.","accounts_in_message":[],"_revision_number":6},{"id":"94ae8f3d359fafa6807216c6c588070b55d917ec","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-08-08 16:45:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"47af92d7122a6aa8cb72e0ff6c57cb260252efbb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-08 16:45:24.000000000","message":"Patch Set 7:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"9d88c750c43bb71985758cc86a3c1b700e70cb6a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-08 18:54:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/7/ : Looks good to me in 1h 33m 20s","accounts_in_message":[],"_revision_number":7},{"id":"4459a447bd6b1b9316a4a2e3ac2ca90dde08ecc3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-08 19:37:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/7/check/gate-nova-pep8/028164c : SUCCESS in 8m 20s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/7/check/gate-nova-docs/3f9fb87/doc/build/html/ : SUCCESS in 24m 06s\n- gate-nova-python26 http://logs.openstack.org/43/92443/7/check/gate-nova-python26/91074ad : SUCCESS in 22m 46s\n- gate-nova-python27 http://logs.openstack.org/43/92443/7/check/gate-nova-python27/8e1b6dc : SUCCESS in 20m 20s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/7/check/check-tempest-dsvm-full/242ec48 : SUCCESS in 41m 24s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/7/check/check-tempest-dsvm-postgres-full/99c6f5b : SUCCESS in 41m 23s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/7/check/check-tempest-dsvm-neutron/4685182 : SUCCESS in 55m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/7/check/check-tempest-dsvm-neutron-heat-slow/24fe6f4 : SUCCESS in 24m 21s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/7/check/check-grenade-dsvm/acf9a24 : SUCCESS in 46m 05s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/7/check/gate-tempest-dsvm-large-ops/64c3db3 : SUCCESS in 21m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/7/check/gate-tempest-dsvm-neutron-large-ops/d156759 : SUCCESS in 17m 05s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/7/check/check-devstack-dsvm-cells/d287053 : SUCCESS in 14m 40s\n- gate-nova-pylint http://logs.openstack.org/43/92443/7/check/gate-nova-pylint/4fb43a4 : SUCCESS in 16m 42s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/7/check/check-grenade-dsvm-partial-ncpu/7876213 : SUCCESS in 46m 08s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/7/check/check-tempest-dsvm-virtual-ironic-nv/b45681a : SUCCESS in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d5a7b7bbe7830fb8307a9c0b2fbbc8e58fe3cfb5","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-08-08 19:38:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"db668ff02aa1b8c8934dab068549a39b118fa014","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-08 19:38:23.000000000","message":"Patch Set 8:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"e128afaca9baa56722957245b9dcc31f97c89f40","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-08-08 19:57:48.000000000","message":"Patch Set 6:\n\n(9 comments)\n\nHi Jay,\n\nThank you for your review and sorry for my late response. Please see my comments inline.","accounts_in_message":[],"_revision_number":6},{"id":"587a32cbee2bd0083b8c2f2be76813401731a576","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-08-08 20:06:02.000000000","message":"Patch Set 8: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/8/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/558b7de101364b2e993f0e05e5d76805/131007_devstack_export.log : SUCCESS in 13m 40s (non-voting)\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/8/check/real-db-upgrade_nova_mysql_user_001:th-mysql/8b4432f7ae364c90a3a6b18cb87b3bcb/user_001.log : SUCCESS in 26m 47s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/8/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/ecebbd4a1a554bfd80861b53218068d0/131007_devstack_export.log : SUCCESS in 14m 34s (non-voting)\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/8/check/real-db-upgrade_nova_percona_user_001:th-percona/ec5edf6ab17e402aa55538429275d3bb/user_001.log : SUCCESS in 23m 34s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/8/check/real-db-upgrade_nova_mysql_user_002:th-mysql/db0892d23f774b3eac2087bd99eb47bf/user_002.log : SUCCESS in 16m 17s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/8/check/real-db-upgrade_nova_percona_user_002:th-percona/7b7d9104c64249cb92ea4be4bf862dc1/user_002.log : SUCCESS in 19m 48s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/Nova/Turbo-Hipster","accounts_in_message":[],"_revision_number":8},{"id":"22846395f4bd5cd03ad57ddc0c6487ee558b8d3c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-08 21:23:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/8/check/gate-nova-pep8/cfa71a6 : SUCCESS in 8m 40s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/8/check/gate-nova-docs/9d57630/doc/build/html/ : SUCCESS in 16m 44s\n- gate-nova-python26 http://logs.openstack.org/43/92443/8/check/gate-nova-python26/c3a6a0e : FAILURE in 18m 13s\n- gate-nova-python27 http://logs.openstack.org/43/92443/8/check/gate-nova-python27/ab0c056 : SUCCESS in 13m 26s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/8/check/check-tempest-dsvm-full/7ec7869 : SUCCESS in 46m 10s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/8/check/check-tempest-dsvm-postgres-full/2a64254 : SUCCESS in 42m 31s\n- check-tempest-dsvm-neutron http://logs.openstack.org/43/92443/8/check/check-tempest-dsvm-neutron/ca3ca56 : SUCCESS in 51m 02s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/8/check/check-tempest-dsvm-neutron-heat-slow/1dc72f3 : SUCCESS in 21m 56s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/8/check/check-grenade-dsvm/92f8f71 : SUCCESS in 38m 21s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/8/check/gate-tempest-dsvm-large-ops/3fa8183 : SUCCESS in 20m 09s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/8/check/gate-tempest-dsvm-neutron-large-ops/aab0129 : SUCCESS in 19m 38s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/8/check/check-devstack-dsvm-cells/ba7b138 : SUCCESS in 14m 02s\n- gate-nova-pylint http://logs.openstack.org/43/92443/8/check/gate-nova-pylint/d5129a4 : SUCCESS in 22m 22s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/8/check/check-grenade-dsvm-partial-ncpu/38c6951 : SUCCESS in 43m 24s\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/43/92443/8/check/check-tempest-dsvm-virtual-ironic-nv/eedfbe0 : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ec982ff3101dee3c1235022a74212db7f5fe6e7a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-08-09 00:22:42.000000000","message":"Patch Set 8:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with recheck hyper-v\n\n- http://64.119.130.115/92443/8/ : Test run failed. in 4h 07m 05s","accounts_in_message":[],"_revision_number":8},{"id":"3f295a4ee972418c62ceb9685f758e004923fc75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-08-10 01:02:40.000000000","message":"Patch Set 8:\n\nBuild failed. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For debugging, check https://wiki.openstack.org/wiki/PowerKVM#Debugging_PowerKVM_Failures\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/8/check-ibm-tempest-dsvm-full/4f59ffc : FAILURE in 2h 53m 05s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/8/check-ibm-tempest-dsvm-postgres-full/9e4abb2 : FAILURE in 3h 05m 11s","accounts_in_message":[],"_revision_number":8},{"id":"3cf15df0e3a9da19447f1e7f413fc4504a396bf3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-08-11 18:02:36.000000000","message":"Patch Set 8: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/8/21185/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":8},{"id":"a91c72cc6a5adc722d0c69f3cb232ed28d20ff2e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-08-12 22:05:50.000000000","message":"Patch Set 8:\n\nrecheck-vmware","accounts_in_message":[],"_revision_number":8},{"id":"94b42f35e265186e789f04fee2c9c7a9ba5c94f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-12 22:06:32.000000000","message":"Patch Set 8: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"bf74c32cbee356fa91b750170050a639eafb18dd","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-11-13 16:51:43.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"f9823742c3c86dc189203538624ad932091c0829","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-11-14 00:51:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b2a78caaa413a30d3374b1ed730e99cd6ac6157c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-14 00:51:55.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"7ea87e392f40a7ffa2190aefafdca167955f8100","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-14 00:52:09.000000000","message":"Patch Set 9:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"ec43fef8f4cc5b864b876bf57cd50e51db5a1b61","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-11-14 00:52:24.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"f15f2f87c0ba4aa6d7a91569a137e7b32c71ffbb","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-11-14 00:53:41.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b4dcfe45f240b21efcbb4da56d459434e582272a","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-11-14 00:54:15.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"4ca10bf0e09a006e0e74c911ea7d0d5e7682bf43","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-14 01:58:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/10/check/gate-nova-pep8/c4f38f5 : SUCCESS in 11m 41s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/10/check/gate-nova-docs/f016014/doc/build/html/ : SUCCESS in 8m 34s\n- gate-nova-python27 http://logs.openstack.org/43/92443/10/check/gate-nova-python27/452f612 : SUCCESS in 18m 53s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-full/f080998 : FAILURE in 47m 32s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-postgres-full/f513e3d : SUCCESS in 52m 00s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-neutron-full/b8e17f1 : SUCCESS in 1h 04m 58s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-neutron-heat-slow/ea46502 : SUCCESS in 17m 19s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/10/check/check-grenade-dsvm/960f5c7 : SUCCESS in 53m 49s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/10/check/gate-tempest-dsvm-large-ops/dcdb836 : SUCCESS in 18m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/10/check/gate-tempest-dsvm-neutron-large-ops/c4542a5 : SUCCESS in 22m 13s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/10/check/check-devstack-dsvm-cells/5b376d7 : SUCCESS in 12m 09s\n- gate-nova-pylint http://logs.openstack.org/43/92443/10/check/gate-nova-pylint/c27c48e : FAILURE in 20m 25s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/43/92443/10/check/gate-nova-tox-functional/e3e268d : FAILURE in 8m 19s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/10/check/check-grenade-dsvm-partial-ncpu/7e4c331 : SUCCESS in 43m 37s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-ironic-pxe_ssh-nv/5f67c3c : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":10},{"id":"e50af9424999e66754b1209920df5fd92bd5d101","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-11-14 02:00:05.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1370496\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"a24d0cc33391df8610bbdf83a957edbb35ed088a","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-14 02:06:35.000000000","message":"Patch Set 10:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/92443/10/ : Test run failed. in 1h 12m 27s","accounts_in_message":[],"_revision_number":10},{"id":"92e304c91c2f2bff9598f6d3e3f7e7ef99576d5b","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-11-14 02:53:24.000000000","message":"Patch Set 10:\n\nrecheck bug 1370496","accounts_in_message":[],"_revision_number":10},{"id":"b5b23f1f378e891d2c2853a69541f7fe5e7844e6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-11-14 03:53:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/10/check/gate-nova-pep8/8c1553e : SUCCESS in 8m 13s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/10/check/gate-nova-docs/80b6faa/doc/build/html/ : SUCCESS in 8m 34s\n- gate-nova-python27 http://logs.openstack.org/43/92443/10/check/gate-nova-python27/acc9136 : SUCCESS in 13m 37s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-full/6d9fe41 : SUCCESS in 51m 37s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-postgres-full/cb7cc74 : SUCCESS in 56m 30s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-neutron-full/5d36329 : SUCCESS in 1h 00m 07s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-neutron-heat-slow/d2752c9 : SUCCESS in 19m 31s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/10/check/check-grenade-dsvm/af50266 : SUCCESS in 47m 41s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/10/check/gate-tempest-dsvm-large-ops/a317ea3 : SUCCESS in 17m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/10/check/gate-tempest-dsvm-neutron-large-ops/ecdf96d : SUCCESS in 18m 26s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/10/check/check-devstack-dsvm-cells/e377be4 : SUCCESS in 12m 38s\n- gate-nova-pylint http://logs.openstack.org/43/92443/10/check/gate-nova-pylint/6cbe6ff : FAILURE in 17m 59s (non-voting)\n- gate-nova-tox-functional http://logs.openstack.org/43/92443/10/check/gate-nova-tox-functional/e3b99bc : FAILURE in 8m 15s (non-voting)\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/10/check/check-grenade-dsvm-partial-ncpu/6259715 : SUCCESS in 43m 22s\n- check-tempest-dsvm-ironic-pxe_ssh-nv http://logs.openstack.org/43/92443/10/check/check-tempest-dsvm-ironic-pxe_ssh-nv/7058b06 : SUCCESS in 27m 21s","accounts_in_message":[],"_revision_number":10},{"id":"5b9ee1e48012750175df160aec3ee54c93a6d536","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-11-14 04:56:05.000000000","message":"Patch Set 10:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/92443/10/ : Test run failed. in 2h 02m 21s","accounts_in_message":[],"_revision_number":10},{"id":"d9721e9e71e360561c1e8a1adfe14210fd5d1b06","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-11-14 05:02:27.000000000","message":"Patch Set 10: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/10/31507/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":10},{"id":"ad3a64e8db0807a59794908ed2db225cb0c426c4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-11-14 15:47:41.000000000","message":"Patch Set 10:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM.\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/10/check-ibm-tempest-dsvm-full/5e99648 : SUCCESS in 58m 34s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/10/check-ibm-tempest-dsvm-postgres-full/bde2ad4 : SUCCESS in 43m 53s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d8519c8c5e5958a405e8affaded04876fa065473","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-11-14 20:42:57.000000000","message":"Patch Set 10: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/10/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/b5d37fb3861b4265aa141498c2bd7044/131007_devstack_export.log : SUCCESS in 22m 28s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/10/check/real-db-upgrade_nova_mysql_user_001:th-mysql/7384d833944245e89ceafcb4f002c67c/user_001.log : SUCCESS in 22m 12s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/10/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/1ccb73322b3a47f9b17ddcbd9c9cc600/131007_devstack_export.log : SUCCESS in 22m 46s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/10/check/real-db-upgrade_nova_percona_user_001:th-percona/d0bc544533144a86b0d1f354a3a32f8e/user_001.log : SUCCESS in 25m 05s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/10/check/real-db-upgrade_nova_mysql_user_002:th-mysql/64ca619d91914ab68092c0e17f8aac45/user_002.log : SUCCESS in 20m 24s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/10/check/real-db-upgrade_nova_percona_user_002:th-percona/c0d38a1e24d94005a9ad0b55811a1ead/user_002.log : SUCCESS in 23m 28s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":10},{"id":"db0250599ce52bb3266b3de7082979e70fea10bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-11-28 20:09:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/92443/10","accounts_in_message":[],"_revision_number":10},{"id":"a9cc177d3a7083f0c5997141ba17788d11940d66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-08 13:07:52.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"85c883fb8d26b0383da235e54a2e38361671563c","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-12-12 01:21:07.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"869dca7da64c1ae859f20204ecf697442a098a09","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-12 01:44:46.000000000","message":"Patch Set 11: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/11/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/8281cc509606467b86dc9c9623f01c7c/index.html : SUCCESS in 19m 17s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/11/check/real-db-upgrade_nova_mysql_user_001:th-mysql/0e65c9eaa6ac431bb35014d121151694/index.html : SUCCESS in 18m 22s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/11/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/cd4d854cbc25410880402888aaf48b8d/index.html : SUCCESS in 20m 12s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/11/check/real-db-upgrade_nova_percona_user_001:th-percona/b349e5e783a64f8d8eab06a2da9a7913/index.html : SUCCESS in 22m 30s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/11/check/real-db-upgrade_nova_mysql_user_002:th-mysql/c0662be07be04f3299054413a7017279/index.html : SUCCESS in 19m 40s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/11/check/real-db-upgrade_nova_percona_user_002:th-percona/4b2e1c6785884cebb61e32e243a21483/index.html : SUCCESS in 16m 45s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":11},{"id":"4abbba9fe89c159525a63b6d2dc10149cc2d6cac","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-12-12 01:57:43.000000000","message":"Patch Set 11:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/11/check-ibm-tempest-dsvm-full/6fe5dfe : SUCCESS in 36m 05s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/11/check-ibm-tempest-dsvm-postgres-full/336b45b : SUCCESS in 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"1e90d003b255329702e656ad08130be505491995","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-12 01:58:58.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/11/check/gate-nova-pep8/7f27c60 : SUCCESS in 7m 40s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/11/check/gate-nova-docs/8c89ccf/doc/build/html/ : SUCCESS in 6m 43s\n- gate-nova-python27 http://logs.openstack.org/43/92443/11/check/gate-nova-python27/1d2c513 : SUCCESS in 14m 58s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/11/check/check-tempest-dsvm-full/bba05e4 : SUCCESS in 35m 12s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/11/check/check-tempest-dsvm-postgres-full/dc43dc4 : SUCCESS in 35m 24s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/43/92443/11/check/check-tempest-dsvm-neutron-full/5584163 : SUCCESS in 37m 29s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/11/check/check-grenade-dsvm/350419e : SUCCESS in 25m 26s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/11/check/gate-tempest-dsvm-large-ops/6164f3e : SUCCESS in 20m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/11/check/gate-tempest-dsvm-neutron-large-ops/97ad75b : SUCCESS in 23m 51s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/11/check/check-devstack-dsvm-cells/c702565 : SUCCESS in 15m 31s\n- gate-nova-tox-functional http://logs.openstack.org/43/92443/11/check/gate-nova-tox-functional/2364475 : SUCCESS in 19m 16s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/11/check/check-grenade-dsvm-partial-ncpu/0eaddef : SUCCESS in 27m 40s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/43/92443/11/check/check-tempest-dsvm-ironic-pxe_ssh/8e57f07 : SUCCESS in 26m 37s\n- check-tempest-dsvm-docker http://logs.openstack.org/43/92443/11/check/check-tempest-dsvm-docker/27ece75 : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c01dc9e5a6a580cc38b506ae1091caf9791bcc77","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-12 02:13:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded.\n\n- http://64.119.130.115/92443/11/ : Looks good to me in 51m 54s","accounts_in_message":[],"_revision_number":11},{"id":"4383c7d0b0fd81ec7e6f22ec02d841daabd7e9ff","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-12-12 18:38:31.000000000","message":"Patch Set 11: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/11/2344/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":11},{"id":"866cde7cd6adeba5c6cdc9ad960f516ec2511d2a","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2014-12-12 21:47:53.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a95d649f174c9440f286f74181769e0e63b08742","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-12-12 21:48:19.000000000","message":"Patch Set 12:\n\nTesting completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with recheck-pkvm. Contact info: kvmpower@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/12/check-ibm-tempest-dsvm-full/f4372da : FAILURE in 1s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/92443/12/check-ibm-tempest-dsvm-postgres-full/3834853 : FAILURE in 2s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"73f7c46857f8b0980d34c1c723b56a38086d0c87","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-12-12 22:13:08.000000000","message":"Patch Set 12: Verified+1\n\nDatabase migration testing successful.\n\n- real-db-upgrade_nova_mysql_devstack_131007:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/12/check/real-db-upgrade_nova_mysql_devstack_131007:th-mysql/16e4c4fb458743b8b6d207fcb80a582c/index.html : SUCCESS in 18m 36s\n- real-db-upgrade_nova_mysql_user_001:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/12/check/real-db-upgrade_nova_mysql_user_001:th-mysql/09db0a1ca8a24a28bf12acacb1e2175b/index.html : SUCCESS in 19m 26s\n- real-db-upgrade_nova_percona_devstack_131007:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/12/check/real-db-upgrade_nova_percona_devstack_131007:th-percona/5b82553a7af04c66b831f849ee675671/index.html : SUCCESS in 17m 47s\n- real-db-upgrade_nova_percona_user_001:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/12/check/real-db-upgrade_nova_percona_user_001:th-percona/787b0c40d4d54fcf976471065944f1d2/index.html : SUCCESS in 24m 08s\n- real-db-upgrade_nova_mysql_user_002:th-mysql http://www.rcbops.com/turbo_hipster/results/43/92443/12/check/real-db-upgrade_nova_mysql_user_002:th-mysql/d355c7701cda4b5c84e8780b04efdd79/index.html : SUCCESS in 19m 13s\n- real-db-upgrade_nova_percona_user_002:th-percona http://www.rcbops.com/turbo_hipster/results/43/92443/12/check/real-db-upgrade_nova_percona_user_002:th-percona/0f292da8e5fb41ea868002ac9e648982/index.html : SUCCESS in 15m 58s\n\nTo recheck, leave \u0027recheck migrations\u0027 as a comment. More information: https://wiki.openstack.org/wiki/ThirdPartySystems/DB_Datasets_CI","accounts_in_message":[],"_revision_number":12},{"id":"c1f626d09c57c49094ffcd82407c96654cf0a2d3","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-12-12 22:37:28.000000000","message":"Patch Set 12: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/43/92443/12/2461/results.html\n\nUse \"xenserver: recheck\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.\n\nDebugging suggestions at https://wiki.openstack.org/wiki/Debugging_XenServer_CI_failures","accounts_in_message":[],"_revision_number":12},{"id":"799017dbb1287bfeb47fa481060aab7c72354584","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-12-12 22:40:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-nova-pep8 http://logs.openstack.org/43/92443/12/check//gate-nova-pep8/d4d670b/ : SUCCESS in 8m 12s\n- gate-nova-docs http://docs-draft.openstack.org/43/92443/12/check//gate-nova-docs/4747b25//doc/build/html/ : SUCCESS in 7m 04s\n- gate-nova-python27 http://logs.openstack.org/43/92443/12/check//gate-nova-python27/51da92c/ : SUCCESS in 16m 18s\n- check-tempest-dsvm-full http://logs.openstack.org/43/92443/12/check//check-tempest-dsvm-full/0f51e27/ : SUCCESS in 34m 12s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/92443/12/check//check-tempest-dsvm-postgres-full/d024d4e/ : SUCCESS in 37m 34s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/43/92443/12/check//check-tempest-dsvm-neutron-full/4491aa3/ : SUCCESS in 47m 16s\n- check-grenade-dsvm http://logs.openstack.org/43/92443/12/check//check-grenade-dsvm/f182b97/ : SUCCESS in 28m 14s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/92443/12/check//gate-tempest-dsvm-large-ops/ec85fd5/ : SUCCESS in 24m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/92443/12/check//gate-tempest-dsvm-neutron-large-ops/5099029/ : SUCCESS in 18m 02s\n- check-devstack-dsvm-cells http://logs.openstack.org/43/92443/12/check//check-devstack-dsvm-cells/b767666/ : SUCCESS in 15m 54s\n- gate-nova-tox-functional http://logs.openstack.org/43/92443/12/check//gate-nova-tox-functional/e73859e/ : SUCCESS in 19m 49s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/43/92443/12/check//check-grenade-dsvm-partial-ncpu/8916876/ : SUCCESS in 27m 55s\n- check-tempest-dsvm-ironic-pxe_ssh http://logs.openstack.org/43/92443/12/check//check-tempest-dsvm-ironic-pxe_ssh/9261102/ : SUCCESS in 29m 27s\n- check-tempest-dsvm-docker http://logs.openstack.org/43/92443/12/check//check-tempest-dsvm-docker/f608448/ : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f9e5da2af43645f01d144e2e9bedd9a53623450d","author":{"_account_id":5170,"name":"Microsoft Hyper-V CI","email":"hyper-v_ci@microsoft.com","username":"hyper-v-ci","tags":["SERVICE_USER"]},"date":"2014-12-13 00:35:23.000000000","message":"Patch Set 12:\n\nBuild failed. For rechecking only on the Hyper-V CI, add a review comment with check hyper-v\n\n- http://64.119.130.115/92443/12/ : Test run failed. in 2h 46m 57s","accounts_in_message":[],"_revision_number":12},{"id":"24a3df4f31b58997171b48ada9e72c909ddfc9ea","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-12-14 20:54:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/92443/11","accounts_in_message":[],"_revision_number":11},{"id":"1b60c27072d40ac4583f16cc361f7a6281976283","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2014-12-16 00:05:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild successful. Logs: http://208.91.1.172/logs/92443/12","accounts_in_message":[],"_revision_number":12},{"id":"7db147e7ecf78fa3a67a8f1ba4c79c1584646bec","author":{"_account_id":7,"name":"Jay Pipes","email":"jaypipes@gmail.com","username":"jaypipes"},"date":"2014-12-17 18:55:08.000000000","message":"Patch Set 12: Code-Review+2\n\n(7 comments)\n\nCode looks good to me, thank you! Left a few notes inline that you might want to address if you respin this. Not +W\u0027ing it until John lifts the blueprint -2.\n\nBest,\n-jay","accounts_in_message":[],"_revision_number":12},{"id":"45783e51ec074e4f50f0ced11038b6d9aca4b123","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2014-12-18 13:12:16.000000000","message":"Patch Set 12:\n\nI am afraid I can\u0027t lift the blueprint -2 until we get the blueprint approved.","accounts_in_message":[],"_revision_number":12},{"id":"5496468094f3b89c8d7b57919032b069f88eb419","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-01-11 08:46:02.000000000","message":"Patch Set 12: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"baeba58f5d024d493c12844b9884c89d632f05ae","author":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"date":"2015-01-13 21:42:05.000000000","message":"Patch Set 12: Workflow-1\n\n@jaypipes\n\nThank you for your review. I\u0027d appreciate it. Unfortunately, I couldn\u0027t get approval from Cinder side and I missed the dead line of driver support for Kilo phase. To avoid the confusion, I added WIP flag for this review.\n\n\n@johnthetubaguy\n\nNo problem. Based on the previous discussion I understood the meaning of this -2 flag.","accounts_in_message":[],"_revision_number":12},{"id":"3c5a51e0d8ef616a4f1f40afa3fdda933d05375b","author":{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},"date":"2015-03-11 06:54:44.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)\n\nSee in line.","accounts_in_message":[],"_revision_number":12},{"id":"d5c2f6f5c288f49ad76975448f6fffc79f034e96","author":{"_account_id":11676,"name":"wuyuting","email":"wytdahu@gmail.com","username":"wuyuting"},"date":"2015-03-11 07:03:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8cf9c310e7eda41874c7df2867d3b7ca75edc4b6","author":{"_account_id":1779,"name":"Daniel Berrange","email":"berrange@redhat.com","username":"berrange"},"date":"2015-03-11 10:06:58.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c21d03e20da3b50a849d0ae8eb1c16964b624e00","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2015-05-13 17:19:29.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":12},{"id":"45807d78983bc726c3d50623306f0b2f0ed2ba7c","author":{"_account_id":1849,"name":"Joe Gordon","email":"joe.gordon0@gmail.com","username":"jogo"},"date":"2015-05-13 17:19:29.000000000","message":"Patch Set 12:\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"ad39a0b61fe715a6f10a5a0a9bbce5153fbc57a0","revisions":{"91785b00c0f50ba4892bab68e0d86b539e4336f9":{"kind":"REWORK","_number":1,"created":"2014-05-06 19:26:35.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/1"}}},"commit":{"parents":[{"commit":"bef55b21adb3485803aa1ef3f414f863aea2bebc","subject":"Merge \"wrap_check_security_groups_policy is already defined\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bef55b21adb3485803aa1ef3f414f863aea2bebc"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-04-04 18:49:44.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-05-06 19:23:44.000000000","tz":-240},"subject":"LVM: Support a volume-gorup on shared storage","message":"LVM: Support a volume-gorup on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91785b00c0f50ba4892bab68e0d86b539e4336f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91785b00c0f50ba4892bab68e0d86b539e4336f9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bef55b21adb3485803aa1ef3f414f863aea2bebc","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"898be5f080c7482f0c125f09170b3857d87fb5bc":{"kind":"REWORK","_number":2,"created":"2014-06-09 15:12:30.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/2"}}},"commit":{"parents":[{"commit":"269e7b42f2bcd4798823430e11ff68a4863c1349","subject":"Merge \"Do not replace the aggregate metadata when updating az\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/269e7b42f2bcd4798823430e11ff68a4863c1349"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-04-04 18:49:44.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-07 00:09:36.000000000","tz":-240},"subject":"LVM: Support a volume-gorup on shared storage v2","message":"LVM: Support a volume-gorup on shared storage v2\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChanges from v1\n - Fixed some typos.\n - Change comment format.\n - Remove unnecessary return.\n - Remove marker function from debug level logs.\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/898be5f080c7482f0c125f09170b3857d87fb5bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/898be5f080c7482f0c125f09170b3857d87fb5bc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"269e7b42f2bcd4798823430e11ff68a4863c1349","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9dc8734e99239a038f62a55be1a24e05afeca38b":{"kind":"REWORK","_number":3,"created":"2014-06-16 21:56:08.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/3"}}},"commit":{"parents":[{"commit":"91765a1b544870d16a2565321185ecbcb51d2f1d","subject":"Create lvm.py module containing helper API for LVM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/91765a1b544870d16a2565321185ecbcb51d2f1d"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-16 15:10:09.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-16 21:39:49.000000000","tz":-240},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9dc8734e99239a038f62a55be1a24e05afeca38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9dc8734e99239a038f62a55be1a24e05afeca38b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"91765a1b544870d16a2565321185ecbcb51d2f1d","is_merged_in_target_branch":false,"change_id":"I4d80f574631d52012152c72d1db3069dd57a5550","change_number":100369,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"61ff488cdcf669fe6257a9b9679d37654dafa501":{"kind":"REWORK","_number":4,"created":"2014-06-18 22:28:16.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/4"}}},"commit":{"parents":[{"commit":"9e00d4cf5ed1a6e61fc41bffef7290c919552f41","subject":"Create lvm.py module containing helper API for LVM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9e00d4cf5ed1a6e61fc41bffef7290c919552f41"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-16 15:10:09.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-18 22:15:18.000000000","tz":-240},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61ff488cdcf669fe6257a9b9679d37654dafa501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/61ff488cdcf669fe6257a9b9679d37654dafa501"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9e00d4cf5ed1a6e61fc41bffef7290c919552f41","is_merged_in_target_branch":false,"change_id":"I4d80f574631d52012152c72d1db3069dd57a5550","change_number":100369,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1a1ff2070f279bf45bfb19fdbadc238f39832d49":{"kind":"REWORK","_number":5,"created":"2014-06-19 23:44:43.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/5"}}},"commit":{"parents":[{"commit":"7c2c0df5419c861fd245fc52dcd8ef49ca89edc9","subject":"Create lvm.py module containing helper API for LVM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c2c0df5419c861fd245fc52dcd8ef49ca89edc9"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-16 15:10:09.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-19 23:39:37.000000000","tz":-240},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a1ff2070f279bf45bfb19fdbadc238f39832d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a1ff2070f279bf45bfb19fdbadc238f39832d49"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7c2c0df5419c861fd245fc52dcd8ef49ca89edc9","is_merged_in_target_branch":true,"change_id":"I4d80f574631d52012152c72d1db3069dd57a5550","change_number":100369,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4a91a1bfe116a13e64e5cf232c1907d25097e7c0":{"kind":"REWORK","_number":6,"created":"2014-06-20 19:16:56.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/6"}}},"commit":{"parents":[{"commit":"fbc678668442e3d29593a17ab5351b7358f37788","subject":"Merge \"xenapi: reorder volume_utils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbc678668442e3d29593a17ab5351b7358f37788"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-16 15:10:09.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-06-20 19:13:13.000000000","tz":-240},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a91a1bfe116a13e64e5cf232c1907d25097e7c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a91a1bfe116a13e64e5cf232c1907d25097e7c0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fbc678668442e3d29593a17ab5351b7358f37788","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"de02f87bda1e8ba9508d177c0ebe140b98a0c283":{"kind":"REWORK","_number":7,"created":"2014-08-08 16:45:21.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/7"}}},"commit":{"parents":[{"commit":"63ed8bed64ae953ce730fc9bcf50ddc9b576e6e1","subject":"Merge \"Add instanceset info to StartInstance response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63ed8bed64ae953ce730fc9bcf50ddc9b576e6e1"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-04 19:03:35.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-08 16:43:25.000000000","tz":-240},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de02f87bda1e8ba9508d177c0ebe140b98a0c283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de02f87bda1e8ba9508d177c0ebe140b98a0c283"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"63ed8bed64ae953ce730fc9bcf50ddc9b576e6e1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"11b009cc2ef8b391a9c92640816f5ad41e530f32":{"kind":"REWORK","_number":8,"created":"2014-08-08 19:38:19.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/8"}}},"commit":{"parents":[{"commit":"63ed8bed64ae953ce730fc9bcf50ddc9b576e6e1","subject":"Merge \"Add instanceset info to StartInstance response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63ed8bed64ae953ce730fc9bcf50ddc9b576e6e1"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-04 19:03:35.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-08 18:00:32.000000000","tz":-240},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nCurrently LVM driver uses software iSCSI target to export a\nvolume-group. This patch adds a feature to support a volume-group on\nshared storage instead of exporting via iSCSI target. When a\nvolume-group on a shared storage (over fibre, iscsi or whatever) is\nvisible from all nodes such as a cinder node and compute nodes, these\nnodes can access the volume-group simultaneously without the target.\n\nGeneral benefits\n- In regard to one storage, use both specific cinder driver and cinder\n  shared LVM driver as the situation demands.\n- Enable cinder to any kinds of shared storages without specific cinder\n  driver.\n- Provide quicker volume and snapshot creation.\n\n*Note that this requires a Cinder lvm SharedLVMDriver as well.\n - https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n - https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/11b009cc2ef8b391a9c92640816f5ad41e530f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/11b009cc2ef8b391a9c92640816f5ad41e530f32"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"63ed8bed64ae953ce730fc9bcf50ddc9b576e6e1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"26efecf1453039438fc6510a9d771bcc42122967":{"kind":"REWORK","_number":9,"created":"2014-11-14 00:51:42.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/9"}}},"commit":{"parents":[{"commit":"08ffe7c1ee46ce9c80f6bd1024e790cb8d0abbb8","subject":"Merge \"Use six.text_type instead of unicode\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/08ffe7c1ee46ce9c80f6bd1024e790cb8d0abbb8"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-04 19:03:35.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-11-14 00:39:36.000000000","tz":-300},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nThis proposal enables LVM driver to support direct data path(I/O path)\nfrom VMs to Storage array using LVM on shared storage instead of using\nI/O path which goes through software SCSI target. As a result, the I/O\npath can be independent of Cinder Node. By enabling direct I/O path\nprovides stable I/O path for VMs and improves throughput for VMs.\n\n*Note: In this feature, both Nova and Cinder pieces are necessary.\n\n Blueprints:\n  * https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n  * https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n\n Spec:\n  * nova:   https://review.openstack.org/#/c/127318/\n  * cinder: https://review.openstack.org/#/c/129352/\n\n Gerrit Review:\n  * nova:   https://review.openstack.org/#/c/92443/\n  * cinder: https://review.openstack.org/#/c/92479/\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26efecf1453039438fc6510a9d771bcc42122967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26efecf1453039438fc6510a9d771bcc42122967"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"08ffe7c1ee46ce9c80f6bd1024e790cb8d0abbb8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"c7bc4137604f1b748f511a91bd41e0cf4927355a":{"kind":"REWORK","_number":10,"created":"2014-11-14 00:53:41.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/10"}}},"commit":{"parents":[{"commit":"ceafd1c5f6f6fc4ffc59f00814314ca02cbc27aa","subject":"Merge \"Remove unused db.api.flavor_extra_specs_get_item\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ceafd1c5f6f6fc4ffc59f00814314ca02cbc27aa"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-04 19:03:35.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-11-14 00:52:59.000000000","tz":-300},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nThis proposal enables LVM driver to support direct data path(I/O path)\nfrom VMs to Storage array using LVM on shared storage instead of using\nI/O path which goes through software SCSI target. As a result, the I/O\npath can be independent of Cinder Node. By enabling direct I/O path\nprovides stable I/O path for VMs and improves throughput for VMs.\n\n*Note: In this feature, both Nova and Cinder pieces are necessary.\n\n Blueprints:\n  * https://blueprints.launchpad.net/nova/+spec/lvm-driver-for-shared-storage\n  * https://blueprints.launchpad.net/cinder/+spec/lvm-driver-for-shared-storage\n\n Spec:\n  * nova:   https://review.openstack.org/#/c/127318/\n  * cinder: https://review.openstack.org/#/c/129352/\n\n Gerrit Review:\n  * nova:   https://review.openstack.org/#/c/92443/\n  * cinder: https://review.openstack.org/#/c/92479/\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c7bc4137604f1b748f511a91bd41e0cf4927355a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c7bc4137604f1b748f511a91bd41e0cf4927355a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ceafd1c5f6f6fc4ffc59f00814314ca02cbc27aa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a7a721fb91f0cd08a38eb482b84b3ffc2e0ec355":{"kind":"REWORK","_number":11,"created":"2014-12-12 01:21:07.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/11"}}},"commit":{"parents":[{"commit":"9a76042eaad034cdd1ffa5346efe117e76aec6e2","subject":"Merge \"VMware: Support volume hotplug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9a76042eaad034cdd1ffa5346efe117e76aec6e2"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-04 19:03:35.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-12-12 01:13:16.000000000","tz":-300},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nThis proposal enables LVM driver to support direct data path(I/O path)\nfrom VMs to Storage array using LVM on shared storage instead of using\nI/O path which goes through software SCSI target. As a result, the I/O\npath can be independent of Cinder Node. By enabling direct I/O path\nprovides stable I/O path for VMs and improves I/O latency.\n\n*Note: In this feature, both Nova and Cinder pieces are necessary.\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7a721fb91f0cd08a38eb482b84b3ffc2e0ec355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7a721fb91f0cd08a38eb482b84b3ffc2e0ec355"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9a76042eaad034cdd1ffa5346efe117e76aec6e2","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ad39a0b61fe715a6f10a5a0a9bbce5153fbc57a0":{"kind":"REWORK","_number":12,"created":"2014-12-12 21:47:53.000000000","uploader":{"_account_id":10115,"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@lycorp.co.jp","username":"mtanino"},"ref":"refs/changes/43/92443/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/92443/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/92443/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/92443/12"}}},"commit":{"parents":[{"commit":"68c7492f6baa7aa4122529af4e73ec4b5be8750f","subject":"Merge \"Re-revert \"libvirt: add version cap tied to gate CI testing\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68c7492f6baa7aa4122529af4e73ec4b5be8750f"}]}],"author":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-08-04 19:03:35.000000000","tz":-240},"committer":{"name":"Mitsuhiro Tanino","email":"mitsuhiro.tanino@hds.com","date":"2014-12-12 21:40:06.000000000","tz":-300},"subject":"LVM: Support a volume-group on shared storage","message":"LVM: Support a volume-group on shared storage\n\nThis proposal enables LVM driver to support direct data path(I/O path)\nfrom VMs to Storage array using LVM on shared storage instead of using\nI/O path which goes through software SCSI target. As a result, the I/O\npath can be independent of Cinder Node. By enabling direct I/O path\nprovides stable I/O path for VMs and improves I/O latency.\n\n*Note: In this feature, both Nova and Cinder pieces are necessary.\n\nChange-Id: Ic280e569d963c25f424d5dbc329bcb08fdaba04c\nSigned-off-by: Mitsuhiro Tanino \u003cmitsuhiro.tanino@hds.com\u003e\nImplements: blueprint lvm-driver-for-shared-storage\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad39a0b61fe715a6f10a5a0a9bbce5153fbc57a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad39a0b61fe715a6f10a5a0a9bbce5153fbc57a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"68c7492f6baa7aa4122529af4e73ec4b5be8750f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
