)]}'
{"id":"openstack%2Fnova~926223","triplet_id":"openstack%2Fnova~master~I247dc0ec03cd9e5a7b41f5c5534bdfb1af550029","project":"openstack/nova","branch":"master","topic":"trival","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-08-13 16:33:02.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-08-16 22:35:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I247dc0ec03cd9e5a7b41f5c5534bdfb1af550029","subject":"hardware: Correct log","status":"MERGED","created":"2024-08-13 16:28:08.000000000","updated":"2024-08-16 22:36:49.000000000","submitted":"2024-08-16 22:35:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926223-trival","meta_rev_id":"9c10f6499f153a1a898d6bd086e4955def7639ea","_number":926223,"virtual_id_number":926223,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-16 22:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2024-08-13 18:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-08-13 16:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2024-08-16 01:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 16:29:56.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-08-13 16:32:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-08-13 16:50:07.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-13 17:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-13 18:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-16 16:07:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"1504f7b3b487e83f6250ec2eb91ff8362f5b51fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-13 16:28:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2720bdf0ba6dcd0ba6046da2ad38c32bcf522a4c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-13 16:29:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"800dc60952ea5ee5846c29d0b6bec05b9a8423c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-13 16:32:21.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"400e53cc45002d2f6a24f74c3282fea3903d4ff1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-08-13 16:32:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9276c3d214173d7944edbe339c8ba346636e7b2a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-13 16:33:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8076ac3eb1e1bc8b132f6bdaa14bae32efa45a7e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-13 16:50:07.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-926223-bjhzx/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"a7fb96446b6fd24ac5183eeebb2b55800f9d3e2f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-13 17:08:47.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-926223-m4z6n/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"cf3127a53fc05fdb60557111604b83b229e548fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:13:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/043666d1b9984833bd825490e7ce6731\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4babbdfa64e84c7c8745f38a1e7db40c : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbe7d4412dd242e6aecf6f671858f5df : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5469da6dc6d16b5379f5edd9c8dcafdeef6630f9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-13 18:18:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a96d240791ec18a4e466616f77c1298ddcb8aa5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:19:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6b80fac25d14886aa68164308dfc48f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66640ee84f1f410f94fb272b3a77b631 : SUCCESS in 1h 02m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6b08ea0f464e4e60a30ee4f4e5002487 : SUCCESS in 1h 44m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e6b655e6448454a93228dae6541bdaa : SUCCESS in 59m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38dc97a70c054f0683a399e6613a444f : SUCCESS in 24m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ebb8b0e721473ea3a6f07ac7ac0468 : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d639bfbc14fe4026895cebb526c12910 : SUCCESS in 15m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb3c495a5df74f349f5d7b8cd6b692ff : SUCCESS in 14m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f42f79aaecb64cd5b4f42f653efeae5a : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de2be10d39d44d8b02e211baca9a4f1 : SUCCESS in 9m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ab06eedbfe694447b7b20dade1e32c6b : SUCCESS in 49m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70aecfff89134773b31ff04a9ae6087f : FAILURE in 1h 36m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8c1115d8e18044e696d2f0af8b23f37d : SUCCESS in 44m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f76ae60ef6ee4b1eac2665467ebee439 : SUCCESS in 41m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/83834f7d8f114068ac67584fc8f004f5 : SUCCESS in 1h 33m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/07e8fa0d0f1a4a58ab311ade245c6044 : SUCCESS in 1h 20m 51s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/7ea46b4787f74c6dbe01b6c2ba2c4237 : SUCCESS in 51m 59s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9f41f4f91f464397b2ae0361f4ca66c1 : SUCCESS in 5m 48s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c5a050432044da09ea7328bdcf1b928 : SUCCESS in 26m 23s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/473b88ac00db45188f4d03aac565483f : SUCCESS in 24m 18s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fb10997722964fc2887235b0bb222240 : SUCCESS in 32m 17s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5176eaf6033f4214b64814996d122905 : SUCCESS in 1h 35m 00s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c43291c00a084c26a0de454678778418 : SUCCESS in 1h 38m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/16603360ca5842b691a749f13edeff7b : SUCCESS in 1h 16m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fe04a8532ed49d697ae7e2a808978dd : SUCCESS in 56m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e551eaed1cc7426f998bd583bb2b463a : SUCCESS in 22m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b33f67c57ac84fa7841808edf2055715 : SUCCESS in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1aee46634e7be24fb74ee22a89f9cb1586cb585","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-16 01:26:42.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f59f39bd946aac5fed5692a5826c19cf5291ebb","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-16 01:46:37.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-926223-mh4xd/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"0bd6630f151d79bee6ceb729fc975c649472b291","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-16 02:06:31.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-926223-t99t4/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"e43f8637a57444ce971e504b0d41ad2536906f65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 02:21:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cfeb57ee25e4a409d831fd192626365\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54c80b742a4c43428f229b1f02b1ec17 : FAILURE in 34m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3743bbf1d17c4e019052ce93e2ea709b : SUCCESS in 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dbfa0f04b11a4970b45a995e5002ee7944ebac76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 03:22:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c1ed87dc9464e259f550737c3e797e0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43c92bbf582c479dbc1a93f6da7a7895 : SUCCESS in 48m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cd7dd1439feb471f89e3d7aa1d686ddf : SUCCESS in 1h 23m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b54dfa1198c34c6bbdca0c80aa387785 : SUCCESS in 47m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc7914ee1b7140088b43c4896521dace : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4340c5c3454ee29cbeaccf4fa6a0fd : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8c8345264b84bddbf1d8f961a73f189 : SUCCESS in 15m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0e3ad74255b4b0f96e126dc00d4b275 : SUCCESS in 13m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c2fc2b7404b4a488155baa5784c99b7 : SUCCESS in 18m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef0d09d1b7c40aa86228ddf3e3b1c87 : SUCCESS in 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d8a6e7f81844de4b1cdc50a1692bf04 : SUCCESS in 1h 00m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/87d22244f89c494abdc6b81507531694 : SUCCESS in 1h 48m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/40911547b4674b87be29748d4e1f1a5d : SUCCESS in 34m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/382215606ccd40b18dc5d3a33f080a3d : SUCCESS in 54m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6db926959cbb4c5882fab93331c5cebd : SUCCESS in 1h 34m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/0de2cc7bc4ab480d84d2fb78efcba21d : SUCCESS in 1h 30m 45s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/09322b92a9f44e35a0429efebed606b2 : SUCCESS in 45m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d2195540d3044580a5123e90f5eee855 : SUCCESS in 4m 27s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b75741209f4141618feea8c2abdcaf75 : SUCCESS in 20m 30s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e42a8f9190b54800aee99cfcd7b0c8a3 : SUCCESS in 26m 30s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dac90b28fc694f1ea1fac269acdb8867 : SUCCESS in 30m 16s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0c284182b99442f691c46818bfc7609c : FAILURE in 1h 20m 59s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/4b71318a1e83449a82a27d8e0e79adbe : SUCCESS in 1h 52m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fdd40a0b5ab476786d560e880a95489 : SUCCESS in 1h 21m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a61c96ac21c04f72a9080aa9bb8fb2cb : SUCCESS in 1h 00m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/61b13acc6a0c433d9e450175f1d95f05 : SUCCESS in 28m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/641fb20291ae4ba897b7bc9164a19b9f : SUCCESS in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b2e37376ade8e7036b93a4c79141dfd4727701c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-16 13:57:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6d867a3a853eb1cac0ac8f71d669e3a60806a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 16:01:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9d1d2e15747497889037c8c0f5291a7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ec0ca092c76418cb91a1b21aa94fdcf : SUCCESS in 1h 02m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a8f4c16d3e2a45a5953b914d5ff8a4e5 : SUCCESS in 1h 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62f419f90d9f4252aca2f9697bbf8e76 : SUCCESS in 59m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec9c5ac5a45544b8a3f0aa8df6473032 : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c23466d8267464580700181e6cc5a62 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d360c67307ec4bf8a9da7c68e8184ca4 : SUCCESS in 14m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/640a513879324d2b9ce85166c9a7143f : SUCCESS in 12m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56740a94fa73426b84388b35154831f9 : SUCCESS in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79984f6ae844b0baad5fbeb1fd45f15 : SUCCESS in 12m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e59956da759a424f93cb10f73641e104 : FAILURE in 1h 04m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/86feebe7a60d489a9f171d2b191a2445 : SUCCESS in 1h 46m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b8befdf4c8de4cb990aba612fa656737 : SUCCESS in 49m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7aacccb139b14a598184251c4accf08d : SUCCESS in 52m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/53897a7c0f434e40be389f8c69df2ea1 : SUCCESS in 1h 46m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/8e3fd953c15b440886a053a542bc469d : SUCCESS in 1h 30m 59s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/439f3a174d18486ab9411d9547fb55ca : SUCCESS in 48m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4ad05730a06d43ab9bca59de1f899fb5 : SUCCESS in 5m 46s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ab8fbde3d943423f819d968e00c1b355 : SUCCESS in 24m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/63f569dd37b54b5da4e448a172595224 : SUCCESS in 25m 07s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/822a28a991d8462cb1f3e4df6975df64 : SUCCESS in 30m 44s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/821bceb722764cfea801c37bb6241c3c : SUCCESS in 1h 51m 27s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/8cef79aa94dd4391a0db4de0734fbcad : SUCCESS in 1h 44m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d67bb5e0181546e8822337976d78914c : SUCCESS in 1h 19m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80979a548b684947a60447c90206537b : SUCCESS in 1h 00m 38s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/63b1cf5ee0f9472c851a322d7c5e91c3 : SUCCESS in 25m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b877894713f0495b82bf955473ecc55e : SUCCESS in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9692a891a91381696b5ef13af5479fa37a917f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 16:02:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a788cd069cfded5a3b95feda70ce8a7ba42e17b0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-16 16:07:40.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"f97cfad39740c7341a0f0b8fb8fed2f22b0fd32b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 16:26:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9627a2ad8d124cca8e42812f94cd3878\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9201b09d956a4d47be9e2ee2e3393138 : FAILURE in 32m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5dc0893f55f64b1893c41b38924c843d : SUCCESS in 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"074fa20b2a53792c4962c2ffc5737d06fa60dcde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 17:45:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8f5277cbff74954b6e8de59722defc1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a75ee306fc549039e49cce373c4756f : SUCCESS in 59m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ba20b30c47884a2b8fb55a15e7ade61a : SUCCESS in 1h 40m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da3c538e6443489f8e38642875004177 : SUCCESS in 1h 09m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6e13bc67fed4c30b2d6290db51e83bf : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efa05fe2ebb545349980414a602df565 : SUCCESS in 14m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1c42a199fac47749cf6cd824d0f7149 : SUCCESS in 12m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fdf0b83e694a089f8cc4c48a034305 : SUCCESS in 10m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6bc1262e27e748ae8b1dc0200d1eee8b : SUCCESS in 47m 10s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f288948600bf44d4836d822f434d886e : SUCCESS in 23m 59s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30e95aae04b34ad6b859210fc6417e68 : SUCCESS in 23m 39s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/56abf8b5563e4d76bf9bc7185928bca4 : SUCCESS in 25m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/78c16c913e924f648df23999c857b327 : SUCCESS in 1h 26m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/06ecc0effc5840ffae5e6da14e845a4e : SUCCESS in 1h 30m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/244d6594fe724f93a242ad8590162925 : SUCCESS in 4m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7a25fdd221c4a289b73264dbb78c1ff : SUCCESS in 1h 30m 23s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cacc156e61ef4b6f88bc7bcb0c6a818c : SUCCESS in 1h 32m 35s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/9ffd273a6f094196885b94fc40a03c0e : SUCCESS in 1h 26m 59s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/96dcef07f23246fcbe84f3a2830ecbbc : FAILURE in 1h 09m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec363782e73d4432b1f791aa07c82bfc : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":2},{"id":"97300622cb8c2d55e7320de849fd16d39aa40b0f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-16 17:56:11.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-926223-jcv8c/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"a760e07d93085684b311077584bad26ceec6689d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-16 18:04:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac3cd71d56f0b95e2b0d33c5069b68541edb974d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 18:50:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5188e7214674b8bb408a978747611af\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56c659fa813140b2af2c9f61e13e322c : SUCCESS in 29m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2dcda084523e4a659a466b61ebded8e3 : FAILURE in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"009fa85fc06438aa5c495e1fdc4b176156eaece1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-08-16 19:12:25.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-926223-zvx6n/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"d8f1781047b7590f8f544cdce80c22cb900a12e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 20:10:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2edf274a767e47c3ba19a76b08b934bf\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1dab836e5539434985c130a41cb80966 : SUCCESS in 1h 05m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4711bd5502d147c8acf27775c95b4d0b : SUCCESS in 1h 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43c4a4254ef84b86aa151cdff25fa5a0 : SUCCESS in 53m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6bf5c0b44bf4ec99534032b0c73decb : SUCCESS in 23m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c582a4553e42cdbf9937661c3390e3 : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f88231da99d74f7aa97a93735a3a611c : SUCCESS in 13m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d086e604b9c84cba8ed0647e23f374c1 : SUCCESS in 13m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/363d6f2327464ee399b46f50d2915f17 : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0296f5e83f84161895bd9ed5fa96037 : SUCCESS in 8m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4e07704c10774601b13978e06961506e : SUCCESS in 57m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bda65fabcd6440e1b33f171425cb399e : SUCCESS in 1h 55m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6ea05ac295164c0fb3b68dc3e5423c6f : SUCCESS in 35m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/457e672382d14c7ba43cc5a5782085dc : SUCCESS in 52m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/423ed7d47b294f4a8dc154b23b6ae0f1 : SUCCESS in 1h 57m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/0cc2330a2c3746099bb6bf9249994208 : SUCCESS in 1h 55m 02s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/49337ef98515478abbab47915e666a2b : SUCCESS in 35m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c6e47841c1184ea2bacd65d6842f56ab : SUCCESS in 4m 49s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e31807d10dff4bfc8da4e9eb12800704 : SUCCESS in 19m 57s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e0960e3cba294d28ac08d293b3edfaad : SUCCESS in 28m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/750b33db2bf8494c8d642316a27062a7 : SUCCESS in 32m 48s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b1e2baba6c1e4a97b23294ef3f554036 : SUCCESS in 1h 57m 49s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/acfed783fc4746e295f4e83d0d491daa : SUCCESS in 1h 49m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/18fdc7bb815d4f1a8731ea18bca56255 : SUCCESS in 1h 25m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/016e01b9614a42ccad227172f87cb264 : SUCCESS in 46m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4639d698c2534b8fa078620b70a58aa9 : SUCCESS in 16m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b213ac95ea1d4b5299f53fff3e6643ce : SUCCESS in 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1ca79ec7fcc7232d97b6dd6262eee0f277e7e9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 20:11:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a8995d7d075cf477d6cb0c2f62118f9d5670d38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-08-16 20:21:11.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"9ceac983108eb66f4042bee409cca60aacf43f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 22:35:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e3bcbea251e456294b4c746cae7252e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b377984bd9d9437eb4f73707e18ab3dd : SUCCESS in 1h 03m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7f1585f29e3844879fcf86d90233816a : SUCCESS in 1h 35m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9be48d33ba934e608f2573ee8301cd44 : SUCCESS in 1h 01m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6eb9771d336424ab08486329ec81809 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7c7017e9f084f019ded8511d262245e : SUCCESS in 14m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70ad2e0a786e489194b11cff2f845bf3 : SUCCESS in 14m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b88a49b5ffc4879946731b36228caae : SUCCESS in 8m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/983bc23dbb154eefb53f30d76b64a127 : SUCCESS in 44m 26s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d059672886774c42956d1633f98b0d8a : SUCCESS in 22m 34s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3cddc82356ec4a83a1cafd4ca698cfb5 : SUCCESS in 23m 21s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/44471d66581847d886bb24da4149f1fc : SUCCESS in 29m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/26d050ea1e1a449bac6f3c4e2c06cd49 : SUCCESS in 2h 14m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/57d8c12a77a04c40bb3cf65a6d323a66 : SUCCESS in 1h 28m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d5f7c9b8104440649244ee62a80d90cf : SUCCESS in 4m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f361d2353b404ca2bea4155bf35f3161 : SUCCESS in 1h 40m 18s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba9e7eae55044c229f7451c903b26396 : SUCCESS in 1h 19m 05s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/300d836cd1f84592be1f6623738f9246 : SUCCESS in 2h 00m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/289493492ff64778b43353be1bd3e2cd : SUCCESS in 1h 15m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fae5cc4c797746beacd42130a3761787 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":2},{"id":"de167e5e5770eb724e505e19447bb92469c643d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 22:35:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9c10f6499f153a1a898d6bd086e4955def7639ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 22:36:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0353268f2ebf4534a11642715e75abfe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4245ee793244322965cd08d5d254ef3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4678bcbb064da580500b1dbeddb0bdfdeac074ef","revisions":{"4e3ed9a7abbb85aa230ae309e7808ab0afec64c8":{"kind":"REWORK","_number":1,"created":"2024-08-13 16:28:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/926223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/926223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/926223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/926223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/926223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/926223/1"}}},"commit":{"parents":[{"commit":"8c4a2e1b4759b9932bcad10810143e61ec9b4c36","subject":"Merge \"tests: Ensure all APIs have a request query schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c4a2e1b4759b9932bcad10810143e61ec9b4c36"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-13 16:24:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-13 16:24:31.000000000","tz":60},"subject":"hardware: Correct log","message":"hardware: Correct log\n\nWe currently get the following error message if attempting to fit a\nguest with hugepages on a node that doesn\u0027t have enough:\n\n  Host does not support requested memory pagesize, or not enough free\n  pages of the requested size. Requested: -2 kB\n\nCorrect this, removing the kB suffix and adding a note on the meaning of\nthe negative values, like we have for the success path.\n\nChange-Id: I247dc0ec03cd9e5a7b41f5c5534bdfb1af550029\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e3ed9a7abbb85aa230ae309e7808ab0afec64c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e3ed9a7abbb85aa230ae309e7808ab0afec64c8"}]},"branch":"refs/heads/master"},"4678bcbb064da580500b1dbeddb0bdfdeac074ef":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-13 16:32:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/926223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/926223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/926223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/926223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/926223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/926223/2"}}},"commit":{"parents":[{"commit":"8c4a2e1b4759b9932bcad10810143e61ec9b4c36","subject":"Merge \"tests: Ensure all APIs have a request query schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c4a2e1b4759b9932bcad10810143e61ec9b4c36"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-13 16:24:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-13 16:32:21.000000000","tz":0},"subject":"hardware: Correct log","message":"hardware: Correct log\n\nWe currently get the following error message if attempting to fit a\nguest with hugepages on a node that doesn\u0027t have enough:\n\n  Host does not support requested memory pagesize, or not enough free\n  pages of the requested size. Requested: -2 kB\n\nCorrect this, removing the kB suffix and adding a note on the meaning of\nthe negative values, like we have for the success path.\n\nChange-Id: I247dc0ec03cd9e5a7b41f5c5534bdfb1af550029\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #2075959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4678bcbb064da580500b1dbeddb0bdfdeac074ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4678bcbb064da580500b1dbeddb0bdfdeac074ef"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
