)]}'
{"id":"openstack%2Fnova~929858","triplet_id":"openstack%2Fnova~master~Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b","project":"openstack/nova","branch":"master","topic":"resize-swap","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-12-12 03:17:48.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2025-01-15 00:33:34.000000000","reason":"removed on reply"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-12-10 16:49:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-12-12 05:48:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b","subject":"Update Nova bdm with updated swap info","status":"MERGED","created":"2024-09-18 23:45:18.000000000","updated":"2025-01-15 00:33:34.000000000","submitted":"2024-12-12 05:48:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"929858-resize-swap","meta_rev_id":"108864fe8479c4b024d13c3cc34ed686d827a75a","_number":929858,"virtual_id_number":929858,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-12 05:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2024-12-12 03:17:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2024-12-10 16:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":1,"date":"2024-12-11 06:49:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2024-12-12 03:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-19 00:07:47.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-19 01:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-19 02:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-20 11:58:00.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-10-02 14:53:36.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-10-14 13:54:02.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2024-12-11 06:49:38.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"}],"messages":[{"id":"64175c5b84d4806739395abd6641797c695e7c91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-18 23:45:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b8084dc42298113eed31dda35c78bcca88684c2","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-18 23:46:04.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3f499ed06a6d6e189daacee5ce4d4077ac2b3138","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-19 00:07:47.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-64549/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"7194e3d6927c4307072fbe4745063905130ea133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 01:34:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34458411f22045f1870d7270b8caa58a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b207434d890411792e31b7bca6171c7 : FAILURE in 34m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9fa7a8922f41405eb70d164ac1663da0 : FAILURE in 44m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d0f869406e52c19522191b00d78aa1a86145cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 02:25:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f484f658d35241b894883200e9345b40\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3ee1e1106104404897647be25c13f08 : SUCCESS in 56m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a239312fede442faa7c88a3d54ccdfec : FAILURE in 1h 29m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ffe8ee1851343e7872d5d8d4f9c0c0f : SUCCESS in 25m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b8e3824112742a6829af738c153c6c4 : FAILURE in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cee5a9d1ee4c4ba604266aa1a97485 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b1a8acb5814c4bbc3f7d01ee988be6 : FAILURE in 12m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df6cc5e7ccac462fb9ca43c4e12b60e0 : FAILURE in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4190a7e3682f4aaa8c50c777bef04543 : FAILURE in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acef42f313a44bf7902179a137c31457 : SUCCESS in 6m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cce063661d7143a8925e03071055358f : SUCCESS in 1h 00m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b54c3d2766d4289aeaafaa056c6b97f : FAILURE in 1h 42m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f94aa83eb6c44c2883f29b922c0c9b1f : SUCCESS in 47m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d8f6517a173c4990acf247db9dc19377 : SUCCESS in 40m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9d981e80deb049a9a4dfa01f7e61aee2 : SUCCESS in 1h 20m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/de44c39f34d846b284f71b52bd2b1bf2 : FAILURE in 2h 39m 27s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/28c3cf8658864d4688a185f02ab3d63d : SUCCESS in 48m 59s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b24401f1e3b84ead804ba742678dc125 : SUCCESS in 4m 37s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f2d37c8ed1d4559aac5ab9452941ef4 : SUCCESS in 34m 25s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b49cfee24f8e4e999e456ec5737511b4 : SUCCESS in 34m 25s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1f6e10db5d894f10bd9ccd2777e15057 : SUCCESS in 31m 07s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/598c960e729e435faec310eb225bae82 : FAILURE in 1h 21m 53s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/214d89352c37476fb60d486a1d586c1b : FAILURE in 1h 27m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f56290afe97e44388cdcd7d24285e302 : SUCCESS in 1h 19m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7131a6a4e7f4486b7f055ccc86a2185 : SUCCESS in 58m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f2205648c69547a2a0904329993f105c : SUCCESS in 17m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9f1b7a913ef2410ea53d7115e8d9bd1a : SUCCESS in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee3c66abd322ea07d1a1061bd9f75d5cf9e47f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-19 15:17:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5601947d76583ce8aa00cff25edef795dc913839","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-19 15:39:15.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-rh2n2/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"fe73b6f6ab2dab68eb4143c4b2d2933cfd4b152f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:11:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b70520a10b34a98b43e51746bfed79d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a2484f8304a4c2ca67b6d452d20c214 : FAILURE in 51m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/504abb53f4b5442186cee7c2f8a3b1cc : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5d048837f86b53fcbcc057a69ee7fb40873d2d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:49:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8efd308a20e74be99a8a302e12177004\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/539f84206b81420f809678240b9dd5d8 : SUCCESS in 52m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/336f04bd44b743f1b1fd1eb9724f3581 : SUCCESS in 2h 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e621add6ed84b24a14c6f8d9ffa89c0 : SUCCESS in 22m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc0652d828154904a025194e112826a2 : SUCCESS in 33m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8787a8230a448ab2f07b21a38c1528 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94ff641be47b4bceab015ccf7d8b8860 : SUCCESS in 15m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f309bf1fbaaf4f769ce591a845102504 : SUCCESS in 9m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc136147807e433bb1ec3a83c7279fd7 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5b962c201f42dba38abb6767d11751 : SUCCESS in 10m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de856ba55b764447ac5cd900f15ef8e3 : SUCCESS in 52m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/495b6bb1a1024a09b98328f32a632cb8 : SUCCESS in 2h 14m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/abc219ea8adc496d98e9400d0ba452bd : SUCCESS in 52m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ae65c0bbc21e4d69af51614e25b27924 : SUCCESS in 55m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7559c41d8dd844f6b3cb6afa254ebd57 : SUCCESS in 2h 00m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/9e823c106d0f4a9d996f26b892d7dc92 : SUCCESS in 2h 08m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3f6c4a30c87044e1aaeb1aa04195a39d : POST_FAILURE in 48m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d13d47ae07d64843a6c17da574f59eb1 : SUCCESS in 4m 22s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8129af8aea99485281dee0faabd302c4 : SUCCESS in 24m 47s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c55026d9a4bf455cbbc70a7315372132 : SUCCESS in 30m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f125b2e38d9d47a28cfec2ed6d25083f : SUCCESS in 31m 49s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8b6fcee122bd4e6c865d89009b14c559 : SUCCESS in 1h 52m 33s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/3ca8203db4d249339ed4ea07d0676e8a : SUCCESS in 1h 51m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a69b7e8855d4284ace6eba98b6c00da : SUCCESS in 1h 27m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10833b4b13bc4f29bb3b5934dd8f95e2 : SUCCESS in 57m 06s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9995d394a00f411fbc5a3cb5fdcd504d : SUCCESS in 30m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a314e3dc8db54e398e446c0d39305cf7 : SUCCESS in 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf2c846a15209e73dd5441a261720e67f2ddbfb4","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-19 19:27:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8e7b9a9ac83fc16d577b6549ac4bc3e4ce3e318","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-19 19:50:43.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-pn2mj/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"07efe1f9c46a1978d1115789777f258f246db96a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 20:36:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835bdc978e7b4926bb92b5a2282c6163\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08423b364c7649bb9559a8b19d490cf0 : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1dd707950bda48bcb0e50e77eba4f0c7 : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"25d36c402f907a6827c1a98b122ef40dc16f5ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 21:52:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/746361e48a8049ea9311b0a4b53e1893\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8438eb96aebf4d748b2c931b07358f7b : SUCCESS in 1h 04m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e24c3316845b46679840d42537ec327f : SUCCESS in 1h 19m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22b77927b7924f89a387e2c7e2510e83 : SUCCESS in 31m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6058819a365f4771bec25f478c7cc7f0 : SUCCESS in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8002aa8c1e47068a21c5173bde729c : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023e5f049f9c4023932ad717414648b7 : SUCCESS in 14m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c70fbba64e5643a78550a7524f75b378 : SUCCESS in 11m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca32535f309e4af2af1779e8b8fe5834 : SUCCESS in 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d30d9ea0eed4171a9d6cfc96aeab7c8 : SUCCESS in 10m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/82e2f9ff2ec64208b4773dbe3af7ab1f : SUCCESS in 54m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2f58598a876448e799c363d14e46ad8d : SUCCESS in 1h 47m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e67b7097dfbb494684aa1f3b97bb162b : SUCCESS in 46m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2cef19cff16443d9adf7b04a08eaa777 : SUCCESS in 57m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b47d4f2f8a4946eb8a3eab72d1d04026 : SUCCESS in 1h 26m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/0882a8d52ba64d7fb3125dc5f1a223fd : SUCCESS in 2h 19m 08s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/09da5e4145814930aba4f54151731c41 : SUCCESS in 46m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bd062307c2fb46f09303c83214b9c9f4 : SUCCESS in 4m 56s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6ae0ba5b6c34da28aa2639c4c5dfa54 : SUCCESS in 31m 29s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/32ad25a44d6549588644580df438533c : SUCCESS in 27m 14s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/044b4813593545bc94bf3fc623db6615 : SUCCESS in 37m 59s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0224e6f6b85e4c7ea9f96292ea552299 : SUCCESS in 1h 52m 53s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/899e05d0dd1342259b09cd3d9386eb52 : SUCCESS in 1h 56m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/91d87c6ead644a5ba486f62eebc12ec2 : SUCCESS in 56m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e2d80bf20a4451baceee366547bd90a : SUCCESS in 1h 03m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/33f0fc0161a94b018d2ad6877c33c349 : SUCCESS in 27m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ae21a5d843774574809aa2f9c6b018b3 : SUCCESS in 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc1ed15ba5842f9b0ef033d8d23be9dd049c0ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 08:28:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"987a94c6545e7a6892f6b1537db11a92f5b4d751","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-20 09:03:04.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-4jhwc/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"e89d62db887f21423d53866e842700d837262255","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-20 09:48:53.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-jlmrj/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"88e41fd4c110948edb38a1a4b9906c5f418f2586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 10:31:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42c43a93359d4746ba9ee1ae3a5601c3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f1b046a21ba4a41a9ff9880397a6abd : SUCCESS in 1h 05m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b76d24a08ded4c4d91f589335d21b1be : SUCCESS in 1h 47m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/343050f61b79498e963c56256717fe9b : SUCCESS in 30m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3973a36a9be941af85bf05c50fc61bae : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59accddbc1ee4cd6aeaaf7663719a7dc : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9ea5c0e6593486fbc805eb765239797 : SUCCESS in 13m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3900ab9cd7054943bc048f9e9f1e17b8 : SUCCESS in 10m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf8f66fb685f49448acd59b3176ca1a7 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a788f65f4e0b47a0a8244e5d5020a49a : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24ca9ba0b1ae4c3795d5bb11f61985ea : SUCCESS in 35m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/882d96e3ae1e44309833a45e8f01e38d : SUCCESS in 52m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20ac574089094db4b1d21a9ac4f94f91 : SUCCESS in 1h 09m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6ae7e2b8d13541f5b9270ad7de7d0057 : SUCCESS in 44m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bc76042f017441018e634063f33de1ed : SUCCESS in 33m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8b6a4cd40848498ba5ad2a3fc520455e : SUCCESS in 2h 01m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/c14ddd3bdb384d0697a333fc6a67b5ba : SUCCESS in 1h 50m 19s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b6d45ec49e9f4cbfbb31416c8669266c : SUCCESS in 39m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c15e5e3643cc4620947715e3d2a2dd2b : SUCCESS in 3m 22s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4c67ded0e6043f5898df689cd9a1159 : SUCCESS in 27m 50s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e183d43a79604b30a5beb3ca5624c275 : SUCCESS in 32m 43s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/86aa9022fe414c3793b8113010aed276 : SUCCESS in 30m 40s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f3876c047f1a4f2d92539f685768f526 : SUCCESS in 1h 19m 03s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/f97b424f7f95498a90a18340390d27b2 : SUCCESS in 1h 55m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9856e7b3119c464999ce6ad0a5984d64 : SUCCESS in 1h 20m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/183be428fb3949248cc7ee56be8ab61d : SUCCESS in 58m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dd5ce2b0c8c44a2081cc6a3e2354d2c1 : SUCCESS in 30m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4e1fd4e8ab7f49f89a0de701254dd9e4 : SUCCESS in 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e47f245a9ef142296ac43c3ecba9ceb929f9bf77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 11:02:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9be9a92d3a3e4698ba5c1694728cac40\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b5a06b2e77e41e8b429b0c10a6cea77 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/91e9a6e2672843a894c6e3400db4f979 : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"57342d7a0820f1ecdd540a34b16b7f71dc58908a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-20 11:58:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d201fbb79a83f3809a64e16019fc5e9103b1b6c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 17:25:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"73149ea89c1dbc874b62c1cd3c1a2cde70a33817","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 17:27:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"41d11617f82dd16b22433c352070a8bbd378beca","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-20 17:29:13.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2a3410400680ea00b0254f67755b5b1465c5c94b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-20 18:18:27.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-xp294/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"cfbcd9e3aae405c054d3c681a5ebd69be1e0a6c4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-20 18:42:39.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-nctg4/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"32b0307ebc6314041dd77562d1114b675a26f607","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-20 19:07:20.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-t628m/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"241df2c407a4d0d51fa7522d4b475a2e3c8ec971","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 20:07:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/103ecd7a1278436c83a71afc5d7e7582\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/234c70ed56274fcc94b9d9ff2a7d3121 : FAILURE in 45m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e13b1b76a4ff454ab27d0d8d8ac77113 : SUCCESS in 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6acd32775b77e7c60fab2c24ffc5c58d82946613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 20:10:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7204c2717365465b9ecfaf967b7c9991\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f5b409aea2514eaabf770787c4b3f5ff : SUCCESS in 1h 00m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/35c53cec3efc4d81abdeae2a734f1d52 : SUCCESS in 1h 44m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acf8a2693f4946f694ed39354462b198 : SUCCESS in 23m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9770e3650b084f78b0c17a480979cf3d : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acde003280c74ffdafe6c936aca0775f : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a95d91420eef4feeb1053b32c7c34cb6 : SUCCESS in 13m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb0a1cf3cd07478ab6f966b7593e31e4 : SUCCESS in 11m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14870cf1ec9b4560b0043e09a781d07f : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945220a63255483088b3cd579895ebbb : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eef3544648704b73a83c620bea281bd2 : SUCCESS in 34m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/abca6f5341894675a004f36ee31ba7c4 : SUCCESS in 1h 08m 37s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ffff4e6b46c24207b6167ba8dcc26b9c : SUCCESS in 1h 39m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5f14b401e86f4ead91a348a87c8140d4 : SUCCESS in 59m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a8b5881c5de941dfa150c8702c558af4 : SUCCESS in 50m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/06e5cf3c3b6d42a286d4aefa6d4af322 : SUCCESS in 1h 44m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/30ebb1052c2349bba4e6a759244898bb : FAILURE in 1h 28m 15s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3ce3da14f4b6408c80d658a5d8b288b3 : SUCCESS in 31m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b59f11a0f86f43efa0bfe4b1fa74c632 : SUCCESS in 4m 45s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21242a37f4eb4a4aae02f1fee66bddaa : SUCCESS in 20m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2de80b7342af488c859dde0412591b50 : SUCCESS in 26m 25s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/86ba933e13664af196df6c8ccc4407a9 : SUCCESS in 30m 23s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3ad375e90fd481093e35f491104e44f : SUCCESS in 1h 46m 13s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/46d5d5dbbf4d42e4a0a12a3bb4216848 : TIMED_OUT in 2h 34m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/58554df9ee88433f9b2f12f5e6ba34fe : SUCCESS in 1h 24m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99931222bc6243bfbf33d96a8cdb0217 : SUCCESS in 1h 00m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/aba926eecadd496888e5591e129d17f3 : SUCCESS in 31m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d493404b80fc4dc6b84452f17ba2d045 : SUCCESS in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c6a68bf3c491ea322321a04b926575f37e411373","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-22 02:29:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0767a790071e3b24c8ce351bc046771bc9defe10","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-22 02:52:20.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-xpjr7/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"f2d84063be1b4db41d24234ec1d56271e02a921b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 03:04:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676e5e83bbe84e718f6deb99a49375ec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33741171e99941779b01c68d098e7cbf : SUCCESS in 30m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b597f7d8a28e4388b4e1124e64a3304d : SUCCESS in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"147557f2ce8b2f98f97feb1f9bd896677889a66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 05:02:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b2e38eb442244c19534cd5530af360a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6acf1767ed4e4bf8a919c0ccbd3d1b7a : SUCCESS in 1h 06m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/670516f84cc94316aa37d5d4f66c4d3a : SUCCESS in 1h 45m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80c92afb17da4cef908d6332ff0c282e : SUCCESS in 28m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b159024205e429b8f779fc24c87f15d : SUCCESS in 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4fb6e81f02542ca9f0bd7c89f2c2803 : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61dd2006c6cc4b3cbb41ea28966eecd3 : SUCCESS in 15m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfc8eef010d04dea97cb6bc314ff8ea9 : FAILURE in 10m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebdc3f33b988471f8bdab49196dcdf85 : SUCCESS in 15m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95efb7574aa4516b794ee3ed6e1668a : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/627dab4eb31444468dec784faa8f69be : SUCCESS in 26m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8f36d324abe1459abbc12b2c97c65eb2 : SUCCESS in 59m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b39aaad8b46741e79601296f3ed2dc58 : SUCCESS in 1h 37m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aa79ef6d9cf24179872ec6d9629d0058 : SUCCESS in 46m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/35075e4a08fb4dfa8c1e5085c9f2a97d : SUCCESS in 44m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b739cfe98d2940dd90720b8db28b62eb : SUCCESS in 1h 36m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/c71faff259af48a1a80de7098c20f748 : SUCCESS in 2h 23m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f5e4083231964015ad274ed3c1589e69 : SUCCESS in 51m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bc1d042733964b6cb5873d72d49a0628 : SUCCESS in 8m 01s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/609c668f3f704dd399b53722065e0545 : SUCCESS in 23m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/80db5e79245c46a794600f05733d3b2f : SUCCESS in 26m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b05f8be5ef694919add6d97fdc1c3f34 : SUCCESS in 30m 30s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/123d2e6c2e6f4b74bbbff42a9c6d63d4 : SUCCESS in 1h 36m 40s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/a98841b748d2402d939f2ab1ebe4b062 : SUCCESS in 1h 36m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/12329ce0f4a64c36b1eae588918f2be2 : SUCCESS in 1h 16m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb533c46252845b9bbe3bbc921b82419 : SUCCESS in 56m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c34cbf2915fd43569dc31d77506765fb : SUCCESS in 19m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0bf21e397c134c099d702801f62ff3c0 : SUCCESS in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d1336b65ebdae80e81c5eb66de85f415febf7171","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-23 03:55:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"05a9a5c165c5b1b92a6a7c451108831901efc1ef","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-23 04:17:17.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-855x7/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"1bc0bd4c568e138aeba9243dc93f2e34e342e2dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 04:25:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b95623378c74e179b4cc3f6e9f772bf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35fcabf0a71b4266be6a48fd8af25be9 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/59f10bcadca54ea8be175818ee2c71bf : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"06d5a5548151a66b4e2082e21b8e5683686a2ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 06:22:48.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/751d3c25123f49d8be13dbaf5db10e71\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2476dca759374cb68b3631941456ad5e : SUCCESS in 1h 08m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8228d71289344268aaf00bf8feff8864 : SUCCESS in 1h 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d18624eb42f74d38b18c1d9831e21218 : SUCCESS in 29m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a05547ea26034681a56aa4e3b14eb29c : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ad1331c7d34fb3bd6e8d0f6243fed0 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/073d68e5f3ad48a6892d0c14a0111080 : SUCCESS in 10m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/267abd2ab46b4b7d8fd59f282fc85c44 : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/385ddc3a4323437fba1cfad4a2ad66dd : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c58212be0984178aa6931672e96c120 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0048509f2511499cadd72b4a41a48d73 : SUCCESS in 26m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0a44081963904f1ebe27706033e2de31 : SUCCESS in 56m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e5713be47f34a1c9b4a61bb5f9baa02 : SUCCESS in 1h 14m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0c7af707a514452cbe52d0ede2891883 : POST_FAILURE in 22m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1e611ef8fa394ba687210a6e03784425 : SUCCESS in 52m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/64ea47e987f0437485c90f26557afb19 : SUCCESS in 1h 53m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/5081987dd3e9446cbb32f513caa44a93 : SUCCESS in 2h 21m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/802ca18cf9a049dc87952d5c7d2496a9 : SUCCESS in 52m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6e9a007e907241748df3927a1d3d6b95 : SUCCESS in 4m 42s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c612d93625ab4d68850aa75c7eff70a0 : SUCCESS in 26m 33s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a86ed3468e99453c84848bcc2061f78f : SUCCESS in 23m 44s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/642e2fbc19f74a38a954afc46f6ec7e1 : SUCCESS in 34m 17s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/538cd87cc99e49f2ac396f573973f5dd : SUCCESS in 1h 45m 36s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/b66f1078b98a441fbfbe4d312594362c : SUCCESS in 1h 55m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/59994c28d3ce41e086e0585ced55f878 : SUCCESS in 1h 22m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f66c3bc1fa1540a0a5f746097b1d7be1 : SUCCESS in 34m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a0c62dcc2de2487687d9ba4b6780dd61 : SUCCESS in 17m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b7cbb8fbd2b04287af474c9f43a17366 : SUCCESS in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a63a497e66cf7bea58e895527c87d7297a3bf0eb","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-09-23 06:37:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a9fca6109b8e07fe7dac95593877d39e7e9fb2bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 06:57:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2a1382a88044de586813fee32e1b342\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c78e742754d846a7ab0107dc8eaf3b5a : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ed90c6d025f4a3b86adb5c5cf620124 : SUCCESS in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"891dc638128b763d59fef370b82f5107df7b3eba","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-09-23 06:58:40.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-d2qj5/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"dcd36cdca93bcd3c946da9824e88b16599548adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 09:08:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4edc81016f45f087d511143a644142\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4207a666d4d4e889dee1487cad8fda1 : SUCCESS in 1h 08m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e121a068aa3e4e199d9d9718ecb03c6e : SUCCESS in 1h 58m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6dc06d1e9524fa9909f770b8bebd301 : SUCCESS in 31m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbb69f80e3d94fcbad4055c07c9a4d00 : SUCCESS in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd982acc05a4c8c98ba4c21676adaaa : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2593434e403349d7a9877a68443965fa : SUCCESS in 14m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e4891d7cf584c95ae8f6e15ce08b6bb : SUCCESS in 10m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70ad0ac1881146f29be904522aaa9a79 : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faadd0d420a54e2e81bbbc4399641eba : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22e868e963840f79faf8e4389087204 : SUCCESS in 33m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3d7e681a26c5469485d95fadbc637a76 : SUCCESS in 47m 51s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8de9a41b1f0d425ebfd1419774c1808e : SUCCESS in 1h 29m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c006f47a6d8c40fcb915d386cf1653fa : SUCCESS in 40m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8d618fc07c0f410d8859002c90ea85e1 : SUCCESS in 44m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9b467ddc518c43f8afa0b2137bf7baec : SUCCESS in 1h 38m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/f2bd5c9eec9b47f5996916f74916bdf5 : SUCCESS in 2h 28m 52s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/93b34646726145499bc78f550c24b971 : SUCCESS in 43m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3f65c2c06f9f42f790e8978599009c7e : SUCCESS in 5m 59s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/112e8ea4539344dd9f9a9d7bb5320d9f : SUCCESS in 24m 52s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/034ecdc7dc91486fb7caa4e0e8780272 : SUCCESS in 25m 26s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6594386a592741b098b975de76d320c4 : SUCCESS in 30m 22s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/182c2188823e492daec196da0c72b000 : SUCCESS in 1h 37m 45s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/4317696c12af4bcdab2909241f730af4 : SUCCESS in 1h 48m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a814ce46cc348f8b7c149bb97e138cf : SUCCESS in 1h 01m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed4ccd1c005343aba2704c2edfce1f36 : SUCCESS in 57m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c37af17b6eae436e8db0c5e883950ac0 : SUCCESS in 33m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c6b80ba027284e9fb321f44dab8f1aba : SUCCESS in 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f326752532c5f4b7d27e4c63ba6cd438a3203ae","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-30 16:36:47.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"15d871dc08ce502b500f5d77c4a9807dd90e3555","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-08 17:53:01.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"78dbccba6807643bf53e7d037c25c0fb8996ed0f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-08 17:54:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b112127bb784ec6052231cb32cc6544f580b46f3","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-08 20:24:25.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ea9b4fd1a513eb8966ca41fc659475078d3c5095","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-08 20:25:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"21de2d1b6651910d247781fedd1e19725e2c5c35","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-08 23:16:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b4fc97a6b0d3b7f968a9a91539fa2a5ac380cac9","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-08 23:33:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf4649c4854f8328280e3da60e4a2a64dc62b25d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-09 00:07:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4ea39b31243801c962ad1f6ac6fa60fb593d69d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-09 00:23:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"11974505937abc12f988331b6143924f75d8d57b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-09 08:20:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac9f633d9b56897f139ccab7c8897b096242c9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-14 11:12:07.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"bb93c1e247e7620dc7132fd620aa843576873071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-14 11:18:21.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"8c1e5a709a6f79a4a595102986ef97521f254a79","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-14 11:19:50.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e789f0eca7849268c83531c3db75e83380d1a1f1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-14 11:38:25.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-nmbbn/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"1a00cc18637ebf24b85bb1b5007b1364e6467ced","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 11:43:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c513f0042f4d3098e234cf0b80a915\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8450d13ca75d426e97cc8d7f2c2207af : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e768e33d0cf48e987c1d5117ec752f8 : SUCCESS in 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"42cd26ef01fd4e1ca5d566472b1e27f832023f95","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-14 12:02:53.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-4rh5q/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"eb879b13bf8361b59e117e5cd0be4e33d8363dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 13:48:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a5da585f4f84325b73b814b5a88892f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f431d85c83e42c798a92a6671854e00 : SUCCESS in 57m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/acd4d2f04a6b49fc9921954d48e4452f : SUCCESS in 1h 59m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b7e78f1b2934d3eb249bb7e82106b0f : SUCCESS in 48m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df64fcc475c47918a1217f0bb23a273 : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e261f690f04388aba21770d07609a9 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3c94af9af8341cb9ce6f37998ca3da1 : SUCCESS in 13m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/130eddd93ec74e7285c7ea8a866662be : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f369ee81bb4e1e8f77c61f07d0c048 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79df513724b14d6a942c89341a6f6ce8 : SUCCESS in 26m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d703ae13bfbc40ad9d499ffd392a9565 : SUCCESS in 47m 43s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c04d97668574cd58a4ac7b3fa4fe0f7 : SUCCESS in 1h 45m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/27b44dc5c41f4808ba8d25b4cc93180b : SUCCESS in 42m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0cc901c7be0b4e9eaf0efbb6a2a07fe7 : SUCCESS in 34m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ef602cf6df564d909ecb5784a417e797 : SUCCESS in 1h 38m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/61928cf921dc413abef3ea4273745d0a : SUCCESS in 2h 23m 23s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/511578756fc54af98a00367569f131ce : SUCCESS in 32m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/71230da4b8b74f329f13d6b6fff97b30 : SUCCESS in 4m 57s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aab77c7856c64ecaa5a96e2dcb72403f : SUCCESS in 26m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0e45baf87eb64b1fb493c9a464a4a879 : SUCCESS in 28m 32s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d7137c851ea402daf571f7568e6ddec : SUCCESS in 31m 05s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c6076e783066456695a0859e262bcc34 : SUCCESS in 1h 47m 25s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/bd8287564684499fa3661f3779d173c9 : SUCCESS in 1h 12m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/802c3ae9163c4e449afc71eb2cbf212e : SUCCESS in 1h 21m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/425b08ce327b47218654e3e1f5e96bcb : SUCCESS in 38m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cc16c2d6f8ae495a94d8ddb95946ad2a : SUCCESS in 18m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9f721b5fe3f840729474ca933a3293a4 : SUCCESS in 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0f07a8a2f36ab786b79d9757ed971530d454ad17","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-14 13:54:02.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"62ccc9a6a8aa6329dc2fe4ce9685c7eac2f87b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-15 16:47:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"dad25db41e455a8d949d106c7edfc7bcab8c570d","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-15 16:50:44.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1b26280b25b8bb0b3d6049ec96b3a38d51908a8e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-15 17:11:17.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-tdznt/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"5993c2a4c6d011a78e7717690fa4d6c6b6937987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 17:27:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3ac6ba76ab4be389c07632da10f356\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2feb221883a14342b758564841900dba : FAILURE in 38m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f33a731d5894b63bf2f19eb0c9e4d9e : FAILURE in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c2cbaed45a6984b4c806dad04874002abd1d48f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:03:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ce1fb69f31d4aa3a21fee61aa89a39b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a3e085bb716468e9409d2310c6ec1c5 : SUCCESS in 54m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/696fca6e22964b06bae8fcfd52999b77 : SUCCESS in 2h 04m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c6fe2d048d14ac1bd114c839c09df06 : FAILURE in 38m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a08bfbba39544c4a95771f192cddbced : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c93f5bbd3c4c65a9fd50a4ae1c4b5f : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6192296b41d24a32ac23b21a16f10010 : SUCCESS in 12m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d97e46e149b64e59ba7f8e842f01db48 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958998b2aeca4ca0a8d1e04ce382fa15 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01b2dc7dd1da49a389fac6e534d61ff9 : SUCCESS in 34m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cc434dd350c24868a67ed32beab45b26 : SUCCESS in 1h 00m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/41a98a2dbccc4d7aa9f839e4d2860d30 : SUCCESS in 1h 11m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/76e9179b78ea4e8cb740e1ca7f65aa15 : SUCCESS in 34m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c9b88c7b6dce4a2591803d7b3a7c5fee : SUCCESS in 49m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ca85baa550ac4fddb5c6fe456f69b9a0 : SUCCESS in 1h 48m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/e8a487986c824998aa5bdb215d9bc52e : SUCCESS in 1h 41m 45s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9e385d7fdd04433aa37b1960bc9f5f4f : SUCCESS in 45m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b6a25a3fd9424646913d859631e70578 : SUCCESS in 4m 56s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9778c45a60984ee0bfdc058e8edcae1b : SUCCESS in 23m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb1f2ed2b36a4fc0943a916672045fcb : SUCCESS in 25m 54s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/eb4ff429061142bca65288949e921560 : SUCCESS in 32m 05s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2dde05621e934625ad76cb6590881beb : SUCCESS in 1h 12m 27s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/6927357c76bd4df99d0cb848f774930e : SUCCESS in 1h 50m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/604ddd696813439ba48b79b920414d8c : SUCCESS in 1h 21m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca5a15a02cfc46dc91e85ff3d167c1fc : SUCCESS in 38m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/411e7ca6ec024448bdcd8895319c68ea : SUCCESS in 31m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/656a6b467e494663adbb931f4fb92fab : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"24ed7b86bfedffd8ae394b3df2f40bace161aba4","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-16 03:30:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f3b58774580b5c6373dda7c2d42566ca69dceeee","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-16 03:59:33.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-whjbv/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"91f7452835e1da5a819174c673b581c98cf71fcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 04:15:25.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d3936476ce14c2aacdc95adfdb0c116\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6d55b0232e840ffae7ae25c264ba8e0 : FAILURE in 39m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5290b4981df9420a8a78686d43463e35 : SUCCESS in 42m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"de2a6181aa3d25588a76dfd0550404502d29503c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 05:27:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb2c1f61a7242f080ede8295409a42e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dac4169e7ca94d0894ae90d198eac7b4 : SUCCESS in 47m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d96cd83479644ae59682923768098f65 : SUCCESS in 1h 15m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ede0f02402c148f1989aca957c9127c3 : SUCCESS in 1h 05m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea284eb28cbe4736a522361a1b99f833 : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c46dfbef30433aa241beb3c0f33f54 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf2b7b9aed38483ea5f1b7d6fdd53b83 : SUCCESS in 12m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/867b756e25ea4fe3b0bb85c2fc9851a4 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbc8638532e477a83bc1b600bd74e2b : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e288e7f86b456fad423886bbc5d8e0 : SUCCESS in 31m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/29d28fe319194c19b4c14a6c58f78f6f : SUCCESS in 37m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/633f9d78afe8423bbcbebae4e09e1ef9 : SUCCESS in 1h 33m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cae80b01736a45939315daa2949391b3 : SUCCESS in 49m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dda3bc7dd6a54b56999a254c215258a4 : SUCCESS in 42m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f7d5020d039742fe8ff5d2e3ea3bd4b3 : SUCCESS in 1h 50m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/fd487fbfd58f4209ab6d08b799be2a0f : SUCCESS in 1h 40m 54s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2b3b082b96b0470286906fa001433580 : SUCCESS in 32m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d0ed1405ec2845128d675c1419c749d6 : SUCCESS in 5m 21s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e13b0869384140b8bf17f454b70e4c04 : SUCCESS in 26m 16s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95d3a380189a449f926a8a19a57daf7d : SUCCESS in 29m 42s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/34c959b78da646edae04748cd3747380 : SUCCESS in 31m 28s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a8f553554fe44033ba622fc639b3fd10 : SUCCESS in 1h 35m 24s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/a40168c4550c4979955d019059be9a68 : SUCCESS in 1h 46m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/72058424c5cc46b48c3b0d2596df9e5a : SUCCESS in 1h 18m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7598cf2af51f49a1befbf5f732796504 : SUCCESS in 47m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2fc6eee27a4c4fc38485835b154b0a63 : SUCCESS in 28m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dee1fe382909487181b3a0f5094fdd1d : SUCCESS in 23m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4cfc2406a2ca17f66353cd2ca6d7eed0967ed620","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-16 05:33:42.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":8},{"id":"f8054c21d8a72dd94755fc14bfdf9954b5ef315e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-17 01:40:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"75ac3005ae9e2a2e77825c56ebbf747e1a0fe289","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-17 03:54:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b49797f26bcbfd91aa1063b1dd3880ba3ba1ede2","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-17 16:00:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6afe7fb0a422ac904e55ae62d158f300b507e577","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-18 01:37:33.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2ad6972b187d2fb4aac41df4d2298fff2d6ac085","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-23 06:32:28.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e1512eaa1cda992640e9b3d4413a6806f18a4ff9","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-23 06:43:20.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"456e0905233e8518a86793bda4580dc98e218b05","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-23 06:58:19.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-flkp4/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"343b4f8db4bd08a22779583b8f39dbd53dc18b56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 07:17:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b1850ff73847e088f960169c93b39a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06f0b17760824970bd48188a95857251 : FAILURE in 42m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f618f19c0eff4f96a60d41fe30fe4352 : FAILURE in 37m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"554a047310826d93c44ffa6c6fc4e2dbcde956c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 09:06:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/640b57bfe999467c90a7d6f71c4b8758\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13109f076bfd4413bb915bb872b60d57 : SUCCESS in 1h 00m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/150ddfb2846f446197a588090559c6a1 : SUCCESS in 1h 48m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31d9104ee4d945d5866eea1058f84703 : SUCCESS in 59m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5157dabb6f72424fb7c950f38a76666c : FAILURE in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c38a1542cb4c308bbcf9f2ebb57089 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebd171d858fc4cd3850c65c852cf6b38 : FAILURE in 10m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7454b183b1c64bb9af9f1a496063367b : FAILURE in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026a67909b5f4bc09ec1061345f44608 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe276e35179141c881978bddb7acb350 : SUCCESS in 35m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5c08be9725384735a7c1ccb8836515a4 : SUCCESS in 1h 04m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3fe40845a24e433b88274a604d74683a : FAILURE in 1h 23m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1fb4788ab08d4e2c9f7b40b750fedc5d : SUCCESS in 49m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/14e062c874ce4fe2966b7e2b887d48ca : SUCCESS in 44m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/42f11517cbc54693bcf93de08511dfaf : FAILURE in 1h 29m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/850a6cfd52a54e4b882423b0446b6564 : POST_FAILURE in 2h 19m 04s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3ed6e058264b4735a8e620329d5c88ea : SUCCESS in 35m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0581b21973a947a8b0c0b968569a4bbd : SUCCESS in 3m 43s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/267134198adc4b50aca11d07e0185802 : SUCCESS in 29m 06s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9540cff4aae8403e97ac616a7e10d805 : SUCCESS in 31m 03s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fbacf9ee692340288198410b063b2eac : SUCCESS in 26m 04s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6412f4caef1f4c4aa7ff0471df022a84 : SUCCESS in 1h 58m 04s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/b7674ea0acae4b6abaef95703f8cb42b : SUCCESS in 2h 04m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/49d5481e96724e40b75711cbfc3449ee : SUCCESS in 1h 04m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73dde477f6ac442c8af082c4ffdee3bc : SUCCESS in 41m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/616e7fe63ab34475ad03757ca9599801 : SUCCESS in 20m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/529f721acdad43bcb51bee87755fa211 : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"35afd82750ded7d2e68019979d747a51362a7aa2","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-26 00:53:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6c1218aeef2301dacc42d5c3f4531b6e1bb7b467","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-08 12:07:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4810193bd93203395da6048b20f3d82879555405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-11-12 09:35:28.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4865b6a1e39b7f3094856afcb7eb96734b9cee56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 10:05:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d961702872f45368c0013a0cd755854\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c5b188676104f20bbc70cb33f1b1038 : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d0e9a58af3d24cc7993421eebb55e35a : SUCCESS in 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"81b28143272f8338a66126595078f524b5640919","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-11-12 10:08:23.000000000","message":"Patch Set 10:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-wr5dz/index.html\u003e","accounts_in_message":[],"_revision_number":10},{"id":"850ece8b52fa89305702c8d7fa173e2e63164d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 11:52:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3da954194d4437ab381582132644f1e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab1c264b34a04f7a9e2ccbbf06eee5ef : SUCCESS in 40m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d582fca88bd64e18a300bc76a3727952 : SUCCESS in 1h 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e6c83d7a4724377bbe909b9d72fb0ee : FAILURE in 38m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d715492576c84d929eb919522038d763 : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/871914a898074a4c8a4cf627a65a19d9 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96d282d8b47e472489e0356b0edf5324 : SUCCESS in 10m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecd5ae4560b845bea663c3cec303b167 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce46815e5dc44fada74cc1c091dbf7a3 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5878d552f7a42c5b2d351e8450eb7f6 : SUCCESS in 40m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c65a51fda0d74a9085f0ed916da9ae67 : SUCCESS in 59m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/105c2ea1507f4ca8bcb1c364d61db73c : SUCCESS in 2h 06m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a5545afa37cb43bf887d078efd492277 : SUCCESS in 43m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cd9164d5f9c7409ba70a3ec8e3d78375 : SUCCESS in 47m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/887cc620f487458989ebec92b853cfdd : SUCCESS in 1h 54m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/4ee90d6bde944bb2afbaf7f89e7ff240 : SUCCESS in 2h 11m 09s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d57e47b056d5407bb1731aa8483eff51 : SUCCESS in 31m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c71b53c215a54f5c9bc5af4819e848bf : SUCCESS in 5m 38s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e12dca410c14a829895dc63a2a772cb : SUCCESS in 29m 35s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/907f80d6b3ee4f0bb0e2af69efcd9ed2 : SUCCESS in 24m 57s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2e01774be5e1481ab59d9b1ab572c222 : SUCCESS in 27m 52s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e8a926c93b5a4b1ab4aa323c0552ae85 : SUCCESS in 1h 23m 01s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/40446a91de0f4884aa525a633c80f745 : SUCCESS in 1h 41m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/af5a647ac00341aca0deea2d0fbfb7d5 : SUCCESS in 1h 11m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29a47fe41adf4421a1a3ecd8c6e9a32e : SUCCESS in 1h 02m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/585d682aa1f54bff8c952dce5cbb5601 : SUCCESS in 16m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7e4c5c7b671048188db2f72a39abaed2 : SUCCESS in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2e2c91ef60044978fa5dd1df6aac5ba499b05388","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-11-12 11:57:55.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bcc7180654032c3a9d71a7282fd0d2f8821ad56b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-11-12 12:27:13.000000000","message":"Patch Set 10:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-9p5j6/index.html\u003e","accounts_in_message":[],"_revision_number":10},{"id":"7de2254db1484882fe3ffbb325104334e0f7d01e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 12:27:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e248522500f40edbe2c7f935d101c9b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61e066c6420344f5b418bed4fedd7c53 : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/77476b9fb5544cda9ba4f1a36bb3a5b8 : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5ce697e2609fbc71d536d27fec5e0854caf062b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 14:03:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3583f14f714a489490f8af39db5fa0e6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0ab2463ee0b4c2c803bd66be904316f : SUCCESS in 1h 05m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c549a8d494f547eca3fec8c95017bfaa : SUCCESS in 1h 56m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c77cdd37d9441ed8a5540a50a56fd34 : SUCCESS in 1h 02m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211114a7567d42938f27c654a7d24948 : SUCCESS in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/158825e4522c4ea5b1b92094f03b263c : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8523a262befe4910b6688e757cdb9696 : SUCCESS in 12m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bd2a9f004e9475093ed2294f51d5f60 : SUCCESS in 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01252ffe73ed404689020116b6c516a2 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed07797f5684974a660749955466f63 : SUCCESS in 24m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/92e8eb49bb0a46808f936fc1df87a1df : SUCCESS in 1h 02m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/18fe4a2eafb04f58bf4d04163e8bf549 : SUCCESS in 1h 31m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/411da836dd2741809907973a318a9807 : SUCCESS in 49m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5f8dd9e27287455dbe57f9e0d29b10c1 : SUCCESS in 39m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2bc2cbee071f46fd9cdfbaddfe6584ae : SUCCESS in 1h 58m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/741a71e160694de29260b4eac9460dc5 : SUCCESS in 1h 54m 24s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/03b2eeae3a0343cca587e5825c542a00 : SUCCESS in 48m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9313038226e24e238b007235ddab5f1a : SUCCESS in 5m 33s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ecc8558cc3e7420dbc27c5594118723f : SUCCESS in 22m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/381eeffaff574509a13d92b56c970b6d : SUCCESS in 22m 51s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c8a31b1d251d48e18a4600854c516f2c : SUCCESS in 27m 19s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f23d6f657b54ae098dc12746f3665c2 : SUCCESS in 1h 47m 40s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/a11fc81f94854fcb87e2e3d7044bbf11 : SUCCESS in 1h 37m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/79be7902bb2248819da69ad020707f24 : SUCCESS in 1h 28m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b1fbe61cb848f29139094c54eadf6b : SUCCESS in 1h 02m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/34b39953945441df93077dbb15c01603 : SUCCESS in 30m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8e9547216ed9420d970a08320021b28d : SUCCESS in 35m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"20c5f81f13a5b1443874bfaeca5eb2eb9ae09cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-11-15 11:30:42.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"553e859f5acc4e7e797821f260abb1125bb73cd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 11:54:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b90ba0bc76bb4af9aa323b59d1c68ccf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f31891c2b93e4c9283b01661d4b02832 : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/00dc70323e2f41a0a50d2b819c33ebeb : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4030de8c6e9e98adfa82e72b6c8094105ee507bf","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-11-15 12:03:42.000000000","message":"Patch Set 11:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-vnwzk/index.html\u003e","accounts_in_message":[],"_revision_number":11},{"id":"f72162b1f1b822d463a488605c2e0731e39c587b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 13:31:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aceb25234324710870307991ed272cd\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d9e1e6ea881462d9847b1035c1c13e5 : SUCCESS in 52m 25s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/014843bc4f014023944aac8a1467f81f : SUCCESS in 1h 27m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a313f5f879345abaf9dcdf552056337 : SUCCESS in 54m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db0eea41e3374d0cae1c5d22ee92d4a0 : SUCCESS in 21m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6172908502147e083ab9b69fc7124dd : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dc81154dcbe49638bee31ac5bb10736 : SUCCESS in 15m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65c02716077e41f9817545899baf8c04 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6189c3fb00430f8fa4660f915c3812 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76d4575c84fc49d4b896cf295110e931 : SUCCESS in 31m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4223fe638c004c79a1bf4b667d068f2f : SUCCESS in 36m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/76ce16c8a7824284918c9eb2ebd021c7 : SUCCESS in 1h 29m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1509e2cad2554df2963fe4d0d63d376f : SUCCESS in 30m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/89c460da73b0490196f44686b48240a8 : SUCCESS in 51m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/913d1145d43e4fbdb453d2371d644b91 : SUCCESS in 2h 00m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/e2bb4291dc0d4349bd91a65a9e6af733 : SUCCESS in 1h 23m 05s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6e7c373f0fd94dfb8a8733d91ead4e6a : SUCCESS in 42m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/719307b3fd7c4a6e832423857b31c4f0 : SUCCESS in 4m 29s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c38995a3061641ef9917066d2f2d95c6 : SUCCESS in 29m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/caad8a6873794b778af772c6eff617c8 : SUCCESS in 25m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fe8ee0a923e84c5794b9e3c7a8a3d381 : SUCCESS in 34m 11s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8d99b56d138d4f55ab9db3c72d2d2507 : SUCCESS in 1h 14m 11s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c882ca25f930436dac268e9ca6e4cbcc : SUCCESS in 1h 23m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6645e7744f014d259ccd90a3b0f54c72 : SUCCESS in 1h 19m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a31e306b2b11483db3fa30ca8b356351 : SUCCESS in 1h 01m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/448c67c8632e4f6cb73636624f08975c : SUCCESS in 18m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b1a5f4c528a9439bb2580471920335e3 : SUCCESS in 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a8c1fd9f120c77b2c942d0866b065111ae2623c6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-09 17:40:09.000000000","message":"Patch Set 11: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"a296d39c3ae3681794a38edc7af63980a8e7619a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-12-10 09:31:45.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"e28cf9abbed9cc4a5831c3c960748b1af806e16e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-12-10 09:38:11.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"0cd65d3432cf3a0796a38f17f0b07049d1a845f5","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-12-10 09:40:27.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b04cefd5ac3621ffa5fa1777ed55d9d7240bf8c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:13:09.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0543cd804aba4b79b2897d27205cdef1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f52795fb74874f268683c87877ceb1e2 : SUCCESS in 33m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f73f49649f844a6be3b1ba5bd430ca9 : SUCCESS in 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cb72399ca33e10191d01b46945b6e326db8e791f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:55:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8adb48ed395d412c8aea96a8dc879e68\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c0672133d184ed8b6a05a4d04bf86c9 : SUCCESS in 1h 00m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/61d16aeaf2b54fe8aa981d2fcccf292f : SUCCESS in 2h 07m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09a267d01bb8413395dfe398967074d1 : SUCCESS in 1h 04m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c40eeadb57174c0da8e51e592a14e933 : SUCCESS in 22m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d771dc55de56477b9323e519bb2d16a7 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f72a47f338cb4721b74cbe7250c16b3e : SUCCESS in 13m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/511d0560b69b461fbf8f919c1acf0d22 : SUCCESS in 12m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd1736d581f348e993a45555beeb6711 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0ffef7d6f8f49119dfb3e59544f5421 : SUCCESS in 29m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/70aae043d9d940dd8a7f9c683fb19ed8 : SUCCESS in 46m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/74776a8495a6469c98d494a6ba7576d2 : FAILURE in 1h 50m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1059d079e3704b2ba07899def0c6e246 : SUCCESS in 42m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/eeda3b609cb246f6b94e9e87d407fef6 : SUCCESS in 55m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/db5ee8301fb54cc0a4e2e92d793fbd05 : SUCCESS in 2h 07m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/8183a59dd4eb4950b308825385577b4c : SUCCESS in 2h 10m 11s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e053ded376a949e49b1a6fde6479fdd2 : SUCCESS in 47m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/88e46282393e4195b25d1ae740841646 : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d32d86a9158947b7abb646d3438c4b9a : SUCCESS in 22m 42s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50efa2de578d4405b64c842be71ae29a : SUCCESS in 24m 51s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59e824380c414115979cffd0b7212a69 : SUCCESS in 28m 24s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2bdebb6b576041469b44504c15f84f9e : SUCCESS in 2h 10m 30s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/ce3313a4bfc24fa29167ef617426c788 : SUCCESS in 1h 48m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5be06218055c498c99305d4fbb60049a : SUCCESS in 1h 30m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e4418a866634b75a85d9576af7c0d30 : SUCCESS in 54m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a260f5f5cf584439975479c63c5873f8 : SUCCESS in 23m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4ba1eedcd5d44490bb33050addd6e55a : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c1c094ee71d1078ee552a9bfb0cd58c47a6209c7","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-12-10 13:18:04.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0d5653a7eee15aa797365b896bf1fb281ebeefa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 14:36:06.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c1c2dbfbdd4165bb83cb541a174b5a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3a9717be2ae45568da9e1a161995883 : SUCCESS in 26m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28a5982e27944d6784b5172bc135b6f3 : SUCCESS in 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9a24434ca812710ba740e1c28846c679fe228b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 16:02:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c9f48b44984cbb9510f21bcff0240a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/058578d55b6a4f19b9969b2ec9ec756c : SUCCESS in 46m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/931950c54812448dbfb0bf61856716a2 : SUCCESS in 1h 41m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85a2b57f92704ce7a3c0c7dbe00ac875 : SUCCESS in 54m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08ecf8ce526e4dffa576d73ed11b5d41 : SUCCESS in 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7205824e068d4059b4a6956f77d6b816 : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdea63d7edf6401088710f0f8b27a121 : SUCCESS in 13m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c9cc435a96c4ae7aa2ec601c3c8620e : SUCCESS in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8d5da1063b24bc58de28fe0774ee17d : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b43dda920e554604a1c6b95e1186c281 : SUCCESS in 25m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/03451ad05c9f4b89bb7242b7c2044819 : SUCCESS in 1h 03m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b6d3c7129fb14813ae022b786fa01974 : SUCCESS in 2h 02m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e29b9b0834d449b7984a63352508e747 : SUCCESS in 56m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1415f0e58460464fb23debc118edb8d0 : SUCCESS in 42m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/88e52ce2325d4d4ca64bcf604e7e18a7 : SUCCESS in 2h 23m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/77f819e4e6a048c1960f0c10d1a7071c : SUCCESS in 1h 58m 48s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ffd2b5eeb88042ab9842e83f4cfa2798 : SUCCESS in 39m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/067611a455664212b93b1ec816dfe1ec : SUCCESS in 4m 46s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24e071df81af4b1bb12170b92ff43a81 : SUCCESS in 24m 19s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62ecd4abab2c4c49a21a859b01d0deb4 : SUCCESS in 24m 56s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8ffc6f488aed4a5b9dbd416f62afe9e7 : SUCCESS in 30m 06s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0777f50ddd9d4169b4d11d244df49d85 : SUCCESS in 1h 48m 12s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/9d919c2376534cf19ce642015458d13a : SUCCESS in 2h 04m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8aafa4faa59649b5bcca2183f4ee47c1 : SUCCESS in 1h 29m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a95123d840554bfc83968abdea07af2a : SUCCESS in 1h 04m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ab4b249d828e4ce48ffc7a17440201bd : SUCCESS in 24m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/18f48d236b7f486b8147a36ae4c378ab : SUCCESS in 36m 52s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3f40aa0b8d48c381715ad6b510c4a91c7775a7c0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-12-10 16:49:21.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"e339acacaa3134aad8351aa664078d6a0c75fcd8","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2024-12-11 06:49:38.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"28bf171d121fce00467186fc1bde984e352ecbb1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-11 10:41:50.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-4rnwz/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"e27ffcf4e990bbc5cb4e66904d33fc93a6994fec","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-11 21:08:59.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-85hbk/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"5c2b7766801e3c091989bf8f62ce085174c7d85b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-11 23:47:34.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-929858-49kcf/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"3266d9994a238a926e36c586587537fb98d04ad0","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-12 03:17:48.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"95ded353fe2e64e31502274bab07aed56903a999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 03:18:35.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"075d23ec6dc627af86d09b3149160dc76d2e30e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 05:48:06.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e424ad6df9de4eee9133736f5fdf4103\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f6985542706242d5939fe7a4a625681f : SUCCESS in 1h 00m 47s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/85b9eb61c8764ccc9fabeff5246288a3 : SUCCESS in 1h 58m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61e9f41881504da3ba6d29e14aabfac3 : SUCCESS in 54m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce880067aa24b09ba3fa57f67b72c1a : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f048f1dbfe7d4329954317be49150fe2 : SUCCESS in 13m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e522a56d208040feaad92d4746f41809 : SUCCESS in 13m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142577adcd9b4e1dae0d4993aea3c1c8 : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfe2c3173fae492eb76e4d8e0e4e501f : SUCCESS in 33m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1161dbcdf6ba4639a2191e266b8e1831 : SUCCESS in 43m 08s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d500a96774214f3c9f87fbb255160963 : SUCCESS in 25m 36s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd1b3d361c2e44b683cbac21d04ae2e8 : SUCCESS in 27m 39s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2428e905557e44b696c06c73cfa52024 : SUCCESS in 24m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fe89d12845b0471b803ce7b07c59ff24 : SUCCESS in 1h 41m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/1e67f5063e894c69a48c0456d41c3ad9 : SUCCESS in 2h 14m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6299f6ea59fb4cd2b19e0fe8c000b69e : SUCCESS in 9m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bbdd6a63267243f38f6047bd553f73cf : SUCCESS in 1h 27m 34s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a347dfad21284db3bc2ec24e7af56c29 : SUCCESS in 1h 34m 08s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/958b76fc9c714650a6e19e54876ba2af : SUCCESS in 1h 31m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/598f0cda376d43179d4a3841539c1b43 : SUCCESS in 1h 12m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d2313c4830e455d92afed9cad863dd8 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":13},{"id":"82505d7647cd2c54d6f20a553ced1c80f0633b63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 05:48:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"74af1abc0c8cc15628dcfc68a23663195f7ca984","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 05:49:24.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/739f8b28f5f047a0b9e9dc552bb6870f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8c13b1599141ed9366b231463c6d41 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed428f7b12a04c3d884ab7267645881e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":13},{"id":"289d11a76c7e9dbf68a6a87e6476a50543c1472e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 16:26:11.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"5f69442fd44654b91ceaae28d0f96e9e35286381","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 21:06:14.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":11},{"id":"108864fe8479c4b024d13c3cc34ed686d827a75a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-15 00:33:34.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"15dccaeed3cd0cc5a4283ec66d7347becabb64df","revisions":{"4c85253bff3f438cd320d96a2ad7d4f08466cbc8":{"kind":"REWORK","_number":1,"created":"2024-09-18 23:45:18.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/1"}}},"commit":{"parents":[{"commit":"b16445df86e76fa4ed299f4002d78f03dbfd0921","subject":"Merge \"Add Dalmatian prelude section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b16445df86e76fa4ed299f4002d78f03dbfd0921"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:45:15.000000000","tz":0},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize functionality\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c85253bff3f438cd320d96a2ad7d4f08466cbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c85253bff3f438cd320d96a2ad7d4f08466cbc8"}]},"branch":"refs/heads/master"},"4c5eeab55fa216d90bc5fe7fb0075b83b5fa74db":{"kind":"REWORK","_number":2,"created":"2024-09-19 15:17:07.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/2"}}},"commit":{"parents":[{"commit":"b16445df86e76fa4ed299f4002d78f03dbfd0921","subject":"Merge \"Add Dalmatian prelude section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b16445df86e76fa4ed299f4002d78f03dbfd0921"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-19 15:16:44.000000000","tz":0},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize functionality\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c5eeab55fa216d90bc5fe7fb0075b83b5fa74db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c5eeab55fa216d90bc5fe7fb0075b83b5fa74db"}]},"branch":"refs/heads/master"},"1631eedbf13e5bf8e1c27e520e8a459717c26886":{"kind":"REWORK","_number":3,"created":"2024-09-20 08:28:45.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/3"}}},"commit":{"parents":[{"commit":"b16445df86e76fa4ed299f4002d78f03dbfd0921","subject":"Merge \"Add Dalmatian prelude section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b16445df86e76fa4ed299f4002d78f03dbfd0921"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 08:28:15.000000000","tz":0},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n- functional tests\n- unit tests\n- docs: releasenote\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1631eedbf13e5bf8e1c27e520e8a459717c26886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1631eedbf13e5bf8e1c27e520e8a459717c26886"}]},"branch":"refs/heads/master"},"daedec609758c132c1fc4dfc03abdf9fec27aa7a":{"kind":"REWORK","_number":4,"created":"2024-09-20 17:25:18.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/4"}}},"commit":{"parents":[{"commit":"b16445df86e76fa4ed299f4002d78f03dbfd0921","subject":"Merge \"Add Dalmatian prelude section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b16445df86e76fa4ed299f4002d78f03dbfd0921"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 17:04:08.000000000","tz":0},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- functional tests\n- unit tests\n- docs: releasenote\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/daedec609758c132c1fc4dfc03abdf9fec27aa7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/daedec609758c132c1fc4dfc03abdf9fec27aa7a"}]},"branch":"refs/heads/master"},"8982adb9bb1bba76120c9300cee9933324024c5f":{"kind":"REWORK","_number":5,"created":"2024-09-20 17:27:12.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/5"}}},"commit":{"parents":[{"commit":"b16445df86e76fa4ed299f4002d78f03dbfd0921","subject":"Merge \"Add Dalmatian prelude section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b16445df86e76fa4ed299f4002d78f03dbfd0921"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-20 17:26:56.000000000","tz":0},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- functional tests\n- unit tests\n- docs: releasenote\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8982adb9bb1bba76120c9300cee9933324024c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8982adb9bb1bba76120c9300cee9933324024c5f"}]},"branch":"refs/heads/master"},"778dd7273eb21e796ba5175126ff2b536d1c6543":{"kind":"REWORK","_number":6,"created":"2024-10-14 11:12:07.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/6"}}},"commit":{"parents":[{"commit":"37076155e6248ec6824ffbe4e26ffcd44994cd70","subject":"Merge \"Support creating servers with RBAC SGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37076155e6248ec6824ffbe4e26ffcd44994cd70"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-10-14 11:11:58.000000000","tz":-240},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- functional tests\n- unit tests\n- docs: releasenote\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/778dd7273eb21e796ba5175126ff2b536d1c6543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/778dd7273eb21e796ba5175126ff2b536d1c6543"}]},"branch":"refs/heads/master"},"d19207c76267257682dfd28174875f2924797c72":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-10-14 11:18:21.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/7"}}},"commit":{"parents":[{"commit":"37076155e6248ec6824ffbe4e26ffcd44994cd70","subject":"Merge \"Support creating servers with RBAC SGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37076155e6248ec6824ffbe4e26ffcd44994cd70"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-10-14 11:17:16.000000000","tz":-240},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- functional tests\n- unit tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d19207c76267257682dfd28174875f2924797c72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d19207c76267257682dfd28174875f2924797c72"}]},"branch":"refs/heads/master"},"6743df82fde16f7b49d810ad7d5b72032073e625":{"kind":"REWORK","_number":8,"created":"2024-10-15 16:47:02.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/8"}}},"commit":{"parents":[{"commit":"37076155e6248ec6824ffbe4e26ffcd44994cd70","subject":"Merge \"Support creating servers with RBAC SGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37076155e6248ec6824ffbe4e26ffcd44994cd70"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-10-15 16:44:33.000000000","tz":-240},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nNot adding test_compute_mgr unit tests, as this require DB\nupdate, which is already covered in functional tests\n\nChange also adds supporting:\n- functional tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6743df82fde16f7b49d810ad7d5b72032073e625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6743df82fde16f7b49d810ad7d5b72032073e625"}]},"branch":"refs/heads/master"},"35d46f7f0ff2c82d656de2796836241cb709461d":{"kind":"REWORK","_number":9,"created":"2024-10-23 06:32:28.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/9"}}},"commit":{"parents":[{"commit":"37076155e6248ec6824ffbe4e26ffcd44994cd70","subject":"Merge \"Support creating servers with RBAC SGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37076155e6248ec6824ffbe4e26ffcd44994cd70"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-10-23 06:32:22.000000000","tz":-240},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nNot adding test_compute_mgr unit tests, as this require DB\nupdate, which is already covered in functional tests\n\nChange also adds supporting:\n- functional tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35d46f7f0ff2c82d656de2796836241cb709461d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35d46f7f0ff2c82d656de2796836241cb709461d"}]},"branch":"refs/heads/master"},"2bbbfb5f3c33790b8485bb7c20ccf4e106773df7":{"kind":"REWORK","_number":10,"created":"2024-11-12 09:35:28.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/10"}}},"commit":{"parents":[{"commit":"3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9","subject":"Merge \"Route shared storage RPC to evac dest at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-11-12 09:34:52.000000000","tz":-300},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- functional and unit tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bbbfb5f3c33790b8485bb7c20ccf4e106773df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bbbfb5f3c33790b8485bb7c20ccf4e106773df7"}]},"branch":"refs/heads/master"},"c1b1df6afab05d607e1bc1a621ba2b38524e7ecb":{"kind":"REWORK","_number":11,"created":"2024-11-15 11:30:42.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/11"}}},"commit":{"parents":[{"commit":"3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9","subject":"Merge \"Route shared storage RPC to evac dest at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-11-15 11:28:13.000000000","tz":-300},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates bdm for swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- unit and functional tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1b1df6afab05d607e1bc1a621ba2b38524e7ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1b1df6afab05d607e1bc1a621ba2b38524e7ecb"}]},"branch":"refs/heads/master"},"493be3a84b0b18fc3eb10112add823e6e0dc68ce":{"kind":"REWORK","_number":12,"created":"2024-12-10 09:31:45.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/12"}}},"commit":{"parents":[{"commit":"3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9","subject":"Merge \"Route shared storage RPC to evac dest at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-12-10 09:31:08.000000000","tz":-300},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates bdm for swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- unit and functional tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/493be3a84b0b18fc3eb10112add823e6e0dc68ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/493be3a84b0b18fc3eb10112add823e6e0dc68ce"}]},"branch":"refs/heads/master"},"15dccaeed3cd0cc5a4283ec66d7347becabb64df":{"kind":"REWORK","_number":13,"created":"2024-12-10 09:38:11.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/58/929858/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/929858/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/929858/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/929858/13"}}},"commit":{"parents":[{"commit":"3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9","subject":"Merge \"Route shared storage RPC to evac dest at startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b530ac15b4bd79167703cb0d71ff7d3b53ec2f9"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-09-18 23:19:15.000000000","tz":0},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-12-10 09:37:51.000000000","tz":-300},"subject":"Update Nova bdm with updated swap info","message":"Update Nova bdm with updated swap info\n\nThis change updates bdm for swap in finish_resize and revert_resize\nfunctionality.\n\nChange also adds supporting:\n- unit and functional tests\n- docs: releasenote\n\nCloses-Bug: #1552777\n\nChange-Id: Ic8aaa0728a43936cd4c6e1ed590e01ba8f0fbf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15dccaeed3cd0cc5a4283ec66d7347becabb64df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15dccaeed3cd0cc5a4283ec66d7347becabb64df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
