)]}'
{"id":"openstack%2Fnova~931479","triplet_id":"openstack%2Fnova~master~Id08761918ccc3477a39d9c778f3ac92679c22511","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2024-10-07 06:48:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id08761918ccc3477a39d9c778f3ac92679c22511","subject":"[doc]Fix the device_spec config doc about placement","status":"MERGED","created":"2024-10-04 11:47:34.000000000","updated":"2024-10-07 06:49:52.000000000","submitted":"2024-10-07 06:48:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"931479","meta_rev_id":"9f4f73d6cc2dbf6da538c227e355955c014671ee","_number":931479,"virtual_id_number":931479,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 06:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2024-10-06 05:48:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2024-10-04 12:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-10-06 10:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-04 15:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2024-10-04 12:44:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 12:12:02.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-04 12:44:32.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-10-04 13:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-04 13:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 15:00:18.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-10-06 05:48:56.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2024-10-06 10:52:51.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2024-10-07 05:02:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"3be05777e2aba86b3075caded321f41bca4e3ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-10-04 11:47:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8be5e7ad503d64d7a2cc296b50d565d10f64ee5b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-04 12:12:02.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-931479-mwxjh/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"76602e6d67b6e56e6ce87cbc46e305cdb638c840","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-10-04 12:44:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46702ad1e8405b784953dd878d6cffbf3759c69e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 13:09:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170f05d01c634009a2013380f2c1cbb0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f74a095e9244ace94add40c3b1086e6 : FAILURE in 40m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/573f68b64ebd43b499532358249269e6 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5175594b29ad4881bd421f687001b28be6c079b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 13:59:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/615ab0df240d4808a47e6045ca052bda\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e3f16f3de864c3a8c0ceea7c62509c8 : SUCCESS in 45m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/18c5978df6db44a382549440a89c0ecc : SUCCESS in 1h 20m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2ff3e8edd7c4db4ad3a2eab8efaa5f0 : SUCCESS in 1h 05m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930077361d0245929f7e4e1e727da890 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5760c8255cef4e709526e63ae8436edc : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2e87ae8516d4595b1981289e9d6cacd : SUCCESS in 12m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8b941360c7746e2ae4c15d268db39ad : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e468ccc1a904a3f81a91b0bd558664f : SUCCESS in 10m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/92119489e42745fab11ab8d12f43ecaf : SUCCESS in 39m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b837df73127a4174bb0fbdea4f1e7556 : SUCCESS in 1h 48m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5eb3425d37b44bf6ba37811f1fd6c054 : POST_FAILURE in 23m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d67ddbf6dc444aa6b91d5b7c8915beb5 : SUCCESS in 56m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/59d8dc0f3928410bba84b489129c5a18 : SUCCESS in 1h 30m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/1fc9433a9cdb4ceb9a3a63af3d88141d : SUCCESS in 2h 11m 10s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/28c6385e501a4afa96bd51b1ff5b3fee : SUCCESS in 47m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e206a164edcc4f7eb36ff3c285994812 : SUCCESS in 4m 45s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b1dc5d129364f1aa52b279dac81f3b3 : SUCCESS in 20m 58s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/60e6e37d273348dbba41935fc6bfecb3 : SUCCESS in 23m 59s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5c45d89da5024a92a93a4e052c919e14 : SUCCESS in 34m 32s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c22d6b7c0bc74e84886b66104126ff37 : SUCCESS in 1h 13m 51s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/f76dde26fe5d4ea2bf514eb0135c5af8 : SUCCESS in 1h 54m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee60cf22d0b14c0590891f650a2ba49e : SUCCESS in 55m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3c7fc8567f48f8833cb8ff1b73fe32 : SUCCESS in 56m 38s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bc3d6fcdade84032b04d8880c342d2be : SUCCESS in 20m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3ac8d8e7cb4a4b0eb3e67ca787439d5d : SUCCESS in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c3c58d8ac68ab1712ff6516753e469cb85775ac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-10-04 14:37:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a894d8c1e4f4ac7e88ae84cd546792c455cccf70","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-04 14:59:58.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-931479-pzhxd/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"9bbb50e2317179de9ae70972203d48509e2ce866","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-10-04 15:00:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0514f018c68cdb57bf4477e6468bf12c1932b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 16:51:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab26ffec2e564614ae18c17430bd6103\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/48dab96fad2a4de18e0bffe40a5a6b39 : SUCCESS in 53m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4261c302aa4242b49e62a5e31a9dc2df : SUCCESS in 2h 06m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a06266cb7b65414fb4e914045f28b1d0 : SUCCESS in 50m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0b3c5e1603d49f0952fe959cda3a3be : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a892ce31d38e48a8898a1405aff193a9 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ab76b24a9324318badb59855ba51dc3 : SUCCESS in 14m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30c648d2b3ed43819037a87787d318d8 : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa321b809794e9cbe0e63e76a05bd1e : SUCCESS in 10m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/04197d00366b49c89192ff5d9ab0219d : SUCCESS in 54m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5fe4934ec2a84a8b829ab5d80286d3bb : SUCCESS in 1h 37m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/da9793d709f14d87a55b5bf354843fcc : SUCCESS in 50m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e6667ef7f3b642afa82adbcdab660b37 : SUCCESS in 47m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fcb5a96628ba4c53bee01e037718eb1a : SUCCESS in 1h 43m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/5ec01d6f1fb24b7fa33c6faa510add03 : FAILURE in 1h 54m 23s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/74aa27b8b0854140ab9d3a88c65c0850 : SUCCESS in 35m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/46b8a30932d34a37a166302c97aaca19 : SUCCESS in 4m 32s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b0c166170ba5495d8fdbb4c1d0d32e5c : SUCCESS in 30m 34s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a970963f7514035a3abf2d0963324f9 : SUCCESS in 28m 08s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6307b1619611458488223b89a866227a : SUCCESS in 30m 28s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8d3d88f8a6e5456889bfc6e93aea28e0 : SUCCESS in 1h 48m 04s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/177d90174fb44bf2a9de2fff04ba4f61 : SUCCESS in 1h 45m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ba9e13e12e84181acb9b1ccc245851a : SUCCESS in 1h 12m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebc442a2cd434783b4bfe0a23c9e1195 : SUCCESS in 56m 59s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c47ac069a8bb4fc38808eb92198e0ae6 : SUCCESS in 20m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c5309c8d0a684d1f9b0fc874ec127c03 : SUCCESS in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c47f50d62c30c558cdc9fd6d52cd599fada52029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 17:56:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ebe81b14b54a82875fd1dcf3e8cdb3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8513a0d5681842df94a644f775ad81d6 : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3280f104b0ee4ddfbe6c1215a776587a : SUCCESS in 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d228a5c1cb5ec9836ff943b4ac28f545e2cdf2b0","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2024-10-06 05:48:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6adafe80527c5c7353d5626ead24dfc475ffc137","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-06 06:12:37.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-931479-dsbwh/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"9386f7008a65856e7959249d2630803c719c3dd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 06:33:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f472283980bd47a78bf55acac3133c76\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bafc13e9b034ba0a9767dc8b394dd06 : SUCCESS in 42m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f28f5eb4d7d435989422aab2e514870 : FAILURE in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf9f6130a7ae128d3251b9d80b4e0cae129fd82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 07:55:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a1b3967f60546f7840454f47b86ab95\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e28c79df9ec4489ba3ecc62b80d46cdd : SUCCESS in 1h 00m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/72dd1f3640694716945af209cae8ae18 : SUCCESS in 2h 00m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f434dd258994d748a613f9e381168db : SUCCESS in 1h 00m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d6ecc783dcd409288a2f3f84eee5ce9 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/120b5670f45a4825866c7b673ac1e154 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67f99bb716b949d0816fbd9cbe1912bf : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c771e9457154ccdbf2111d03274ead1 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab6681cfaa24f3ebbde85c2c1872041 : SUCCESS in 6m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8b002135016540cbbdfb52ff1ed0f1a9 : SUCCESS in 35m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7dee4aeed7d4db5b19d0fe0a7c0a1f9 : SUCCESS in 1h 09m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/41a8d324b9a24ecaadf92a2e01f2683d : SUCCESS in 35m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ec0a58f8a96a4e8789b8bc7879cae1b9 : SUCCESS in 34m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fb588dd594da4b44af8a29bb1aad03bd : SUCCESS in 1h 22m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/564abf8b1ccc49369e76786ebb2f51a6 : FAILURE in 1h 33m 49s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/0762280835d345809e7bea1288050cb4 : SUCCESS in 51m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/287623eb566d4a1b8ae4db5a55cb1529 : SUCCESS in 4m 17s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/931bae1aaab942729bbd4fd6a1fa1fd9 : SUCCESS in 22m 59s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ceb36a09d0e419ba519dae1b37e0420 : SUCCESS in 26m 46s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b627d4e140984e6b86f26a231496d62f : SUCCESS in 28m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ee3e18362fcc4388a10dc396eef89bf0 : SUCCESS in 2h 03m 10s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/063040d36e8a4a4088426bd2741ac6f2 : SUCCESS in 1h 59m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/473dd4cc02ce43808eb7a6f7b2731be8 : SUCCESS in 1h 28m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c00bee35eac4d99b542ce5b2e6d92ca : SUCCESS in 36m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/762c660363d644ae8a5c029c84c4529d : SUCCESS in 29m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/87de62644b2b4ff68e8fed32e04dc519 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20e6456e444751cd3591485c7d78dd284ef7c08c","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-06 10:52:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d80dadcd925b725e9cd524d22c133f609078df3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 11:16:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a242a645134fc9a60c3206afd63216\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/105e23fec065451c9950c56988521983 : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9054e8d457d34cf8bdde7b414735a919 : SUCCESS in 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f220f42931d28af34303c2db3bbe1b3a0941b26","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-06 11:39:32.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-931479-d6hzd/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"c9bce85b16da7a090ad5491434162f7f083fb038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-06 12:46:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2290c0af276b44e09ab9aff25a83fc76\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0945b063fba44e16bcaab512e334d6b5 : SUCCESS in 45m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f35d46a961d4494ab80f826c5f0a754b : SUCCESS in 1h 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98be328ab70e4b32840c1b4432ee5e34 : SUCCESS in 48m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9aa97a7606d43278dba358993decf27 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9c17268cc942a5b767945dd8825508 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/747d3b3766b042af8f6d828ff8a94605 : SUCCESS in 13m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8b262e68f8c4cc7aae86d48b0f82b77 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b78beabe2d74c73b87bc4a8f8281538 : SUCCESS in 9m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/86ccd18c7bad4594a221df87f746ab2b : SUCCESS in 43m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aad05806255d4828beb55518ac914cf3 : SUCCESS in 1h 23m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b5bf9034b41c4a91806eefc9e9f98b17 : POST_FAILURE in 24m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7ada24533c894cb1a97acff339db2525 : SUCCESS in 54m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/87f6895d99b54196bf5e246615801852 : SUCCESS in 1h 26m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/fd54606e44d54f9fa6b3d0cf1738844f : FAILURE in 1h 24m 50s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4b4e2b0dc23249dab8967a393647cdc9 : SUCCESS in 44m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f8d7ac367b21417db56960adfa18a7d2 : SUCCESS in 5m 07s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/23fbbcdb729d48d48b1e4168eaf3aa4f : SUCCESS in 25m 22s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1cc249f4fcdc4bda9af852ba97577bb5 : SUCCESS in 27m 37s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9e7608746c0b40878332a506193b2c7a : SUCCESS in 23m 46s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c12188fac71e497b840b1bc6c0641ae4 : SUCCESS in 1h 12m 09s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/d67aec3a52584278af21bb9267ab8339 : SUCCESS in 1h 44m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/51870a477d7344548f55b4ae48df87dc : SUCCESS in 1h 04m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e3e4fb33d04cb4bfa350ace6f285df : SUCCESS in 40m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1cc965b76fa84122a0d3d58b2ca5d590 : SUCCESS in 18m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8bb71d80e5424b41844c9a5a4e18af57 : SUCCESS in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30fd0626df75868085465945ef5b139585cb61c3","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-10-07 01:55:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"133c446bacd58aa7503adfa2c79598f80e1b7963","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-07 02:16:37.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-931479-2g8qf/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"7ae45146e39a206abfc6336beaa7bf2331d314fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 02:31:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3836d31545141b1b62475adbd2c4d19\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cfcc8e4eb87490caceea1d3ce0cb898 : SUCCESS in 30m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e5f40ad6497419cb4fba5ac9ef846d2 : FAILURE in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1903a5441154ae4fdc8d32248766a1b2138dc501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 04:57:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97fc75642a834e87a946477bd8e8ce60\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db46382eee1944d8b3db7756dd149972 : SUCCESS in 1h 04m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1b6a15eedf474a0da4430b4e20e101c1 : SUCCESS in 1h 38m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6da40869e6b40d8b2a7ed882fd37900 : SUCCESS in 50m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02ccdc1466b14ed6944afcac462fcc95 : SUCCESS in 19m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad97185110df4347832be28878cf687e : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a0673ff95946d1be6105b24a30dd5f : SUCCESS in 13m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac75a6ec0fde4214bf355147b1c4e845 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dab1d7560164afb8be9ff595ec15f29 : SUCCESS in 8m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/464643bcb345412285424eb7e7afc0af : SUCCESS in 1h 05m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/772c6652d81e4a5899f28db1843996e6 : SUCCESS in 1h 40m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ab529963b57842a79c786d1253670fc9 : SUCCESS in 45m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8c32cc78caf24bf7a72b2674717863de : SUCCESS in 1h 10m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a25d58875fe54f8082f96052af0f4124 : SUCCESS in 1h 40m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/e73feb22fb2e44c38989661830c89316 : SUCCESS in 2h 54m 08s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d9257d2903194c07906cf5b819cf1f26 : SUCCESS in 37m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/659fb3fc0f824e10ad3891a9e2af4404 : SUCCESS in 4m 22s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f3fea9982d84d45be1d0314345bcda7 : SUCCESS in 25m 09s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f207687c341d46d1bb043736466a1ac1 : SUCCESS in 27m 49s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6121fbf45ad943d8ac1a841814fb6a44 : SUCCESS in 28m 54s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/920c7bc3dce041b9a3aa60bca6ced5b3 : SUCCESS in 1h 11m 02s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/396e28446a0849c6973ad181bd920033 : SUCCESS in 1h 50m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce132a367e7d4a63a132674aaa5be81d : SUCCESS in 1h 16m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30e3ef3f39224b0da6dc5560bbfea058 : SUCCESS in 57m 07s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3498d41b92104327b3430a11471a7fc7 : SUCCESS in 21m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9ba5670fa4d648189708c3a2615eba47 : SUCCESS in 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4a7909fed8536b2b41a108b624736419a198d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 04:57:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"daf24ce2dbc55ce3bffeede92062ff8c6acaa474","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-07 05:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4d7e23c691a5fd6aaa71e8113368dc296b0693e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 06:48:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e535b00dc594b5189dd513a43666ce1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ae7e14efd4ca4ed6927a1896e84facbf : SUCCESS in 1h 15m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/12d26dee5a6c4a02a07c4ad42d48bf36 : SUCCESS in 1h 50m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08eb8c9cd9b24b538c39f41abbc5d0f5 : SUCCESS in 43m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b690c0050a4b28b08376bb1b60953a : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/296ac9882b284ccfb13952afd3862323 : SUCCESS in 10m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81846b08f35e4bcc92c58079afc90b85 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cb7998d7f04f0a84644cd481454921 : SUCCESS in 8m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/12f34833705a4e04a9d443a1cbf3da0e : SUCCESS in 31m 52s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f67e3bba7f5f429b9e4d6601ee710dd8 : SUCCESS in 19m 32s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d91357e579a4537a92a38c2c2cddea4 : SUCCESS in 18m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bd532416cb2f4792a98c9d111788b5b6 : SUCCESS in 22m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5c9cce05b6424e14b75f0344a56b7394 : SUCCESS in 1h 20m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/32c7ee8a2a434b768596cd71300f32af : SUCCESS in 1h 26m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cac8c11a553049bb90978397d75a279b : SUCCESS in 4m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e43e6693aa6d4521bfb695d0ade1ba35 : SUCCESS in 1h 12m 29s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e683340952494ff4967b19a193c23775 : SUCCESS in 1h 11m 41s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/1b6c92ca1b5e4f2c86056e22de5af353 : SUCCESS in 1h 22m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/603b016308cb454b86ea9bf8dc0b59ba : SUCCESS in 1h 22m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a35c0c74969e44f7888cd4acb244f3ef : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8dd2ea598d1f437bd47baed73277ed4af6bd1ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 06:48:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f4f73d6cc2dbf6da538c227e355955c014671ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 06:49:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4525e40fab20408aa0b8467cb8636482\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cffa243ac04790b2f2c7854b97c34f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fca941adb73b49ec272850126c1c98c6c8ef4b01","revisions":{"fca941adb73b49ec272850126c1c98c6c8ef4b01":{"kind":"REWORK","_number":1,"created":"2024-10-04 11:47:34.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/79/931479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/931479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/931479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/931479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/931479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/931479/1"}}},"commit":{"parents":[{"commit":"df5454021543e8aa3c73289224158c9090f0da6a","subject":"Merge \"Vmware: Remove uuid parameter from get_vmdk_info call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df5454021543e8aa3c73289224158c9090f0da6a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-10-04 11:46:44.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-10-04 11:46:44.000000000","tz":120},"subject":"[doc]Fix the device_spec config doc about placement","message":"[doc]Fix the device_spec config doc about placement\n\nChange-Id: Id08761918ccc3477a39d9c778f3ac92679c22511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fca941adb73b49ec272850126c1c98c6c8ef4b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fca941adb73b49ec272850126c1c98c6c8ef4b01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
