)]}'
{"id":"openstack%2Fnova~931906","triplet_id":"openstack%2Fnova~master~I88e5e12d18264ce848457191ba3de2fbd8d8bf5c","project":"openstack/nova","branch":"master","topic":"reqs","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-10-09 13:49:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88e5e12d18264ce848457191ba3de2fbd8d8bf5c","subject":"docs: Remove unused sphinx deps","status":"MERGED","created":"2024-10-09 11:50:35.000000000","updated":"2024-10-09 13:50:49.000000000","submitted":"2024-10-09 13:49:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931906-reqs","meta_rev_id":"164ade026935aaf0db30888ee110b7d75bf2616e","_number":931906,"virtual_id_number":931906,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-09 13:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2024-10-09 11:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2024-10-09 11:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 11:52:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-10-09 12:13:49.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-09 12:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 12:50:51.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"4d1533043924fd9d68c557183626bd8ac415b48f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-09 11:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"053f3d1d58acd998d91e023824acd0f3905c2417","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-10-09 11:52:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c05372c0e34c213afab2ea900477a38441088131","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-09 12:13:49.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-931906-l82hv/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"bd8da7249841e480a8c8adeeca653a6938e764a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 12:45:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda002dfdad646b4b6db0b82cb6c4857\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/1f961900ac884a6585be2aa811498d3b : SUCCESS in 3m 33s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de06a7ff8a2f49f195e3a9ccd123c82b : SUCCESS in 3m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e03804fe624448fbe66450c42bf9a7c : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2786d12e0ae743e7a7e253021fd30294 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d817fae9b454fe0ad6036f88dd30e7c : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72427b2ac2e34dc687d2e36b6e5531de : SUCCESS in 25m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3df9e49bc21341f79c21b0ea91ef5a9c : SUCCESS in 4m 31s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7b16cafda8c4c08993b780a562e9781 : SUCCESS in 26m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/850b14fb9db74c93bb9bccde8144810f : SUCCESS in 23m 06s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5992bea2b7904f229314a8ce24292c6f : SUCCESS in 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"37f4a15a5445063a2c0fdf99a19d6cd1f628b999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 12:45:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ab48339ba8ad887aa2a38c035306a5bf3121756","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2024-10-09 12:50:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"22155e174bfbb7762a49a6cad9eaaa6a4616866d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 13:49:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f2f61b957b848f593e6a19349906f4d\n\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/1b1a518279ed4ef4ac11dbdd8e4985b3 : SUCCESS in 5m 17s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ecba0e93b3f74d0a9970be677cdd3d43 : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7c274fcdc65446fdad4498d7eeb1d7ab : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cdce066302746ef9db3b17536aee708 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/997b83bf5e314281b6276c82a3048ebc : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c644e5b6f24c739631e607e5895865 : SUCCESS in 37m 16s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8b7f8ab319a4b5d965783e8a14c2c10 : SUCCESS in 25m 57s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b98aa843ed648f884cbbeb29552aab1 : SUCCESS in 27m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fc3dfad1166f48edbae57cb452812ae5 : SUCCESS in 30m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/acdc64b545a642e1a0a0de363ff86b99 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"03ae5dbc2a2c4b1ca97437ddef3c587cc2f1809a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 13:49:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"164ade026935aaf0db30888ee110b7d75bf2616e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 13:50:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0b18fc267d649eb967422d723d48fcf\n\n- promote-api-guide https://zuul.opendev.org/t/openstack/build/8e1d8249471d4e3d95229722c361c2e1 : SUCCESS in 49s\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/140aad317cd54e04a7f7926c6dc0da59 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd3ad31b2c5436d9384b077bfd4cc12 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10cb189b7f754e9cb04e2f1c3429bbd1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be4f4133a0ab391f55d26c0d91cd17d21870bf72","revisions":{"be4f4133a0ab391f55d26c0d91cd17d21870bf72":{"kind":"REWORK","_number":1,"created":"2024-10-09 11:50:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/931906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/931906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/931906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/931906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/931906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/931906/1"}}},"commit":{"parents":[{"commit":"1e57c08559d2b1e7348701e79ecb292d28f195fd","subject":"Merge \"Fix image format error on vol-backed snapshot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1e57c08559d2b1e7348701e79ecb292d28f195fd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-10-09 11:46:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-10-09 11:46:44.000000000","tz":60},"subject":"docs: Remove unused sphinx deps","message":"docs: Remove unused sphinx deps\n\nWe dropped use of these some time ago but forgot to remove them from the\n\u0027doc/requirements.txt\u0027 file. Fix that oversight now.\n\nChange-Id: I88e5e12d18264ce848457191ba3de2fbd8d8bf5c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/be4f4133a0ab391f55d26c0d91cd17d21870bf72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/be4f4133a0ab391f55d26c0d91cd17d21870bf72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
