)]}'
{"id":"openstack%2Fnova~933311","triplet_id":"openstack%2Fnova~master~I322302375392bffd9f8001282deb14a91b34349c","project":"openstack/nova","branch":"master","topic":"bug/1913016","attention_set":{"33847":{"account":{"_account_id":33847,"name":"Mina Shahsavan pour","display_name":"Mina Shahsavan","email":"m.shahsavanpour@gmail.com","username":"m.shahsavan"},"last_update":"2024-11-12 13:27:52.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I322302375392bffd9f8001282deb14a91b34349c","subject":"nova-api: Prevent os-resetState API action on instances in disallowed VM states.","status":"NEW","created":"2024-10-24 13:43:11.000000000","updated":"2024-11-12 14:21:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"f86de1a3eebb6078765966ea0283fb6903b3a19c","_number":933311,"virtual_id_number":933311,"owner":{"_account_id":33847,"name":"Mina Shahsavan pour","display_name":"Mina Shahsavan","email":"m.shahsavanpour@gmail.com","username":"m.shahsavan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-24 16:16:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2024-11-12 14:21:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2024-11-12 13:27:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 14:17:20.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-24 14:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-24 16:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 13:27:52.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-11-12 14:21:34.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"b5324d9b198a542d22b8660c5e14a10bfa295e1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33847,"name":"Mina Shahsavan pour","display_name":"Mina Shahsavan","email":"m.shahsavanpour@gmail.com","username":"m.shahsavan"},"date":"2024-10-24 13:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"935ce4ca0c8efce33f63573a1baaae3306965e4c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-10-24 14:17:20.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-933311-w542v/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"bb4b3f03f6212eed383d2d0e9de8fcf6fa522e2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:42:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d82793413c14923b6ab0625d8b2dc2f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88d4ced536254e90863090faa995d121 : FAILURE in 46m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3cacaa6e84bd481b91e05742fdcfc1eb : FAILURE in 53m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37a76a9eb2567bb7e208a8adbf21b11359197c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 16:16:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db20385eead04cd39674dc93b2d2d343\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8757b538ef3c456b8a1c5dc64f2bcd8d : SUCCESS in 53m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0357a7516d62435f98e47cf425af3a7a : SUCCESS in 1h 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27bc367fe5c1424bb91a0fa29f1c5810 : SUCCESS in 48m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f10fca4f9924313955642cce1285588 : SUCCESS in 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b6cb0e37d564912a59932bdf5be6e53 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b7989b6b4c485eb202146d3196b492 : SUCCESS in 13m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64eee4353c814c8eaa563301f45b59dc : SUCCESS in 14m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5ca86042be4e238c7473a718ab9b7c : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b836281e784299b0fbd75488d2ec9d : SUCCESS in 36m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/842365f845874ed8b38f00baea542091 : SUCCESS in 1h 00m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f143572c47f441796b05d213fa25706 : SUCCESS in 1h 50m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0aa36c9632de4a50aa4d6499244c6dea : SUCCESS in 1h 13m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/43a029197a04491796350e69a4d71383 : SUCCESS in 46m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0549e67bb2f94c50b206bcf0a262e362 : SUCCESS in 1h 45m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/3d7e9367ae6a4ee7ae865d34713a781b : SUCCESS in 2h 02m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/775d6f1ae7bd40a5ba8a3afd575b6954 : SUCCESS in 44m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b43fe65e013d4882ad89216569c28d12 : SUCCESS in 4m 48s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f3e4abe637048db87c555a75ec3cf32 : SUCCESS in 24m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc73df1aeeb84a8596b025b36aadb947 : SUCCESS in 23m 52s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ef2ad57326848a1b7897af47f6e3c06 : SUCCESS in 30m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8d8b6634311044cca427707ff776cdbf : SUCCESS in 2h 11m 16s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/8efab1092fc3434fa5e214ad1017ecd0 : SUCCESS in 2h 04m 26s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1443facf8df47798bc5b8829174503c : SUCCESS in 1h 19m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d356aa09c68948628196e3a97c04e4ed : SUCCESS in 59m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b2e782e27e6541c38c49752c18748bd3 : SUCCESS in 30m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a98291d494544967827ebd2100dba7db : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ba19f0e412427358f5385bcb5d40da82c8eaaea","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-11-12 13:27:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f86de1a3eebb6078765966ea0283fb6903b3a19c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-12 14:21:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc191b82908560e10f6f505762c7986f5f051ad4","revisions":{"cc191b82908560e10f6f505762c7986f5f051ad4":{"kind":"REWORK","_number":1,"created":"2024-10-24 13:43:11.000000000","uploader":{"_account_id":33847,"name":"Mina Shahsavan pour","display_name":"Mina Shahsavan","email":"m.shahsavanpour@gmail.com","username":"m.shahsavan"},"ref":"refs/changes/11/933311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/933311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/933311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/933311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/933311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/933311/1"}}},"commit":{"parents":[{"commit":"ed2bf3699ddc360fb646b895560fa9dbcfd6beba","subject":"Merge \"VMware: updates resource provider trait list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed2bf3699ddc360fb646b895560fa9dbcfd6beba"}]}],"author":{"name":"Mina ShahsavanPour","email":"m.shahsavanpour@gmail.com","date":"2024-10-24 13:28:06.000000000","tz":210},"committer":{"name":"Mina ShahsavanPour","email":"m.shahsavanpour@gmail.com","date":"2024-10-24 13:43:03.000000000","tz":210},"subject":"nova-api: Prevent os-resetState API action on instances in disallowed VM states.","message":"nova-api: Prevent os-resetState API action on instances in disallowed VM states.\n\nCloses-Bug: #1913016\nChange-Id: I322302375392bffd9f8001282deb14a91b34349c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc191b82908560e10f6f505762c7986f5f051ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc191b82908560e10f6f505762c7986f5f051ad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
