)]}'
{"id":"openstack%2Fnova~935565","triplet_id":"openstack%2Fnova~master~I8d6de19f0f66e9536671da23eea1f81faf71b16a","project":"openstack/nova","branch":"master","topic":"bug/2088831","attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-08-26 22:24:18.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8d6de19f0f66e9536671da23eea1f81faf71b16a","subject":"tests: Functional reproducer for bug 2088831","status":"NEW","created":"2024-11-18 21:24:41.000000000","updated":"2026-08-27 21:30:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee2de429e7b242ba7046ef6eb7382c7372d65d94","_number":935565,"virtual_id_number":935565,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-26 22:24:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2026-08-26 20:54:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-18 21:55:30.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-18 22:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-13 22:51:57.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-26 20:54:50.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-26 21:41:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ef97e3f01dc770e25bd1057ce8a6d5059730adbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-11-18 21:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"436808b774e20990c19a18937b4968acbe587c4a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-11-18 21:55:30.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-935565-vcv6x/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"e61625d9237d3cab56c61d4aeb28db8e139e32ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 22:05:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c4505c90ec3489f9461459da33379e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71fd3257a15042b79f2f1da9ea85592a : SUCCESS in 27m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53398303f85d4b3f9cda0b14ad64722e : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b39a17caa90486e9e9f7f2d500660a9 : SUCCESS in 12m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3df471a9abea49c29e7fa1bf0f4802a4 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998171002a994b6f90ae9886249d180a : SUCCESS in 10m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c92469f9d9cf4d72a0cd25fc36d1b705 : SUCCESS in 4m 06s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6555a4aebcf44e43978fd7129e5e0fbd : SUCCESS in 25m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/311f18d4d126484ab9f825f6ff284d99 : SUCCESS in 20m 44s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c1c98fee0a3e4adb800b1a741bb4855c : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dd0a49da198c2925bb75d517cebe77a2dd1ecfa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 22:12:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/402b3536a84a43bca875ac96f2d1e7de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6ccbbf8c5e34df1afe58a9c65f453e2 : FAILURE in 45m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66f6c2193116457e92c6e9fca6b6e97c : FAILURE in 43m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a646b726d8310ea093fbf4f86a6ef0142c2e63f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-13 22:51:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"59514c9f579f9a9cc1fe68011336d5037745f49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-23 18:20:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1e75fe0455c28b6b0e2babe5314aa52ddbbc8d28","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-23 19:13:20.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-935565-9zg9n/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"909dbc11ed1821edfaab16157b90ca5840c424bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 19:29:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0075b19d7bb4eb9982bfab1ccc4eab7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4816c8e8a0c4edbab272a80a2b34ae1 : SUCCESS in 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c433cab02db451eabbdcad86a3a9277 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/867c2e711e114ae6840572d6dfa8cb85 : SUCCESS in 12m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9999d64af2514538a7ba9b8115e50709 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d6fd3a2c254ee391522c5107d07b8f : SUCCESS in 10m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a1daecda747d4c26afb5ee086595fc90 : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/927c96a05a764835a70762d69fb0ddae : SUCCESS in 27m 30s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be29bc64aeb449db94c5cb3d3b6eb001 : SUCCESS in 26m 30s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8e4b7abd5cd742f284b628800f3677d3 : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c62325ac1547da2861f180ad1cb8ce40077c9ca4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-23 19:34:04.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"c7e81feb89031f199b4d4abe0db0eb8ec7997b4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 20:12:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d20c0e1ef6489896d8bf7a340003f9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35c616ebc34a470e80a8317d0318f1ac : FAILURE in 48m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a8f839c6eba4de286a96ce3a478f206 : FAILURE in 1h 00m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7aef68fe6487311667cee9381674bc4257918713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-26 20:46:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"16bbe84374c706f2dbc264908ba14552d655f95c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 20:54:50.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/687fde5fa2b4497fa1d88b409d17718b\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1307a661e5664d57a65cbc4e8f537d0e : SUCCESS in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be95a373847488f7e9b82a4686db2d3a3694cdac","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 21:41:20.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8d6de19f0f66e9536671da23eea1f81faf71b16a .","accounts_in_message":[],"_revision_number":3},{"id":"3444f36015733c5786d045b65fe724a48842bd3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 22:24:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2901697289249b8981990572d43e78e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06558e36ddb34d529398948d9b0d87e0 : FAILURE in 1h 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06c0c7d22e8b40d19f8c0a3a43f3a3c7 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61020a0dff8b4859a41a67c8d3f9c263 : SUCCESS in 14m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/765ad48f4f4740469825c57441535177 : SUCCESS in 9m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a452e77148f14d18ab742864a842a520 : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8627384a73eb466bbfafb9361fd8c0a6 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f0b35feed7e4ecdb6e9f9928883e6ec : SUCCESS in 3m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f4a6542ab6044c66bd63a9ddf00f42b9 : SUCCESS in 4m 20s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b619032806134d52af9e547a9de53f99 : FAILURE in 28m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/febb86a8186e4d0ea7210daf9538f896 : SUCCESS in 26m 52s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6492e638602d44e189d0814ac307421c : SUCCESS in 29m 40s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fcccdc80abbe4ef79c1e6241a9121f21 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":3},{"id":"924f927a50b02de99f0393f6c585d685119f3f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 00:16:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/126f2e37d65142d7ba818f4a8a6e731e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21a9af43fb3741bc91ea4aeb87e04788 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a785dad60534702acec1944088444ec : SUCCESS in 21m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/62d7e9c57d1347f3a828c393bb3444f3 : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee2de429e7b242ba7046ef6eb7382c7372d65d94","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 21:30:48.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-935565-lw8bb/index.html\u003e","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4bc7c6fcd4b20cdc9a4c14218058f1f3f9684d5b","revisions":{"507aaaf5b1647207fcaf4d6c8aa7abf118695f74":{"kind":"REWORK","_number":1,"created":"2024-11-18 21:24:41.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/65/935565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/935565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/935565/1"}}},"commit":{"parents":[{"commit":"1acaf899a6964484e5b5be4337618ebbe6ca8dbb","subject":"Merge \"Clean up the remaining logic for Windows OS Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1acaf899a6964484e5b5be4337618ebbe6ca8dbb"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-18 21:21:19.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-18 21:22:35.000000000","tz":0},"subject":"tests: Functional reproducer for bug 2088831","message":"tests: Functional reproducer for bug 2088831\n\nThis reproduces a bug where a server create with hw:mem_encryption\nfails when unified limits quotas are enabled:\n\n  {\"badRequest\": {\"code\": 400, \"message\": \"Memory encryption requested\n  by hw:mem_encryption extra spec in zqdxonddgkblwxarzpjj flavor but\n  image None doesn\u0027t have \u0027hw_firmware_type\u0027 property set to \u0027uefi\u0027 or\n  volume-backed instance was requested\"}}\n\nRelated-Bug: #2088831\n\nChange-Id: I8d6de19f0f66e9536671da23eea1f81faf71b16a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/507aaaf5b1647207fcaf4d6c8aa7abf118695f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/507aaaf5b1647207fcaf4d6c8aa7abf118695f74"}]},"branch":"refs/heads/master"},"d760c9c274ff277ec6439c1e7c8fb6d65b4b2d9a":{"kind":"REWORK","_number":2,"created":"2025-01-23 18:20:34.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/65/935565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/935565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/935565/2"}}},"commit":{"parents":[{"commit":"6f537e756bdcaa1d86a629940abf3775b3d56c80","subject":"Restore nova.virt.__init__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f537e756bdcaa1d86a629940abf3775b3d56c80"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-18 21:21:19.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-01-23 18:20:15.000000000","tz":0},"subject":"tests: Functional reproducer for bug 2088831","message":"tests: Functional reproducer for bug 2088831\n\nThis reproduces a bug where a server create with hw:mem_encryption\nfails when unified limits quotas are enabled:\n\n  {\"badRequest\": {\"code\": 400, \"message\": \"Memory encryption requested\n  by hw:mem_encryption extra spec in zqdxonddgkblwxarzpjj flavor but\n  image None doesn\u0027t have \u0027hw_firmware_type\u0027 property set to \u0027uefi\u0027 or\n  volume-backed instance was requested\"}}\n\nRelated-Bug: #2088831\n\nChange-Id: I8d6de19f0f66e9536671da23eea1f81faf71b16a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d760c9c274ff277ec6439c1e7c8fb6d65b4b2d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d760c9c274ff277ec6439c1e7c8fb6d65b4b2d9a"}]},"branch":"refs/heads/master"},"4bc7c6fcd4b20cdc9a4c14218058f1f3f9684d5b":{"kind":"REWORK","_number":3,"created":"2026-08-26 20:46:16.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/65/935565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/65/935565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/65/935565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/65/935565/3"}}},"commit":{"parents":[{"commit":"b2002fef0b180f00ca1d00f235742b9cc8577289","subject":"Merge \"doc: Add documentation for Intel TDX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2002fef0b180f00ca1d00f235742b9cc8577289"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-11-18 21:21:19.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-08-26 20:43:28.000000000","tz":-420},"subject":"tests: Functional reproducer for bug 2088831","message":"tests: Functional reproducer for bug 2088831\n\nThis reproduces a bug where a server create with hw:mem_encryption\nfails when unified limits quotas are enabled:\n\n  {\"badRequest\": {\"code\": 400, \"message\": \"Memory encryption requested\n  by hw:mem_encryption extra spec in zqdxonddgkblwxarzpjj flavor but\n  image None doesn\u0027t have \u0027hw_firmware_type\u0027 property set to \u0027uefi\u0027 or\n  volume-backed instance was requested\"}}\n\nRelated-Bug: #2088831\n\nChange-Id: I8d6de19f0f66e9536671da23eea1f81faf71b16a\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bc7c6fcd4b20cdc9a4c14218058f1f3f9684d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4bc7c6fcd4b20cdc9a4c14218058f1f3f9684d5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
