)]}'
{"id":"openstack%2Fnova~937223","triplet_id":"openstack%2Fnova~master~Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f","project":"openstack/nova","branch":"master","topic":"bug/2091138","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-06-12 08:02:20.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-06-11 18:38:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-06-12 08:02:20.000000000","reason":"Change was submitted"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2026-06-12 08:02:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f","subject":"libvirt: Add conf option for live migration monitor interval","status":"MERGED","created":"2024-12-06 02:54:06.000000000","updated":"2026-06-16 10:46:37.000000000","submitted":"2026-06-12 08:02:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937223-bug/2091138","meta_rev_id":"5a4be8a723f1998ba8a5a4c701faedf9176f7fd2","_number":937223,"virtual_id_number":937223,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-12 08:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-06-11 18:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-06-11 17:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-06-11 18:38:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 03:37:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-06 06:17:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-10 20:19:08.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-14 21:29:07.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-03-05 07:47:18.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2026-06-04 19:07:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-06-05 04:45:38.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-06-11 04:10:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-11 16:02:11.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"a66ea163c4ebbbbe98048b198515945f1996959b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-06 02:54:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c67f24952b49cade996e677a6cb2ce98fd92028","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 03:37:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629dd21fdfed4512a3106fff1ce87814\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4e2e5f7bbd7477aa4374a7a97f495c3 : SUCCESS in 31m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6742d36e07f44e9a9157ad4003f99a5c : FAILURE in 40m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5753fd592c773338eeca9bd090d10fabda941d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 06:17:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd41c895b8654ddda1b0fdfe90196ae8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14acfbfa977b4a33a50db9b5b8bd6248 : SUCCESS in 1h 09m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c2e67d8427b143e29c93b9130685eadc : SUCCESS in 1h 28m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3211b54e5bdb487b9e4f01f5087d7736 : SUCCESS in 1h 05m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86314c03b4984be7b5beef90753a4841 : SUCCESS in 25m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a8504b6f5049d4a2c8eccfd34b63e5 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85b855e8fd304150bb84f1706ed9a4e2 : SUCCESS in 14m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd279f98d2884393b458e76e44e14abc : SUCCESS in 15m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cafae4a9379454793a06619bf423837 : SUCCESS in 10m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/495399c07b5d4389a082a0ec5bbfd37b : SUCCESS in 1h 12m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a77f44334e0f4240b1d13b57cad0c84c : SUCCESS in 1h 56m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/caafd0029d0e43c29074816a1c759083 : SUCCESS in 1h 01m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e931d010f30542639459010dd731884c : SUCCESS in 1h 11m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f4e5c87d67024717aec147e122dca755 : SUCCESS in 59m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/da2856c0341b4aeaab0ab1b76d81f775 : SUCCESS in 1h 57m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/dfc293dfcf1f4fb7b3628827bcc5972b : SUCCESS in 2h 27m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/09cc4d2187ee4578947830b9f311b538 : SUCCESS in 34m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/932075b1f69a4488aab6b1ae832527bf : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9963f20ab455441fafec507e24e0fc5f : SUCCESS in 25m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe4e45be463e44b8b0ed2458c7a614bf : SUCCESS in 27m 37s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/19a7ccf62c3043efb5b93d327a56a497 : SUCCESS in 29m 20s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e6e97b36dcc44cac8ad28b8d7b6a7296 : SUCCESS in 1h 39m 21s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/79c9edf0d6d8497b9e890898f8a8cf30 : SUCCESS in 1h 52m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0003488b2d5345adb68a623fe66c1385 : SUCCESS in 1h 39m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0ad1d9c0d4445e9886a5b0c914735b3 : SUCCESS in 1h 04m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0af8dc6500164ee986f704463423e9f4 : SUCCESS in 22m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9fcdb0dbae3b42e8af4ce87a6ca8818d : SUCCESS in 38m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff2350abfec3ec68ece3fd2dacc701c8bab8e255","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-10 20:19:08.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937223-62qdw/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"fbcefdf29494c443dad0f4be38f6b61f2cef299b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-14 21:29:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"09742743b4ee85713e4c6c59695441180565bacf","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2025-03-05 07:47:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"74b6ae808100957b276bece15a43565470911824","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-04 19:07:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e29b2279c1dd6d361957f7c8a8f8a4727ec8e8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:31:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128224f54f42424b95b558d1575671cf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4d1387c192441c58c7531378fcebd7e : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d6f8d1eec8249a4aed8d371d302d57d : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6c9f4229a4b2494aa00f444d53a95d58 : SUCCESS in 21m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32ec88e0745ea4fd9357f9f5351686e29ee7e59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 21:54:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15a36454c3ac42938ef6aa1881a326c3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44dceb22489646f7a594ba83b6c9cdf6 : SUCCESS in 58m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/73f08621e1f9410dabedb4d4e90f768b : TIMED_OUT in 2h 39m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d595a47e23bd42a29e8c170f47ba9670 : SUCCESS in 55m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fde169a916f499f9cd5bbd9e7de0f80 : SUCCESS in 1h 24m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5f4680e7c6404281f8701cb135b60d : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69f302b77e6e4b82af654bae20af684b : SUCCESS in 15m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da733fa89b744d8a9887a5df6954cf87 : FAILURE in 14m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b34c43c892b34dcb9b5cfffecb0d9ac7 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cfa700089404eda8681f382d2fecee3 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a769736b900a41848a61bedb5e739f41 : SUCCESS in 14m 05s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/84aacaef734f40f79d1f1136bb24714e : SUCCESS in 56m 30s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8a62ffa548f347b0b3ef17a367840a9a : SUCCESS in 2h 13m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0a37a584b8084f4f9dd6763d5ad5b755 : SUCCESS in 54m 57s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/e93134a076a949dcbd233d34faa8d8d6 : SUCCESS in 36m 24s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a12af12d0bd546628778821fff69329f : SUCCESS in 1h 07m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2c063af1a4a349c297356273a9307090 : SUCCESS in 57m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/ac8fdb92efaa48a4abd86b7eb4eed942 : SUCCESS in 1h 32m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/47a2e5d45fbe454d9bc6ae9b12ad0a38 : SUCCESS in 2h 16m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5650a965f0d42b68890d28f4391c733 : SUCCESS in 2h 29m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f1b9ee207d2645b6a4fd78c5c86ff859 : SUCCESS in 58m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/25f17ed07682419b85023164916dc42a : SUCCESS in 5m 40s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/be713da413cc4b438808ec9b9ee6ba88 : SUCCESS in 28m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/df3627d5057840d2bbef16d4c6acdbda : SUCCESS in 29m 58s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/05b8ba211f0247ae9f4e3520164501b6 : FAILURE in 1h 15m 12s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdfc42fe718248fd9e301e6b4289f35a : SUCCESS in 1h 45m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62d559603aae4ccb9e8cabc309f72107 : SUCCESS in 1h 04m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b8fbf454a1084cd9b79ca4235b51ee52 : SUCCESS in 16m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7aba4813cc38499a83b8bffb485d319f : SUCCESS in 34m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a460fa4d4eb64dd5a80e7f3de3295f16 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"279c9e1192a4c258dd2854830bd1d0e886f4227a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-05 04:45:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"204d5129fca7e4665eea71b9435aafcc41dcbec4","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-11 03:52:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3e40743025a216b91d8781ef1cb769bed372a26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-11 03:53:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"da490b0ebf14e0c426929d766dcd0827b5ec25d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-11 03:57:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de663efa79c5d2cc9c66ff9fb75452dea60bbb21","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-11 04:10:53.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f .","accounts_in_message":[],"_revision_number":2},{"id":"fa1420c63b8886eebebb3cea0229b5327a8f7c4e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-11 04:15:53.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f .","accounts_in_message":[],"_revision_number":3},{"id":"d02ff0ffbe6676ded5dd8c9dd2df81726016d435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 06:33:56.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4980a1237f1b4cc8acd346cc17fe927a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06c34b5958b44adfbbe212d9f1ba3590 : SUCCESS in 1h 03m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3c5d19e1a97f4541917c18535560966c : SUCCESS in 1h 57m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bae3ca2212844ff8a899caca82c64685 : SUCCESS in 1h 06m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5fae941fe9a450c8eb16d23380b1c14 : SUCCESS in 50m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9290f8f31db349ea9cf6cc7cf44b8a80 : FAILURE in 7m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb8c4fd3aa864863b79a29975b0f34d7 : SUCCESS in 14m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0c61915f01145209bfebd86a866f17d : SUCCESS in 13m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/871d3ee481bc483f86a20b7298759eb8 : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a532ad443446ae883d01427870efc3 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05198b67f1c64493a61c6a9530ca3c86 : SUCCESS in 14m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/60d5be94e15d45dcbe4ea6c2c5119337 : SUCCESS in 1h 08m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e237bfffb304b86b76ab531195a1a25 : SUCCESS in 1h 39m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ae52de678bce48f38f4b23a120f3aadf : SUCCESS in 59m 20s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/c293942f61fd42609b69bae2d847c03b : SUCCESS in 36m 17s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/84017514dbdd411a91f2e893c6750412 : SUCCESS in 41m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2c289ffb9f884429be7a90783bffedf5 : SUCCESS in 56m 20s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e50a1801ab054b0c8a840a17744e9157 : SUCCESS in 1h 27m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e3805aadf43c415e92d5fa4ad12e05a5 : FAILURE in 2h 30m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/b581abd43f6b49a5acf1b4ad8cfd1e65 : SUCCESS in 2h 22m 13s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/24f1375ca0f04400b5ec7848d0242147 : SUCCESS in 49m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d2af7bddc8604315ba093bd5e1d564c8 : SUCCESS in 3m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8fcc4f42ee73480b89ecf22a2ed9cbed : SUCCESS in 27m 05s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b8a586ec4c4345f2ac09db784269bc0c : SUCCESS in 24m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/02eaffb7fc96410a83a7974b721f2f68 : TIMED_OUT in 2h 01m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/861fb7b974d14a1b82692d7a6889d67e : SUCCESS in 1h 35m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8bfd7b0d4684fda861eec923a02b140 : SUCCESS in 31m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/96b0ad1993da4ff0af9e3ab2508238a1 : SUCCESS in 13m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e5d8f13608d04bbc987bec9a3d58b7d5 : SUCCESS in 21m 51s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/658229fd0129493daac9e6cc27ad4089 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":3},{"id":"6356fc780c35688b80b8d41db379f8f37bea93c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-11 15:39:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fc7da2db29a3c655d958da78c6e9f3ebdc8f30bf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-11 15:57:06.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f .","accounts_in_message":[],"_revision_number":4},{"id":"3992c4136fa2fe581ed599d50beef6f29067aeaa","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-11 16:02:11.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"465d34841d307b4d5653e74ecf4255581d6ffee6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-11 17:36:59.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"918aa9640aaa250606ade9f404bed01c0828c061","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-11 18:30:38.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"71f8c8dc33ae58c3f05bbd74a5dedddd09295331","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-11 18:38:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c7227d45438f7d3c4bc21dd532f86961761b3309","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-11 18:38:42.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"54cad192201cc1c7a7a342cf0b21ae11b1fd7e39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 18:40:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4015d80ec2c640f699aa40b67c0bbdba\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d31b9e8303ef4c6a937adfaafdb4f100 : SUCCESS in 1h 33m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/deb3906cc8a4417caa711c671954f18c : SUCCESS in 2h 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44c446701de54111831ef5aa1c9a2142 : SUCCESS in 1h 04m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e108437ac4e4eb9821be987874f9114 : SUCCESS in 1h 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d659ac18efc435a8e5448b454446ff9 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ea56c32737a427899e93c7a1739fdf8 : SUCCESS in 14m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42cce1642a0e4b4a9dce848f726d9d0f : SUCCESS in 15m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/505f0860776d4dc4bb2e767a4f78620d : SUCCESS in 22m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd66206b8f5d49abad3b22136a42bea9 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8f8492238b8482aa2f7be048b2cb2ba : SUCCESS in 17m 46s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/dfcdeef059a94cfdaa9e9ae515ba3c07 : SUCCESS in 1h 07m 35s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9660afec3cc465793fa0f3fa432a559 : SUCCESS in 1h 47m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6e7262b07a4c4accaa64a626bc00421f : SUCCESS in 55m 23s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/61bb76f9dc44499eb90d5f83395d7cfd : SUCCESS in 35m 37s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3657e563c60e452fb62f4a99c5b9964d : SUCCESS in 1h 12m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0dde93e64e844aa193c56f25fe1b3121 : SUCCESS in 54m 59s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d03c2ce4c7864e7490f31d8b1c1460ee : SUCCESS in 1h 08m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/af50226316f94174beff5691ebc2f958 : SUCCESS in 2h 07m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/16b07fffe9fb4c4aa6aad3fd13121359 : SUCCESS in 2h 46m 48s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6b9ebd35841f4bbe8cc420d6f5fc54cc : SUCCESS in 55m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/caee5212580c4f9c9c1aa70019869feb : SUCCESS in 5m 18s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/668dea7fa1e840e59cb69305626680d4 : SUCCESS in 23m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/455c639c109c4e1388359bb42d26e3fa : SUCCESS in 21m 44s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e77c94c5d6cf44429c237ca36e4bb04e : TIMED_OUT in 2h 01m 42s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bea9d670837b44c88af9c33c88674e82 : SUCCESS in 1h 09m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6dc8217bbd949b88a087a4eb7f193b0 : SUCCESS in 49m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5fef0e823ccc4617acde657a186bbf81 : SUCCESS in 21m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/df39139495654ac49a657c08fc05f751 : SUCCESS in 24m 17s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d19f103753694da9960c2cef529e450f : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":4},{"id":"48884a039b021c406e654753112ce192c0b77427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 18:41:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5a2c32620e079216460badfeb2192636609989d5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-11 19:01:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"af251aed17200f6522a6410d944f808fba252a98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 21:09:18.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f91ac4cd8f047abb2d339a50e0468d8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1fe822d6ebcc4d3e80ae04d2a9c06c12 : SUCCESS in 1h 03m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c1eae542b1f74b9e83eb2a1442b62592 : SUCCESS in 1h 52m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3d8319062de4ff7996fb98263d7b3e3 : SUCCESS in 46m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e4c106c86f4a58aa090e27a6f374b4 : SUCCESS in 7m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6443394c20d45ecb2146f1490d4dafc : SUCCESS in 15m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf16dd738fb3406aa2b21cbd4a05099e : FAILURE in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d723bd18574d019df86968834d3c64 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f26e23330d254c069bdbe0531b0d46bf : SUCCESS in 16m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/662a93f9e2724a28917def587f7e3e6a : SUCCESS in 33m 40s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4406fca70c4f489fa24b7932ffe80417 : SUCCESS in 27m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b0313a86ecae43e5a059ceac1debfb59 : SUCCESS in 28m 21s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8d5e0798cb14402cb9508421675f55e1 : FAILURE in 1h 10m 18s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d2154b626028417093e66ff5d3eccf2f : SUCCESS in 1h 46m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/b6b2eb4929d9437f820cad978802b0df : SUCCESS in 2h 22m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ac0aea3b394d44a3aec07623dfde2e3a : SUCCESS in 4m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/525e9dae89df4f4d90312f1ff9c471ec : SUCCESS in 1h 35m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b63a5b0a2de44d5d9190672827f089b2 : SUCCESS in 1h 37m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1ab3ace1dd94b79a39f05c3f7279d13 : SUCCESS in 47m 16s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b9548ca388184c0498fb61b1260e0767 : SUCCESS in 15m 00s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ffb3e27d6b514ccead3a4e458d3829a4 : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":4},{"id":"8b4fdeca616f3168f0d808aca8358126e0ba7ed5","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-12 02:39:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f113b61c7e56be25c1bf3680408dd65697b3b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 05:17:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64f4fe7d5f844ac39ed69d1e2442cd19\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff5f1f0812b046e193ac739874139106 : SUCCESS in 1h 05m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c437733ea777492eaaaa76da3bfe53b6 : SUCCESS in 1h 55m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27f009aa0c084fde8045bf586dd63ebb : SUCCESS in 1h 12m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/549f9f1560a94940bab464d6145c3855 : SUCCESS in 51m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76bafaf39bc94a908866073aaf7e2098 : SUCCESS in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3ab5a01804a477e9e0da1347a02335b : SUCCESS in 12m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8423918b77ac48b1914568efec4ac699 : SUCCESS in 12m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/75da0b0bde384f9fb561bf585b7db56d : SUCCESS in 24m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f52e3342b4b491e976a95e3b9fa1266 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4b89168cfd48bc81aca503fe7aa111 : SUCCESS in 17m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4ae87fb005ae4b60a1717f81bedff0f2 : SUCCESS in 57m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb8e1ee8e2e34ea28a6210192f166799 : SUCCESS in 1h 38m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a138229ac49a4c5c943c849ba195cd92 : SUCCESS in 59m 46s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/cd5630eda92a476d8a525d7d1125ae0d : SUCCESS in 41m 32s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/7f9086bc46f54bcd86faaf332f0face8 : SUCCESS in 1h 03m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bad138511cf542a29c2da41327f5f7ac : SUCCESS in 44m 00s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fe4483f453da4c35b16511e2fff97efa : SUCCESS in 1h 13m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/266cda7b804c4b6bb2f81bba7bd50c65 : SUCCESS in 2h 18m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/472c05af5e664c89941da6406c685a45 : SUCCESS in 1h 56m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9d234d4af8654069a541eae1f71ff6c0 : SUCCESS in 1h 27m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f365f18a3d074f8a927cf484a685a3c8 : SUCCESS in 5m 40s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2432b6a3fc74db4886ebae0b6ed1cea : SUCCESS in 23m 37s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/38e979df47144b1bb6c07c2908f6246a : SUCCESS in 26m 30s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e01891bfe2814b6080535622484f7d7e : FAILURE in 1h 11m 06s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b55cad346ef6403099cb464cccc34417 : SUCCESS in 1h 16m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52c24465974b48f19b0828dab5368689 : SUCCESS in 54m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9e229fbc042742f38ab6e94e64e1ae2b : SUCCESS in 21m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c38d7f74daa646c2afafe05d535f07cf : SUCCESS in 32m 36s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c218842032264269942e1d5bf2997fa7 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6a4f239b135a2e073cb2bb92da083a88b5475484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 05:18:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"836fbcf29aa20b47ff1784596ebb86c513347ae6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-12 05:30:14.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"465f80acfda570efb35dc4cbaf22a394ace79ba5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:02:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf9be299a2d4511a6ad459a16ba69b5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86f055d40f3d42448467a3410b87d7d0 : SUCCESS in 32m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0c2cadb07c684dd9b9c05a51d429bc7f : SUCCESS in 1h 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7feb9fa8814d416ea9831c7b36493998 : SUCCESS in 47m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb3b7d8a0454af099d36b16fb6dec7c : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/077329c142234d75b715937a24eca5d1 : SUCCESS in 13m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56e23e5d28f8429aa0f37d441d6b2e24 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb72d1418ff642cb954cf7a2d3e5ea9f : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7de8e039b9e843eba166eb1f4a10492b : SUCCESS in 10m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/840f2819c78a416a8f677be234421d8f : SUCCESS in 46m 49s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/63c2d189827347ae873ae64d740a36a0 : SUCCESS in 27m 48s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cc99e7e935ef43c4b42bc1f557edb90d : SUCCESS in 20m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dee04f5b5e254270b6157173caebe042 : TIMED_OUT in 2h 01m 12s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/96247d96c65e47bf9ed18d7c47837ef0 : SUCCESS in 2h 03m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/a61fb2320e1c4795abd7d194d5002b1b : SUCCESS in 2h 37m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ced4aa4451ea4f47a58bd574fcd690a7 : SUCCESS in 5m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/43bf1ea10b434b7d831d0e88414b24ae : SUCCESS in 1h 37m 49s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/24a14be9adc646e491b12f37208e3ab8 : SUCCESS in 1h 05m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a44b27f327e4d7fbe37a10a29d055fd : SUCCESS in 58m 26s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3cad8495bf4d4bfeb18578a920afc6de : SUCCESS in 10m 10s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/1532c2fec3ba426cba2bb99bbd1f5041 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":4},{"id":"19c20e45eb8bd99fa48c4bfe5aca35c193be7923","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:02:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8736ad8c4d7cf8ff922b3b93e2a68e94b7d92f5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 08:06:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb37e9e4e044487aaacef5ba98e58a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79cb74742dcd4df9a9ed188c08d5073f : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ebb77da1ec24ecc95f6f52e21a20c03 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4},{"id":"5a4be8a723f1998ba8a5a4c701faedf9176f7fd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 10:46:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54cc0837bf7a434592ceb95600b9596c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/43800b6c85e44fef92269c4b957f217f : POST_FAILURE in 25m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ba8155486cd344a69fe7bcf7f5dc4209 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8e36070a20f34878ab90ce9fbb8e4862 : SUCCESS in 24m 00s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"48ea13af2f5f2771e6717c496333f56926d197dd","revisions":{"67fd665c6a997864c473cd6cb16a5790a440e140":{"kind":"REWORK","_number":1,"created":"2024-12-06 02:54:06.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/23/937223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/937223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/937223/1"}}},"commit":{"parents":[{"commit":"d3b5080483c09ebf72fd972bab431fba9cab7355","subject":"Merge \"api: Add response body schemas for extensions API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d3b5080483c09ebf72fd972bab431fba9cab7355"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-12-06 02:50:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-12-06 02:52:35.000000000","tz":0},"subject":"libvirt: Add conf option for live migration monitor interval","message":"libvirt: Add conf option for live migration monitor interval\n\nCurrently the live migration monitoring thread saves migration stats to\nthe database at a hard-coded interval of 5 seconds.\n\nBecause nova-compute does not have direct access to the database, all\ndatabase writes it initiates must go over RPC to nova-conductor and\nnova-conductor will connect to the database. In certain deployment\nenvironments, long running live migrations can produce a substantial\nnumber of database writes and if the writes are sufficiently frequent,\nthe message queue could become overwhelmed and culminate in\nMessagingTimeout errors for nova-compute.\n\nThis adds a [libvirt] config option to provide a way to control the\nfrequency of database writes during running live migrations to suit the\ndeployment environment.\n\nCloses-Bug: #2091138\n\nChange-Id: Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/67fd665c6a997864c473cd6cb16a5790a440e140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/67fd665c6a997864c473cd6cb16a5790a440e140"}]},"branch":"refs/heads/master"},"0531fdb29101c5a9f7660923ae9eab6ef623a3a2":{"kind":"REWORK","_number":2,"created":"2026-06-11 03:53:56.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/23/937223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/937223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/937223/2"}}},"commit":{"parents":[{"commit":"d8c997c350d745a118b34dfce62701f0013d7a16","subject":"Merge \"libvirt: Remove unused utility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8c997c350d745a118b34dfce62701f0013d7a16"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-12-06 02:50:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-06-11 03:53:39.000000000","tz":-420},"subject":"libvirt: Add conf option for live migration monitor interval","message":"libvirt: Add conf option for live migration monitor interval\n\nCurrently the live migration monitoring thread saves migration stats to\nthe database at a hard-coded interval of 5 seconds.\n\nBecause nova-compute does not have direct access to the database, all\ndatabase writes it initiates must go over RPC to nova-conductor and\nnova-conductor will connect to the database. In certain deployment\nenvironments, long running live migrations can produce a substantial\nnumber of database writes and if the writes are sufficiently frequent,\nthe message queue could become overwhelmed and culminate in\nMessagingTimeout errors for nova-compute.\n\nThis adds a [libvirt] config option to provide a way to control the\nfrequency of database writes during running live migrations to suit the\ndeployment environment.\n\nCloses-Bug: #2091138\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.172\nChange-Id: Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0531fdb29101c5a9f7660923ae9eab6ef623a3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0531fdb29101c5a9f7660923ae9eab6ef623a3a2"}]},"branch":"refs/heads/master"},"d65711b4f92aff304c5a7cb998238afbe38d5f11":{"kind":"REWORK","_number":3,"created":"2026-06-11 03:57:12.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/23/937223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/937223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/937223/3"}}},"commit":{"parents":[{"commit":"d8c997c350d745a118b34dfce62701f0013d7a16","subject":"Merge \"libvirt: Remove unused utility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8c997c350d745a118b34dfce62701f0013d7a16"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-12-06 02:50:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-06-11 03:56:49.000000000","tz":-420},"subject":"libvirt: Add conf option for live migration monitor interval","message":"libvirt: Add conf option for live migration monitor interval\n\nCurrently the live migration monitoring thread saves migration stats to\nthe database at a hard-coded interval of 5 seconds.\n\nBecause nova-compute does not have direct access to the database, all\ndatabase writes it initiates must go over RPC to nova-conductor and\nnova-conductor will connect to the database. In certain deployment\nenvironments, long running live migrations can produce a substantial\nnumber of database writes and if the writes are sufficiently frequent,\nthe message queue could become overwhelmed and culminate in\nMessagingTimeout errors for nova-compute.\n\nThis adds a [libvirt] config option to provide a way to control the\nfrequency of database writes during running live migrations to suit the\ndeployment environment.\n\nCloses-Bug: #2091138\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.172\nChange-Id: Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d65711b4f92aff304c5a7cb998238afbe38d5f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d65711b4f92aff304c5a7cb998238afbe38d5f11"}]},"branch":"refs/heads/master"},"48ea13af2f5f2771e6717c496333f56926d197dd":{"kind":"REWORK","_number":4,"created":"2026-06-11 15:39:37.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/23/937223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/937223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/937223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/937223/4"}}},"commit":{"parents":[{"commit":"d8c997c350d745a118b34dfce62701f0013d7a16","subject":"Merge \"libvirt: Remove unused utility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8c997c350d745a118b34dfce62701f0013d7a16"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-12-06 02:50:47.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-06-11 15:37:39.000000000","tz":-420},"subject":"libvirt: Add conf option for live migration monitor interval","message":"libvirt: Add conf option for live migration monitor interval\n\nCurrently the live migration monitoring thread saves migration stats to\nthe database at a hard-coded interval of 5 seconds.\n\nBecause nova-compute does not have direct access to the database, all\ndatabase writes it initiates must go over RPC to nova-conductor and\nnova-conductor will connect to the database. In certain deployment\nenvironments, long running live migrations can produce a substantial\nnumber of database writes and if the writes are sufficiently frequent,\nthe message queue could become overwhelmed and culminate in\nMessagingTimeout errors for nova-compute.\n\nThis adds a [libvirt] config option to provide a way to control the\nfrequency of database writes during running live migrations to suit the\ndeployment environment.\n\nCloses-Bug: #2091138\n\nAssisted-by: Claude claude-opus-4-6 claude-code-2.1.172\nChange-Id: Id3fc2f56c9e3f1d9250c412f95481ff9b4aeac8f\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48ea13af2f5f2771e6717c496333f56926d197dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/48ea13af2f5f2771e6717c496333f56926d197dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
