)]}'
{"id":"openstack%2Fnova~937275","triplet_id":"openstack%2Fnova~master~Icbfc568bead950a3be22784746421cc11382b2d7","project":"openstack/nova","branch":"master","topic":"bug/2126751","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-10-31 16:29:56.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-10-13 11:55:31.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"34420":{"account":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"last_update":"2025-10-13 11:25:06.000000000","reason":"removed on reply"},"37702":{"account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"last_update":"2025-10-31 16:29:56.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-10-31 13:32:41.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-10-31 16:29:56.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-10-31 16:29:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbfc568bead950a3be22784746421cc11382b2d7","subject":"[CI][nova-next]test with placement ac optimizations","status":"MERGED","created":"2024-12-06 17:41:30.000000000","updated":"2025-10-31 16:33:06.000000000","submitted":"2025-10-31 16:29:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"937275-bug/2126751","meta_rev_id":"0951c6c30b910c19c3a90b8159680ea49d24e88c","_number":937275,"virtual_id_number":937275,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 16:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-10-13 11:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-10-31 13:32:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-10-31 13:32:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-06 18:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-06 20:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-10 01:38:37.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2024-12-11 05:45:25.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-13 15:01:25.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-23 16:14:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-01-31 13:12:34.000000000","updated_by":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"reviewer":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"state":"CC"},{"updated":"2025-02-05 17:10:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-02-05 17:13:11.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"c071ef5c019bd9014a3bd5479f8b3d4b00c32293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-06 17:41:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"368c774206d4212438dd45fb08c1fc7217513a47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 18:30:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d3d9a886a724c3395c95ab62b02acf6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23c8d8cfdd5f4c2788972bff1ca6314a : SUCCESS in 30m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05258515f93641d2b769b163a17f4593 : SUCCESS in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d361bca6b5052877571d1a45a2ed1dacfde8290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 20:33:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16a6284f32884393a64e298b155aa47d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70b08fe68a654d5289ca8f2000989eed : SUCCESS in 2m 58s\n- build-openstack-api-guide https://zuul.opendev.org/t/openstack/build/768479919a594fce96b1d6ccea0443ab : SUCCESS in 5m 21s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/643555e7d69749c58d0184ce26e07853 : SUCCESS in 5m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d5d385dcc9c428391a35efdb0550a8e : SUCCESS in 3m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a4fdb9791f44cc989fa3dd15ce44086 : SUCCESS in 58m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/217981a76c484c01afc8d144944e0441 : SUCCESS in 2h 05m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c09f5038cc0f4e0996b1aa46532fb2e8 : SUCCESS in 1h 06m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e52520416b34a9ca9cc4a80e452d256 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d67a9715e14e3a9c23423ad49d9e50 : SUCCESS in 13m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/514334f06b5643e695124dce34740eb6 : SUCCESS in 16m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95e851b904fc422ebdcf274283f85338 : SUCCESS in 16m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae308f3ad1b34452b5343cc72816d008 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dce40d4bb3bc47d5b460c6caaea31e6b : SUCCESS in 34m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/318fff6b7ecc4118940c369de3e055ed : SUCCESS in 48m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a01a7b5b22814263be913efa7571ecc4 : TIMED_OUT in 2h 36m 00s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/75d382089c7f45038fe81416292201e1 : SUCCESS in 1h 22m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9c1dd3fac22240ff9545c03413467c28 : SUCCESS in 45m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/100a0eb4d00f444386ee96bf6ee0d35d : SUCCESS in 58m 47s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fb2d36d1d8ed4897a01dac0338bdd4c2 : SUCCESS in 1h 06m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5fa2f9bfd10246e4813311e31802125e : SUCCESS in 1h 54m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/e4985d288d8a4c8e8fa5961ae5e7cafb : SUCCESS in 1h 25m 52s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2fe46b6f4fee40d5beb4b326c78ea9c0 : SUCCESS in 36m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d46334e682ef461191db0b93c5121728 : SUCCESS in 5m 37s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94ae94dbacee4b17927e0c90a30b545f : SUCCESS in 23m 10s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3c2a1857e6af4c209ddcf2811ccea08a : SUCCESS in 27m 14s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/79cf3a6060ef4c20baba2f63188edb28 : SUCCESS in 25m 38s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9908efd394ec4b87a2cbab03584728ad : SUCCESS in 1h 45m 20s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/66210dc7c3e1488ba1613b24f9a756ba : SUCCESS in 1h 53m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b5c824d10d9407599d4577b741a75c2 : SUCCESS in 1h 19m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13dd2d264b8843f3a86f068f4eb19954 : SUCCESS in 1h 38m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d8e0ac59d4af4f2c96ea5c4487385912 : SUCCESS in 20m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/67e83cb8690d4b489253c9fc16ada6ed : SUCCESS in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fc4f3c153902ef2f531640a70b596082b6bb3e1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-09 12:21:56.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"346dac8e351a029804cab1b5adf032239c1f503c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-09 12:26:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ddfffd97a694a51fc26ffabeb9a2fdfff1e2c81e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-09 12:27:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb7d04aef4668cbf8528fbc0ab5f62ab7e250ad8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:15:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a63a46015335458b8510563ffc238898\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bad04dff83fc45388fa6ade7e72c8460 : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52694dd9975643bd9177ed753beb27b8 : SUCCESS in 29m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6199789d280a5bd0684f2d7642d6f94f9ad57f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:40:59.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/936658 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"a15e954a21b5f603d3867d12d4a15bc38c5f9caa","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-10 01:38:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec583751a0fd5b34c28397dd6d9b2a19fe1bab77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 02:27:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce6d78976c049e4b9562ad307876b3d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f74aa701fd694ae8ac16b3fdaa47d0ba : FAILURE in 38m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45ee5aa7be2e454d9fc3d40ddb5e8ecf : FAILURE in 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8e954313450fdf9e096c724731e0ef7fafd988c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 04:01:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e7d2524fafa448bb18c65bf70c13984\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab401ce96c3646f6b6f8715f328799e1 : SUCCESS in 1h 10m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aceea46d8578490c961ec885dfff1c73 : SUCCESS in 1h 48m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/871fd471db7e4792ad00f05003fe1eaa : SUCCESS in 1h 12m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d75658490b5f42a184e2b6d4856f5fa9 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f52297e5b9343b988c11970d1c1a869 : FAILURE in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8344ca508ccb44cca274ee98d7d44dfe : SUCCESS in 13m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/781365ab95a04b70b1aef438d548682a : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80869715a7d44e77b784e8c69919172a : SUCCESS in 7m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/88459a79fe1942f48a8cc0751a3d441b : SUCCESS in 52m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd8a424ea5ed4564837735aaad90fcfc : SUCCESS in 1h 35m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/75319630907445e2903cc2c0d740cb98 : SUCCESS in 49m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2c3b576561964b01b648acb5b3ad781b : SUCCESS in 48m 21s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/0be513897f884fa3b78a73d2f4221435 : SUCCESS in 1h 05m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d6fcb2bab78d45bc8b783a1365c7ddf0 : SUCCESS in 1h 57m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/36f3f9fd144b4fc4b85cec05a2bb29da : SUCCESS in 2h 15m 53s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a3cb602ee70e49b88b1b84bcaebecc03 : SUCCESS in 33m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/41181eef41ce4cd3836439ca36c0b390 : SUCCESS in 6m 08s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e34c7635c9b4d569062467d21a1a215 : SUCCESS in 25m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f50a0377aed4012ae8e199fa1d4c719 : SUCCESS in 19m 15s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f49cf17237424ee5ab9fcf27e9559fb3 : SUCCESS in 24m 04s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2dc04c6fca00448c8c3978c15c3fa42a : FAILURE in 1h 44m 14s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/57c1f9ab9c83443bb2a5344ef7155ba6 : SUCCESS in 1h 11m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6290e9a517d44aad872232f84a1fb276 : SUCCESS in 1h 22m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99accef81d5b4e6b95c5457d13d9ab4c : SUCCESS in 1h 04m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6b572f0ff55f4c8da796428e4b22b53f : SUCCESS in 19m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8c3b430681c64be5b7207208e398ef24 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5df2a4f15bcb5e61b479ca7259804f00e7b1ea2","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-11 05:45:25.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-87nr9/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"676dd16af50ec24e8f3a0547804e85c27953ce8b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-11 17:08:01.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-lrvd7/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"ce7cb48e492d4c5aaaaea644a05ef183a7d038a6","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-11 23:05:22.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-64rjr/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"83fdf0fc89e7254bce8801b2fd5547aef5daae01","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-13 09:41:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f91c3b2047b1b9a581dff16c4046303264d7110","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-12-14 01:52:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdb671b74716f8c569d8d0530370c85dae9c681f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 12:41:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b04329f38349593c9bc64c2605d8be076eedf603","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 13:18:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88b03fe7254424190b78e41823372de\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4479f07512da482aab92924d24d8c8a9 : SUCCESS in 31m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/247e059a35a749b8a8950ebef6fb62f0 : FAILURE in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0c999af20446aca0a33731bb5b52c28f52cffdc","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-13 13:27:39.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-vpt29/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"20939931c787a112c51ca8aa3aca3542d3dab6c3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 14:43:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cac087b7c3f3d323d4287b29c8f5ab83a3a20645","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 14:46:41.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6832428b8a927d942093bdc190c487a6d99d4502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 14:53:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a9c0994345ce50a2cabe40b7dfe85896d09f1ec8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-13 15:01:25.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c39ed0a83001a450a54a06f0615e15440d5f9bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 15:06:46.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"71fab5f1767c4b9d901bc5739415016ce2a1ff20","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 15:07:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f0e8bd54ae93dc3985662444f2ec4c861c65c962","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-13 15:46:49.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-qt2gh/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"214c493767347722be3c58924609bd2327e4cbd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 15:55:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f660cb10668407880f052cd5d202a41\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00288dd8d9e3480c8b954d94edbbeb96 : FAILURE in 46m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10249439e3f248d6b222f85774debb41 : FAILURE in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4eb10e1c0f675ded61e6938a01f57ee035fb1eda","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-13 16:32:33.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-qq8x9/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"1096242d0675d358f05ef1bcc6706d1fee04bcfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 17:43:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71c0a6def5b043428e8eac896c61b73e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65d43c9ce6364e57a4d141e6772d152e : SUCCESS in 1h 00m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6821d4015cf040d5881d9baf1e9bf93e : SUCCESS in 1h 42m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88724b2e7e47498c997fe6f9ccb36fa8 : SUCCESS in 49m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a49a5e0b152b4249b7c184cfcbd64e42 : SUCCESS in 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e01230c35546df9013afd1c9c4de51 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f1c9e21fbc14ac9964ba48251b3e772 : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/565ddc94bac24d19b6b6716fec686700 : SUCCESS in 16m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5b08c641bb42c2bdee66ac39b35386 : SUCCESS in 11m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9009f042252048af85a62e93d43504f7 : SUCCESS in 59m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77617bf2ade84134b54b68745541c2c6 : SUCCESS in 1h 44m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0724c2f869964b05b40b183fcb3f13b8 : SUCCESS in 1h 09m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3aa6ab5eb93c42909de7267feecfa738 : SUCCESS in 55m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/69f72024137940d3a3596d57dec65480 : SUCCESS in 1h 17m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/93a3fae537c74189a45553614a729152 : FAILURE in 2h 14m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/10bce1bf131e435ba7565914752652a4 : FAILURE in 2h 28m 19s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e5eedd2446db4bdc896b4709285f45fa : SUCCESS in 41m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/91e942ad7fcd436fae55edbcaeb3024e : SUCCESS in 8m 13s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f3c44bd086a4c49888cbf2397322bd0 : SUCCESS in 27m 11s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/383ad3ddd3734aa4b32b917564ae3096 : SUCCESS in 28m 36s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d39414b252664f77b98c049949462d6e : SUCCESS in 25m 09s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1071c588a6fa45bba1ef3ca2364e68c3 : SUCCESS in 1h 49m 33s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/fab7c430c1df4dbb8496b853fc22ed60 : SUCCESS in 1h 52m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b197252a32cb44668905c4d648920711 : SUCCESS in 1h 15m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2802d45d8f664cbfb083aa70fb7f3df9 : SUCCESS in 1h 00m 07s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b48bdffd71784d2aa3745d9f360b183f : SUCCESS in 20m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6b4da630bd1642b384a165817a8295b1 : SUCCESS in 31m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a0dcfaa2b7473c5f625d56d75891827f7a056d34","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-13 17:50:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"72a77a6a0e8cf45357218c9ec15396b4e2fbb0ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 19:47:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade4b5f374584cd68d829f0e786a298d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5ebf40fd6e34c558781ee2f12975685 : SUCCESS in 42m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c49974828a0d494eaf216f9df765c5a0 : SUCCESS in 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4eeaecf492132f83d6e8932bf1cb4a48fecae0cd","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-13 19:54:35.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-cz4tc/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"1874da61cb5687f4ac74a6dd23f828a414c721be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 20:26:42.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb96a555ca4243ea99dfb5e17d1e7067\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4276ebcf6c8545998311d1fd09fd308e : SUCCESS in 52m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8ff958204346400392a62a9a85da6f1e : SUCCESS in 1h 53m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fd894d914eb48b994e78758040f224e : SUCCESS in 1h 01m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d9d359e125d44ee80d8f65765a0b6b1 : SUCCESS in 20m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e090fdcc6abe43f6ac4ee10290157486 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/160a9228838d47529710e2a57daa5c8f : SUCCESS in 13m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f47ddd888a124930944602c4d85dbdf6 : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45aff1652bbd480088443a1943b7a740 : SUCCESS in 6m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/32ab594d59f847a6807a5ca5e106c076 : SUCCESS in 1h 03m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4553d62f68c14a218edd23a692fd7b87 : SUCCESS in 1h 16m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/343520e252fa425684227a193df454e4 : SUCCESS in 55m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e953feb4d4fb4c118f957f9b80b0b104 : SUCCESS in 59m 44s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/391b0319933746fdb404e1563232f4d6 : SUCCESS in 59m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e2b3157660354d6b982949b40c6fedb0 : SUCCESS in 2h 08m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/76da5f42bc62443fb9040332f7953fdb : FAILURE in 1h 12m 41s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5678f4791ab046d2ad2e587f27d57a72 : SUCCESS in 28m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a5476bc041694582bdd811c4281acfb5 : SUCCESS in 3m 46s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/790133d82c57425fa58901a01dbfde81 : SUCCESS in 28m 50s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c71c736b96ef4a829d9d6bfc15f97b52 : SUCCESS in 26m 35s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/25d04093f0a14ea980dc58774f68058f : SUCCESS in 34m 04s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d6ed7812f244e26bbd6b5407cc9f676 : SUCCESS in 2h 26m 13s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/a079be13c39643d3bd02cccaee83124b : SUCCESS in 1h 45m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/903aeb1b0c98461ea2345ef165b02ffe : SUCCESS in 1h 25m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36fe523c9495480dbe5de10951a11d5c : SUCCESS in 52m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e35f2e26e1f44bf0b734413dd1d5d16c : SUCCESS in 32m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/71a9d855b0ce4efd8623b8c34f95fcc1 : SUCCESS in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0a04adf3ca65ac7d12be38472cee8796b72740bb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-17 13:51:30.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04e033794df12b78323d2df8ee18e9da46281ce5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-17 14:04:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"365f0ad0f4d1a5a0d6f0b60c1793a74e2c5a8ede","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-17 14:06:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5e49df03fe233dbf39aa20ab4a3fc9cf222b62b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:43:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47ac3fe5d0d8415d8ffd92be020e48e7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52488e8fdf774d3e95e47eb01b1670d4 : FAILURE in 31m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e891e026eb5a41b6bc7b809ba9339047 : SUCCESS in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0817e24e96ad32ebe5e34faaf2922ef9e986b387","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-17 15:56:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7194814cab2443948a294cb5d4f9b3c791fbab82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-17 15:59:26.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5cd2b0dbd5ee6cb5508cf17e357d191088b3159d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 17:58:47.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40856806aa7f4601b107058d1b24cfd6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b36be2edf8304cb2a28e4484f80e40ca : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7a6fb80b1dc0496785ceccbda8f6233e : TIMED_OUT in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f45f92662c86ed221874b495326aafc32630759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 18:48:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e21cb9fcbcc45728b4b65a4605d3b49\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e72a503988bb4dd8a4c8d7aeeaee18b9 : SUCCESS in 1h 01m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c4079dd3258149999350679876cc0f11 : SUCCESS in 1h 52m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/306c75f79c724771b2c0defaf5526482 : FAILURE in 1h 06m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0d127c2a3d24cb7aa58b226fdaadcc7 : SUCCESS in 23m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48112eeefca47dfa5583d716dc2bf03 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff3835f5c23425c86212824a3d5ec8b : SUCCESS in 13m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62df84ff35f74ec694422caef3b44a59 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45fd1d3ab312460ab63a7a7f19b599e6 : SUCCESS in 11m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b71c9eb8ea3f436da53994576a9e1cb6 : SUCCESS in 38m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/66b950026e96494fa3501c2cc06d79a1 : SUCCESS in 1h 21m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a56f3aa7daf84cabbe8cbbbdc37d59ae : SUCCESS in 55m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/332d475f8719416ab8db5e417ec2fc51 : SUCCESS in 52m 14s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/86ea9c5cb1a74f19b6aa9595a84d7c9f : SUCCESS in 1h 11m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b4ff5d3b47f349abbc53bc44fd437127 : SUCCESS in 1h 42m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/75273138712f4fa8ae6d49281e53526d : FAILURE in 1h 59m 54s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/bfe95f1a16f1491a907e8f5f708f87c9 : SUCCESS in 34m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a68f756fa1c241efa269c3804d7bc97a : SUCCESS in 3m 22s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3e553b0e21644209775e1d248dea02d : SUCCESS in 28m 31s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e958ca9c56a346c4948ec9b6e89ebd36 : SUCCESS in 24m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3bb0ebb2173349b1b4bd5670cfd6a47f : SUCCESS in 32m 58s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2fea6aa687ad42598080b2d3cd22ec8a : SUCCESS in 1h 45m 13s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/5459342ec7d740268ef6805ca0d68352 : SUCCESS in 1h 39m 54s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9454fdfeaab04c9aaf807d1d7456ecae : SUCCESS in 1h 32m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d246711312134c11a2d5f47a9e11ff43 : SUCCESS in 46m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/eefae4add573400c9c02c21c24e72dad : SUCCESS in 23m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b75c405499064ea499034316b5900c08 : SUCCESS in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8eb657a4e94d37eded3910d3688d027c1311360c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-20 15:38:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3be7fd7ac566f493c7bcc0030655daebf237b9c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-20 15:58:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7723d6ddbf20432921347d8c9f8181956609a31e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-20 16:00:18.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fcc78aff089e39ca0a0015bf7fd312a0c6e216cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 17:31:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99d4d89668a423da383a1ca35bdb349\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/237c4abe333c46518575b2918529da7e : FAILURE in 33m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c17ba7dfc65f4654b3d84572b418d02f : FAILURE in 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a172a2bb03d181443c62640f63d5f8b846bb5fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 18:28:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dccb7db5bab34991b155ece1c0a5a407\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/66ddf436771240e089d249301e72a08b : SUCCESS in 1h 02m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/42b185a160d64ce396344908e8f36d8f : SUCCESS in 2h 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fa434b18df3419f88fe4f9158cc7c1a : SUCCESS in 1h 04m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9c34b8b34a34889ad1e8cc1a6423774 : SUCCESS in 27m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a646f9f50343f2b93297329c026365 : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6fcc980fb2545748bf45b0625ee0901 : SUCCESS in 11m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c96477266e8f4e7d880f593ee96a1653 : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5468539aa7024a0a8d5371a3ac28552b : SUCCESS in 10m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a59b9eaab1924c3fb6177b1ae405edae : SUCCESS in 44m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9269013089c8419fb3e8fbcf8252175c : SUCCESS in 2h 03m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/053e9b2be23842ca9f5de3084f9ca777 : SUCCESS in 54m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5a780212c8cd4a6889855b7a6d039fa4 : SUCCESS in 1h 18m 59s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9aa8a18556ed4cfcacf03db3ebd0d065 : SUCCESS in 1h 10m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9ac3f1f23ac14dfaa510a9769088235c : SUCCESS in 1h 52m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/f79a0a71a7bd46b2bf35bd66f3b037f9 : FAILURE in 2h 18m 17s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/37e13b2650624b75b272efcdd16a6fa4 : SUCCESS in 40m 45s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5128227db769434d8bf51e52cbf438ba : SUCCESS in 5m 33s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96cad2e6855f4901b2ff414665406278 : SUCCESS in 25m 25s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c5d9a22f14f4d069d75919adfd5657c : SUCCESS in 26m 46s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/43b33d17f9de4ce0874fad0d4a9c5475 : SUCCESS in 34m 16s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/801d588b88d64a9f8baea7ea2e077b2e : SUCCESS in 1h 54m 26s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/feb2606fe09c4b5e88c79cee13d66a84 : SUCCESS in 2h 04m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/537242d4484b4dcf8a8a81613536a0e6 : SUCCESS in 59m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1bf2e314d2a4b3fb2b7dfff22926021 : SUCCESS in 48m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0ec75d86c34f442c881cdc61c400c0ef : SUCCESS in 23m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8b3882f71c1045de9c7f5dba71461848 : SUCCESS in 25m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4adda4be7ebefcb80d11505ab0302959dd7068f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-21 13:39:54.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3f21dada998bcf9aa8211d2d454842ba13f5d89d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-21 15:55:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6d5d3dca577cd3e8f7d65c3f7289d732d04b1207","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-21 15:57:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f5cac95d77f2cff0b632d203670ceb7b893018ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:13:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9d153414724d5f901ee4017cd59d87\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed2d09201ac54866ae33ed879e2be900 : FAILURE in 37m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8d243b300354c46ab7e18ab7411a2ac : FAILURE in 56m 48s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4a36c88ac7d0c94b4cdb38365e88a3a17c7f31e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:15:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/622e5fd0dc774481aee72dd4d7d26b8d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6078e6461af4c76b931e561c64931d8 : SUCCESS in 1h 04m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e61d4846b57f47c3b88e100e7c0bde8e : SUCCESS in 1h 44m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db48fe33b9904305afaf138f62ccb1c5 : SUCCESS in 1h 08m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf357ce1d3fa41be82211b2c1fae0617 : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e00a096632664e37925eaae37b757b0f : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b4a51672cd946398150d8111dacfe9f : SUCCESS in 10m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/825f9cfeda124e9891be48670ec80da1 : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb91610ee7aa4fc480b2fa30c80f7916 : SUCCESS in 9m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5980eed57ab84523bbe28e4656cc1051 : SUCCESS in 1h 14m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c1d0d544cdd41feae978717f9395722 : SUCCESS in 1h 39m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d2abac6ed8ec40df8c9b692b7e09a1e9 : SUCCESS in 51m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cc586d0c05bb4a99a63ce7fd95b34150 : SUCCESS in 33m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/df8bb74e6f1741089523ef2cb87ad51f : SUCCESS in 1h 06m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5ab0533201be452ebd84d5a0d5e1e8a7 : SUCCESS in 1h 55m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/9081f28624864a39878e9fde7a5527cc : FAILURE in 2h 28m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f354b5cbd77a42d2bb817b778d2d0210 : SUCCESS in 28m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/38d918905fce4ea385e4e80cbd1bb45c : SUCCESS in 5m 49s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbc8c780ef794b0e9d491de3afbda5c8 : SUCCESS in 24m 46s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5995c31be7d444769130c3cd92621639 : SUCCESS in 29m 17s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7a8f27514eba46f6bd11ed9aeea22bd5 : SUCCESS in 29m 42s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/259913a95e834d3ab67988c4191cf864 : SUCCESS in 1h 57m 43s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/2e3dce3a38464e0faa104c9f2d36c19a : SUCCESS in 1h 46m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b859fed6aaf4f62bef254dc2364a4f3 : SUCCESS in 1h 29m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bad927e654934640bc82e72c80d34042 : SUCCESS in 1h 04m 32s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e66e4602f66942f8acd22d4a452f7b40 : SUCCESS in 29m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f1cb8f24f1154d74bbc187b76cb9eae7 : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a75099f84a6e1b0a9aeba458ff93a58a872c61cc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-21 17:20:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f8c6e6e1c47ec92c3a93bbcf33ee1698acaafa54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-22 16:08:06.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"a9964fd6560e5b66c4289178bb36fe80bf9b0cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:45:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d125d6e9fdb54bac904123b748df26c4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98cf9cc520304838a798a474d808700e : SUCCESS in 58m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2bf6b7cc1fc341b4a83bb89d92b7a654 : SUCCESS in 1h 43m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37c3a68d0bd448da8d7ce5354819c71f : SUCCESS in 1h 03m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9fa881aea054fd0a08b13d0f8bf9780 : SUCCESS in 31m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f330c5a6e934551936cd36db0919a84 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2f78cb1044e4dfca3ef8724ee9a3e67 : SUCCESS in 11m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/830d49a1b7af4377800a3e251f684c02 : SUCCESS in 12m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac36d7144c484533b0b73ce70013a33e : SUCCESS in 10m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f5cfef9bd5b4406a87c4b2037785ba19 : SUCCESS in 58m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/820335e951c646cdb8372b5b12e773fa : SUCCESS in 1h 51m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b57158dbdf2448a4aedbc1d3e4bc41c4 : SUCCESS in 1h 11m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f30f772d8dd84b4db64b4dc3ff37d36c : SUCCESS in 42m 46s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fa3fa7f3d7a8486ca9bed376ce5f231d : FAILURE in 1h 55m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/72ac3780d88f423a975ab386d06bb461 : SUCCESS in 2h 18m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/3d49a0468c8f4f2c90fa48a5cf04a41f : FAILURE in 2h 30m 54s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4a6065becdb640e4965b451f31f3d7b1 : SUCCESS in 42m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/157afcb98005405798d0b9294380235e : SUCCESS in 6m 34s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfd9b60d3a99437aad391ff818b3b2c6 : SUCCESS in 29m 12s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/393dae2982b54c2cb5430c12fd511925 : SUCCESS in 23m 05s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f3e8046cec9948a6baa3710ca49624cc : SUCCESS in 29m 52s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf929a8a9db343d8ad2616a315a3d3d6 : SUCCESS in 1h 33m 34s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c015d3dbde0a4e029b1104095c762a92 : SUCCESS in 1h 46m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1126fd1eb83c4a53a68a48c7153f2068 : SUCCESS in 1h 38m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26709eaf57b840dca6f76a844f9b03a2 : SUCCESS in 1h 08m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ae61bffdfbe043d1a6556727aa2af974 : SUCCESS in 42m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d7b89ba6220a4c49b290cd6b0c6deffc : SUCCESS in 38m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dce1f91b64ab3c52a132336057a59fb00d3bfc20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 19:19:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/310e1c4cdd7c4c6ba19d3bc6e7e3c16a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f4ae40801eb44248e1e0dd345d07c4b : FAILURE in 40m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/02da97cb6879457f92bdeecf88a7cf4e : FAILURE in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cef97a1b1a989bea36da0658cc9d36ccfa7e9490","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-23 09:07:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0f91284923ba2509f773fb60f9ecdffc969a2ea1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 09:50:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66d93427ee8a43dd837309aacfa83a10\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2badd387393f40dcb511ec3f444ccb1d : FAILURE in 38m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ca16390b5dac4d029fc18e1d1865cee8 : FAILURE in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9456339bbe6016b110968b00e9fe82411b354174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 12:10:01.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf87be0748174efda690a14cea6b01e6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8aa77b1a4d82446bab58cb47dcc22a48 : SUCCESS in 1h 28m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d5db79ccd45b4882a08e138b4dfb4b1f : SUCCESS in 1h 52m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69c9329a70604615873cf94c1d7ef2b4 : SUCCESS in 1h 19m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a86c5aff134d70b35f04e8e303810b : SUCCESS in 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ae02357a4c74c179a85e2955cd61205 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3df9bba6ea64883bd66a98b206cf5ad : SUCCESS in 16m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acd91221817d4ce1b525dcbb6b1f6e23 : SUCCESS in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62eab90a00be4c3ba60d8f014d264342 : SUCCESS in 11m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d5db6cb91c547d2b4ffffbbd82b8789 : SUCCESS in 58m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb90e9466a18430283f720f3f0a075d2 : SUCCESS in 1h 42m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/06859f7c499e402cb7600f86953efb0d : SUCCESS in 41m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ae0a82075f5746f3a7e4ee59813deaa9 : SUCCESS in 45m 46s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/dc4d460d25ee44718fbcea4d07046512 : SUCCESS in 1h 06m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e1ede99b901946adb1f202997b54a86b : SUCCESS in 2h 03m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/d78420c6cda24ee78ebfb78ca521f811 : FAILURE in 2h 20m 53s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ae6f88b0707d4210bf3b9c27798295b5 : SUCCESS in 39m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ab88fb7d6a65411a82fb296574a724cf : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/987c0b9d08ad448994cb2f746b7daf7c : SUCCESS in 26m 56s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26fd370b7fb746899535ed29d3c3ae4e : SUCCESS in 24m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/822e3194220645298faba2ff73a76ca1 : SUCCESS in 36m 13s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/edc3eb01e14f4e21894118261b9f1ae7 : SUCCESS in 1h 57m 10s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c34572f1a08843a78d83b22d9d3d5bed : SUCCESS in 2h 14m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/698cc4f9a461408aaf63ed6cabe1c502 : SUCCESS in 2h 02m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42e858ee8355429290346049fa4a2751 : SUCCESS in 58m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0afaab254e26488faa9e9c86d1d74e70 : SUCCESS in 22m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/477fd0061f584413b325b1931d62163f : SUCCESS in 33m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"84b0889bbfbe5402ffd5f600b55a18aaa1d19e95","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-23 13:52:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"75b817eea3e2dafff90a9347e5fdde95548064e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 14:59:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d905c5277149dda37e11b0f474fcf2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84db6764655e49ce9042b0fd2b2c88fe : FAILURE in 48m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4af6a1a06924feaa671a62c257f4dae : FAILURE in 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"477d5ab4c9a11e691e95657184cb2feca6122dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 16:09:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f80f4d5bbc645ed90420372a39b8843\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b16a5350568e4cb6964533dca86292ac : SUCCESS in 37m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8bc63ed647954ae0b352575e42216f5c : SUCCESS in 1h 56m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90c70aabf75b4b36b62bd1bbe5dfc9de : SUCCESS in 1h 05m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aa9f0b3ebb34ca3a227870c0ae9dcdb : SUCCESS in 28m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd64994038a497383f30f2a8e4ba783 : SUCCESS in 8m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a4ca982336448d785b13fee28be9689 : SUCCESS in 10m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1defd158e16b4176a7ca4c64d7211879 : SUCCESS in 15m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfb81161326e43a791cc487aef36fd79 : SUCCESS in 10m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bf969b1489964dd4b73beba251b4b361 : SUCCESS in 1h 13m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5767988a217845a2950f837b2e016e6a : SUCCESS in 1h 42m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fb2c0708aa724387b2b45dc231ef5884 : SUCCESS in 56m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2d67aa9de89b44c389becb11d50778f4 : SUCCESS in 55m 46s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/0048066c3f20406a9f9effb01652eedd : SUCCESS in 58m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6fc74501a148451a90dc52987a26e36b : SUCCESS in 1h 56m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/5a646167a3854cb697669b41d9cb989d : SUCCESS in 2h 10m 37s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ca5ec53b4520479991035869de66a5a0 : SUCCESS in 41m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f39bc67567be4a278fcf247a86d82dde : SUCCESS in 5m 30s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee0363436a584cf1854a69830bc23ada : SUCCESS in 26m 35s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6c015b8fcc54d8eb5926667c31a0067 : SUCCESS in 26m 53s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/97ba69541d894af3a308e03b58c5117e : SUCCESS in 32m 08s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/81fffdcfe6d3471ba16253de0769f5e2 : SUCCESS in 1h 54m 24s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/62bd78dc40c84ecfa6657a512c354918 : SUCCESS in 1h 37m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/55f3304c90b4489fb0d73512de547b6e : SUCCESS in 1h 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f257e8e029c64904b69251f8e00010e0 : SUCCESS in 1h 11m 54s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8a2063dd0239499085947f906882ddd2 : SUCCESS in 31m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a8a0872f310642009673d39186e3a31e : SUCCESS in 25m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4c6872a8f884d37ba112e4bf16dfff8177526b16","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-23 16:14:22.000000000","message":"Patch Set 10:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":10},{"id":"9b1102384c4ec54f7caea1e8163786a90c2d9f5e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-24 12:56:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4d936a397986d12c3c06f5a05d4b86ea0ba92c1e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-01-24 13:00:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c4d575ed8fd49d766cc595309ede35e52762cb30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:33:21.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ea17b99562d450280e8ddf45af94d6b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/074b7dc4ac85492abe8f3df1fd521177 : FAILURE in 40m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f5d866f4912e42c89efe5380fada86ff : FAILURE in 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a28533efad07cdcdc169f3196fd6422530091cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:52:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cccb22622984f25902c316d5b155c19\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4cbac0bf656541c5a0b2260213be1c2a : SUCCESS in 49m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a29965698d9c4cce9ebc1d33e809be87 : SUCCESS in 2h 32m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4467a6dfcab64b37ade0b77a77f612b4 : SUCCESS in 1h 38m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50f58f60e6df4108a29697e92ee80889 : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ed6c2b21a84769a8c6be6ef38167f3 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e740895cb48843b4b188a078182c8f03 : SUCCESS in 14m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2daa5b0b8bc245b2934fe63feecaae19 : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7516723d62364d41ab726213b357f074 : SUCCESS in 11m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bafa7de60b99473883a93b436830055f : SUCCESS in 1h 19m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4987c94b03094f3cbcb7e21d16264066 : SUCCESS in 1h 50m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/36c694b0a1bb496c9480fa9d3a94a769 : SUCCESS in 46m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/48da96385fe2434f921b26bfb1286afb : SUCCESS in 55m 45s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e1e2e53d7bb7463096981359b0567acf : SUCCESS in 54m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/61bd1baa5eeb42098f5da238fd4cbd2e : SUCCESS in 2h 11m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/2cc49dc3f847481a9a82af78705e0d0e : SUCCESS in 2h 18m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/42b870efe925491d96cc2cd6d6a64c63 : SUCCESS in 39m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1090e4e06f3e442ea965d6acbe5f4c57 : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c9c3725cf8443e1a42b9a2c4fbaf29b : SUCCESS in 25m 41s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8010d05db44c4970ad39b4a0516362be : SUCCESS in 25m 45s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ffa00d58da8844e59a956284e4af3810 : SUCCESS in 31m 03s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b935eb350ad4b719d947caa7f272371 : SUCCESS in 2h 26m 43s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/1b2487c2e7b04655ba26e948bc1664a1 : SUCCESS in 2h 09m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e33bbb030a7d4f0281d23ebf3ff1ae68 : SUCCESS in 2h 45m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1022a1f14924e4c9d3667242b525322 : SUCCESS in 1h 30m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e4e7e6958f574cb1a067aeaa7a07b395 : SUCCESS in 1h 00m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3379b37fc4064442aba83172051e5d72 : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"971b5ad826b63d6d2ac073375250b3104264a512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-01-31 13:13:31.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"328256cf7bd7a9eeaf678fa8e33468d288da0a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 13:13:52.000000000","message":"Patch Set 11: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 937275,11","accounts_in_message":[],"_revision_number":11},{"id":"8ab0422ee93a5181ca98a2d1278a312dc95a28de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 13:13:53.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 937275,11","accounts_in_message":[],"_revision_number":11},{"id":"4337636aea07775c01c3c083811aa84f14127647","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-01-31 13:14:38.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"48666582b35dedb8dcacc8f912c40fb98d8dfd3e","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-02-02 09:06:28.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7d6c703b4f2d6ee2bf25eb55283661eb8f55d25c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-02-02 09:06:53.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.yaml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6a45df81f8fe50469911b9f52097c2e581b8a182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 09:07:17.000000000","message":"Patch Set 12: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/nova on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/nova/.zuul.yaml@master\", line 384, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/nova/.zuul.yaml@master\", line 388, column 1","accounts_in_message":[],"_revision_number":12},{"id":"917c0bf332e5794421f231c548ecd5f93831cded","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 09:07:23.000000000","message":"Patch Set 12:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/nova on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/nova/.zuul.yaml@master\", line 384, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/nova/.zuul.yaml@master\", line 388, column 1","accounts_in_message":[],"_revision_number":12},{"id":"8a934a7e8da06657f9aeecbd127f3de8fa008f37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-02-02 09:07:31.000000000","message":"Patch Set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"855e8f1f8d9510516ac85cef717c330079256a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 09:07:54.000000000","message":"Patch Set 13: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/nova on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/nova/.zuul.yaml@master\", line 384, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/nova/.zuul.yaml@master\", line 388, column 1","accounts_in_message":[],"_revision_number":13},{"id":"7cd9ff40aa1178cc915c3fbba2fec9989ce2b13b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-02 09:08:01.000000000","message":"Patch Set 13:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/nova on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/nova/.zuul.yaml@master\", line 384, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/nova/.zuul.yaml@master\", line 388, column 1","accounts_in_message":[],"_revision_number":13},{"id":"fe4c8019f2666e3518b27fd345e6ac8bd9e29a57","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-02-02 09:08:23.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e9f40faa55ad63f1f674bfaf244ff2142822b807","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-03 14:29:21.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-kgxnv/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"c61d6bad19b60d0024289a59a2c56439192291fc","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-03 20:23:29.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-gwp87/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"b3422a95e07154a702bdf16206976d9612e2fdf2","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-03 20:51:20.000000000","message":"Patch Set 11:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-pf7dn/index.html\u003e","accounts_in_message":[],"_revision_number":11},{"id":"ed34c65a3f867314a5bf12b6fd214f670337b056","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-02-04 18:00:52.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"66169ceef614309a3d16d3d4259aa4f0e5c23a15","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-02-04 18:02:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":14},{"id":"3f377a8d716618c26f4d4673b1dab621f69aaafd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 18:49:27.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79ee73cc980e440984935fcd96c35a0d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5b3094dcb5646de9ff2cb47e92552b7 : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b4203191cb54d7498d4d593ddefc265 : FAILURE in 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0a43abd1e1453bbc731ad343e57f19e6a89fdf48","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-04 19:45:14.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-kj6pj/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"4a8297a7f0946a512a6557b924f3e609a1a7b095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 20:09:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32174c2d7a974abebb8f9eda59c17b96\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73c872297da44c3d8abb94ef9b57cf25 : SUCCESS in 57m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9bc8130fabc0411aab54b775a2bd53a6 : SUCCESS in 1h 30m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83d3f3fc6a1d46b68aa878e43dffa484 : SUCCESS in 1h 10m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c676e24318904d86b735e768d25821ec : SUCCESS in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e13c157a84c466fb4a4b72ce85933ce : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2d919b995c944cbab033810444c47e3 : SUCCESS in 12m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb71ed73dc7b4ee29a5c502ab4ea5e86 : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65f714a289c4ec5859aa519b25a58ef : SUCCESS in 11m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eb4bb68d72f94152b9e45ecef1882e4f : SUCCESS in 44m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e8760cbabd14728b533f25711cb11b0 : SUCCESS in 2h 02m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3093d34e3e4b408b98762172d5cb8f89 : SUCCESS in 50m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7395d4fa82d046a4898c4c462d53ea12 : SUCCESS in 52m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/577b4153c1584d1ab8d53f51cc9a8b9b : SUCCESS in 53m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ca611f4833884f12ab6e1dcacf4ecfd3 : SUCCESS in 2h 02m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/35041fa770a14e5ba21290ffda4cb907 : SUCCESS in 1h 21m 38s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d9c423f435a3406b88045d8c383edfac : SUCCESS in 29m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/329dc51f9357458087dbded7fc09c8a7 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64cab2573d8c47d3a2d9b18b8ae85978 : SUCCESS in 22m 29s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b9400b2e4ca429682b53b9949315a3a : SUCCESS in 23m 12s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c40cab05e8184abe93b63ed567b22ac5 : SUCCESS in 24m 22s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c4db18a65e55496b989b23b9c95dcb4f : SUCCESS in 1h 53m 24s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/9554e68504994540b78c39aefe6da5c6 : SUCCESS in 1h 48m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3e8735628de43aea4337b29bcb2265f : SUCCESS in 1h 01m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd650f74bab3495a80d21ad41fbd5f37 : SUCCESS in 48m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4f907478cf9149d4864c1bf7ae88396a : SUCCESS in 21m 12s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7deef73e0cc04d839f9687196a6c1f8f : SUCCESS in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cea5dcf6e02db9acc66ec547f239fd2124d155aa","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-05 17:10:59.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"1ec795e4528348ff3cf3426e55e39ce68c65ef21","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-02-05 17:13:11.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"dabb318f7ac0f4c63df5f68079247a51e8d464a6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-14 08:40:45.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a03f6ebe8ed6b4cb2bf4f7e2626580a2907b40e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 09:25:18.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49cb2177c3a5492c92f2d90c1343fc6e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b85188dc1ec347ddaa23f60230bcc31f : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d7081ea1adeb49e992a6395a1883b70a : FAILURE in 38m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef1fff3f093849cfae7de6e238d2af21 : SUCCESS in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6e8a14a70d9f365ccf3a2b4abc1352a654d2507a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 11:22:02.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eabf4b3d1f8241808d0667e76787160b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/65e7cf1cce2240eaac894ff9f8e1b2af : SUCCESS in 40m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1814e5a59071401d88cfb09618468ecd : SUCCESS in 1h 43m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c2ea2240d44a57a041c3d3b81e407b : SUCCESS in 1h 04m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98c048d870c04f0bb49c2e586b093fe0 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba0300ef2f904181876c48e73f252b87 : SUCCESS in 9m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/275fd9289b2447ffaad6353de0aeb949 : SUCCESS in 10m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b7ca6ac2b0b47fb91ef2f42f6b807ad : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3835ff1727744b6ba1fe749d7134740 : SUCCESS in 19m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d576d0d6404d89bf3996b91b2140ad : SUCCESS in 10m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d65eb124753a4e70b32e73fedda5189d : SUCCESS in 55m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5c6b6ef16b24627ba34ccd052affe3e : SUCCESS in 1h 57m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7343c92626834fafbd75199988248ede : SUCCESS in 48m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ea32d14eae244c4684bd9ebaf2ebc925 : SUCCESS in 52m 15s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f936c2c9ce07471791252ed7ceab8d90 : SUCCESS in 1h 09m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/542805a8f1144bd09ae9b8b6b0863387 : FAILURE in 30m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/61b3b63bd9ae42c38510d42330933959 : SUCCESS in 2h 36m 00s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c4d95bc8beb544338142cc9526707f9d : SUCCESS in 40m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/33b7dd74a2304150a2b573927b789f98 : SUCCESS in 5m 04s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/996abbcac24748a392a5181e9d38b6f3 : SUCCESS in 27m 17s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4b5787047ce4b128fdf257e2e4c561f : SUCCESS in 26m 40s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2b89c2ec18f34afd825433d419436da8 : SUCCESS in 25m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe81aba2279744f1806d9b89122ff643 : SUCCESS in 46m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/574718ccbfac4becb9eb856eb7f4bdf8 : SUCCESS in 51m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e7e3f7500e68497f8bea83a2d1cd7adb : SUCCESS in 30m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/902158ffe98b4834bc4fee2820aae942 : SUCCESS in 28m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b53b4064d4bf73f1b40e198a3bf860ae6db116ad","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 16:24:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"fef35ecf6cd39e07a6a0f88154c3d088cb8a0bfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 16:51:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a677d4af7f1947c5b5ad656977a1f5ba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cdc40228556d4c7ca0a6392c38d9dc24 : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e3637216f0b4472b5d11dc969c45078 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1be1a374b9af44eeba1ca640663f73e5 : SUCCESS in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"99044562c08b70930a51b671d58b2d766278cbc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:35:59.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71228b07c03f44d596e1c86589428728\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f4e983dce73495cba6f887c03456e41 : SUCCESS in 43m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d09513b12adb40a89bbdb2932e08c68c : SUCCESS in 1h 59m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/999eb5667b2147429bf0a5f8cea0ed1c : SUCCESS in 1h 04m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f014c99f248e4a1a85a0e0f019b6f89e : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74619abeb034ca1b2a800bd2263c753 : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/674b0bd51c3045a1b8e39bb310b88882 : SUCCESS in 8m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bb8a1d82d2243d2af9df442bbc63889 : SUCCESS in 13m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a6656db8874482a9d5c6a2b2e532ddc : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9abc2ffdd8ac43f9b1652177f20d1e72 : SUCCESS in 13m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/71e7099907914d48880e468a34430d80 : SUCCESS in 48m 29s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f64dda10fb2c4e618d3503b9de5c7094 : SUCCESS in 1h 50m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/64be9d9fefb74263b6b98ecdb3b44399 : SUCCESS in 54m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9e233582391c469f8df4e07739e123be : SUCCESS in 25m 41s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/bb9b7e49153c432a889d2b43d942a8ab : SUCCESS in 1h 03m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f434d552e0a74875beab0e2127608c2e : SUCCESS in 1h 51m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/710753410c344edc86f24a82a07f7797 : SUCCESS in 2h 08m 04s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/16611097f9b241c799a8259baeb52c98 : SUCCESS in 45m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b9d86e37ffac425bab444e9d3e8865d0 : SUCCESS in 4m 39s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9274d0adef324a7da98456631b9e2a1e : SUCCESS in 27m 23s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e468d4b76724d9dbef379a81bda066c : SUCCESS in 26m 53s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/673208823b58429aa5247c62972b5f58 : SUCCESS in 30m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a65d80e9020411493402fc56101092d : SUCCESS in 1h 14m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89e60e7f8c444c8a809dfd304a621611 : SUCCESS in 55m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/05cd7dbeb1c9419dba13f414363659ad : SUCCESS in 11m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/235e4c452ea34923907a1ef3039a2c11 : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a058f6005c3559af380d1322e921992f27e408c3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-15 19:24:14.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-937275-jpdw4/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"a0749bd20d580a1b1a446317e0807f7d9e616b10","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-13 08:55:43.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"00cbd3609c7903a49a9c7e36b81f06c1e91d0831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-13 09:08:36.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":15},{"id":"38936f256e707c9cd57bf12027c04c20b969a65f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-10-13 09:09:36.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8233f10c9791a7119c35efcde29e12e6974cdef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 11:18:12.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008416dec3bd46fab6c7b00dcd0c1249\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4067990cf4084cd291eb9e0bc8d62b06 : SUCCESS in 1h 04m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/17b1a54e89464ebc9b4f33c159bab764 : SUCCESS in 2h 08m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5504f1d26ee498d8772c02746417af2 : SUCCESS in 1h 03m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8814bafd06a1420780714bcda03dcb54 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cfa9136246b421eb16f6856fc1d0bcc : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8776735fd0346859a1c720056fe2bbd : SUCCESS in 14m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b64caf869ca14611ba9e9730a24681d7 : SUCCESS in 12m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ab15ac7b1ca4b179e1804a912747a37 : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e71583bff67445b1b758a8881de032bf : SUCCESS in 13m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2a994609101f4f07bba2dc73ac433a39 : SUCCESS in 27m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29e430fb07844ca18a6c6f4013eec369 : SUCCESS in 1h 41m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/23095e205e5e43a58d30cd4e43389094 : SUCCESS in 28m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6fb8b39c4400450885cc7b977aeb4275 : SUCCESS in 44m 39s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/13df4169c4c7429ba0c1a2f434d08099 : SUCCESS in 1h 03m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a036973833284ba485ef85b181e3d8de : SUCCESS in 1h 33m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/ce5b70a209e64dbbb1afa914e320b964 : SUCCESS in 1h 12m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2234f4aa80f64e92877fe9b107b550c9 : SUCCESS in 48m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cda8ceb71ae342c7840002af11a1a3b1 : SUCCESS in 3m 11s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca5c3f8b3f104bf689fad8d66b6d286f : SUCCESS in 25m 19s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ae9ac60682bb46b78b50dd806a9018af : SUCCESS in 32m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e7a19a3c1da4f5caa62103c65148a83 : SUCCESS in 1h 22m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92b2c00a5ba44042a26245694461e1a5 : SUCCESS in 58m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/eaf6ea819eeb4a758acc6b51408cf8fa : SUCCESS in 32m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/645c44017e1b4c72a9a6306b108a1615 : SUCCESS in 35m 24s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/69cf1a2ec18f477aa839babe150e2137 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":15},{"id":"07f156cc103ee5de574b6237b7cfef6bb2b9f358","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-10-13 11:25:06.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":15},{"id":"e085ad80544db6e0ae46101b3e3bc868dd1a737b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-10-13 11:55:31.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"20031d33975d1302164bd159a0f47c193f6bbf60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:00:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/961d3b701854484e8efe06ac208f157c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f66a059f6ee141fbabe6cdff8fe4ae70 : SUCCESS in 29m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/078b63ea78744c9280ad04ad58f609d0 : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/af08bee39cbf4e47b483d1c19585be88 : SUCCESS in 40m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a17a6d68efa4dbd572abdba6f60f679a3a403034","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-10-31 13:32:41.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"7846ddc5ae3f53169021a2f8c21a14d55fc48b5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 13:33:30.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"047ab22f93f64e15155f80f3bf6e03b1ac495515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:29:56.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f5d6e75fbd40508ceb4ba55acd0e41\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d3f87c438f94fdcae4a6428f2a84ae2 : SUCCESS in 51m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1c74223239ba4727b3eef2b6531db321 : SUCCESS in 59m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0370798c031b4baf907368d820211484 : SUCCESS in 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78c733db8761412394b673e2e80589aa : SUCCESS in 9m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc88a102c5254b24938f19f2aca1c23f : SUCCESS in 10m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb4708b506ef457db6dff0e1dfbd56cf : SUCCESS in 15m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff35242bb79d4ba3b43b98b6edc49219 : SUCCESS in 14m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4ec4ac2318ca4335b2d0833b639cd74b : SUCCESS in 39m 05s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aac9e821be134a249e4437f082ee36f2 : SUCCESS in 25m 17s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fd405f5be47b4122aef1aab1d0f6abe0 : SUCCESS in 30m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/09ef514f514f4b9cbf8433aeff72a021 : SUCCESS in 2h 06m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/9c05bb3f730740838019f31be12c566e : SUCCESS in 2h 31m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3ada2c9314cc4fcbba17f9e1cb00ed20 : SUCCESS in 5m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85520723280c4f2b9c36e2a536c2c9b7 : SUCCESS in 1h 53m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cea1d09ea88498dabc97919e5536601 : SUCCESS in 52m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/269223b29f7c46f0890c005aa4ecd7ad : SUCCESS in 30m 44s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/47210abb65e8497cb3aabfb30b19af6f : SUCCESS in 9m 35s","accounts_in_message":[],"_revision_number":15},{"id":"6a203b6bc42757b886705c4c817d173ba12c7b73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:29:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"0951c6c30b910c19c3a90b8159680ea49d24e88c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:33:06.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72cc4c3ea57043638ccf9055bced00db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43d46ee85fa549b3890b95d86c8678b8 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"da1fb487e7c5be0d4906cae22de1cbe9a5819ba0","revisions":{"552d26e7badb0256818ab0936a72742904f08432":{"kind":"REWORK","_number":1,"created":"2024-12-06 17:41:30.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/1"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"subject":"DNM: test with placement ac breadth-first","message":"DNM: test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/936658\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/552d26e7badb0256818ab0936a72742904f08432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/552d26e7badb0256818ab0936a72742904f08432"}]},"branch":"refs/heads/master"},"27f414f1731ce27441dce252c615314dcb1f2b3d":{"kind":"REWORK","_number":2,"created":"2024-12-09 12:26:52.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/2"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-09 12:26:39.000000000","tz":60},"subject":"DNM: test with placement ac breadth-first","message":"DNM: test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/placement/+/936658\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/27f414f1731ce27441dce252c615314dcb1f2b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/27f414f1731ce27441dce252c615314dcb1f2b3d"}]},"branch":"refs/heads/master"},"47d2cfb9381528aa6ede20048d04478e2d2f513b":{"kind":"REWORK","_number":3,"created":"2025-01-13 12:41:37.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/3"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-13 12:40:49.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/47d2cfb9381528aa6ede20048d04478e2d2f513b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/47d2cfb9381528aa6ede20048d04478e2d2f513b"}]},"branch":"refs/heads/master"},"f1ad789d763dca7ae45cc5f647568a08af51e0c2":{"kind":"REWORK","_number":4,"created":"2025-01-13 14:53:21.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/4"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-13 14:52:36.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f1ad789d763dca7ae45cc5f647568a08af51e0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f1ad789d763dca7ae45cc5f647568a08af51e0c2"}]},"branch":"refs/heads/master"},"5850e2c5850c751b3de57bcc03e1a893b4a24c7c":{"kind":"REWORK","_number":5,"created":"2025-01-13 15:06:46.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/5"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-13 15:06:39.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5850e2c5850c751b3de57bcc03e1a893b4a24c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5850e2c5850c751b3de57bcc03e1a893b4a24c7c"}]},"branch":"refs/heads/master"},"327d55ca64bbcf7f692ce4c4d99655369aadeee0":{"kind":"REWORK","_number":6,"created":"2025-01-17 14:04:17.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/6"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-17 14:01:17.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nI needed to duplicate the vars:devstack_local_conf section in the\nnova-next job definition to be able to reconfigure devstack on\nthe controller only. It cannot be done via vars as the is fails when\nit tries to apply on computes without placement service. Also it cannot\nbe selectively reconfigure placement via group-vars as then we loose the\nglobal vars config on the controller.\n\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/327d55ca64bbcf7f692ce4c4d99655369aadeee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/327d55ca64bbcf7f692ce4c4d99655369aadeee0"}]},"branch":"refs/heads/master"},"e8b44dd2e56d9b697019f46ff565b40d55166c4a":{"kind":"REWORK","_number":7,"created":"2025-01-17 15:59:26.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/7"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-17 15:58:36.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/922846\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e8b44dd2e56d9b697019f46ff565b40d55166c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e8b44dd2e56d9b697019f46ff565b40d55166c4a"}]},"branch":"refs/heads/master"},"3a9e56c39f997f4c55ed8d2affa40c0b7c146319":{"kind":"REWORK","_number":8,"created":"2025-01-20 15:58:41.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/8"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-20 15:58:25.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/922846\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a9e56c39f997f4c55ed8d2affa40c0b7c146319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3a9e56c39f997f4c55ed8d2affa40c0b7c146319"}]},"branch":"refs/heads/master"},"a193a79f0802766d09ff6a8c428a07ac8bd5ff0a":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-01-21 13:39:54.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/9"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-21 13:39:35.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a193a79f0802766d09ff6a8c428a07ac8bd5ff0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a193a79f0802766d09ff6a8c428a07ac8bd5ff0a"}]},"branch":"refs/heads/master"},"df708a2a074844e4de0ef0f6b6997ad21afa03c9":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2025-01-22 16:08:06.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/10"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-01-22 16:07:16.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939825\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df708a2a074844e4de0ef0f6b6997ad21afa03c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df708a2a074844e4de0ef0f6b6997ad21afa03c9"}]},"branch":"refs/heads/master"},"fdc32d917a760e2d0254b4af3ea78b1b09325a21":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-01-31 13:13:31.000000000","uploader":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"ref":"refs/changes/75/937275/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/11"}}},"commit":{"parents":[{"commit":"693515bc16014875a347d7d9bf1a1ec62292ef3a","subject":"Merge \"api: Add response body schemas for assisted volume snapshots APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/693515bc16014875a347d7d9bf1a1ec62292ef3a"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Salah Hazaa","email":"dev.salahhazaa@gmail.com","date":"2025-01-31 13:13:31.000000000","tz":0},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939825\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n5afe03440248483bbdb0fcb1f2fbf114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdc32d917a760e2d0254b4af3ea78b1b09325a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdc32d917a760e2d0254b4af3ea78b1b09325a21"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d6dab1b0cd8d306739d6118a100cbd3503d5f91d":{"kind":"REWORK","_number":12,"created":"2025-02-02 09:06:53.000000000","uploader":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"ref":"refs/changes/75/937275/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/12"}}},"commit":{"parents":[{"commit":"d267ede98f1b0aa25630b1efe07db5a8b478d473","subject":"Merge \"api: Allow min/max_version arguments to expected_errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d267ede98f1b0aa25630b1efe07db5a8b478d473"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Salah Hazaa","email":"dev.salahhazaa@gmail.com","date":"2025-02-02 09:06:53.000000000","tz":0},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939825\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n5afe03440248483bbdb0fcb1f2fbf114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6dab1b0cd8d306739d6118a100cbd3503d5f91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6dab1b0cd8d306739d6118a100cbd3503d5f91d"}]},"branch":"refs/heads/master","description":"Rebase"},"5c8d3efca65880c196511bc6507ecb33673fc523":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2025-02-02 09:07:31.000000000","uploader":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"ref":"refs/changes/75/937275/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/13"}}},"commit":{"parents":[{"commit":"d267ede98f1b0aa25630b1efe07db5a8b478d473","subject":"Merge \"api: Allow min/max_version arguments to expected_errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d267ede98f1b0aa25630b1efe07db5a8b478d473"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Salah Hazaa","email":"dev.salahhazaa@gmail.com","date":"2025-02-02 09:07:31.000000000","tz":0},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939825\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c8d3efca65880c196511bc6507ecb33673fc523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c8d3efca65880c196511bc6507ecb33673fc523"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fa5ec296a4d70b98c0ac8b967adab4e1060c5cd9":{"kind":"REWORK","_number":14,"created":"2025-02-04 18:00:52.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/14"}}},"commit":{"parents":[{"commit":"d267ede98f1b0aa25630b1efe07db5a8b478d473","subject":"Merge \"api: Allow min/max_version arguments to expected_errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d267ede98f1b0aa25630b1efe07db5a8b478d473"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-02-04 18:00:42.000000000","tz":60},"subject":"[CI][nova-next]test with placement ac breadth-first","message":"[CI][nova-next]test with placement ac breadth-first\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939825\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa5ec296a4d70b98c0ac8b967adab4e1060c5cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa5ec296a4d70b98c0ac8b967adab4e1060c5cd9"}]},"branch":"refs/heads/master"},"da1fb487e7c5be0d4906cae22de1cbe9a5819ba0":{"kind":"REWORK","_number":15,"created":"2025-10-13 09:08:36.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/75/937275/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/937275/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/937275/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/937275/15"}}},"commit":{"parents":[{"commit":"d267ede98f1b0aa25630b1efe07db5a8b478d473","subject":"Merge \"api: Allow min/max_version arguments to expected_errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d267ede98f1b0aa25630b1efe07db5a8b478d473"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2024-12-06 17:40:55.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-10-13 09:07:06.000000000","tz":120},"subject":"[CI][nova-next]test with placement ac optimizations","message":"[CI][nova-next]test with placement ac optimizations\n\nThe nova-next job will run tempest with allocation candidate generation\noptimizations enabled:\n\n  [placement]\n  allocation_candidates_generation_strategy \u003d breadth-first\n  max_allocation_candidates \u003d 100000\n  [workarounds]\n  optimize_for_wide_provider_trees \u003d True\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/939825\nDeepnds-On: https://review.opendev.org/c/openstack/placement/+/962776\nChange-Id: Icbfc568bead950a3be22784746421cc11382b2d7\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da1fb487e7c5be0d4906cae22de1cbe9a5819ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da1fb487e7c5be0d4906cae22de1cbe9a5819ba0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
