)]}'
{"id":"openstack%2Fnova~938223","triplet_id":"openstack%2Fnova~master~I1ea5e23038d37e1d37c8a4e764d56d93087a1189","project":"openstack/nova","branch":"master","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-23 07:25:50.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"7556":{"account":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"last_update":"2024-12-24 11:39:48.000000000","reason":"\u003cGERRIT_ACCOUNT_24243\u003e replied on the change","reason_account":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-02-07 07:38:22.000000000","reason":"\u003cGERRIT_ACCOUNT_24243\u003e replied on the change","reason_account":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"}},"26286":{"account":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"last_update":"2025-01-23 07:25:50.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"24243":{"account":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"last_update":"2025-02-13 08:23:03.000000000","reason":"\u003cGERRIT_ACCOUNT_24243\u003e replied on the change","reason_account":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"}}},"hashtags":[],"change_id":"I1ea5e23038d37e1d37c8a4e764d56d93087a1189","subject":"Abort live migration task when stop nova compute service","status":"NEW","created":"2024-12-24 11:38:43.000000000","updated":"2025-03-21 03:15:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ad72c6970019c84aecfd9385a3afa0a4c73eb76","_number":938223,"virtual_id_number":938223,"owner":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-21 03:15:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2025-03-21 00:57:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-24 11:39:48.000000000","updated_by":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-12-24 11:39:48.000000000","updated_by":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"reviewer":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"state":"REVIEWER"},{"updated":"2024-12-24 12:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-24 12:21:56.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-24 12:39:36.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"state":"REVIEWER"},{"updated":"2024-12-24 13:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-01-07 07:13:51.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2025-01-23 09:30:01.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"bab9d2be8722807f8470d1b5adb7accca4521277","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2024-12-24 11:38:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d494362c1ee2f619ed77cfb75755c23820d7398a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 12:03:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957a293e84f14d7ea4e3e22cb98714dd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c3813bd3a79454b9d38f7c1a2f8fd0d : FAILURE in 19m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2932b741be244a7d94c98939660c771c : FAILURE in 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e308e40e981fed1e90378aa31eb512224a26aca6","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-24 12:21:56.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-nq2mf/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"938cb42204ea04d3e415e03a70fbb664adb0c679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 13:49:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5552a920246146e797a9f402c9437bda\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec4c7ee172234ba1afb85dd751def9cc : SUCCESS in 54m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/890a23e476404b3c938e5a50266854eb : SUCCESS in 1h 44m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8338234451a742198ef4bd0a5151cbee : SUCCESS in 42m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c47fabf3cbc048aab8defd794bfa13bb : FAILURE in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4537421c0a594f9fa8bec55433e1d014 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5001c62506ec4ba2849bcd80ee5de2a1 : FAILURE in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd556960790f431f806276d9df8752e6 : FAILURE in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f91b217264b45f6bdb06a830a809843 : SUCCESS in 7m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f49e1aa3dab4427588b0b98e22b142ab : SUCCESS in 37m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8270d4eddbbd4e09ad89f5a4a046e079 : SUCCESS in 1h 11m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/88485dcb31b1426b936192175140a9ef : FAILURE in 1h 18m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/27e6b1b329ee4243a448042372ef044a : SUCCESS in 35m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6a8039e9275d4384900f6e3ea3d08e0b : SUCCESS in 1h 55m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/ce712c698f0a4d18867627d4b1f8e681 : SUCCESS in 1h 24m 03s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/953b86ce62ab460bbe3f7e28f62b7c7a : SUCCESS in 41m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/07b5caa7ea0044c28e5ad93ce906bf38 : SUCCESS in 3m 31s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d80cf9d82c7847d18337782ecb6ee79f : SUCCESS in 30m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ecc1fcf2d154cb28bb52ea2cb389860 : SUCCESS in 32m 59s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d3289a7ad51d4703a3d90769b7d02790 : SUCCESS in 33m 20s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2d847cd26b834aa1ae639ba18a5dd862 : SUCCESS in 1h 46m 39s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/72e2a764d51143ec8fc76f956c33e140 : SUCCESS in 1h 42m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd1329bd74c94c589d0a6157a3c9dd89 : SUCCESS in 1h 24m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8106e4814ca648d2a7065a6af1186731 : SUCCESS in 54m 23s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/76eecf422a534b17b55b8a76623c2904 : SUCCESS in 17m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f02af3f1acc5466c8d6d24dd83ca2a20 : SUCCESS in 26m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d197d0b642a2380dc13344187ed16707af7748be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2024-12-25 02:21:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b50ac686b153588d496fef5ab2b66cd1a403d2d1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-25 03:01:01.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-6hwtp/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"826a8649140042f52f9574960f7c96b516af3616","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 03:05:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db20e2bd325486b989b9fb8490fbab2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e6305da3de442c29e3d11a81f1d2593 : FAILURE in 32m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/268f98d65fea4ddbab5f0b5d67df699b : FAILURE in 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e13302b88bde509c30bd2f20d10239ff924592f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-25 05:17:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4a681c8d7454d1a8a743847ca0f9216\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0befdd08a23543289018fa25e857678a : SUCCESS in 59m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/97df454ae20543a18cca55269c1479f2 : SUCCESS in 1h 28m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8830b16e078c4b46869c58c756e94f34 : SUCCESS in 1h 31m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b1ba0e2c26d452eb40444fac1ba393d : FAILURE in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0403c7d394224bd3ab817d4a8ad9ba97 : SUCCESS in 19m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71e50b93e17e474a927982c6418525b1 : FAILURE in 12m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/734da44d076f495ab03673a8b27bca6d : FAILURE in 18m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c89dc0a61649d9ae94d5e563a3266b : SUCCESS in 11m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/519227eef0dd48c98751df4d2d591c1a : SUCCESS in 57m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed61c7a3364b4a54b86e77542e1420dc : SUCCESS in 1h 49m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aa7f737932624c4d8478c06ee1c7669d : SUCCESS in 42m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7b5f43b85ed74cd3b135010b798c9921 : SUCCESS in 46m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d8f208265f8c449e9eedc69b38c462b9 : SUCCESS in 2h 16m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/627f94ee68e74e53910cf50a54138e94 : SUCCESS in 2h 04m 03s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b9c69f66ba474b88a01092ffbf6aa1e4 : SUCCESS in 46m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/527ca67a31d146d9bf99576426194ffd : SUCCESS in 5m 46s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b39d5dd176b246b695b7ab9298aa5649 : SUCCESS in 31m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2b384307015494588f72c2cd3713ca0 : SUCCESS in 29m 10s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d6f2d34aa2fd4ce29dc9833a0707b9e9 : SUCCESS in 30m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1234a6eed1484706901abec0e0832ec8 : SUCCESS in 2h 08m 38s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/4d2de1054df14c92bd4534b2e507307c : SUCCESS in 1h 53m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/de2cebdff8474e7f8d8ba1933e509d56 : SUCCESS in 1h 31m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbdebe823f8347a3b805d516ca47bca9 : SUCCESS in 1h 11m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1fcade5c9074442a8ffbb77f9bf9cdac : SUCCESS in 27m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9bc5b9452c224ad183e95c8325fc5c33 : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8925cd0f28871e3a8ab586498d565ca83741dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2024-12-27 09:34:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fc7dc8366e318dd79c9b3a1c91b70f7068a03f16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 10:03:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a48e6660017049e3b0b92e5680b84523\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf98ec5236184ffa95f1b7f6bb8ef0a9 : FAILURE in 23m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d6e70826ba5b4068aeecfda2f96e1902 : FAILURE in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"27eece5709f2661df4039f1ad37f1d06452b54c9","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-27 10:17:25.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-rdm6p/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"09b475bd881097dedc8f8127af2e90ca99ca5792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 11:58:13.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f8f8c25db1743fb8530bcfff6158f80\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed1e2e91ff7c4acfb404500d5311ab53 : SUCCESS in 52m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1dc8c11b566c4b8897a6e42f0121a6fa : SUCCESS in 1h 35m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50f3e44d3b9f4eb48404509a2c94996d : SUCCESS in 53m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef62058441e491c99ef9553449fefed : FAILURE in 19m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2b290893dcc48c297b561da13c027bb : FAILURE in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4dc8a8062443e1b44a5605b3866e9d : FAILURE in 10m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b277ec8d2f0d4065b3ae7280b5540984 : FAILURE in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de4c5a00dad465db1bb424fb9f2e2e7 : SUCCESS in 10m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7bea16acc7a542799774e3e33e0d2929 : SUCCESS in 1h 03m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/070e00ef99d24080bba105e8aff8108e : SUCCESS in 1h 22m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1a095f079ade4fadaf0fef1d3d18443a : SUCCESS in 1h 03m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c93809a2c8e94b5fbae2558907a6edc1 : SUCCESS in 55m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dc653ecbb9ef4fa19ac671281ca5ca99 : SUCCESS in 2h 22m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/318c6dda41fa4d649ff0f3a6eab6064d : SUCCESS in 1h 46m 33s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5dd7779b302b4834b779dbc88d350b9d : SUCCESS in 37m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e7a05ae268ed41699cf6af23cdca2de2 : SUCCESS in 6m 06s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bad5d17b94ee436aa94c25233b62e0f5 : SUCCESS in 33m 35s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73a1c81c5363458f8e4d880e417d7458 : SUCCESS in 40m 07s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/12ed3e25bfda4d84a83ace29de706114 : SUCCESS in 39m 19s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8092619f317646f8bc5be817eaf14535 : SUCCESS in 1h 31m 54s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/ca1cbf05e23b4b2ba73e954c5335c2ea : SUCCESS in 1h 21m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c70f47cabe94d7d839531aabb912af8 : SUCCESS in 1h 04m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ceac974fe55f4aa8addb2632edb2d145 : SUCCESS in 51m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a6504c283f2a4445aa9b1bba266f1580 : SUCCESS in 25m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/32b46f32afe3471b8596cf17a69b5ae6 : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"33e145c7c64cf619a2f396ef99fa855b98da9e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-01-06 05:44:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ca36aeff5034e5cc1f0015963249379657da8c81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 06:16:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e9e14eaf5640b2aca080a64d03042c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/376d13d57aa741c9a8a24c63a283068d : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebbce4fa784042cb80e6abfb98fc1145 : SUCCESS in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1f64b35ed3e25bc9ee3e7b52e00880f2a57e6a08","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-06 06:26:02.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-wwlp5/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"8b4923a7fa7b71eb8ee23baca701e4a67f8d70aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 07:44:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7702be3ef5d4faa9da444ec967cd354\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ab49aa7242443ce88431529314c3b43 : SUCCESS in 42m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a7f4b07b42e64940819646b4a79cb835 : SUCCESS in 1h 58m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de723eb1ad644368a55cee5c53c35eda : SUCCESS in 1h 03m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b74c0e8a17f4179a9701f669b78ca86 : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b45905f3ede4a2b8a59a8bf0f6c8cbd : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df7a5194b6b84e81811e85a9fb6ce670 : SUCCESS in 13m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf4a90d147904476a8b4a97d4888e6b9 : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f9e1f6d428542c0b18eec837b7d03be : SUCCESS in 10m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9ab52cc014574611b9c5828cedac5cc9 : SUCCESS in 42m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cef90ec200d144c19a25150b9b2fe113 : SUCCESS in 1h 10m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2fe79e4d467046a2bb4a420ceba13117 : SUCCESS in 37m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/018e987ea601465e8988186d97318bea : SUCCESS in 32m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a920834196104e3285b26ea585953b69 : SUCCESS in 1h 30m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/d8749896493a4b97a6c7c679799eb1b3 : SUCCESS in 1h 25m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/44e72ee3d41841a4967077dc548bcaab : SUCCESS in 38m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f2531e5ffe2a42579293e9aec2158b72 : SUCCESS in 5m 14s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16764ca2fddf42d5ae1508e28ddbb2e3 : SUCCESS in 29m 15s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00a3e8bb27a649cbb2edd7b64781be8c : SUCCESS in 26m 02s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa0394ecb3cf4f268e519778df14bfa3 : SUCCESS in 36m 50s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7c997b79cbd94479b11b3bd6275f5c00 : SUCCESS in 1h 30m 32s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/a658add2f56a4cb3b8f55b9eb5361a03 : SUCCESS in 1h 09m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2ea8d7f6f0445a6baeccc490b5d60c6 : SUCCESS in 1h 01m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b4b212a20124ff98633d642a7dcfd20 : SUCCESS in 36m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/37976b2e069447da805b1241615c0e23 : SUCCESS in 19m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8260e99e8b6a4aec999d03f13b8cce5f : SUCCESS in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"edacd0e62a1b159ec37ff6bf33d8d0fc7e44d7dc","author":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"date":"2025-01-07 05:48:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"044f91e3a122a32a4127762aa9443f9dde40a769","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-01-07 06:00:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"16af0aa15cbd28009d9a4e2585808d32592e82a5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-18 01:42:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4131847631e41a0d17fc5902c16adf756a9b517e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-21 15:09:25.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"19796a54959b2570ad0cbd6dd546a9d0727a0f66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-01-23 07:25:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cb9f5ea5730c18a1c3f2c11f2fd4fdc91a3bc840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 07:58:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d765713c716d45f698284c5b74eed7d9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe6b5aa5d085453aaf28d52e87400dcf : SUCCESS in 27m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/124830eb899949c0aafb8b217a36405b : SUCCESS in 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"80e31fa8cc87943c531f29b76ee35d126652e468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 09:25:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ea793c487ad4b08a8f22da69629bae9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/604fda8cab6b49a091931b9213b729ed : SUCCESS in 36m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c975b0241bfc4079a724563ad12c672d : SUCCESS in 1h 15m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fda684b961e5462485af54b6b3fd22f7 : SUCCESS in 40m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ceed59f45004a1d82baedee05d3826c : SUCCESS in 24m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3e3a4bd5e44416bb749294b1d0f308 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03b4580077084780b9e8f1e0f0a6bfa6 : SUCCESS in 12m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a108c0b7e95a4637bbc2c39d65f1620b : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ed9f02a1a14a849b36ad2ee54ffd9f : SUCCESS in 11m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/951dc94edfb1424b892f60889475ba7e : SUCCESS in 49m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac834dbfd35f412a9af0085e1608daba : SUCCESS in 1h 28m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f3625c0421e346e09d8c521c8b8f0303 : SUCCESS in 40m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ac978e74186349a9bd9c1c41fa400588 : SUCCESS in 43m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7515ca752280454c949b6dd692a08901 : SUCCESS in 1h 37m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/ec11adbed94445f897b2e04a69906dae : SUCCESS in 1h 54m 11s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/7f37e9638fda49088f1db764bcf9fe21 : SUCCESS in 37m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6b4195afaeac4a3e9ebd58bfd8b4e4b5 : SUCCESS in 5m 44s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d442ddd47b664587b49b6cc665f6a89c : SUCCESS in 24m 43s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e67d09daf73a4c38a3a40f9e44a1c8e3 : SUCCESS in 26m 43s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/60018841d441435eb662342ec7a187bb : SUCCESS in 28m 23s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/732f3b1b76734d2a89d2113c2de912fd : SUCCESS in 1h 53m 24s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/08c91468dc154fc2b86214725996dcaa : SUCCESS in 1h 57m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ae060b87bfe4b69866a0ba61145c2d7 : SUCCESS in 1h 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a232c9ea6594649b58ae3ca649186f1 : SUCCESS in 1h 04m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9430b4ed37bc4b5d976d49dbdd56dab8 : SUCCESS in 30m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c14921a243874269b8f8bf95e357fe38 : SUCCESS in 35m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"20048536825b77fcbab4ad298ff278b2bba3a3c8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-23 09:30:01.000000000","message":"Patch Set 5:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"07ef427c0032f2658524ee3d6a08a0ff82c2204f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-02-06 08:25:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a452ee4a6972046c2a6a6fcfa5039db29c7389c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 09:06:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e270899861674bb4a28a103aaab3ff95\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46bb67a3f37a44a0b15e4e85d5ceb27a : SUCCESS in 38m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4edcc577f260453dad201a907572531e : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"659b73c7cd40d26cb27c7f3915129587c8b4e0c3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-06 10:31:12.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-5brcs/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"aee4a8692f025e2400fbedfdbba29039d67fa1df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 10:49:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca10c862285b45fc8f66c9685c9bd339\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c5d8ea86eb3441eeb6b1040308931905 : SUCCESS in 58m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/03178a611a3140cf9b7eae3e625263c7 : SUCCESS in 1h 56m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab61b1610fa24836ac7d6e751f8e32e3 : SUCCESS in 1h 06m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d27367b643da4e539a94a25e0f048f14 : SUCCESS in 24m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c789f49006e42bfa6037f2324d15e58 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5f386f3719e4b6698603abba07bfdfe : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/869e8206862e47f887739ab050fddd5b : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862f85c9a5ea4cad8a09daecac73f8e0 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e30c50eaa21e4dc1999d93fc59d54b31 : SUCCESS in 25m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ca183583e86e46eeb291573e71b0fbc9 : SUCCESS in 47m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0d40a84845744b82992d0e2f119cebed : SUCCESS in 1h 44m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c9dabe797dac44369811d018a8956d7b : SUCCESS in 41m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2a9a032cb17649b38b7781b16f11c779 : SUCCESS in 43m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/74b81494d6d24ff585e842a24502281b : SUCCESS in 1h 49m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/1f16fb73b097413c989e47be026a4bb8 : SUCCESS in 2h 14m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/ccc787461cae4454a38cda855bc011a2 : SUCCESS in 49m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3a477f7ddfe7461bb5564f328d2782d4 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a3a85bc05e348d49e07242d3f33da3b : SUCCESS in 25m 04s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd623856992c48df9f4f4aca67b2fb3b : SUCCESS in 24m 21s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a56e556da83e4615a2f9c350d87befdf : SUCCESS in 30m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/945d2d3f10894a9699e4862ca955949f : SUCCESS in 1h 22m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c90f0f32dc4478097bd3c05b44ef23e : FAILURE in 56m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c7a11955a0df43e2b552fb1da10c375b : SUCCESS in 24m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/879f3549a4fb40eea22d68d58ae62863 : SUCCESS in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e42181583d8776d5f5437f5c39deb5f1285f9f4f","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-02-06 11:29:25.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3f084d1ac262ee32b55bfa9f41bd8f857279e7b8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-02-06 15:25:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eb742a9e1d5704291e152d15923e6a15a8245c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-02-07 07:37:57.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a6649024324e71163bad32904dc04c75800832e6","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-02-07 07:38:22.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"187d2bdee9956b22d683765fa5674a71f1eaf466","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-07 08:01:57.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-qbb57/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"1b1de334bab47e7dd25bfae36957e2269e2ea68a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 08:09:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa14b34871304859aa6eb7db20660f77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15e984257ad8426f9752463308c250cb : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/399d6c3d1df944868db0d046b8a672f6 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"36a49ec98c6fcd5ad860664e833062c4aa01a721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 09:59:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca519081e86f42ffa73465895a26c4c7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ede7823aef6340c4a48d37dd0b11ba35 : SUCCESS in 59m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dfec9159e9774bd69f8567f6e632e279 : SUCCESS in 1h 18m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87d396044c0a4c8bb6e5b632ae93f3c3 : SUCCESS in 1h 11m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d04e86914c94848b5167f737dc3f0e3 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/376462f90bbb4084827a28d54d9fe363 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b57cbdb45fcb42d79fa2a92090805217 : SUCCESS in 13m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/636ae43785bc4e87869a2b9b13be4268 : SUCCESS in 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdcd9caa89bf4e39955d27fe175b7312 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c83d7b0c0e484c11b5de3272c42b233d : SUCCESS in 37m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2254ab6a24b54cd2ad3eb29cd9373301 : SUCCESS in 1h 06m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e183efc75dfc4251955565c9c368d9e8 : SUCCESS in 1h 45m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ed512713b79e4b8f879fd293e8655412 : SUCCESS in 56m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a1f3aeaaeaf54c3dab9ea61812d1aa1e : SUCCESS in 44m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e0174234c48436eb27213ff2247659c : SUCCESS in 1h 36m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/867d6d229f13428b806ec17ec1499787 : SUCCESS in 2h 08m 03s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/01ed370fe6084918be5c126784fcfd79 : SUCCESS in 31m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fee347b638e34557af30fe47b3b8c24c : SUCCESS in 4m 38s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e700241be7864cf1bcd5329841a764af : SUCCESS in 21m 43s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b73c26c32f674fadbf352749c9be4f46 : SUCCESS in 28m 00s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fb419092e0484f0b9c5a41ce3466e1f2 : SUCCESS in 33m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e437befff1a446809254aaeb4874f10c : SUCCESS in 1h 29m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9feda12e748d4a6ab85cf932a19fcd92 : SUCCESS in 59m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/662b4897878841e3b728aadfe8096cb7 : SUCCESS in 31m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f16f0d6c2f16491c922992ca92a11bdc : SUCCESS in 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ff1393d05fb024a5699e45ab8d4fe8fcc1906acc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-07 10:05:10.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"d46be71f655116b3e299eb35b665eda006a5935f","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-02-13 08:23:03.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"84b8f309e22dda85a272d03e0ff16acdc585f09c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"date":"2025-03-21 00:57:48.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8a68a545dabbcece5ed018020f641ba15ed79666","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 01:14:19.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb4944e9720141dbb88b550d518c37d7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/581a41e0154f4cbe9bb3a4b4545eb564 : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7fc947d897d4f5e98a6fb426fb62d02 : SUCCESS in 14m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3b56f753ec33c3b8100af03203618f5a6a05515b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-03-21 01:52:47.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938223-pqp55/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"0ad72c6970019c84aecfd9385a3afa0a4c73eb76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 03:15:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1585cb6907724ac6b9cf038d16ff5443\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56975cd1accf4df4bd55f076ec88c2a0 : SUCCESS in 25m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4f6204409f1642aa9e1f77b15acadf46 : SUCCESS in 1h 04m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/952c595b2c6c4d549ca5073dbabd13ce : SUCCESS in 1h 01m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70346442ffed4ac8ab2263a03467a45d : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b60423a08242f9b7556d88796986bf : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef9a920729444aaa56758f864323a73 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab11611d101845069102d2551fa971c2 : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9e3d3b98c04b199b5fc3dd1e4e264a : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658028ef7d9a47dea1358d38f49f2f85 : SUCCESS in 16m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cb836cecde3f4380b506cf05763c6f49 : SUCCESS in 43m 48s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb3bb5df6a814833854d3e943719734f : SUCCESS in 1h 58m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5e6caccd4ca343b8a4520c0acb4820ec : SUCCESS in 27m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8c681749f79244c286f4cd22e27f9b68 : SUCCESS in 28m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7445242832434ad897209c870fd01d93 : SUCCESS in 2h 03m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/0356c64fd26f4e2c938132d404a31024 : SUCCESS in 2h 04m 28s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/736d1cc1e682454093498319521fa0ba : SUCCESS in 39m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e41752b4ef9a4e8287bc6780f2a38cb9 : SUCCESS in 4m 30s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c246ac622654e27a508fd97a5993ac9 : SUCCESS in 28m 15s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75f2f2ee3b6940daa2b25435f38c8a02 : SUCCESS in 29m 19s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7f213d3d35364e368bfda6ec5e45927d : SUCCESS in 24m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0599e50bbd104d3eaa355fc72ca5e0d7 : SUCCESS in 1h 21m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afb990cf0b0240a39008eeb71b36538c : SUCCESS in 40m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/21fd99de4aa84d8abb7e8a1953fdf8be : SUCCESS in 18m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/71dba15c16ce45508cf2d03ee25fa0ac : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"2aaae543a5793f409bf202d7953f3b4d2e18a8d7","revisions":{"1b8bb84676595baf3dbc7e5356653a7ba41417c3":{"kind":"REWORK","_number":1,"created":"2024-12-24 11:38:43.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/1"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2024-12-24 11:38:27.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2024-12-24 11:38:27.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b8bb84676595baf3dbc7e5356653a7ba41417c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b8bb84676595baf3dbc7e5356653a7ba41417c3"}]},"branch":"refs/heads/master"},"63580700c3e53827c882f43e06a7aa71b98e29bb":{"kind":"REWORK","_number":2,"created":"2024-12-25 02:21:21.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/2"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2024-12-25 02:21:10.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2024-12-25 02:21:10.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63580700c3e53827c882f43e06a7aa71b98e29bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/63580700c3e53827c882f43e06a7aa71b98e29bb"}]},"branch":"refs/heads/master"},"a29a45bbf3f56973f5b2e58a5bd2814b91baa6a5":{"kind":"REWORK","_number":3,"created":"2024-12-27 09:34:36.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/3"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2024-12-27 09:34:21.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2024-12-27 09:34:21.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a29a45bbf3f56973f5b2e58a5bd2814b91baa6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a29a45bbf3f56973f5b2e58a5bd2814b91baa6a5"}]},"branch":"refs/heads/master"},"5dd3f2442819afd4574446482ccb1b8f0c16d876":{"kind":"REWORK","_number":4,"created":"2025-01-06 05:44:25.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/4"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2025-01-06 05:44:12.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2025-01-06 05:44:12.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5dd3f2442819afd4574446482ccb1b8f0c16d876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5dd3f2442819afd4574446482ccb1b8f0c16d876"}]},"branch":"refs/heads/master"},"fd306417f3764d59c8b9e56fe8289f58cbc4491d":{"kind":"REWORK","_number":5,"created":"2025-01-23 07:25:50.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/5"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2025-01-23 07:25:36.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2025-01-23 07:25:36.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd306417f3764d59c8b9e56fe8289f58cbc4491d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd306417f3764d59c8b9e56fe8289f58cbc4491d"}]},"branch":"refs/heads/master"},"13af296fcf219653db805ed18b9fe011930df8ae":{"kind":"REWORK","_number":6,"created":"2025-02-06 08:25:34.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/6"}}},"commit":{"parents":[{"commit":"e27bbe72e0d293e55c30d4f90ca0afcf47427419","subject":"Merge \"Cleanup RBAC jobs in check/gate pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e27bbe72e0d293e55c30d4f90ca0afcf47427419"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2025-02-06 08:24:31.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2025-02-06 08:25:25.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13af296fcf219653db805ed18b9fe011930df8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13af296fcf219653db805ed18b9fe011930df8ae"}]},"branch":"refs/heads/master"},"9fc5d07ea36b40c280d77752015d5421646fe25c":{"kind":"REWORK","_number":7,"created":"2025-02-07 07:37:57.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/7"}}},"commit":{"parents":[{"commit":"e27bbe72e0d293e55c30d4f90ca0afcf47427419","subject":"Merge \"Cleanup RBAC jobs in check/gate pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e27bbe72e0d293e55c30d4f90ca0afcf47427419"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2025-02-07 07:37:45.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2025-02-07 07:37:45.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9fc5d07ea36b40c280d77752015d5421646fe25c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9fc5d07ea36b40c280d77752015d5421646fe25c"}]},"branch":"refs/heads/master"},"2aaae543a5793f409bf202d7953f3b4d2e18a8d7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-03-21 00:57:48.000000000","uploader":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"},"ref":"refs/changes/23/938223/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/938223/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/938223/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/938223/8"}}},"commit":{"parents":[{"commit":"76c3c4c1bd03298794e1e3fb471c4506c9a38ee0","subject":"Merge \"Ignore metadata tags in pci/stats _find_pool logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/76c3c4c1bd03298794e1e3fb471c4506c9a38ee0"}]}],"author":{"name":"benlei","email":"15201676659@163.com","date":"2025-02-07 07:37:45.000000000","tz":480},"committer":{"name":"benlei","email":"15201676659@163.com","date":"2025-03-20 03:27:00.000000000","tz":480},"subject":"Abort live migration task when stop nova compute service","message":"Abort live migration task when stop nova compute service\n\nAdd an abort live migration function in cleanup_host method\nwhen gracefully stop nova compute service, this will avoid\nduplication instances that when libvirt is execute instance\nmigration but nova compute service restarted and the same\ntime libvirt completed the migration, because the _init_instance\nwill resume instance that has migrating status.\n\nRelated-Bug: #2092391\nChange-Id: I1ea5e23038d37e1d37c8a4e764d56d93087a1189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2aaae543a5793f409bf202d7953f3b4d2e18a8d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2aaae543a5793f409bf202d7953f3b4d2e18a8d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":24243,"name":"benlei","email":"15201676659@163.com","username":"xubenlei","status":"easystack employee"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
