)]}'
{"id":"openstack%2Fnova~938311","triplet_id":"openstack%2Fnova~master~I21b59a886e66069f3abd9e059c14115dcd7b2a62","project":"openstack/nova","branch":"master","topic":"bug/2092529","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-21 02:28:21.000000000","reason":"\u003cGERRIT_ACCOUNT_37469\u003e replied on the change","reason_account":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"}},"37469":{"account":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"last_update":"2026-03-27 06:15:27.000000000","reason":"\u003cGERRIT_ACCOUNT_36725\u003e replied on the change","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I21b59a886e66069f3abd9e059c14115dcd7b2a62","subject":"Fix volume size mismatch in PowerFlex during concurrent extensions","status":"NEW","created":"2024-12-30 01:16:24.000000000","updated":"2026-04-06 08:02:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"35f9b3364d631d9f64a10fe7d2c823298aa339de","_number":938311,"virtual_id_number":938311,"owner":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-13 10:28:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2025-01-18 01:27:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2026-03-27 06:15:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-30 01:50:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-30 01:56:02.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-30 03:58:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 13:57:06.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"},{"updated":"2025-01-18 01:27:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-01-18 01:27:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-03-27 06:15:27.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"state":"REVIEWER"},{"updated":"2026-04-06 08:02:41.000000000","updated_by":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"reviewer":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"state":"CC"}],"messages":[{"id":"b1b851d8456ba8d44fa6c198283d7e997d5ab40d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2024-12-30 01:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a0935ce9892a275ca651bca017281a55804acf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2024-12-30 01:19:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2410e938123b36668d572acadf8d889e500c527","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 01:50:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74bf7e74fbff438a8c963850abe9f2cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79b2836e4a1c427c8835f94a32e0c381 : SUCCESS in 29m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a482c75771e94161a01d92a98b3e9ee7 : SUCCESS in 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf9ab0b01ac351e7386288f1935daa9ceb6d7f31","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-30 01:56:02.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938311-4xqxf/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"02edb5bc9bf3bd4da30faffbd5630668cc49bf1d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-30 02:35:40.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938311-lrv5x/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"ed170ba7d5cfd3aa2fba622e3df40ee7d8cc2b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 03:58:21.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53d991fa8aa244a0bc30bccf6c6b7885\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d59d3d83b8814abbb9d5507f6d346117 : SUCCESS in 38m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d99235643fd548b2871b7610cdce151e : SUCCESS in 1h 40m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae501f8eea0340fea768cb3f3ab62a1e : SUCCESS in 1h 02m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/219d0d27add84ab58c9c4c0cf803dec6 : SUCCESS in 23m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21751d3ff06a435fb36f8a00a5052a4e : FAILURE in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/534d34c53a9e4b87ac3f0cfe893dd40f : SUCCESS in 13m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/499d4d137195424b8607fb09e2c61db6 : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b33698e2be04c059f7a5f6b3c224498 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/801fa29bb8bb4b4baaf4d6e985915dbb : SUCCESS in 33m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8542e8330862426ca70da536ad5f5ec1 : SUCCESS in 1h 09m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/955216ad4bca45e3b27874961c2b0854 : FAILURE in 10m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5cb09b057a0a4c48b7b5dd012979ebc1 : SUCCESS in 59m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dbd95c407dcd42078bae1ed276b2ac36 : POST_FAILURE in 13m 23s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5f78ee8398c741dbaed149142870d070 : SUCCESS in 45m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/58135bec268844cd87b6cc37643478bf : SUCCESS in 2h 10m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/a5c5636a388740b783e23d1a3217a890 : SUCCESS in 2h 07m 59s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/69fe7b1d1b664adf9c42f43ff62645d5 : SUCCESS in 29m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dfb24d19f2e84544b9ab46882d1eb988 : SUCCESS in 5m 20s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c8db5af892a49c19a2719465c71e51e : SUCCESS in 25m 35s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90bee33810c8401fa2d0948059c1415c : SUCCESS in 23m 44s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/042e73e464fa42218a46bda94f7b80da : SUCCESS in 29m 06s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/52336543902746aa8f05778ec0959e91 : SUCCESS in 1h 43m 37s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/2dd566abee3b482181a3ae18299a63eb : SUCCESS in 1h 50m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa11d2a8e3624a2d96dd3353e93cdea1 : SUCCESS in 1h 16m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11bfad4eecb04b728607c440eb196162 : SUCCESS in 54m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d0364c69873c4d5cb3de3a8ce622f85a : SUCCESS in 26m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fd936a0ca86148a3892c66f25d6e5959 : SUCCESS in 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"576a6fcad9bc99172c0e4b7230ad05af6867b630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2024-12-30 04:32:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"248d60698f8e1eefb7105006287b170a4c59695a","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2024-12-30 04:34:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b6068e8c8b54158c463d0f13a98c4c7ba35d02c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2024-12-30 04:43:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"151657a930603bfa22710daff1f818e55da26ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 05:18:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9260dc2d6bc54eb7ad4ad46c7ef3bb02\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01436a6dbb644ed085ca72602fbf6f55 : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bc219303c3094d3daebeed90c453e276 : SUCCESS in 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9caccbfbff83ade4453b6dd9db0bd87ab85dde2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-30 06:55:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9a480070ea442d183e9f3963da2ff58\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/07e67cfcf4414852a1555ab9ec691d0e : SUCCESS in 56m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dca36f403f6645febd13fff9b115d272 : SUCCESS in 1h 37m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7013a33796df4c7197e6532caf15ceda : SUCCESS in 1h 03m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a63d32b22d23458dba8389d53af44fcd : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f41f5964c19740e08819a2fc92ef2076 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66b3bb445b6947739b12d1da6336bfbf : SUCCESS in 13m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35538d9c89a640dfab55dc2cb2a1e840 : SUCCESS in 13m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f950edee484066b5711fcf70ef3f7e : SUCCESS in 13m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf21d4fbbe774acc9a8f1dd683790b14 : SUCCESS in 26m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e6500c1a4e544095a12bb74494d649e9 : SUCCESS in 53m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/afca3926df3e421e98ecca6e43c2c716 : FAILURE in 11m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4b285e26907e4831896174fbd945bf08 : SUCCESS in 49m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/82e7c05c7539405cab9d984e02fb5090 : POST_FAILURE in 13m 02s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/13c502bce57f4ad09d61f22b3432ec52 : SUCCESS in 56m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/95720410d06f44a7a3ad32c0c9e9d470 : SUCCESS in 1h 29m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/aedb254d6afc47b9a9c154dab841dbef : FAILURE in 2h 08m 26s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/3f8233a177984d45a5f2e22316a86883 : SUCCESS in 33m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/eb245d4183b144bbb0f175379065ed64 : SUCCESS in 4m 45s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a35ad5aa9f394654a03660823744b964 : SUCCESS in 25m 43s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1657eea892b442148bd6c4d25b1dd8b6 : SUCCESS in 20m 10s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9baccc4dfa314df2974a47c13c46b66a : SUCCESS in 32m 48s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f667c80d204455ebf20d04e4355a731 : SUCCESS in 1h 57m 20s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/0a01d98c41024250884e0c0c6f3b44d3 : SUCCESS in 1h 46m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7280b3d65a9a41b995921c5260780e8b : SUCCESS in 1h 40m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9afbfd6f5efc40f8afd10f7560a34cce : SUCCESS in 1h 00m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1cf611edff1a4652a0bde8908378492b : SUCCESS in 31m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a3ad7fcbe4e2480a8c50b2aef364e902 : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1e83c99c46958a07c1a3a9071d1c172b116dbc2f","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2024-12-31 01:15:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"08f4a05e24427366dafcb4113d48e7311f9e9e89","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2024-12-31 01:57:08.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938311-m5997/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"1210e44482b7a57db0262db333837e87371ee37a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 02:06:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/958030121a7c457f843c9c9bc7cb5e77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/605ac5ab70074b4787546ecbe81db1d8 : FAILURE in 34m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/931ad4135d0348b7a9a5d6c0395eb24c : SUCCESS in 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c4dd0a9bda99e388313b2253b4a38cc58a24bba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 03:37:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04b248bf993c4864a4e4ee09d76d1199\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18520725283f4876ae2f13b2e6098c34 : SUCCESS in 31m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/88f24cbd9ab84c14899e9eca9fe0b6db : SUCCESS in 1h 55m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a35349119ec24a82a31fb9ced14129da : SUCCESS in 37m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73540e9b3b5c4368841adaf169f4a467 : SUCCESS in 28m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42b203e0fed4ae18f2792feb9a109ae : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a86d588853443508218ddc3af472c1d : SUCCESS in 14m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6d49a2a9efa41b5a0f32157f392b130 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c672a8b07449b08ef1cbeaa0b1d743 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d4372cd27a4289a97a589042e7df78 : SUCCESS in 21m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4eb6102682f49dea0b58017ad3ffd63 : SUCCESS in 41m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6f0adbbf853445679c7ae57202f20ecf : SUCCESS in 1h 57m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bfb047903b774af88bb2f59edbcc2e35 : SUCCESS in 54m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6d2786ef64db4b9bbd6c8030c888d861 : SUCCESS in 58m 55s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2ad871754831454e839f9a6fbd5759d2 : SUCCESS in 1h 18m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/614d484e38f24e58bfe6059f8c3fe07f : SUCCESS in 2h 03m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/f57782c2104a4668adb195e11bca277f : SUCCESS in 2h 15m 02s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/497d8a64a5ed414a96a5293151d556cf : SUCCESS in 47m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f86024509e3646488b90d9e4283ffe95 : SUCCESS in 6m 01s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3cd69837a7bf4198aa866ad857eadc27 : SUCCESS in 24m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/258a1430cefb4b749a89171f361b62b6 : SUCCESS in 29m 52s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c32a27742d0346609e8a486b174b0aa0 : SUCCESS in 32m 06s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3ad6a124c5b249f1b8bedb5b48b2117e : SUCCESS in 1h 40m 49s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/033cfc806bbb44e5beca85453f0dc7ee : SUCCESS in 1h 38m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c899f71b9c6f46d0aac329d22436fcc6 : SUCCESS in 1h 19m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf02c687127744589585a3ec407726c4 : SUCCESS in 33m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f2abed1fd07e4aa39aeafe3fdaf4fd3b : SUCCESS in 37m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cf80da0ce72647f9ab1c99b0f69d5142 : SUCCESS in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"25bf4d1fa459d227f727421307bda042275ab00a","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-10 13:57:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5fad7c819698d5f6b7de3116e7d18e4ed0e4b834","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2025-01-10 13:57:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e35fafc67dcbd40f7245833b579760493c483a33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:51:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf64a4d9a194561b05fdda0b9ab77b1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe15af5189fb4bc59f5c0c6fd9029ec1 : SUCCESS in 34m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a0c3fd6eb0646c691e1275c72a18371 : FAILURE in 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"581fc8fd6eed9b1969495436af7617949aa43f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:00:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1806e1c9625747f3b1e41b468d892187\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87771641d9034e1696a69c3b488fb6df : SUCCESS in 46m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6d6f860c35244e238804d57673bf89c8 : SUCCESS in 1h 47m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc2b92b16d27448c89b5264ed5f72930 : SUCCESS in 40m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04444b966a104ad9b4376ee7bde36248 : SUCCESS in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eee4396bc224d238bc45fd32399f1d3 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f654bbb0b614b3fba923b785040a250 : SUCCESS in 14m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc740704defd49359b31ef6f5747d05d : SUCCESS in 13m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2452a2cf5314b18944598e9e49358e2 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8cfe8997eb0435a8ac4040e96a69c18 : SUCCESS in 34m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/935813d861014907be64f0b9989449f9 : SUCCESS in 1h 08m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b7892ac947a472d86f551af856f5b7e : SUCCESS in 1h 59m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/282c846b8af34f54a86c47f7c41c6cde : SUCCESS in 56m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/32c1f9bc63af4da9a0a524058e16dfcf : SUCCESS in 1h 04m 56s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c48750a4a87040afbd812267d283155c : SUCCESS in 1h 00m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e94b130bb84940cf8cf23901f4ac255b : SUCCESS in 1h 59m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/be5368ba812f4886823075ef179301bd : SUCCESS in 1h 55m 02s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b963961c0154405bb7b40c9e5e28ead5 : SUCCESS in 31m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/abdab943be8e492ab8c48fa5a017ddc6 : SUCCESS in 6m 40s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7fd968e182fe4c2799586802aa87f7a2 : SUCCESS in 25m 27s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57924217ab064a27b76910b00269a962 : SUCCESS in 24m 35s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/aba294fa9a4845e9935c4356664b6867 : SUCCESS in 32m 56s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d387283727fa440b9945111468bcf31f : FAILURE in 1h 28m 12s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c5915bf9b5a04ed28324fd5fe9e0ec33 : SUCCESS in 1h 56m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cedc133134ca4a648b65b1fd61ef8af9 : SUCCESS in 1h 24m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d849f3eec6bd457bb96387ff2114607f : SUCCESS in 1h 04m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dcac4ccbaef74672ab84f5bca947de4e : SUCCESS in 21m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f57509803acf457281644b335b56f20b : SUCCESS in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"521f612827e800f12a2c5b01e50d8ce58fbea226","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-10 16:14:52.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938311-m69gz/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"ec0be079d70b0fab16897de475bcedc614dd4c8a","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2025-01-13 08:06:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da3f46d82cc1644880b96f9b4537670e0e18b58d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 08:38:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee2ed1499624947a92307e3effa9e3d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fe232246fc84fc99af330ef00fefc10 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a6ce408cb454a96bc2d8bc21511d608 : SUCCESS in 30m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2bcd3485e92db4c6d7f7b9e5afe8b72e22bfe402","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-13 09:43:23.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-938311-cd22r/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"dade2b75a695de36f51077f0a4d4d3091c59e64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 10:28:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03b8aebe9f9418e80e9c98f7ab326c5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/180802782ac945be82d632da51b23d87 : SUCCESS in 1h 04m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/28949762554944b497270afab89331d2 : SUCCESS in 2h 03m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e630285e6e94156b6539df0825075f1 : SUCCESS in 1h 05m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f05bf328014b48d5b85918056ff43d4a : SUCCESS in 29m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4f1e20ba844ced9c1d56a55b6340d4 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2c4dab051314adea21f85abfc8c6c89 : SUCCESS in 12m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e78ddb1f8b040aca5afcc74cd98a405 : SUCCESS in 12m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d092963ee3bb4b23b309316e1f9d4275 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b95b634e5104d5699313154aca75ef4 : SUCCESS in 30m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/94f3301bada046cdb03871e56def9f67 : SUCCESS in 49m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/866944f9affe4935872b4e9bf9aa1d7b : SUCCESS in 1h 50m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/20f6fd3afe6b48bdb48cb2aac9f3b43d : SUCCESS in 42m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/09272abc7a694173855d086f9697004f : SUCCESS in 56m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/45c3517055d24c61a6524314604d2cab : SUCCESS in 1h 18m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/47d0acf7261c4cc1a41c526ba22e292b : SUCCESS in 1h 23m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/44f658ce4c6d444b9877b9a80cc2885c : SUCCESS in 2h 15m 38s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2f2f687284be425e9712945fc308db14 : SUCCESS in 52m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/41bc43bf10164a7f906df92d93075bda : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e40192a22d24e458a506308f1c3bb56 : SUCCESS in 28m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11cdf6b12b794895b76820414b49029a : SUCCESS in 25m 44s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/564e7ab6ddd649d4b4ffa6f043678f9f : SUCCESS in 31m 25s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e936bb25aa414af3a3514f343f449dae : SUCCESS in 1h 42m 31s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/634eab43ec2a4cc9a546e746cc3cb9a6 : SUCCESS in 2h 05m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4383e93417474d45857bf80ef856c3e2 : SUCCESS in 1h 28m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05567acb1f8947868e1a967168edc54a : SUCCESS in 57m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1f68bc6bc7fb488e823226c64efce389 : SUCCESS in 32m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/42c0eabea1f648a88c2a256e29c53efb : SUCCESS in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"130d1bdeb5a80a0775b2c6b7580cc66a9e9b6920","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-18 01:27:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5480b7de2eb66259fed433bf67f520c63b488603","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-18 01:27:23.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"0a13f811757016f01893ca2c308a83376b48132e","author":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"date":"2025-01-21 02:28:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1188a83d2873e7a5ecb174f29f8a7c4158a5464c","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-27 06:15:27.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"35f9b3364d631d9f64a10fe7d2c823298aa339de","author":{"_account_id":37437,"name":"Amit Zauber","display_name":"Zaubea","email":"amit.zauber@dell.com","username":"amit.zauber@dell.com"},"date":"2026-04-06 08:02:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7d09fe4c5adeab3a188b598e40a553c1db277a1e","revisions":{"9b6985048967e0b82c0f08a1d5e49fd67f8867cb":{"kind":"REWORK","_number":1,"created":"2024-12-30 01:16:24.000000000","uploader":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"ref":"refs/changes/11/938311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/938311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/938311/1"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 01:12:06.000000000","tz":540},"committer":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 01:12:44.000000000","tz":540},"subject":"Fix volume size mismatch in PowerFlex during concurrent extensions","message":"Fix volume size mismatch in PowerFlex during concurrent extensions\n\nAddressed an issue in the ScaleIO (PowerFlex) driver where concurrently extending\nmultiple volumes attached to separate running VMs resulted in a mismatch between\nthe physical and capacity sizes of the volumes. This discrepancy required a VM\nreboot to resolve, which is not acceptable for high-availability configurations.\n\nChanges made:\n- Implemented retry logic in the `extend_volume` method to handle size mismatches\n  without necessitating a VM reboot.\n- Added logging to track the extension process and any discrepancies encountered.\n\nThis fix ensures that volume extensions are handled gracefully, maintaining\nconsistency between physical and capacity sizes without requiring VM reboots.\n\nCloses-Bug: #2092529\nChange-Id: I21b59a886e66069f3abd9e059c14115dcd7b2a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b6985048967e0b82c0f08a1d5e49fd67f8867cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b6985048967e0b82c0f08a1d5e49fd67f8867cb"}]},"branch":"refs/heads/master"},"1a36de5672ce54f426ce23d4e39f39165ef89153":{"kind":"REWORK","_number":2,"created":"2024-12-30 01:19:05.000000000","uploader":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"ref":"refs/changes/11/938311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/938311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/938311/2"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 01:12:06.000000000","tz":540},"committer":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 01:15:05.000000000","tz":540},"subject":"Fix volume size mismatch in PowerFlex during concurrent extensions","message":"Fix volume size mismatch in PowerFlex during concurrent extensions\n\nAddressed an issue in the ScaleIO (PowerFlex) driver where concurrently extending\nmultiple volumes attached to separate running VMs resulted in a mismatch between\nthe physical and capacity sizes of the volumes. This discrepancy required a VM\nreboot to resolve, which is not acceptable for high-availability configurations.\n\nChanges made:\n- Implemented retry logic in the `extend_volume` method to handle size mismatches\n  without necessitating a VM reboot.\n- Added logging to track the extension process and any discrepancies encountered.\n\nThis fix ensures that volume extensions are handled gracefully, maintaining\nconsistency between physical and capacity sizes without requiring VM reboots.\n\nCloses-Bug: #2092529\nChange-Id: I21b59a886e66069f3abd9e059c14115dcd7b2a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a36de5672ce54f426ce23d4e39f39165ef89153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a36de5672ce54f426ce23d4e39f39165ef89153"}]},"branch":"refs/heads/master"},"4b46142372856a7eeaa16ebe4d2690e3d4ccee86":{"kind":"REWORK","_number":3,"created":"2024-12-30 04:32:01.000000000","uploader":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"ref":"refs/changes/11/938311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/938311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/938311/3"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 01:12:06.000000000","tz":540},"committer":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 04:28:11.000000000","tz":540},"subject":"Fix volume size mismatch in PowerFlex during concurrent extensions","message":"Fix volume size mismatch in PowerFlex during concurrent extensions\n\nAddressed an issue in the ScaleIO (PowerFlex) driver where concurrently extending\nmultiple volumes attached to separate running VMs resulted in a mismatch between\nthe physical and capacity sizes of the volumes. This discrepancy required a VM\nreboot to resolve, which is not acceptable for high-availability configurations.\n\nChanges made:\n- Implemented retry logic in the `extend_volume` method to handle size mismatches\n  without necessitating a VM reboot.\n- Added logging to track the extension process and any discrepancies encountered.\n\nThis fix ensures that volume extensions are handled gracefully, maintaining\nconsistency between physical and capacity sizes without requiring VM reboots.\n\nCloses-Bug: #2092529\nChange-Id: I21b59a886e66069f3abd9e059c14115dcd7b2a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b46142372856a7eeaa16ebe4d2690e3d4ccee86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b46142372856a7eeaa16ebe4d2690e3d4ccee86"}]},"branch":"refs/heads/master"},"7d09fe4c5adeab3a188b598e40a553c1db277a1e":{"kind":"REWORK","_number":4,"created":"2024-12-30 04:43:35.000000000","uploader":{"_account_id":37469,"name":"DongHun, Kim","display_name":"DongHun, Kim","email":"bigheadzacky@gmail.com","username":"donghunkimzack"},"ref":"refs/changes/11/938311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/938311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/938311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/938311/4"}}},"commit":{"parents":[{"commit":"3c038645dbd2bac9b53a2e15c3fa938dd0b204b9","subject":"Merge \"Refactor test_server_shares: Mock in Base Class and trait verification\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c038645dbd2bac9b53a2e15c3fa938dd0b204b9"}]}],"author":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 01:12:06.000000000","tz":540},"committer":{"name":"donghunkimzack","email":"bigheadzacky@gmail.com","date":"2024-12-30 04:39:44.000000000","tz":540},"subject":"Fix volume size mismatch in PowerFlex during concurrent extensions","message":"Fix volume size mismatch in PowerFlex during concurrent extensions\n\nAddressed an issue in the ScaleIO (PowerFlex) driver where concurrently extending\nmultiple volumes attached to separate running VMs resulted in a mismatch between\nthe physical and capacity sizes of the volumes. This discrepancy required a VM\nreboot to resolve, which is not acceptable for high-availability configurations.\n\nChanges made:\n- Implemented retry logic in the `extend_volume` method to handle size mismatches\n  without necessitating a VM reboot.\n- Added logging to track the extension process and any discrepancies encountered.\n\nThis fix ensures that volume extensions are handled gracefully, maintaining\nconsistency between physical and capacity sizes without requiring VM reboots.\n\nCloses-Bug: #2092529\nChange-Id: I21b59a886e66069f3abd9e059c14115dcd7b2a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d09fe4c5adeab3a188b598e40a553c1db277a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d09fe4c5adeab3a188b598e40a553c1db277a1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
