)]}'
{"id":"openstack%2Fnova~939163","triplet_id":"openstack%2Fnova~stable%2F2023.2~I4e423bac26990fc3538e840990294b178c30e374","project":"openstack/nova","branch":"stable/2023.2","topic":"proxy-libvirt-calls-stable/2024.2-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-01-14 20:14:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4e423bac26990fc3538e840990294b178c30e374","subject":"libvirt: Wrap un-proxied listDevices() and listAllDevices()","status":"ABANDONED","created":"2025-01-13 20:34:23.000000000","updated":"2025-01-15 07:12:02.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":939162,"cherry_pick_of_patch_set":1,"meta_rev_id":"5132626c53c87c32994cc1448e6d7376951c4b89","_number":939163,"virtual_id_number":939163,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-13 22:31:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-01-14 15:00:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 22:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 15:00:20.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-01-15 07:12:02.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"76defe29fc4b3e8a679eef85950da7280f1c2429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-13 20:34:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"31b80e9180d7fe33975820fed8d9a7dd46fe55d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 22:31:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3db0c6e8aab4607910a87de896f9761\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a31b62d7bcd457fa0070dd641934c91 : SUCCESS in 1h 08m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9bb937c5ff8c4c4cbedf7f382341b097 : SUCCESS in 1h 41m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78e8d62fc2e74ea686c44f7da89b4eae : SUCCESS in 53m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62902ab2a1794ee18dacbf7262bb1266 : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151492334686452fb9ff9eb44bfd717c : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ae654d527c4b23ad56c64a75e8add2 : SUCCESS in 14m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6c13a1042fd4a97ba610fcde0db54af : SUCCESS in 14m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b43aa5664d404bd49e574ba730b61aab : SUCCESS in 14m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3fb33ee857a4f2387aebf5071030b6f : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd2d73b142c4c7a811a3e1015b60d72 : SUCCESS in 11m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5355e080edea4d03ac937aee5315160d : SUCCESS in 55m 24s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3a209d3cf9242bab9d4965cd1ec4e17 : SUCCESS in 1h 55m 05s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/694921225f7c4d88aee2bc1c87994957 : SUCCESS in 1h 01m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8b0adf91358047f482d0c1213e580669 : FAILURE in 5m 43s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eadf2d16016b40179f7c90f1f3b75666 : SUCCESS in 20m 31s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc1fe725a27a4a5d9803a9c6b4b4d928 : SUCCESS in 24m 52s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e61ab352947f40f1b321d771b4f0758f : SUCCESS in 24m 44s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41b75f1effe5441290ee96f516f1444c : SUCCESS in 1h 32m 22s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/b880f283a6df43a9b85df345ab3e2469 : SUCCESS in 1h 50m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0a014b4ede640e89b403ac0082856c5 : SUCCESS in 55m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f17b297cdf3a4df6a362747e10ce2012 : SUCCESS in 58m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f4db416f6ba840a3a90ed4da18c4d049 : SUCCESS in 29m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b51a6c9a79584bdf9f6bfbf27009f7b3 : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a387d2a8aadd2b4322a39f4005379ff5efe78cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 22:42:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef72406758f4e16865e63f6ddfcad89\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99bbef75a47b4377a03502d5364055ba : FAILURE in 33m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b04505c64324892aca81a05f8942685 : FAILURE in 35m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/79efbc7b94dc4b8b867d49bbad313a27 : FAILURE in 47m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3af7575dff91433d8b9dfa00389a790e : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a381296a92a504993f7facbcc3f0b681016d4722","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-14 15:00:20.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3030f3e35abbc45e22c49100ff7ae5595836d47a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-14 20:14:55.000000000","message":"Abandoned\n\nThis needs to be redone on master","accounts_in_message":[],"_revision_number":1},{"id":"5132626c53c87c32994cc1448e6d7376951c4b89","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-15 07:12:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fae5ab2b808c840657cea18f12d003e78320d2c","revisions":{"1fae5ab2b808c840657cea18f12d003e78320d2c":{"kind":"REWORK","_number":1,"created":"2025-01-13 20:34:23.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/63/939163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/939163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/939163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/939163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/939163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/939163/1"}}},"commit":{"parents":[{"commit":"ed7371f4f9f1fd69ff6d1594f65c21a89cbfdd41","subject":"[doc]Fix the device_spec config doc about placement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed7371f4f9f1fd69ff6d1594f65c21a89cbfdd41"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-18 02:54:02.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-01-13 20:34:23.000000000","tz":0},"subject":"libvirt: Wrap un-proxied listDevices() and listAllDevices()","message":"libvirt: Wrap un-proxied listDevices() and listAllDevices()\n\nThis is similar to change I668643c836d46a25df46d4c99a973af5e50a39db\nwhere the objects returned in a list from a libvirt call were not\ntpool.Proxy wrapped. Because the objects are not wrapped, calling\nmethods on them such as listCaps() can block all other greenthreads\nand can cause nova-compute to freeze for hours in certain scenarios.\n\nThis adds the same wrapping to libvirt calls which return lists of\nvirNodeDevice.\n\nCloses-Bug: #2091033\n\nChange-Id: I4e423bac26990fc3538e840990294b178c30e374\n(cherry picked from commit 0fccb365ddea10b9d6d082c3d95dba24b7fec435)\n(cherry picked from commit bdb66c067546b625217466d19647c1b0e8bf91e6)\n(cherry picked from commit 5ce0f731d2c6dc83e71fc0068ad637dadf914ce2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fae5ab2b808c840657cea18f12d003e78320d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fae5ab2b808c840657cea18f12d003e78320d2c"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
