)]}'
{"id":"openstack%2Fnova~939453","triplet_id":"openstack%2Fnova~master~Ic90949cae840290dc376af7a7a3056babe2539f2","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-01-16 19:22:47.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-01-17 15:10:03.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-01-22 18:30:38.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-22 18:30:38.000000000","reason":"Change was submitted"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2025-01-22 18:30:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic90949cae840290dc376af7a7a3056babe2539f2","subject":"Run nova-next without periodic cache healing","status":"MERGED","created":"2025-01-16 14:19:48.000000000","updated":"2025-01-22 18:33:34.000000000","submitted":"2025-01-22 18:30:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939453","meta_rev_id":"d23d5fe576569fce644e3eff2ae676c0288d6104","_number":939453,"virtual_id_number":939453,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-22 18:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2025-01-16 19:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-01-17 15:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2025-01-17 15:10:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-16 14:20:35.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-01-16 14:20:35.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-16 14:20:35.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-16 14:21:28.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-01-16 15:08:14.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-16 16:43:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-16 16:49:08.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"b6fdee843ab33099eafdfadd97fe24a9c5f72c2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2025-01-16 14:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b32d588c4e56bef91678cdaf6bca18225bb89200","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-01-16 15:08:14.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-939453-c6f74/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"ef17692646da495296210905e59fb2d70a681474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 16:43:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/520c59f95ab14495a5391df99ad5bd55\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1fc01b2a52a1445cad1d609318ec06c0 : SUCCESS in 56m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d9a0080761c543249ce44bd7e632d9ed : SUCCESS in 1h 51m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8efecd82c0c34804886076cf3f623657 : SUCCESS in 1h 00m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7907fb125f5b4bb9907439c8964047aa : SUCCESS in 25m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc42a26210549968abdfcefe9b982b3 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f52ed9d767f4d97bcb73b652ae650de : SUCCESS in 13m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf5f7643a6684102829b0f14d9d91c12 : SUCCESS in 13m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4e4dc3048848dfb595d0c35f267796 : SUCCESS in 10m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/06bdcbc024fb498a8c2834c46a5217c0 : SUCCESS in 53m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c72fbaff8414fe09d1eee0982b7c997 : SUCCESS in 1h 42m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/71124a512ce14fa58d0de5e0ef55b711 : SUCCESS in 59m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d87c84ea254a430ca9719659cf4b2ff6 : SUCCESS in 44m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/ed6c7eb10e8243adb2eeea4b49bd19bc : SUCCESS in 1h 00m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a8c8a7c097124bfbbcd2dd65e3a1680f : SUCCESS in 1h 49m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/44ef0e3c3cb3424daad68e370cc0fad3 : SUCCESS in 2h 02m 53s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/7f07a9fae0c649ddbfb932087ca86fe7 : SUCCESS in 49m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f988d9587686405281b08944ce234904 : SUCCESS in 5m 26s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff3f4f8985754dd5bc52acaca2064f9b : SUCCESS in 24m 54s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a4a1091c72947d5804021890c9a4edf : SUCCESS in 26m 11s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9d2ee2969a5c49b0b2d50cf1c7a19636 : SUCCESS in 33m 04s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87c84892fc2b4c1ab33cceb9f4ee9727 : SUCCESS in 2h 12m 00s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c77454062f2d43d6ad7475626552cf02 : SUCCESS in 2h 07m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8222b51ab65d41398ead121f37245aad : SUCCESS in 1h 36m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff88de27022641778315718bf9cf9386 : SUCCESS in 54m 38s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f272b060890b4746bf784c86f54af405 : SUCCESS in 28m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c49e018014574186b73096a069d69536 : SUCCESS in 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1178938d6d999beaebbca2eb3af7277d19357d7f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-16 16:49:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"296253a867f45862e59f6778a264eebbad840ebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 18:04:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ded22d058b4a471497ddd96241570c91\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c376eb9af6bb4f03942c37bb9b132184 : FAILURE in 55m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba28aed698994c0294f5274b90f65c6f : FAILURE in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27c2b267d3fef11afe4061545594dc8fb8be894c","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-16 19:22:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55c62f171fb6aa5948d4d4447c6e971f30b4cd7d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-17 15:10:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68b686ef74b60be3f29ac125d4036a0563579db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 15:11:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"955a8017c0ff04eae2cfdaadf648e6063f2c0398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 18:40:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ffd6667a9dc4f739cc7641f83f1946f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9720dd0e4194c9fba879755f6ad0e90 : SUCCESS in 54m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ee0a4bc4602449ef9ba77fda51b9c544 : SUCCESS in 1h 36m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53e64a18bc2a4178bd78cd77f6f6bcd4 : SUCCESS in 56m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c49e962b2dc4531a512d4c0354ccca8 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fddd5c851a1d4850bfc0079b9cee4e49 : SUCCESS in 15m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca3f0c26f0ca4a75be3ab78f50902f4e : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a90de25d1a43cf9587e4f478a573f0 : SUCCESS in 9m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f0ae073df36d4663987ccc2dba4ed195 : SUCCESS in 44m 21s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0918ffc13b7c4cf69fbb37c2f77e12fd : SUCCESS in 40m 32s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6a3e7f5c73047cca0f7893e2e4dfee0 : SUCCESS in 26m 11s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/02ac86506f9c439cb5990ebe54b8106c : SUCCESS in 35m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e3a3e05b5e9e46799d173520e806e962 : SUCCESS in 1h 49m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/c573a115e7624138a3171531cc0315f7 : SUCCESS in 2h 13m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/284c2a7853d644f8880c840b5d3896ff : SUCCESS in 18m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe13e49dee334c65ba7ba3493a4f0eb5 : SUCCESS in 2h 16m 15s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f03c679143384a9f9dcb0bab8b98f943 : SUCCESS in 2h 33m 41s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/0a76570ffc374faaadb1e774f754a436 : SUCCESS in 1h 37m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dfb2c4faf0940909cf2a0a575f73ce5 : FAILURE in 47m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd9fe90a4f5d49c8b64b248361d02c56 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"75341b9f9a0755f28d83f2a9524ded5b8e93d0c4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-01-22 13:00:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37f71a62dca5765ac35ab06c7428aad4c58af698","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 13:50:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb8168c4ef87441bb9c4fb3873aa8395\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b4aec223ef74027bffa7c58ad1f6631 : SUCCESS in 32m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8832b1ce3847493c87a7fefcb5d709f7 : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"637463f98ef4484bccf246ba8e4c8e087055d97e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 15:49:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d812aa9ca54845b88097b79a6b356e1e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f7a4721289d4a4cbcb2cd3391e4cfb4 : SUCCESS in 53m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/841f8c0f43094bcda590527bb865ab77 : SUCCESS in 1h 20m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05a2ed8cb8e24f15b2a368579bff827b : SUCCESS in 42m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72eb355ba9084f92b62d8b4bd1302e65 : SUCCESS in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85d7f3de0454d09ad977e1d61243318 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d08a57c843384ec6b30965da9cc3740f : SUCCESS in 14m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/299b1da7a7554288afb33320a4d76236 : SUCCESS in 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c817b640be474f2585052c7af7db4028 : SUCCESS in 9m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a9f72d72d4b642e3a3d7e213b2693ed2 : SUCCESS in 1h 11m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ed94a7dca484423a38b5d8dc1b86a21 : SUCCESS in 2h 06m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/25b9f9783236436ab80ae2c4e9f0b540 : SUCCESS in 44m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a2c68086d8f646e69c9c367e3589242b : SUCCESS in 54m 15s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/80781845be534c789308c1378b906355 : SUCCESS in 1h 02m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6f30b21bf8e545f8af5b2e72b51d7411 : SUCCESS in 2h 27m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/9c7d1d48a82344f2be65b7697f1c9ce7 : SUCCESS in 2h 18m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/fc9612a3b99f4e35b1b299a9892e0349 : SUCCESS in 48m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f024d7ff676b498d992bfff4fe05cef0 : SUCCESS in 4m 43s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca417548b5e64341a23d6a05b35cb460 : SUCCESS in 27m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/653ae345016c44aeb57aadfef652a6ae : SUCCESS in 25m 29s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c35227493e4447d396fc7b8b93dff3fb : SUCCESS in 28m 14s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/362722bbd91040e983a1036a2f9f26eb : SUCCESS in 1h 52m 49s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/25d9da2ae34740dc8470312d2ad2dbd4 : SUCCESS in 1h 45m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/194159ba5de74444a6fe8aadd60b54fe : SUCCESS in 1h 32m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d40592b121d041f3a61406d5ff779753 : SUCCESS in 54m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bfbe446a5b2f4438879088900fc991ef : SUCCESS in 20m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7b8f9a5670394b4f81fc2aebbbefda8a : SUCCESS in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e78a698ea15bd4d24e01b7d6878a24589a68c624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 15:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8e9e7eaf790b4fe45bd58421ced0ab939531fe3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-01-22 15:53:59.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"58c0c6c570871ddddc14fb03d555184115ba5cc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:30:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e150795c234a13958f61645d362cff\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5907c4b7c125486d9b5e951599338cfe : SUCCESS in 1h 12m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/34c77f3784794d37be5d0d053ea45c7c : SUCCESS in 2h 07m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e42cb08cc95a4cf3a850dbceb9d246d4 : SUCCESS in 1h 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778f927faad947d5adf29835f110779c : SUCCESS in 14m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfe614817ba948d3ac7a305e900da5f6 : SUCCESS in 12m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/466ef0bc7c104b76a4c8ae04d033e1de : SUCCESS in 12m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb8a6d49f864f0f8b3e0ab431d7b2aa : SUCCESS in 9m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/55334b5652ae4d4c8f6495a7eb465a41 : SUCCESS in 50m 53s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/686bac9af26e49529c57791555501a8b : SUCCESS in 26m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9def3f861ece4e7ab27823a039254ed4 : SUCCESS in 28m 13s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/96dd18e46ad34b138957c0990a69e7c7 : SUCCESS in 29m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7d2b3d4023994295a817f977e20f370c : SUCCESS in 1h 53m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/b8c3d5af9e034bf68f105a0ec8963d3d : SUCCESS in 2h 32m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/80fc76f792ba4da3a142a49306ec40a3 : SUCCESS in 6m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a8ccd9f5feb1482d8a93ac37f4f20987 : SUCCESS in 2h 12m 04s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9c6a6f7e45642718b61a5d6f864d477 : SUCCESS in 2h 32m 47s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/98c2350e2c8647fd8193d31d81c53e14 : SUCCESS in 1h 58m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d837d37c7e684e21855b784b416ed020 : SUCCESS in 1h 38m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c1b4cc9a25641e09bfc56bb7e825193 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"25db57a82e19eabcff826051dc5659338b6cd0f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:30:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d23d5fe576569fce644e3eff2ae676c0288d6104","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:33:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73cdcedb1d274387ad661164229e05b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d09af98fc344b60ac80ac200a8a33e4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26d174b65d89976f448eb5b4cdf9930fbce3b132","revisions":{"26d174b65d89976f448eb5b4cdf9930fbce3b132":{"kind":"REWORK","_number":1,"created":"2025-01-16 14:19:48.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/53/939453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/939453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/939453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/939453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/939453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/939453/1"}}},"commit":{"parents":[{"commit":"e4efead033fc841ff6a93e75445c87dd8f23f7de","subject":"Merge \"Revert \"libvirt: Wrap un-proxied listDevices() and listAllDevices()\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e4efead033fc841ff6a93e75445c87dd8f23f7de"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2025-01-16 14:17:34.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2025-01-16 14:19:44.000000000","tz":60},"subject":"Run nova-next without periodic cache healing","message":"Run nova-next without periodic cache healing\n\nNeutron networking backends today are expected to work without\nthe periodic healing of the cache in Nova. Turn it off to gain\nadditional performance and in order to start testing this\nsetup in advance, before we are ready to change the default\nvalue to it.\n\nChange-Id: Ic90949cae840290dc376af7a7a3056babe2539f2\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26d174b65d89976f448eb5b4cdf9930fbce3b132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/26d174b65d89976f448eb5b4cdf9930fbce3b132"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
