)]}'
{"id":"openstack%2Fnova~939560","triplet_id":"openstack%2Fnova~master~Ib00ada67ae489f56c7b988d017cea32df39212da","project":"openstack/nova","branch":"master","topic":"dnm-gate-check","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-20 15:03:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib00ada67ae489f56c7b988d017cea32df39212da","subject":"[DNM] Checking gate health","status":"ABANDONED","created":"2025-01-17 20:48:33.000000000","updated":"2025-01-20 15:03:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"461900700dbaf440214588c25f2ed2498e4688af","_number":939560,"virtual_id_number":939560,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-17 23:24:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-17 21:25:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-17 23:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d9163128a55077a38e53360eafa2c37d4fd2dd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-17 20:48:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ec2fee615ad598bab2722236f454259535b6279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 21:25:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4bcc4cc59794eb8b4fd60c45df20538\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15fe2e33f62b45dca9d7bd795879d37e : FAILURE in 35m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/966a0fde91b446559c2c1ed162f99449 : FAILURE in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49a028c152815457ece6714720ac2780c44f3001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 23:24:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2026aab4d52941429f24af34e794b2c4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1e11f5a9da8849d683cb73432b0edfd4 : SUCCESS in 45m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/679552086305474a8f27cbdda62dd989 : SUCCESS in 1h 54m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdbc8ce8236347f7bf30ea425b2cd565 : SUCCESS in 45m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e376493b8f49457bb2671acc46eb234c : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d5ad0b8b564037ac05f014532243b4 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d0ea25798474e99996a84b7bc976f75 : SUCCESS in 12m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb123c6ba0ed4bae8d17d93ed4bf5c85 : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9fc18f0c7b84616a6943b27d4a8654e : SUCCESS in 5m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/abbdcd4622984202a91f7247d3600d75 : SUCCESS in 44m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a02d5ac28074a29818a3cef9c775a96 : SUCCESS in 1h 37m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/890f2ee242d14a32a161dbe979967ada : SUCCESS in 31m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/de06b5c026a7424bb1fc9db9a0395d20 : SUCCESS in 40m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3e76f603376c4ffcb7047387945783d8 : SUCCESS in 1h 58m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/f7bf025d1c2046d1ac696c73795ff4cf : SUCCESS in 2h 29m 36s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/00700f7e950a4bd6acb101cfe7497ed3 : SUCCESS in 47m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cebaba0a0f4145da83bd348a40606de4 : SUCCESS in 5m 14s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b55a7c691184d6695591dddbb89ed62 : SUCCESS in 27m 56s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba1cb7b83770474c84abfd16c26cdb5f : SUCCESS in 23m 12s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b48a542832044da79eb5b3164685b458 : SUCCESS in 29m 47s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/38cc15fa14ff4a4e9cfc49c8c4fb9027 : FAILURE in 1h 26m 26s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/129c426ba02445aeb5a14cf02d7b3781 : SUCCESS in 2h 01m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/da837daae00b4f3884d114eb33c5f7bb : SUCCESS in 1h 00m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b78a68570144c619a235c8736c7962c : SUCCESS in 45m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/684fb90d77bc498892f78466a844f56f : SUCCESS in 26m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5c56d345d8bc4feebca4661eba53501c : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"461900700dbaf440214588c25f2ed2498e4688af","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-20 15:03:32.000000000","message":"Abandoned\n\nGate is fixed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a01642fc09a5933e1d2e11ffa2a52a755ecbfbdd","revisions":{"a01642fc09a5933e1d2e11ffa2a52a755ecbfbdd":{"kind":"REWORK","_number":1,"created":"2025-01-17 20:48:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/939560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/60/939560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/939560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/939560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/60/939560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/60/939560/1"}}},"commit":{"parents":[{"commit":"e4efead033fc841ff6a93e75445c87dd8f23f7de","subject":"Merge \"Revert \"libvirt: Wrap un-proxied listDevices() and listAllDevices()\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e4efead033fc841ff6a93e75445c87dd8f23f7de"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-17 20:47:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-17 20:47:15.000000000","tz":60},"subject":"[DNM] Checking gate health","message":"[DNM] Checking gate health\n\nDepends-On: https://review.opendev.org/c/openstack/keystone/+/939559\nChange-Id: Ib00ada67ae489f56c7b988d017cea32df39212da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a01642fc09a5933e1d2e11ffa2a52a755ecbfbdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a01642fc09a5933e1d2e11ffa2a52a755ecbfbdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
