)]}'
{"id":"openstack%2Fnova~941347","triplet_id":"openstack%2Fnova~master~I8a70052d062316e85a182c389e84e09a19aca085","project":"openstack/nova","branch":"master","topic":"bp/policy-manager-role-default","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-07-15 00:45:01.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-10 05:51:24.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-07-15 00:45:01.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-07-14 18:33:15.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I8a70052d062316e85a182c389e84e09a19aca085","subject":"Add project manager role in Nova API policy rule","status":"ABANDONED","created":"2025-02-12 09:33:38.000000000","updated":"2025-07-15 00:45:01.000000000","total_comment_count":46,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"2fd5c48bcf0067092960e1b2f253c5fba6cd3e3e","_number":941347,"virtual_id_number":941347,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-29 20:37:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":-1,"date":"2025-07-09 20:33:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 10:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-12 10:14:45.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-12 12:15:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-15 22:25:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-02-18 18:36:03.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-03-12 19:24:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-07-09 09:30:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"71cd1681d6552a4d1c7d81984ce41c6172aed31c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-12 09:33:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d68aefcd7aef6c29d54afbab308102aaf9a7e688","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-12 09:34:34.000000000","message":"Topic manager-role removed","accounts_in_message":[],"_revision_number":1},{"id":"8918fb5ae5cf028a3d78cfc680ec1a36df3fb406","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-12 09:34:42.000000000","message":"Topic set to bp/policy-service-and-manager-role-default","accounts_in_message":[],"_revision_number":1},{"id":"532f78850e669b6dafab2fcdd6535876e0a79305","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 10:00:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af12079068d8415e90587fe65c261aa7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/910f33de2d3c49e48b336bae1f01de8a : FAILURE in 25m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea637e0251a84fe8b0a926d0473d347e : FAILURE in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"239841961dcfde4588914619c5cf9fbb3fd882a0","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-12 10:14:45.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-941347-5qbw2/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"1a5c2e1847b993d74aae7b914bec30896f7ee834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 12:15:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3e9f2b5a53742bb82cb79ed4621c137\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e48e881d6f44e5f8b5de185a54f24ee : SUCCESS in 56m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4c255397bb53486b99afedcb7957fcce : FAILURE in 1h 27m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fee65f5ad85c48568903cbaf984b590d : SUCCESS in 57m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6a63804384a4fde9660f11c37c82b42 : FAILURE in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2ca2e4065d41bb931bee54fbdc6793 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1ffc3676b2b4a4cb97abaca82d7527e : FAILURE in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f4670d1d8f14874a08b386520160366 : FAILURE in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7b8f1224364801bfd37e2b75f0f211 : SUCCESS in 6m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7ecae50038914249ba0bd54bda4b3284 : SUCCESS in 1h 02m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b5dd26b442a4527bcc03d09a8a8ab55 : FAILURE in 1h 58m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c4991c82a8164bde99027e4f3e517423 : SUCCESS in 59m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0371833cbf9c4eab847fee3ef94976ea : SUCCESS in 1h 01m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3cde458c23a345148448cb90653ea428 : FAILURE in 2h 34m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/4600a2f116aa491d868eb4d95d5daa6f : FAILURE in 2h 18m 37s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4c2500a5d98541d898df060c7755ecfb : SUCCESS in 51m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8c70c8d984614f03a179118941f3d96d : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea93e5ff18bf44ccb0f387d70cd9778b : SUCCESS in 26m 03s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d153020111054e0c9674ae483d8c9a34 : SUCCESS in 26m 13s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9c0b4539f0a94ffa996baf59a1a4d40d : SUCCESS in 29m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/014403f26fea44b5888be6134cf39006 : SUCCESS in 1h 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/356567f020b74508880de4ca714fa3c5 : SUCCESS in 1h 28m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/26f6cb9385f34a6c870de27d80d9df62 : SUCCESS in 27m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b2978f19afa04315a9eec57df76f9cc5 : SUCCESS in 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78a5befd19c3cbba0c35ba527ee6cff1d20172a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-13 20:28:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b7e1ff881762de97487e9a17e26430c196c626aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:57:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1d23600e1741f99c4ebee338a53913\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf0d19cae81d4c9c8ef500b76a2336aa : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b45a43a1385943089c36d9018ee88e2a : SUCCESS in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"432808c421793c437a68393f1fc37d83f7d46db9","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-13 21:17:54.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-941347-8p27g/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"8c95de612e8cac6f0906cc11651d821d9c6f95d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 23:12:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2ad9d634d8240deb71d066dd9de2fac\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b443b5a670944eb88424319bb6238264 : SUCCESS in 58m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d84c81ac078b4f329f147e00504d7d95 : FAILURE in 1h 35m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80dda445a5674c8a8aa75214a1690c1f : SUCCESS in 1h 10m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c890c35d5c744a28dc5dc5026ea72da : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c31f6070dee47118c393b2aaf179f53 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df19f7c6951b45bba4026667809c9ba4 : SUCCESS in 10m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1c41e9e7f4d4355a0791c703f52797d : SUCCESS in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be05121fc5848c584f000557d178463 : SUCCESS in 9m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cd43ae431f4b4fc9a9b0de83d0ad75f3 : SUCCESS in 55m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d16512c8a019440ba11ff8b3bd97003f : FAILURE in 1h 49m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b905c13b4a174fe89562f66b55b3744e : SUCCESS in 43m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4ca93cf251de41549b5ded4ff9bd6b5e : SUCCESS in 55m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a664f05e8d084fb9a611e8134b28b3fc : FAILURE in 1h 57m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/80698589ac2b4c02bcd7e8dfa71f51a7 : FAILURE in 2h 41m 00s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e900afacc11e49658e96e5a8cbb6f3fb : POST_FAILURE in 13m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bd81d45148274635a49ce7a522ceb7d5 : SUCCESS in 5m 21s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7160ee4b8fd24bdf99f4b2c49abcc077 : SUCCESS in 24m 37s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa67b00d9f4c40e9be7237f19d156921 : SUCCESS in 26m 37s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/10ab8ddfe68b4cb99a887fe418d87290 : SUCCESS in 43m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/19058014370448989ad27407797729d5 : SUCCESS in 1h 18m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/887ec565b2eb4449b609e55d5edcc8ea : SUCCESS in 47m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/db12dc12cbcc4de7a62b145f7ed249c3 : SUCCESS in 23m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/984de1eaaca54065abf05ce3b8d69928 : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"117fee5e922c4a196408201481d291ae51b67bd6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-14 18:35:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c79284d57fc48487b89d2dc464b97c0e873d88aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 18:36:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/323e98cfd6a24024849dc9161e935fcd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a00d2ff4f60d41a98db0e8105d363fef : NODE_FAILURE Node(set) request 300-0026341614 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d61cce3280b45709b113b0710bdd1af : NODE_FAILURE Node(set) request 300-0026341615 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd781596119c5f7114274361cac7a2fbdba15ebe","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-14 19:04:33.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-941347-2p8zf/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"4ec64368e5a6b2e50ece9b0420cec75c4393d32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 20:55:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2db3e29777c44f94a1aad223bc6c28e5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c38b7adb9ce43349c20f617f9da5c4b : SUCCESS in 34m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d3308cbb89b74795b853bcadd1f8b135 : FAILURE in 1h 04m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e902d4281ac84a518946ef77073240d3 : SUCCESS in 56m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fe8e441762e4167bfdf2c7b1b6e5089 : SUCCESS in 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/658e568f0f1e4dea8b107e29a4eab220 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d791e5d91bf4f1b9dae9b09ceea3f01 : SUCCESS in 13m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5dbb8866c7ee48138bb5eecac2ff459f : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827db26c7264498193f048167932010e : SUCCESS in 10m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dcf2c8c08a164a9eafbeb5180aa5a4e9 : SUCCESS in 52m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1edcb4d2d0af4321836564c408345c91 : SUCCESS in 2h 08m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b493407ce99d43078ae7f642b9ebb112 : SUCCESS in 46m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/240ccb562ad445f78bae54ae6caa9580 : SUCCESS in 52m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/253229f050344e6bb754742ee0a9cfdf : SUCCESS in 1h 51m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/78df2d9955294a6fbdae75c18d70e71b : SUCCESS in 2h 12m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/960469173d98418fbf8ebf4535935acd : SUCCESS in 39m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e5da342111bc4386b005f2f984ded79b : SUCCESS in 5m 36s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e0423b4fe4045e49151b61e7794a8fa : SUCCESS in 32m 22s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9185d64fb8134baebd659bd8824e42fd : SUCCESS in 29m 37s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3116cede0f634a93bcda7a03d07eb42f : SUCCESS in 30m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/55692295d1aa4498be05e93c4dc48330 : SUCCESS in 1h 15m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdc94a80f8b2495aa40e24df93283788 : SUCCESS in 50m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8915bebfa4d8411a88d260f3cb01ab5e : SUCCESS in 36m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/be23975fcb5a4886a7eedfb1fb5da7c2 : SUCCESS in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ba84a6d3609240359e2277f1dbe4f6b9636dff65","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-15 00:37:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e38a18bf166c69f0b6d323fd5b56d4a5c0160e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 00:37:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3de079d089ef487086919f1c9d2a57ac\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f36df1887209443f89d4068d25842cb1 : NODE_FAILURE Node(set) request 300-0026343858 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/882dc6feb53f4ea8817430e914557f48 : NODE_FAILURE Node(set) request 300-0026343859 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ca5b7f35c6b1be6c09a4ece6122e5a9d8b3885e5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-15 00:59:09.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-941347-52mws/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"5e49c7249538e7d2a40b4da34bc21e5f0df114a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 03:02:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d74c026c57d486492eec6ac9dba1fd0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b41bf99fecaa44d38be1474081101cde : SUCCESS in 41m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/69f1f26e4dd541849e609e61012e1cde : SUCCESS in 1h 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6a979eab612462d8ab94fe087206259 : SUCCESS in 49m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45fd448d4a59422e8094ab7fa040955c : SUCCESS in 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adbf8088fd174a1393631a6d96e044b2 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc53eb4951645a29053fec82eacfcfa : SUCCESS in 13m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecf13ad4d62a46ceb003f18abf41e1ee : SUCCESS in 13m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbeb04262844cc0b6da320c6fd1d450 : SUCCESS in 9m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7c6e1189e5c740aa8d6caf75fb1a013f : SUCCESS in 1h 05m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34e6591ad9bd4a7da8218e2fffa031d9 : SUCCESS in 1h 41m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/835a508863214ce68fa35e29625c174f : SUCCESS in 56m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a47bc225dcaa46aa9a008a1dea4d2eba : SUCCESS in 56m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c14921b6107947d383ed397cc22bcf94 : SUCCESS in 1h 53m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/95a1ebb79f52434f9c44f6bcdfb1b981 : FAILURE in 2h 22m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5e214d3bbbbd492896c94344ea306843 : SUCCESS in 50m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/47f214e7c4bd4df9b3ae18a815594de1 : SUCCESS in 4m 33s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9df0bb492f042baadeb18dcf5aa14a9 : SUCCESS in 26m 20s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28488ee7d0ea4975af709fcd1f93e277 : SUCCESS in 34m 08s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1b78ca325c3c401eae801f8af0e9dcf1 : SUCCESS in 33m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cddf2ed9d854c4690cdd22f0f8a40b1 : SUCCESS in 1h 01m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/381cf2fe2829402ab5206ccef22570be : SUCCESS in 39m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fe3d82c0c38d4011aa394a0af84524e8 : SUCCESS in 20m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8f28241824e742228936523c08002802 : SUCCESS in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cd5af71517f8ec2a92cd5c43abc5ba83ec37a3c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-15 19:54:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae458e46103354d537c5a65e994aa8b1e6d302d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 19:55:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b7663033a44461c87a6b7a511df19c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a9c022eae1c488bb189ec021ec7f77d : NODE_FAILURE Node(set) request 300-0026348954 failed in 0s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/138702d7a26c481487285eede5d22eb2 : NODE_FAILURE Node(set) request 300-0026348955 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ba72c2df5751fa5ea846bb1b10203782acc6391","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-02-15 21:01:42.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-941347-fc6zx/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"a93d3f739109ac9d91baaad37c9b7ccacc622764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 22:20:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f633a251fa4494b15b93b1c4507cb7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/471a0f90cc2f44ed9354b76cb2fea497 : SUCCESS in 44m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cdd54685b95b40c6bd7f0a8e8a00bc91 : SUCCESS in 1h 23m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe2c1c0fa27049ee86c64ad003de7a2f : SUCCESS in 46m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32f2ebd801144d8fab70b44ef72581cb : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d33bdc6ffd94bba83b53502569b170d : SUCCESS in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/067c3e9f4c404637b6258e3e36b2551c : SUCCESS in 16m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4db88f73e1534d2ca9f8f90153e0ff35 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04cfc818fa846098d5ed2d9e3aec7b3 : SUCCESS in 13m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1853a7d8385344a08caaeb07b7005ffe : SUCCESS in 54m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c2e01a6b2404f1fbe97dd2be50a77cf : SUCCESS in 1h 53m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bfc87b6b461542519aaebf325a677233 : SUCCESS in 1h 00m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b85ed39aac694e3db78d7761cc9db775 : SUCCESS in 52m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8475c0a1e7684c1499f39a3d8d42ecc2 : SUCCESS in 2h 17m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/b8ea98427dfa4cc1beeba1adac0f08cf : SUCCESS in 2h 15m 46s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5189551759a0444d8d9c17d389b95294 : SUCCESS in 43m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f32ea1568e94477daadd8b7a535e9d45 : SUCCESS in 4m 54s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f02b4c0d76c4e18a62c0e6e9eace785 : SUCCESS in 26m 18s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1db469a995284a639620abfb3ad5547c : SUCCESS in 28m 06s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9208f6a4ebd94497aa2aae23db1bc68a : SUCCESS in 32m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2a264c34586462085db70edd22cac79 : SUCCESS in 1h 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dea497f8dcd64ad698804fc81ef30486 : SUCCESS in 37m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0fd576504ea9414ebe120fc76cfb4d1f : SUCCESS in 32m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/361f9aa6b0c24b7c8b038bbef8188874 : SUCCESS in 19m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2bd7c5dd4bdde10ed1902f9d2591ad0e5c9484d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-02-15 22:25:49.000000000","message":"Patch Set 2:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":2},{"id":"61d3dd393548dfd228dc6e7db5cd7532f59c5db1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-02-18 18:36:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d9e9dcca5d81748c65c839e78992a74df4d42da4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-02-18 19:35:37.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0c9964525bd1f007305381aa8e0e617ffc4db6c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-12 20:16:44.000000000","message":"Topic bp/policy-service-and-manager-role-default removed","accounts_in_message":[],"_revision_number":2},{"id":"059638bb30c316099c1320cd132ee477185676ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-12 20:16:46.000000000","message":"Topic set to bp/policy-manager-role-default","accounts_in_message":[],"_revision_number":2},{"id":"5dc69028f5da40475a1681bc6532b13ca3eef261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-23 23:25:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5dda8e939ca1980359101144ba3201180038a8b2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-23 23:25:36.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"acfac83b2529e6afc88ae2960045125ff9e7b513","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 23:50:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aaa7b2922964c13a3bec44424480308\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d7cc56103d54cd2a42ccffcb99cd507 : SUCCESS in 22m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/31c5f58101ee4eee8bdefb0546a1ceb6 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c94d4ab2697d497d93801313ef89bfad : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a94f102cd8c7d50bbd68c4bdd80158a59141671b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 01:27:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6db62d5db414c24a89df0529901faea\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57645b5f1a5142a0abbb0fed460b0afa : SUCCESS in 1h 01m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/58415a5f042945d3b5a174f4d33555bb : SUCCESS in 2h 01m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74cef5e52585499fb33d92d774b4cc33 : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9e1c3adb1e430d8e838235ec214556 : SUCCESS in 7m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d9040207fb14a8ebaf977904cc22833 : SUCCESS in 9m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6beaba5a9caf4016ab6a9199473d6e92 : SUCCESS in 16m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7f95c1783904346be17e64c909ad4e2 : SUCCESS in 13m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3860337c7f92436283ac8a1cd146fc41 : SUCCESS in 10m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c42f91021fd6446c9e581bfb13c8442d : SUCCESS in 4m 45s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/311bb63be6204716b2dd97dfef678910 : SUCCESS in 23m 07s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/96234b74670a43d0a018f72a5805cba7 : SUCCESS in 25m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/620a39c580304412823a363b613721d4 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ae10445ab16a12f85bb81a2f2fe251dfffec3730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-25 02:27:29.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"75ce18cfe515e795a55a0e29f389bf7492b8ad88","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-25 02:27:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"620d50ca9c603d36cc6f02f338a9e65846f0dced","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 03:47:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f39d413ffc7b4f8ea81a09bb33e0e754\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67ebc2c9630f43988f2bbd0088799f2c : SUCCESS in 31m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37b60a63135344be9f3b5040b085e95b : SUCCESS in 34m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/69ec6f1526c54fd8ad1047afbe389327 : SUCCESS in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c0db74f3b715ff1595e6b89e3f71142f1e9da42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 04:09:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77b4640919ea44bebf7e1c7150085bc6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18125de5e0b640da91aa1dc5f1ce28d2 : SUCCESS in 53m 00s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/03e4bfcc61cf40e4b1c234d100576dba : SUCCESS in 1h 30m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dfbdea08bb145b68bd9d830f1e070be : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3684bac74c14a42b451b3c5c8680d2c : SUCCESS in 8m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5c0f3cef4cd4dbf9a00c252de4f8d75 : SUCCESS in 11m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0119bfc0be3d4ed3a203f695594e9a10 : SUCCESS in 13m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56b3469e76314cffaef9ed434d432f81 : SUCCESS in 16m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25da42359f4043f5bba23fe351869736 : SUCCESS in 13m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/74afe47c980f49c0ad8feb4f16984ccc : SUCCESS in 6m 32s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/252a781a1f8c40af84b14238d82346fe : SUCCESS in 27m 53s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/cc9ba25f2a7543708105db4c67bf8951 : SUCCESS in 33m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbd4bbce7a3e448cb0d9d630d2673d5b : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":4},{"id":"2b5b48c0ddc6615616c2457f961274d7f612d46d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-26 00:19:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc46704cef9ec5cdb433db24a177a84e5aafef00","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-26 18:32:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c04a32202a5095042e086dd62ccf01fcb2623f6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-29 16:25:06.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5a32f4d3d2f186e5403bb89ff8697aa38f5ac48d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-29 17:00:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a61eab2469a48d692eab0d825789eee\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/005282257d1845ec9b5e5ff071878ef9 : SUCCESS in 26m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbde587102ff4fa681f83a5e902953ff : SUCCESS in 29m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d4b75b528b624208b1597b98a289783a : SUCCESS in 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7849f1108b0aae0195401cbe485f5343afdae6e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-29 18:16:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d988355910474433ab6e6970255afc1f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b89db8354dd848178a585d7a40e06ed3 : SUCCESS in 59m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8b1136a530f14540ac6978465e32067c : SUCCESS in 1h 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0333bd562058427a98d32fe64d62b8e4 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce7f5a0af3e49cba3281fb856c7d3a4 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8639f686611c412db594e8e463aa93c6 : SUCCESS in 12m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce30ee505ee64a09a095106d1d436f45 : SUCCESS in 8m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30d9a65f97ca4a3ab12aa3c0421e2767 : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01cccb2f5f8a4a26a0feadacdaa102d8 : SUCCESS in 12m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a6b29667bea549d6925296354b4a99a0 : SUCCESS in 2m 42s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc98bcf941f0400dbfe477324f348353 : SUCCESS in 20m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ca10d5d0dd40455a81578d840988e09a : SUCCESS in 29m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb932b9e80834274a4849e0c6c3f929a : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":5},{"id":"315622c0be6588e12436d539b7b0c197672dc553","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-06-29 18:22:06.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"1b45a819215b5437836f6536fde3742239119c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-29 19:00:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4ccbd599cd48d40ddf01d4c09ff3facd72a98583","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-29 19:27:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2414d7dfce53448c8dd76385c22c0c86\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac4f11cc29c04bedb8a6fd3f97ccb8a7 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b97318b0ad9542dc89a015d5a7e9c024 : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72a5bcbe61704c589981b22fa6586574 : SUCCESS in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c032e5d2be54215e34d590c11ca6ecebea5cdbbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-29 20:37:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5280ff58268f44e5ba7f20be99ff7986\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f96a14fc16d541bd9c4ce13e4a8c6fb0 : SUCCESS in 48m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a7c95669df2b49f7a1bb25f456ed2918 : SUCCESS in 1h 36m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df1777f1e979448285afac2051c67aa4 : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632fec5730d84f07bdc3e83f7c5b15d1 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd3946d7a07a4fdfb34f1eed9975579d : SUCCESS in 11m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df061ecb34244dc1b913872f7c7d4db5 : SUCCESS in 8m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d29ee31c14b44ab813150dcd3ca1c2b : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03199814b174ee88c2d416b57c5a7c4 : SUCCESS in 5m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/157b24c3a2f24aeaad504026a4e2963b : SUCCESS in 2m 41s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90ea25525acb404f990e05e3ae5df3a5 : SUCCESS in 27m 18s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/21982352592a4ec9bc7b152000fd18c2 : SUCCESS in 31m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6087a55aeab94e4eaad0476125acf9a2 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":6},{"id":"709be439fbcd32b3ff062257e6ea85b34cbee41f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-06-29 20:42:50.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":6},{"id":"2c7f82fa00529ecb107437dc863ca6ed7bfb08b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-09 09:30:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"89048a70dcef20666caa30109b186e23a71885c9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-09 17:07:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"03ee6ac1e5b719ea77414c5a201c25805f43db89","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-07-09 20:33:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"87499f85cea36492b7ea8980f8343783d5caddca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-14 18:33:15.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c5da7ebf646f6c199a040532e7c4271f1d682a30","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-15 00:24:20.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2fd5c48bcf0067092960e1b2f253c5fba6cd3e3e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-15 00:45:01.000000000","message":"Abandoned\n\nthis is not needed as changes are squashed in https://review.opendev.org/c/openstack/nova/+/953063","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"524d01ffff6fec3dfb37dddc8ee9ae409aa1c4ff","revisions":{"b509efb8e7508e448586717f1682180937ddec4f":{"kind":"REWORK","_number":1,"created":"2025-02-12 09:33:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/941347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/47/941347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/47/941347/1"}}},"commit":{"parents":[{"commit":"37b5d1d6fd401366b4245565e888470389c57f3f","subject":"Add Project Manager role context in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37b5d1d6fd401366b4245565e888470389c57f3f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-12 09:12:57.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-12 09:29:42.000000000","tz":-480},"subject":"Add project manager role in Nova API policy rule","message":"Add project manager role in Nova API policy rule\n\nKeystone project manager role can be used for the\nproject-level management APIs and is denoted by\nsomeone with the manager role on a project. It is\nintended to perform more privileged operations than\nproject-member on its project resources.\n\nNova adding the project-manager role as default for\nthe below API policies:\n\n- \u0027os_compute_api:os-migrate-server:migrate\u0027 (\"Cold migrate a server without\n  specifying a host\")\n- \u0027os_compute_api:servers:migrations:force_complete\u0027 (\"Force an in-progress\n  live migration for a given server \")\n- \u0027os_compute_api:servers:migrations:delete\u0027 (\"Delete(Abort) an in-progress\n  live migration\")\n- \u0027os_compute_api:os-instance-usage-audit-log:list\u0027 (\"List all usage audits.\")\n- \u0027os_compute_api:os-instance-usage-audit-log:show\u0027 (\"List all usage audits\n  occurred before a specified time for all servers\")\n- \u0027os_compute_api:os-admin-password\u0027 (\"Change the administrative password for a\n  server\")\n- \u0027os_compute_api:os-deferred-delete:restore\u0027 (\"Restore a soft deleted server\")\n- \u0027os_compute_api:os-deferred-delete:force\u0027 (\"Force delete a server before\n  deferred cleanup\")\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nChange-Id: I8a70052d062316e85a182c389e84e09a19aca085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b509efb8e7508e448586717f1682180937ddec4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b509efb8e7508e448586717f1682180937ddec4f"}]},"branch":"refs/heads/master"},"a6af1b9bc5f90313722bdf4c7d38fe36a4cbb76d":{"kind":"REWORK","_number":2,"created":"2025-02-13 20:28:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/941347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/47/941347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/47/941347/2"}}},"commit":{"parents":[{"commit":"37b5d1d6fd401366b4245565e888470389c57f3f","subject":"Add Project Manager role context in unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/37b5d1d6fd401366b4245565e888470389c57f3f"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-12 09:12:57.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-02-13 20:28:07.000000000","tz":-480},"subject":"Add project manager role in Nova API policy rule","message":"Add project manager role in Nova API policy rule\n\nKeystone project manager role can be used for the\nproject-level management APIs and is denoted by\nsomeone with the manager role on a project. It is\nintended to perform more privileged operations than\nproject-member on its project resources.\n\nNova adding the project-manager role as default for\nthe below API policies:\n\n- \u0027os_compute_api:os-migrate-server:migrate\u0027 (\"Cold migrate a server without\n  specifying a host\")\n- \u0027os_compute_api:servers:migrations:force_complete\u0027 (\"Force an in-progress\n  live migration for a given server \")\n- \u0027os_compute_api:servers:migrations:delete\u0027 (\"Delete(Abort) an in-progress\n  live migration\")\n- \u0027os_compute_api:os-instance-usage-audit-log:list\u0027 (\"List all usage audits.\")\n- \u0027os_compute_api:os-instance-usage-audit-log:show\u0027 (\"List all usage audits\n  occurred before a specified time for all servers\")\n- \u0027os_compute_api:os-admin-password\u0027 (\"Change the administrative password for a\n  server\")\n- \u0027os_compute_api:os-deferred-delete:restore\u0027 (\"Restore a soft deleted server\")\n- \u0027os_compute_api:os-deferred-delete:force\u0027 (\"Force delete a server before\n  deferred cleanup\")\n\nPartial implement blueprint policy-service-and-manager-role-default\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/941594\n\nChange-Id: I8a70052d062316e85a182c389e84e09a19aca085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6af1b9bc5f90313722bdf4c7d38fe36a4cbb76d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6af1b9bc5f90313722bdf4c7d38fe36a4cbb76d"}]},"branch":"refs/heads/master"},"1c17f11242ea77c45dd4d960c12e429a38493f99":{"kind":"REWORK","_number":3,"created":"2025-06-23 23:25:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/941347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/47/941347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/47/941347/3"}}},"commit":{"parents":[{"commit":"470efebc5e382b4c25fbd3efb178964d84f19b50","subject":"Add new policies for live migrate APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/470efebc5e382b4c25fbd3efb178964d84f19b50"}]}],"author":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-06-23 23:17:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-06-23 23:24:42.000000000","tz":-420},"subject":"Add project manager role in Nova API policy rule","message":"Add project manager role in Nova API policy rule\n\nKeystone project manager role can be used for the\nproject-level management APIs and is denoted by\nsomeone with the manager role on a project. It is\nintended to perform more privileged operations than\nproject-member on its project resources.\n\nNova adding the project-manager role as default for\nthe below API policies:\n\n- \u0027os_compute_api:os-migrate-server:migrate\u0027 (\"Cold migrate a server without\n  specifying a host\")\n- \u0027os_compute_api:os-migrate-server:migrate_live\u0027 (live migrate server\n  without specifying host)\n- \u0027os_compute_api:servers:migrations::index\u0027 (Lists in-progress live\n  migrations for a given server without host info)\n- \u0027os_compute_api:servers:migrations:force_complete\u0027 (\"Force an in-progress\n  live migration for a given server \")\n- \u0027os_compute_api:servers:migrations:delete\u0027 (\"Delete(Abort) an in-progress\n  live migration\")\n- \u0027os_compute_api:os-deferred-delete:restore\u0027 (\"Restore a soft deleted server\")\n- \u0027os_compute_api:os-deferred-delete:force\u0027 (\"Force delete a server before\n  deferred cleanup\")\n\nPartial implement blueprint policy-manager-role-default\n\nChange-Id: I8a70052d062316e85a182c389e84e09a19aca085\nSigned-off-by: Ghanshyam Mann \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c17f11242ea77c45dd4d960c12e429a38493f99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c17f11242ea77c45dd4d960c12e429a38493f99"}]},"branch":"refs/heads/master"},"2035a0b2ee0ab11c66b37b6700698f21d468b8a9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-25 02:27:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/941347/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/47/941347/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/47/941347/4"}}},"commit":{"parents":[{"commit":"e687feafa820b197e6f7d5f60727df45093b9e68","subject":"Add new policies for live migrate APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e687feafa820b197e6f7d5f60727df45093b9e68"}]}],"author":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-06-23 23:17:50.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-06-25 02:27:29.000000000","tz":0},"subject":"Add project manager role in Nova API policy rule","message":"Add project manager role in Nova API policy rule\n\nKeystone project manager role can be used for the\nproject-level management APIs and is denoted by\nsomeone with the manager role on a project. It is\nintended to perform more privileged operations than\nproject-member on its project resources.\n\nNova adding the project-manager role as default for\nthe below API policies:\n\n- \u0027os_compute_api:os-migrate-server:migrate\u0027 (\"Cold migrate a server without\n  specifying a host\")\n- \u0027os_compute_api:os-migrate-server:migrate_live\u0027 (live migrate server\n  without specifying host)\n- \u0027os_compute_api:servers:migrations::index\u0027 (Lists in-progress live\n  migrations for a given server without host info)\n- \u0027os_compute_api:servers:migrations:force_complete\u0027 (\"Force an in-progress\n  live migration for a given server \")\n- \u0027os_compute_api:servers:migrations:delete\u0027 (\"Delete(Abort) an in-progress\n  live migration\")\n- \u0027os_compute_api:os-deferred-delete:restore\u0027 (\"Restore a soft deleted server\")\n- \u0027os_compute_api:os-deferred-delete:force\u0027 (\"Force delete a server before\n  deferred cleanup\")\n\nPartial implement blueprint policy-manager-role-default\n\nChange-Id: I8a70052d062316e85a182c389e84e09a19aca085\nSigned-off-by: Ghanshyam Mann \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2035a0b2ee0ab11c66b37b6700698f21d468b8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2035a0b2ee0ab11c66b37b6700698f21d468b8a9"}]},"branch":"refs/heads/master","description":"Rebase"},"4f9568ff9e43934f49714a281df095a3726e44a8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-06-29 16:25:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/941347/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/47/941347/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/47/941347/5"}}},"commit":{"parents":[{"commit":"bb0a8a1b3d543fad1ab8dce2ba04fd239e5400c9","subject":"Add new policies for migration APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb0a8a1b3d543fad1ab8dce2ba04fd239e5400c9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-06-23 23:17:50.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-06-29 16:25:06.000000000","tz":0},"subject":"Add project manager role in Nova API policy rule","message":"Add project manager role in Nova API policy rule\n\nKeystone project manager role can be used for the\nproject-level management APIs and is denoted by\nsomeone with the manager role on a project. It is\nintended to perform more privileged operations than\nproject-member on its project resources.\n\nNova adding the project-manager role as default for\nthe below API policies:\n\n- \u0027os_compute_api:os-migrate-server:migrate\u0027 (\"Cold migrate a server without\n  specifying a host\")\n- \u0027os_compute_api:os-migrate-server:migrate_live\u0027 (live migrate server\n  without specifying host)\n- \u0027os_compute_api:servers:migrations::index\u0027 (Lists in-progress live\n  migrations for a given server without host info)\n- \u0027os_compute_api:servers:migrations:force_complete\u0027 (\"Force an in-progress\n  live migration for a given server \")\n- \u0027os_compute_api:servers:migrations:delete\u0027 (\"Delete(Abort) an in-progress\n  live migration\")\n- \u0027os_compute_api:os-deferred-delete:restore\u0027 (\"Restore a soft deleted server\")\n- \u0027os_compute_api:os-deferred-delete:force\u0027 (\"Force delete a server before\n  deferred cleanup\")\n\nPartial implement blueprint policy-manager-role-default\n\nChange-Id: I8a70052d062316e85a182c389e84e09a19aca085\nSigned-off-by: Ghanshyam Mann \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f9568ff9e43934f49714a281df095a3726e44a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f9568ff9e43934f49714a281df095a3726e44a8"}]},"branch":"refs/heads/master","description":"Rebase"},"524d01ffff6fec3dfb37dddc8ee9ae409aa1c4ff":{"kind":"REWORK","_number":6,"created":"2025-06-29 19:00:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/941347/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/47/941347/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/47/941347/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/47/941347/6"}}},"commit":{"parents":[{"commit":"bb0a8a1b3d543fad1ab8dce2ba04fd239e5400c9","subject":"Add new policies for migration APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb0a8a1b3d543fad1ab8dce2ba04fd239e5400c9"}]}],"author":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-06-23 23:17:50.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmaan@ghanshyammann.com","date":"2025-06-29 18:59:05.000000000","tz":-420},"subject":"Add project manager role in Nova API policy rule","message":"Add project manager role in Nova API policy rule\n\nKeystone project manager role can be used for the\nproject-level management APIs and is denoted by\nsomeone with the manager role on a project. It is\nintended to perform more privileged operations than\nproject-member on its project resources.\n\nNova adding the project-manager role as default for\nthe below API policies:\n\n- \u0027os_compute_api:os-migrate-server:migrate\u0027 (\"Cold migrate a server without\n  specifying a host\")\n- \u0027os_compute_api:os-migrate-server:migrate_live\u0027 (live migrate server\n  without specifying host)\n- \u0027os_compute_api:servers:migrations:index\u0027 (Lists in-progress live\n  migrations for a given server without host info)\n- \u0027os_compute_api:os-migrations:index\u0027 (Lists migrations without host info)\n- \u0027os_compute_api:servers:migrations:force_complete\u0027 (\"Force an in-progress\n  live migration for a given server \")\n- \u0027os_compute_api:servers:migrations:delete\u0027 (\"Delete(Abort) an in-progress\n  live migration\")\n- \u0027os_compute_api:os-deferred-delete:restore\u0027 (\"Restore a soft deleted server\")\n- \u0027os_compute_api:os-deferred-delete:force\u0027 (\"Force delete a server before\n  deferred cleanup\")\n\nPartial implement blueprint policy-manager-role-default\n\nChange-Id: I8a70052d062316e85a182c389e84e09a19aca085\nSigned-off-by: Ghanshyam Mann \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/524d01ffff6fec3dfb37dddc8ee9ae409aa1c4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/524d01ffff6fec3dfb37dddc8ee9ae409aa1c4ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
