)]}'
{"id":"openstack%2Fnova~942654","triplet_id":"openstack%2Fnova~stable%2F2024.1~Ib5befdd3c13367daa208ff969f66cba693ae2c76","project":"openstack/nova","branch":"stable/2024.1","topic":"bug/2098892","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-03-25 01:33:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib5befdd3c13367daa208ff969f66cba693ae2c76","subject":"libvirt: Fix regression of listDevices() return type","status":"MERGED","created":"2025-02-24 23:37:39.000000000","updated":"2025-03-25 01:35:44.000000000","submitted":"2025-03-25 01:33:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942654-bug/2098892","meta_rev_id":"97a729abccad5297b004680c79a0c2259f8e50ba","_number":942654,"virtual_id_number":942654,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-25 01:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-03-24 23:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-03-05 13:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-03-24 23:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-25 01:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-04 22:25:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-03-05 13:14:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-03-06 02:09:00.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-24 23:49:55.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"73cb2ed8a0376367b2000a131831abab158b0cc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-02-24 23:37:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f1f1d36766492e5cc87a683ae2563f4d661d73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 01:57:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd9aae103f1a4431b35b89cf5da38a87\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e43be8ef8d56404b8d7eb2d66f127a00 : SUCCESS in 53m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/60bc119295134bd784f27f0aeda57a9d : SUCCESS in 1h 40m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c04608f5581b4643b6fe65450293fb59 : POST_FAILURE in 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405de91ed8274266846fefd8a29b6ddb : SUCCESS in 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0856ccd9726493687295ea121cc8350 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54da34885e93410f970d8c2a276cf8a2 : SUCCESS in 13m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01ab099ab23a46319fbb60973837421a : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1089b68c485b4fe89a8a0a268564d2e2 : SUCCESS in 10m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/480ccf8765a54b689fc8f7b4fe0a79f6 : SUCCESS in 1h 17m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5abb33e7b7a64510a0da17b3bcf6b61e : SUCCESS in 1h 46m 25s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d6630f8f9c5c44e98cebb861c6104e82 : SUCCESS in 1h 12m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/955ccf4dea8d4d95b5e72f802ca08974 : FAILURE in 4m 32s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d5682d6bc95848f0b8c35497a7dfb6a4 : SUCCESS in 24m 14s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3b184088f93f4a03ba24fd929d5b0d52 : SUCCESS in 30m 26s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6fa7522d4c9045d4b9d4931b1b1775f1 : SUCCESS in 24m 03s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a1f3b7dc05b345708f3f881e431ff96b : SUCCESS in 1h 25m 58s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/fdc13f33ee724c9aac9c99966c58f3cb : SUCCESS in 1h 35m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fff76b98a1f04aaea15ade716511298d : SUCCESS in 2h 05m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ea778496524423d94ba92d170a500af : SUCCESS in 43m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4ce86ae1e3674e48be0438a4caaa7555 : SUCCESS in 24m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/455cb6f8a2c4486a91d7aaf64195fcdc : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9477c7ea05457baa76ef36da1ebe0cca95b029a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 03:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de0b97eb02f4a5aa867bfacab0e1474\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/057ad0017c954d1c9c020cf61ced9461 : TIMED_OUT in 1h 01m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee7ac2663e734d5280ba80b60264eaae : FAILURE in 52m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb87c0d3dce9e13a2f37943c123e89ba53df568d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-03-04 17:30:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa1221d518cb6b2d639814428edb713840222e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 19:19:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97145ce0846245f49882c947d22ef825\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c486ddd1d36f4ca49e089d4cbba7d312 : SUCCESS in 29m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d6aab971220409287d45e651c078eea : SUCCESS in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58150694aa1bdacc75505287dc7265632c17d8e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 19:46:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bbaad2fcb154a599d3c33f570964500\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19261a4bac924e8a9ace81b18c5cb973 : SUCCESS in 51m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7e5160481043408798c300642a3a412d : SUCCESS in 2h 04m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33b30fbae6c8406b85771cfac7975d80 : SUCCESS in 58m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f9cd2301cd24a6a81c97677883a843a : SUCCESS in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03f22886f614853b754c98f941d1bc7 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d32c676536cc44aa89491f2a68812e81 : SUCCESS in 16m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/071646aa2db243a8994969345c851fd7 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f044dd34f0cd4efdae8a2772131d30b4 : SUCCESS in 10m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f167cf2a7afe46fd9b769f1547ad3359 : SUCCESS in 1h 01m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9dcbb6f61a844d019c0e92a524a62bd7 : SUCCESS in 2h 07m 42s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4609eed9255548d9ad2034d30764c540 : FAILURE in 1h 06m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f6be2983766e4863891c20d4d5affd7e : SUCCESS in 5m 42s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cccca326f8b74a90baac62fb75563d1a : SUCCESS in 25m 22s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a0332c7f84e479195bb81e279d779b8 : SUCCESS in 24m 23s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bcc7622cc2944f76a54db7657be15c0f : SUCCESS in 26m 44s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5cef6e55de1945619336b170d4f305e0 : SUCCESS in 1h 51m 58s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/dfc25f5f71064017ae7ac4048eacd92f : SUCCESS in 1h 29m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/64f71ad8821f44c9bddc62036cebec0b : SUCCESS in 1h 09m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cdcde227d8e4082b7801d2eb9ca29ea : SUCCESS in 50m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f4e5d86aeeb3488fb58c44fe1c1edc3a : SUCCESS in 20m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2fe81b7a16ef4197a8aa39369b6d83d9 : SUCCESS in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c25aa06ca99ef38ceaf881259ccd2d0b73bac666","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-03-04 20:29:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bcbe0709efc4e07631b90a87e4b6da8f11dd199","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 21:25:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3f66b638bb4d99b74083a2bde67a9e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f6154dee01f46fb8597f5058fdac89d : SUCCESS in 38m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b55c8bfa9e6f4df89cdde5c66a597e8c : FAILURE in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7227b6543798a66aa59815200697f100ecfbb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 22:20:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b74739c5be18427d840d3bf489c5f75f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88d78cb236f444c99e8ad480e2728319 : SUCCESS in 54m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ba398a469cb94836901a14a8a33354fa : SUCCESS in 57m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/851c3aa9bbb54d61b4269a0c61d8821b : SUCCESS in 59m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20307d2653724b87bbe2702b64031f97 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5268d5a943494936b04b7f8cc85490fd : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8ca88adbe764ccd90b2c3b6c63d0c31 : SUCCESS in 13m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a19fc6acd5346d891387d67c11df8e8 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53642be5d3ef44a7b3e6eb04796b106f : SUCCESS in 9m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/105fb4fd2397481384ce9746f979be54 : SUCCESS in 1h 08m 57s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b50078180baf418faf3782b88dc39ecc : SUCCESS in 1h 32m 13s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fa2ee1826d1f42feae0000e7d522a795 : SUCCESS in 53m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/37eea947e7054171872ce9851182fc4f : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/75b5fd5a02ff44218de5f003b01dbdbb : SUCCESS in 28m 10s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a561c17410524111a870bc9425857d74 : SUCCESS in 23m 50s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f619deae6964912972d0a4bd45948a3 : SUCCESS in 25m 32s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ecba7e96ff34487a7a5796dec01ace3 : SUCCESS in 1h 24m 35s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/2712722e0fc6421b9a894ab8daa3ce1e : SUCCESS in 1h 44m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/755a4c39bfe74d8196967102abbbd0a9 : SUCCESS in 1h 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebd2ba3ff94a47ce9e76747b5024d02a : SUCCESS in 1h 17m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc412ecba1334776b70e825a59e364d6 : SUCCESS in 31m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/89a27bbc6bb749168d6b280be36da2f8 : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1194a518c59096f4c36eb4a7a82e10e11e76d6a6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-03-04 22:25:49.000000000","message":"Patch Set 1:\n\nBuild FAILED. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"5ee3c09cede7f38311245d009275ff349186fd5e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 13:14:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"683941b4184abf3c38a7a8dd963667ac0d30cdc3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-03-06 02:09:00.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-942654-dsdcm/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"563741fa37d0b83ef5373ada2eecff7fc7de5695","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-03-24 23:49:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83e16582b56a9a61577acb9f82d04d0472aadfff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:50:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b578db1ce2cf251f2db4912a29b3d9743905e647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 01:33:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/691eb04f9e2042e0902bf37885a42825\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c385d234be4b4ec5aad7468a67e32345 : SUCCESS in 50m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/58372178b23b4a4983fc2de843305f66 : SUCCESS in 1h 22m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30801a9283a14562b4c6b0ac49e75232 : SUCCESS in 52m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f98f157f6d24d858871c7a7ab26cd49 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ac509b93274ac09faec5430dbdc142 : SUCCESS in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc64778cbfcf42479b6edf58e7457d5f : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010b60fddcd84baf8f6e52a3232b37b4 : SUCCESS in 9m 31s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cf230a3445e46ab82e6632f0db51a00 : SUCCESS in 24m 32s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38633725fbfe432ab482b2f943a05d37 : SUCCESS in 27m 11s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45e3e7e0de494f75a74dd54c204d05b5 : SUCCESS in 23m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/60fec28507ca4b5892e892ee9aaf999f : SUCCESS in 3m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5e71606f04c4269bb556764634246a2 : SUCCESS in 57m 35s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eae69c2507af4253906cb648e1f4e1ef : SUCCESS in 1h 29m 17s\n- tempest-integrated-compute-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/835749e22f8d453b8948bc021cb362ab : SUCCESS in 1h 23m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d9a0b537929482eb521eb1396e54142 : SUCCESS in 1h 13m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4df6e9a46f244ce982ea6c87cfa8eb42 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"71e8bc33a25c84e472a917f3624509fdb9afc6ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 01:33:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"97a729abccad5297b004680c79a0c2259f8e50ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 01:35:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cafa7ca979cf45e89942911e89a4ad1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b8231f8c234acc81048e106768fd15 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8bfa1efbb71be1909897b39ecc12e687f177e89","revisions":{"b8bfa1efbb71be1909897b39ecc12e687f177e89":{"kind":"REWORK","_number":1,"created":"2025-02-24 23:37:39.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/54/942654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/942654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/942654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/942654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/942654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/942654/1"}}},"commit":{"parents":[{"commit":"f516a0c45cc638c2114e0a496092e4565d3aa1e3","subject":"Reproducer for bug 2098892","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f516a0c45cc638c2114e0a496092e4565d3aa1e3"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-02-20 04:09:44.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-02-24 23:37:20.000000000","tz":0},"subject":"libvirt: Fix regression of listDevices() return type","message":"libvirt: Fix regression of listDevices() return type\n\nThis a partial revert of change\nI60d6f04d374e9ede5895a43b7a75e955b0fea3c5 which added tpool.Proxy\nwrapping to the listDevices() and listAllDevices() methods.\n\nThe regression was caught during downstream testing with vGPUs and the\nupdate_available_resource() periodic task was failing with:\n\n  TypeError: virNodeDeviceLookupByName() argument 2 must be str or\n    None, not Proxy\n\nIt turns out that while the listAllDevices() method returns a list of\nvirNodeDevice objects [1], the listDevices() method returns a list of\nstring names [2] and is generated from the corresponding function in C\n[3].\n\nThe error was not caught by unit or functional testing because those\ntest environments intentionally do not import the libvirt Python\nmodule -- so mocked code in the LibvirtFixture runs instead. Also, the\nupdate_available_resource() method has a \u0027except Exception:\u0027 at the end\nwhich logs an error but does not re-raise. So it would not cause a\nfunctional test to fail.\n\nThis reverts the change that caused the regression, updates potentially\nconfusing docstrings, adds type annotations to the methods that use\nlistDevices(), and moves the nodeDeviceLookupByName type checking into\nthe LibvirtFixture.\n\nCloses-Bug: #2098892\n\n[1] https://github.com/libvirt/libvirt-python/blob/408815a/libvirt-override-virConnect.py#L520-L524\n[2] https://github.com/libvirt/libvirt-python/blob/408815a/libvirt-override-api.xml#L448-L453\n[3] https://libvirt.org/html/libvirt-libvirt-nodedev.html#virNodeListDevices\n\nChange-Id: Ib5befdd3c13367daa208ff969f66cba693ae2c76\n(cherry picked from commit 2c07aa06458a2e70d79f8b3c68960bdccd11ab57)\n(cherry picked from commit c4f4ae784f2078b652d07053d5a69a81dba1a8f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8bfa1efbb71be1909897b39ecc12e687f177e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8bfa1efbb71be1909897b39ecc12e687f177e89"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
