)]}'
{"id":"openstack%2Fnova~946082","triplet_id":"openstack%2Fnova~stable%2F2025.1~I641140ebbe8aad149e403dce0d11c599a67e7a22","project":"openstack/nova","branch":"stable/2025.1","topic":"ceph_squid_test-stable/2025.1","hashtags":[],"change_id":"I641140ebbe8aad149e403dce0d11c599a67e7a22","subject":"DNM: test dependency on devstack-plugin-ceph changes","status":"NEW","created":"2025-04-01 19:42:36.000000000","updated":"2025-04-03 16:18:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":918957,"cherry_pick_of_patch_set":3,"meta_rev_id":"08173ab10512631259f64c7ea214d4229698435f","_number":946082,"virtual_id_number":946082,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-01 22:18:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":-1,"date":"2025-04-01 19:42:49.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-01 19:59:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-01 22:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-03 16:18:28.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"00b393e509035a5286ca49ab276c70c849f52434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-04-01 19:42:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5f3a007ff9ab714c8023ac134fdc41ae4c2f050c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-04-01 19:42:49.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3e0334057bcb20c92ee84d9e848cdd66939d6c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 19:59:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaed32274faf498b84a90e4c89c0f8ea\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c483d247021a42ccb7184dd078a14206 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40276298c0e94f999bcaa7e5904f4668 : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5684ef9d09a68028b9017b412b3b9f44470d5e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 22:18:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7cc2580b616481ab9f4d281f32a991b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ae54c36a8af4eb3b73ab4a58b0abc0a : SUCCESS in 28m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/11e45178e9a245a58af970e8e9eab7c5 : SUCCESS in 1h 34m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e964a5716408421fb3b7c4fe4cf669c5 : SUCCESS in 57m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7910c5352d44db78a26a6a68d692c22 : SUCCESS in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db191351049f47b68ae901d92ee8aff4 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cee2f65b8267423caa30b144bab98aa3 : SUCCESS in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5005278552f74d71a1bb7df9c981d93d : SUCCESS in 17m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6be91516f6ec41fa8bb47b332afbaba7 : SUCCESS in 13m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a669fdd407e744f094749b50acc0ed27 : SUCCESS in 1h 16m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dc31d7d0819f46fda751bf4f63398979 : SUCCESS in 1h 42m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/810fe99d9a064a35b529cefc25a9ca35 : SUCCESS in 47m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/faf717dde9a74489b3acab282d3ca65d : SUCCESS in 41m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/39e55ed36c4e4b349b115985ad4bff86 : SUCCESS in 1h 52m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/aebeb993b12c4d8284156c29b8d3d1a5 : SUCCESS in 2h 28m 53s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/97604bee6cf7422aa95c434d7a30cc3a : SUCCESS in 44m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/223296da5b6d42cdb63623b1ce656605 : FAILURE in 5m 19s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aecb7f29f94347b89131354fffe6503f : SUCCESS in 29m 09s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5355b10f228648f4b9e6eaaa6b66049b : SUCCESS in 26m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/47554a6b4cf442ec9d579a4d0d5bb359 : SUCCESS in 32m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/34e17790325b41fc8ff2823d03a90a48 : SUCCESS in 1h 18m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/753ad8f4a518413dbff0c75a671ee17f : SUCCESS in 53m 38s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/978279734e0848dfb512db7e119c20ae : SUCCESS in 22m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f8b59d5346c04f168f6999acbcf30de0 : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08173ab10512631259f64c7ea214d4229698435f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-03 16:18:28.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946082-9w777/index.html\u003e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06e5e3f37bd471ad89ad4ea263aea4078dee5aba","revisions":{"06e5e3f37bd471ad89ad4ea263aea4078dee5aba":{"kind":"REWORK","_number":1,"created":"2025-04-01 19:42:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/82/946082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/946082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/946082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/946082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/946082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/946082/1"}}},"commit":{"parents":[{"commit":"6422ebea0b152a2b34abdacf4453c70bf0ee37cf","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6422ebea0b152a2b34abdacf4453c70bf0ee37cf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-05-10 01:18:35.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-04-01 19:42:36.000000000","tz":0},"subject":"DNM: test dependency on devstack-plugin-ceph changes","message":"DNM: test dependency on devstack-plugin-ceph changes\n\nDepends-On: If6009392198b1d9e9c35e4fabae344a550a07796\nChange-Id: I641140ebbe8aad149e403dce0d11c599a67e7a22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/06e5e3f37bd471ad89ad4ea263aea4078dee5aba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/06e5e3f37bd471ad89ad4ea263aea4078dee5aba"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
