)]}'
{"id":"openstack%2Fnova~946223","triplet_id":"openstack%2Fnova~master~I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27","project":"openstack/nova","branch":"master","topic":"bug/2106085","attention_set":{"30099":{"account":{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},"last_update":"2025-12-16 04:11:36.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-07-15 07:08:49.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-16 04:13:52.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}},"26250":{"account":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"last_update":"2025-11-27 15:31:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-11-27 18:20:12.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2026-07-16 02:16:02.000000000","reason":"Someone else replied on the change"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-04-03 08:46:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27","subject":"Add check_instance_state validation to the four action API","status":"NEW","created":"2025-04-03 03:52:55.000000000","updated":"2026-08-31 18:11:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":26,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"70176d35d538cd4311e5fb5890fb815fa18276b9","_number":946223,"virtual_id_number":946223,"owner":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-16 02:15:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":-1,"date":"2026-08-31 18:11:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-03 04:34:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-03 05:51:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-03 07:07:24.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-04-04 09:18:35.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-20 17:55:04.000000000","updated_by":{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},"reviewer":{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},"state":"REVIEWER"},{"updated":"2025-11-27 14:44:36.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"CC"},{"updated":"2025-11-27 15:25:44.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2025-11-27 15:31:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-12-16 05:12:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-16 06:24:18.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-29 16:38:07.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-07-14 13:12:18.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-31 18:11:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"8d4885f8738b026a29a148e52a34391a6db07673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-03 03:52:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7d5a52c7b9aea64c73e90a0a57964de86bb9584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-03 04:14:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e80dab65a9e4b43e1977edb1e2b24d8da211c567","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 04:34:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d5ca305cf449c2ba91e3de70cd0d29\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df42f25cc99649d9a3caf22e028022ce : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d082a12214d48a4b5525b60ed84dc1a : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/37c7810aa01e4f5db9eab1a6205fa1cf : SUCCESS in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"377b7a09e6a2b5193721571fad3eb5000ed1b355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 05:51:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4253c71ab8c445d19804d11a6228b631\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b76efb7b8454ef48f3f266df336e0fe : SUCCESS in 26m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cb86bbb8db8449e5a178076bc131ad8e : SUCCESS in 1h 07m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c74f56bf709b40febe8ccf97d3f422e9 : SUCCESS in 31m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13e803f5ebc5400294ec7c10b4dcd06e : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a423a3f737fe4b12ba05b8e992f0d2b5 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/879f8c27f58a42179e0d03d7dcbfdaef : SUCCESS in 12m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e470c63e3ba48339d25e439d1c005d0 : SUCCESS in 12m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/120222e9041e46e48ad9d89941bc8576 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7485dde379624f7fb5d1126349d0c712 : SUCCESS in 10m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2b2666997d754f748b4b6fc6cf4b297d : SUCCESS in 42m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a091f6f7cbb4e0f85518509c59e1ba7 : SUCCESS in 1h 28m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/450c071c77e14ddfb87e372cdee2d733 : SUCCESS in 39m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7baa130a05cb4798b234dbd1488cf2fa : SUCCESS in 41m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bc28bd0dc1e14622a6504674dafe91c6 : SUCCESS in 1h 26m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/65b4351e91dc41d487b2503dc06a9b9e : SUCCESS in 1h 35m 04s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/90f60426562b4f5b80584245a1b97e7d : SUCCESS in 30m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/906add06674342e68c90617273ce46bb : SUCCESS in 4m 48s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f9d5eed3abe4dcabee02e47050d753b : SUCCESS in 28m 49s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4052215a8a24739ac2207af7999e0c1 : SUCCESS in 31m 58s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7346e3882c5647ac9fc4490a9362d365 : SUCCESS in 33m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e91e9d8f1854d9fbcd40ccc9735f954 : SUCCESS in 1h 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11c6100c986344a484e036e27619301f : SUCCESS in 36m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cab822adbb48403cab9241f4ada5405b : SUCCESS in 15m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6c2f09e508c94173bfc1e003feb3e928 : SUCCESS in 30m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c8ff128d6dcd669599bfe9718b1bbfe357bbc98","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-04-03 07:07:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"95096765f5e20735d16a4dd1259f3c12a6a7e231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-03 08:46:05.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2f2bb2dc7e83563fac52a001e543768b0eb9398b","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-03 08:46:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"55c73d1758dccb0cc41ab8fef0c34ce175148d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 09:10:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a482d7bb92574832aa55c77fb2296a35\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61573b6a6cbb4d4dab051edab26b3247 : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0b32bd89b914b669ce5920916514438 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c4bff411e6df4c91a31a9a55ea06518c : SUCCESS in 22m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a73e9ce6dd2132171216a862caf6c41406569601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 11:06:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f948eb2d99d4011bbdd3af36371b28c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec7f70b69dbc4ee7bba71d6b517abe62 : SUCCESS in 48m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/11607cf7a9de4f1ebc4ce56402651757 : SUCCESS in 1h 02m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5c884d58f024f689aa392233df7ab2c : SUCCESS in 58m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7628e977c3d44e5bedc20a3088e35fd : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0531868d63748518559dc01c626cd1b : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b91a6687329c44e38439113f4f122678 : SUCCESS in 11m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acc60c0cfec64f269676786313f97696 : SUCCESS in 8m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e552f001e21445aa66de426141e204c : SUCCESS in 16m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736a081f44c3485cb6e9ee19e42450b9 : SUCCESS in 11m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f894b6ea3df54135a1109579bb7454c2 : SUCCESS in 54m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/926224e1f3c94ff2b373e93c9f2bc427 : FAILURE in 1h 04m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9b05ac36a9b14dbd9af0932cecb29027 : SUCCESS in 53m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/75b18f6a71f947de8633950cfc5325ec : SUCCESS in 34m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/adf87680ac8f4987af93d9520e57db2d : SUCCESS in 1h 55m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/71d7190f75584c0bb5dd92f06b77f60c : SUCCESS in 2h 14m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/09c87a3118e248479f876a2ec36958fc : SUCCESS in 39m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ea35efd198bf42a1910c34e98cd1d608 : SUCCESS in 4m 43s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/366f2f9042ec44e68f5b2266af1bfa60 : SUCCESS in 30m 19s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5cff4acfab5471f98852f1c176e5c8d : SUCCESS in 29m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5cf391bf3e414fbaa7446b56bb30e797 : SUCCESS in 34m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/55143a7b75f64f76966a14af5ddd43aa : SUCCESS in 1h 14m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad4bbcc8cb754cd181e2eca256192b15 : SUCCESS in 48m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/af02567a8f784366960bfe6186bb2c8b : SUCCESS in 15m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/42be5f7ff2864d56b1f494d198b688b7 : SUCCESS in 36m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff16b476b3a3e9f5f4e0cf804fe9b75df065a1ef","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-03 14:21:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97b27982d6602cb6743aaa214fb273ffe1a5ae37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 14:50:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28262334aef247a6b43ca2ba589e59a6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/226b88cc0996483a9991189c9816d8f1 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/39b56e48e2984195bf78770c2ff5e3da : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/749eaf31ef914d0ba1eedc5f99336a59 : SUCCESS in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0f9fe6fffc59aacfda104aaed588579355aeb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 16:40:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a31799ef9ce7485eb3df246f04cf1830\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d5863d7930343718c035eb8c7f59e5d : SUCCESS in 1h 01m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b7bbb42fcdd6408b94e75086e2a424a1 : SUCCESS in 1h 50m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71f86972fca74405a23161e9a9133d17 : SUCCESS in 1h 00m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/406ec1fd5cf9462dbda32700c52c970a : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8c7a873ac34d749a9448c0a67fbf9e : SUCCESS in 9m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69f16af6c9cb4ecf837c6525ea473f85 : SUCCESS in 13m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d81f4f2146a647c89d59c17d8bea0320 : SUCCESS in 14m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebf9afcdc40c4a779b91d5cc9d0508ff : SUCCESS in 16m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7118f249fd9b45bb9d2c571369bddc41 : SUCCESS in 13m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fdd2961f4c9c44f7a0c4a6c5e08b2e06 : SUCCESS in 51m 30s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d8b81a48228434185d01fcf132ddbfb : SUCCESS in 1h 58m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a30debf49613496f93a4f7cd3e44c838 : SUCCESS in 38m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/502923617fb748cfa7164fd081567a77 : SUCCESS in 46m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/287e6e3d33b54fc1a46aac417aa782ee : SUCCESS in 2h 14m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f0401b1bc4e4e98b3f94f39c48e700f : SUCCESS in 1h 16m 05s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/041c10a994c341db918292f06eccb4a3 : SUCCESS in 38m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3a5c8b7ac9a14d71bb92fcbce6df722c : SUCCESS in 5m 01s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ccf41b155544754ac171909345f9b60 : SUCCESS in 22m 25s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc34bcc42d4b439f86507400405d092d : SUCCESS in 26m 32s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/65cac4f3358a4a428221e634fa3cb76b : SUCCESS in 29m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/761f0f0ef1724d498210e98845427c66 : SUCCESS in 1h 09m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/790d598c5ea94a0397b8900ce00b61e9 : SUCCESS in 45m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e29479ca553a445ea64afac3f9bc7920 : SUCCESS in 24m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2b92b4724062495cb1fca2fccc30c58b : SUCCESS in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b7b9c343260949417c1db8e7b8e5d44c3d61c1a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-04 09:18:35.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-kq2z6/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"7addd46678c4c3fee2338a8e46a965c995978365","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-04 09:51:10.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-f78zw/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"5cb1b8fb7fc4e0c521d446bf9d7888126140f7d0","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-04 10:23:07.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-s6t7k/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"33456d0f3495351f1ed539bcae1f6b20cff41c8c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-04 10:52:14.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-s52b4/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"38c3dfa5851d7f4b7d5b2890e3adbade4ea298cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-10 18:04:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f0d897effb6a0f251301a66ea142d1442bd9f2be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 18:30:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f15cc951b58441549dafc6591ebf93d7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31d407cd26d643d0b1c6b40989608035 : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e8934da9c63a40c18b3263012a91ceba : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0feafc63d2bc495086927b1c08dfcaf9 : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6453a74e7fd92b839d1e4724c4e8cfd5ae51f284","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-10 20:11:19.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-s7j7s/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"69dbf69ef21b955f3fcce18ba9a40519f9712f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 20:35:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27427fee94c74eabae4f573f94725cb6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d31479c4125a48c5b022d56e5f77a618 : SUCCESS in 53m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/78c8c95c3a304a2885f483df1a8a6c13 : SUCCESS in 1h 07m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d8b1070264b45dea3864ad7c7a02cb2 : SUCCESS in 52m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcf78b0257264e6691b199d4d2338ab7 : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be6f0d19e2f43f0b55c7b971314c59c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/883ec2bda786473ea16660f05429ec07 : SUCCESS in 11m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59b44bbf30f441aca18f0ba2ce9ac322 : SUCCESS in 8m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abfaed35e42c4847bcf860ab5fb11f2d : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a15069bcc9b42109bb52a96503b5796 : SUCCESS in 5m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c405ff1391a7475085df5165a4e37731 : SUCCESS in 28m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1f62dfd1c3e46ce962916efb09d2777 : SUCCESS in 1h 52m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e6296380aad84156a21d43c9164df265 : SUCCESS in 43m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/25ca0141671a4172a7195cac2a836015 : SUCCESS in 45m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0368299c12c14790b299421309cb1ade : SUCCESS in 1h 43m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/914557cd94734685b5481375e0433f95 : SUCCESS in 2h 26m 27s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/df13f822bc04445a8c9991bf9648fc87 : SUCCESS in 41m 08s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/466ac8505c5b46f9aa54269dc94eb54c : SUCCESS in 4m 41s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/433fbbe58f264cbfb54029af054b59b2 : SUCCESS in 26m 14s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/17163db4ef9f4bd4af48ccb1ef7cf4c0 : SUCCESS in 28m 04s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/879bec8b15944a16b4224476b71e87ca : SUCCESS in 29m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2baf1f914834dc991fccdda857f1838 : SUCCESS in 1h 12m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47622ebdf9fe430d9d21a53c74f4f78d : SUCCESS in 34m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bf348e04b55d41719168c4d258f7d79a : SUCCESS in 22m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2245246206864b8d9d1efafb921bdba4 : SUCCESS in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d3b81323fb1e9bcce6f68710d15d4c6a93da2f83","author":{"_account_id":30099,"name":"Nisha Brahmankar","email":"nishagb017@gmail.com","username":"nishagb","status":"Currently, I work in the OpenStack Team of LY Corporation (formerly LINE Corp.) in Tokyo, Japan."},"date":"2025-05-20 17:55:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e8787f9761ba935569f78488ea51a9800ee793cf","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2025-11-27 14:44:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a44db78b9ba21e67460fc88edb3f00325035d75b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-11-27 15:25:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"748aa27b1b605bba7304fe54ce20161112eb2003","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-27 15:31:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"75c7d86341da7fd037845f1f957170cae4380887","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-27 18:20:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"092ca6dbb15364446f17a1fbd62dc425555fa2f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-12-16 04:11:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1c76bd11fa7cae898e8bc583eb2c4a534e54d5a0","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-12-16 04:13:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b361b9e19ad87d82e8aae6d9175a1630d87f662","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 04:52:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60e92588c42448b0839973629483d255\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d475ba824e2042189addca82950b6cf5 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92e5d371d8f0491185c4aefd565ad9c4 : SUCCESS in 35m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c83f1567c014370a5b013c49ec1dc99 : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"639348ec412420c4ccea6098a9b761e329c733cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-16 05:12:18.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27 .","accounts_in_message":[],"_revision_number":5},{"id":"76ceef6c3adc179a771b8687afe87f7cc8be897b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 06:17:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3153d7ecb054cd6a2fb3f501bd18ba5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19ff184415c241b7906fa17da1502c63 : SUCCESS in 56m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/614ee8a7631f4aa3b51a57681252606c : SUCCESS in 1h 59m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/597b204afd384d0487198490ee4fba26 : SUCCESS in 1h 05m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33202241120c4e43bd6e4db197bf23b5 : SUCCESS in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1181582cbe9475c8ebd772a726e1c1d : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6a773f8f88548498746ee0b23cdc303 : SUCCESS in 8m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df66b0295dce49dbb3bb1debd2d8ca6f : SUCCESS in 13m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ac1c62e39744f1d8cb190768e9dd774 : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fdbbc651e87495b8eb221c9b50a066d : SUCCESS in 11m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/848db99c44da49b28cfa65a369617fdc : SUCCESS in 35m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2eba88fd9bf14d62b8ad55a77fdfd88a : SUCCESS in 1h 24m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ebe8b1391e5f44ee812e4aa5cc942329 : SUCCESS in 40m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/97dc32582f3147c2a73df35412c4f783 : SUCCESS in 52m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/baf261222a0a469f8fe1cdbf76ba687b : SUCCESS in 1h 32m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/09d2145ae2c84c4884345d5fb4125d0b : SUCCESS in 1h 18m 13s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/779a9a0b79ea43dab0f09ed46d4533eb : SUCCESS in 29m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/742959e4d8694a3584d013eb7f8d5f8c : SUCCESS in 6m 21s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1df3f39ca3384e39b4357f6d1982e970 : SUCCESS in 23m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dc90d50498ef4206aface394de02259b : SUCCESS in 30m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/81cc4900417e45869b164bee94664d3e : SUCCESS in 1h 21m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80c1b17d649a43ac99e4483b5a7abacd : SUCCESS in 55m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/640f3f1ebe6d41bd89cae3d44634c02a : SUCCESS in 33m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d116fda90a454a4d81d0f12edc1dbafb : SUCCESS in 30m 04s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/758dcac8d31e4c7aa77bb13164aaca9d : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":5},{"id":"b595b792c982e426c1da107e23d5b2baa5535544","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-16 06:24:18.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":5},{"id":"580115530e844ce98a1e945f5a3c90c86af11ba2","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-29 16:38:07.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9c12761f7b697a18d71f76c1dbe78e5cc8d8fdfa","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-29 16:40:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2838982272855e83babaa755fb4a173cc10f5627","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-14 04:20:02.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0654361778fc68f7bcd686b4d4df60651774ee5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-14 04:20:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"9d3ea0594ac0fd363733626a07b9497d2ae587a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 04:58:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c53a92ce384442cbaf0c5d780a1de9a8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f2b2e3c808549dcbf566d615043cb0c : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f519c40587a476d94376d1d4662ae27 : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bd08f57b88d14c3ca4d7125ba5fdb903 : SUCCESS in 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7933acfdef22e35a70cf01ff71b5484b943a2555","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-14 05:14:23.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27 .","accounts_in_message":[],"_revision_number":6},{"id":"69abd2d099197b97d1c1c3cd1a4f33be6b81e225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 06:20:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92a837e3455d482f8f453dbe6cbcf819\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/139d52dbecf844ec9ae9cc8b1c995d3c : SUCCESS in 5m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b9fa472a36a428b80e4efea978182f4 : SUCCESS in 1h 01m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f3991a7fddfd4925b04834670ea1b651 : SUCCESS in 1h 31m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bafbdb57f61144c98ffcf44f719870fb : SUCCESS in 28m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6127c1f19c9a4ffeaa3ae445af2bf387 : SUCCESS in 1h 02m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315784de9df945ef9940f4792d059878 : SUCCESS in 7m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e52d9150e94143a28d954ef4ec129738 : SUCCESS in 14m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8834218478b4d58b769d32b18e56d2b : SUCCESS in 13m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/76ee3cbaf2d840e692d796a34928bb0d : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb85f25c0434abfb58ff3ea12f11e2e : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8126df3ed5ca495a9a4791004a2ec468 : SUCCESS in 7m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/525d2ab09c2143c699381a85c3be1652 : SUCCESS in 1h 01m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a663f51329d427a9daa7322e7cba919 : SUCCESS in 45m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bb6fabeb845d4fdd90857e4d2e2e7917 : SUCCESS in 23m 58s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/7eba30abdc00480d9d28ab03fa334e98 : SUCCESS in 38m 13s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e2c58fa457b9421db2e167b50dc7c26a : SUCCESS in 1h 04m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a5722129c9a640c7a00c6fa9e88df86e : SUCCESS in 24m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/85c5937d3c194644950a3f4e7aba3d2c : SUCCESS in 1h 53m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/afbd537f10da474ea86e7cf340c6136f : SUCCESS in 1h 05m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f6e32ef58c144d929cd48dd53d3078e2 : POST_FAILURE in 24m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1898136be56342f8af970b84f7bb9fca : SUCCESS in 3m 32s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/06d2e6ee5a3147b9b9ff416e6e2a12b6 : SUCCESS in 29m 25s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/91cd5d6d226340afb5e427ef75a4b1ac : SUCCESS in 28m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e6e648947ab64e71a3dd9f3526135bee : SUCCESS in 47m 40s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4507c6f4951458b97b809208877ba33 : FAILURE in 1h 04m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b896604a71647858e73038a11484b04 : SUCCESS in 55m 52s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ed0cfa4ccb64bc88542fdfb8175141d : SUCCESS in 47m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0a7350a7bde243d58a16b136c33533a9 : SUCCESS in 33m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fd582b67c87c4a1ba4eecccfaa824728 : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":6},{"id":"af8c920fccf031b16ad4e7d392bee63d598e94e0","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-14 06:52:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"abb472e61f70929d9173a7897ac3acdf741c6877","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 07:32:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b6e0865671d4f4999cb3cfc4f5cdfef\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2a38b87c40314461a861a072c53ed1d2 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bcaf8aa6d5548b8bf9c0cc92574b750 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8e97061de767451abc64b28017890da0 : SUCCESS in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4ed2409f20ef75fdcaa0f3c7eba0639a3c8d7541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 08:46:20.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ec99b4672ea44bd86fdf26d7ad33846\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0213ac6755442009c7c384e11106172 : SUCCESS in 5m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce043abd23134a1594c0e9d946bce45a : SUCCESS in 30m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/186ac641330842a5a0d8994d67616de3 : SUCCESS in 1h 37m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5f7ddf3a3424081ae6a35b4f14257e9 : SUCCESS in 44m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a792164a34f4b4a9b0cd8eb4f0e4c43 : SUCCESS in 56m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a764838b5db40d196e1dc73b7072287 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a58554e8624e403cab7b1c48846dda2e : SUCCESS in 9m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68baf58c840747bb95f1c4cd8d12f3c5 : SUCCESS in 9m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38031aa5de91441c8cb99d3eaf427593 : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee381309328e4296b76bf0efac721092 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/649ace06cfa0461397f8981b0f46aa82 : SUCCESS in 4m 07s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3596a5809f9d41d68aeb5ce71fd57537 : SUCCESS in 1h 06m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38dff2d81b614b60a52a39d8ecdfe152 : SUCCESS in 1h 47m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/15f05b9bb23242a0af20a3196423733a : SUCCESS in 51m 49s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/042a7192d20a446fa7adead4eb370cdc : SUCCESS in 35m 00s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/8aec17dae9d1484fb200a7d7e476337c : SUCCESS in 1h 02m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d9e285dcab234df0b9838b823b3b0c69 : SUCCESS in 24m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/50d32fd4c5264aa8b54bb9262a8a2a24 : SUCCESS in 50m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/8f3e647186444b52bfae2e02accc594a : SUCCESS in 1h 29m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6867238b4c7f4d2e97d0b1e21dcfca67 : POST_FAILURE in 20m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9bc819148d994c96bd32ca1be176ee25 : SUCCESS in 4m 59s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f6608e68cd6c44c9a1c5fad4c3414666 : SUCCESS in 28m 34s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/45f82fd695f046128cf81dd57edeefa2 : SUCCESS in 25m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/bc6b34296b3047f6a3b880d0fb005090 : FAILURE in 33m 44s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/05b9fbd06841442e801488630090edbe : FAILURE in 1h 43m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0e2e373240b48d3bd12154a4a236c76 : SUCCESS in 28m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/01e7d8166e10472e99b6398264489ef3 : SUCCESS in 37m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c1be885d31254610b2622e851a1d458d : SUCCESS in 25m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/419fc51d51d144b5b4e6d33fd6dfbfa5 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":6},{"id":"2238164caa326e1e2cebfb9ed45a664b7370a936","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-14 13:12:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c34ca9e2ff1e703f35340364361dd50d897bcd5c","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-15 07:08:04.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0e7c41002bc079e47ea26e1fa1bb673f328362a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-15 07:08:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a3b8f478909d5e5da83ab4eb9d3c818a336a9dd9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-15 08:04:49.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27 .","accounts_in_message":[],"_revision_number":7},{"id":"fc626c46dbb22be033a8a46d3df66331847caaf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 08:42:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c3d9bacb5d43509cdee2af2988af9e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0244c844dd044a3c9080f5701c4487b0 : SUCCESS in 34m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7ce54b4022a4de280a1026c4e76448c : SUCCESS in 31m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8b7c0aec357d4357aadc082fa12dd5ad : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b552d2949e253f484fba621ed4a71f736e55b17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 09:04:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/981607cc18414c75a8f770bf76933b09\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0592038720aa4d4b807e076360123e58 : SUCCESS in 6m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf841a6eff04437f85809ad75da389ac : SUCCESS in 1h 13m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d7474f8ddf874152a51c75ae74095d06 : SUCCESS in 1h 39m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fdaf7e07dd54181b418cc957b19d8bf : SUCCESS in 27m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d03fdc1af3e4e409d8425606bf20415 : SUCCESS in 1h 35m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49c209844abb413c981ed63bc2a851ba : SUCCESS in 9m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4483bf9c9b334d75b694e1637f3d4439 : SUCCESS in 8m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4ffe3e54b2b4007a697dbd8dca473ca : SUCCESS in 13m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a3ed7e0c2134f9bac7518352efe112a : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7e14e5973a46febf6982c6fb4588f3 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5d1ff77181f4c3a9c735e961e0d71f2 : SUCCESS in 12m 31s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e64f727d970843128097bedb53bcba11 : SUCCESS in 1h 07m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0fe1a936a994d929fbc31df405700e3 : SUCCESS in 1h 20m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e21400752b354f97ab6c52b459d3e2e4 : SUCCESS in 49m 24s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/b3ac0e332acf474da8bfa203302118eb : SUCCESS in 38m 31s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/699826ba9c0d4f00b655474aa4da92e8 : SUCCESS in 1h 01m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a41f996201994b9d85169f1842c396cd : SUCCESS in 50m 44s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2253ee0e67f846a6bc7c92111ef7baf9 : SUCCESS in 1h 48m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/cd57d2f6245947078cceaf5d67c38d40 : SUCCESS in 1h 51m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/974674c39cf24bf6be5944def1d2a6fd : POST_FAILURE in 24m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/29e51c4ad9f84d0492f77ecf159da6c8 : SUCCESS in 6m 52s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b0f72478ba746aa81722fe1d1b45e40 : SUCCESS in 29m 21s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/be619dda9e534b418b2c2251b0b0705d : SUCCESS in 24m 47s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/713ce58ffd044cc8bbba5597f2f5c9a2 : FAILURE in 31m 32s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d550604f46a74db8ad8946ca1518686e : SUCCESS in 57m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/519286dfabd14fdd986faa47b4940e76 : SUCCESS in 1h 01m 30s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6601005899924690ac47d920907f8a3b : SUCCESS in 27m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8cd671992e274b32bb48457f0237a347 : SUCCESS in 15m 07s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/122a21950ba449a6b61385fd8d90e145 : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":7},{"id":"914c153adcd79fd0686598b7ac82035f95bac963","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-15 14:18:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"acc372d048b8a7af0d032686f6934f70b053d7de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 15:01:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0ac9a9e289d46e0b8186b89e399e804\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bc103aa746f49e8a6b412e3ff3da31d : SUCCESS in 27m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6b2698e958d6400d8ce8484e924ddac0 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/22adbf0e58274251bee09a6594bcad88 : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a80940edac9ce49bc3f4e6f3e0d68a315dff1218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 16:19:00.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88921ef269554820ba896e1b2c1666a6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7f849588bac470592a7fae54b3b4127 : SUCCESS in 6m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f83d7b7b9f448a1a5fa93cf1f29ddbd : SUCCESS in 1h 03m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8310de29769646f785a423a3d84b042d : SUCCESS in 1h 49m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce377e02f8ca4e8ebbe7d2e0b8a7a8b0 : SUCCESS in 1h 01m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11007add1b2e49609c2758ff0f65a094 : FAILURE in 54m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89c6f910bc2427894b1c3caa640a581 : SUCCESS in 9m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8f314ab33ac487f96832914d3eb94e0 : SUCCESS in 14m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/927883b625bb473c8a1776b6b13ca3a6 : SUCCESS in 14m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94c02f01a756458d8c78d1e3572320d5 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fc96c92dc644738a66153e216d23b8 : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec0eacccc4c8495c9b956e63ee87caac : SUCCESS in 8m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/af56acc6c2e9496791ef7cb6830e0a15 : SUCCESS in 1h 07m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/54f4d27f3d254b64abea3fdb613a1f8f : SUCCESS in 1h 30m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a8f608b3a138477b9fef259cd980570a : SUCCESS in 48m 17s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5fc1a9b07997420e91f2665f2e5a93ee : SUCCESS in 41m 16s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5f420a45507a42a48e3a39f6b796c4e1 : SUCCESS in 1h 07m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/58f27f08fc8c40119054ff41742cbff5 : SUCCESS in 53m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4f7ac7d119bd4a448e37836e1f7b24ac : SUCCESS in 1h 45m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/0220f01c920e40038929e688cdaa3f33 : SUCCESS in 1h 53m 43s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/593b75e30c0c4e5a8fc705d8a69548f5 : SUCCESS in 55m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3e7830cbd62b4cdab16996311e38c2bf : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0edd790baf14b2088c1ef4664b4372c : SUCCESS in 28m 28s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ba791be780ab4b0e9ef33852f1a6b1cc : SUCCESS in 26m 47s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f90d7ee515804221a45c331268b0ea73 : SUCCESS in 40m 34s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/88bc069a284c4400b7ce57e1644e0385 : SUCCESS in 1h 33m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bf6952ec11345a0b1cf23820fb327ff : SUCCESS in 1h 04m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ad8e9eb9664848f1a1140d836aa03042 : SUCCESS in 43m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c9b2e9a8838b4224a785d95bde294859 : SUCCESS in 35m 32s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b6451693cf9948daab5bf74a572ca0fb : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":7},{"id":"123bb06c9fe9eba4b99bf8d0abf0f36440fa412c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-15 18:26:09.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-k9c96/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"ff5b99198bd06dc7d20a54e9f9e544e78a5ef456","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2026-07-16 01:03:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0ccc1b7cf54b11566a5543a1aa43ba0eede4485e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 01:37:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13bad4004b264b52ae135ddb6b29f2b8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b5ce9cf352b24f178c3f6e50e4a8e042 : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e827d54bb7de48cd8557dffcd8c9e857 : SUCCESS in 22m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/99b66760626246f0ad948edb9ad07cc2 : NODE_FAILURE Node(set) request 5023461d2b054927bfbad61f4e1737e8 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5cd511f91d9faf2008cc0257391311450d43c5e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 02:15:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9875a6767a24455d949362bd38c14199\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84e04e11f2d74437b167ea1a4537de87 : SUCCESS in 5m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2c8abf84455415696be837fa748f8bb : SUCCESS in 35m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f84bca53226041a9b1ff9f3158f2c616 : SUCCESS in 52m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42d4b5d4702b4c98b3f02158c060c0e6 : SUCCESS in 40m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cc8ad2bd41b48a1b07bdd1acec59aa5 : SUCCESS in 56m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75cc8b7e2e31422586e68c45b4b28170 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/082dd7a4fcd948058b1dc9f8ea7ded0c : SUCCESS in 8m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef7c5bf65040484fb2dee6038cd5d7e8 : SUCCESS in 10m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c45859109fe4fc1a7f02b7d499625f2 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735af2d2700c4ec4b561a496622e2b30 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f90789f88b64625b0a599f31646e002 : SUCCESS in 4m 52s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a00d25c691ec49f6b99499730ea14126 : SUCCESS in 39m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b531a700a951459ab7f1604d7cd8606e : SUCCESS in 41m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/15503933b6ad493b8bae6e420831452d : SUCCESS in 50m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f04d99ff9cc94e6a88bf25a0ea5992ed : SUCCESS in 38m 52s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/608d3c397cde4323974208c7694ef63a : SUCCESS in 40m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c812eaa7ec4e4e75942a8469ef79fb5e : SUCCESS in 51m 00s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e99ebf0ee7c4e018a0aeb8435f4fa99 : SUCCESS in 59m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/626d4ea8500347beb7947a1678b87c31 : SUCCESS in 1h 00m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a76bf4f2cc33414dbd3c32291aef8a65 : SUCCESS in 29m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c478ba0c50ed4c8bb92fc9f92ec845d1 : SUCCESS in 3m 42s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8b98393ecef248049319875a5b44e683 : SUCCESS in 27m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/34e4aa342fde4160b73c162725ebf791 : SUCCESS in 25m 44s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ddb27b6f192348f096f5cf7bc5e7e942 : SUCCESS in 30m 04s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/501076134e224d8e8c08abc74d942a68 : SUCCESS in 58m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04804acb5a2e4331a58767f75dac338e : SUCCESS in 53m 42s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a5a9ca65a23440d1967162185392b79f : SUCCESS in 20m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/39c0bbcdad584687b7368fc41e574305 : SUCCESS in 18m 27s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cbaa5f769a51458696ead206f0b80147 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":7},{"id":"883fc232d9a655287b17be16ec0890b677df0a72","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-16 02:26:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"15375e849a785df5a6dc12435fe8a7c37a05d500","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-17 00:47:59.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-946223-hcbj7/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"70176d35d538cd4311e5fb5890fb815fa18276b9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-31 18:11:30.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7bec05f57c95ec57871d37d2c09c0a2d2139b352","revisions":{"7d106aee8bb0b093834c28deb69bcc59b3557f28":{"kind":"REWORK","_number":1,"created":"2025-04-03 03:52:55.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/1"}}},"commit":{"parents":[{"commit":"9d910ec4bf2a12baf3b5f0ec3bc41686413538fb","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d910ec4bf2a12baf3b5f0ec3bc41686413538fb"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:52:26.000000000","tz":540},"subject":"Add check_instance_state validation to lock API","message":"Add check_instance_state validation to lock API\n\nThe lock API makes instance action event record in the beginning.\nThe instance action table has foriegn key to the instance uuid in\nthe instances table. If the lock API is called before instance record\nis not registered in the main DB, the lock API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the lock\nmethod in the nova.compute.api. It prevent the no instance record\nsituation.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d106aee8bb0b093834c28deb69bcc59b3557f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d106aee8bb0b093834c28deb69bcc59b3557f28"}]},"branch":"refs/heads/master"},"1505f098e24d91722e14b31b2aa66c64140725e6":{"kind":"REWORK","_number":2,"created":"2025-04-03 04:14:20.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/2"}}},"commit":{"parents":[{"commit":"9d910ec4bf2a12baf3b5f0ec3bc41686413538fb","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d910ec4bf2a12baf3b5f0ec3bc41686413538fb"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 04:14:10.000000000","tz":540},"subject":"Add check_instance_state validation to lock API","message":"Add check_instance_state validation to lock API\n\nThe lock API makes instance action event record in the beginning.\nThe instance action table has foriegn key to the instance uuid in\nthe instances table. If the lock API is called before instance record\nis not registered in the main DB, the lock API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the lock\nmethod in the nova.compute.api. It prevent the no instance record\nsituation.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1505f098e24d91722e14b31b2aa66c64140725e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1505f098e24d91722e14b31b2aa66c64140725e6"}]},"branch":"refs/heads/master"},"5f5ec41f4746263e2bef81efb01cff3c75d6719b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-03 08:46:05.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/3"}}},"commit":{"parents":[{"commit":"9d910ec4bf2a12baf3b5f0ec3bc41686413538fb","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d910ec4bf2a12baf3b5f0ec3bc41686413538fb"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 08:45:06.000000000","tz":540},"subject":"Add check_instance_state validation to lock API","message":"Add check_instance_state validation to lock API\n\nThe lock API makes instance action event record in the beginning.\nThe instance action table has foreign key to the instance uuid in\nthe instances table. If the lock API is called before instance record\nis not registered in the main DB, the lock API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the lock\nmethod in the nova.compute.api. It prevent the no instance record\nsituation.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f5ec41f4746263e2bef81efb01cff3c75d6719b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f5ec41f4746263e2bef81efb01cff3c75d6719b"}]},"branch":"refs/heads/master"},"3042858006ee033fdff202c0cac8f216113454ad":{"kind":"REWORK","_number":4,"created":"2025-04-10 18:04:22.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/4"}}},"commit":{"parents":[{"commit":"9d910ec4bf2a12baf3b5f0ec3bc41686413538fb","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d910ec4bf2a12baf3b5f0ec3bc41686413538fb"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-10 18:01:43.000000000","tz":540},"subject":"Add check_instance_state validation to lock API","message":"Add check_instance_state validation to lock API\n\nThe lock API makes instance action event record in the beginning.\nThe instance action table has foreign key to the instance uuid in\nthe instances table. If the lock API is called before instance record\nis not registered in the main DB, the lock API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the lock\nmethod in the nova.compute.api. It prevent the no instance record\nsituation. If instance\u0027s launched_at is None, the lock API returns\n409 Conflict.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3042858006ee033fdff202c0cac8f216113454ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3042858006ee033fdff202c0cac8f216113454ad"}]},"branch":"refs/heads/master"},"33c01eb54bae0ac60ee1bf77330117ab4aea3588":{"kind":"REWORK","_number":5,"created":"2025-12-16 04:11:36.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/5"}}},"commit":{"parents":[{"commit":"fa46ec9b1cdc22060b4e87043bba2ff0adc3a6dc","subject":"Add bug 2106085 reproducer test case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa46ec9b1cdc22060b4e87043bba2ff0adc3a6dc"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-12-16 04:10:15.000000000","tz":540},"subject":"Add check_instance_state validation to lock API","message":"Add check_instance_state validation to lock API\n\nThe lock API makes instance action event record in the beginning.\nThe instance action table has foreign key to the instance uuid in\nthe instances table. If the lock API is called before instance record\nis not registered in the main DB, the lock API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the lock\nmethod in the nova.compute.api. It prevent the no instance record\nsituation. If instance\u0027s launched_at is None, the lock API returns\n409 Conflict.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33c01eb54bae0ac60ee1bf77330117ab4aea3588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33c01eb54bae0ac60ee1bf77330117ab4aea3588"}]},"branch":"refs/heads/master"},"07148105110301dae1279cca820109788d5b12fb":{"kind":"REWORK","_number":6,"created":"2026-07-14 04:20:16.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/6"}}},"commit":{"parents":[{"commit":"ed16482c1c5f7be058990979567e612b530b86f3","subject":"Add bug 2106085 reproducer test case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed16482c1c5f7be058990979567e612b530b86f3"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-14 04:19:16.000000000","tz":540},"subject":"Add check_instance_state validation to the four action API","message":"Add check_instance_state validation to the four action API\n\nThe four action API, lock, unlock, inject network info and reset state\n, try to make instance action event record in the beginning.\nThe instance action table has foreign key to the instance uuid in\nthe instances table. If any of the four API is called before instance record\nis not registered in the main DB, the API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the four API\nin the nova.compute.api. It prevent the no instance record\nsituation. If instance\u0027s launched_at is None, the four API returns\n409 Conflict.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07148105110301dae1279cca820109788d5b12fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07148105110301dae1279cca820109788d5b12fb"}]},"branch":"refs/heads/master"},"7bec05f57c95ec57871d37d2c09c0a2d2139b352":{"kind":"REWORK","_number":7,"created":"2026-07-15 07:08:49.000000000","uploader":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"ref":"refs/changes/23/946223/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/946223/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/946223/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/946223/7"}}},"commit":{"parents":[{"commit":"6eac583ec31aedba5c3221d37bb26f6940f941f2","subject":"Add bug 2106085 reproducer test case","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6eac583ec31aedba5c3221d37bb26f6940f941f2"}]}],"author":{"name":"Masahito Muroi","email":"masahito.muroi@linecorp.com","date":"2025-04-03 03:41:20.000000000","tz":540},"committer":{"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","date":"2026-07-15 07:07:45.000000000","tz":540},"subject":"Add check_instance_state validation to the four action API","message":"Add check_instance_state validation to the four action API\n\nThe four action API, lock, unlock, inject network info and reset state\n, try to make instance action event record in the beginning.\nThe instance action table has foreign key to the instance uuid in\nthe instances table. If any of the four API is called before instance record\nis not registered in the main DB, the API returns 500 Internal\nError.\n\nThis commit adds the check_instance_state validation to the four API\nin the nova.compute.api. It prevent the no instance record\nsituation. If instance\u0027s launched_at is None, the four API returns\n409 Conflict.\n\nCloses-Bug: #2106085\nChange-Id: I64d1eaa18d65ff6903f83905b3ad4d02f8fb5b27\nSigned-off-by: Masahito Muroi \u003cmasahito.muroi@linecorp.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bec05f57c95ec57871d37d2c09c0a2d2139b352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7bec05f57c95ec57871d37d2c09c0a2d2139b352"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
