)]}'
{"id":"openstack%2Fnova~947212","triplet_id":"openstack%2Fnova~master~I31b1aa854d8c95e47ba476051a650937b739a52b","project":"openstack/nova","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-04-24 15:58:24.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"11082":{"account":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"last_update":"2025-04-28 10:17:55.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-04-24 11:48:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2025-04-28 10:17:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31b1aa854d8c95e47ba476051a650937b739a52b","subject":"Replace eventlet sleep with time.sleep","status":"MERGED","created":"2025-04-15 11:45:39.000000000","updated":"2025-04-28 10:18:53.000000000","submitted":"2025-04-28 10:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"947212-eventlet-removal","meta_rev_id":"3084185c50d9d80ef918709c76f5b550669abe39","_number":947212,"virtual_id_number":947212,"owner":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-28 10:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2025-04-24 15:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-24 13:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":2,"date":"2025-04-28 08:44:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":1,"date":"2025-04-26 07:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-04-24 15:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 11:46:39.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-04-15 11:46:39.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-04-15 12:16:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 12:30:19.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 14:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-21 07:00:45.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2025-04-26 07:11:44.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-04-28 08:44:40.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"34c8e48c1ad8cb81a89ad12d6cfde7a726c9363a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-15 11:45:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"009dd5aa4958fdf10c1316bd33ceab58f45f83ab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-15 11:45:52.000000000","message":"Topic replace-eventlet-time removed","accounts_in_message":[],"_revision_number":1},{"id":"6ac8d500a0506f58e3dcce3d1a9936738fb72e5e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-15 11:45:57.000000000","message":"Topic set to eventlet-removal-part-1","accounts_in_message":[],"_revision_number":1},{"id":"280948ca9040e57748dd59109add7b0d75da5b36","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 12:03:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52d86f5e17a0651defc8267c703127fb45b85cd2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 12:04:20.000000000","message":"Topic eventlet-removal-part-1 removed","accounts_in_message":[],"_revision_number":1},{"id":"de1af398a7344014e54fff5ea7e0f55c28c1c70b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 12:16:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8b0e4cc1859405faba28942adf5f6cd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98601b71f34040d9a42ad623a4abe654 : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32403e07783842b59487dc721d3405f6 : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/52108f262e0e416994a7c5c1340e34c1 : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fef0d383a85b22ba771f2afc746c9bd107ea36e7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-15 12:22:48.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":1},{"id":"f4a3a50505a36212e3a8d293f942a1f5d51f4a8c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-15 12:30:19.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-fkn4m/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"1e844ea009870c04b7dc84a0d99b5f30d11c54c8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-15 13:00:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1dba8f684a96d1583a68137adc141a81d264c589","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 13:28:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8da0656a0f572e42c33e0fa279e1913da2c745c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:09:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b692fbbfb82d489dae74f69658ed9b7c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf7ab691b1fa4763ab6ca879116f2a2f : SUCCESS in 32m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8923d3063c2f416e8bcc38478a1a01c6 : SUCCESS in 1h 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4a623cfd478481b81e1d2c2257ef59f : SUCCESS in 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/078c6be70de94248a998694344b69f28 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01f16f454794721a5046885d69f4bec : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53c9170743134d4783e8f1cf1f4661ae : SUCCESS in 8m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58040b2c2fd043a9ba32677eda9713e0 : SUCCESS in 13m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5beb89a3fa5f42d2a0b84ee026514aff : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a4e66c6d9c04fc0b7fcd096ae7dc512 : SUCCESS in 10m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8ac821bb89d04387b2b450d7a45bd93c : SUCCESS in 55m 57s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3f2f4f008344a878de1869bc2f61a91 : SUCCESS in 1h 28m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/976588b80a01465d886f404642dcf34b : SUCCESS in 49m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cf3599f227954381908a7b6e5dba70f5 : SUCCESS in 51m 28s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d93341b241574f348c985c3112ed1777 : SUCCESS in 1h 13m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/35e3c19167bb40b196c2634628eb645b : SUCCESS in 2h 00m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/8b623ddd6ae14a568b7fdce3717842a2 : SUCCESS in 1h 54m 31s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/deeb2d71d8104128a0fd95cca118e476 : SUCCESS in 37m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c5c6a3ce1be9458093329add493005ef : SUCCESS in 4m 44s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb735a4b12394221b477494ffc7473c6 : SUCCESS in 29m 36s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5175edf9a2754defa76d7d47ffa3a7de : SUCCESS in 22m 08s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/06f0e8116ab941b4895d008ba310ee3f : SUCCESS in 32m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/17516b3120d6486c8ffd0de2b0d382da : SUCCESS in 1h 04m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/021d51b536e9456ba141563cacce8a6f : SUCCESS in 53m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5ce7b70b8d21460e9b6aebac59981d50 : SUCCESS in 21m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/69325225a18c4f48a40b2117c3772a8d : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"741dfc89991c43edce771b3d7298afca956133d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-18 10:14:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"caa7027916daaae10402b8da0a1de202b989e76d","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-18 10:15:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b2e2b2f80f8c0f20c2c46c3a668a759f1e47c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 10:38:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf684c237e841cea9ba9a1a953650b0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21cbe3cc5db1489a9ee3bc5973e367d2 : SUCCESS in 17m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/48448fc96e3a45cf9089864b6891daee : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d81b67545f6421fb2dede0bb345f7c9 : SUCCESS in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42cc5fcf623c2730c35786a420bcd4c7d4efb0f4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-18 10:42:08.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-4d4xg/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"b31c2aa87ecb9391445265029aa3604e684b8c1a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-18 12:20:16.000000000","message":"Topic replace-eventlet-time removed","accounts_in_message":[],"_revision_number":2},{"id":"5589dc86ea224ab419cdd72e2f0dc10cc868c866","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-18 12:20:40.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":2},{"id":"47288857f92e0dfab2583a792f1dd889d4787039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 12:20:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aa5a936a2994bd1bec59abc680e329b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1edaf0837fea41e0961a73b12b733385 : SUCCESS in 41m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3f8d9efe798940aea9a3434e5e146dc0 : SUCCESS in 1h 34m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/810624d0fef64320a6718fdb952d675e : SUCCESS in 46m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec467f30b9104613ae168ca7ca55798a : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08553e45f800472aad001dbf9bc986ee : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ce94b2305e54bd08673e96f50349d22 : SUCCESS in 11m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13a1c742d69f47ee93fb305fc39a7590 : SUCCESS in 7m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57a2534f0ca1428d9a164fba66567686 : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b9e8a395684042ba28b29f9318236a : SUCCESS in 5m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/27b8dd5f7ad44e2f97e6c01cf7aca8a8 : SUCCESS in 46m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bcee528eccec4e97a9960d1e23c8599a : SUCCESS in 1h 25m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5427c042f64b47fbb9b715765502db96 : SUCCESS in 36m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5fe464561f9143d281a72193d2ee395f : SUCCESS in 27m 14s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/36f1482762f240999d7e8b66418bce80 : SUCCESS in 1h 07m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d596e442b8b742ca853880b49e57b72f : SUCCESS in 2h 00m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/3a239a7495b94e7a98a01dc64bbb44fa : SUCCESS in 1h 13m 13s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8e03e1b40a7c4eada0f3242776a154d2 : SUCCESS in 41m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0b9f04ca73eb4d5697272492630d7c62 : SUCCESS in 4m 43s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ed3e7bdbe183495399a9a262bd25ac65 : SUCCESS in 26m 06s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03ba6a9a31b64d298411e4ba8208b3c6 : SUCCESS in 22m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4226531937e041c49df77052f7ebba7f : SUCCESS in 29m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9440d05b727049b784984a089b070dc3 : SUCCESS in 1h 08m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88f125c79cee4fc6b44d8bbb5006905f : SUCCESS in 30m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ff020e1c84c64ea5b52b9dcf84b91e3e : SUCCESS in 24m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/24e0764789904adcb48956c9d40cb8c7 : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fea09718d74ae5293fb36b7423ef166646fb309b","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2025-04-21 07:00:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6212cb9c94b7f763d299483f594cdbe77d313d22","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-22 07:55:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"829516a6b0c80147e2b97b068b2bbf557a18d38e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-22 08:14:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"374ca7adaa6985b8ecaad95bb41bbac220d4619b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-22 10:06:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"634491cb4ae2b0aa145f09fe14daffa576491b64","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-22 10:07:32.000000000","message":"Patch Set 3:\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4643cf258141dc20631a6a19bbb85bfa0859f2b3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-22 10:36:06.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-jvj8w/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"e1ca58c236b79609bb1cf54a79c5e29f4c6c4914","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 11:19:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf57c8fdc764c5aa06483eb2ec74b39\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5047f602f45043409a1f5a461470d52f : FAILURE in 28m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce14852efca04861bc6138c0e9cda2be : FAILURE in 30m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d9a212c8f39d48fb80c9574d4841d7e1 : FAILURE in 47m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"021fa8d0fd64acd0dd2e68c54a185dcb44aa123c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:12:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/877e91dd35ed46fd8d70faa1a9952106\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e472a39993bd4020bc536add6dc200c6 : SUCCESS in 55m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/996b8a9bb10a4da09eb7498b32d9b62f : SUCCESS in 1h 53m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/189b522cfe8241b0a2a5a817dc63c239 : SUCCESS in 57m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a682ad7461f94d1ab45607ae26470c9f : FAILURE in 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef668ee81f949a29cfae590308436f4 : SUCCESS in 8m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd841d5897da4411930e17cbd53bd61b : FAILURE in 14m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1461572a21e464ea8b44760971aed20 : FAILURE in 13m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b383a8f4baa64fe786b2aa716fb49cf7 : FAILURE in 16m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ae4725cf7d47c7bf69386cd369b9a4 : SUCCESS in 10m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a3506dd9b0be4cafba39431dff997395 : SUCCESS in 53m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75a45fb1b08143dea8efa1b4c65a1ffe : SUCCESS in 1h 03m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/60e12a0b89e04fcfb4eb75839335d15a : SUCCESS in 37m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/994b677536e44e25a5f354de39771ffd : SUCCESS in 29m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6798cf6a7bd14deaa0e947887a84772b : SUCCESS in 56m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/89ceb6d8f75a44c59ae98d660c47187b : SUCCESS in 2h 04m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/d4cf52f7391a45e8949103484ddfee93 : SUCCESS in 1h 28m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5fd749b496e94de6bb3072ab3a2a4433 : SUCCESS in 36m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/40fa5891328b4f67b79301d19ca96dd5 : SUCCESS in 6m 16s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05b55ca8c8154e85949168e1dab669ec : SUCCESS in 22m 47s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d51e9f3e67ec4ef3ba1caece51726620 : SUCCESS in 24m 47s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a834a83b4fb642ce8b16cf43eca16e10 : SUCCESS in 30m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/abb93d895f95497b976c38a6b10d1f95 : SUCCESS in 1h 20m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8640ea6e98245a683d6ae5758070245 : SUCCESS in 56m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e51a2f5552a94d5f9aed6ab7d019630c : SUCCESS in 26m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1b845e29f1304d6d9d36941f4aa7bb37 : SUCCESS in 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"264acc202527218258ed3d931c2784b8fa1058b1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-22 12:47:09.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"751337763f9a400cf2b560ece85aabee1d751acd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-22 13:18:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79dc8da670dc97c052419ee59fd0d61ec7a762c0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-22 20:04:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fdb04145358999ec46ceca1c77ce1d6efbf114ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-23 13:24:12.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"175424b0c6cab7f2301f06e8a9032c7625c6cec1","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-23 13:32:33.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"56222ee5bf2bfa74b520e21de9ffa0b6e7355b29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 13:47:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db826cd377314bfeb8343527b2063fbe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/503232d6e2f645d68bb29b4f62ec85fe : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/012e6d49503f49b990ee0373e366f980 : FAILURE in 17m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/429141f4e45747009f78c6fce7d11694 : FAILURE in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af5574a6441fc16d17763fe58bc3f2b3dea28716","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-23 13:53:06.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-djwxg/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"aebbd8e9ada260acf695ec1c14ba5e44db11347a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 15:39:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/300798d48d314a3ca2e17579727e4832\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/badb1a63f63d4b3f9bd1468c64197241 : SUCCESS in 27m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1ddc957da62b45d0a4b71b979d83c6ab : SUCCESS in 56m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a8baaaf9d6c4e37be9d5e2c578f4db6 : SUCCESS in 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37ceac8067ef403a9a1738f18471877c : FAILURE in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e70ff0d18a54e8dbfc42ab61adde56d : SUCCESS in 7m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/788fd78fbf9f442ab435d0e71f3001f2 : FAILURE in 7m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9930d087e10e4023bf501ac29bdd8c09 : FAILURE in 9m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b21cbd921604dc2a25466717a313346 : FAILURE in 15m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606f0128bf1e4a06bdf18d30a9fbd525 : SUCCESS in 6m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fdfe51a34ab74ea7b2a0d0be832f5531 : SUCCESS in 28m 29s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/49593b09b25a4be0b0b08671d30a420f : SUCCESS in 2h 09m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e70fbf1957ce422c8fcf401cb4c9f406 : SUCCESS in 26m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/127fbf74cfd84630a5406b92b4fb69f4 : SUCCESS in 47m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7707bf9a5dad4cdc8c9829875ca7afee : SUCCESS in 1h 20m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b862ed841965448a9f104a60be4a845e : SUCCESS in 2h 01m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/e9db3882a7a7440bb5b8550ddd9aca49 : SUCCESS in 1h 45m 50s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5e5c65447e054316a6500d00997ab205 : SUCCESS in 41m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/63287409b1174c95b299830d70c3937f : SUCCESS in 4m 50s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e085ef898b0f4eb3a5dfeb03591b25bc : SUCCESS in 25m 41s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5eb330ceb133401bae0232b6302e40ee : SUCCESS in 29m 41s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/194c2f397bf74b83a6f20ade306c03aa : SUCCESS in 30m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/462244d0c4704d84bdb3d5f84c01570d : SUCCESS in 1h 19m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb7591f4eaf44801ad08e06384f854b6 : SUCCESS in 49m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8a9066a3ac204622b290b2235c50bcc7 : SUCCESS in 22m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0e25d78234394b398eac16ce0c6c0ca9 : SUCCESS in 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5571bfeeb3ba9751c88ac7f5e2a83f19a1880482","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-24 10:04:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d59f107dd62efd07f66f42bb91e2f76714ff2af1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-24 10:07:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8877ddc628de52144c17302114aad021ac79c896","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-24 10:09:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9105e97a7c2e77b0076510e445ba04dff278a717","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-24 10:53:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c1179594438896755130d352412afb4a27dc7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-24 11:14:06.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0a70431ac95571a24b4834957f18d422e5afc094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 11:46:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48f8742f3c234b3bb54e0f8382171406\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27745fc2bdbd432b9150dc8ede88c98c : SUCCESS in 28m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d5869ac112a4313a1034872e4df3177 : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0e620fb69334dc089741d1406ececa1 : SUCCESS in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f28297310a7c238e01f1198388a9592915b5485b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-24 11:48:41.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cd48c32c55128fe6ace8498706fda577ba6eb2ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2025-04-24 12:09:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"77862a690f31583a5537b616476b6445d0b3672f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 12:30:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/618b99c453ef4b47b7dda9935116d688\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14ca9b4886f54f06b9e3b273979b45df : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6ae813efcd694a74899ce24efd04791c : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3e88fdd33b9404699c99739bae7098e : SUCCESS in 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d4b2c00ea6ac06949dd7387f1ab6deefae01c5a7","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-24 12:42:21.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-6fdsv/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"705be84292a2ccd75f49ffe14399d8a908b445be","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-24 13:11:42.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-qkggh/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"111b2a0ee9fb3d5dc77fec7a5a921a3afe4bf5a4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-24 13:39:51.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"64f3564280aeea06025e512c466dd4ffdd984cdf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-24 13:41:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fdced4ca7c13c8992a923c64b268328a3e6e55ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 14:33:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d766771b55524192a922086a096dda63\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed8e78134b374f738459f966d4d4bbbc : SUCCESS in 51m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b2f52416f2af4555a172bfa0d793f2fb : SUCCESS in 1h 30m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2672d73dec6045418e0a673b679190de : SUCCESS in 46m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1dfb1491644456f9f8a9b479d0c0dfa : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056e9419de054e6c9a9c7db11767c5ea : SUCCESS in 8m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1d96543f8ad42228bceda5f1e37736a : SUCCESS in 11m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d874b52c5164359992620e26b99f767 : SUCCESS in 12m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89626d5a26b84849a2bcee9cb76dbb93 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aae37c109354754974f31c8115ba9b5 : SUCCESS in 11m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dea3d3f5879f456baa8d5cc3405190f2 : SUCCESS in 56m 43s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b3b50af514b46cca778b551bfbc8051 : SUCCESS in 2h 17m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e083580668fe42698a080891ac1747e7 : SUCCESS in 38m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c50f38f5bcf3445eb029e8f8e1a0cede : SUCCESS in 49m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fff237f33e314d9f96ef25b81720bb5c : SUCCESS in 53m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3d6eac3c802f4ba4b3ad53311b453ae4 : SUCCESS in 1h 58m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/9b001321a760408ca827ce3f0cb9f978 : SUCCESS in 1h 20m 03s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b365d15e7b6440abab8d611eca59c49e : SUCCESS in 40m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bb4e3e935831499c94b79543a9c1e732 : SUCCESS in 4m 42s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bc9774de57a425387ed01f7dbfdacea : SUCCESS in 25m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cf185416c44c48d3bbfc918fdb0b98ce : SUCCESS in 27m 40s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e06c1e1edd424924936101a448682f59 : SUCCESS in 27m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbd700d0dfb14936a5ecbf26d1be8fbf : SUCCESS in 1h 12m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/503d8e6784f541b09c6256259bf9f5cd : SUCCESS in 49m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6ab0c5a759fd4de7bf17ada912cee4b0 : SUCCESS in 20m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/22558db08573481385a61ab7a5a36685 : SUCCESS in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d805e1d3e969401edcd3b576c9e52952fd648039","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-24 15:58:24.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e14b57a34e8da9be4e3c4f5bf1f9fbffaa3b590b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 15:58:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b410da419fab2fea05587c0a5300e556cdbd2675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 18:06:48.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d741a9de81ba49ccb140f06d529dd61c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22f7b8ec77f643d5981a22ae65db0534 : SUCCESS in 55m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/21fd754fe3ce4882850524a919fe4a6d : SUCCESS in 59m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7659ea93773f47be91a7c3b5dc513441 : SUCCESS in 54m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69caccff6bcf4caba9598fe10f7d9434 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a353033429cc4a2587c5167d28cda357 : SUCCESS in 7m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35d193c5999a4acba34d66073d477cbb : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8581f07adff64153b6fa3fa6c68f1e9a : SUCCESS in 10m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e13bbc92af814d05af2e581a5a6c6240 : SUCCESS in 37m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fafa8a511074bce972596038aa68afe : SUCCESS in 28m 38s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec3e8760ccb84b8c9f5a022f9d073f27 : SUCCESS in 28m 14s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e34300e6f8c94137bed9d427c164a559 : SUCCESS in 32m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/acad63a9208b4297b04d8acb9a56d8e6 : FAILURE in 1h 13m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/1d60a10641734a88a6302fbdda223b18 : SUCCESS in 2h 06m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/645bb9573c1246f4925ab1af413ec012 : SUCCESS in 7m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd06b4b92c634c84a6d36c0428582d72 : SUCCESS in 2h 01m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/20f54298e5b84340afeff1d5003b2373 : SUCCESS in 1h 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b6a4e2a92924bca9f604d51cd35d474 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":6},{"id":"b31c7c5e5d189bd301200ac37079d9875e4fd2e2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-25 06:53:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"df8e8a62f6b0e9fc09fbba33934f9ede5529609f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-25 07:18:51.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-fxt62/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"48c0c6cea7912f1845754d77026d6ee546357022","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 07:28:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edaf0e3cac384a41a746f5d6e47054e5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9c602c8ae7f40d3a29a1c36591e823d : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cc624c7a6634e1e9ded70c1004a9a34 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/487ee66fa2e54dbead0bd2e3a02f0ddc : SUCCESS in 32m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1abaea737ba414179a3d92b7f015eefe9fda2823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 09:20:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82221bf0769f418eb9e7642ffba8b05b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89ad91e76ba44cde957ed9165b239ed9 : SUCCESS in 48m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/da0d1dd7038f408db786c82a994fba8d : SUCCESS in 1h 39m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df5d0ddfd5c044d7a136f14f5ec4e5ef : SUCCESS in 51m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/826bbb33011746209a512177fcc2ebae : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8a14e8487f446199a1b2961442df9a : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9e8ef748bcb45359ef63625a8703e2f : SUCCESS in 10m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a87ff3a12b514ea18f6910d21af8dc36 : SUCCESS in 7m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2cddaa05ca843b793d0c3d4d5b9a30b : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a173a70c7024ff79c0835bf4a85de7a : SUCCESS in 5m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4da5208ff27946779186cf04c634050c : SUCCESS in 49m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e53881b2e694ff8996b55bf7c477e79 : FAILURE in 1h 31m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/509f10bc4fc04dc6acea10e0bea27317 : SUCCESS in 38m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/efd5e8c8b46f450ba193bb2c3eefbbf3 : SUCCESS in 40m 25s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3114d6092f5643a28fd80fd4ca4e11bc : SUCCESS in 54m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3130735dd55e47ccb619b166af46d9bd : SUCCESS in 2h 04m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/b4b48c4213114bbb863a3eb6e0d59416 : SUCCESS in 2h 20m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/4509d2255a714e07ab0a8063b678883b : SUCCESS in 33m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/adb94248cd7e478d923ef0107935dffb : SUCCESS in 5m 09s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8eea81a9e1704963bd2f70f2da32d52e : SUCCESS in 27m 44s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4319077e6954a45b6561282a77c2fd1 : SUCCESS in 22m 51s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6a2b9a64120848d98b20e7fe5b9be8cd : SUCCESS in 24m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb5964ae3e7348758d4779b39dde2285 : SUCCESS in 1h 16m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/109adf04705a4691bc00240a27c1bbbb : SUCCESS in 56m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7692bc76dac24a239b7c4b1ec2ad4c10 : SUCCESS in 18m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5f4dc6255d1d42d291dd5aea345627e9 : SUCCESS in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"19d44811ca1ac6adfe18c43815155b400463825c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-25 09:34:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"752b5eab322667e1813608ccccbb2c2d81011831","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-25 09:58:03.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-9cfh5/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"269c44c6e4816451a88a29e148bbb8f0475b8cd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 10:05:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f55487d031054342aea2367a2433d66f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86c97635a1b14a62a45f070e6f7e14f8 : SUCCESS in 28m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a06dcbe59bf44f59aa89f73a8ee0b6f9 : SUCCESS in 21m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21bf5d1b31fc4616ba9f9ddd8d87431c : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b50097d62b53cdcc1044e98005b14da479997c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 11:37:16.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e57f0a98f97c4a6e959b945f3ed31abc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b36419d235d4dd59823166e0d372496 : SUCCESS in 1h 05m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/78ad21568e0c4fa68616e7297819a19e : SUCCESS in 1h 58m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3654acbf90cc4b0cb49299d23aa4203b : SUCCESS in 56m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d2444bfabcc4b5294ee8ee062c8aca9 : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7da9e152744e8e8e4a43034e45b207 : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3908d51b5bad4180b0527f3a1696fc09 : SUCCESS in 13m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca6b4834851c425c83a9f26f506b8086 : SUCCESS in 13m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4954398c438d4ac2898bc02612730981 : SUCCESS in 17m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/165b837254fe4166bd9decd6c036ef64 : SUCCESS in 13m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6655a5f48dd2472ba2ee0a6d939d746d : SUCCESS in 32m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a2556ace9dc466087988980ee538a93 : SUCCESS in 1h 08m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4910549d17eb4695a2c029c3ab05c4e3 : SUCCESS in 36m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/57c99ba422f9406c9ee5e7c29ee2e7af : SUCCESS in 40m 33s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b7a17ac9237b45949ba49e0131a7285f : SUCCESS in 56m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e8322ee79f442378f252a7cdf9aec2f : SUCCESS in 1h 30m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/22371d5b7e054393b1a95c0b0d07a169 : SUCCESS in 1h 58m 43s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/be44b5866c7444569c93756e33cf4714 : POST_FAILURE in 22m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/889653792f4249c487e701a40fac4cfe : SUCCESS in 4m 54s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7dfd5bbe350e4da49a11f3cfcec6c69e : SUCCESS in 22m 13s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9608fbafc32649469efc9f6469176945 : SUCCESS in 26m 53s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/08bec1d498f3492eb37216570e349dd4 : SUCCESS in 27m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b49765b620614b44bd325ebb677b7ff1 : SUCCESS in 58m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/040942393a8f4aa381967bf98f3ffa47 : SUCCESS in 42m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ecf6d51eab814a85abbd1dc56c9c7ab0 : SUCCESS in 27m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/064528b6e7db453d9056da5f45d31537 : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e06565a8163e21c470a23c2bebf1aeff2a063885","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-25 12:29:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5f5cbcc62ed7c2de5fb6b32dc9afb29fe7021a75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 13:07:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/677b51d8e0d74e77901629e7751094d3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e50843aebb4e4c71b570c7d9e9a63fdb : SUCCESS in 31m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/96d1507ed0dd4db1be6f77457d1c5c4f : SUCCESS in 29m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/931b4a90100146e69866fcdc735a422e : SUCCESS in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e60565d9699aaaf2b031ed11974cf8073fc09195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 14:51:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f818928d24444aab8583dda41354fc58\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f56ed043b9724bfe914935dc2206c2d7 : SUCCESS in 49m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d24f73aa92ec442bb6c421fd21f5abc5 : FAILURE in 1h 32m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49afa6f5d58f45609ded09254e336c44 : SUCCESS in 32m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a5dec17e5a54b84baf78f7717d9c8f1 : SUCCESS in 24m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5630d460de44b4b0edba89dc4baa2d : SUCCESS in 8m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b169b0ceab64097b6ca822b7d33da59 : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b1c5253f2bd433f998cc3ace604e57d : SUCCESS in 12m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f58be479dce4028a54f3a1c2752e55d : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1464bd628d3f445bb0c63f395b9659a9 : SUCCESS in 9m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/61ef8543dddd45ba859213b5aff757fc : SUCCESS in 52m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/71f177363ffa48e6b919d9496425bcb0 : SUCCESS in 1h 27m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/20cff456e68140c4a11af8775511b981 : SUCCESS in 40m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5e4cedf309e64eecb2b3dbd03a6f5d76 : SUCCESS in 49m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f522e82dd30c447c9ab3546d55a9bd1c : SUCCESS in 1h 10m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cdb8b64fd9304ecd84d94685b046ca34 : SUCCESS in 1h 51m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/2960f4b101364560bbbd7bf765737504 : SUCCESS in 2h 09m 44s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/95d2cb26e78740c79c46c24c1524edda : SUCCESS in 43m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bfe74cb5c32f4264a698aa1aae225a1a : SUCCESS in 4m 25s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1274cb957c31442bae816b6d228f0e4e : SUCCESS in 25m 28s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a84f062f0344e37a022ce7d21e2058c : SUCCESS in 25m 29s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6e2a1b4930b944a9a11fb32029bc8fd8 : SUCCESS in 38m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/868841270beb446b885f674394240fb9 : SUCCESS in 1h 00m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a410d283a7b049dab2f67819da400f2c : SUCCESS in 50m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/15f2e09785bc4183a11cefebb182f1fb : SUCCESS in 12m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/30499323334e486cb3170be81d4967b8 : SUCCESS in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b395a903c575c9703baec5227a268ba0c32fce6a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-25 17:12:26.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-rxxwx/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"c1cecc70683d519a67218c35b0aa76ef2d83d94c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-04-26 07:11:44.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b66f217dbaf7a2f6a4c4443f0bbf49848833ece2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-26 11:00:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f18b7a1e74ab8914265afd68e41a3b97126ae281","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-26 11:13:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9315b7f2f41e2429916b764f3322d49e5bc07b54","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-26 11:19:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"caca5fdcdcde823e08b289405984990f896f41d8","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-26 11:39:32.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-jzn9c/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"ede24491f7d315b69c77c7508c200f0f0ae54613","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 11:45:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c482da2e9347e288402a0b4f3f1fc0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cdf72f6442954cdbb0863db6b9f862a3 : SUCCESS in 25m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4f341a23240c4b63ad9e21dadf7069f7 : SUCCESS in 25m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/475da1066bd74a179f45126335dfe087 : SUCCESS in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ace96b35be2508f7176f83d0cfd21b0c9d8da566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 12:55:12.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4eee549e548492e9d7410d81a20b147\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2603c01d974428dacc9e5c4fbac0b0e : SUCCESS in 28m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/db9c9c03694740c08b2fb40843e154fc : SUCCESS in 1h 05m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e605c074e4c4b1895f28b34b0594ca0 : SUCCESS in 32m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b585d27f4e0487699a44fd4d1c67647 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42272ed7aec4d0882559eb3d379f67a : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf5e95df10284e3c9eeb7817d67038d5 : SUCCESS in 8m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c92ee5086ee4554953e159d2e39f310 : SUCCESS in 10m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e5adc3118de458eb3b6d153bba34374 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e687f1b2aa4469a1d9fc4f436e0461 : SUCCESS in 9m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3e8b3f4dbcf34bdd86c0b2283bb4bea9 : SUCCESS in 44m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7f28ff023b747849862c96817ffccdb : FAILURE in 1h 21m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/063fa8f7f1064eceb7b186dd4e07def1 : SUCCESS in 34m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/66402fd675d14fa4a23b15e299919302 : SUCCESS in 26m 08s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c75b82ba7a914a43ba42ab55c456ec92 : SUCCESS in 1h 00m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a523bb1d40ba4f5581dbc2ec92c8d9f8 : SUCCESS in 1h 40m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/cb47b65c98e949198dafb7d42df0d9c3 : SUCCESS in 1h 36m 39s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/22fb98b9e59148beb8fb61130a6ca981 : SUCCESS in 21m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5e7fa68bb08940f1896d5447281ffe6e : SUCCESS in 4m 25s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ca8203e38d442fdac21a79b1392668a : SUCCESS in 21m 01s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59381c77af93465394a351d2a30da090 : SUCCESS in 21m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/34cffc97df6a47fe870f2b5d35c9d05f : SUCCESS in 25m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0cfe2c28ab341d19fd4539527df93bf : SUCCESS in 1h 02m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc87fd0f2dcb4c72be14075eefe266fd : SUCCESS in 39m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4092671f2213488eb710bb0fc05ae6e2 : SUCCESS in 23m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a40e50344955457e9566e71eb7f2d3d7 : SUCCESS in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d5b0ddaf700efb37646f86e4015698512de7d462","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-26 14:54:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"14bb56c2d375d22ada2da408e83cd189751fd7c1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-26 14:54:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ce20e6342c4ad964efe2b8e0a2959a394bc6c1a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 15:18:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20dd1569a2924b37a406252a7966672b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aae1bbcdbc664de19c1011d41968c49a : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c66deef69d7d4dac8968145c5ef07339 : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4350fc516ce4f8aa2866bc920944d64 : SUCCESS in 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a7d4e3f47200307c7b04eb0461d585c92ede58c1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-26 15:25:22.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-w8g9h/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"adbd8f7283e71440ff319574fe033a3daae924be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-26 16:28:24.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b941b149020414fa2507fd9ec39cc0c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/231049b50a294cc38e4d4a2bc81e7f59 : SUCCESS in 28m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3cf4f13bf5bc4be0986ac5001457c6ab : SUCCESS in 1h 02m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4c67ffeb13d4653ae3d97072221fdf0 : SUCCESS in 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b27f556fdbd040fda6cb251314c055c3 : SUCCESS in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ccb552c05194929811b37cb60ec9e99 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3373dad9aa794922b45884dad1566d07 : SUCCESS in 10m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c432c09010af443abcfcaaea68334350 : SUCCESS in 12m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79284f9cd51845ff9d83fb53559ec459 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1913570f41c14e2a923a174966da72ab : SUCCESS in 11m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/266e7f957e324545812aaaf25944de80 : SUCCESS in 32m 29s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ab07a39620e4fed8b238c36f7c6846e : FAILURE in 1h 00m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/634e583664ff437abc10a47e9354c183 : SUCCESS in 47m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/999e4b07a2234efbb15f633f05bd9539 : SUCCESS in 41m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5dac7509b10e46c9a2ca7208ed11076f : SUCCESS in 53m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4b8d51a5b9904ed4ae05e4ddae82b058 : SUCCESS in 1h 32m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/e8dd5f15e6b642cf9cafb700d647b912 : SUCCESS in 1h 29m 09s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/95aff47d78814de1882a29d7834f1030 : SUCCESS in 30m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8b1ef084efd24b57af01dd11879da3e2 : SUCCESS in 4m 42s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/699063498bcc4d01bb0ea3eb1c4afa27 : SUCCESS in 21m 52s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9337edf68712422cabfd2589cea5d775 : SUCCESS in 23m 05s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fde9044721a64da7aaffebbcff180d6b : SUCCESS in 25m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e67609ad55d04776a704818170e2561e : SUCCESS in 1h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a489bec9d0b040fc9fd3f43851fbca5d : SUCCESS in 42m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/afb395c5a54b41c8950d24f3396dbe53 : SUCCESS in 19m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1b78ab6fc41e401387957ea616add85f : SUCCESS in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6bdcf4aca4839f2ab55577622fb51a3cec13d95d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-27 11:01:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"feaf810abf635f5af99f54bb28f2c03a0a13b933","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-27 11:27:01.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-tt675/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"3e16d419fecf51f4623403839312772367073618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 11:36:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7530632b4d6a42ba8c8dccbb3ab7b786\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5265eafe6904718b68e0d0d390ec376 : SUCCESS in 27m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0a70ea972133439589165495a56f8c02 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb50113ba4d84a4b81a2c29cdb29eedb : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"48ca64d5af9cf9d2bdfbc92e2969621cf682cc1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 12:57:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e283efe9141d451fb6e085d8d9d30d9a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3420e4c33bcc45debbc861097f0d4ef1 : SUCCESS in 42m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e21b853b903d41d09d6f58e23b4d5529 : SUCCESS in 59m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f5ff1ccaebc48439ca58df7e5b87968 : SUCCESS in 45m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3815fc3d3834861bccdf5989120ffe3 : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b4a74b484e44ccbfdc57a977f0c76a : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/176646e8ba0f4f5c820837ec07166365 : SUCCESS in 8m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bffb37fe7d0c49f3bcb1c7c40e1ce378 : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f21acfda63d4d0f8afbc9be7b493ad6 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4755686ed334c12812aef309578ac7e : SUCCESS in 6m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f743cde15f904fbca157700597c330d5 : SUCCESS in 43m 37s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5f7dccf493cf4b9d87d87a8f0a052192 : FAILURE in 1h 33m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8d808bc2792c424c95360419144cda79 : SUCCESS in 39m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/211f2bdb730947999648cd6c7895cd31 : SUCCESS in 41m 56s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e6c23783f10f4e6db94ce42e66cea0ce : SUCCESS in 44m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/124e81aeb3144ea887d44d7a68a143c4 : SUCCESS in 1h 34m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/dc1da5cc65af49a1aed13c79a249f5d8 : SUCCESS in 1h 41m 17s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/0eba71edbe5f43b5ac805d98716f22c3 : SUCCESS in 27m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c243a0dde59842dc85c8fe6430e070b4 : SUCCESS in 4m 41s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48fdce1dd6474bbfaf06129cf1c99584 : SUCCESS in 21m 38s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4320d228d9447959d2f25489be56403 : SUCCESS in 21m 18s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f1c8bc35d18e4e94848f3fb8e7f130ed : SUCCESS in 24m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e36f331f94c4fe69f161756a6288d96 : SUCCESS in 58m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c999e5a1bcb94f9facfbff1e95b0f70a : SUCCESS in 41m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/67cda76bb47d4356b9b2d50f43e6f950 : SUCCESS in 11m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a762329e561b4e239ea0b2a0c08b1d6d : SUCCESS in 23m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9d233f8c49a8918ef1a0c9a04c23f77081b82cda","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-27 14:24:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"65c86458370f3ccad053512de740621b9d09d3d3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-27 14:54:22.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-zbndt/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"d84762d25fa8dd9786fd61896402627fef9c8cd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 14:59:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918eeb06d19a4201be65d521cdd2a56c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28c56506eb6a4952a905f6d252b736e2 : SUCCESS in 27m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf4ec7265514437d8e6e582142ad1861 : SUCCESS in 27m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8bbdf0b472964d24b25be35b6b9f66b9 : SUCCESS in 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4ec6e2945f89268dabaf2a4952429aedf8b5ec89","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-27 15:12:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2c5e80feb4c5eb9fe30a52563049b39995737c61","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-27 15:33:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"46c5e9307df42c18c981ddd72d06492a7b2930b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 16:40:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/474ba79bcc1148be8878fcdd88fec159\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74cda93e63eb4c639a505394d435bd17 : SUCCESS in 36m 59s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/87ec35a421674452b26e596c9a37d729 : SUCCESS in 1h 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0f7b7b30d4045aeb0039a429237a398 : SUCCESS in 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/794025e341874724be2cb59273f5c2ce : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a105bfbc03fe4600bc2e20fa071f0154 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0381c969c81844da8ff7bf82b7628c30 : SUCCESS in 10m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05785de571b942e1a02226be4e040b8d : SUCCESS in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3ae95b07cfbf4291947cc3262c9bf073 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3fb96ecf3547d3a7c87da23015149c : SUCCESS in 5m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7367927bfbf54abebc92998b8bd9055c : SUCCESS in 41m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36b738d0c6ac4e488a3a9d729e29f58b : SUCCESS in 1h 29m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/79d94649dba84383a39025fc7594d168 : SUCCESS in 36m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/812328a01c3846a9a65a269a2a4ae56b : SUCCESS in 23m 45s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/813b6197ffde44af8905dd5960bbd4fe : SUCCESS in 1h 04m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4a0e04304b92483fab26046fd30615eb : SUCCESS in 1h 40m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/c59e55591a2c4de3b24b5d3e2cb9b52e : SUCCESS in 2h 10m 05s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/1e0c9e45a97f4377ade7dfe5d3dc5c22 : SUCCESS in 30m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b1150a38d9da4953947eb0b9b658fc78 : SUCCESS in 4m 30s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69f1cb25101d47c6b70a9cc193c0b511 : SUCCESS in 21m 20s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1879a339d0f4e5f96acb2368f625d3b : SUCCESS in 20m 32s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e1050970bbfc483c9b5e27dd4823b548 : SUCCESS in 24m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/000c3b4e674a4a85858eac8ab2896eb0 : SUCCESS in 1h 00m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aef098111f8b44288e351865d88e34e6 : SUCCESS in 42m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b9abd44424c74ab2973486d0d0337b1c : SUCCESS in 10m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/53066a1aef214788a786d8497e375d97 : SUCCESS in 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2623db2018b8e2381a6c41da27539b9b2e563ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 16:41:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9e9a749feb67acb53f17f49d6b82a831ab8a4535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 18:21:37.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b890e625839432b86d1074fce8d5ad5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3baacdb065942b784322c9b37c62838 : SUCCESS in 38m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3d9006082af3476d96e659372daae5c9 : SUCCESS in 1h 26m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b869a7d95cc444eda5ae541fec727e39 : SUCCESS in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a41d97ff00440938f2ae80382c0c6b5 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2446d86cce04875b72d5844da9c5685 : SUCCESS in 11m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/637f3c518b174834b7fdb684dfc42e02 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fac5efd1ec94cecb21c915fcbf46643 : SUCCESS in 9m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b67ad3763bd945cd958fa3a1a9db2eb9 : SUCCESS in 40m 32s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6410887d62b545bea33ac78cff058055 : SUCCESS in 23m 53s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20b44239596f461998a86c47bb61f070 : SUCCESS in 23m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4b6caaf4e0384508bf365265c41475e9 : SUCCESS in 27m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1c68ecea7e424531a1124da7f69976d1 : SUCCESS in 1h 02m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/d8ede4f74270433681d29c79ad161bbd : SUCCESS in 1h 35m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7f434d81c1f9446e985a356de6401b81 : SUCCESS in 4m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fff562c5959841b1948aa2fcf3a36ce4 : FAILURE in 39m 38s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/732551be9c534f129262361967f8d24a : SUCCESS in 48m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df0475003b574112baf6a58c057ba807 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":6},{"id":"41ed67592df649dfd88d5e498afcdbf2912e5ddc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-27 19:13:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"983520a43dd522d66a79692f2b67fb40b28ccffb","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-27 19:43:23.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-fmxb9/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"b22e1f2f22deeb3049b38ec4484f24527d0a2d93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 19:48:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a952f7b79564fd0b8b0db09ff9d5e0d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/519dbb7f7bee440a8e9931978ba72187 : SUCCESS in 31m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2f303ace98f4aaaa2c4ec556618797b : SUCCESS in 32m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/843e4dc684ce41049e6122adad2d0018 : FAILURE in 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d6847793ec579ec1f32e52db67bfb061a8072c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 20:52:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05014090fc6466aaf6b427e8f4794ae\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7bccd5002fb45d2aec5a9b4c902103e : SUCCESS in 41m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e5999c413292453b9c3a184ed28d56e0 : SUCCESS in 1h 27m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/586ae9cf01194329934c7bd2020c8e1c : SUCCESS in 46m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e83921ce240646a1a62cdd69eeebabf7 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0775183782a341fe872626bb256d00c3 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e996a30175a4b76b004b7353169becf : SUCCESS in 11m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68d13c90d1fd4880bfbc7e1d3fc6ca5f : SUCCESS in 11m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36b02197c2c947388e214d27fdeba682 : SUCCESS in 13m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65395458baba4b819185205d3c3ec665 : SUCCESS in 5m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/47c9271792f54bdb9022c7279c772ead : SUCCESS in 43m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8f3e2a4de9524606b94eadb552ab3ed9 : SUCCESS in 1h 38m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/102a72e1e1b24ed4b711f1e9676decf7 : SUCCESS in 38m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0d4d10aa753e4df1ae82f2557faa1228 : SUCCESS in 30m 05s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6ade44a53b124082a3e7a28d859aa09d : SUCCESS in 55m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1273939249e74b7695ea390ae9789423 : SUCCESS in 1h 10m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/8248eb88c77949e6b24b68b4ba11617a : SUCCESS in 1h 24m 26s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/878c5f36369f4019b83493dd08e27aa2 : SUCCESS in 40m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fa3fb336d4f047938a8500402834d0b2 : SUCCESS in 3m 01s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/725615a9d732496d9e0a1ef3c52534bb : SUCCESS in 27m 24s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a8e236870a684b3b9061551905350cd3 : SUCCESS in 24m 38s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1baccfc90d2c4282bd45d105aa899915 : SUCCESS in 33m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2128f8a202bb44cba12e660f39e38134 : SUCCESS in 50m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d96e380e28f4d16bb80fbdccf58930e : SUCCESS in 1h 04m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c9674bee4f224506a4dbd4fd8a0917f9 : FAILURE in 13m 15s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fba32c060e55419ab44ec4491cf3f8d4 : FAILURE in 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9fae8dc4334a40cdc1ff59a75c26c8f99e861c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 20:52:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ac70636e39c19980033e761ae0e95ee4a67aa661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-27 22:53:50.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62dbac27c782483eb9741a247347b351\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b88b28f44e01490e90bc18dc85abca8d : SUCCESS in 55m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a76ae32abc1b4acd843fd3fb0b9963d9 : SUCCESS in 1h 56m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ccbae54188949a78b14178a2b3b98df : SUCCESS in 1h 03m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d150665351d4ec0a86f2b6d71e15d55 : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17c8232dfb6c4ae8823a8867bf3e9ee3 : SUCCESS in 8m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b1ba202e33c4b1b8cd4afa51c7c5e16 : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5071222b4a814c9a8feb69e3f875fbb8 : SUCCESS in 6m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/745236ed29924120b14c2768e746e756 : SUCCESS in 27m 02s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce27d5dcaecc43da9bcf5a30beadbde3 : SUCCESS in 19m 55s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01f8b8435947494db65aacc499545e29 : SUCCESS in 20m 50s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9aa66512dc934e65a82ea1e131e5624f : SUCCESS in 27m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/588af48dcb8c4f25ba19a2bcb7c566e8 : SUCCESS in 1h 34m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/1a18af765fba4d15976be62707fecfda : SUCCESS in 1h 15m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3a6ee102def74026aa82fd2ec81aa8c4 : SUCCESS in 4m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/810c98edc501437c89b0bf6f38386c7b : SUCCESS in 1h 19m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ec4f49498514e3a9f7be2d3e15b43e2 : SUCCESS in 57m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6adf0e4cb70940f3ac15f8fe504382ae : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":6},{"id":"82c93c317503005987c22792fcf09946ab2cd7b9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-28 06:01:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a8c1d0a5ecedbfefd67fa041d59bb1b302949aee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 06:27:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a476b64f50084bee96706d933638abcd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7823ccc5490a4b94863d9626671d7df7 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/63b0bd28e58d43e2b58caaa1fca3de3b : SUCCESS in 19m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9c411a20ab7247f7ab0cd20b550a6b92 : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cbfc112a51ca7ed11a13533b14b33ac365709c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 08:09:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc820e34ce6445dfb15e043b57279fd8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69e8cae44d9a4d9483fe1f0153d87b10 : SUCCESS in 28m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e60e923a33d645519950ecc13d47632d : SUCCESS in 59m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2227c4deccd49ce837eb65681fad3e5 : SUCCESS in 32m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c18627c91cf54a57873132668d85d2bc : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1533b935426d4f92be128aa40ae3da24 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b48aab99f1b4989a7fe50b6335744e4 : SUCCESS in 8m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9694999ac2474da19449541da5ed1cdc : SUCCESS in 14m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d552acbc7f247ada5ec4e8cb92c5e57 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368117e4ea4a4fc592150519f383e1ea : SUCCESS in 12m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1d896801e84e46b3b2ef56d48a605b3f : SUCCESS in 28m 30s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/452e6a15ce834cb4919121e9345e171b : SUCCESS in 2h 02m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fd713fcf8aef40ba8fd50f3a8ad1bcc5 : SUCCESS in 25m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/31c66cc94df4440b85c9fec0ccc20d7b : SUCCESS in 43m 40s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f9c839e1f34444289eddb7267bb9bf42 : SUCCESS in 54m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7e6534b2b74d4940981fa0c97dbbbb2b : SUCCESS in 1h 46m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/86e60aab426f48feae879011b1ac7a8e : SUCCESS in 1h 46m 01s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/542dc6b1e0f441d293cfec2d9ee812cc : SUCCESS in 33m 08s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3c2948b24aa0403293f988f51b461cc4 : SUCCESS in 5m 39s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/97001dd516114e9c880ccd21a6f6be2e : SUCCESS in 22m 30s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d469218b1347434195d4d75cb6a6aeeb : SUCCESS in 22m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/11a28b9b2527430fb2794e3682d4211c : SUCCESS in 23m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/503f4db710f84867a3295ce78de9e269 : SUCCESS in 1h 06m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b222e937dc8646cbb58d64ec39843af0 : SUCCESS in 43m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/485b9aa6f5aa4c928e4d9c0e19ba47fd : FAILURE in 21m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/91939b720cfc4370810b4f5638ddd9c0 : FAILURE in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f89178d17a3f1ee11052b94a03e22c55e78f7948","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 08:10:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"407f05e3add0ef32b6574b8d4b6585396fc02216","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-04-28 08:44:40.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"454accf260af48589860938281d9bc05b56c39fe","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-28 10:12:36.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947212-7ghrr/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"0ac90dfb2337604ab7c385008d6c1299af7d164e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:17:54.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc23c7db6550492e9bdeb4928b399710\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34d5df176bee4fb79734c058158b163b : SUCCESS in 1h 09m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3aeaf8c1c89a4cdb9401987879e1beed : SUCCESS in 1h 52m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45abe5d2b1144168a399ca8c04a77e73 : SUCCESS in 54m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4eabb2f6844c6c930ce3f8667739d4 : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0825503113e481fa2e5296a438b1b78 : SUCCESS in 11m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b78ab4172e4c4348a5f1498e3e81de4c : SUCCESS in 16m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a8b249e9e3f45f9995efb1ef9718612 : SUCCESS in 11m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6213628715994ec9a19de59a2320524f : SUCCESS in 49m 54s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/debb4ce459954d708add797d55bc4097 : SUCCESS in 28m 14s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c9b1b5e3b5cc4be99a0198c28da6ad1a : SUCCESS in 25m 15s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d92fc1ae341f4e4ea2304d710c86c4b3 : SUCCESS in 30m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cf429054cfb34377b86e5171911d8b3b : SUCCESS in 2h 02m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/c31b3d80b6734851a380f65653c08fbd : SUCCESS in 1h 55m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/063ef5717dbc4b8293cfad3fb56de267 : SUCCESS in 4m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3368098ad0946cab74c510efeef893f : SUCCESS in 1h 55m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/92f839173a334b7eb47739fa9e11d285 : SUCCESS in 1h 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8afa0573966492e8d94cdc4071fac16 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":6},{"id":"00cdbe23b0436ed53baa6b57ae1dce34990b285a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:17:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3084185c50d9d80ef918709c76f5b550669abe39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:18:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee61de22d89a4319a9d956db8b9b632d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a133f9c1d62b4fb6903828705e381c83 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3946a94538840df90836ab18c42c6f56490efe7a","revisions":{"a95dcecda943cab9d2d76cb3292b9893058c434d":{"kind":"REWORK","_number":1,"created":"2025-04-15 11:45:39.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/12/947212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/947212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/947212/1"}}},"commit":{"parents":[{"commit":"1ad11b13884baeaa6ed9f8f5818f4d176f4d3134","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ad11b13884baeaa6ed9f8f5818f4d176f4d3134"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"subject":"Replace eventlet sleep with time.sleep","message":"Replace eventlet sleep with time.sleep\n\nAs part of the Eventlet removal, this patch replaces eventlet.sleep\nwith the equivalent time.sleep, which should work the same\nwith Python threads\n\nChange-Id: I31b1aa854d8c95e47ba476051a650937b739a52b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a95dcecda943cab9d2d76cb3292b9893058c434d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a95dcecda943cab9d2d76cb3292b9893058c434d"}]},"branch":"refs/heads/master"},"01154f34de915c1a3f142706b9300fd9e76acbc5":{"kind":"REWORK","_number":2,"created":"2025-04-18 10:14:46.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/12/947212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/947212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/947212/2"}}},"commit":{"parents":[{"commit":"1ad11b13884baeaa6ed9f8f5818f4d176f4d3134","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ad11b13884baeaa6ed9f8f5818f4d176f4d3134"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-18 10:14:32.000000000","tz":120},"subject":"Replace eventlet sleep with time.sleep","message":"Replace eventlet sleep with time.sleep\n\nAs part of the Eventlet removal, this patch replaces eventlet.sleep\nwith the equivalent time.sleep, which should work the same\nwith Python threads\n\nChange-Id: I31b1aa854d8c95e47ba476051a650937b739a52b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01154f34de915c1a3f142706b9300fd9e76acbc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01154f34de915c1a3f142706b9300fd9e76acbc5"}]},"branch":"refs/heads/master"},"74127f1c5328aae9b96faee745fb2f81fe857ede":{"kind":"REWORK","_number":3,"created":"2025-04-22 10:06:58.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/12/947212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/947212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/947212/3"}}},"commit":{"parents":[{"commit":"1ad11b13884baeaa6ed9f8f5818f4d176f4d3134","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ad11b13884baeaa6ed9f8f5818f4d176f4d3134"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-22 10:06:13.000000000","tz":120},"subject":"Replace eventlet sleep with time.sleep","message":"Replace eventlet sleep with time.sleep\n\nAs part of the Eventlet removal, this patch replaces eventlet.sleep\nwith the equivalent time.sleep, which should work the same\nwith Python threads\n\nChange-Id: I31b1aa854d8c95e47ba476051a650937b739a52b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74127f1c5328aae9b96faee745fb2f81fe857ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74127f1c5328aae9b96faee745fb2f81fe857ede"}]},"branch":"refs/heads/master"},"fd8ef6fecfcdf4d77aaa54f93acfd28940e46a03":{"kind":"REWORK","_number":4,"created":"2025-04-23 13:24:12.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/12/947212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/947212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/947212/4"}}},"commit":{"parents":[{"commit":"09e8f6e47ac63fbda8be848494060ecc44d7c93a","subject":"Merge \"Fix description of [pci] alias\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09e8f6e47ac63fbda8be848494060ecc44d7c93a"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-23 13:02:21.000000000","tz":120},"subject":"Replace eventlet sleep with time.sleep","message":"Replace eventlet sleep with time.sleep\n\nAs part of the Eventlet removal, this patch replaces eventlet.sleep\nwith the equivalent time.sleep, which should work the same\nwith Python threads\n\nChange-Id: I31b1aa854d8c95e47ba476051a650937b739a52b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd8ef6fecfcdf4d77aaa54f93acfd28940e46a03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd8ef6fecfcdf4d77aaa54f93acfd28940e46a03"}]},"branch":"refs/heads/master"},"4aa0409d8bd395279a4c0252f1540b001a1591bd":{"kind":"REWORK","_number":5,"created":"2025-04-24 11:14:06.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/12/947212/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/947212/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/947212/5"}}},"commit":{"parents":[{"commit":"09e8f6e47ac63fbda8be848494060ecc44d7c93a","subject":"Merge \"Fix description of [pci] alias\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09e8f6e47ac63fbda8be848494060ecc44d7c93a"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-24 11:13:46.000000000","tz":120},"subject":"Replace eventlet sleep with time.sleep","message":"Replace eventlet sleep with time.sleep\n\nAs part of the Eventlet removal, this patch replaces eventlet.sleep\nwith the equivalent time.sleep, which should work the same\nwith Python threads\n\nChange-Id: I31b1aa854d8c95e47ba476051a650937b739a52b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4aa0409d8bd395279a4c0252f1540b001a1591bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4aa0409d8bd395279a4c0252f1540b001a1591bd"}]},"branch":"refs/heads/master"},"3946a94538840df90836ab18c42c6f56490efe7a":{"kind":"REWORK","_number":6,"created":"2025-04-24 12:09:30.000000000","uploader":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"ref":"refs/changes/12/947212/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/947212/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/947212/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/947212/6"}}},"commit":{"parents":[{"commit":"09e8f6e47ac63fbda8be848494060ecc44d7c93a","subject":"Merge \"Fix description of [pci] alias\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09e8f6e47ac63fbda8be848494060ecc44d7c93a"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-15 11:40:08.000000000","tz":120},"committer":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2025-04-24 12:09:19.000000000","tz":120},"subject":"Replace eventlet sleep with time.sleep","message":"Replace eventlet sleep with time.sleep\n\nAs part of the Eventlet removal, this patch replaces eventlet.sleep\nwith the equivalent time.sleep, which should work the same\nwith Python threads\n\nChange-Id: I31b1aa854d8c95e47ba476051a650937b739a52b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3946a94538840df90836ab18c42c6f56490efe7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3946a94538840df90836ab18c42c6f56490efe7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
