)]}'
{"id":"openstack%2Fnova~947245","triplet_id":"openstack%2Fnova~master~I79b725f3b3569e9c1460a93ac40ca92269e7d003","project":"openstack/nova","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-04-17 18:21:57.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-04-17 15:51:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-04-22 17:38:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I79b725f3b3569e9c1460a93ac40ca92269e7d003","subject":"Remove WSGIService and WSGIServer classes","status":"MERGED","created":"2025-04-15 13:39:55.000000000","updated":"2025-04-22 17:39:23.000000000","submitted":"2025-04-22 17:38:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947245-eventlet-removal","meta_rev_id":"e64c3f4ebdc881da2bb65376ac87c1bff265eab9","_number":947245,"virtual_id_number":947245,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 17:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2025-04-17 18:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2025-04-17 15:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2025-04-17 18:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 14:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 14:39:53.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2025-04-15 14:46:29.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 16:51:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-17 10:28:37.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-04-17 18:21:57.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"1fe7bfed45c993e525445be491d49512814b9408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 13:39:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c33bfe8ba248d3da1fc4e9fd784b0250485c9f2a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 13:40:29.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89562a59c3ec3c56474a8b1abf4e4d725d09dcf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e592ea4b714bd29210c207a5f071d0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dbb1c5cd61cc4b1dbc269baea0d787c2 : SUCCESS in 25m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b50cc46b294741869a1d4daadbf9a999 : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8398a712be0a4348917237fb39d301f2 : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acafb7ac9c8269d516d2ed0baafe43ba659ab340","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 14:30:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8f249054e21d336049b24a10079227ca21c69da6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 14:37:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c2faa7f50f2f5bf710cab0686a4b73024888d2bf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-15 14:39:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"abd741d8f9a531b1d8a403278b97a9a708312487","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-15 14:46:29.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947245-4ndvf/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"a06ee1e9788fd33e98c2c2401ff01fc84fd3ef06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:57:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee97bfec1214417b9b47a7aeefb36bba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b048e74a8504ed4bee82792708a73c7 : SUCCESS in 14m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95828d1a921b49df808995ab48fac67e : SUCCESS in 15m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8f657decbd6411c9809944fc2886401 : SUCCESS in 18m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c17f5a998663b7656407929fc07b496a17153db6","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-15 15:56:08.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947245-xkjbb/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"5427d87e4daed7e25a0ff633e9d8b570dec3cef5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-15 16:27:39.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947245-s2v88/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"395d662e6fb593870a57641badc3b5caab12d233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 16:51:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68bbef2ffd9f4f5ab932f37d869019ee\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/499819578df04666827181b9fd46256e : SUCCESS in 45m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6f01443def21459c9043a8c3873b195e : SUCCESS in 1h 40m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77b5dab7dafc44e5a49787e8e451d76e : SUCCESS in 59m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ae3e834b03c4ba39e8f2a5b8eeb6b32 : SUCCESS in 26m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf7377344a14e199b31fa943c1d58bc : SUCCESS in 7m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7dbf7d7b4b41419f83df56444202470a : SUCCESS in 19m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/550bbf6fc9054e148c3e4af7d2bfd342 : SUCCESS in 13m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c41dd8fcac74b24a5fba4e028b29a2a : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4824655a1d994924b1f3a7f92c485d8a : SUCCESS in 11m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c9e1264dc5624d1a8c62f7319e42b896 : SUCCESS in 56m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e04d0b26fce1475197a7060c2c7cd570 : SUCCESS in 1h 41m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/716a8c6497b449b3b291c2b9e6b7a8dd : SUCCESS in 1h 02m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/75b98f05a1c14e1d803396bdfa83b034 : SUCCESS in 28m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2a00a330035e4398bfa1645ff977b5af : SUCCESS in 2h 01m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/a4bdce0311c04b60ba3b220ff9892db6 : SUCCESS in 2h 01m 43s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/10ccfc7b3ca4415894f54dc6e69217b0 : SUCCESS in 30m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c6abf26bf28c496db228df084440e2e0 : SUCCESS in 7m 17s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ec6d276debd4a74a2891ab3b3bccec6 : SUCCESS in 26m 02s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/497495a24fdd48ddb24ae9f2a5965fc1 : SUCCESS in 24m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c81da5bd14504cf6869f734e1460d51a : SUCCESS in 25m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c65baa3851f4befa6be586567eb2130 : SUCCESS in 1h 01m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9efd5392cf4c4b9392f6e60a4e4d2eb2 : SUCCESS in 49m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/990b4153713147d7a54e69acc83e4cf6 : SUCCESS in 26m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/14dfab8ad3544ec19c1641ae1731cfd3 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e7ebd8d86c07693cc120efd86e38d544e03c8a5f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-16 19:13:58.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5b205891dc141af1b76f53a103d6fce952fc2a46","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-17 10:28:37.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9938c40a9a38a24576cbb82eecb890fe8479ab77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-17 13:30:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"182fc716395f7684b16efce5aa5c0a372bcddec9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-17 13:30:51.000000000","message":"Patch Set 3:\n\n(13 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5b7ffffc59e13ce89434fc0632e75a5bdc62cebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 13:51:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8dfe1492a4948ea9ad97e37b1a4ce37\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/250ed849b55745bfa14000b0c1787739 : SUCCESS in 18m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7fa51fc00ce8469ba8f798a48c580258 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7505a530b99a484c8cc7872c392cabf8 : SUCCESS in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e005a75a3930897294cedc78947aadbfc244af9c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-04-17 14:01:49.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-947245-xvhpr/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"ac0d994e29bed4b1a94ec280f4a6e291f4adda14","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-04-17 14:49:54.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fd9e5cf9359a87329dd216c5dac9d5ef502c3996","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-04-17 15:51:15.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"18348128ddacaf3c5af3ce513601ad200a2aefd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 15:56:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5376a71bb84ba79205dd2b6c8a3d33\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a8b0d9e8488141a99b539a92904b6106 : SUCCESS in 31m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1daef5c1aa414b51902c72eee6411224 : SUCCESS in 1h 49m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b92b42f1f2c4fa5a11ae8b4dadbc499 : SUCCESS in 56m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ebef0aa3224b7fa2c8f0631dace863 : SUCCESS in 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395e26ee629f479b9be2f3ee9b48d6d7 : SUCCESS in 7m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cb615fd66504cb1b3734fa2893e0381 : SUCCESS in 12m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bffdc7f5712c43e796b589827f5d8cc6 : SUCCESS in 13m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/472479c0ea924c8eb27fb8d554d2e8b5 : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d6e56a8fae4b92b458333a8ceeea3b : SUCCESS in 11m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4b9451ee1a5e4a60a754958520f1b7cb : SUCCESS in 27m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6bf438339d764bf8a5c95e632c52d93c : SUCCESS in 1h 59m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f841db3ee5584c3181e525d15ed4c379 : SUCCESS in 23m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/843c720417b74a029eef40dbc9158945 : SUCCESS in 52m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/11f1da6eacfd491cbb36c47d729bca72 : SUCCESS in 2h 08m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/77daade65ad7482a8783d477ff49fbf5 : SUCCESS in 2h 20m 40s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/6000187bcbb04bff9eaaba0a9aa0e6a2 : SUCCESS in 26m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ef861d5973ba4ec8b6e0443fe029908a : SUCCESS in 4m 40s (non-voting)\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e1e32b8008749c8b4670c7224ae4cd9 : SUCCESS in 26m 26s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0818b85a81d4efb8bfe23f2a9722822 : SUCCESS in 25m 31s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/76b960822e0247eab0c61ea59b6ea4a1 : SUCCESS in 34m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d6aad1c78f347d5ada834f10f620938 : SUCCESS in 50m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/917546ac87c444409e234026a93e89bb : SUCCESS in 44m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/28bdba83f3d14b6c96fa3114e18f6903 : SUCCESS in 19m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/968616e943c34212a89ebcdec0bddc14 : SUCCESS in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"068fbafcc17548d7a9e2bfc3f9e40242ff2df01f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-04-17 18:21:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"49d3faba2b605811edd41cd0dae6681a9c35070f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 15:18:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e81ed1630f6c2dbe91c72358b895985ffd4a8a7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:38:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3bb517d4c3e4b5399801d593ed8c109\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b62ec72ce7e141009c1ac651d0669ab9 : SUCCESS in 57m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f7c0be74b0304316afb87b25355abf92 : SUCCESS in 1h 34m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c710321d518e49d2a9f79c9d9c9e9f27 : SUCCESS in 47m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c53eb937cfb4bbcaade9e5e7a23eb9f : SUCCESS in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc7df64f2ed04bdfb1d1ff3caa160468 : SUCCESS in 13m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67a7981d66424bed888168eb7c90e505 : SUCCESS in 16m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06dd963c6b6437ca6ea0cf0bb6a0b8a : SUCCESS in 13m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bc520c33f2ee49878ebebbd45680c02a : SUCCESS in 44m 38s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc7bf673b676461f8c9b2711f50f2bda : SUCCESS in 27m 38s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2809fdc335cb4ff98cfe30f3a86badc4 : SUCCESS in 29m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/812ea69ae88e4fecb6789587bf1723e9 : SUCCESS in 27m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/581202068441441187d8179cd91b68ec : SUCCESS in 1h 54m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/e5f1e6faa26244c3ac08ebde3b46289e : SUCCESS in 2h 14m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/47370ec2b5414a36b00b1e3261040a99 : SUCCESS in 4m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/128dab699bef4004b02cc6bfcebd3299 : SUCCESS in 1h 49m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a93529e36fc48c9a85114e38f3ab1fd : SUCCESS in 1h 19m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fc5a8df363b42b2bb92ce4acbfe398e : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":4},{"id":"316581d8631321b2f6c60850a7c5b84fbe21749f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:38:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e64c3f4ebdc881da2bb65376ac87c1bff265eab9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:39:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4d31a09f424d08aa45275b783486ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9285d93f8447ed91b4f76709c87516 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"51eb60063fa09e742508af18da3513c5fabb7de8","revisions":{"4c9b6be87a9bfe50d6a212f604b6def41264bc57":{"kind":"REWORK","_number":1,"created":"2025-04-15 13:39:55.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/947245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/45/947245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/45/947245/1"}}},"commit":{"parents":[{"commit":"05b219746f38a2b4ad7a44eb3125111d417ca303","subject":"Remove eventlet based WSGI server entry points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05b219746f38a2b4ad7a44eb3125111d417ca303"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 13:12:56.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 13:39:42.000000000","tz":120},"subject":"Remove WSGIService and WSGIServer classes","message":"Remove WSGIService and WSGIServer classes\n\nThe previous patch[1] removed the Eventlet based WSGI entry points, and\nthat code was the only real user of the in tree WSGIService and\nWSGIServer classes, we can remove those too. This removes a good chunk\nof eventlet dependency from our tree.\n\nThere is a catch though. The functional test env used these to start the\nnova-metadata-api service. We need to re-implement the fixture to use\n\u003cput tech here\u003e instead.\n\n[1] Ie758550c0b8fb02aeb398396961467d9f845fcc9\n\nChange-Id: I79b725f3b3569e9c1460a93ac40ca92269e7d003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c9b6be87a9bfe50d6a212f604b6def41264bc57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c9b6be87a9bfe50d6a212f604b6def41264bc57"}]},"branch":"refs/heads/master"},"1cbe749a74b9816d86bee31309651fa1796c9628":{"kind":"REWORK","_number":2,"created":"2025-04-15 14:30:03.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/947245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/45/947245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/45/947245/2"}}},"commit":{"parents":[{"commit":"05b219746f38a2b4ad7a44eb3125111d417ca303","subject":"Remove eventlet based WSGI server entry points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05b219746f38a2b4ad7a44eb3125111d417ca303"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 13:12:56.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 14:28:08.000000000","tz":120},"subject":"Remove WSGIService and WSGIServer classes","message":"Remove WSGIService and WSGIServer classes\n\nThe previous patch[1] removed the Eventlet based WSGI entry points, and\nthat code was the only real user of the in tree WSGIService and\nWSGIServer classes, we can remove those too. This removes a good chunk\nof eventlet dependency from our tree.\n\nThere is a catch though. The functional test env used these to start the\nnova-metadata-api service. We re-implemented the fixture to use\nload the wsgi app and use the wsgi intercept instead. This also showed\nthat while the Eventlet based API service could be reset via the\noslo.service interface the wsgi APP based API service cannot. So the\nrelated cell caches reset testing is removed.\n\n[1] Ie758550c0b8fb02aeb398396961467d9f845fcc9\n\nChange-Id: I79b725f3b3569e9c1460a93ac40ca92269e7d003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cbe749a74b9816d86bee31309651fa1796c9628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1cbe749a74b9816d86bee31309651fa1796c9628"}]},"branch":"refs/heads/master"},"b1d32e5e53f1cc6ee26bf7dd6f4f9ef3b84843ff":{"kind":"REWORK","_number":3,"created":"2025-04-15 14:37:14.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/947245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/45/947245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/45/947245/3"}}},"commit":{"parents":[{"commit":"05b219746f38a2b4ad7a44eb3125111d417ca303","subject":"Remove eventlet based WSGI server entry points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05b219746f38a2b4ad7a44eb3125111d417ca303"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 13:12:56.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 14:37:09.000000000","tz":120},"subject":"Remove WSGIService and WSGIServer classes","message":"Remove WSGIService and WSGIServer classes\n\nThe previous patch[1] removed the Eventlet based WSGI entry points, and\nthat code was the only real user of the in tree WSGIService and\nWSGIServer classes, we can remove those too. This removes a good chunk\nof eventlet dependency from our tree.\n\nThere is a catch though. The functional test env used these to start the\nnova-metadata-api service. We re-implemented the fixture to use\nload the wsgi app and use the wsgi intercept instead. This also showed\nthat while the Eventlet based API service could be reset via the\noslo.service interface the wsgi APP based API service cannot. So the\nrelated cell caches reset testing is removed.\n\n[1] Ie758550c0b8fb02aeb398396961467d9f845fcc9\n\nChange-Id: I79b725f3b3569e9c1460a93ac40ca92269e7d003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1d32e5e53f1cc6ee26bf7dd6f4f9ef3b84843ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1d32e5e53f1cc6ee26bf7dd6f4f9ef3b84843ff"}]},"branch":"refs/heads/master"},"51eb60063fa09e742508af18da3513c5fabb7de8":{"kind":"REWORK","_number":4,"created":"2025-04-17 13:30:38.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/45/947245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/45/947245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/45/947245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/45/947245/4"}}},"commit":{"parents":[{"commit":"05b219746f38a2b4ad7a44eb3125111d417ca303","subject":"Remove eventlet based WSGI server entry points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05b219746f38a2b4ad7a44eb3125111d417ca303"}]}],"author":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-15 13:12:56.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-04-17 13:29:59.000000000","tz":120},"subject":"Remove WSGIService and WSGIServer classes","message":"Remove WSGIService and WSGIServer classes\n\nThe previous patch[1] removed the Eventlet based WSGI entry points, and\nthat code was the only real user of the in tree WSGIService and\nWSGIServer classes, we can remove those too. This removes a good chunk\nof eventlet dependency from our tree.\n\nThere is a catch though. The functional test env used these to start the\nnova-metadata-api service. We re-implemented the fixture to use\nload the wsgi app and use the wsgi intercept instead. This also showed\nthat while the Eventlet based API service could be reset via the\noslo.service interface the wsgi APP based API service cannot. So the\nrelated cell caches reset testing is removed.\n\n[1] Ie758550c0b8fb02aeb398396961467d9f845fcc9\n\nChange-Id: I79b725f3b3569e9c1460a93ac40ca92269e7d003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51eb60063fa09e742508af18da3513c5fabb7de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51eb60063fa09e742508af18da3513c5fabb7de8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
